)]}'
{"id":"openstack%2Fnova~647957","triplet_id":"openstack%2Fnova~master~I1c7e2d5ff2e7b5158f8f2c4b2ebce5c20e99a984","project":"openstack/nova","branch":"master","topic":"bug/1821824","hashtags":[],"change_id":"I1c7e2d5ff2e7b5158f8f2c4b2ebce5c20e99a984","subject":"Adding tests to demonstrate bug #1821824","status":"MERGED","created":"2019-03-27 00:59:28.000000000","updated":"2019-04-02 18:58:30.000000000","submitted":"2019-04-02 16:23:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"647957-1554222228158-3bab15b6","meta_rev_id":"531de6fbd893e36b09a5993899f20d9aba7b523a","_number":647957,"virtual_id_number":647957,"owner":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-01 15:28:48.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-03-30 22:52:32.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-02 12:35:18.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-01 09:27:43.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-02 12:50:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2019-04-02 16:23:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-02 13:13:17.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"date":"2019-04-02 13:04:37.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"date":"2019-04-02 10:09:36.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-04-02 09:56:25.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-04-01 08:02:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-03-30 20:30:41.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":1,"date":"2019-04-01 08:02:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-03-27 02:57:12.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2019-03-27 08:18:19.000000000","updated_by":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"reviewer":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"state":"REVIEWER"},{"updated":"2019-03-29 17:16:59.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-03-30 20:30:41.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-03-30 22:52:32.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-01 08:02:35.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-04-01 09:27:43.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-01 15:28:48.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-02 09:56:25.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-02 10:09:36.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-04-02 12:35:18.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-02 12:50:41.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-04-02 13:04:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-02 13:13:17.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-02 16:23:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad9f0647616eb9f21b9e6083ab3465677cfe2fd6","author":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"date":"2019-03-27 00:59:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a438ee35b5249b3c00a93fdca551ac0e0fe5ce57","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 01:00:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b8fabcdcb9d434684dc56be66e2ca8714679820","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-27 02:05:47.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29571 : FAILURE in 1h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"017ec8fe5ab3e905bcb41db741f14fe8e867c2fc","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-27 02:14:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/1/check/tempest-dsvm-full-xenial/5407d1b/ : SUCCESS in 1h 11m 28s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/1/check/tempest-dsvm-full-xenial-py3/54a5632/ : SUCCESS in 1h 14m 58s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/1/check/grenade-dsvm-xenial/4face1b/ : SUCCESS in 1h 00m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ecd6aae3f65af336ccd09f9d1d4e334d5d245ed","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-27 02:16:39.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/647957/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5d86800 : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"359bc491b349d57e0e07b6b1bbe04746df14ab91","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-27 02:54:19.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/647957/1/check/nova-out-of-tree-pvm/3521027 : FAILURE in 1h 52m 08s\n- nova-in-tree-pvm http://184.172.12.213/57/647957/1/check/nova-in-tree-pvm/1de89b8 : SUCCESS in 1h 54m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3d6f2105aaf9a5b83d526182370b4755120c6516","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-03-27 02:57:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nUnit tests for this change should be added.","accounts_in_message":[],"_revision_number":1},{"id":"23525086b2cf6548285ce2a76a344606d251ae8f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-27 03:05:10.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/647957/1 : FAILURE in 2h 05m 00s","accounts_in_message":[],"_revision_number":1},{"id":"7d63ed77ffafd6808d8694b43b6f3e8ac7aa8430","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-27 03:23:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/57/647957/1/check/grenade-py3/f44a03c/ : SUCCESS in 57m 19s\n- tempest-full-py3 http://logs.openstack.org/57/647957/1/check/tempest-full-py3/217847d/ : SUCCESS in 1h 33m 25s\n- openstack-tox-cover http://logs.openstack.org/57/647957/1/check/openstack-tox-cover/511404b/cover/ : SUCCESS in 15m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/1/check/openstack-tox-lower-constraints/2aabf8b/ : SUCCESS in 13m 44s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/1/check/openstack-tox-pep8/73747cb/ : FAILURE in 16m 15s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/1/check/openstack-tox-py27/ff837ba/ : SUCCESS in 12m 00s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/1/check/openstack-tox-py35/2598396/ : SUCCESS in 14m 49s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/1/check/openstack-tox-py36/a303b76/ : SUCCESS in 11m 57s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/1/check/openstack-tox-py37/1d18d76/ : SUCCESS in 12m 27s\n- openstack-tox-docs http://logs.openstack.org/57/647957/1/check/openstack-tox-docs/8bc0e17/html/ : SUCCESS in 6m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/362bd05/ : SUCCESS in 48m 03s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/1/check/devstack-plugin-ceph-tempest/29eebc0/ : SUCCESS in 1h 29m 16s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/1/check/neutron-grenade-multinode/d202f94/ : SUCCESS in 1h 08m 05s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/1/check/nova-grenade-live-migration/05cfee0/ : SUCCESS in 1h 00m 45s\n- nova-live-migration http://logs.openstack.org/57/647957/1/check/nova-live-migration/6689ef4/ : SUCCESS in 45m 44s\n- nova-next http://logs.openstack.org/57/647957/1/check/nova-next/f3ff7b4/ : SUCCESS in 1h 36m 23s\n- nova-tox-functional http://logs.openstack.org/57/647957/1/check/nova-tox-functional/50c8fd2/ : SUCCESS in 21m 44s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/1/check/nova-tox-functional-py35/a0561c4/ : SUCCESS in 19m 03s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/1/check/tempest-slow-py3/0ff390f/ : SUCCESS in 2h 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"2fe8bebb799797a6dd438a07191b675c9debd208","author":{"_account_id":15888,"name":"Zhenyu Zheng","email":"zheng.zhenyu@outlook.com","username":"Kevin_Zheng"},"date":"2019-03-27 08:18:19.000000000","message":"Patch Set 1: Code-Review-1\n\nHmm interesting, seems true if a flavor only have forbindden traits will be ignored. -1 for pep8 and tests.","accounts_in_message":[],"_revision_number":1},{"id":"9b00440e2e042a9b46e977a158e50f3d7fed8d58","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-03-27 08:45:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91bfc1c47e1dc42a9ad6e3a616cfec11fd3ef777","author":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"date":"2019-03-27 12:57:35.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nWill get some tests together.","accounts_in_message":[],"_revision_number":1},{"id":"1f247ce8eac09f7a31d01203043d042f40ffbbe6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-27 13:45:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"419c41f18211e307f189e3c1bc0198798c010ec2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-27 20:08:45.000000000","message":"Patch Set 1:\n\nDiscussed this in IRC today [1]. TL;DR: Magnus to propose a change with a test case demonstrating the bug, and then stick this fix on top of it.\n\n[1] http://eavesdrop.openstack.org/irclogs/%23openstack-nova/%23openstack-nova.2019-03-27.log.html#t2019-03-27T15:15:28","accounts_in_message":[],"_revision_number":1},{"id":"ce79c1f261f47bca7eba379f70727a10ae50cb34","author":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"date":"2019-03-29 09:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1c29bdf49430e8f3d7335fa691f903b801fe07e5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 09:11:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"31c8a771c50aa352837046c3e685371ddf983428","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-03-29 09:47:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nThank you for working on this. Your tests seems good to me. But we need to complicate the things a bit to be able to merge the reproduction test.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":2},{"id":"98f419554880604eed966688f19513ff890f27c3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 10:31:13.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/2/check/tempest-dsvm-full-xenial/e200986/ : SUCCESS in 1h 11m 53s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/2/check/tempest-dsvm-full-xenial-py3/75c26fa/ : SUCCESS in 1h 19m 46s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/2/check/grenade-dsvm-xenial/e0a9f09/ : SUCCESS in 58m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"40d653571992da258a16138fd88b0ddf5e4c613a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 11:32:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/57/647957/2/check/grenade-py3/5369ead/ : SUCCESS in 57m 43s\n- tempest-full-py3 http://logs.openstack.org/57/647957/2/check/tempest-full-py3/272ec10/ : SUCCESS in 1h 34m 45s\n- openstack-tox-cover http://logs.openstack.org/57/647957/2/check/openstack-tox-cover/08fb348/ : FAILURE in 12m 47s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/2/check/openstack-tox-lower-constraints/0555c16/ : FAILURE in 14m 20s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/2/check/openstack-tox-pep8/ba1d276/ : SUCCESS in 9m 09s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/2/check/openstack-tox-py27/5aa19b3/ : FAILURE in 11m 52s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/2/check/openstack-tox-py35/9465873/ : FAILURE in 12m 19s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/2/check/openstack-tox-py36/aaaa354/ : FAILURE in 11m 24s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/2/check/openstack-tox-py37/889ed4e/ : FAILURE in 12m 59s\n- openstack-tox-docs http://logs.openstack.org/57/647957/2/check/openstack-tox-docs/28ceeb6/html/ : SUCCESS in 7m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f7807e6/ : SUCCESS in 51m 50s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/2/check/devstack-plugin-ceph-tempest/075b8ae/ : SUCCESS in 1h 24m 41s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/2/check/neutron-grenade-multinode/c241b12/ : SUCCESS in 1h 05m 15s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/2/check/nova-grenade-live-migration/167b250/ : SUCCESS in 1h 04m 44s\n- nova-live-migration http://logs.openstack.org/57/647957/2/check/nova-live-migration/3d74443/ : SUCCESS in 45m 11s\n- nova-next http://logs.openstack.org/57/647957/2/check/nova-next/aacecc3/ : SUCCESS in 1h 22m 00s\n- nova-tox-functional http://logs.openstack.org/57/647957/2/check/nova-tox-functional/6ea5d07/ : FAILURE in 18m 41s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/2/check/nova-tox-functional-py35/fc7ff06/ : FAILURE in 20m 35s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/2/check/tempest-slow-py3/35b4807/ : SUCCESS in 2h 13m 56s","accounts_in_message":[],"_revision_number":2},{"id":"014dea9afbdd9039b8f0564ae6a8b0db70b44898","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-29 12:24:05.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/647957/2/check/nova-out-of-tree-pvm/68369b0 : FAILURE in 1h 52m 54s\n- nova-in-tree-pvm http://184.172.12.213/57/647957/2/check/nova-in-tree-pvm/7bac306 : FAILURE in 2h 12m 09s","accounts_in_message":[],"_revision_number":2},{"id":"4045c129c00b543568fb4f22fa81e0f3fbd2c12e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 13:15:07.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29687 : SUCCESS in 2h 25m 33s","accounts_in_message":[],"_revision_number":2},{"id":"23b06e19d36841bed488026fdabfe0c2befc3554","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-29 13:17:55.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/647957/2 : FAILURE in 2h 19m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a26ff7eceea304566cf22f3acc626fdc50c7252d","author":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"date":"2019-03-29 15:18:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"54a1cf82664137f236c1c68ced29acca3b258c26","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 15:18:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f8ac8e3683b78830cb27e410f58aa9776f727b1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-03-29 15:26:29.000000000","message":"Patch Set 3: Code-Review+2\n\nGood job! Looks good to me!\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":3},{"id":"1c2abdd13ebe2915171eeaa7c20dc8e4d256f0b9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-29 16:06:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"980221af3d7268d782431d9710c3c3e85de44785","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-29 16:08:20.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22047930e5615eb80b1148fa676607b8f4317920","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-29 16:13:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e9d703d8ed2ed207d4792b4b7fcd22b8d8c0e798","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-29 16:25:52.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29701 : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":3},{"id":"129c7b8208f5ab12733980ec4eb0b25314ff1b33","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-03-29 17:16:59.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/57/647957/3/check/nova-out-of-tree-pvm/38760fc : FAILURE in 1h 58m 10s\n- nova-in-tree-pvm http://184.172.12.213/57/647957/3/check/nova-in-tree-pvm/d36a72f : FAILURE in 1h 46m 38s","accounts_in_message":[],"_revision_number":3},{"id":"8f1f831d9ad4696a1ec8c6f67ebd0956c98748d6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-29 17:33:15.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/647957/3 : FAILURE in 2h 14m 20s","accounts_in_message":[],"_revision_number":3},{"id":"04779621f41785805d9371c97a1c5c92dbfdcb96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-29 17:38:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/57/647957/3/check/grenade-py3/efddf86/ : SUCCESS in 1h 05m 52s\n- tempest-full-py3 http://logs.openstack.org/57/647957/3/check/tempest-full-py3/bf4c172/ : SUCCESS in 1h 27m 50s\n- openstack-tox-cover http://logs.openstack.org/57/647957/3/check/openstack-tox-cover/b3f7507/cover/ : SUCCESS in 18m 03s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/3/check/openstack-tox-lower-constraints/6cd3b1d/ : SUCCESS in 14m 06s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/3/check/openstack-tox-pep8/703bafc/ : SUCCESS in 10m 16s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/3/check/openstack-tox-py27/e9155dd/ : SUCCESS in 11m 40s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/3/check/openstack-tox-py35/e35b285/ : SUCCESS in 13m 44s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/3/check/openstack-tox-py36/49bdcd6/ : SUCCESS in 12m 11s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/3/check/openstack-tox-py37/9056063/ : SUCCESS in 13m 07s\n- openstack-tox-docs http://logs.openstack.org/57/647957/3/check/openstack-tox-docs/fe2b55b/html/ : SUCCESS in 6m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/3/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8183857/ : SUCCESS in 51m 10s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/3/check/devstack-plugin-ceph-tempest/46a1546/ : SUCCESS in 1h 40m 12s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/3/check/neutron-grenade-multinode/05f3217/ : SUCCESS in 1h 09m 54s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/3/check/nova-grenade-live-migration/6941066/ : SUCCESS in 1h 08m 04s\n- nova-live-migration http://logs.openstack.org/57/647957/3/check/nova-live-migration/0e40361/ : SUCCESS in 52m 37s\n- nova-next http://logs.openstack.org/57/647957/3/check/nova-next/d963bbc/ : SUCCESS in 1h 31m 12s\n- nova-tox-functional http://logs.openstack.org/57/647957/3/check/nova-tox-functional/4b23559/ : SUCCESS in 19m 24s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/3/check/nova-tox-functional-py35/31769a8/ : SUCCESS in 23m 01s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/3/check/tempest-slow-py3/1931f23/ : SUCCESS in 2h 13m 25s","accounts_in_message":[],"_revision_number":3},{"id":"b5f149689a23dacdc303102de9f66883941d87b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 17:38:58.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/647957/3/check/EMC_VxFlexOS_NOVA/e860cfb/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"59e0357dc19c373962abc3520d2dd85daf6cd3e6","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-03-29 18:43:05.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-647957-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"51306817d69e35b79742cdf9ea54147dd4ae31ef","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-29 21:56:28.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/3/check/tempest-dsvm-full-xenial/b957dad/ : SUCCESS in 1h 12m 38s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/3/check/tempest-dsvm-full-xenial-py3/7c51bb1/ : SUCCESS in 1h 21m 25s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/3/check/grenade-dsvm-xenial/02bc9b6/ : SUCCESS in 1h 00m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf3c9ea7b69216f53167cfbb4a33991ae85ad66d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-29 23:52:14.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/647957/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/884e563 : FAILURE in 20m 38s","accounts_in_message":[],"_revision_number":3},{"id":"ab39eab38a3450f7b6ba237e9b3228e5fad1494e","author":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"date":"2019-03-30 17:27:54.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e4b57f1700f46300fe28560271352a7d0776dbbe","author":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"date":"2019-03-30 18:24:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"59d5c1b559b588cd974e5f29c05b03be2eeb28fc","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-30 18:24:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f2faad1c6c5cc66330180f30d1b7f50b0627807b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-30 19:28:48.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29731 : SUCCESS in 1h 04m 02s","accounts_in_message":[],"_revision_number":4},{"id":"70aa498c0097f66753c4258a9134eac2f2aba4e7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-30 19:42:20.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial/d147e4a/ : SUCCESS in 1h 13m 13s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial-py3/408e088/ : SUCCESS in 1h 17m 21s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/grenade-dsvm-xenial/d71b391/ : SUCCESS in 57m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4da82cc791180e4a55fde5acaa0b5b87276ef3c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-30 20:30:41.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks for the updates Magnus.","accounts_in_message":[],"_revision_number":4},{"id":"67d73f5f885b193397b52e58d7701d3acb3a6c12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-30 20:37:04.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/57/647957/4/check/grenade-py3/211b25b/ : SUCCESS in 44m 51s\n- tempest-full-py3 http://logs.openstack.org/57/647957/4/check/tempest-full-py3/85d3ea8/ : SUCCESS in 1h 46m 34s\n- openstack-tox-cover http://logs.openstack.org/57/647957/4/check/openstack-tox-cover/170823e/cover/ : SUCCESS in 20m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/4/check/openstack-tox-lower-constraints/f0952e0/ : SUCCESS in 12m 09s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/4/check/openstack-tox-pep8/dd7a1d1/ : SUCCESS in 8m 58s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/4/check/openstack-tox-py27/e1c078a/ : SUCCESS in 13m 08s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/4/check/openstack-tox-py35/4cac9c9/ : SUCCESS in 13m 19s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/4/check/openstack-tox-py36/819943f/ : SUCCESS in 12m 31s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/4/check/openstack-tox-py37/40442a9/ : SUCCESS in 11m 46s\n- openstack-tox-docs http://logs.openstack.org/57/647957/4/check/openstack-tox-docs/a925610/html/ : SUCCESS in 6m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/16444ee/ : SUCCESS in 52m 26s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/4/check/devstack-plugin-ceph-tempest/b1dbfcf/ : SUCCESS in 1h 38m 30s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/4/check/neutron-grenade-multinode/8d16892/ : SUCCESS in 1h 44m 53s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/4/check/nova-grenade-live-migration/f5639d9/ : SUCCESS in 1h 03m 47s\n- nova-live-migration http://logs.openstack.org/57/647957/4/check/nova-live-migration/f52b73e/ : FAILURE in 46m 17s\n- nova-next http://logs.openstack.org/57/647957/4/check/nova-next/2802537/ : SUCCESS in 1h 21m 14s\n- nova-tox-functional http://logs.openstack.org/57/647957/4/check/nova-tox-functional/f541b5d/ : SUCCESS in 19m 09s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/4/check/nova-tox-functional-py35/09afd5f/ : SUCCESS in 18m 49s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/4/check/tempest-slow-py3/5b5d665/ : SUCCESS in 2h 06m 54s","accounts_in_message":[],"_revision_number":4},{"id":"56018a9e3c5a74f0f5dadd6c7acbc584716bffd7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-30 20:44:02.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/647957/4 : SUCCESS in 2h 18m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ef6763c922a85cb3b686d15f04679d1b23943f07","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-03-30 22:52:32.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/647957/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/07c1206 : SUCCESS in 1h 23m 42s","accounts_in_message":[],"_revision_number":4},{"id":"e0d9992df25721127e999cfa9811c5d96c46d22b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-03-31 00:12:30.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"84de6fdf61c89fa4d1d63f458826ef197c6c2f7c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-03-31 00:12:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4b900e62bcd30eedb2e1b4c94ac0b866d31cff25","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-03-31 01:15:48.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29733 : FAILURE in 1h 02m 59s","accounts_in_message":[],"_revision_number":4},{"id":"2f760a55bb8e2a421c7da8405e5c2c55ca5fe28e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-03-31 01:43:42.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial/adcaa8b/ : SUCCESS in 1h 08m 48s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial-py3/c968895/ : FAILURE in 1h 30m 41s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/grenade-dsvm-xenial/0c3a809/ : SUCCESS in 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7b1c654adc1d3b94dd301f5980545460a308da43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-03-31 02:25:27.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/57/647957/4/check/grenade-py3/53ae108/ : SUCCESS in 51m 47s\n- tempest-full-py3 http://logs.openstack.org/57/647957/4/check/tempest-full-py3/fb20c6a/ : SUCCESS in 1h 31m 56s\n- openstack-tox-cover http://logs.openstack.org/57/647957/4/check/openstack-tox-cover/b75be2b/cover/ : SUCCESS in 14m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/4/check/openstack-tox-lower-constraints/cc0ada2/ : SUCCESS in 10m 25s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/4/check/openstack-tox-pep8/21a6e2f/ : SUCCESS in 9m 19s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/4/check/openstack-tox-py27/e56d30e/ : SUCCESS in 11m 56s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/4/check/openstack-tox-py35/e769d26/ : SUCCESS in 14m 08s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/4/check/openstack-tox-py36/59428a5/ : SUCCESS in 11m 23s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/4/check/openstack-tox-py37/65c9565/ : SUCCESS in 11m 21s\n- openstack-tox-docs http://logs.openstack.org/57/647957/4/check/openstack-tox-docs/aad7503/html/ : SUCCESS in 11m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/6cede6c/ : SUCCESS in 48m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/4/check/devstack-plugin-ceph-tempest/87e6f56/ : SUCCESS in 1h 30m 18s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/4/check/neutron-grenade-multinode/f9c98a3/ : SUCCESS in 1h 17m 03s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/4/check/nova-grenade-live-migration/c39aac3/ : SUCCESS in 1h 03m 47s\n- nova-live-migration http://logs.openstack.org/57/647957/4/check/nova-live-migration/98a5df3/ : FAILURE in 44m 48s\n- nova-next http://logs.openstack.org/57/647957/4/check/nova-next/8a13f71/ : SUCCESS in 1h 36m 52s\n- nova-tox-functional http://logs.openstack.org/57/647957/4/check/nova-tox-functional/011e4fd/ : SUCCESS in 17m 08s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/4/check/nova-tox-functional-py35/f0d2bd0/ : SUCCESS in 17m 31s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/4/check/tempest-slow-py3/025eb3a/ : SUCCESS in 2h 05m 20s","accounts_in_message":[],"_revision_number":4},{"id":"91f2adfba959480fae2209322f7d470abc5723db","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-03-31 02:30:53.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/647957/4 : FAILURE in 2h 17m 52s","accounts_in_message":[],"_revision_number":4},{"id":"26ce007057cc6066cc7f9a106111a627511d2975","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-01 08:02:35.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n@Magnus: Thanks for the update. Good job!","accounts_in_message":[],"_revision_number":4},{"id":"d34d8e28b3860cc1293bde010119923fef942d3d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-01 08:03:37.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5b02c993f01c606bdbaf1fe2d3fe340f2c9d8c09","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 08:04:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9fdd2aed14b29831bbb8db877aeaf8c3e225dd92","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-01 09:27:43.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial/a28d19a/ : SUCCESS in 1h 10m 50s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial-py3/10b5c87/ : SUCCESS in 1h 16m 40s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/grenade-dsvm-xenial/4674873/ : SUCCESS in 59m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3508425db76c1a9f077fea87af17c38c723d9bfc","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-01 10:24:18.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/647957/4 : FAILURE in 2h 20m 09s","accounts_in_message":[],"_revision_number":4},{"id":"29a8d34bb2392518847b0fed0b0bb7781496d7b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-01 10:27:04.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/57/647957/4/check/grenade-py3/c5121c1/ : SUCCESS in 57m 56s\n- tempest-full-py3 http://logs.openstack.org/57/647957/4/check/tempest-full-py3/9af7c5f/ : SUCCESS in 1h 30m 26s\n- openstack-tox-cover http://logs.openstack.org/57/647957/4/check/openstack-tox-cover/8b93bf1/cover/ : SUCCESS in 15m 58s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/4/check/openstack-tox-lower-constraints/86fd5cc/ : SUCCESS in 13m 04s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/4/check/openstack-tox-pep8/44078cf/ : SUCCESS in 10m 38s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/4/check/openstack-tox-py27/2cb6d53/ : SUCCESS in 11m 27s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/4/check/openstack-tox-py35/c915cc8/ : SUCCESS in 16m 05s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/4/check/openstack-tox-py36/5fbea88/ : SUCCESS in 11m 07s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/4/check/openstack-tox-py37/d2837d0/ : SUCCESS in 12m 27s\n- openstack-tox-docs http://logs.openstack.org/57/647957/4/check/openstack-tox-docs/8b792ac/html/ : SUCCESS in 8m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/af00832/ : SUCCESS in 45m 29s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/4/check/devstack-plugin-ceph-tempest/86d3bc2/ : SUCCESS in 1h 57m 52s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/4/check/neutron-grenade-multinode/0f6d131/ : SUCCESS in 1h 06m 34s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/4/check/nova-grenade-live-migration/2b60a9a/ : SUCCESS in 1h 08m 07s\n- nova-live-migration http://logs.openstack.org/57/647957/4/check/nova-live-migration/ff0e764/ : FAILURE in 47m 42s\n- nova-next http://logs.openstack.org/57/647957/4/check/nova-next/4c5a626/ : SUCCESS in 1h 43m 05s\n- nova-tox-functional http://logs.openstack.org/57/647957/4/check/nova-tox-functional/46b5926/ : SUCCESS in 16m 28s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/4/check/nova-tox-functional-py35/82149eb/ : SUCCESS in 20m 52s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/4/check/tempest-slow-py3/674295e/ : SUCCESS in 2h 17m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b2c898ff17c02afde8b9efd9e4b3fe7e66c4f4b8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-01 10:35:01.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29753 : SUCCESS in 2h 31m 05s","accounts_in_message":[],"_revision_number":4},{"id":"72529fa447ef4ba265b984fbd80fd6055f2789d4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-01 15:28:48.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/57/647957/4/check/check-dsvm-tempest-vz7-exe-minimal/aa6f614 : FAILURE in 1h 18m 06s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":4},{"id":"f845c47f27c34471fc890332e9da16565d641534","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-02 07:24:48.000000000","message":"Patch Set 4:\n\nnova-live-migration job has been fixed","accounts_in_message":[],"_revision_number":4},{"id":"cb7c9cd90716cb7d77f6c78b78a8abea6b573919","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-02 07:24:52.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"87c73282241aa66ea5d9be68b194f7fb5d6fc6d0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 07:25:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"67341d232cfba6da72d8b7020bc5b3b25fdad372","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 09:40:28.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/57/647957/4/check/grenade-py3/ca31681/ : SUCCESS in 50m 58s\n- tempest-full-py3 http://logs.openstack.org/57/647957/4/check/tempest-full-py3/6cab583/ : SUCCESS in 1h 28m 25s\n- openstack-tox-cover http://logs.openstack.org/57/647957/4/check/openstack-tox-cover/6a730a9/cover/ : SUCCESS in 18m 24s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/4/check/openstack-tox-lower-constraints/40eab06/ : SUCCESS in 14m 27s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/4/check/openstack-tox-pep8/38d7747/ : SUCCESS in 9m 14s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/4/check/openstack-tox-py27/8407dea/ : SUCCESS in 13m 02s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/4/check/openstack-tox-py35/218e44a/ : SUCCESS in 16m 21s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/4/check/openstack-tox-py36/ad781da/ : SUCCESS in 13m 34s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/4/check/openstack-tox-py37/1c58f8a/ : SUCCESS in 14m 24s\n- openstack-tox-docs http://logs.openstack.org/57/647957/4/check/openstack-tox-docs/f0dd197/html/ : SUCCESS in 6m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/f4bbbd8/ : SUCCESS in 47m 46s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/4/check/devstack-plugin-ceph-tempest/362f8ec/ : SUCCESS in 1h 31m 19s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/4/check/neutron-grenade-multinode/05c7dcc/ : SUCCESS in 59m 23s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/4/check/nova-grenade-live-migration/d011eb8/ : SUCCESS in 59m 35s\n- nova-live-migration http://logs.openstack.org/57/647957/4/check/nova-live-migration/3c96f5c/ : FAILURE in 42m 17s\n- nova-next http://logs.openstack.org/57/647957/4/check/nova-next/6aeeeda/ : SUCCESS in 1h 29m 52s\n- nova-tox-functional http://logs.openstack.org/57/647957/4/check/nova-tox-functional/e2bd412/ : SUCCESS in 20m 09s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/4/check/nova-tox-functional-py35/39c1b3b/ : SUCCESS in 19m 18s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/4/check/tempest-slow-py3/10b7ba5/ : SUCCESS in 2h 13m 28s","accounts_in_message":[],"_revision_number":4},{"id":"8364b58a9e3414ec4649d4a177685e91a12add0c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-04-02 09:55:46.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"076083552596688365514bf2fd4e5e3e83fecb14","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 09:56:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"84c86fd8a6719a06b9f67c0c16daf3d875875610","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-02 10:09:36.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/647957/4 : FAILURE in 50s","accounts_in_message":[],"_revision_number":4},{"id":"a29d5017f2638e4ff58ce67bbfa2f8e0263bbaf4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 12:34:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/57/647957/4/check/grenade-py3/c457a6d/ : SUCCESS in 1h 09m 02s\n- tempest-full-py3 http://logs.openstack.org/57/647957/4/check/tempest-full-py3/76be211/ : SUCCESS in 1h 31m 45s\n- openstack-tox-cover http://logs.openstack.org/57/647957/4/check/openstack-tox-cover/712457b/cover/ : SUCCESS in 16m 46s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/4/check/openstack-tox-lower-constraints/239d989/ : SUCCESS in 17m 31s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/4/check/openstack-tox-pep8/8b5858a/ : SUCCESS in 10m 26s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/4/check/openstack-tox-py27/bf9d350/ : SUCCESS in 13m 15s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/4/check/openstack-tox-py35/ebc14a5/ : SUCCESS in 19m 04s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/4/check/openstack-tox-py36/4997325/ : SUCCESS in 13m 26s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/4/check/openstack-tox-py37/bfa1331/ : SUCCESS in 12m 15s\n- openstack-tox-docs http://logs.openstack.org/57/647957/4/check/openstack-tox-docs/e0fe586/html/ : SUCCESS in 7m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/57/647957/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/cd6326b/ : SUCCESS in 54m 01s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/57/647957/4/check/devstack-plugin-ceph-tempest/5c6f9ed/ : SUCCESS in 1h 28m 36s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/57/647957/4/check/neutron-grenade-multinode/5a12ad0/ : SUCCESS in 1h 13m 58s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/4/check/nova-grenade-live-migration/5509c88/ : SUCCESS in 1h 12m 12s\n- nova-live-migration http://logs.openstack.org/57/647957/4/check/nova-live-migration/23e953d/ : SUCCESS in 1h 28m 52s\n- nova-next http://logs.openstack.org/57/647957/4/check/nova-next/448fe18/ : SUCCESS in 1h 12m 37s\n- nova-tox-functional http://logs.openstack.org/57/647957/4/check/nova-tox-functional/f892003/ : SUCCESS in 18m 52s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/4/check/nova-tox-functional-py35/1a53002/ : SUCCESS in 19m 49s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/4/check/tempest-slow-py3/e124c87/ : SUCCESS in 2h 10m 28s","accounts_in_message":[],"_revision_number":4},{"id":"3e809294b996b7b96d565e756458bd364374fe23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 12:35:10.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"65cacca997db087c92ce7077ff6c9dd7dc187b36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-02 12:35:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/57/647957/4/check/EMC_VxFlexOS_NOVA/56515d6/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"b48e9be961b69a882225aea4b7709bfcaca38c1b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-02 12:50:41.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-29855 : SUCCESS in 2h 48m 00s","accounts_in_message":[],"_revision_number":4},{"id":"3704b38959d4614b93c4e01660c29d21a7aa497c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-02 13:04:37.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/57/647957/4/check-vote/ext-nova-zuul/4bdf8f1 : FAILURE in 1h 15m 00s","accounts_in_message":[],"_revision_number":4},{"id":"c8e0730a86379833fbe13a7113e110a5caa32191","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-02 13:13:17.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-57-647957-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"570865e74662de4a255ab2688920f62ff1b3e59b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 16:23:46.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/57/647957/4/gate/grenade-py3/8352052/ : SUCCESS in 1h 01m 26s\n- tempest-full-py3 http://logs.openstack.org/57/647957/4/gate/tempest-full-py3/9266606/ : SUCCESS in 1h 38m 33s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/647957/4/gate/openstack-tox-lower-constraints/36f5385/ : SUCCESS in 17m 15s\n- openstack-tox-pep8 http://logs.openstack.org/57/647957/4/gate/openstack-tox-pep8/3ea641d/ : SUCCESS in 10m 43s\n- openstack-tox-py27 http://logs.openstack.org/57/647957/4/gate/openstack-tox-py27/5232705/ : SUCCESS in 14m 52s\n- openstack-tox-py35 http://logs.openstack.org/57/647957/4/gate/openstack-tox-py35/4f4400a/ : SUCCESS in 13m 57s\n- openstack-tox-py36 http://logs.openstack.org/57/647957/4/gate/openstack-tox-py36/9b6cb6b/ : SUCCESS in 12m 02s\n- openstack-tox-py37 http://logs.openstack.org/57/647957/4/gate/openstack-tox-py37/3bb137d/ : SUCCESS in 12m 55s\n- openstack-tox-docs http://logs.openstack.org/57/647957/4/gate/openstack-tox-docs/a96dc65/html/ : SUCCESS in 7m 35s\n- nova-grenade-live-migration http://logs.openstack.org/57/647957/4/gate/nova-grenade-live-migration/743f711/ : SUCCESS in 1h 08m 19s\n- nova-live-migration http://logs.openstack.org/57/647957/4/gate/nova-live-migration/7aff231/ : SUCCESS in 59m 15s\n- nova-tox-functional http://logs.openstack.org/57/647957/4/gate/nova-tox-functional/bb17c98/ : SUCCESS in 25m 00s\n- nova-tox-functional-py35 http://logs.openstack.org/57/647957/4/gate/nova-tox-functional-py35/0160638/ : SUCCESS in 21m 05s\n- nova-next http://logs.openstack.org/57/647957/4/gate/nova-next/b8dd193/ : SUCCESS in 1h 28m 15s\n- tempest-slow-py3 http://logs.openstack.org/57/647957/4/gate/tempest-slow-py3/4994010/ : SUCCESS in 2h 11m 01s","accounts_in_message":[],"_revision_number":4},{"id":"0b0135d33572ea64b1a2995cb391711106e5294b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-02 16:23:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ca7c2e8739910cb6948b52916241d402f63094f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-02 18:58:30.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial/b05015c/ : SUCCESS in 1h 15m 49s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/tempest-dsvm-full-xenial-py3/33b8b17/ : SUCCESS in 1h 29m 37s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/57/647957/4/check/grenade-dsvm-xenial/0b4c09f/ : SUCCESS in 58m 49s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"13e661792d73c27bcc7616b6bf612b5784e994e3","revisions":{"2233d69fdb89f3b146553c059de57cd425bc29f8":{"kind":"REWORK","_number":1,"created":"2019-03-27 00:59:28.000000000","uploader":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"ref":"refs/changes/57/647957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/647957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/647957/1"}}},"commit":{"parents":[{"commit":"64f6cbc9120e3c288f312eddc59452dee4998f93","subject":"Fix incomplete instance data returned after build failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64f6cbc9120e3c288f312eddc59452dee4998f93"}]}],"author":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-27 00:48:29.000000000","tz":60},"committer":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-27 00:48:29.000000000","tz":60},"subject":"Fix bug preventing forbidden traits from working","message":"Fix bug preventing forbidden traits from working\n\nModifies _clean_empties function to take forbidden traits into account\nin addition to required traits.\n\nChange-Id: I1c7e2d5ff2e7b5158f8f2c4b2ebce5c20e99a984\nCloses-Bug: #1821824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2233d69fdb89f3b146553c059de57cd425bc29f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2233d69fdb89f3b146553c059de57cd425bc29f8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64f6cbc9120e3c288f312eddc59452dee4998f93","is_merged_in_target_branch":true,"change_id":"If966eb1161c842ff49aa530e4482dbca87b61a3e","change_number":645546,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7593ddcdab4022299476d2ed6d39774fca49b579":{"kind":"REWORK","_number":2,"created":"2019-03-29 09:10:49.000000000","uploader":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"ref":"refs/changes/57/647957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/647957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/647957/2"}}},"commit":{"parents":[{"commit":"64f6cbc9120e3c288f312eddc59452dee4998f93","subject":"Fix incomplete instance data returned after build failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64f6cbc9120e3c288f312eddc59452dee4998f93"}]}],"author":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-27 00:48:29.000000000","tz":60},"committer":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-29 09:08:52.000000000","tz":60},"subject":"Adding tests to demonstrate bug #1821824","message":"Adding tests to demonstrate bug #1821824\n\nFixed functional test test_flavor_traits_based_scheduling as it\nwould pass 50% of the times even if the required trait was not\ntaken into account.\n\nAdded functional test test_flavor_forbidden_traits_based_scheduling\nto test that forbidden traits are taken into account.\n\nAdded unit test test_resources_from_request_spec_flavor_forbidden_trait\nto test that a single forbidden trait doesn\u0027t get lost in the\nresources_from_request_spec function.\n\nThe last two tests should at the time fail until the next commit where\nthe actual bug will be fixed.\n\nChange-Id: I1c7e2d5ff2e7b5158f8f2c4b2ebce5c20e99a984\nRelated-Bug: #1821824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7593ddcdab4022299476d2ed6d39774fca49b579"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7593ddcdab4022299476d2ed6d39774fca49b579"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64f6cbc9120e3c288f312eddc59452dee4998f93","is_merged_in_target_branch":true,"change_id":"If966eb1161c842ff49aa530e4482dbca87b61a3e","change_number":645546,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"edc130bc226e0e073b9fa102e8e77d6ba0834440":{"kind":"REWORK","_number":3,"created":"2019-03-29 15:18:06.000000000","uploader":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"ref":"refs/changes/57/647957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/647957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/647957/3"}}},"commit":{"parents":[{"commit":"64f6cbc9120e3c288f312eddc59452dee4998f93","subject":"Fix incomplete instance data returned after build failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64f6cbc9120e3c288f312eddc59452dee4998f93"}]}],"author":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-27 00:48:29.000000000","tz":60},"committer":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-29 14:58:42.000000000","tz":60},"subject":"Adding tests to demonstrate bug #1821824","message":"Adding tests to demonstrate bug #1821824\n\nFixed functional test test_flavor_traits_based_scheduling as it\nwould pass 50% of the times even if the required trait was not\ntaken into account.\n\nAdded functional test test_flavor_forbidden_traits_based_scheduling\nto test that forbidden traits are taken into account.\n\nThe second test temporarily passes while it should not until the\nnext commit where the bug is actually fixed.\n\nChange-Id: I1c7e2d5ff2e7b5158f8f2c4b2ebce5c20e99a984\nRelated-Bug: #1821824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/edc130bc226e0e073b9fa102e8e77d6ba0834440"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/edc130bc226e0e073b9fa102e8e77d6ba0834440"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64f6cbc9120e3c288f312eddc59452dee4998f93","is_merged_in_target_branch":true,"change_id":"If966eb1161c842ff49aa530e4482dbca87b61a3e","change_number":645546,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"13e661792d73c27bcc7616b6bf612b5784e994e3":{"kind":"REWORK","_number":4,"created":"2019-03-30 18:24:23.000000000","uploader":{"_account_id":23922,"name":"Magnus Bergman","email":"mb@citynetwork.se","username":"magnusbe"},"ref":"refs/changes/57/647957/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/647957/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/647957/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/647957/4"}}},"commit":{"parents":[{"commit":"64f6cbc9120e3c288f312eddc59452dee4998f93","subject":"Fix incomplete instance data returned after build failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64f6cbc9120e3c288f312eddc59452dee4998f93"}]}],"author":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-27 00:48:29.000000000","tz":60},"committer":{"name":"mb","email":"mb@citynetwork.se","date":"2019-03-30 18:07:59.000000000","tz":60},"subject":"Adding tests to demonstrate bug #1821824","message":"Adding tests to demonstrate bug #1821824\n\nFixed functional test test_flavor_traits_based_scheduling as it\nwould pass 50% of the times even if the required trait was not\ntaken into account by adding a new part that disables the compute\nhost that has the required trait and retries, expecting a no valid\nhost error.\n\nAdded functional test test_flavor_forbidden_traits_based_scheduling\nto test that forbidden traits are taken into account.\n\nThe second test temporarily passes where it should not until the\nnext commit where the bug is actually fixed.\n\nChange-Id: I1c7e2d5ff2e7b5158f8f2c4b2ebce5c20e99a984\nRelated-Bug: #1821824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13e661792d73c27bcc7616b6bf612b5784e994e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/13e661792d73c27bcc7616b6bf612b5784e994e3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"64f6cbc9120e3c288f312eddc59452dee4998f93","is_merged_in_target_branch":true,"change_id":"If966eb1161c842ff49aa530e4482dbca87b61a3e","change_number":645546,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
