)]}'
{"id":"openstack%2Fnova~651299","triplet_id":"openstack%2Fnova~master~Ia8da8cd040d104d745bd7adf20438b8eec558680","project":"openstack/nova","branch":"master","topic":"bp/remove-cells-v1","hashtags":[],"change_id":"Ia8da8cd040d104d745bd7adf20438b8eec558680","subject":"Remove \u0027instance_info_cache_update_at_top\u0027","status":"MERGED","created":"2019-04-09 16:15:52.000000000","updated":"2019-04-30 22:11:51.000000000","submitted":"2019-04-30 22:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"651299-1556662311638-b3e99d47","meta_rev_id":"c49042482eb7c1725b6e239236d80a3a097cdb9b","_number":651299,"virtual_id_number":651299,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-30 20:09:44.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 17:46:47.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-29 17:03:54.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 19:57:05.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-04-30 19:07:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"date":"2019-04-29 18:06:30.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"date":"2019-04-30 19:08:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-04-30 22:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"date":"2019-04-30 21:09:57.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 18:52:45.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-04-30 20:13:34.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-04-30 17:46:09.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2019-04-29 21:43:17.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-04-29 21:43:17.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-13 02:30:59.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-04-13 07:01:37.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-04-13 11:45:18.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-04-18 13:46:07.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2019-04-18 14:43:53.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-04-29 17:03:54.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 18:06:30.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-29 21:43:17.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-04-30 17:46:09.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 17:46:47.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 18:52:45.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 19:07:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 19:08:29.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-04-30 19:57:05.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 20:09:44.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 20:13:34.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-04-30 21:09:57.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-30 22:11:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d243cd37a74e20e8b11870f57b5ddfd270542193","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-09 16:15:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0df45dd9f1782f659eabf92bf48831b9cdfd6e73","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-09 18:10:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/1/check/tempest-dsvm-full-xenial/dda0b9e/ : SUCCESS in 1h 15m 13s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/1/check/tempest-dsvm-full-xenial-py3/914893a/ : SUCCESS in 1h 20m 49s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/1/check/grenade-dsvm-xenial/8e54bef/ : SUCCESS in 1h 04m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da930cb91c5009737171ab861cccba643cdc1470","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-09 19:22:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/99/651299/1/check/grenade-py3/45743b7/ : SUCCESS in 54m 30s\n- tempest-full-py3 http://logs.openstack.org/99/651299/1/check/tempest-full-py3/3b13160/ : SUCCESS in 1h 33m 39s\n- openstack-tox-cover http://logs.openstack.org/99/651299/1/check/openstack-tox-cover/3779fa2/cover/ : SUCCESS in 16m 52s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/1/check/openstack-tox-lower-constraints/26b0d46/ : SUCCESS in 14m 53s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/1/check/openstack-tox-pep8/82ed275/ : SUCCESS in 9m 59s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/1/check/openstack-tox-py27/7c31f83/ : SUCCESS in 13m 53s\n- openstack-tox-py35 http://logs.openstack.org/99/651299/1/check/openstack-tox-py35/1544d37/ : SUCCESS in 16m 54s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/1/check/openstack-tox-py36/1d9c8d8/ : SUCCESS in 13m 36s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/1/check/openstack-tox-py37/cafe02b/ : SUCCESS in 20m 31s\n- openstack-tox-docs http://logs.openstack.org/99/651299/1/check/openstack-tox-docs/d721c77/html/ : SUCCESS in 8m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/5f42e6a/ : SUCCESS in 57m 17s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/1/check/devstack-plugin-ceph-tempest/158e72b/ : SUCCESS in 1h 32m 38s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/1/check/neutron-grenade-multinode/a0bba44/ : SUCCESS in 1h 09m 56s (non-voting)\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/1/check/neutron-tempest-linuxbridge/e06b938/ : SUCCESS in 1h 31m 08s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/1/check/nova-grenade-live-migration/a521d40/ : SUCCESS in 1h 04m 33s\n- nova-live-migration http://logs.openstack.org/99/651299/1/check/nova-live-migration/c0a2300/ : SUCCESS in 47m 50s\n- nova-next http://logs.openstack.org/99/651299/1/check/nova-next/eb87960/ : SUCCESS in 1h 42m 00s\n- nova-tox-functional http://logs.openstack.org/99/651299/1/check/nova-tox-functional/2a42a70/ : SUCCESS in 20m 58s\n- nova-tox-functional-py35 http://logs.openstack.org/99/651299/1/check/nova-tox-functional-py35/26aaff1/ : SUCCESS in 19m 29s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/1/check/tempest-slow-py3/eb45230/ : SUCCESS in 2h 21m 32s","accounts_in_message":[],"_revision_number":1},{"id":"59e17b5abf0c8d701dda6bd2d373ff0fe6a1d066","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-09 19:23:56.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/99/651299/1/check/EMC_VxFlexOS_NOVA/ef34810/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"a8c1bd654b1fcd61ce4878d043f0cfe7d30e42d1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-09 20:55:42.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30289 : SUCCESS in 3h 33m 32s","accounts_in_message":[],"_revision_number":1},{"id":"85308c928cdd6bbb8de1f2f59a9b288fe4ca36f6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-09 20:56:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/651299/1 : SUCCESS in 2h 15m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ee62c374c813472677f299c8538f607837a4a70f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-10 00:04:41.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-99-651299-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"a8b2c4b3fa989f247bc88c1e4ba2239235ac02a4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-10 05:00:41.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/99/651299/1/check-vote/ext-nova-zuul/96c51a0 : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":1},{"id":"92ee2595bdd0a109a6b97ddf21e69df24776e590","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-11 01:46:58.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/651299/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/469d9e0 : SUCCESS in 1h 23m 22s","accounts_in_message":[],"_revision_number":1},{"id":"313c122be830496a2a8293ccfc5d2a06bb211966","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-12 22:16:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ce19fa0ce43b7560b3de22ebb0c4c96c4fce7ac0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-13 02:27:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/1/check/check-dsvm-tempest-vz7-exe-minimal/da0359d : SUCCESS in 1h 13m 36s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":1},{"id":"8bb4fbf0e9dd41cc6b2607fa2267b298e86163a1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-04-13 02:30:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f243a1aad872001b9113d8c0816140e040d68201","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-04-13 07:01:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e2037b4df2e26c750cee5f2c757386b089e23bf5","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-04-13 11:45:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4ff03c5be1133dfdcb685b9eaa108f46ef0ac24","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-16 08:52:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"055942b1b964a3043071edf1127e9a7851f4ac31","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-16 10:51:14.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/2/check/tempest-dsvm-full-xenial/5c03ed6/ : SUCCESS in 1h 38m 20s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/2/check/tempest-dsvm-full-xenial-py3/7c847f2/ : SUCCESS in 1h 14m 23s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/2/check/grenade-dsvm-xenial/e1ee85d/ : SUCCESS in 53m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"294d125e712c51136b216bdaf21e52247d962e3c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-16 12:48:13.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/651299/2 : FAILURE in 2h 34m 08s","accounts_in_message":[],"_revision_number":2},{"id":"e4f5eae60b108f6277ed97e20a3128c646bdde0a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-16 13:10:23.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30537 : FAILURE in 3h 27m 02s","accounts_in_message":[],"_revision_number":2},{"id":"394137bb9be0791637490d1a53a8ce59aef3f5a7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-16 14:22:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"49649ebe801c42264253cabf248cbf6930ed4943","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-16 16:15:21.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/3/check/tempest-dsvm-full-xenial/2c978ef/ : SUCCESS in 1h 29m 24s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/3/check/tempest-dsvm-full-xenial-py3/814c08b/ : SUCCESS in 1h 31m 27s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/3/check/grenade-dsvm-xenial/3ad2b5c/ : FAILURE in 9m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09c1db24c252cbfdedb89cb00be6396e1bb5124c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-16 17:26:44.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"4b90aaf6cd87561c60ee55b8439c758288ca9aaf","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-16 17:37:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f3f3da152c0a23db648eed0dd3d210a09c68f9af","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-16 19:13:11.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/4/check/tempest-dsvm-full-xenial/d9e3339/ : SUCCESS in 1h 25m 25s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/4/check/tempest-dsvm-full-xenial-py3/d5ebd26/ : SUCCESS in 1h 26m 20s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/4/check/grenade-dsvm-xenial/fe713d7/ : FAILURE in 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f8a1ec3919527c4a506058bedd49a2b6e8a30e55","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-16 22:06:21.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30578 : FAILURE in 2h 51m 26s","accounts_in_message":[],"_revision_number":4},{"id":"20507ad573ea03778e2f4d812f524a8c14fa3928","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-16 22:14:47.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/651299/4 : SUCCESS in 2h 32m 01s","accounts_in_message":[],"_revision_number":4},{"id":"a4a1aa70c8580a6c55f351c9933be818c4384ee9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-16 23:09:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/99/651299/4/check/grenade-py3/33fdce5/ : SUCCESS in 1h 04m 43s\n- tempest-full-py3 http://logs.openstack.org/99/651299/4/check/tempest-full-py3/184549a/ : SUCCESS in 1h 37m 37s\n- openstack-tox-cover http://logs.openstack.org/99/651299/4/check/openstack-tox-cover/6fd5604/cover/ : SUCCESS in 14m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/4/check/openstack-tox-lower-constraints/8db3b43/ : FAILURE in 15m 00s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/4/check/openstack-tox-pep8/ffd1bb3/ : SUCCESS in 10m 34s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/4/check/openstack-tox-py27/318cb38/ : SUCCESS in 13m 29s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/4/check/openstack-tox-py36/c1ab393/ : SUCCESS in 13m 51s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/4/check/openstack-tox-py37/6e41cf6/ : SUCCESS in 12m 13s\n- openstack-tox-docs http://logs.openstack.org/99/651299/4/check/openstack-tox-docs/e94f5bd/html/ : SUCCESS in 6m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/e94fe48/ : SUCCESS in 50m 29s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/4/check/devstack-plugin-ceph-tempest/09c18f3/ : SUCCESS in 1h 37m 06s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/4/check/neutron-grenade-multinode/83c1afa/ : SUCCESS in 1h 15m 04s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/4/check/neutron-tempest-linuxbridge/77c0356/ : SUCCESS in 1h 23m 36s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/4/check/nova-grenade-live-migration/509c651/ : SUCCESS in 1h 25m 16s\n- nova-live-migration http://logs.openstack.org/99/651299/4/check/nova-live-migration/f674ae8/ : SUCCESS in 55m 36s\n- nova-next http://logs.openstack.org/99/651299/4/check/nova-next/ab58921/ : SUCCESS in 1h 17m 21s\n- nova-tox-functional http://logs.openstack.org/99/651299/4/check/nova-tox-functional/b94d241/ : SUCCESS in 23m 21s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/4/check/nova-tox-functional-py36/1dd50c9/ : SUCCESS in 18m 01s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/4/check/tempest-slow-py3/efe73ee/ : SUCCESS in 2h 26m 04s","accounts_in_message":[],"_revision_number":4},{"id":"f818f8995a7a53de798b0af414ab9a3591273789","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-17 09:48:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"a587a30170989bdfa9f73f3b78057a20527a971f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 09:56:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b4a875ba1613bc19b6fa3397d30f242089fce879","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-17 11:29:29.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/4/check/tempest-dsvm-full-xenial/24ddc47/ : SUCCESS in 1h 20m 53s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/4/check/tempest-dsvm-full-xenial-py3/2aeb6d9/ : SUCCESS in 1h 19m 37s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/4/check/grenade-dsvm-xenial/a5c60e8/ : SUCCESS in 1h 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d98929b58af3ff3ebe9cbfa8a66c47f49a514c35","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-17 12:55:57.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-30619 : SUCCESS in 2h 11m 15s","accounts_in_message":[],"_revision_number":4},{"id":"77939cb08194df515712eb825e64fecb84b7ee00","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-17 13:46:46.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/651299/4 : FAILURE in 2h 16m 44s","accounts_in_message":[],"_revision_number":4},{"id":"6cfb062e6f5639145294c92e99c411ea7fb39588","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-17 20:53:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/99/651299/4/check/grenade-py3/f7b6ef8/ : SUCCESS in 1h 02m 23s\n- tempest-full-py3 http://logs.openstack.org/99/651299/4/check/tempest-full-py3/075d3d7/ : SUCCESS in 1h 33m 52s\n- openstack-tox-cover http://logs.openstack.org/99/651299/4/check/openstack-tox-cover/652801e/cover/ : SUCCESS in 20m 18s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/4/check/openstack-tox-lower-constraints/7282da4/ : SUCCESS in 15m 42s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/4/check/openstack-tox-pep8/8d8c0e2/ : SUCCESS in 11m 30s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/4/check/openstack-tox-py27/5a7ebf4/ : SUCCESS in 17m 58s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/4/check/openstack-tox-py36/e450ae6/ : SUCCESS in 12m 46s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/4/check/openstack-tox-py37/ceb0857/ : SUCCESS in 13m 52s\n- openstack-tox-docs http://logs.openstack.org/99/651299/4/check/openstack-tox-docs/e4f3391/html/ : SUCCESS in 7m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/4/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/0f649ba/ : SUCCESS in 50m 41s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/4/check/devstack-plugin-ceph-tempest/4565952/ : SUCCESS in 1h 34m 20s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/4/check/neutron-grenade-multinode/71e4522/ : SUCCESS in 1h 11m 48s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/4/check/neutron-tempest-linuxbridge/1605c2e/ : SUCCESS in 1h 39m 05s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/4/check/nova-grenade-live-migration/d1ff5e2/ : SUCCESS in 1h 05m 38s\n- nova-live-migration http://logs.openstack.org/99/651299/4/check/nova-live-migration/c703bd4/ : SUCCESS in 50m 52s\n- nova-next http://logs.openstack.org/99/651299/4/check/nova-next/c1535e7/ : SUCCESS in 1h 28m 27s\n- nova-tox-functional http://logs.openstack.org/99/651299/4/check/nova-tox-functional/2ba7fd4/ : SUCCESS in 19m 15s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/4/check/nova-tox-functional-py36/5476804/ : SUCCESS in 19m 13s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/4/check/tempest-slow-py3/33798e9/ : SUCCESS in 2h 15m 03s","accounts_in_message":[],"_revision_number":4},{"id":"a5b66e87ec3bb87a4ad0500a55807fd9c96106b5","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-17 20:54:29.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/99/651299/4/check/EMC_VxFlexOS_NOVA/27e939b/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck, for more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"81f0f80edc6f912aab6bc5b9e31a03da0f484040","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-17 22:20:51.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-99-651299-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"580108bc0ef3a7b071246823b921b78ada107c6e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2019-04-18 13:46:07.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ca00e61e7e84219c576e8bb25713d6545d23894e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-18 14:43:53.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"816e5b8d03e3e46b6a24ddcfc5ba1068e04c05c9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-19 09:45:23.000000000","message":"Patch Set 4:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/4/check/nova-out-of-tree-pvm/04d30c0 : FAILURE in 13m 23s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/4/check/nova-in-tree-pvm/77e6f15 : FAILURE in 3h 33m 16s","accounts_in_message":[],"_revision_number":4},{"id":"76a7fb14f6e87cac6e52c70ff0d0e92f73839158","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 12:40:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"10b64a367a9b4ba053f22a84a5be288be90a26a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 14:57:18.000000000","message":"Patch Set 4: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":4},{"id":"42f30758a4d3785cffffb4d33815e2d7eb304ba2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-04-29 16:55:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3969c72eab9d7f331f4ff4919e962bd3ea7eee86","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 16:56:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5a5bac512f3f42a80009cd14940950458a557a73","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 17:01:29.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/5/check/check-dsvm-tempest-vz7-exe-minimal/417db97 : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"180ee87cdeb08abee860c3444cc199004866c7a4","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 17:03:54.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/651299/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/4db8594 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":5},{"id":"b3f015b1a749206b80d2c3ac8b436c1f5bcdbd63","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2019-04-29 18:06:30.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/651299/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"b4f51f781879a9dd0d4f98cf32856c0df0fa169d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-29 18:32:19.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial/2e8f0c6/ : SUCCESS in 1h 18m 37s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial-py3/667894c/ : SUCCESS in 1h 23m 19s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/grenade-dsvm-xenial/bff9f46/ : SUCCESS in 1h 34m 18s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a6eb6db9823105dc82d09e38b14e59412d3c8a70","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-29 19:39:45.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31244 : FAILURE in 2h 32m 33s","accounts_in_message":[],"_revision_number":5},{"id":"f75551eb3d618dea62bb1ffe2ff16b9e236b6084","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-29 20:16:06.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/5/check/nova-out-of-tree-pvm/8d3752d : FAILURE in 2h 49m 36s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/5/check/nova-in-tree-pvm/a294d26 : FAILURE in 3h 19m 25s","accounts_in_message":[],"_revision_number":5},{"id":"d01f15ccaea73f4500b08732a5b328aecf47bd37","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-29 20:42:18.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/99/651299/5/check-vote/ext-nova-zuul/06a4476 : SUCCESS in 56m 08s","accounts_in_message":[],"_revision_number":5},{"id":"2917416626069d7ace8c3819b4929a96d246601f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-29 20:49:46.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/651299/5 : FAILURE in 2h 24m 32s","accounts_in_message":[],"_revision_number":5},{"id":"dd6cce2daa691c3417998b9a10aee7e2037e483f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 20:55:06.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/check/grenade-py3/ec76a9f/ : SUCCESS in 1h 02m 17s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/check/tempest-full-py3/da4cdd2/ : SUCCESS in 1h 34m 49s\n- openstack-tox-cover http://logs.openstack.org/99/651299/5/check/openstack-tox-cover/bbbb590/cover/ : SUCCESS in 16m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/check/openstack-tox-lower-constraints/f069ee1/ : SUCCESS in 14m 35s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/check/openstack-tox-pep8/22b0430/ : SUCCESS in 11m 35s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/check/openstack-tox-py27/d6c99ab/ : SUCCESS in 17m 52s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/check/openstack-tox-py36/b89c599/ : SUCCESS in 19m 51s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/check/openstack-tox-py37/410f00c/ : SUCCESS in 13m 36s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/check/openstack-tox-docs/ccc3100/html/ : SUCCESS in 9m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/2d17b65/ : SUCCESS in 51m 15s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/5/check/devstack-plugin-ceph-tempest/ebf0412/ : SUCCESS in 1h 48m 38s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/5/check/neutron-grenade-multinode/f7f02c8/ : SUCCESS in 1h 11m 08s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/5/check/neutron-tempest-linuxbridge/e5f6659/ : SUCCESS in 1h 39m 55s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/check/nova-grenade-live-migration/c315b26/ : SUCCESS in 1h 13m 12s\n- nova-live-migration http://logs.openstack.org/99/651299/5/check/nova-live-migration/fe95d36/ : SUCCESS in 59m 20s\n- nova-next http://logs.openstack.org/99/651299/5/check/nova-next/8ee20f5/ : FAILURE in 1h 21m 29s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/check/nova-tox-functional/0833402/ : SUCCESS in 17m 12s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/check/nova-tox-functional-py36/b71621b/ : SUCCESS in 16m 58s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/check/tempest-slow-py3/3659075/ : SUCCESS in 2h 13m 29s","accounts_in_message":[],"_revision_number":5},{"id":"ce1da027180a2563455b4e2fdb2a233295f028a4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-29 21:17:26.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"c878fd099b0c2f23e18cb19c6f7fbf195d320aa9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 21:18:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"4a20109b60ddce320980ba15de50bfc905729818","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-29 21:22:15.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/5/check/check-dsvm-tempest-vz7-exe-minimal/4e3f97b : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"019558913799a36946536623c961e2f339a33c27","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-29 21:43:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nrecheck\n\nFast reapproval after trivial manual rebase","accounts_in_message":[],"_revision_number":5},{"id":"31c871e823040a917d7707d8c435b4047bbda8dd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-29 22:50:40.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial/74b45bb/ : SUCCESS in 1h 15m 59s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial-py3/272b925/ : SUCCESS in 1h 32m 13s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/grenade-dsvm-xenial/b63b69d/ : SUCCESS in 1h 05m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5ecf303e5bbe27b2740df13d52eb3c87fa83aa09","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-29 23:03:19.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31260 : SUCCESS in 1h 45m 15s","accounts_in_message":[],"_revision_number":5},{"id":"6202fcaef755ef24bf5cf1afcb20497970fa37a3","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-29 23:17:18.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/5/check/nova-out-of-tree-pvm/36c1bcb : FAILURE in 1h 57m 57s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/5/check/nova-in-tree-pvm/cf45094 : FAILURE in 1h 59m 20s","accounts_in_message":[],"_revision_number":5},{"id":"5f6179438cd80667ae1f4ea814da962789b56286","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 23:33:16.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/check/grenade-py3/5a8e5b4/ : FAILURE in 1h 01m 15s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/check/tempest-full-py3/b2acb45/ : SUCCESS in 1h 34m 30s\n- openstack-tox-cover http://logs.openstack.org/99/651299/5/check/openstack-tox-cover/bd1e9ef/cover/ : SUCCESS in 14m 08s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/check/openstack-tox-lower-constraints/febeb65/ : SUCCESS in 11m 41s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/check/openstack-tox-pep8/b2c9000/ : SUCCESS in 10m 09s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/check/openstack-tox-py27/f8b543c/ : SUCCESS in 11m 26s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/check/openstack-tox-py36/56766e7/ : SUCCESS in 14m 51s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/check/openstack-tox-py37/ad145e2/ : SUCCESS in 11m 40s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/check/openstack-tox-docs/bcd8fa0/html/ : SUCCESS in 6m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/b577fea/ : SUCCESS in 47m 44s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/5/check/devstack-plugin-ceph-tempest/69e9b98/ : SUCCESS in 1h 44m 57s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/5/check/neutron-grenade-multinode/32c09b8/ : SUCCESS in 1h 08m 02s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/5/check/neutron-tempest-linuxbridge/5ff5b61/ : SUCCESS in 1h 31m 21s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/check/nova-grenade-live-migration/c93ede8/ : SUCCESS in 1h 07m 25s\n- nova-live-migration http://logs.openstack.org/99/651299/5/check/nova-live-migration/062608a/ : SUCCESS in 48m 31s\n- nova-next http://logs.openstack.org/99/651299/5/check/nova-next/512f9a9/ : SUCCESS in 1h 25m 57s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/check/nova-tox-functional/cbdf240/ : SUCCESS in 19m 47s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/check/nova-tox-functional-py36/7272ebf/ : SUCCESS in 16m 19s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/check/tempest-slow-py3/afa3ac0/ : SUCCESS in 2h 14m 47s","accounts_in_message":[],"_revision_number":5},{"id":"d86f622fcc3bb7b7e98db659ebfe611d50e02731","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-30 00:14:09.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/651299/5 : SUCCESS in 2h 18m 03s","accounts_in_message":[],"_revision_number":5},{"id":"17b33ff9747a89401e5d71eb032583dd5e8d0148","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 00:45:39.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"a2a3610ba1615c6a2a7be5ff3289d3b60c00cb0d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 00:46:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0c23cea3ee7953f40ee3b4b76fa943e4a574189a","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 00:48:46.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/5/check/check-dsvm-tempest-vz7-exe-minimal/1963123 : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"4b07df28b2b964130f8809913b6e41f652dbe578","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 01:46:09.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/99/651299/5/check-vote/ext-nova-zuul/895d6d6 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":5},{"id":"944d5db360b8b22760f5afe9286a3b8b67d38939","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-30 01:50:21.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31271 : SUCCESS in 1h 04m 05s","accounts_in_message":[],"_revision_number":5},{"id":"b8c7c2ce6e1607ce88b4028d9daf9a7ea5a1bf7a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-30 02:08:02.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial/d071126/ : SUCCESS in 1h 13m 52s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial-py3/99336e3/ : SUCCESS in 1h 21m 23s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/grenade-dsvm-xenial/bbd5cfa/ : SUCCESS in 1h 01m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0bd12aed25e7bb3ccb9b0a1151cca27c7f55d777","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-30 02:18:28.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/5/check/nova-out-of-tree-pvm/cb695a4 : FAILURE in 1h 32m 16s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/5/check/nova-in-tree-pvm/9555f47 : FAILURE in 1h 29m 59s","accounts_in_message":[],"_revision_number":5},{"id":"9675402ad074cd082fe11e0c97f110c6b052d715","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-30 02:56:47.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/651299/5 : SUCCESS in 2h 10m 26s","accounts_in_message":[],"_revision_number":5},{"id":"c9d93b85314459a318caf14875f27a150d778390","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 03:02:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/check/grenade-py3/3522264/ : FAILURE in 30m 50s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/check/tempest-full-py3/b6013db/ : SUCCESS in 2h 04m 08s\n- openstack-tox-cover http://logs.openstack.org/99/651299/5/check/openstack-tox-cover/6cf7dbb/cover/ : SUCCESS in 15m 05s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/check/openstack-tox-lower-constraints/b616fed/ : SUCCESS in 15m 55s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/check/openstack-tox-pep8/209f083/ : SUCCESS in 9m 47s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/check/openstack-tox-py27/abd8aec/ : SUCCESS in 13m 17s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/check/openstack-tox-py36/6be2fa7/ : SUCCESS in 13m 05s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/check/openstack-tox-py37/3030c54/ : SUCCESS in 12m 18s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/check/openstack-tox-docs/fcbf9ac/html/ : SUCCESS in 5m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/ee602b9/ : SUCCESS in 52m 51s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/5/check/devstack-plugin-ceph-tempest/4481dce/ : SUCCESS in 1h 44m 40s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/5/check/neutron-grenade-multinode/f56eef3/ : SUCCESS in 1h 07m 56s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/5/check/neutron-tempest-linuxbridge/f822b6a/ : SUCCESS in 1h 31m 50s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/check/nova-grenade-live-migration/6eb43fa/ : SUCCESS in 1h 10m 16s\n- nova-live-migration http://logs.openstack.org/99/651299/5/check/nova-live-migration/c50391c/ : SUCCESS in 48m 12s\n- nova-next http://logs.openstack.org/99/651299/5/check/nova-next/be9ac37/ : FAILURE in 1h 59m 22s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/check/nova-tox-functional/5339442/ : SUCCESS in 20m 03s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/check/nova-tox-functional-py36/91be34f/ : SUCCESS in 15m 39s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/check/tempest-slow-py3/331049d/ : SUCCESS in 2h 15m 23s","accounts_in_message":[],"_revision_number":5},{"id":"b4ef9480258eb008e620ca3987554957e9754da2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 13:23:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1c2c66e37db45e2e1b8c585116fe42285e896e00","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 13:23:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ff95ee84a91977dd7bdeb887f8b0ca582eb6d6b7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 13:26:50.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/5/check/check-dsvm-tempest-vz7-exe-minimal/7058d43 : FAILURE in 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"4d2c20573db09290165a8a8f7cdb5cfdf43b4b31","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-30 14:41:46.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31300 : SUCCESS in 1h 17m 57s","accounts_in_message":[],"_revision_number":5},{"id":"58b96c4c89c646e5e66270f08a0c79c306b15d6c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-30 14:52:30.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial/c6a9c57/ : SUCCESS in 1h 17m 45s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial-py3/3855dfe/ : SUCCESS in 1h 27m 47s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/grenade-dsvm-xenial/a9b1c45/ : SUCCESS in 1h 05m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef2191fe90b547b3143cd71c4e47e064cef153fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 15:04:13.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/99/651299/5/check-vote/ext-nova-zuul/86f573d : SUCCESS in 58m 14s","accounts_in_message":[],"_revision_number":5},{"id":"ba601b8e0f5f83282be2dc31da01edcb5821c43e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 15:17:34.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/check/grenade-py3/b840eaf/ : SUCCESS in 1h 05m 24s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/check/tempest-full-py3/8d4f718/ : SUCCESS in 1h 30m 58s\n- openstack-tox-cover http://logs.openstack.org/99/651299/5/check/openstack-tox-cover/76e5a00/cover/ : SUCCESS in 15m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/check/openstack-tox-lower-constraints/2da641d/ : SUCCESS in 19m 49s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/check/openstack-tox-pep8/a4f8386/ : SUCCESS in 9m 26s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/check/openstack-tox-py27/fa7e488/ : SUCCESS in 15m 10s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/check/openstack-tox-py36/04b7ba8/ : SUCCESS in 17m 00s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/check/openstack-tox-py37/0aff8d2/ : SUCCESS in 17m 55s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/check/openstack-tox-docs/34c1d72/html/ : SUCCESS in 7m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/75dd945/ : SUCCESS in 51m 28s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/5/check/devstack-plugin-ceph-tempest/01b1341/ : SUCCESS in 1h 53m 01s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/5/check/neutron-grenade-multinode/559f2c5/ : SUCCESS in 1h 07m 20s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/5/check/neutron-tempest-linuxbridge/294eea7/ : SUCCESS in 1h 23m 43s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/check/nova-grenade-live-migration/ee9b6fa/ : SUCCESS in 1h 14m 56s\n- nova-live-migration http://logs.openstack.org/99/651299/5/check/nova-live-migration/ca0463f/ : SUCCESS in 57m 06s\n- nova-next http://logs.openstack.org/99/651299/5/check/nova-next/cc93434/ : SUCCESS in 1h 28m 42s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/check/nova-tox-functional/b436d42/ : SUCCESS in 17m 47s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/check/nova-tox-functional-py36/a23d762/ : SUCCESS in 18m 58s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/check/tempest-slow-py3/efd95fd/ : FAILURE in 28m 11s","accounts_in_message":[],"_revision_number":5},{"id":"48e0de5bfe9b4194f97e02cb56e29e7866617100","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-30 15:32:25.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/5/check/nova-out-of-tree-pvm/3c05165 : SUCCESS in 2h 08m 40s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/5/check/nova-in-tree-pvm/8610dc7 : FAILURE in 1h 33m 44s","accounts_in_message":[],"_revision_number":5},{"id":"ed3a0a3e2c6412b87ce248127482bb0519070ed8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 15:45:23.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"e6778b6e0703dfdb8cda328ce2dc3124ad30c669","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 15:46:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ae215c01b9336556bbf5ba8229ffe321b6c278ad","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 15:46:53.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/5/check/check-dsvm-tempest-vz7-exe-minimal/2794cd6 : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"fb785270b7441e9d39c635737a9920feef9da91f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-30 16:23:21.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31305 : FAILURE in 37m 18s","accounts_in_message":[],"_revision_number":5},{"id":"72ae39745a38e3ee72e7d11d62b247530eff12a5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-30 16:28:02.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/651299/5 : SUCCESS in 2h 22m 14s","accounts_in_message":[],"_revision_number":5},{"id":"599b7d35c2c41e998e13acfd76e39bf420b576f9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 16:54:29.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/99/651299/5/check-vote/ext-nova-zuul/129c00c : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":5},{"id":"745b5bc54faa5b86e7faa09e3727f634f3465ff7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-30 17:10:35.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial/3322a45/ : SUCCESS in 1h 14m 29s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial-py3/d67a9be/ : SUCCESS in 1h 24m 20s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/grenade-dsvm-xenial/0cf77c3/ : SUCCESS in 1h 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b49a954fe9c704704c2db930bd5c5c145785d05a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-30 17:22:41.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/5/check/nova-out-of-tree-pvm/b233b3e : FAILURE in 1h 36m 43s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/5/check/nova-in-tree-pvm/88f95f6 : SUCCESS in 1h 30m 18s","accounts_in_message":[],"_revision_number":5},{"id":"eae3f156432c149bed967e3ac1345e06bd6e94f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 17:31:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/check/grenade-py3/736acec/ : SUCCESS in 1h 01m 44s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/check/tempest-full-py3/b202dc4/ : SUCCESS in 1h 40m 40s\n- openstack-tox-cover http://logs.openstack.org/99/651299/5/check/openstack-tox-cover/e68cc12/cover/ : SUCCESS in 17m 26s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/check/openstack-tox-lower-constraints/00c3598/ : SUCCESS in 15m 44s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/check/openstack-tox-pep8/d8bfbac/ : SUCCESS in 9m 56s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/check/openstack-tox-py27/9a6c285/ : SUCCESS in 15m 48s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/check/openstack-tox-py36/f95a6c1/ : SUCCESS in 11m 57s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/check/openstack-tox-py37/d972898/ : SUCCESS in 11m 52s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/check/openstack-tox-docs/074bb68/html/ : SUCCESS in 6m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/59f60ab/ : SUCCESS in 52m 36s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/5/check/devstack-plugin-ceph-tempest/645f8df/ : SUCCESS in 1h 41m 16s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/5/check/neutron-grenade-multinode/ad581e7/ : SUCCESS in 1h 02m 29s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/5/check/neutron-tempest-linuxbridge/85ed3ac/ : FAILURE in 24m 28s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/check/nova-grenade-live-migration/f5ad13a/ : SUCCESS in 1h 02m 05s\n- nova-live-migration http://logs.openstack.org/99/651299/5/check/nova-live-migration/7847ea8/ : SUCCESS in 42m 51s\n- nova-next http://logs.openstack.org/99/651299/5/check/nova-next/3cdfb64/ : FAILURE in 29m 02s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/check/nova-tox-functional/c32576b/ : SUCCESS in 18m 28s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/check/nova-tox-functional-py36/c474557/ : SUCCESS in 15m 12s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/check/tempest-slow-py3/a1f0a47/ : FAILURE in 38m 49s","accounts_in_message":[],"_revision_number":5},{"id":"83e6bbacb493cf7f8dad262668513a1d9ce89318","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-04-30 17:44:36.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"f783d63cf3153e35610233725be2670b44bd90e3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 17:46:09.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bfd9bd63d0732169200fb5f812fabda58cf5cb97","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 17:46:47.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/99/651299/5/check/check-dsvm-tempest-vz7-exe-minimal/3049cfc : FAILURE in 27s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"9dc4e06e3f5b18480404b0f1b0245e1122ee25ec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-04-30 18:52:45.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/99/651299/5/check-vote/ext-nova-zuul/6370442 : SUCCESS in 54m 30s","accounts_in_message":[],"_revision_number":5},{"id":"685ca9d741f238c490d9c3ce11c7612bcbb30ec9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-30 19:07:32.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial/064fdbd/ : SUCCESS in 1h 16m 14s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/tempest-dsvm-full-xenial-py3/1243e30/ : SUCCESS in 1h 22m 01s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/99/651299/5/check/grenade-dsvm-xenial/14d1fa9/ : SUCCESS in 1h 03m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b1ca94d6bb9476bc4c44da53dfe4a6c3a721c7f0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-04-30 19:08:29.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-31310 : FAILURE in 1h 23m 14s","accounts_in_message":[],"_revision_number":5},{"id":"2df1afd078b5d3dbb50660a95d28ec86e2ed1e48","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 19:56:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/check/grenade-py3/e5e40e3/ : SUCCESS in 56m 45s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/check/tempest-full-py3/0e85d57/ : SUCCESS in 1h 28m 21s\n- openstack-tox-cover http://logs.openstack.org/99/651299/5/check/openstack-tox-cover/8ab80e6/cover/ : SUCCESS in 15m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/check/openstack-tox-lower-constraints/a002d1a/ : SUCCESS in 14m 07s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/check/openstack-tox-pep8/ce00d14/ : SUCCESS in 10m 45s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/check/openstack-tox-py27/75e4e97/ : SUCCESS in 12m 36s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/check/openstack-tox-py36/9842c56/ : SUCCESS in 11m 31s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/check/openstack-tox-py37/b184e1d/ : SUCCESS in 12m 02s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/check/openstack-tox-docs/205b29d/html/ : SUCCESS in 5m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/99/651299/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8420f5c/ : SUCCESS in 49m 18s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/99/651299/5/check/devstack-plugin-ceph-tempest/30a0483/ : SUCCESS in 1h 24m 42s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/99/651299/5/check/neutron-grenade-multinode/cb1ae98/ : SUCCESS in 1h 07m 50s\n- neutron-tempest-linuxbridge http://logs.openstack.org/99/651299/5/check/neutron-tempest-linuxbridge/1d591f9/ : SUCCESS in 1h 33m 17s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/check/nova-grenade-live-migration/0ce6bfb/ : SUCCESS in 1h 06m 36s\n- nova-live-migration http://logs.openstack.org/99/651299/5/check/nova-live-migration/240c12c/ : SUCCESS in 47m 14s\n- nova-next http://logs.openstack.org/99/651299/5/check/nova-next/bed43b3/ : SUCCESS in 1h 20m 24s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/check/nova-tox-functional/2f4d41e/ : SUCCESS in 18m 54s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/check/nova-tox-functional-py36/c3ccf41/ : SUCCESS in 22m 45s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/check/tempest-slow-py3/4b1d6af/ : SUCCESS in 2h 06m 52s","accounts_in_message":[],"_revision_number":5},{"id":"73f5f503c1987a394096b06b6cd58b513939a604","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-04-30 19:57:05.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/99/651299/5/check/EMC_VxFlexOS_NOVA/8dda459/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-dell-emc-vxflexos\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/Dell_EMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"7be9162873e958f459a6a9dffb142811275dc6a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 19:57:15.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c3cca563ad47dba449714bbf4ad5f9781de4bfd2","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-04-30 20:09:44.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/99/651299/5/check/nova-out-of-tree-pvm/a35c147 : SUCCESS in 2h 16m 59s\n- nova-in-tree-pvm http://184.172.12.213/99/651299/5/check/nova-in-tree-pvm/ab16254 : FAILURE in 2h 24m 16s","accounts_in_message":[],"_revision_number":5},{"id":"2db9d706aa480e872a2526fdf7b1586db911f066","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-30 20:13:34.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/651299/5 : FAILURE in 2h 16m 40s","accounts_in_message":[],"_revision_number":5},{"id":"040015d83c0a3f5c0d52839e355c580ccf6f9183","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-04-30 21:09:57.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-99-651299-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"d0ee376afd83d049425826edf43b842856b08d17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 22:11:51.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/99/651299/5/gate/grenade-py3/fbb776c/ : SUCCESS in 50m 26s\n- tempest-full-py3 http://logs.openstack.org/99/651299/5/gate/tempest-full-py3/48b141d/ : SUCCESS in 1h 15m 11s\n- openstack-tox-lower-constraints http://logs.openstack.org/99/651299/5/gate/openstack-tox-lower-constraints/a179ea7/ : SUCCESS in 12m 55s\n- openstack-tox-pep8 http://logs.openstack.org/99/651299/5/gate/openstack-tox-pep8/2ae9f4b/ : SUCCESS in 13m 24s\n- openstack-tox-py27 http://logs.openstack.org/99/651299/5/gate/openstack-tox-py27/dbe5740/ : SUCCESS in 25m 26s\n- openstack-tox-py36 http://logs.openstack.org/99/651299/5/gate/openstack-tox-py36/af69967/ : SUCCESS in 18m 41s\n- openstack-tox-py37 http://logs.openstack.org/99/651299/5/gate/openstack-tox-py37/4c16e16/ : SUCCESS in 18m 25s\n- openstack-tox-docs http://logs.openstack.org/99/651299/5/gate/openstack-tox-docs/cdaffc6/html/ : SUCCESS in 5m 29s\n- nova-grenade-live-migration http://logs.openstack.org/99/651299/5/gate/nova-grenade-live-migration/15a16c8/ : SUCCESS in 1h 14m 42s\n- nova-live-migration http://logs.openstack.org/99/651299/5/gate/nova-live-migration/de644a2/ : SUCCESS in 56m 14s\n- nova-tox-functional http://logs.openstack.org/99/651299/5/gate/nova-tox-functional/48cf92a/ : SUCCESS in 24m 46s\n- nova-tox-functional-py36 http://logs.openstack.org/99/651299/5/gate/nova-tox-functional-py36/afaab28/ : SUCCESS in 29m 12s\n- nova-next http://logs.openstack.org/99/651299/5/gate/nova-next/4913883/ : SUCCESS in 1h 41m 52s\n- tempest-slow-py3 http://logs.openstack.org/99/651299/5/gate/tempest-slow-py3/c442c0c/ : SUCCESS in 2h 12m 13s","accounts_in_message":[],"_revision_number":5},{"id":"1289f8d6995875cb24d846e6a5c7e24b10e3b727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-30 22:11:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e3a8188e21b409f1720fd2a9b40a66c51571b210","revisions":{"72515fc0203d1dcf8ac9e0fc7d650a7ced0fcd45":{"kind":"REWORK","_number":1,"created":"2019-04-09 16:15:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/651299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/99/651299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/99/651299/1"}}},"commit":{"parents":[{"commit":"0f53a8837893a653d1c29c146af94014e155ff92","subject":"Remove \u0027instance_fault_create_at_top\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0f53a8837893a653d1c29c146af94014e155ff92"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-04 15:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-09 16:15:37.000000000","tz":60},"subject":"Remove \u0027instance_info_cache_update_at_top\u0027","message":"Remove \u0027instance_info_cache_update_at_top\u0027\n\nThis was being called within the \u0027InstanceInfoCache\u0027 object. Remove it\nin its entirety, given that nothing should be calling it now.\n\nAn \u0027update_cells\u0027 argument is removed from a number of functions as\nthese all resulted in a call to the cells v1 RPC API, a call that is now\nremoved.\n\nPart of blueprint remove-cells-v1\n\nChange-Id: Ia8da8cd040d104d745bd7adf20438b8eec558680\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72515fc0203d1dcf8ac9e0fc7d650a7ced0fcd45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72515fc0203d1dcf8ac9e0fc7d650a7ced0fcd45"}]},"branch":"refs/heads/master"},"d6ce99e321dd511f0b4373afe0d7ee985f1ae3ca":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-04-16 08:52:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/651299/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/99/651299/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/99/651299/2"}}},"commit":{"parents":[{"commit":"dd189c0214008b80594db7e7afb39cc899c30d53","subject":"Remove \u0027instance_fault_create_at_top\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd189c0214008b80594db7e7afb39cc899c30d53"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-04 15:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-16 08:47:27.000000000","tz":60},"subject":"Remove \u0027instance_info_cache_update_at_top\u0027","message":"Remove \u0027instance_info_cache_update_at_top\u0027\n\nThis was being called within the \u0027InstanceInfoCache\u0027 object. Remove it\nin its entirety, given that nothing should be calling it now.\n\nAn \u0027update_cells\u0027 argument is removed from a number of functions as\nthese all resulted in a call to the cells v1 RPC API, a call that is now\nremoved.\n\nPart of blueprint remove-cells-v1\n\nChange-Id: Ia8da8cd040d104d745bd7adf20438b8eec558680\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6ce99e321dd511f0b4373afe0d7ee985f1ae3ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d6ce99e321dd511f0b4373afe0d7ee985f1ae3ca"}]},"branch":"refs/heads/master"},"2b9938036d7774f23a266d1d1e1cf6a6af160f5b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-04-16 14:22:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/651299/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/99/651299/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/99/651299/3"}}},"commit":{"parents":[{"commit":"dcba539f72f06565cb532f060bb9a612e6f6a6d2","subject":"Remove \u0027instance_fault_create_at_top\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dcba539f72f06565cb532f060bb9a612e6f6a6d2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-04 15:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-16 14:22:33.000000000","tz":60},"subject":"Remove \u0027instance_info_cache_update_at_top\u0027","message":"Remove \u0027instance_info_cache_update_at_top\u0027\n\nThis was being called within the \u0027InstanceInfoCache\u0027 object. Remove it\nin its entirety, given that nothing should be calling it now.\n\nAn \u0027update_cells\u0027 argument is removed from a number of functions as\nthese all resulted in a call to the cells v1 RPC API, a call that is now\nremoved.\n\nPart of blueprint remove-cells-v1\n\nChange-Id: Ia8da8cd040d104d745bd7adf20438b8eec558680\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2b9938036d7774f23a266d1d1e1cf6a6af160f5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2b9938036d7774f23a266d1d1e1cf6a6af160f5b"}]},"branch":"refs/heads/master"},"54cde449516bc4823c6be301fd9ad553c63ac388":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-04-16 17:26:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/651299/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/99/651299/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/99/651299/4"}}},"commit":{"parents":[{"commit":"0b48d9148bbde88bb6b50329f76b6a25db9ac986","subject":"Remove \u0027instance_fault_create_at_top\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0b48d9148bbde88bb6b50329f76b6a25db9ac986"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-04 15:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-16 17:26:17.000000000","tz":60},"subject":"Remove \u0027instance_info_cache_update_at_top\u0027","message":"Remove \u0027instance_info_cache_update_at_top\u0027\n\nThis was being called within the \u0027InstanceInfoCache\u0027 object. Remove it\nin its entirety, given that nothing should be calling it now.\n\nAn \u0027update_cells\u0027 argument is removed from a number of functions as\nthese all resulted in a call to the cells v1 RPC API, a call that is now\nremoved.\n\nPart of blueprint remove-cells-v1\n\nChange-Id: Ia8da8cd040d104d745bd7adf20438b8eec558680\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/54cde449516bc4823c6be301fd9ad553c63ac388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/54cde449516bc4823c6be301fd9ad553c63ac388"}]},"branch":"refs/heads/master"},"e3a8188e21b409f1720fd2a9b40a66c51571b210":{"kind":"REWORK","_number":5,"created":"2019-04-29 16:55:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/651299/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/99/651299/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/99/651299/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/99/651299/5"}}},"commit":{"parents":[{"commit":"65848706079b10a2989a83d1c5c4bb872afa7098","subject":"Remove \u0027instance_fault_create_at_top\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65848706079b10a2989a83d1c5c4bb872afa7098"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-04 15:40:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-04-29 16:55:41.000000000","tz":60},"subject":"Remove \u0027instance_info_cache_update_at_top\u0027","message":"Remove \u0027instance_info_cache_update_at_top\u0027\n\nThis was being called within the \u0027InstanceInfoCache\u0027 object. Remove it\nin its entirety, given that nothing should be calling it now.\n\nAn \u0027update_cells\u0027 argument is removed from a number of functions as\nthese all resulted in a call to the cells v1 RPC API, a call that is now\nremoved.\n\nPart of blueprint remove-cells-v1\n\nChange-Id: Ia8da8cd040d104d745bd7adf20438b8eec558680\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3a8188e21b409f1720fd2a9b40a66c51571b210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3a8188e21b409f1720fd2a9b40a66c51571b210"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
