)]}'
{"id":"openstack%2Fnova~653811","triplet_id":"openstack%2Fnova~stable%2Fpike~Ib28dc7feb40470919f8a60f7b05ebc0f85f60856","project":"openstack/nova","branch":"stable/pike","topic":"nova-grenade-live-migration","hashtags":[],"change_id":"Ib28dc7feb40470919f8a60f7b05ebc0f85f60856","subject":"Drop legacy-grenade-dsvm-neutron-multinode-live-migration from Pike","status":"ABANDONED","created":"2019-04-18 16:35:37.000000000","updated":"2019-04-22 12:56:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca6b46be7e84e539b82f813df3541e636fb0e20a","_number":653811,"virtual_id_number":653811,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"date":"2019-04-18 18:53:43.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-04-18 17:52:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":-1,"date":"2019-04-19 06:43:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2019-04-18 17:13:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":-1,"date":"2019-04-18 18:09:20.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-18 17:13:12.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-04-18 17:52:49.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-18 18:53:43.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-04-19 06:43:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"790876a27ad9d3fe4255ca4fb588a7996e9eb7e1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-18 16:35:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"969f1383c5fa44825214a0813cca4dc074a6ff32","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-04-18 17:13:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac85d378249ec8e1dc17824e062a275fe37b280f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-04-18 17:52:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/11/653811/1/check/tempest-dsvm-full-xenial/0b229f2/ : SUCCESS in 1h 16m 17s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/11/653811/1/check/tempest-dsvm-full-xenial-py3/2e9b3d4/ : FAILURE in 56m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c9f1ba90ea32a7aee08ff8c56ea4400ae130ea5","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-18 18:07:57.000000000","message":"Patch Set 1:\n\nHmm yeah I didn\u0027t realize we already had ceph disabled in this job in pike. Given that, maybe we don\u0027t really need this since on queens I\u0027m just disabling ceph to keep the job.","accounts_in_message":[],"_revision_number":1},{"id":"f1db02e027b2b8ed3a81feabd7a0eaa70d473792","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-18 18:09:20.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027m going to restore and rebase https://review.openstack.org/#/c/640207/ to see why the job was failing there, maybe it was unrelated.","accounts_in_message":[],"_revision_number":1},{"id":"3321b8c57a19511dabc08606d7401be1808a737f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-04-18 18:53:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/653811/1 : FAILURE in 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"012e98d93a6038720ad40e58c7c9edbf03771800","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-19 06:43:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/11/653811/1/check/openstack-tox-pep8/adb7668/ : SUCCESS in 9m 31s\n- openstack-tox-py27 http://logs.openstack.org/11/653811/1/check/openstack-tox-py27/4990e9a/ : SUCCESS in 14m 58s\n- openstack-tox-py35 http://logs.openstack.org/11/653811/1/check/openstack-tox-py35/cf84c0f/ : SUCCESS in 14m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/11/653811/1/check/build-openstack-sphinx-docs/517c253/html/ : SUCCESS in 6m 04s\n- tempest-full finger://ze12.openstack.org/498d92db61d1427aac0a35ed6c77a21b : RETRY_LIMIT in 41m 11s\n- neutron-grenade http://logs.openstack.org/11/653811/1/check/neutron-grenade/bc62f74/ : SUCCESS in 56m 21s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/11/653811/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/1f7b738/ : SUCCESS in 46m 50s (non-voting)\n- nova-live-migration http://logs.openstack.org/11/653811/1/check/nova-live-migration/9898b03/ : SUCCESS in 41m 19s\n- tempest-slow finger://ze08.openstack.org/3935d424cbcf48098a73b869bb58d496 : RETRY_LIMIT in 34m 57s\n- nova-tox-functional http://logs.openstack.org/11/653811/1/check/nova-tox-functional/7e0032e/ : SUCCESS in 21m 27s\n- nova-tox-functional-py35 http://logs.openstack.org/11/653811/1/check/nova-tox-functional-py35/2442ef6/ : SUCCESS in 17m 10s\n- neutron-grenade-multinode http://logs.openstack.org/11/653811/1/check/neutron-grenade-multinode/9da0eda/ : SUCCESS in 1h 02m 43s\n- devstack-plugin-ceph-tempest finger://ze11.openstack.org/8999a5d04d6e45d094a9c2c519336cf4 : RETRY_LIMIT in 38m 42s (non-voting)\n- tempest-multinode-full finger://ze06.openstack.org/614413f66819422db22a9c9d6b472fca : RETRY_LIMIT in 41m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c1e8b10db7ccdeffc0d06898db288736d7170b8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-04-22 12:56:58.000000000","message":"Abandoned\n\nhttps://review.opendev.org/#/c/640207/ is working.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ececd0929799a5203bf5fc70b6330f6427e1ed71","revisions":{"ececd0929799a5203bf5fc70b6330f6427e1ed71":{"kind":"REWORK","_number":1,"created":"2019-04-18 16:35:37.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/11/653811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/11/653811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/653811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/653811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/11/653811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/11/653811/1"}}},"commit":{"parents":[{"commit":"919108321180e36c45abf58f5659c07d0f5f69d7","subject":"Merge \"Do not persist RequestSpec.ignore_hosts\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/919108321180e36c45abf58f5659c07d0f5f69d7"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-18 16:32:14.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-18 16:32:14.000000000","tz":-240},"subject":"Drop legacy-grenade-dsvm-neutron-multinode-live-migration from Pike","message":"Drop legacy-grenade-dsvm-neutron-multinode-live-migration from Pike\n\nPer change Ie9b61775dbb92b10237688eaddaca606c1c73a23 the\nlegacy-grenade-dsvm-neutron-multinode-live-migration is being\nmoved in-tree and renamed to nova-grenade-live-migration but\ndoing that in Pike at this point is complicated because of\nbug 1813216. Trying to make this work in Pike is also very\nlow priority since we will soon be putting the branch into\nextended maintenance mode where a reduction in test coverage\nis expected. As a result, this change simply removes the (broken)\nnon-voting job from running no stable/pike changes.\n\nChange-Id: Ib28dc7feb40470919f8a60f7b05ebc0f85f60856\nRelated-Bug: #1813216\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ececd0929799a5203bf5fc70b6330f6427e1ed71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ececd0929799a5203bf5fc70b6330f6427e1ed71"}]},"parents_data":[{"branch_name":"refs/heads/stable/pike","commit_id":"919108321180e36c45abf58f5659c07d0f5f69d7","is_merged_in_target_branch":false}],"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
