)]}'
{"id":"openstack%2Fnova~656666","triplet_id":"openstack%2Fnova~stable%2Fstein~I91c9c1e88afa035c757f692c78ad72d69cc3c431","project":"openstack/nova","branch":"stable/stein","topic":"bug/1825020","hashtags":[],"change_id":"I91c9c1e88afa035c757f692c78ad72d69cc3c431","subject":"Fix volume-backed resize with a smaller disk flavor","status":"MERGED","created":"2019-05-01 16:17:19.000000000","updated":"2019-05-02 19:48:30.000000000","submitted":"2019-05-02 19:48:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656666-1556826510212-54e3a629","meta_rev_id":"a465eab882c69d70b1b166f90e8f2bafdd874e88","_number":656666,"virtual_id_number":656666,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-05-02 02:40:08.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-05-02 03:04:30.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-05-02 11:19:18.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2019-05-02 19:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-02 15:15:34.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-02 15:15:34.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-02 02:40:08.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-02 03:04:30.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-02 11:19:18.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-02 15:15:34.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-05-02 19:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f646a467b1d317045c76700476ad777c9b69f764","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-05-01 16:17:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29f66b59d609e5111067833dfe89101dd7a02557","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-01 20:00:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/66/656666/1/check/grenade-py3/440b2d3/ : SUCCESS in 54m 43s\n- tempest-full-py3 http://logs.openstack.org/66/656666/1/check/tempest-full-py3/2af3be9/ : SUCCESS in 1h 30m 58s\n- openstack-tox-cover http://logs.openstack.org/66/656666/1/check/openstack-tox-cover/53c7da2/cover/ : SUCCESS in 14m 36s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/656666/1/check/openstack-tox-lower-constraints/5c1e979/ : SUCCESS in 12m 07s\n- openstack-tox-pep8 http://logs.openstack.org/66/656666/1/check/openstack-tox-pep8/3251db7/ : SUCCESS in 9m 29s\n- openstack-tox-py27 http://logs.openstack.org/66/656666/1/check/openstack-tox-py27/6002bfd/ : SUCCESS in 12m 31s\n- openstack-tox-py35 http://logs.openstack.org/66/656666/1/check/openstack-tox-py35/fcdd9e5/ : SUCCESS in 14m 56s\n- openstack-tox-py36 http://logs.openstack.org/66/656666/1/check/openstack-tox-py36/0a416dd/ : SUCCESS in 10m 32s\n- openstack-tox-py37 http://logs.openstack.org/66/656666/1/check/openstack-tox-py37/f8123c7/ : SUCCESS in 14m 30s\n- openstack-tox-docs http://logs.openstack.org/66/656666/1/check/openstack-tox-docs/a11c24d/html/ : SUCCESS in 6m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/66/656666/1/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/8883413/ : FAILURE in 28m 17s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/66/656666/1/check/devstack-plugin-ceph-tempest/170e7b2/ : SUCCESS in 1h 32m 31s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/66/656666/1/check/neutron-grenade-multinode/2eef1a2/ : SUCCESS in 1h 09m 34s\n- nova-grenade-live-migration http://logs.openstack.org/66/656666/1/check/nova-grenade-live-migration/62ba0a2/ : SUCCESS in 1h 03m 34s\n- nova-live-migration http://logs.openstack.org/66/656666/1/check/nova-live-migration/d473af1/ : SUCCESS in 48m 30s\n- nova-next http://logs.openstack.org/66/656666/1/check/nova-next/166e238/ : SUCCESS in 1h 18m 07s\n- nova-tox-functional http://logs.openstack.org/66/656666/1/check/nova-tox-functional/892d9a7/ : SUCCESS in 31m 12s\n- nova-tox-functional-py35 http://logs.openstack.org/66/656666/1/check/nova-tox-functional-py35/f378caf/ : SUCCESS in 33m 24s\n- tempest-multinode-full http://logs.openstack.org/66/656666/1/check/tempest-multinode-full/4d77549/ : FAILURE in 1h 54m 18s (non-voting)\n- tempest-slow-py3 http://logs.openstack.org/66/656666/1/check/tempest-slow-py3/44068dd/ : SUCCESS in 2h 09m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4ca30386da8fe2909c79852773779887ff7fee38","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-02 02:40:08.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/656666/1 : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"44f3da9f1638ad76ff246f7567abab9852925e64","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-05-02 03:04:30.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/656666/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/a593877 : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e548fea7480d98e505d968bc7121e889b89dae25","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-02 11:19:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/66/656666/1/check/tempest-dsvm-full-xenial/c458107/ : SUCCESS in 1h 16m 58s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/66/656666/1/check/tempest-dsvm-full-xenial-py3/a211a32/ : SUCCESS in 1h 19m 59s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/66/656666/1/check/grenade-dsvm-xenial/90f71f9/ : SUCCESS in 58m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb25aaab4603b2d6b2c38299b8ccfef80095501f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-05-02 15:15:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM, clean and useful. Approving as this is from another Nova stable core.","accounts_in_message":[],"_revision_number":1},{"id":"609c48e7f949c6aadec839f4b98b96abcae3917f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 15:16:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a49c45bf61f8aacf6622c85781a0ee1265059d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 19:48:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/66/656666/1/gate/grenade-py3/126cf24/ : SUCCESS in 1h 00m 34s\n- tempest-full-py3 http://logs.openstack.org/66/656666/1/gate/tempest-full-py3/10e3b48/ : SUCCESS in 1h 35m 06s\n- openstack-tox-lower-constraints http://logs.openstack.org/66/656666/1/gate/openstack-tox-lower-constraints/d2e74d0/ : SUCCESS in 11m 39s\n- openstack-tox-pep8 http://logs.openstack.org/66/656666/1/gate/openstack-tox-pep8/3ea3cde/ : SUCCESS in 9m 50s\n- openstack-tox-py27 http://logs.openstack.org/66/656666/1/gate/openstack-tox-py27/57cd381/ : SUCCESS in 11m 59s\n- openstack-tox-py35 http://logs.openstack.org/66/656666/1/gate/openstack-tox-py35/13e64f4/ : SUCCESS in 13m 09s\n- openstack-tox-py36 http://logs.openstack.org/66/656666/1/gate/openstack-tox-py36/281a532/ : SUCCESS in 14m 06s\n- openstack-tox-py37 http://logs.openstack.org/66/656666/1/gate/openstack-tox-py37/bc902fe/ : SUCCESS in 12m 27s\n- openstack-tox-docs http://logs.openstack.org/66/656666/1/gate/openstack-tox-docs/ad1f753/html/ : SUCCESS in 6m 04s\n- nova-grenade-live-migration http://logs.openstack.org/66/656666/1/gate/nova-grenade-live-migration/3b30f1d/ : SUCCESS in 1h 12m 36s\n- nova-live-migration http://logs.openstack.org/66/656666/1/gate/nova-live-migration/69e137f/ : SUCCESS in 1h 06m 11s\n- nova-tox-functional http://logs.openstack.org/66/656666/1/gate/nova-tox-functional/b4c770e/ : SUCCESS in 30m 13s\n- nova-tox-functional-py35 http://logs.openstack.org/66/656666/1/gate/nova-tox-functional-py35/33afbb0/ : SUCCESS in 25m 52s\n- nova-next http://logs.openstack.org/66/656666/1/gate/nova-next/0aa259b/ : SUCCESS in 1h 42m 20s\n- tempest-slow-py3 http://logs.openstack.org/66/656666/1/gate/tempest-slow-py3/279499f/ : SUCCESS in 2h 25m 35s","accounts_in_message":[],"_revision_number":1},{"id":"46b05b76adeeeea1b1cf4a10ead98529ff4dae4f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-02 19:48:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3285210add2758ad0182358317fabca626cdb0ee","revisions":{"3285210add2758ad0182358317fabca626cdb0ee":{"kind":"REWORK","_number":1,"created":"2019-05-01 16:17:19.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/66/656666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/656666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/656666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/656666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/656666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/656666/1"}}},"commit":{"parents":[{"commit":"a1db4507ac832913a911afbee90f76f576d8ea20","subject":"Add functional regression recreate test for bug 1825020","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a1db4507ac832913a911afbee90f76f576d8ea20"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-04-17 03:53:03.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-01 16:13:07.000000000","tz":-240},"subject":"Fix volume-backed resize with a smaller disk flavor","message":"Fix volume-backed resize with a smaller disk flavor\n\nChange I06fad233006c7bab14749a51ffa226c3801f951b in Stein started\ncalling _validate_flavor_image_nostatus during resize but did not\npass the root_bdm since the underlying volume does not change. This\ncan, however, lead to FlavorDiskSmallerThanMinDisk being raised\nif the new flavor has a disk size smaller than the current flavor.\nThis is wrong in the case of a volume-backed server because we don\u0027t\ncare about the root disk size in the flavor in that case.\n\nThis fixes the bug by splitting the pci/numa validation logic out\nof _validate_flavor_image_nostatus into a separate method so that\nthe resize method can call that directly for a volume-backed server\nrather than deal with the complicated disk size logic in\n_validate_flavor_image_nostatus (see bug 1646740 for details, but\ntl;dr if the image min_disk is less than the flavor root_gb, the\nmin_disk stored in the instance image system metadata is the\nflavor[\u0027root_gb\u0027] which could be larger than the root volume size).\nTo see why trying to use _validate_flavor_image_nostatus during\nresize for a volume-backed server is a mess, look at PS1 of this\nchange.\n\nChange-Id: I91c9c1e88afa035c757f692c78ad72d69cc3c431\nCloses-Bug: #1825020\n(cherry picked from commit 8aae3e39e58a05ed50b891aeae261641f544ad1f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3285210add2758ad0182358317fabca626cdb0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3285210add2758ad0182358317fabca626cdb0ee"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
