)]}'
{"id":"openstack%2Fnova~658533","triplet_id":"openstack%2Fnova~stable%2Frocky~I8036dc4351aa4b77f41b44622ae030cc8ae51d2f","project":"openstack/nova","branch":"stable/rocky","topic":"bp/use-nested-allocation-candidates","hashtags":[],"change_id":"I8036dc4351aa4b77f41b44622ae030cc8ae51d2f","subject":"Enable nested allocation candidates in scheduler","status":"ABANDONED","created":"2019-05-11 08:43:09.000000000","updated":"2019-05-30 04:48:01.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cc8889a28e71b15e69c074c71f6448f877417b35","_number":658533,"virtual_id_number":658533,"owner":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-05-12 02:56:45.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"date":"2019-05-12 02:54:06.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"date":"2019-05-12 01:13:21.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"date":"2019-05-14 15:33:23.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-05-12 03:50:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2019-05-13 08:21:09.000000000","_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"date":"2019-05-13 07:52:19.000000000","_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2019-05-12 03:03:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"all":[{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":-1,"date":"2019-05-14 07:51:11.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-11 08:44:03.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-05-12 01:13:21.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-12 02:54:06.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-05-12 02:56:45.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-12 03:03:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-12 03:50:11.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-13 07:52:19.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-05-13 08:21:09.000000000","updated_by":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"reviewer":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"state":"REVIEWER"},{"updated":"2019-05-14 15:33:23.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"}],"messages":[{"id":"571c214e1b13c6fa7c7a46cd0d808a624fbadd23","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-11 08:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4228544b2540b3a8a484eac439870a0a8786b33","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-11 08:53:37.000000000","message":"Patch Set 1:\n\nNote for reviewers: According to the placement profiler using https://review.opendev.org/#/c/643269, I\u0027ve found that _exclude_nested_providers() consumes more than half of the execution time (in my local environment experiment) using microversion 1.21. This is not a problem for microversion \u003e 1.29 which is the case for nova in stein+ releases, but is a problem for nova in rocky release so...","accounts_in_message":[],"_revision_number":1},{"id":"9ab046abae2bf4ef20740421e67cc8ec4410e107","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-05-11 09:08:35.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/658533/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1a5e53a : FAILURE in 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"91eeb4f91d057935b6ef84cc830efc92d497333a","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-11 09:33:05.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/33/658533/1/check/nova-out-of-tree-pvm/c70d109 : FAILURE in 49m 02s\n- nova-in-tree-pvm http://184.172.12.213/33/658533/1/check/nova-in-tree-pvm/4390e9e : FAILURE in 48m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c194708228b8d00f28789f3df1aaaa3f93227490","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-11 09:48:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/33/658533/1/check/openstack-tox-pep8/a5e2701/ : SUCCESS in 9m 01s\n- openstack-tox-py27 http://logs.openstack.org/33/658533/1/check/openstack-tox-py27/57d5af1/ : SUCCESS in 11m 37s\n- openstack-tox-py35 http://logs.openstack.org/33/658533/1/check/openstack-tox-py35/5cd3ffa/ : SUCCESS in 13m 05s\n- build-openstack-sphinx-docs http://logs.openstack.org/33/658533/1/check/build-openstack-sphinx-docs/bb7f74c/html/ : SUCCESS in 6m 19s\n- tempest-full http://logs.openstack.org/33/658533/1/check/tempest-full/c17077e/ : FAILURE in 57m 41s\n- neutron-grenade http://logs.openstack.org/33/658533/1/check/neutron-grenade/01dbe14/ : FAILURE in 50m 45s\n- tempest-full-py3 http://logs.openstack.org/33/658533/1/check/tempest-full-py3/e7fe8a4/ : FAILURE in 53m 01s\n- build-openstack-releasenotes http://logs.openstack.org/33/658533/1/check/build-openstack-releasenotes/ea454df/html/ : SUCCESS in 38m 41s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/33/658533/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/887fc3a/ : FAILURE in 31m 53s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/33/658533/1/check/nova-cells-v1/70e59ac/ : FAILURE in 52m 30s\n- nova-grenade-live-migration http://logs.openstack.org/33/658533/1/check/nova-grenade-live-migration/1f0812e/ : FAILURE in 45m 57s (non-voting)\n- nova-live-migration http://logs.openstack.org/33/658533/1/check/nova-live-migration/114330f/ : FAILURE in 36m 09s\n- nova-multiattach http://logs.openstack.org/33/658533/1/check/nova-multiattach/ed04788/ : FAILURE in 29m 50s\n- nova-next http://logs.openstack.org/33/658533/1/check/nova-next/f3ea6b3/ : FAILURE in 57m 32s\n- tempest-slow http://logs.openstack.org/33/658533/1/check/tempest-slow/bae6157/ : FAILURE in 1h 02m 11s\n- nova-tox-functional http://logs.openstack.org/33/658533/1/check/nova-tox-functional/8996653/ : FAILURE in 16m 48s\n- nova-tox-functional-py35 http://logs.openstack.org/33/658533/1/check/nova-tox-functional-py35/513165b/ : FAILURE in 16m 43s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/658533/1/check/openstack-tox-lower-constraints/b2b6a7c/ : SUCCESS in 11m 06s\n- neutron-grenade-multinode http://logs.openstack.org/33/658533/1/check/neutron-grenade-multinode/e58ae1d/ : FAILURE in 59m 04s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/33/658533/1/check/devstack-plugin-ceph-tempest/2bfa218/ : FAILURE in 52m 24s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/33/658533/1/check/tempest-multinode-full/d5ae351/ : FAILURE in 1h 02m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1be56c4e7d0c18c3a1698a097a2f6ade6662d7e9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-11 09:55:15.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/33/658533/1/check/tempest-dsvm-full-xenial/20da62e/ : FAILURE in 1h 11m 12s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/33/658533/1/check/tempest-dsvm-full-xenial-py3/33ef141/ : FAILURE in 1h 09m 03s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/33/658533/1/check/grenade-dsvm-xenial/0227632/ : FAILURE in 1h 09m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f153ba64c64b6453a03ad74322448f482652c191","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-11 09:57:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/658533/1 : FAILURE in 1h 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ddc2acb1f3c339655613edc747875f89a3d3dda4","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-12 00:48:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"259b73498542062df838bec60694617fee3aceb9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-05-12 01:13:21.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/658533/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/d5a9dde : FAILURE in 24m 21s","accounts_in_message":[],"_revision_number":2},{"id":"8831f9a313bea8ed972459391e490de86ec702bf","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-05-12 02:54:06.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/658533/2 : SUCCESS in 2h 05m 10s","accounts_in_message":[],"_revision_number":2},{"id":"89ad51a4621185b635a9527e085dbae574200d4e","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-05-12 02:56:45.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/33/658533/2/check/nova-out-of-tree-pvm/5ce1629 : FAILURE in 14m 01s\n- nova-in-tree-pvm http://184.172.12.213/33/658533/2/check/nova-in-tree-pvm/4d02228 : FAILURE in 1h 33m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1afc9813e802167e18b1cb71ce23a8efdfe860e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 03:03:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/33/658533/2/check/openstack-tox-pep8/04ec764/ : SUCCESS in 9m 46s\n- openstack-tox-py27 http://logs.openstack.org/33/658533/2/check/openstack-tox-py27/7763000/ : SUCCESS in 13m 59s\n- openstack-tox-py35 http://logs.openstack.org/33/658533/2/check/openstack-tox-py35/ed0d631/ : SUCCESS in 14m 19s\n- build-openstack-sphinx-docs http://logs.openstack.org/33/658533/2/check/build-openstack-sphinx-docs/9d42e0e/html/ : SUCCESS in 7m 25s\n- tempest-full http://logs.openstack.org/33/658533/2/check/tempest-full/e9fb676/ : SUCCESS in 1h 37m 16s\n- neutron-grenade http://logs.openstack.org/33/658533/2/check/neutron-grenade/83647c1/ : SUCCESS in 50m 48s\n- tempest-full-py3 http://logs.openstack.org/33/658533/2/check/tempest-full-py3/5a2ea92/ : SUCCESS in 1h 04m 18s\n- build-openstack-releasenotes http://logs.openstack.org/33/658533/2/check/build-openstack-releasenotes/0d10c5d/html/ : SUCCESS in 32m 14s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/33/658533/2/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/492d6e1/ : SUCCESS in 41m 09s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/33/658533/2/check/nova-cells-v1/cc3ac63/ : SUCCESS in 51m 23s\n- nova-grenade-live-migration http://logs.openstack.org/33/658533/2/check/nova-grenade-live-migration/0ae7ac6/ : SUCCESS in 1h 00m 22s (non-voting)\n- nova-live-migration http://logs.openstack.org/33/658533/2/check/nova-live-migration/10163a7/ : SUCCESS in 47m 22s\n- nova-multiattach http://logs.openstack.org/33/658533/2/check/nova-multiattach/9a9a60f/ : SUCCESS in 56m 05s\n- nova-next http://logs.openstack.org/33/658533/2/check/nova-next/25f5052/ : SUCCESS in 1h 24m 11s\n- tempest-slow http://logs.openstack.org/33/658533/2/check/tempest-slow/8713d18/ : SUCCESS in 2h 01m 20s\n- nova-tox-functional http://logs.openstack.org/33/658533/2/check/nova-tox-functional/22465be/ : SUCCESS in 15m 57s\n- nova-tox-functional-py35 http://logs.openstack.org/33/658533/2/check/nova-tox-functional-py35/67e4fe1/ : SUCCESS in 15m 15s\n- openstack-tox-lower-constraints http://logs.openstack.org/33/658533/2/check/openstack-tox-lower-constraints/495c98c/ : SUCCESS in 13m 28s\n- neutron-grenade-multinode http://logs.openstack.org/33/658533/2/check/neutron-grenade-multinode/72ee84d/ : SUCCESS in 51m 23s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/33/658533/2/check/devstack-plugin-ceph-tempest/a8691d1/ : SUCCESS in 1h 22m 08s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/33/658533/2/check/tempest-multinode-full/6f02e2e/ : SUCCESS in 1h 30m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b656d3af837c1e2c478456652bda68717ddff250","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-05-12 03:50:11.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/33/658533/2/check/tempest-dsvm-full-xenial/e5e72cf/ : SUCCESS in 1h 31m 07s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/33/658533/2/check/tempest-dsvm-full-xenial-py3/0fee711/ : SUCCESS in 1h 30m 16s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/33/658533/2/check/grenade-dsvm-xenial/0d00a1d/ : SUCCESS in 1h 17m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9f0cc94618a5f0f290febcddfbed3233eaf5013d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-05-13 07:52:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef7f750ce7388e3f23d945bbacb7bf5d8e7388df","author":{"_account_id":26936,"name":"Surya Seetharaman","email":"suryaseetharaman.9@gmail.com","username":"tssurya"},"date":"2019-05-13 08:21:09.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"fe6ccd5ca469a5a2d9bc6c413ec351aa33958b90","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-14 07:51:11.000000000","message":"Patch Set 2: Workflow-1\n\nI\u0027d like to abandon this if there is no objection giving preference to https://review.opendev.org/#/c/658977/, which is backportable towards rocky.\n\nBug reported in\n* https://storyboard.openstack.org/#!/story/2005669\n* https://bugs.launchpad.net/nova/rocky/+bug/1828937\n(same report)","accounts_in_message":[],"_revision_number":2},{"id":"69022b150cbe3c7cd94415648206bb31e1cb354f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-05-14 15:33:23.000000000","message":"Patch Set 2:\n\n\u003e preference to https://review.opendev.org/#/c/658977/\n\nI\u0027m +2 on ^ - a very clean solution to the problem.\n\n\u003e which is backportable towards rocky.\n\nThis could be tricky (see comments on above patch), but I would be in favor.\n\nSo yes, I think we can abandon this one.","accounts_in_message":[],"_revision_number":2},{"id":"7b893bc8d0b1397e536847755bbafee77f7f239f","author":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"date":"2019-05-30 04:48:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2a15e3476550b3a26c4f8e95e862688b8ed6c7a","revisions":{"d88b59cd8f24d4d64514b0bf676de032e00c639a":{"kind":"REWORK","_number":1,"created":"2019-05-11 08:43:09.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/33/658533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/658533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/658533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/658533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/658533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/658533/1"}}},"commit":{"parents":[{"commit":"8e130e2be7c2eb604d53fd6053254c6b54d47524","subject":"Use migration_status during volume migrating and retyping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e130e2be7c2eb604d53fd6053254c6b54d47524"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2018-07-25 07:51:00.000000000","tz":540},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-11 08:37:27.000000000","tz":0},"subject":"Enable nested allocation candidates in scheduler","message":"Enable nested allocation candidates in scheduler\n\nThis patch changes scheduler client to use 1.29 microversion,\nin which nested allocation candidates feature is enabled,\nwhen getting allocation candidates.\n\nNOTE: The nested feature is not needed for the release since none of\nthe projects including nova had started to mode nested resource\nproviders, but this was backported because of the following reason.\n\nWhen scheduling an instance, getting allocation candidates were slow.\nThis was because the scheduler\u0027s placement client used older\nmicroversion, ``1.25`` where placement had to filter the nested\nresource providers and all the resource information for the backward\ncompatibility which were added in the newer version, ``1.26`` and\n``1.29`` respectively, for the backward compatibility.\n\nThis patch fixes it to use ``1.29`` for getting allocation candidates\nto remove the filtering the process improving the performance.\n\nThis doesn\u0027t break the API. In ``1.26`` the returned resource in\nprovider summaries from placement is exposed to nova but not to end\nusers. In ``1.29`` nested providers are returned, but again, in rocky\nrelease none of the projects includeing nova started to model nested\nresource providers.\n\nBlueprint: use-nested-allocation-candidates\nChange-Id: I8036dc4351aa4b77f41b44622ae030cc8ae51d2f\n(cherry picked from commit dc8ff63fe4f9c4d916c1a4aaeed202e186f8090e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d88b59cd8f24d4d64514b0bf676de032e00c639a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d88b59cd8f24d4d64514b0bf676de032e00c639a"}]},"branch":"refs/heads/stable/rocky"},"a2a15e3476550b3a26c4f8e95e862688b8ed6c7a":{"kind":"REWORK","_number":2,"created":"2019-05-12 00:48:15.000000000","uploader":{"_account_id":25625,"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","username":"tetsuro0907"},"ref":"refs/changes/33/658533/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/658533/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/658533/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/658533/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/658533/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/658533/2"}}},"commit":{"parents":[{"commit":"8e130e2be7c2eb604d53fd6053254c6b54d47524","subject":"Use migration_status during volume migrating and retyping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e130e2be7c2eb604d53fd6053254c6b54d47524"}]}],"author":{"name":"Tetsuro Nakamura","email":"nakamura.tetsuro@lab.ntt.co.jp","date":"2018-07-25 07:51:00.000000000","tz":540},"committer":{"name":"Tetsuro Nakamura","email":"tetsuro.nakamura.bc@hco.ntt.co.jp","date":"2019-05-12 00:45:45.000000000","tz":0},"subject":"Enable nested allocation candidates in scheduler","message":"Enable nested allocation candidates in scheduler\n\nThis patch changes scheduler client to use 1.29 microversion,\nin which nested allocation candidates feature is enabled,\nwhen getting allocation candidates.\n\nNOTE: The nested feature is not needed for the release since none of\nthe projects including nova had started to model nested resource\nproviders, but this was backported because of the perfomance issue.\n\nWhen scheduling an instance, getting allocation candidates were slow.\nThis was because the scheduler\u0027s placement client used older\nmicroversion ``1.25`` where placement had to filter out nested\nresource providers from allcation requests and filter out some of the\nresource information from provider summaries, which were added in\nmicroversion ``1.26`` and ``1.29`` respectively for the backward\ncompatibility.\n\nThis patch fixes nova side to use microversion ``1.29`` for getting\nallocation candidates to skip the filtering process to improve the\nperformance.\n\nThis doesn\u0027t break nova API. This is because the additional resource\ninformation placement started to return in provider summaries from\n``1.26`` is not used in nova or exposed to end users and from ``1.29``\nplacement started to return nested providers, but again, in rocky\nrelease, none of the projects including nova had started to model\nnested resource providers.\n\nBlueprint: use-nested-allocation-candidates\nChange-Id: I8036dc4351aa4b77f41b44622ae030cc8ae51d2f\n(cherry picked from commit dc8ff63fe4f9c4d916c1a4aaeed202e186f8090e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2a15e3476550b3a26c4f8e95e862688b8ed6c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a2a15e3476550b3a26c4f8e95e862688b8ed6c7a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
