)]}'
{"id":"openstack%2Fnova~665788","triplet_id":"openstack%2Fnova~stable%2Frocky~Ia597594e0469c0e83d7ad22b0678390aaebaffe7","project":"openstack/nova","branch":"stable/rocky","topic":"bug/1804062","hashtags":[],"change_id":"Ia597594e0469c0e83d7ad22b0678390aaebaffe7","subject":"Fix double word hacking test","status":"MERGED","created":"2019-06-17 23:18:27.000000000","updated":"2019-07-24 19:09:57.000000000","submitted":"2019-07-24 19:09:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665788-1563995397652-67437c28","meta_rev_id":"5d21a8fb06e2da1345f43062c80f1431f44c40bd","_number":665788,"virtual_id_number":665788,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-07-02 15:41:00.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2019-07-01 18:33:00.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-07-24 19:09:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-24 17:47:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2019-07-04 13:39:48.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-24 17:47:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-17 23:18:27.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-07-01 16:17:44.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-07-01 18:33:00.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-07-02 15:41:00.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 13:39:48.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2019-07-24 17:47:37.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2019-07-24 19:09:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f73f6f2985f3d7d3c471ac8965827fd633213295","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-17 23:18:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"800db05e1ebe2f237f0584aeb1dc3fb766d413f9","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-17 23:41:07.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/88/665788/1/check/nova-out-of-tree-pvm/24d33dc : FAILURE in 21m 15s\n- nova-in-tree-pvm http://184.172.12.213/88/665788/1/check/nova-in-tree-pvm/a30c313 : FAILURE in 22m 00s","accounts_in_message":[],"_revision_number":1},{"id":"984310837e8ab560ab777af575b1072f6eb2627d","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-06-18 00:00:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"65360d686dd2fc28b228791f78796d2cec2fa08a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-18 01:16:51.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/665788/1 : FAILURE in 1h 57m 32s","accounts_in_message":[],"_revision_number":1},{"id":"379e1e8bf3c3ed7e6e98988fd81f63c6a0db5877","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 01:44:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/665788/1/check/openstack-tox-pep8/ceb4a02/ : SUCCESS in 10m 55s\n- openstack-tox-py27 http://logs.openstack.org/88/665788/1/check/openstack-tox-py27/852bce4/ : SUCCESS in 12m 48s\n- openstack-tox-py35 http://logs.openstack.org/88/665788/1/check/openstack-tox-py35/8453e04/ : SUCCESS in 16m 09s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/665788/1/check/build-openstack-sphinx-docs/aac7f70/html/ : SUCCESS in 8m 23s\n- nova-tox-functional http://logs.openstack.org/88/665788/1/check/nova-tox-functional/dac0ce9/ : SUCCESS in 18m 06s\n- nova-tox-functional-py35 http://logs.openstack.org/88/665788/1/check/nova-tox-functional-py35/8968df8/ : SUCCESS in 17m 32s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/665788/1/check/openstack-tox-lower-constraints/8ce608d/ : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":1},{"id":"894c47337cd4fe5afabd2a8ba18287d1bcc66d0a","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2019-06-24 09:32:24.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"74ac9c0d04a46bb4641b91fcbd97a7586babb865","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-24 09:54:49.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/665788/2 : FAILURE in 21m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2cdba9745d004a6056add6c2b9052c5cdf74040d","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-24 09:55:01.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/88/665788/2/check/nova-out-of-tree-pvm/33b9478 : FAILURE in 20m 58s\n- nova-in-tree-pvm http://184.172.12.213/88/665788/2/check/nova-in-tree-pvm/4cb47fd : FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":2},{"id":"36351901808839dd5301f0ede7150520c52452e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-24 10:24:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/665788/2/check/openstack-tox-pep8/cf733cd/ : SUCCESS in 11m 16s\n- openstack-tox-py27 http://logs.openstack.org/88/665788/2/check/openstack-tox-py27/709ffa2/ : SUCCESS in 14m 23s\n- openstack-tox-py35 http://logs.openstack.org/88/665788/2/check/openstack-tox-py35/10f55b6/ : SUCCESS in 12m 44s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/665788/2/check/build-openstack-sphinx-docs/bc3baec/html/ : SUCCESS in 7m 29s\n- nova-tox-functional http://logs.openstack.org/88/665788/2/check/nova-tox-functional/97d4b3c/ : SUCCESS in 23m 46s\n- nova-tox-functional-py35 http://logs.openstack.org/88/665788/2/check/nova-tox-functional-py35/9dff341/ : SUCCESS in 21m 09s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/665788/2/check/openstack-tox-lower-constraints/3b83c96/ : SUCCESS in 12m 26s","accounts_in_message":[],"_revision_number":2},{"id":"72f1cf92ffdd69ee22b0047572c6e3392027f472","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-01 16:17:28.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c087971a7185bc6876d97c2765e1a44860177f1b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-07-01 16:17:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e4400bcfd3c4ab026e4e70a0a93a2562f6bcf9e7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-07-01 18:33:00.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/665788/3 : SUCCESS in 2h 02m 15s","accounts_in_message":[],"_revision_number":3},{"id":"fe2360d41e338dd1940d8cfdff6ef14087b40b3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 20:43:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/665788/3/check/openstack-tox-pep8/1f99022/ : SUCCESS in 17m 03s\n- openstack-tox-py27 http://logs.openstack.org/88/665788/3/check/openstack-tox-py27/1664e83/ : SUCCESS in 13m 02s\n- openstack-tox-py35 http://logs.openstack.org/88/665788/3/check/openstack-tox-py35/476aa43/ : SUCCESS in 14m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/665788/3/check/build-openstack-sphinx-docs/4aa10f4/html/ : SUCCESS in 13m 36s\n- nova-tox-functional http://logs.openstack.org/88/665788/3/check/nova-tox-functional/25d3a5c/ : SUCCESS in 17m 57s\n- nova-tox-functional-py35 http://logs.openstack.org/88/665788/3/check/nova-tox-functional-py35/57f0a37/ : SUCCESS in 20m 50s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/665788/3/check/openstack-tox-lower-constraints/e990780/ : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":3},{"id":"c6ddfd33ffb4008dfbd44c98e2cd58be70526747","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-07-02 15:41:00.000000000","message":"Patch Set 3:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/88/665788/3/check/nova-out-of-tree-pvm/b98a204 : FAILURE in 21m 54s\n- nova-in-tree-pvm http://184.172.12.213/88/665788/3/check/nova-in-tree-pvm/ab6f790 : FAILURE in 1m 18s","accounts_in_message":[],"_revision_number":3},{"id":"ebf972efc2768fc71241d0dba874f5404c0684b8","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-07-04 13:39:48.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"9bf3b29bb2caa9f24f385a0bb144c25fd32c3548","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2019-07-24 17:47:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"53a82cec77f504f3998bdf904b14b3f58cad88f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 17:49:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1feabfd55294af4dce0afb8da9ff1e9892f1ad0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:09:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/88/665788/3/gate/openstack-tox-pep8/716792a/ : SUCCESS in 17m 55s\n- openstack-tox-py27 http://logs.openstack.org/88/665788/3/gate/openstack-tox-py27/073ba8a/ : SUCCESS in 17m 06s\n- openstack-tox-py35 http://logs.openstack.org/88/665788/3/gate/openstack-tox-py35/cabaf8e/ : SUCCESS in 12m 40s\n- build-openstack-sphinx-docs http://logs.openstack.org/88/665788/3/gate/build-openstack-sphinx-docs/38a7b52/html/ : SUCCESS in 8m 38s\n- nova-tox-functional http://logs.openstack.org/88/665788/3/gate/nova-tox-functional/5423801/ : SUCCESS in 16m 51s\n- nova-tox-functional-py35 http://logs.openstack.org/88/665788/3/gate/nova-tox-functional-py35/25c6ffa/ : SUCCESS in 14m 54s\n- openstack-tox-lower-constraints http://logs.openstack.org/88/665788/3/gate/openstack-tox-lower-constraints/d3d65be/ : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":3},{"id":"26deeb0463a257f98cd101b80f801ab400163582","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 19:09:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4858074c89838eadeb9eaf9f39917e9fb90acd93","revisions":{"9955b08fe5f103f445ba814d3b012f66823dc095":{"kind":"REWORK","_number":1,"created":"2019-06-17 23:18:27.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/88/665788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/665788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/665788/1"}}},"commit":{"parents":[{"commit":"d8f1bef63ab4357af2105431d3ea367100518aa2","subject":"Merge \"[ironic] Don\u0027t remove instance info twice in destroy\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8f1bef63ab4357af2105431d3ea367100518aa2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-12 14:10:59.000000000","tz":60},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-17 23:18:15.000000000","tz":600},"subject":"Fix double word hacking test","message":"Fix double word hacking test\n\nAt present, \u0027pycodestyle\u0027 feeds the following string into the \u0027tokenizer\u0027\nlibrary:\n\n    [\"\u0027This is the the best comment\u0027\"]\n\n(note the added quotes because this isn\u0027t valid Python otherwise)\n\nOn previous versions of Python, this tokenizer would parse the string like so:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\")\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nHowever, with the fix [1] backported to recent versions of Python, this now\nresolves to:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\")\n  (4, \u0027\u0027, (1, 30), (1, 31), \u0027\u0027)\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 4 \u003d \u0027NEWLINE\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nTypically, \u0027pycodestyle\u0027 will run physical line checks on each line as it\nparses the token:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2036\n\nFor the former case above, the line doesn\u0027t include a newline which\nmeans we never parse a \u0027NEWLINE\u0027 token with a logical line (the fifth\nelement of the token tuple) corresponding to our full line. This means\nwe don\u0027t here but that wasn\u0027t an issue previously since there\u0027s a\nfallthrough case that handled tokens remaining at the end of the parse:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2114-L2116\n\nUnfortunately, because we now have an additional newline character to\nparse, one that\u0027s on a separate line to our test string no less, we run\nlogical checks on it:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2105-L2107\n\nThis is an issue since the logical check wipes stored tokens meaning\nwe\u0027ve nothing to check when we get to the fallthrough case:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2012\n\nThis fixes changes things so that a newline is included (and also adds\nquotes so it\u0027s valid Python, but that\u0027s mostly unrelated). This means we\nend up with the following instead:\n\n  [\"\u0027This is the the best comment\u0027\\n\"]\n\nOn both Python without the bugfix and with it, this parses as:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\\n\")\n  (4, \u0027\\n\u0027, (1, 30), (1, 31), \"\u0027This is the the best comment\u0027\\n\")\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 4 \u003d \u0027NEWLINE\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nWhich triggers things in \u0027pycodestyle\u0027 correctly.\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2044-L2046\n\nThis isn\u0027t _really_ a fix since there\u0027s clearly still a bug in either\n\u0027pycodestyle\u0027 or Python (I think the latter, since it\u0027s adding a newline\nto a file that explicitly doesn\u0027t have one), but the chances of us\nhitting this bug in practice are rather low - you\u0027d need to make a\nmistake on the very last line of a file without a newline at the end\nwhich is something Vim, for example, won\u0027t even let you do without\nsetting special flags - and therefore it can be reasonably ignored.\n\nConflicts:\n\tnova/tests/unit/test_hacking.py\n\nNOTE(stephenfin): Conflicts are because we don\u0027t have change\nI35c654bd39f343417e0a1124263ff31dcd0b05c9 (\"Bump to hacking 1.1.0\") or\nchange I8826c3fb89690805baae6b9b7b48985abb8d62d3 (\"Skip\ntest_check_doubled_words hacking check UT\") on this branch.\n\n[1] https://bugs.python.org/issue33899\n\nChange-Id: Ia597594e0469c0e83d7ad22b0678390aaebaffe7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1804062\n(cherry picked from f545a25cc443c41dcd9bdd028064c28b53f56037)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9955b08fe5f103f445ba814d3b012f66823dc095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9955b08fe5f103f445ba814d3b012f66823dc095"}]},"branch":"refs/heads/stable/rocky"},"825c64738bf0164b5206ef9a285fc6599f32da96":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-24 09:32:24.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/88/665788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/665788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/665788/2"}}},"commit":{"parents":[{"commit":"d8f1bef63ab4357af2105431d3ea367100518aa2","subject":"Merge \"[ironic] Don\u0027t remove instance info twice in destroy\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d8f1bef63ab4357af2105431d3ea367100518aa2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-12 14:10:59.000000000","tz":60},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2019-06-24 09:32:09.000000000","tz":600},"subject":"Fix double word hacking test","message":"Fix double word hacking test\n\nAt present, \u0027pycodestyle\u0027 feeds the following string into the \u0027tokenizer\u0027\nlibrary:\n\n    [\"\u0027This is the the best comment\u0027\"]\n\n(note the added quotes because this isn\u0027t valid Python otherwise)\n\nOn previous versions of Python, this tokenizer would parse the string like so:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\")\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nHowever, with the fix [1] backported to recent versions of Python, this now\nresolves to:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\")\n  (4, \u0027\u0027, (1, 30), (1, 31), \u0027\u0027)\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 4 \u003d \u0027NEWLINE\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nTypically, \u0027pycodestyle\u0027 will run physical line checks on each line as it\nparses the token:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2036\n\nFor the former case above, the line doesn\u0027t include a newline which\nmeans we never parse a \u0027NEWLINE\u0027 token with a logical line (the fifth\nelement of the token tuple) corresponding to our full line. This means\nwe don\u0027t here but that wasn\u0027t an issue previously since there\u0027s a\nfallthrough case that handled tokens remaining at the end of the parse:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2114-L2116\n\nUnfortunately, because we now have an additional newline character to\nparse, one that\u0027s on a separate line to our test string no less, we run\nlogical checks on it:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2105-L2107\n\nThis is an issue since the logical check wipes stored tokens meaning\nwe\u0027ve nothing to check when we get to the fallthrough case:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2012\n\nThis fixes changes things so that a newline is included (and also adds\nquotes so it\u0027s valid Python, but that\u0027s mostly unrelated). This means we\nend up with the following instead:\n\n  [\"\u0027This is the the best comment\u0027\\n\"]\n\nOn both Python without the bugfix and with it, this parses as:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\\n\")\n  (4, \u0027\\n\u0027, (1, 30), (1, 31), \"\u0027This is the the best comment\u0027\\n\")\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 4 \u003d \u0027NEWLINE\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nWhich triggers things in \u0027pycodestyle\u0027 correctly.\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2044-L2046\n\nThis isn\u0027t _really_ a fix since there\u0027s clearly still a bug in either\n\u0027pycodestyle\u0027 or Python (I think the latter, since it\u0027s adding a newline\nto a file that explicitly doesn\u0027t have one), but the chances of us\nhitting this bug in practice are rather low - you\u0027d need to make a\nmistake on the very last line of a file without a newline at the end\nwhich is something Vim, for example, won\u0027t even let you do without\nsetting special flags - and therefore it can be reasonably ignored.\n\nConflicts:\n\tnova/tests/unit/test_hacking.py\n\nNOTE(stephenfin): Conflicts are because we don\u0027t have change\nI35c654bd39f343417e0a1124263ff31dcd0b05c9 (\"Bump to hacking 1.1.0\") or\nchange I8826c3fb89690805baae6b9b7b48985abb8d62d3 (\"Skip\ntest_check_doubled_words hacking check UT\") on this branch.\n\n[1] https://bugs.python.org/issue33899\n\nChange-Id: Ia597594e0469c0e83d7ad22b0678390aaebaffe7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1804062\n(cherry picked from f545a25cc443c41dcd9bdd028064c28b53f56037)\n(cherry picked from commit 0cb6106b83c33bded9e6cdec7737964c36be8de5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/825c64738bf0164b5206ef9a285fc6599f32da96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/825c64738bf0164b5206ef9a285fc6599f32da96"}]},"branch":"refs/heads/stable/rocky"},"4858074c89838eadeb9eaf9f39917e9fb90acd93":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2019-07-01 16:17:28.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/88/665788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/88/665788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/88/665788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/88/665788/3"}}},"commit":{"parents":[{"commit":"7f483517df997e5ba22b4aedaecc01804943d952","subject":"Merge \"libvirt: flatten rbd images when unshelving an instance\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f483517df997e5ba22b4aedaecc01804943d952"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-12 14:10:59.000000000","tz":60},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-07-01 16:17:28.000000000","tz":0},"subject":"Fix double word hacking test","message":"Fix double word hacking test\n\nAt present, \u0027pycodestyle\u0027 feeds the following string into the \u0027tokenizer\u0027\nlibrary:\n\n    [\"\u0027This is the the best comment\u0027\"]\n\n(note the added quotes because this isn\u0027t valid Python otherwise)\n\nOn previous versions of Python, this tokenizer would parse the string like so:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\")\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nHowever, with the fix [1] backported to recent versions of Python, this now\nresolves to:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\")\n  (4, \u0027\u0027, (1, 30), (1, 31), \u0027\u0027)\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 4 \u003d \u0027NEWLINE\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nTypically, \u0027pycodestyle\u0027 will run physical line checks on each line as it\nparses the token:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2036\n\nFor the former case above, the line doesn\u0027t include a newline which\nmeans we never parse a \u0027NEWLINE\u0027 token with a logical line (the fifth\nelement of the token tuple) corresponding to our full line. This means\nwe don\u0027t here but that wasn\u0027t an issue previously since there\u0027s a\nfallthrough case that handled tokens remaining at the end of the parse:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2114-L2116\n\nUnfortunately, because we now have an additional newline character to\nparse, one that\u0027s on a separate line to our test string no less, we run\nlogical checks on it:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2105-L2107\n\nThis is an issue since the logical check wipes stored tokens meaning\nwe\u0027ve nothing to check when we get to the fallthrough case:\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2012\n\nThis fixes changes things so that a newline is included (and also adds\nquotes so it\u0027s valid Python, but that\u0027s mostly unrelated). This means we\nend up with the following instead:\n\n  [\"\u0027This is the the best comment\u0027\\n\"]\n\nOn both Python without the bugfix and with it, this parses as:\n\n  (3, \"\u0027This is the the best comment\u0027\", (1, 0), (1, 30), \"\u0027This is the the best comment\u0027\\n\")\n  (4, \u0027\\n\u0027, (1, 30), (1, 31), \"\u0027This is the the best comment\u0027\\n\")\n  (0, \u0027\u0027, (2, 0), (2, 0), \u0027\u0027)\n\nwhere (3 \u003d \u0027STRING\u0027, 4 \u003d \u0027NEWLINE\u0027, 0 \u003d \u0027ENDMARKER\u0027)\n\nWhich triggers things in \u0027pycodestyle\u0027 correctly.\n\n  https://github.com/PyCQA/pycodestyle/blob/2.5.0/pycodestyle.py#L2044-L2046\n\nThis isn\u0027t _really_ a fix since there\u0027s clearly still a bug in either\n\u0027pycodestyle\u0027 or Python (I think the latter, since it\u0027s adding a newline\nto a file that explicitly doesn\u0027t have one), but the chances of us\nhitting this bug in practice are rather low - you\u0027d need to make a\nmistake on the very last line of a file without a newline at the end\nwhich is something Vim, for example, won\u0027t even let you do without\nsetting special flags - and therefore it can be reasonably ignored.\n\n[1] https://bugs.python.org/issue33899\n\nChange-Id: Ia597594e0469c0e83d7ad22b0678390aaebaffe7\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\nCloses-Bug: #1804062\n(cherry picked from f545a25cc443c41dcd9bdd028064c28b53f56037)\n(cherry picked from commit 0cb6106b83c33bded9e6cdec7737964c36be8de5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4858074c89838eadeb9eaf9f39917e9fb90acd93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4858074c89838eadeb9eaf9f39917e9fb90acd93"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
