)]}'
{"id":"openstack%2Fnova~666631","triplet_id":"openstack%2Fnova~master~I5235751b1dbc77ecc919eec7f3e022cd70085051","project":"openstack/nova","branch":"master","topic":"bug/1789074","hashtags":[],"change_id":"I5235751b1dbc77ecc919eec7f3e022cd70085051","subject":"vif: Resolve a TODO and update another","status":"MERGED","created":"2019-06-20 16:18:41.000000000","updated":"2019-08-11 12:13:05.000000000","submitted":"2019-07-17 04:15:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666631-1563336953856-14776782","meta_rev_id":"97c3918141fde76598068eab7a08b36d7fdd9f41","_number":666631,"virtual_id_number":666631,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-16 22:19:58.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-12 21:27:39.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-17 01:26:57.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-16 23:23:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-07-16 22:58:24.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2019-07-17 04:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2019-07-12 16:43:27.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-17 03:13:46.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"date":"2019-07-12 18:17:02.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-15 16:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-07-12 15:40:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2019-07-16 18:32:52.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2019-07-16 18:32:52.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-21 10:47:42.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2019-06-21 11:06:59.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-21 21:50:28.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-06-25 21:59:07.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-04 15:12:57.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2019-07-12 16:43:27.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-12 18:17:02.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-07-12 21:27:39.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-15 16:13:54.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2019-07-16 18:32:52.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-07-16 22:19:58.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-16 22:58:24.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-07-16 23:23:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 01:26:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 03:13:46.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-17 04:15:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9652981874d4905569b9ab982dd42c098f336bd1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-20 16:18:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"519ee38e7c64637a7b3fa28dd44f1a8e326bc7f9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-20 16:21:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ce6deef9103351e27e10963933d1ef21e71067a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-20 16:48:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20fcde5b0b075e31290e76afa76d264ee76d34b1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-20 16:49:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a39e6f84ee077d9b99122e9063c67e7834fe80fe","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-20 18:49:43.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/2/check/tempest-dsvm-full-xenial/2ee23cd/ : SUCCESS in 1h 23m 08s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/2/check/tempest-dsvm-full-xenial-py3/d5b650b/ : SUCCESS in 1h 40m 16s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/2/check/grenade-dsvm-xenial/090fc66/ : SUCCESS in 1h 12m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"414cfaa94eabaf1940f29febe0d9406f2af00975","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-20 21:14:50.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/666631/2 : SUCCESS in 2h 20m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2bb531456412c631d7a553571b42a01aa1d05a55","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-20 21:19:54.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/31/666631/2/check-vote/ext-nova-zuul/7790999 : FAILURE in 1h 01m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e74bdcd1441c4cca3e6f84fb57494e209435a03c","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-20 23:10:20.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/2/check/check-dsvm-tempest-vz7-exe-minimal/ac5eb5e : FAILURE in 1m 33s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":2},{"id":"99683b839dee3810b45d364b9268107b54ea010c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-21 01:36:03.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-32973 : FAILURE in 3h 45m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a7bb7906a21c57b3a456f574bb8238a4bf5557f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 01:40:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/31/666631/2/check/grenade-py3/e42d6ec/ : SUCCESS in 58m 16s\n- tempest-full-py3 http://logs.openstack.org/31/666631/2/check/tempest-full-py3/0e8720c/ : SUCCESS in 1h 30m 04s\n- openstack-tox-cover http://logs.openstack.org/31/666631/2/check/openstack-tox-cover/825702a/ : FAILURE in 28m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/2/check/openstack-tox-lower-constraints/b80007c/ : TIMED_OUT in 40m 56s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/2/check/openstack-tox-pep8/b17e8ce/ : FAILURE in 8m 56s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/2/check/openstack-tox-py27/d7f198e/ : SUCCESS in 16m 12s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/2/check/openstack-tox-py36/18f9fe3/ : SUCCESS in 12m 10s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/2/check/openstack-tox-py37/287a295/ : SUCCESS in 12m 11s\n- openstack-tox-docs http://logs.openstack.org/31/666631/2/check/openstack-tox-docs/58fb40a/html/ : SUCCESS in 8m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/2/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/68fc0fd/ : SUCCESS in 1h 00m 16s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/2/check/devstack-plugin-ceph-tempest/f9f9626/ : SUCCESS in 1h 28m 28s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/2/check/neutron-grenade-multinode/642cc58/ : SUCCESS in 1h 11m 20s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/2/check/neutron-tempest-linuxbridge/34d5beb/ : SUCCESS in 2h 27m 17s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/2/check/nova-grenade-live-migration/c21a38c/ : SUCCESS in 1h 13m 12s\n- nova-live-migration http://logs.openstack.org/31/666631/2/check/nova-live-migration/1b619ec/ : SUCCESS in 49m 22s\n- nova-lvm http://logs.openstack.org/31/666631/2/check/nova-lvm/300e58e/ : SUCCESS in 1h 14m 13s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/2/check/nova-multi-cell/3b96d3d/ : SUCCESS in 1h 40m 18s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/2/check/nova-next/0d125b1/ : SUCCESS in 1h 24m 54s\n- nova-tox-functional http://logs.openstack.org/31/666631/2/check/nova-tox-functional/c06e5d5/ : SUCCESS in 21m 03s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/2/check/nova-tox-functional-py36/a76e367/ : SUCCESS in 17m 36s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/2/check/tempest-slow-py3/b346bb6/ : SUCCESS in 2h 21m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c6108994a2b10dff333eb3157887c311950cb2ed","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 07:39:35.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/666631/2/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/85c7468 : SUCCESS in 2h 16m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4ba53fe3242ce75afa4611fe91a2e0a12b3a3c7c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-21 10:05:58.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"19d9f836db736d83d172d8f688111d1c5996de1a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 10:07:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d4bc40079cda48c56284b41e7238a728d1c009bf","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-06-21 10:47:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nya this makes sense we are just removing an old todo and updating an existing one.","accounts_in_message":[],"_revision_number":3},{"id":"f63e5a9f832edb7a312f26d1378ca23628db9eb4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-21 10:48:17.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"b312e9d139e70ca7fd9cc9814a7d271357f2c408","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 10:50:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"af6588c31f52a9fb406783c6bcda4a8ad8e064f0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-21 11:05:14.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"cdcb1f1c4151525c54621b7e1978fc2e00ab011f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 11:06:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ef1837271bebf643b9e15090f24bddc9efe70bf5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-21 12:36:42.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/5/check/tempest-dsvm-full-xenial/ccc6c0a/ : SUCCESS in 1h 23m 29s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/5/check/tempest-dsvm-full-xenial-py3/9f0d13d/ : SUCCESS in 1h 28m 54s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/5/check/grenade-dsvm-xenial/2c87ddf/ : SUCCESS in 1h 14m 23s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2930ec7a92d31d3060ad3d1df87acf597e1c0f43","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-21 14:19:31.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/666631/5 : FAILURE in 2h 46m 10s","accounts_in_message":[],"_revision_number":5},{"id":"e121a9b9310b396cc594b4ec4b9441e26c7375ef","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 14:25:25.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/5/check/check-dsvm-tempest-vz7-exe-minimal/7ce0588 : FAILURE in 12m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"c5297360c63f5b679cb723db57ab63dba1db4f0a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-21 14:34:13.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/31/666631/5/check-vote/ext-nova-zuul/5f1a43a : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":5},{"id":"551a91e64a56c020bddde8ee41ef8245a2d17cce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-21 14:52:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/31/666631/5/check/grenade-py3/7fd360f/ : SUCCESS in 56m 26s\n- tempest-full-py3 http://logs.openstack.org/31/666631/5/check/tempest-full-py3/aad5fa3/ : SUCCESS in 1h 31m 34s\n- openstack-tox-cover http://logs.openstack.org/31/666631/5/check/openstack-tox-cover/6326b84/cover/ : SUCCESS in 26m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/5/check/openstack-tox-lower-constraints/5c05414/ : SUCCESS in 13m 38s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/5/check/openstack-tox-pep8/78b7575/ : SUCCESS in 10m 31s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/5/check/openstack-tox-py27/eac85d3/ : SUCCESS in 22m 58s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/5/check/openstack-tox-py36/6bd485c/ : SUCCESS in 21m 36s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/5/check/openstack-tox-py37/5d15c69/ : SUCCESS in 11m 10s\n- openstack-tox-docs http://logs.openstack.org/31/666631/5/check/openstack-tox-docs/ca91900/html/ : SUCCESS in 5m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/39363bd/ : SUCCESS in 48m 54s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/5/check/devstack-plugin-ceph-tempest/933cc62/ : SUCCESS in 1h 26m 37s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/5/check/neutron-grenade-multinode/ac3c2c7/ : SUCCESS in 1h 20m 51s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/5/check/neutron-tempest-linuxbridge/90751dc/ : SUCCESS in 1h 19m 33s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/5/check/nova-grenade-live-migration/7f2527f/ : SUCCESS in 1h 07m 08s\n- nova-live-migration http://logs.openstack.org/31/666631/5/check/nova-live-migration/f79d289/ : FAILURE in 19m 35s\n- nova-lvm http://logs.openstack.org/31/666631/5/check/nova-lvm/fe5db13/ : SUCCESS in 1h 11m 57s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/5/check/nova-multi-cell/2e5f20a/ : SUCCESS in 1h 54m 52s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/5/check/nova-next/484b54e/ : SUCCESS in 1h 46m 02s\n- nova-tox-functional http://logs.openstack.org/31/666631/5/check/nova-tox-functional/53a5081/ : SUCCESS in 20m 51s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/5/check/nova-tox-functional-py36/1aa80be/ : SUCCESS in 17m 17s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/5/check/tempest-slow-py3/6674411/ : SUCCESS in 2h 32m 37s","accounts_in_message":[],"_revision_number":5},{"id":"f2f693237a6cb17ef6bfeea218401541a7c6c064","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-21 16:07:22.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/31/666631/5/check/nova-out-of-tree-pvm/ba80b41 : FAILURE in 3h 24m 12s\n- nova-in-tree-pvm http://184.172.12.213/31/666631/5/check/nova-in-tree-pvm/78ea9b7 : FAILURE in 5h 00m 55s","accounts_in_message":[],"_revision_number":5},{"id":"cc00fdc2f4375231a1e04428c103bc37604f1291","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-21 16:24:36.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33007 : FAILURE in 3h 29m 19s","accounts_in_message":[],"_revision_number":5},{"id":"c7df576783700aaebf9de5bdf3c8b06ddc4b692d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-06-21 21:50:28.000000000","message":"Patch Set 5:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/666631/5/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/448d367 : SUCCESS in 2h 03m 50s","accounts_in_message":[],"_revision_number":5},{"id":"5ad069c8d51f7bef341041cba4c7fd7abd63c4a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-06-25 16:16:03.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"ee4b721b64451cc12367e430819c6beb465ddaec","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-06-25 17:59:51.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/5/check/tempest-dsvm-full-xenial/7f6fbce/ : SUCCESS in 1h 25m 56s\n- tempest-dsvm-full-xenial-py3 https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/5/check/tempest-dsvm-full-xenial-py3/24f4fbf/ : SUCCESS in 1h 35m 39s (non-voting)\n- grenade-dsvm-xenial https://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/nova/31/666631/5/check/grenade-dsvm-xenial/2e97300/ : SUCCESS in 1h 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f226a654a239ba3a2c393a132991553ba8050dd7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-06-25 19:07:55.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/31/666631/5/check-vote/ext-nova-zuul/136ff52 : SUCCESS in 56m 41s","accounts_in_message":[],"_revision_number":5},{"id":"7486302262bc37617fc4625c490ab0821f49fdd1","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-06-25 19:57:49.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/5/check/check-dsvm-tempest-vz7-exe-minimal/a7b4431 : FAILURE in 17m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":5},{"id":"dc03f81bd0cf55623ecaeea2d4d41bc7eec62096","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-06-25 20:37:36.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33146 : FAILURE in 2h 41m 34s","accounts_in_message":[],"_revision_number":5},{"id":"e487e298112127d4a48af9c303815fcb2a5e9376","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-06-25 21:59:07.000000000","message":"Patch Set 5:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/31/666631/5/check/nova-out-of-tree-pvm/c299a59 : FAILURE in 5h 05m 52s\n- nova-in-tree-pvm http://184.172.12.213/31/666631/5/check/nova-in-tree-pvm/6241309 : FAILURE in 3h 01m 59s","accounts_in_message":[],"_revision_number":5},{"id":"b29a6656c9ad53bcb8007ff2865372d3f665a054","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-06-25 22:28:49.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/666631/5 : SUCCESS in 2h 26m 17s","accounts_in_message":[],"_revision_number":5},{"id":"54401bb22f12faa7a47f4df94a69cfa385a65082","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-26 05:42:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/31/666631/5/check/grenade-py3/e3a6211/ : SUCCESS in 55m 36s\n- tempest-full-py3 http://logs.openstack.org/31/666631/5/check/tempest-full-py3/a114afa/ : SUCCESS in 1h 42m 34s\n- openstack-tox-cover http://logs.openstack.org/31/666631/5/check/openstack-tox-cover/cd065bc/cover/ : SUCCESS in 15m 04s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/5/check/openstack-tox-lower-constraints/92f0b3a/ : SUCCESS in 15m 25s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/5/check/openstack-tox-pep8/08afbf5/ : SUCCESS in 10m 08s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/5/check/openstack-tox-py27/8ebad1e/ : SUCCESS in 12m 06s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/5/check/openstack-tox-py36/645732a/ : SUCCESS in 12m 56s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/5/check/openstack-tox-py37/61ceba8/ : SUCCESS in 11m 48s\n- openstack-tox-docs http://logs.openstack.org/31/666631/5/check/openstack-tox-docs/8c148f5/html/ : SUCCESS in 7m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/5/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/af586f1/ : FAILURE in 37m 55s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/5/check/devstack-plugin-ceph-tempest/156bfe9/ : SUCCESS in 1h 19m 25s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/5/check/neutron-grenade-multinode/0f43843/ : SUCCESS in 1h 04m 11s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/5/check/neutron-tempest-linuxbridge/e431216/ : SUCCESS in 1h 43m 32s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/5/check/nova-grenade-live-migration/c632825/ : SUCCESS in 1h 13m 52s\n- nova-live-migration http://logs.openstack.org/31/666631/5/check/nova-live-migration/1fda0e8/ : SUCCESS in 53m 36s\n- nova-lvm http://logs.openstack.org/31/666631/5/check/nova-lvm/c283619/ : SUCCESS in 1h 10m 22s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/5/check/nova-multi-cell/00f47f2/ : SUCCESS in 1h 23m 21s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/5/check/nova-next/820222a/ : SUCCESS in 1h 18m 56s\n- nova-tox-functional http://logs.openstack.org/31/666631/5/check/nova-tox-functional/ea163b7/ : SUCCESS in 18m 39s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/5/check/nova-tox-functional-py36/7cd7856/ : SUCCESS in 17m 08s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/5/check/tempest-slow-py3/b2543a7/ : SUCCESS in 2h 55m 22s","accounts_in_message":[],"_revision_number":5},{"id":"10dcccdceae206937eb2c75381931328d590c11d","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-06-26 05:43:32.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/31/666631/5/check/EMC_VxFlexOS_NOVA/ba96b00/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"ee3e37fdca31b4c9d5f9ab252132b1fe66494ca1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-06-26 08:41:59.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-31-666631-5 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":5},{"id":"a282a5397bf20a430643048b072f8e7a4d23d91c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-10 15:19:06.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"2ec968376554af26ebf2ee7d675af11c55ad6306","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-07-10 17:53:02.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33727 : SUCCESS in 1h 59m 39s","accounts_in_message":[],"_revision_number":6},{"id":"fabf9ace45e80558cefed3705529e0933fd5ef6a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-07-10 18:26:13.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/666631/6 : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":6},{"id":"995cc82cb160884721effde4eb7c671c1b3ec97e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:27:58.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/31/666631/6/check/grenade-py3/9dfb2c6/ : SUCCESS in 1h 04m 52s\n- tempest-full-py3 http://logs.openstack.org/31/666631/6/check/tempest-full-py3/7c5c339/ : SUCCESS in 1h 39m 25s\n- openstack-tox-cover http://logs.openstack.org/31/666631/6/check/openstack-tox-cover/d20eaae/cover/ : SUCCESS in 21m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/6/check/openstack-tox-lower-constraints/d05ffbb/ : SUCCESS in 18m 29s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/6/check/openstack-tox-pep8/1292a0f/ : SUCCESS in 10m 44s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/6/check/openstack-tox-py27/552b5d5/ : SUCCESS in 16m 38s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/6/check/openstack-tox-py36/6f376c0/ : SUCCESS in 15m 37s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/6/check/openstack-tox-py37/ffcc4df/ : SUCCESS in 20m 52s\n- openstack-tox-docs http://logs.openstack.org/31/666631/6/check/openstack-tox-docs/76618a8/html/ : SUCCESS in 6m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/441703c/ : SUCCESS in 50m 52s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/6/check/devstack-plugin-ceph-tempest/e52502d/ : SUCCESS in 1h 42m 31s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/6/check/neutron-grenade-multinode/41cfe1a/ : SUCCESS in 1h 17m 34s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/6/check/neutron-tempest-linuxbridge/4b33d61/ : SUCCESS in 1h 28m 47s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/6/check/nova-grenade-live-migration/7bdedd1/ : SUCCESS in 1h 06m 49s\n- nova-live-migration http://logs.openstack.org/31/666631/6/check/nova-live-migration/193c591/ : SUCCESS in 1h 02m 15s\n- nova-lvm http://logs.openstack.org/31/666631/6/check/nova-lvm/9d546d6/ : SUCCESS in 1h 07m 14s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/6/check/nova-multi-cell/9ec5788/ : SUCCESS in 1h 46m 56s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/6/check/nova-next/c419938/ : FAILURE in 1h 32m 18s\n- nova-tox-functional http://logs.openstack.org/31/666631/6/check/nova-tox-functional/708ffc5/ : SUCCESS in 23m 15s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/6/check/nova-tox-functional-py36/aadbefb/ : SUCCESS in 21m 13s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/6/check/tempest-slow-py3/5ba6f45/ : SUCCESS in 2h 18m 07s","accounts_in_message":[],"_revision_number":6},{"id":"bdbc365bb6d402bb857d9548e1af9664a3cd0447","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 21:39:04.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/6/check/check-dsvm-tempest-vz7-exe-minimal/6888709 : FAILURE in 19m 02s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"36c32f18dade438828e7a6cbd527eb674bd27cc4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-10 22:04:32.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/6/check/tempest-dsvm-full-xenial/c498d58/ : SUCCESS in 1h 24m 26s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/6/check/tempest-dsvm-full-xenial-py3/4e33cfa/ : SUCCESS in 1h 28m 13s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/6/check/grenade-dsvm-xenial/2d60105/ : SUCCESS in 1h 13m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e7d3841bf113814e67abd52b3826bc1b173e7c3c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-11 02:14:48.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/31/666631/6/check-vote/ext-nova-zuul/b8aa79e : SUCCESS in 1h 00m 50s","accounts_in_message":[],"_revision_number":6},{"id":"a9fb63c7064c70af4937c3d3cb08b09cdbe134a3","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2019-07-12 12:03:39.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"52d23f6a9c569c980f34b43f8abf835593579804","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-07-12 12:15:44.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/6/check/check-dsvm-tempest-vz7-exe-minimal/7e9d29c : FAILURE in 10m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":6},{"id":"1aca033e3f4e2d74c701190e2782765488914c5d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-12 13:02:54.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/31/666631/6/check-vote/ext-nova-zuul/702d764 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":6},{"id":"0089e5c2797767a6f58c887e5ab0d969a4b297b8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-07-12 13:08:19.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33829 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":6},{"id":"3917190fde64a33cb8c1c731877eb49764cbe111","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-12 13:46:15.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/6/check/tempest-dsvm-full-xenial/35d9390/ : SUCCESS in 1h 41m 37s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/6/check/tempest-dsvm-full-xenial-py3/368fdd2/ : FAILURE in 59m 37s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/6/check/grenade-dsvm-xenial/995320f/ : SUCCESS in 1h 35m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f422c4842727d3905d4949a341a56da781feb95c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-07-12 14:29:21.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/666631/6 : FAILURE in 2h 24m 54s","accounts_in_message":[],"_revision_number":6},{"id":"f7a6e3288fa47da9e90ba783b64908b79c571a8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 14:33:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/31/666631/6/check/grenade-py3/7dc2ddc/ : SUCCESS in 59m 18s\n- tempest-full-py3 http://logs.openstack.org/31/666631/6/check/tempest-full-py3/d105779/ : SUCCESS in 1h 34m 26s\n- openstack-tox-cover http://logs.openstack.org/31/666631/6/check/openstack-tox-cover/6bd71f4/cover/ : SUCCESS in 22m 01s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/6/check/openstack-tox-lower-constraints/215b35e/ : SUCCESS in 23m 48s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/6/check/openstack-tox-pep8/e41bbe4/ : SUCCESS in 10m 33s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/6/check/openstack-tox-py27/a34bd4a/ : SUCCESS in 17m 05s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/6/check/openstack-tox-py36/2d79a0d/ : SUCCESS in 17m 40s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/6/check/openstack-tox-py37/6c05ddc/ : SUCCESS in 18m 58s\n- openstack-tox-docs http://logs.openstack.org/31/666631/6/check/openstack-tox-docs/33b5f29/html/ : SUCCESS in 9m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/6/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/48e7ef6/ : SUCCESS in 55m 57s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/6/check/devstack-plugin-ceph-tempest/de3d042/ : SUCCESS in 1h 22m 56s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/6/check/neutron-grenade-multinode/d914c05/ : SUCCESS in 1h 15m 15s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/6/check/neutron-tempest-linuxbridge/ea16b90/ : SUCCESS in 1h 40m 01s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/6/check/nova-grenade-live-migration/c92d1da/ : SUCCESS in 1h 12m 49s\n- nova-live-migration http://logs.openstack.org/31/666631/6/check/nova-live-migration/3d67cdb/ : SUCCESS in 48m 43s\n- nova-lvm http://logs.openstack.org/31/666631/6/check/nova-lvm/43feca6/ : SUCCESS in 1h 12m 27s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/6/check/nova-multi-cell/c8a924e/ : SUCCESS in 1h 43m 00s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/6/check/nova-next/168ba17/ : SUCCESS in 1h 24m 24s\n- nova-tox-functional http://logs.openstack.org/31/666631/6/check/nova-tox-functional/7166d5e/ : SUCCESS in 24m 57s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/6/check/nova-tox-functional-py36/ac6d145/ : SUCCESS in 23m 06s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/6/check/tempest-slow-py3/2c0b229/ : SUCCESS in 2h 14m 48s","accounts_in_message":[],"_revision_number":6},{"id":"f5c45b2a14493af150ace0f30d57370b05d8549b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-12 14:33:39.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/31/666631/6/check/EMC_VxFlexOS_NOVA/825d6d5/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"0c954f60e8f4a5327e17ba7a504ba6174aa315d0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-07-12 15:40:17.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"77a9f943623ee4b8d3755f19ae5dc1cdeed7e321","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-07-12 16:13:13.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/7/check/check-dsvm-tempest-vz7-exe-minimal/d946ea6 : FAILURE in 17m 22s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"39126379a36dcf6032890995761ddbbd92a5d7cc","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-12 16:32:54.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-31-666631-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"c89f5328d1f8853ce62de85ebdf7d53db8a47610","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-07-12 16:43:27.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/31/666631/7/check-vote/ext-nova-zuul/386a2b0 : SUCCESS in 1h 01m 18s","accounts_in_message":[],"_revision_number":7},{"id":"d92a61e26b0da1dfeef2b8a1f0440245103c0d88","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-12 17:09:53.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/7/check/tempest-dsvm-full-xenial/082ad1e/ : SUCCESS in 1h 26m 28s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/7/check/tempest-dsvm-full-xenial-py3/62f9f21/ : SUCCESS in 1h 19m 47s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/7/check/grenade-dsvm-xenial/d1cfdb5/ : FAILURE in 1h 07m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d88880f65476e26cee5de4b0e699b7209c197773","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-07-12 17:36:25.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33834 : FAILURE in 1h 54m 13s","accounts_in_message":[],"_revision_number":7},{"id":"ea8ac19bb1fb9f52d160658a80aa622ab333ac55","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-07-12 18:17:02.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/666631/7 : SUCCESS in 2h 34m 39s","accounts_in_message":[],"_revision_number":7},{"id":"9ad569a66dc0a56e87c3c758619a4de00c8ecb8c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 18:39:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/31/666631/7/check/grenade-py3/8f5fdb3/ : SUCCESS in 1h 07m 08s\n- tempest-full-py3 http://logs.openstack.org/31/666631/7/check/tempest-full-py3/957d985/ : SUCCESS in 1h 45m 15s\n- openstack-tox-cover http://logs.openstack.org/31/666631/7/check/openstack-tox-cover/9b749c8/cover/ : SUCCESS in 20m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/7/check/openstack-tox-lower-constraints/1342d72/ : SUCCESS in 21m 50s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/7/check/openstack-tox-pep8/a36bc4e/ : SUCCESS in 10m 38s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/7/check/openstack-tox-py27/9d73e92/ : SUCCESS in 22m 44s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/7/check/openstack-tox-py36/edeb2f4/ : SUCCESS in 18m 00s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/7/check/openstack-tox-py37/006a65e/ : SUCCESS in 15m 35s\n- openstack-tox-docs http://logs.openstack.org/31/666631/7/check/openstack-tox-docs/57a5d61/html/ : SUCCESS in 6m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/64c98be/ : SUCCESS in 51m 36s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/7/check/devstack-plugin-ceph-tempest/185bde6/ : SUCCESS in 1h 24m 20s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/7/check/neutron-grenade-multinode/444810e/ : SUCCESS in 1h 16m 56s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/7/check/neutron-tempest-linuxbridge/1d11ee3/ : SUCCESS in 1h 35m 10s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/7/check/nova-grenade-live-migration/e681239/ : SUCCESS in 1h 06m 55s\n- nova-live-migration http://logs.openstack.org/31/666631/7/check/nova-live-migration/3b99a20/ : SUCCESS in 53m 24s\n- nova-lvm http://logs.openstack.org/31/666631/7/check/nova-lvm/95beb7f/ : SUCCESS in 57m 01s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/7/check/nova-multi-cell/08349f9/ : SUCCESS in 1h 46m 37s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/7/check/nova-next/ca09a02/ : SUCCESS in 1h 35m 40s\n- nova-tox-functional http://logs.openstack.org/31/666631/7/check/nova-tox-functional/b00af4f/ : SUCCESS in 51m 01s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/7/check/nova-tox-functional-py36/1651113/ : SUCCESS in 21m 23s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/7/check/tempest-slow-py3/0be0f32/ : SUCCESS in 2h 33m 38s","accounts_in_message":[],"_revision_number":7},{"id":"3caf7aa478d0669ed0c805154c1469a0274f9e85","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-12 18:39:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/31/666631/7/check/EMC_VxFlexOS_NOVA/04a319b/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"c3f0cffe2039e0963730b718cc9a6ff78c9efb7a","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-07-12 21:27:39.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/31/666631/7/check-nova/Nova-ML2-Sriov/6d62157 : SUCCESS in 1h 20m 31s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/31/666631/7/check-nova/Nova-MACVTAP-ML2-Sriov/3f75b58 : FAILURE in 1h 58m 41s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":7},{"id":"ad340b23605dfc94946d75f8ecf370bb2d0c9511","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-13 02:37:56.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-31-666631-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"d33a1f07aed0bf86b6f1d6b4de2a9d0cc6c0b723","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2019-07-15 16:13:54.000000000","message":"Patch Set 7: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":7},{"id":"70a6cd27aca4b6c4ba363f712b500957508fc2db","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-16 18:32:52.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"60471fd6b6272dfece11bf9ffa94ad5c50e87afb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 18:34:50.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"40b7e9c5b2fb964fbd43ffb2be9538db7815f1c6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 21:43:37.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 http://logs.openstack.org/31/666631/7/gate/grenade-py3/e117562/ : SUCCESS in 1h 09m 23s\n- tempest-full-py3 http://logs.openstack.org/31/666631/7/gate/tempest-full-py3/f7528c2/ : SUCCESS in 1h 36m 44s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/7/gate/openstack-tox-lower-constraints/adf00d9/ : SUCCESS in 23m 33s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/7/gate/openstack-tox-pep8/8ff8854/ : SUCCESS in 8m 35s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/7/gate/openstack-tox-py27/acfb5bb/ : SUCCESS in 16m 12s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/7/gate/openstack-tox-py36/b7e8648/ : SUCCESS in 18m 18s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/7/gate/openstack-tox-py37/5b488f4/ : SUCCESS in 16m 09s\n- openstack-tox-docs http://logs.openstack.org/31/666631/7/gate/openstack-tox-docs/b0785b2/html/ : SUCCESS in 7m 31s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/7/gate/nova-grenade-live-migration/f052473/ : SUCCESS in 1h 12m 16s\n- nova-live-migration http://logs.openstack.org/31/666631/7/gate/nova-live-migration/e3ed2fa/ : SUCCESS in 47m 37s\n- nova-tox-functional http://logs.openstack.org/31/666631/7/gate/nova-tox-functional/34ead7b/ : SUCCESS in 24m 14s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/7/gate/nova-tox-functional-py36/560e93d/ : SUCCESS in 27m 40s\n- nova-next http://logs.openstack.org/31/666631/7/gate/nova-next/121a2b6/ : SUCCESS in 2h 15m 36s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/7/gate/tempest-slow-py3/5483c11/ : TIMED_OUT in 3h 05m 33s","accounts_in_message":[],"_revision_number":7},{"id":"bac6d4297c131f48e735ceb842630c87f041540c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-07-16 21:55:28.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"b7cfa377fc0cf91ef4d20f245c6105e360ddd335","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-07-16 22:19:58.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/31/666631/7/check/check-dsvm-tempest-vz7-exe-minimal/5740889 : FAILURE in 16m 54s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":7},{"id":"94aad1e58bf61a158ab9fc74480114ebd342cf29","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-07-16 22:58:24.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-33950 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":7},{"id":"7aa876afea854f6ff9d79735a20d6caab0f2d44e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-16 23:23:09.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/7/check/tempest-dsvm-full-xenial/27d0046/ : SUCCESS in 1h 25m 58s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/7/check/tempest-dsvm-full-xenial-py3/34a1610/ : FAILURE in 1h 26m 55s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/31/666631/7/check/grenade-dsvm-xenial/08d1718/ : SUCCESS in 1h 14m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"da908e14e3d0637e79bd05f3257c00129a2e360d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 01:26:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 http://logs.openstack.org/31/666631/7/check/grenade-py3/7663a96/ : SUCCESS in 1h 04m 16s\n- tempest-full-py3 http://logs.openstack.org/31/666631/7/check/tempest-full-py3/447de74/ : SUCCESS in 1h 50m 14s\n- openstack-tox-cover http://logs.openstack.org/31/666631/7/check/openstack-tox-cover/33c4d4b/cover/ : SUCCESS in 16m 30s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/7/check/openstack-tox-lower-constraints/e3b5138/ : SUCCESS in 17m 54s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/7/check/openstack-tox-pep8/07522f6/ : SUCCESS in 9m 54s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/7/check/openstack-tox-py27/8caf511/ : SUCCESS in 17m 05s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/7/check/openstack-tox-py36/2de8e2c/ : SUCCESS in 18m 25s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/7/check/openstack-tox-py37/c257291/ : SUCCESS in 18m 02s\n- openstack-tox-docs http://logs.openstack.org/31/666631/7/check/openstack-tox-docs/1743dc9/html/ : SUCCESS in 6m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/31/666631/7/check/ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa/c4a5a7e/ : SUCCESS in 58m 42s (non-voting)\n- devstack-plugin-ceph-tempest http://logs.openstack.org/31/666631/7/check/devstack-plugin-ceph-tempest/737103b/ : SUCCESS in 1h 08m 15s (non-voting)\n- neutron-grenade-multinode http://logs.openstack.org/31/666631/7/check/neutron-grenade-multinode/641ac7b/ : SUCCESS in 1h 07m 28s\n- neutron-tempest-linuxbridge http://logs.openstack.org/31/666631/7/check/neutron-tempest-linuxbridge/8e11926/ : SUCCESS in 1h 24m 17s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/7/check/nova-grenade-live-migration/4c48b63/ : SUCCESS in 1h 06m 00s\n- nova-live-migration http://logs.openstack.org/31/666631/7/check/nova-live-migration/5fa8dae/ : SUCCESS in 1h 01m 04s\n- nova-lvm http://logs.openstack.org/31/666631/7/check/nova-lvm/ff3aa56/ : SUCCESS in 50m 06s (non-voting)\n- nova-multi-cell http://logs.openstack.org/31/666631/7/check/nova-multi-cell/d414d8a/ : SUCCESS in 1h 30m 13s (non-voting)\n- nova-next http://logs.openstack.org/31/666631/7/check/nova-next/7868423/ : SUCCESS in 2h 30m 53s\n- nova-tox-functional http://logs.openstack.org/31/666631/7/check/nova-tox-functional/c70e802/ : SUCCESS in 32m 55s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/7/check/nova-tox-functional-py36/a2918ea/ : SUCCESS in 28m 53s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/7/check/tempest-slow-py3/541997c/ : SUCCESS in 2h 59m 28s","accounts_in_message":[],"_revision_number":7},{"id":"d5f3ae770e3d05ed4277e9fe683a93374bc853ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 01:26:49.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"5babd608cece0d6380c6aed8a29b68a2359d637e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-07-17 01:26:57.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/31/666631/7/check/EMC_VxFlexOS_NOVA/8eb4cd8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":7},{"id":"45ac3bbac7ee2c06650ecba5b09a52bc9cffbb78","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-07-17 03:13:46.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-31-666631-7 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":7},{"id":"15f361e07fd56d5ab2f924b3c95fc4e079e89b67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 04:15:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d8ff94f0f88849b4125970d38ba2508a27f7d581","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 04:15:53.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 http://logs.openstack.org/31/666631/7/gate/grenade-py3/552da5f/ : SUCCESS in 1h 00m 43s\n- tempest-full-py3 http://logs.openstack.org/31/666631/7/gate/tempest-full-py3/065f09e/ : SUCCESS in 1h 31m 35s\n- openstack-tox-lower-constraints http://logs.openstack.org/31/666631/7/gate/openstack-tox-lower-constraints/bd61257/ : SUCCESS in 18m 47s\n- openstack-tox-pep8 http://logs.openstack.org/31/666631/7/gate/openstack-tox-pep8/3df3dd4/ : SUCCESS in 10m 21s\n- openstack-tox-py27 http://logs.openstack.org/31/666631/7/gate/openstack-tox-py27/47dc9cc/ : SUCCESS in 16m 03s\n- openstack-tox-py36 http://logs.openstack.org/31/666631/7/gate/openstack-tox-py36/9596e11/ : SUCCESS in 16m 05s\n- openstack-tox-py37 http://logs.openstack.org/31/666631/7/gate/openstack-tox-py37/4722a9c/ : SUCCESS in 19m 06s\n- openstack-tox-docs http://logs.openstack.org/31/666631/7/gate/openstack-tox-docs/5db2111/html/ : SUCCESS in 6m 52s\n- nova-grenade-live-migration http://logs.openstack.org/31/666631/7/gate/nova-grenade-live-migration/d155a6c/ : SUCCESS in 1h 14m 30s\n- nova-live-migration http://logs.openstack.org/31/666631/7/gate/nova-live-migration/58941cd/ : SUCCESS in 1h 01m 25s\n- nova-tox-functional http://logs.openstack.org/31/666631/7/gate/nova-tox-functional/f00cb79/ : SUCCESS in 25m 58s\n- nova-tox-functional-py36 http://logs.openstack.org/31/666631/7/gate/nova-tox-functional-py36/f55578b/ : SUCCESS in 22m 32s\n- nova-next http://logs.openstack.org/31/666631/7/gate/nova-next/61a1932/ : SUCCESS in 1h 16m 12s\n- tempest-slow-py3 http://logs.openstack.org/31/666631/7/gate/tempest-slow-py3/fb00fd7/ : SUCCESS in 2h 44m 40s","accounts_in_message":[],"_revision_number":7},{"id":"c25a12eec2b6c6497582f6dcf40ef1d59791f70e","author":{"_account_id":14807,"name":"Lenny","email":"lennyb@mellanox.com","username":"lennyb"},"date":"2019-08-11 12:13:05.000000000","message":"Created a revert of this change as Ieb901b802f7e3f08e0b5a2443bd8d3f1783260eb","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"3f56e44b843ab8c0ee017eb58a8d77679aeb7b3b","revisions":{"7e07c40e5c63b48202781f42b643c1a5fd4bd8e6":{"kind":"REWORK","_number":1,"created":"2019-06-20 16:18:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/1"}}},"commit":{"parents":[{"commit":"736d58b178c07e868a37a326738444768dace8a8","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/736d58b178c07e868a37a326738444768dace8a8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 16:18:21.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e07c40e5c63b48202781f42b643c1a5fd4bd8e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e07c40e5c63b48202781f42b643c1a5fd4bd8e6"}]},"branch":"refs/heads/master"},"6df58f3ab1635e2e2e1b7318a2f8387211738c9a":{"kind":"REWORK","_number":2,"created":"2019-06-20 16:48:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/2"}}},"commit":{"parents":[{"commit":"32c9f7a26d93fb0cca2e9b2d1fca395619eda61a","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/32c9f7a26d93fb0cca2e9b2d1fca395619eda61a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 16:47:48.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6df58f3ab1635e2e2e1b7318a2f8387211738c9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6df58f3ab1635e2e2e1b7318a2f8387211738c9a"}]},"branch":"refs/heads/master"},"3d74b6436d00babe28ff4643779841276bf9c478":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-21 10:05:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/3"}}},"commit":{"parents":[{"commit":"d77b0a4cdd6a459c123978821b557ac95f49baf8","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d77b0a4cdd6a459c123978821b557ac95f49baf8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-21 10:05:08.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d74b6436d00babe28ff4643779841276bf9c478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d74b6436d00babe28ff4643779841276bf9c478"}]},"branch":"refs/heads/master"},"6f49d5c760cd8f0823874f0490e897dae57a2317":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-06-21 10:48:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/4"}}},"commit":{"parents":[{"commit":"07e2d1f2ba1d6107112d205c26d9f46413ef932b","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/07e2d1f2ba1d6107112d205c26d9f46413ef932b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-21 10:46:57.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f49d5c760cd8f0823874f0490e897dae57a2317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6f49d5c760cd8f0823874f0490e897dae57a2317"}]},"branch":"refs/heads/master"},"d2e899f5fcbd4de57e2f5813a922d815b31bd571":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-06-21 11:05:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/5"}}},"commit":{"parents":[{"commit":"bb82b16baa70cb764edffd064dc9abe4d2731139","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb82b16baa70cb764edffd064dc9abe4d2731139"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-21 11:04:50.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2e899f5fcbd4de57e2f5813a922d815b31bd571"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d2e899f5fcbd4de57e2f5813a922d815b31bd571"}]},"branch":"refs/heads/master"},"fdfaf7791708152e2aeb0ed289f63a4aa08cfe85":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-07-10 15:19:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/6"}}},"commit":{"parents":[{"commit":"821df7c27f76d9ca2f9631ead7e35d3cc8a1019e","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/821df7c27f76d9ca2f9631ead7e35d3cc8a1019e"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-10 15:17:59.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdfaf7791708152e2aeb0ed289f63a4aa08cfe85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdfaf7791708152e2aeb0ed289f63a4aa08cfe85"}]},"branch":"refs/heads/master"},"3f56e44b843ab8c0ee017eb58a8d77679aeb7b3b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-07-12 15:40:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/666631/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/31/666631/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/31/666631/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/31/666631/7"}}},"commit":{"parents":[{"commit":"b4f3b2b09f8c836c470065be09be9281c336dec3","subject":"vif: Stop using getattr for VIF lookups","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4f3b2b09f8c836c470065be09be9281c336dec3"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-06-20 13:24:48.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-07-12 15:39:58.000000000","tz":60},"subject":"vif: Resolve a TODO and update another","message":"vif: Resolve a TODO and update another\n\nOnce TODO noted that a block could be removed once we bump to libvirt\n1.3.8 or greater. We require 3.0.0 now so that\u0027s resolved. Another one\nlooks like it should be resolved in 3.2.0 so the TODO is updated to\nhighlight this for future reviewers.\n\nChange-Id: I5235751b1dbc77ecc919eec7f3e022cd70085051\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3f56e44b843ab8c0ee017eb58a8d77679aeb7b3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3f56e44b843ab8c0ee017eb58a8d77679aeb7b3b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
