)]}'
{"id":"openstack%2Fnova~669169","triplet_id":"openstack%2Fnova~stable%2Fqueens~I67f341bf895d6cc5d503da274c089d443295199e","project":"openstack/nova","branch":"stable/queens","topic":"bug/1834048","hashtags":[],"change_id":"I67f341bf895d6cc5d503da274c089d443295199e","subject":"libvirt: Add a rbd_connect_timeout configurable","status":"ABANDONED","created":"2019-07-04 12:42:27.000000000","updated":"2020-01-21 13:39:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e8db9ca9294d6ec8986d1838b9621ace6e1cb52a","_number":669169,"virtual_id_number":669169,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-02 20:13:14.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-03 01:57:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"date":"2019-12-02 18:02:46.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-02 19:17:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-10 12:09:03.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-09 13:37:03.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-02 18:02:46.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-12-02 19:17:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-02 20:13:14.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-03 01:57:58.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"363603c1af543a117542190d40dc1f3c73b35d54","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-07-04 12:42:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1c31feebc04da7d345bca7d583722663f3e48ffa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 20:23:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/69/669169/1/check/openstack-tox-pep8/4003d63/ : SUCCESS in 8m 34s\n- openstack-tox-py27 http://logs.openstack.org/69/669169/1/check/openstack-tox-py27/e962850/ : SUCCESS in 13m 10s\n- openstack-tox-py35 http://logs.openstack.org/69/669169/1/check/openstack-tox-py35/e022f47/ : SUCCESS in 17m 32s\n- build-openstack-sphinx-docs http://logs.openstack.org/69/669169/1/check/build-openstack-sphinx-docs/fe92139/html/ : SUCCESS in 7m 03s\n- tempest-full http://logs.openstack.org/69/669169/1/check/tempest-full/82ab209/ : SUCCESS in 1h 30m 11s\n- neutron-grenade http://logs.openstack.org/69/669169/1/check/neutron-grenade/22de394/ : SUCCESS in 57m 20s\n- tempest-full-py3 http://logs.openstack.org/69/669169/1/check/tempest-full-py3/cc5120d/ : SUCCESS in 1h 10m 50s\n- build-openstack-releasenotes http://logs.openstack.org/69/669169/1/check/build-openstack-releasenotes/c21fc60/html/ : SUCCESS in 36m 41s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa http://logs.openstack.org/69/669169/1/check/ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa/38f8dd4/ : SUCCESS in 49m 19s (non-voting)\n- nova-cells-v1 http://logs.openstack.org/69/669169/1/check/nova-cells-v1/96a6f18/ : SUCCESS in 51m 26s\n- nova-grenade-live-migration http://logs.openstack.org/69/669169/1/check/nova-grenade-live-migration/86222ad/ : SUCCESS in 1h 02m 34s (non-voting)\n- nova-live-migration http://logs.openstack.org/69/669169/1/check/nova-live-migration/66d7405/ : SUCCESS in 46m 03s\n- nova-multiattach http://logs.openstack.org/69/669169/1/check/nova-multiattach/5659505/ : SUCCESS in 1h 09m 45s\n- nova-next http://logs.openstack.org/69/669169/1/check/nova-next/2104029/ : SUCCESS in 1h 27m 13s\n- tempest-slow http://logs.openstack.org/69/669169/1/check/tempest-slow/428aa5f/ : SUCCESS in 1h 58m 34s\n- nova-tox-functional http://logs.openstack.org/69/669169/1/check/nova-tox-functional/673a418/ : SUCCESS in 20m 47s\n- nova-tox-functional-py35 http://logs.openstack.org/69/669169/1/check/nova-tox-functional-py35/3549ec4/ : SUCCESS in 19m 10s\n- neutron-grenade-multinode http://logs.openstack.org/69/669169/1/check/neutron-grenade-multinode/239979b/ : SUCCESS in 1h 14m 13s\n- devstack-plugin-ceph-tempest http://logs.openstack.org/69/669169/1/check/devstack-plugin-ceph-tempest/3ae79a8/ : SUCCESS in 59m 42s (non-voting)\n- tempest-multinode-full http://logs.openstack.org/69/669169/1/check/tempest-multinode-full/4a646f3/ : SUCCESS in 1h 40m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"acb0cf3c185d7585c74b19ac5f25c80eff573fe7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-07-05 01:34:14.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/669169/1 : FAILURE in 44m 03s","accounts_in_message":[],"_revision_number":1},{"id":"7d991ebf646e271a66cbb335cbc98324bf81d9f5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-07-05 10:10:01.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/69/669169/1/check/tempest-dsvm-full-xenial/413564e/ : SUCCESS in 1h 00m 47s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/69/669169/1/check/tempest-dsvm-full-xenial-py3/bc82443/ : FAILURE in 1h 16m 28s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/69/669169/1/check/grenade-dsvm-xenial/fd4dd19/ : SUCCESS in 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9116018a44fd721b274e4e6061c7e763db6f01bc","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-07-10 12:09:03.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/669169/1/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/ae3a729 : EXCEPTION","accounts_in_message":[],"_revision_number":1},{"id":"820f9859ac2dbd3873cbcf91df104ecb9158ee4f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-09 13:37:03.000000000","message":"Patch Set 1: Workflow-1\n\nI\u0027m not sure about this, especially on Queens, but waiting for rocky and stein anyway.","accounts_in_message":[],"_revision_number":1},{"id":"18f37b5f430c1dd191d7474146fdd24626b3c15e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-09-18 14:49:33.000000000","message":"Abandoned\n\nHolding off as we didn\u0027t get any positive feedback about backporting this downstream, I might reopen these once the change has been released in Train for a while.","accounts_in_message":[],"_revision_number":1},{"id":"a38b5c5b22d8a4caf02b3b01f9e518751fc92557","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-02 17:12:24.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"078dd7bcc73a626a6e06ab2479e7018fc0fb43e6","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-02 17:12:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a557ca2935317ce6ccc185d29d8408139b679a75","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-02 18:02:46.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/69/669169/2/check/pmem-tempest-plugin-filtered/59a8667/ : POST_FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"db9a9d17d65583ab526479b54190539c36d0c126","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 19:17:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fd659352fc14abaa7da7b202c4bdf82 : SUCCESS in 10m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61bd09d1d394470d9e46c057757de87a : SUCCESS in 14m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fb35c581fd2d4ab69b0660c87bae18af : SUCCESS in 13m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7add0e79fdb9461eb46dcab3721258a2 : SUCCESS in 6m 48s\n- tempest-full https://zuul.opendev.org/t/openstack/build/684a60df38d04e4eb4a1b39f9a27e637 : SUCCESS in 1h 28m 38s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5020e0492e4f4fe69f2b23dfa0392106 : SUCCESS in 1h 10m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02ec0aa8129346768150c6e76b58ad02 : SUCCESS in 1h 13m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de61826d2ec8485cb81c55548793781a : SUCCESS in 58m 59s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7cc76a089b954bf299a9705654568c3a : SUCCESS in 49m 35s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/5e0e37a502c742e89f85c975d6264579 : SUCCESS in 1h 06m 58s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/657ddd7349f445a58a7f0b0a82247cee : FAILURE in 1h 11m 50s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a91699653687462490028dd2b7c2bbdb : SUCCESS in 38m 25s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/c6343368c9b14ecba88b104f05479e3e : FAILURE in 1h 10m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/934e10f1e13542cea991e465dc44390d : SUCCESS in 1h 18m 56s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/b75dca9200cf415aac5d2f98e846ce86 : SUCCESS in 1h 48m 16s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/9c532267e58e4b7a851549c229a94bf0 : SUCCESS in 23m 45s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/048520300e4d4736b561077d47115f68 : SUCCESS in 19m 42s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/abb1f3b7c8874bc79d49b828faf9a9cd : SUCCESS in 1h 01m 44s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0fdb01928c23455588633f2d5a49c1f4 : SUCCESS in 1h 01m 05s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/95de643ab5574b2b85175784a9dc9db3 : SUCCESS in 1h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f3d591fa45df62342116bae31bbe9a4594e6580a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-02 20:13:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/669169/2 : FAILURE in 1h 08m 13s","accounts_in_message":[],"_revision_number":2},{"id":"6377e117abcdff4205303b93ebd4239d9bfed6b1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 01:57:58.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/69/669169/2/check/tempest-dsvm-full-xenial/5032986/ : SUCCESS in 2h 33m 06s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/69/669169/2/check/tempest-dsvm-full-xenial-py3/9a9b1cc/ : FAILURE in 2h 35m 32s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/69/669169/2/check/grenade-dsvm-xenial/94a9522/ : FAILURE in 2h 56m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35f1e4afe57d4ed6fed7a1cfdbbde7ea0b73b82b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-01-21 13:39:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f1d0850484ce798dbaf262091413dda96ba699b1","revisions":{"3692f7773c3d25cb9d9ab1a6fd91aea7b95d782f":{"kind":"REWORK","_number":1,"created":"2019-07-04 12:42:27.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/69/669169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/669169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/669169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/669169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/669169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/669169/1"}}},"commit":{"parents":[{"commit":"de7a8f9e443b7cf20c18e8637b43e0f2bfada31c","subject":"Merge \"fix up numa-topology live migration hypervisor check\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/de7a8f9e443b7cf20c18e8637b43e0f2bfada31c"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-06-25 17:20:24.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-07-04 12:42:01.000000000","tz":60},"subject":"libvirt: Add a rbd_connect_timeout configurable","message":"libvirt: Add a rbd_connect_timeout configurable\n\nPreviously the initial call to connect to a RBD cluster via the RADOS\nAPI could hang indefinitely if network or other environmental related\nissues were encountered.\n\nWhen encountered during a call to update_available_resource this can\nresult in the local n-cpu service reporting as UP while never being able\nto break out of a subsequent RPC timeout loop as documented in bug\n\nThis change adds a simple timeout configurable to be used when initially\nconnecting to the cluster [1][2][3]. The default timeout of 5 seconds\nbeing sufficiently small enough to ensure that if encountered the n-cpu\nservice will be able to be marked as DOWN before a RPC timeout is seen.\n\n[1] http://docs.ceph.com/docs/luminous/rados/api/python/#rados.Rados.connect\n[2] http://docs.ceph.com/docs/mimic/rados/api/python/#rados.Rados.connect\n[3] http://docs.ceph.com/docs/nautilus/rados/api/python/#rados.Rados.connect\n\nCloses-bug: #1834048\nChange-Id: I67f341bf895d6cc5d503da274c089d443295199e\n(cherry picked from commit 03f7dc29b75d1099ef44a034ed7e23d2a4444ac6)\n(cherry picked from commit a27ed51cb31912a5159f0cad804fdc447c6b0407)\n(cherry picked from commit 0f29cea0d4d692060c92cdd6e48d93f0f029cb4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3692f7773c3d25cb9d9ab1a6fd91aea7b95d782f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3692f7773c3d25cb9d9ab1a6fd91aea7b95d782f"}]},"branch":"refs/heads/stable/queens"},"f1d0850484ce798dbaf262091413dda96ba699b1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-02 17:12:38.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/69/669169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/69/669169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/669169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/669169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/69/669169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/69/669169/2"}}},"commit":{"parents":[{"commit":"4fcb7816bc88fd513debe70b95aa60bff74e37fb","subject":"Fix false ERROR message at compute restart","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4fcb7816bc88fd513debe70b95aa60bff74e37fb"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-06-25 17:20:24.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-02 17:12:38.000000000","tz":0},"subject":"libvirt: Add a rbd_connect_timeout configurable","message":"libvirt: Add a rbd_connect_timeout configurable\n\nPreviously the initial call to connect to a RBD cluster via the RADOS\nAPI could hang indefinitely if network or other environmental related\nissues were encountered.\n\nWhen encountered during a call to update_available_resource this can\nresult in the local n-cpu service reporting as UP while never being able\nto break out of a subsequent RPC timeout loop as documented in bug\n\nThis change adds a simple timeout configurable to be used when initially\nconnecting to the cluster [1][2][3]. The default timeout of 5 seconds\nbeing sufficiently small enough to ensure that if encountered the n-cpu\nservice will be able to be marked as DOWN before a RPC timeout is seen.\n\n[1] http://docs.ceph.com/docs/luminous/rados/api/python/#rados.Rados.connect\n[2] http://docs.ceph.com/docs/mimic/rados/api/python/#rados.Rados.connect\n[3] http://docs.ceph.com/docs/nautilus/rados/api/python/#rados.Rados.connect\n\nCloses-bug: #1834048\nChange-Id: I67f341bf895d6cc5d503da274c089d443295199e\n(cherry picked from commit 03f7dc29b75d1099ef44a034ed7e23d2a4444ac6)\n(cherry picked from commit a27ed51cb31912a5159f0cad804fdc447c6b0407)\n(cherry picked from commit 0f29cea0d4d692060c92cdd6e48d93f0f029cb4b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f1d0850484ce798dbaf262091413dda96ba699b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f1d0850484ce798dbaf262091413dda96ba699b1"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
