)]}'
{"id":"openstack%2Fnova~670226","triplet_id":"openstack%2Fnova~stable%2Fqueens~I141c90ac27d16f2e9c033bcd2f95ac08904a2f52","project":"openstack/nova","branch":"stable/queens","topic":"bug/1836005","hashtags":[],"change_id":"I141c90ac27d16f2e9c033bcd2f95ac08904a2f52","subject":"doc: Fix a parameter of NotificationPublisher","status":"MERGED","created":"2019-07-11 04:05:31.000000000","updated":"2019-08-10 03:31:07.000000000","submitted":"2019-08-10 03:31:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670226-1565407867053-c4f9ddb8","meta_rev_id":"8943df578ddfe2a0f496bb565c439a2b646453bb","_number":670226,"virtual_id_number":670226,"owner":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-08-09 23:49:04.000000000","_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-08-09 16:52:30.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-08-10 03:31:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-08-09 16:05:44.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-08-09 16:05:47.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-09 16:05:47.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-08-09 16:52:30.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-08-09 23:49:04.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-10 03:31:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"071f06af09238fdde87da750adbbdfc629749187","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2019-07-11 04:05:31.000000000","message":"Patch Set 1: Cherry Picked from branch stable/rocky.","accounts_in_message":[],"_revision_number":1},{"id":"11c7493eafe5a20cf56a99fb2e9b40d2da143d21","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-07-11 04:07:00.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/670226/1 : FAILURE in 39s","accounts_in_message":[],"_revision_number":1},{"id":"4d67bfca3b7774fe934ee58a3cc0ea9ae7dc9cfe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 04:19:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/26/670226/1/check/openstack-tox-pep8/970b479/ : SUCCESS in 10m 20s\n- build-openstack-sphinx-docs http://logs.openstack.org/26/670226/1/check/build-openstack-sphinx-docs/a03b2a9/html/ : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":1},{"id":"af1201a613d8652bcab00a0c0b00996a4a4af4e3","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-09 13:33:44.000000000","message":"Topic changed from bug/1836005-stable/queens to bug/1836005","accounts_in_message":[],"_revision_number":1},{"id":"fc44a1a0a72dc1832a42f60e9bb7a7d72da4f36a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-09 13:34:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nClean backport from rocky, trivial docs fix, I\u0027m going to fast approve.","accounts_in_message":[],"_revision_number":1},{"id":"4f8fa461d94e2835929763ae812f7ce18ba3d7ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:42:18.000000000","message":"Patch Set 1: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"6f9d8041ead8547380faf5f222ccaa77f7c9791d","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-09 16:05:44.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a1e81a227f34cfcf179d42da9de8219b4348cf9f","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-08-09 16:05:47.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f20fa8584ff100d5815ff5d296f3573f1ec83ade","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-09 16:52:30.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/670226/2 : FAILURE in 43m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a3dd345cab5831c6f9db37d1ee37c59159d15269","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 19:24:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/26/670226/2/check/openstack-tox-pep8/9e6a2b7/ : SUCCESS in 8m 38s\n- build-openstack-sphinx-docs https://logs.opendev.org/26/670226/2/check/build-openstack-sphinx-docs/2035760/html/ : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":2},{"id":"5155e1ddf4bfc0f0c94e3468cdc480b70a4c0ec2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 22:45:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5e30e2e04920e3a1f011ed133b5ce8ae8195e24f","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-08-09 23:49:04.000000000","message":"Patch Set 2:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/26/670226/2/check/nova-out-of-tree-pvm/7423238 : FAILURE in 20m 41s\n- nova-in-tree-pvm http://184.172.12.213/26/670226/2/check/nova-in-tree-pvm/19b5786 : FAILURE in 21m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f9ec476a14de131197967db94209b6a40f51b376","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:31:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/26/670226/2/gate/openstack-tox-pep8/f7efcea/ : SUCCESS in 8m 40s\n- build-openstack-sphinx-docs https://logs.opendev.org/26/670226/2/gate/build-openstack-sphinx-docs/1758149/html/ : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8943df578ddfe2a0f496bb565c439a2b646453bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-10 03:31:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b6108ebb3f31fe835e2337d0a7449a69ab613c04","revisions":{"d0fb978ac90ce2a5f359f873367fdd43f9e1854d":{"kind":"REWORK","_number":1,"created":"2019-07-11 04:05:31.000000000","uploader":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"ref":"refs/changes/26/670226/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/670226/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/670226/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/670226/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/670226/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/670226/1"}}},"commit":{"parents":[{"commit":"95a35cbee44f8e97aa6d3a04820175c052f03cb3","subject":"Merge \"Init HostState.failed_builds\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95a35cbee44f8e97aa6d3a04820175c052f03cb3"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-07-10 07:06:22.000000000","tz":540},"committer":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-07-11 04:05:31.000000000","tz":0},"subject":"doc: Fix a parameter of NotificationPublisher","message":"doc: Fix a parameter of NotificationPublisher\n\nThe \u0027binary\u0027 parameter has been changed to the \u0027source\u0027\nsince I95b5b0826190d396efe7bfc017f6081a6356da65.\nBut the notification document has not been updated yet.\n\nReplace the \u0027binary\u0027 parameter with the \u0027source\u0027 parameter.\n\nChange-Id: I141c90ac27d16f2e9c033bcd2f95ac08904a2f52\nCloses-Bug: #1836005\n(cherry picked from commit de31466fdb09dd367b23660f71b4ca06f83271a2)\n(cherry picked from commit 5ba20d128cd0dee8b9b2a429a71e1bac5b0220df)\n(cherry picked from commit 7793f4d759d35841848597e27faadb1903df21e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0fb978ac90ce2a5f359f873367fdd43f9e1854d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d0fb978ac90ce2a5f359f873367fdd43f9e1854d"}]},"branch":"refs/heads/stable/queens"},"b6108ebb3f31fe835e2337d0a7449a69ab613c04":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-08-09 16:05:44.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/26/670226/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/26/670226/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/670226/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/670226/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/26/670226/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/26/670226/2"}}},"commit":{"parents":[{"commit":"e04268f0dcf7a3c5805e29a9eaa5604374ded69e","subject":"Merge \"Cleanup when hitting MaxRetriesExceeded from no host_available\" into stable/queens","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e04268f0dcf7a3c5805e29a9eaa5604374ded69e"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-07-10 07:06:22.000000000","tz":540},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-08-09 16:05:44.000000000","tz":0},"subject":"doc: Fix a parameter of NotificationPublisher","message":"doc: Fix a parameter of NotificationPublisher\n\nThe \u0027binary\u0027 parameter has been changed to the \u0027source\u0027\nsince I95b5b0826190d396efe7bfc017f6081a6356da65.\nBut the notification document has not been updated yet.\n\nReplace the \u0027binary\u0027 parameter with the \u0027source\u0027 parameter.\n\nChange-Id: I141c90ac27d16f2e9c033bcd2f95ac08904a2f52\nCloses-Bug: #1836005\n(cherry picked from commit de31466fdb09dd367b23660f71b4ca06f83271a2)\n(cherry picked from commit 5ba20d128cd0dee8b9b2a429a71e1bac5b0220df)\n(cherry picked from commit 7793f4d759d35841848597e27faadb1903df21e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6108ebb3f31fe835e2337d0a7449a69ab613c04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6108ebb3f31fe835e2337d0a7449a69ab613c04"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
