)]}'
{"id":"openstack%2Fnova~678453","triplet_id":"openstack%2Fnova~master~Idd49b0c70caedfcd42420ffa2ac926a6087d406e","project":"openstack/nova","branch":"master","topic":"bp/virtual-persistent-memory","hashtags":[],"change_id":"Idd49b0c70caedfcd42420ffa2ac926a6087d406e","subject":"libvirt: Enable driver discovering PMEM namespaces","status":"MERGED","created":"2019-08-26 03:25:40.000000000","updated":"2019-09-17 05:45:49.000000000","submitted":"2019-09-15 21:22:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":101,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678453-1568582569631-65f41938","meta_rev_id":"01e6a562df64ec494da5a8b8e9ba65fc49cf0e87","_number":678453,"virtual_id_number":678453,"owner":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-15 18:49:33.000000000","_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-14 04:54:53.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-15 18:56:03.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-15 08:04:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-09-15 08:04:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2019-09-15 17:33:54.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":2,"date":"2019-09-15 21:22:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},{"value":0,"date":"2019-09-15 18:48:18.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-09-15 10:59:21.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"date":"2019-09-15 19:17:58.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-09-15 16:38:37.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-09-13 08:55:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-09-13 08:55:20.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-09-13 09:20:39.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-05 13:17:31.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2019-09-11 10:35:27.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-11 12:29:22.000000000","updated_by":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"reviewer":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"state":"REVIEWER"},{"updated":"2019-09-11 16:51:31.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-12 21:04:39.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-09-13 09:20:39.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-09-14 04:54:53.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 08:04:00.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 08:04:34.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-09-15 10:59:21.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 16:38:37.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 17:33:54.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-09-15 18:48:18.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 18:49:33.000000000","updated_by":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 18:56:03.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-15 19:17:58.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-09-15 21:22:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f86a8895f0bb8faa2652a6204fa50948e85fe42f","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-26 03:25:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9375963fa98ea5b0cd44613450e73f005e2b330","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 03:32:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22803ec81116f96e60c75541c77a78498e057844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 05:43:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8791e1c453c849269d3c7f055b21e5cd : SUCCESS in 1h 02m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/98adda176b2e4f5b99cd28accfb35c59 : SUCCESS in 1h 21m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cda32e38ca5401fb25b0388f8d9e84b : SUCCESS in 16m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/84f05c63dc0f444095ae90468fa94d53 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be24306d76c4f2e9c0b71a2f3378599 : SUCCESS in 11m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8ac3b4979c7479c93f8aceeed77cd8b : SUCCESS in 14m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c78e918866bd41799ba92d1d96e98051 : SUCCESS in 13m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/65e6187a6d4742268471da9d117052fa : SUCCESS in 15m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00e4efffe2c147cf8b232837759b929c : SUCCESS in 7m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/158893a475f5485e868c29aff37434e7 : SUCCESS in 49m 17s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a56b2948474f4357af802475d94fcd2d : SUCCESS in 1h 01m 03s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/75506f489b6347ae82a780cd79d94790 : SUCCESS in 1h 09m 10s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/e519a87afb4741c19937d0d4ae956432 : SUCCESS in 1h 19m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/205bb9fa4c62408bb7a5bd4324e54ad4 : FAILURE in 1h 02m 52s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fbb1df40d9684aa6976b8cda9e8a8b04 : SUCCESS in 56m 05s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9276bee5badf4da3b030cbc76c492765 : SUCCESS in 1h 23m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/cd64097ec00a46839854cb403248c7e7 : SUCCESS in 1h 27m 18s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/2365231713dd4c229a41b9b8544c251c : SUCCESS in 21m 35s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b56f0c04fc864dcf95e8c53ef5034ba7 : SUCCESS in 20m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5dcb06d9f7c24e0c87c00b9cef2a5a3a : SUCCESS in 2h 12m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f2b07d68d314508b7ade0830be1be9c : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":1},{"id":"aacaddfb71aeb3adafd7357a49e79cb63d17d787","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-26 06:45:20.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"f688332dc205813102a9ab07387a9c1913613b11","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 06:52:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e4d0794008f5091f94810d8e3c564bad206202b2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-26 09:23:33.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35848 : FAILURE in 1h 26m 09s","accounts_in_message":[],"_revision_number":2},{"id":"19515567fdffe29a48b9b9110a4248b53cca148d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 09:55:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/202d891286cd4ed3a2ca6c93ce2e9597 : SUCCESS in 1h 15m 12s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3c7d0a7840d54af2baf96373f2cd78ea : FAILURE in 28m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d79ac48ba1684e80b8c9242df9f2a167 : SUCCESS in 39m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0876450d836442fa8fbca111b3ab75e : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3b0759186cf4369aae8a13ef1893d11 : SUCCESS in 11m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c500e8802bdf4fae97d50faf2b4f594d : SUCCESS in 12m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1c85209c0f241ada1bbeb098a0722ba : SUCCESS in 11m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b00aed5eebca4714b0cafc49a73d1cdc : SUCCESS in 12m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e455e7b7f3e4f6a9e0fb771bfdccc74 : SUCCESS in 15m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f2894b03f33a4a3aa3eca358ced7fe7f : SUCCESS in 53m 46s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0b97b40b0b9741caba04800e0774b60f : SUCCESS in 1h 10m 09s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3f3ff4478af14bd0a76589ddda55474c : SUCCESS in 1h 25m 45s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/5359285e5eb3419fa23ffcffa7d92192 : SUCCESS in 1h 12m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4f6c94bbbea1488db2e2b0dba5e14352 : FAILURE in 1h 04m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/962c0c6e35ca4e0f9ac1a31624953543 : SUCCESS in 1h 11m 16s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/57a1197ab1874dedb63d13b97ee941cd : SUCCESS in 1h 42m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/ec7936b61dfb4d32b7acfca137e4f2ee : SUCCESS in 1h 34m 22s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/22bd68f8dc244759a8ee89bd409f3aa1 : SUCCESS in 21m 46s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3cbc738088c4787ad6848fee8e30e9d : SUCCESS in 32m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/997968ca8eaf46d48091129db80a3a55 : SUCCESS in 2h 26m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37c9cdbaec48458e97da85cb66d31f16 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":2},{"id":"937520b3e98af60b5f6e42e70130f6ad0432e888","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-26 10:54:16.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/2/check/tempest-dsvm-full-xenial/952285c/ : SUCCESS in 1h 56m 05s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/2/check/tempest-dsvm-full-xenial-py3/66b6ec9/ : SUCCESS in 2h 14m 54s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/2/check/grenade-dsvm-xenial/877d4d4/ : SUCCESS in 2h 00m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8737bf53164da76bd800a1d77369028cf368aaa","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-26 12:39:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1d23083e6a89f513952fde17a91ef49d4986619f","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-26 12:46:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"47921ce83010864c1f5a1c5af2b897020b4e973d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-26 14:33:55.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35876 : FAILURE in 1h 39m 59s","accounts_in_message":[],"_revision_number":3},{"id":"4db8f4b455439520daab9f458de97474709d7ea0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-26 15:23:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c07ba641684f46339a5753763debc198 : SUCCESS in 1h 10m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f75068656c6d44408629dba03a4f9e8f : SUCCESS in 1h 31m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cd87bc9d98a40f6b52e645d8d9cbeae : SUCCESS in 18m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb4c1a5dc92d47b1bbeec2bedeea1985 : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a8c5e78449e4069b31727f53e927232 : SUCCESS in 10m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5e2af2132f7f4741984e2dc2436e04ba : SUCCESS in 17m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7741e41b409f4aa69539c8d36a71e7de : SUCCESS in 13m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b109d0b6fb9b490dbfa8185be2b9499f : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ed089e5032a4cb2975b9a3ce5478c39 : SUCCESS in 9m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/17372b4192d24a41a097cb4562774bb2 : SUCCESS in 56m 36s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1ace373d4b7d42fe8e16dfa6cf22a254 : SUCCESS in 1h 16m 56s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c40b54ad86ea41c191b71e67fb5e2c26 : SUCCESS in 1h 14m 00s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/17f74ec59d494d60863fe1833a987a9a : SUCCESS in 1h 14m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a6f80019fc494886bdbe0e2a5faa6026 : FAILURE in 1h 11m 47s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/42b64c15a815443bac7c1b505e176a8c : SUCCESS in 51m 27s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1da58eba96cb4de189095f18765367a4 : SUCCESS in 1h 17m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/637c36b62c764a8dbd8fa93690970918 : SUCCESS in 1h 31m 13s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/cd4ccbf1e6f84b188b2956e4d6ab02ff : SUCCESS in 22m 49s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ba1809cd9f84a85be15422a6d185672 : SUCCESS in 21m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9db58ba64ccc4e33a63f6b2203be4693 : SUCCESS in 2h 29m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50226613c5894e68bca3cab87a5c41c7 : SUCCESS in 1h 06m 43s","accounts_in_message":[],"_revision_number":3},{"id":"cee3722b811ec21ec0b2909df657554c366e02cd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-26 15:23:25.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/3/check/tempest-dsvm-full-xenial/5074320/ : SUCCESS in 2h 15m 49s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/3/check/tempest-dsvm-full-xenial-py3/7a6a5a6/ : SUCCESS in 2h 13m 23s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/3/check/grenade-dsvm-xenial/f831463/ : SUCCESS in 2h 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9b28c2cba4fc29df3140bf926ab04c2d2817059","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-26 19:38:55.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/3 : SUCCESS in 2h 48m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4c66c8cdd404abf869022e4b1a38bfddfbe9d383","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-27 06:52:33.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/3/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/1dbf16c : SUCCESS in 1h 46m 23s","accounts_in_message":[],"_revision_number":3},{"id":"1856bc43ce82a1fdc45963a964ec9b3ea801fc6a","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-27 09:44:23.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"5f7e6b61bf08aefba1dcb411baf10815d2ef16f5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-27 09:52:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2bf811ef656881c1ee8e661a2cec880a70a9e04c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-27 11:03:57.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-35923 : FAILURE in 1h 05m 37s","accounts_in_message":[],"_revision_number":4},{"id":"cb5ab17a29f7076d3212a4fe32508fc08d8a2e90","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-27 12:24:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1e96541169014c7fa8d05e3f51bc323a : SUCCESS in 56m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e7838fff9812468ab4a5406ccf4a0124 : SUCCESS in 1h 20m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f845b48c7e641c19c4619418d101115 : SUCCESS in 16m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/729895cac1b64b2bb842936d0aeaaff7 : SUCCESS in 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c87268c37b4bc3ada4e59c56b5a2f9 : SUCCESS in 11m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/62116d37c392482a9c1f2c28d521d56b : SUCCESS in 14m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f8ff96c7834535a73755801be6ba85 : SUCCESS in 13m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d71906016b77445fa2adcd1512419cae : SUCCESS in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478c9ee703e44e0ca38c451e24ba36bb : SUCCESS in 8m 50s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/562d1f5f92f240bb98bf469ab997ad3b : SUCCESS in 52m 10s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1e5683222c5c459a9b25238eab21eddb : SUCCESS in 1h 24m 57s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d5a18baa88e9432fac357920e68dbced : SUCCESS in 1h 12m 28s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/9bc27b3dea304db4814f9c6170dde0aa : SUCCESS in 1h 16m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/affa3e15958b4d468934225fdf01fdfb : FAILURE in 1h 01m 35s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/22aa926d4ed14e7a9d91093607c99d87 : SUCCESS in 54m 44s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e28ca61c91204a4f91200a3aa4a8ccc0 : SUCCESS in 1h 23m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/0a3e93ef6b0941b89730addb7fdbab19 : SUCCESS in 1h 39m 15s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/59a5b2b23ffc453785021de08910fcb6 : SUCCESS in 24m 45s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8fe0b052b46b46c0ae26042c2f5a03e7 : SUCCESS in 22m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/95c69c4ce52748ceb28afa6d54c006e9 : SUCCESS in 2h 36m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6c0c195aa24419f9861416e8996174e : SUCCESS in 1h 05m 33s","accounts_in_message":[],"_revision_number":4},{"id":"ee5bfcc2470e25fabb3e6dadf70b9f2e4495090e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-27 12:25:58.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/4/check/tempest-dsvm-full-xenial/0462d4a/ : SUCCESS in 2h 11m 18s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/4/check/tempest-dsvm-full-xenial-py3/4bc12c9/ : SUCCESS in 2h 14m 07s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/4/check/grenade-dsvm-xenial/993f86e/ : SUCCESS in 2h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eee6ea7c0696be88c372dc5e104c3393fd56d068","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-27 12:31:09.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/4 : FAILURE in 2h 35m 34s","accounts_in_message":[],"_revision_number":4},{"id":"faac589a5300593cc3110100ea8cb287836bf510","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 01:35:23.000000000","message":"Patch Set 4:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/4/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/060fd24 : SUCCESS in 1h 55m 44s","accounts_in_message":[],"_revision_number":4},{"id":"39fee06a704e95e4ec75b9c4137f4a0311a00a97","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-28 05:45:09.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"049e2718f6ca59749dc157c6a4c90ae0ec2fe535","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 05:51:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"05d1b06942becb30ab83829bc505c71670910e37","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-08-28 05:57:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a57dc2d91cbad4a4ff33a795d62d65c60d6a73a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 10:15:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/592d28179caf4c209e2d4a51ac3f83b1 : SUCCESS in 1h 05m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7d2ba19b9ad74a7eaacd536c73674b9d : SUCCESS in 1h 27m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a3cc7c094874e67b1a41e0e8e5d077c : SUCCESS in 20m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f86405d064a740a7acf6ff9b0e8a746f : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d7860a0b514835b0f947b99409f3f7 : SUCCESS in 11m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ad6c13c8132245a08f858099c032c2ef : SUCCESS in 16m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a49025f87bb45fdb603c49dc7aa08ad : SUCCESS in 18m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/322b4e5dd272417fbc8f3c81093e03bd : SUCCESS in 16m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a38a9fb6531745eea6b1d6bd969117df : SUCCESS in 10m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/69048fd826dc4014aa454b43e4dd3824 : SUCCESS in 55m 42s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d4b97878243a4bd693a1feeeab676ff6 : SUCCESS in 1h 20m 19s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d9e5c91b086410e85d3d6e622fde676 : SUCCESS in 1h 14m 53s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/f19d94d6f04d48b491719127ae46c51e : SUCCESS in 1h 24m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bef8eeef3da84dfca831683dbcb6b7b9 : FAILURE in 1h 01m 54s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d112c2fe7dad47729f0f9abc879ca2fe : SUCCESS in 1h 08m 28s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2ed391c3dc96499d806ee45ba88d8198 : SUCCESS in 1h 38m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/61f5c14cb17c43cc91108d66041a6045 : SUCCESS in 1h 30m 28s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/c67305760cda4e7fa1e67f5e20b5ffe1 : SUCCESS in 18m 54s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/550e9d9cc6f9482bab8b77d83ea47133 : SUCCESS in 18m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a6d7c2cd9ba0473d8d33335cce20fe81 : SUCCESS in 2h 21m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d40c3ca2805b4058afe01ae54ecb0e7f : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":5},{"id":"36112e6e739be57347b1bf5c58a5197d749f5011","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-28 10:32:28.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36049 : FAILURE in 2h 05m 47s","accounts_in_message":[],"_revision_number":5},{"id":"59f5f54e26b036c6dc116a45d1103964b695a448","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-28 10:45:47.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a776fbe64a6ec0f2423ba20f0a58906545c690c3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-28 10:49:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a01f0fb16a97b430368a18f6ae5feb50f735405a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-28 12:50:46.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36072 : FAILURE in 1h 09m 42s","accounts_in_message":[],"_revision_number":6},{"id":"800969ee155e4f05b4b8553590ab3727ce93db2a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-28 14:18:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/65850b1a1ac2441baa1ec2a233104977 : SUCCESS in 1h 02m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fbd0b05ebac34960b0ae90e43732bef2 : SUCCESS in 1h 20m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5d3ec367454bc5a94c0ac1aee9dae8 : SUCCESS in 19m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/022a86aaaaec4091b443006d43d5a20b : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82a1471529fa422aa584b4e101a464c0 : SUCCESS in 11m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e52eb803b490414491c2833ba5069f78 : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc7b4963c9f245629b95876434be113e : SUCCESS in 14m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6cfa2945299e43078e5d9f47d55be4b7 : SUCCESS in 13m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/467415ea0a634e2eba86cc785640e920 : SUCCESS in 7m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4eb159221a1b478fa6914b4ec5d67978 : SUCCESS in 1h 11m 54s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/cc9b1c6d86d44123b225e77e2f92a1c9 : SUCCESS in 1h 19m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/78bf0123d3c94520835f7c1494b9afdc : SUCCESS in 1h 01m 15s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/1122d8717e764687b17da5a81e4a5f3d : SUCCESS in 1h 08m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/38b6eca0265b4e5d89fe4b6cf66a3670 : FAILURE in 1h 05m 38s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a5f65ad212224d86a51470c31dee160c : SUCCESS in 48m 16s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/75f9c1de40df47b4b470ee2221582e14 : SUCCESS in 1h 36m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/a6ea2bd880314b1ea4c91d36e65285d8 : SUCCESS in 1h 24m 45s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/cf73260e47a9474ebcfb5b4c90afc2f0 : SUCCESS in 22m 19s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7e5502a8b09a45a590d7c86f4a43c90d : SUCCESS in 17m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b93ce5853464fb6a4843055b4829986 : SUCCESS in 2h 22m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bf65ad214ed44609ddf4ee665ec3952 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":6},{"id":"3eba9ba95288ce94b32746c7c3c7583012aac21c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-28 23:53:06.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/6 : FAILURE in 2h 34m 37s","accounts_in_message":[],"_revision_number":6},{"id":"3dcdab060c30ff2625c5928881d10369a16f4587","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-29 03:21:36.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"1d39a6c86ec38b09c2b1c685aa9d166dc464be0b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-29 03:25:04.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"54312c8fcf429634e054d3ac6d689f3a3ed99df1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-29 05:01:05.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36118 : FAILURE in 1h 26m 47s","accounts_in_message":[],"_revision_number":7},{"id":"e8f09ab3a1743dbeae6e25e5988db69ed6f9f98a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-29 05:51:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d13e72d0f2ac497ea6845e95b4271362 : SUCCESS in 1h 07m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fc2f8e5b833b4e318592beea7c5bfde7 : SUCCESS in 1h 27m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a3f26f994754064a9f6ea9e72aab449 : SUCCESS in 18m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b696c7fc36d41a1a41c80facba83bed : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7df9fb670666429aa03108693cac9aec : SUCCESS in 12m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f9ac9ac64da4edeb88ef286137c6414 : SUCCESS in 15m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c801718de7fa43478ec8fa5e9c4dc1e3 : SUCCESS in 13m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e657f56b6fed429cb02e659a5ae67788 : SUCCESS in 15m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a0deb9d5c4e4a41bf61804d02e9d71e : SUCCESS in 7m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1ba13c446a8d4d0bb5c7ae81f4b4d2d6 : SUCCESS in 45m 58s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/30ce0794d41a4cc19029f597fe781a51 : SUCCESS in 1h 13m 25s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/30b08eaaf4004d9da66675a6b6ca7e46 : SUCCESS in 1h 09m 23s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/6442dd5c9fff4daf9ecee23614c0f3bb : SUCCESS in 1h 11m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/db013207afb1474e9faeec21395ee849 : FAILURE in 1h 05m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d79bcb549e1a45ba996454e8307c11d5 : SUCCESS in 1h 02m 08s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a9f3290183de427e97c18f684809e85e : SUCCESS in 1h 33m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/85fccae1ab5941469481520c9219f85c : SUCCESS in 1h 07m 58s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/42ef47d6746444d3aa15a7391d43c10c : SUCCESS in 20m 26s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/11c4aa87f6ca45e6b4d2186ab02aec4f : SUCCESS in 19m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54ec10b6ac254cb6954c977a3f0d9e09 : SUCCESS in 2h 25m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e4c122e8cca429bb780506c790e33a8 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":7},{"id":"30e0d2f862248a60b417a7c415e241d469103b9d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-29 10:10:12.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/7 : SUCCESS in 2h 35m 58s","accounts_in_message":[],"_revision_number":7},{"id":"6c79a697d11cfd52813ea03244655b2944915bcb","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-30 09:22:49.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"69d9ab71a49e5acc59034ae75fd18922fe3be9d3","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-30 09:26:04.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"cf509b6613b4c7e1818f6f46ecae913ae21df635","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 09:43:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"db3ec33e099f77c0ad769d78520ab6ad73548d22","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-08-30 10:53:24.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"833451a0eb54e306effd2205512766caa4f919ca","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 10:59:41.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"eb0adfc5c6c88571071452463dd7d8910a47ffb6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-08-30 11:20:17.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/10 : FAILURE in 17m 46s","accounts_in_message":[],"_revision_number":10},{"id":"4d4c941d2459ce9f3f091fc0a857e8e77e637130","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-08-30 13:31:36.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36237 : FAILURE in 1h 22m 41s","accounts_in_message":[],"_revision_number":10},{"id":"546b81301b0b1dacd3f00e612144eb96523c8496","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-30 14:07:25.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/992edbf5d29249dbbe7c5f3e09c2232f : SUCCESS in 1h 02m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/75c8eae5a06f42a2b147ff5a2f13270b : SUCCESS in 1h 37m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/275b8f9ea4964d94843ad754fe1e81b8 : SUCCESS in 20m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1ed15ae565c4bf2b8051d1268bc8afd : SUCCESS in 19m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69be39c423024e70aa90665a15241ca9 : SUCCESS in 9m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/321ab31180e743f88f6111558722a492 : SUCCESS in 15m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/064aee58d1994b3ea8a1a96aec68c021 : SUCCESS in 15m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dad03c80168740e38e54f4bd858934c8 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f194f6add6994262b240a5d0914d1891 : SUCCESS in 9m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fce7134edabb4a74a80d7cfe94edfb5c : FAILURE in 21m 30s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/ce5961e5af604425bc86649cf5bdfdd8 : SUCCESS in 1h 24m 21s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/daecb68ef4b5449a84437622d23c164b : SUCCESS in 1h 08m 53s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/f3d12051c3b042dc85c502604cd11e87 : SUCCESS in 1h 10m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cb11db94917c493e93c73fb4892eebcd : SUCCESS in 52m 40s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/72e01f532d554550bade2b525ec0e7e1 : SUCCESS in 58m 19s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ba8238eb72ba4409927856682d74ceb9 : SUCCESS in 1h 42m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/49ffe0f31feb49d59ec8d87133376ab3 : SUCCESS in 1h 21m 47s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/627bbf3e4805496fa30f2957347718c7 : SUCCESS in 22m 10s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/15946d555b534cc48d44e7db2f96d42d : SUCCESS in 20m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/71fd3b22f3914cb5a89feb9068825d08 : SUCCESS in 2h 34m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da60969f51c04bba8b64c3f3bcbf2f01 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":10},{"id":"cdd55d7cd42f984d2dc26ad49d80c459cb695032","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-08-30 14:08:31.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/10/check/EMC_VxFlexOS_NOVA/de76f86/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"a75acfa88cb59a8ffb8522f0fd043c7b393308c7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-08-30 14:38:19.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/10/check/tempest-dsvm-full-xenial/f7189b8/ : SUCCESS in 1h 35m 31s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/10/check/tempest-dsvm-full-xenial-py3/72299b8/ : SUCCESS in 1h 42m 34s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/10/check/grenade-dsvm-xenial/cd4c85f/ : FAILURE in 29m 56s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fae7be72410a0d127886d5e1da8db687cb5fd97d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-08-30 16:18:33.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-10 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":10},{"id":"c6d42b760e932513fae690f3c97e2faa6f8b25c7","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-30 19:17:20.000000000","message":"Patch Set 10: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":10},{"id":"8cb3848b533fc9fc56b5154a74f1d12b2f29d4e3","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-30 20:44:56.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"85f9d4ac9ca464ab5e8e921569f9d3c73f17c44e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-08-30 21:08:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9b17df9c23b9cf40494bc008911c8bb8c9a0f8f3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-09-01 02:26:10.000000000","message":"Patch Set 10:\n\nTesting failed ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenial-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/10/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/559cbbe : FAILURE in 42m 34s","accounts_in_message":[],"_revision_number":10},{"id":"cf98778ddead05146e559575187c1ca11f43a79c","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-02 07:10:24.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9020b7d5fa2a59be1580a9b87f33e47f1bf448ef","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-02 07:15:43.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"928ebb515e27c73d7b08e0b2d0fc88e61e97a575","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 09:33:03.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/22c4c382848a48b18c7156efc1669f39 : SUCCESS in 1h 04m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0fdf33de77e347b4b2c613a4036f1a6d : SUCCESS in 1h 29m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c425404ceb164e7f82c86a76db59da56 : SUCCESS in 18m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f1811707ac04381a6f78a866dc03e7d : SUCCESS in 17m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6cabf99c1ba4ba48906d7759ff4a8b5 : SUCCESS in 9m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df9e44a774eb4695b164be7115a7133c : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f19004540ccb4e0b962d7e78bd0870f6 : SUCCESS in 13m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af6a333d79bb42fba293e0470a8168aa : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/345ccc30f6b14f2c937a91a655b44195 : SUCCESS in 8m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e4a5a577c157485cb912f4bc0247cd40 : SUCCESS in 56m 13s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/629cd8fffa9c443193981aee8e73c455 : SUCCESS in 1h 10m 05s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fed9fa39c9e8414491b2768c5c19a72e : SUCCESS in 1h 02m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/47f5f2dae4654ca7aeef30e66a948493 : SUCCESS in 1h 16m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/66f7e0a0d2dd47119bc93af8668f9460 : SUCCESS in 49m 21s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/120b834bc63b4dff8987e52f58e3daee : SUCCESS in 48m 23s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3ec20934d2f54963b9846035e72c25e7 : SUCCESS in 1h 36m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/5c5c41127032431ca50392810046b22e : SUCCESS in 1h 28m 15s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/d59dd9770e0d4af2bc19743caf75f636 : SUCCESS in 19m 01s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7fc6f09334854c95b813eda1aecb15c7 : SUCCESS in 17m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ab6ff2e0a23b4bddbe7ec42278be3a55 : SUCCESS in 2h 15m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae881272f1ad4c42934acb9076b5f2b0 : SUCCESS in 57m 16s","accounts_in_message":[],"_revision_number":11},{"id":"0462637e8cd3b8b61196b3481e0408d06ef013ba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-02 09:34:15.000000000","message":"Patch Set 11:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/11/check/EMC_VxFlexOS_NOVA/5013091/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":11},{"id":"fffc8358c916ddf3a6cbf22df3796152112bdfce","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-02 09:41:49.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36339 : ABORTED in 2h 19m 48s","accounts_in_message":[],"_revision_number":11},{"id":"85a37bc1d31a3b712a211aaf30628d4c0d64640c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-02 10:05:38.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/11 : SUCCESS in 2h 47m 30s","accounts_in_message":[],"_revision_number":11},{"id":"875e1baf0af39a4904bbc4f48b82213d98b3e22c","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-02 10:38:11.000000000","message":"Patch Set 11:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-11 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":11},{"id":"72e6bf3f0aeaee4c59b130fa3b7fa95e3567b0f1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-02 12:11:52.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/53/678453/11/check-nova/Nova-ML2-Sriov/ea08e0a : FAILURE in 25m 18s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/53/678453/11/check-nova/Nova-MACVTAP-ML2-Sriov/fc5437a : FAILURE in 16m 10s (non-voting)\n- NVMe http://13.74.249.42/53/678453/11/check-nova/NVMe/8ffa346 : SUCCESS in 1h 09m 18s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":11},{"id":"fd6ad9b09117ac14443bf866a03b383597d1dc1d","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-02 13:00:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"197b7e46752fc0c94fc8a37cc6aadfffe0dea945","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-02 13:01:40.000000000","message":"Patch Set 11:\n\n(11 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8fe62878f7c7ce1ed9bcaa55fbf80608caee1de7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-02 13:06:20.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"cddecd028466831bd7846a090a81819a79a67875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-02 15:38:26.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/49f8c19a22d44923b2a66ae5eb477a69 : SUCCESS in 1h 09m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c8def58cff4146edbcb44b304f736ce2 : SUCCESS in 1h 35m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02234ae9b0ec486e83286c5e9c1d4875 : SUCCESS in 17m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb182d86225249749aedf8ab896c6ea8 : SUCCESS in 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a708937119b6483e8c3169bbc92ce38b : SUCCESS in 10m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/28c17a5f643646898cd337f9f47fdd2a : SUCCESS in 16m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4774b17ab2f4374a462a6548675d418 : SUCCESS in 14m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4eba9489e554988947c598d484b24cf : SUCCESS in 13m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52fce37366d842759cf131ac57d11add : SUCCESS in 8m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5c636196bbba46e7bd124d07138d5d3d : SUCCESS in 59m 23s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0abbf9f11d7347798b9b76155fbb9555 : SUCCESS in 1h 12m 34s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7d8ad35b2694039ab4bdc9f09e6bc74 : SUCCESS in 1h 23m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/855358db536647b5b562048d82e55ba7 : SUCCESS in 1h 18m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b4153f015f9f41a5bef3be3671da6464 : SUCCESS in 56m 41s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2fe30b7320be424890a356cf60b4322a : SUCCESS in 1h 05m 49s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6b1056e66e694e82a28556b1d3d1948e : SUCCESS in 1h 33m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/a147879907d743b789e0ddfa5634a8b8 : SUCCESS in 1h 24m 20s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/1f7b313d62374ceda1dd7d9739333de7 : SUCCESS in 21m 32s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1aec14b348634d589b98ec5fc514b05a : SUCCESS in 22m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4e6a7d3d04a444a18ac93473d28f5391 : SUCCESS in 2h 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38e989581d2c466d83cd952559be5015 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":12},{"id":"2d6eff21df1fc7fbfd2d8c9d7a9f3b40b0da02d1","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-02 15:39:47.000000000","message":"Patch Set 12:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/12/check/EMC_VxFlexOS_NOVA/3963dab/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":12},{"id":"5d1c307625f53f39ba6941116b2fbf000dc97fdf","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-02 16:38:15.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36373 : FAILURE in 2h 39m 04s","accounts_in_message":[],"_revision_number":12},{"id":"3e0afc6e49d7985a2e61921361fd576038c2ff54","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-02 18:22:02.000000000","message":"Patch Set 12:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/12 : SUCCESS in 2h 23m 38s","accounts_in_message":[],"_revision_number":12},{"id":"9f65afe7de1b45b1d432d1f30e6a25383709e5b2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-02 21:24:27.000000000","message":"Patch Set 12:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-12 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":12},{"id":"c644aca68992662e9649ba848b19ecdb47af2cd6","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-03 06:21:41.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"d0fdd8f34b3e2092a86fae03a74c5f380485822c","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 06:27:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"fe2c365411fa2f4674ec4cf574bee4bc967f4931","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-03 06:32:05.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- ext-nova-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":13},{"id":"93552bd42c9a7aec0e01ef5de288164b2a73d348","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-03 07:44:58.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36427 : FAILURE in 1h 09m 20s","accounts_in_message":[],"_revision_number":13},{"id":"b61afbcc5fc7cf66d7e86a8944245556b67aff60","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-03 09:09:30.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/13 : SUCCESS in 2h 39m 44s","accounts_in_message":[],"_revision_number":13},{"id":"2e5cc7b698863c9c6b5dd30046529f5a45106616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-03 09:36:44.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2b75cb19e1ed4579b1b9b53cf124d0f9 : SUCCESS in 1h 01m 03s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/aeb17ef1abd5458080ce1b576eba07da : SUCCESS in 1h 31m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72b2914f7d35451397e16d846a987285 : SUCCESS in 19m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c207aa6b3e8549c29b4171b9bcb07b19 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0fedcb923040e9b55b6340c4ca95a7 : SUCCESS in 11m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1de69949168452dad3a5030caf989c7 : SUCCESS in 17m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d5a20dbbbf34636bb38bb41e111c3b8 : SUCCESS in 12m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/156beb0252d345c1a071f99303b19f2e : SUCCESS in 15m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9dcbb8f46794f9997b69761b00a8769 : SUCCESS in 6m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bbe0ab195d884486af8a018c19c457a9 : SUCCESS in 1h 03m 28s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f46b6606291e4baea8057771a706b808 : FAILURE in 1h 18m 57s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5a82e0d80c344c289a6b0c3eb928818 : SUCCESS in 1h 18m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/838216077d544ad49237c93de1cad573 : SUCCESS in 1h 19m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/10c10c73dde8484cb8721d28325d8751 : SUCCESS in 44m 43s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/0dd03d78922a443684e1dea33b12d1aa : SUCCESS in 51m 09s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6737c1df92784db88463c3ff930e9c45 : SUCCESS in 1h 41m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/123b047412cc4cf3ab31ac0c0d6d4e40 : SUCCESS in 1h 32m 27s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/7fe4e0d2f7044f79847a4912287525b7 : SUCCESS in 19m 33s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9a18ea99fd544b1492fdfb7b140f1112 : SUCCESS in 19m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/baec9a149d434a8caffb8a82540f7b02 : SUCCESS in 2h 09m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d3a40dc64d14b2aaab9b383b73d5623 : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":13},{"id":"bd79943722315f1273b17f0a584a4dd72b5f374f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-03 09:37:55.000000000","message":"Patch Set 13:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/13/check/EMC_VxFlexOS_NOVA/020d7c6/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":13},{"id":"eb38fb20b26cd3589321fd2c82a1863ce558f32f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-03 10:51:24.000000000","message":"Patch Set 13:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-13 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":13},{"id":"287829176674ba80a17b4f1f26f8fb4e615564c9","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-03 13:58:33.000000000","message":"Patch Set 13:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/53/678453/13/check-nova/Nova-ML2-Sriov/5aaee3b : SUCCESS in 1h 22m 51s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/53/678453/13/check-nova/Nova-MACVTAP-ML2-Sriov/57f723b : FAILURE in 2h 01m 07s (non-voting)\n- NVMe http://13.74.249.42/53/678453/13/check-nova/NVMe/bd49992 : SUCCESS in 49m 11s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":13},{"id":"341fb0e663e0b38c8562d41e76089e68c00c8890","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 01:58:04.000000000","message":"Patch Set 13:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/13/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/5e44d5f : SUCCESS in 1h 54m 38s","accounts_in_message":[],"_revision_number":13},{"id":"997ca41a81688c4c54822d25f80fe38344b7f69e","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-04 12:49:50.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"94902ddea97a4e940c3d0535a04cb9d0296e70d3","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 12:55:52.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"363e8f8e37979367c5f47b940bd247bfb1bc4ffd","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-04 14:26:25.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36520 : FAILURE in 1h 22m 05s","accounts_in_message":[],"_revision_number":14},{"id":"f68a8778d546fd361cd3082f59776644170c0f49","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-04 14:58:12.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"c5b4290e8861b2c7d531eb56972ccbb39c5064d7","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-04 15:04:19.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"eff683ebc3ce975f16be017fda0017200bac95ff","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 15:04:33.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"de1b7a97b5ef104fcae3f91239582fae18807ae8","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-04 15:12:21.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"cbd306b0f6cd4928f24f71e2cd8df5e9ec7476ad","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 15:39:55.000000000","message":"Patch Set 17:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"9c7cadeb6c793303794a7318acdc291675276bba","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 17:23:04.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/17/check/check-dsvm-tempest-vz7-exe-minimal/a27f795 : FAILURE in 11m 23s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/17/check/check-dsvm-tempest-vz7-exe-multinode/02d8b5b : FAILURE in 11m 19s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":17},{"id":"66ed71a313898cd76c8b7eae90ca7aadb45d7fa1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-04 18:04:00.000000000","message":"Patch Set 17:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36534 : SUCCESS in 2h 01m 03s","accounts_in_message":[],"_revision_number":17},{"id":"5bbeea0767d10ad1bbadb2f16274d42dfc016fd6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-04 19:02:03.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/17 : SUCCESS in 3h 00m 46s","accounts_in_message":[],"_revision_number":17},{"id":"a024323c4687957668d7b3973721ce9dd1fd0741","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-04 21:21:46.000000000","message":"Patch Set 17:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/17/check/tempest-dsvm-full-xenial/2284f39/ : SUCCESS in 2h 09m 49s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/17/check/tempest-dsvm-full-xenial-py3/f63c2f6/ : SUCCESS in 2h 11m 34s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/17/check/grenade-dsvm-xenial/5ac7db0/ : SUCCESS in 2h 34m 37s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"13fdd8fd3d8c4de5e280ca2fd5007cdaadda4927","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-04 23:47:50.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1789c96b66d14570b390bea4b1d5067f : SUCCESS in 1h 06m 06s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c897937a2f8f407c90af22b7ae04646b : SUCCESS in 1h 26m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0ba1d714c874164bc6c9bb27cfc2fbf : SUCCESS in 21m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d1a5db74fb843c6859df2522793fc20 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd4eb9229b9d438683bae4701639efde : SUCCESS in 13m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e70b78366434f61935966a275d5d7f3 : SUCCESS in 16m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9039cbfe75b440fbb071bdacd22267e0 : SUCCESS in 13m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0b3617893c7149b7acde88585261bc72 : SUCCESS in 14m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b81f34a937c430f935622316ab36cf7 : SUCCESS in 11m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ba7150ee05524d6fba2c4090399dcf06 : SUCCESS in 55m 54s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a2486761a6d5416982ea45d7672e0b35 : SUCCESS in 1h 12m 06s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/74e1ecfa10e8448e95aad46406b707cc : SUCCESS in 1h 17m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae9a3f6d650643c180aedc45728c33c2 : SUCCESS in 1h 11m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4c9e0592104d478ebda97d01c85b30a9 : SUCCESS in 1h 01m 50s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d093c30c674847ef928baa8a3286ef54 : SUCCESS in 56m 57s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/536e0a6fb09e4cae90a095100f5a7b60 : SUCCESS in 1h 31m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/0860977f2f42420a85f597dda9bceab4 : SUCCESS in 1h 24m 41s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/7d9c818d2e594894a9cd4bdf129895bd : SUCCESS in 21m 06s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97a99becb1f34a60b09a12a554215311 : SUCCESS in 20m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed8977fc8bf143b982a93c0f08b5570b : SUCCESS in 2h 42m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffb3f69dc66f46a7a54851e4e417dfe1 : SUCCESS in 1h 06m 56s","accounts_in_message":[],"_revision_number":17},{"id":"4c67df680cc43d9cb10d594131f37285f35f5560","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-04 23:49:00.000000000","message":"Patch Set 17:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/17/check/EMC_VxFlexOS_NOVA/f4bee38/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":17},{"id":"87221f221b650d37718343db145944c5190d1113","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-05 02:10:13.000000000","message":"Patch Set 17:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-17 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":17},{"id":"07a9b13e64cdff62daefbefad7edc17846beb621","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-09-05 05:55:32.000000000","message":"Patch Set 17: Code-Review+1","accounts_in_message":[],"_revision_number":17},{"id":"1bd5b75eafd29956acdf5738c5ddcbe3c6afe87a","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-09-05 06:49:00.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d8aebec1dd5d7ce3ea908c924b09f188fc04922b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-05 09:10:04.000000000","message":"Patch Set 17:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/17/check-vote/ext-nova-zuul/f50030a : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":17},{"id":"248005b801aa1cca50cba3a6edb83b0c7261f247","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-05 10:29:58.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"7d328ad0f26cc17dacb4deffbe3b633c906bfd3e","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-05 10:35:16.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"fda373190b17e9abe67688bcee022269efbe6799","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 10:35:32.000000000","message":"Patch Set 18:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"a01ef55d0c40d2da160f05e53364b7aa96b845d8","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-05 11:39:10.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"ae1e9c186e079f913f5e7060596698d12f2aa9ed","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 11:44:59.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5c95346a661aa591de4a1c314cc081ba472c4664","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2019-09-05 13:17:31.000000000","message":"Patch Set 19: Code-Review+1","accounts_in_message":[],"_revision_number":19},{"id":"15fa00f999006df6ca8d4cf77f89220bdd810cea","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-05 13:31:23.000000000","message":"Patch Set 19: Code-Review+2\n\n(2 comments)\n\nThis looks good, thanks. A couple nits for a followup.","accounts_in_message":[],"_revision_number":19},{"id":"8db55ac688e207cc9234a2ba710752d8fbd569b7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-05 13:52:06.000000000","message":"Patch Set 19:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36651 : FAILURE in 1h 20m 02s","accounts_in_message":[],"_revision_number":19},{"id":"e6080ff31d841206cb65591f07e32a66b8578739","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-05 21:04:52.000000000","message":"Patch Set 19:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/19/check/check-dsvm-tempest-vz7-exe-minimal/55e268f : FAILURE in 17m 42s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/19/check/check-dsvm-tempest-vz7-exe-multinode/f8f24e5 : FAILURE in 12m 36s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":19},{"id":"d847dce4b4d47e65ffbfe727d491abaac399067e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-06 00:15:13.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/19 : SUCCESS in 2h 32m 07s","accounts_in_message":[],"_revision_number":19},{"id":"888ff389ab9a070c2d91b35df392870beab55ab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 06:22:21.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c29bb5eea2b44ecab4c39c3b6497b8e1 : SUCCESS in 57m 57s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c176b1503b454d78978e0bb3eb1a1e52 : SUCCESS in 1h 26m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc0a43ae527c48a588576bdfd499eeae : SUCCESS in 18m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cf81a1b15c2414797541a065c430d37 : SUCCESS in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22955c16f1cf42448b168bb44352f243 : SUCCESS in 12m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b1ed790da689466ab9c6a71300db74d9 : SUCCESS in 16m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7e254a1ba734bcf9e5402daf97f3e8d : SUCCESS in 11m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ade755811384552a5b0efc0713f5ca0 : SUCCESS in 11m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e78d63b3827e47f89ded4eeea1875ac6 : SUCCESS in 7m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8dc01af8c1ea40b7990c355e4488b1d8 : SUCCESS in 50m 06s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/67bcc92d674a42c299595a433ee08807 : SUCCESS in 1h 11m 51s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1fea14a9b2af41d48e0808f27bffde89 : SUCCESS in 1h 13m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/24ffb859a33c46348d4c13dfbea2b688 : SUCCESS in 1h 14m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a9a724639d3d4cafa2847df5eb617d91 : SUCCESS in 58m 01s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/3b28e8fa67ef40c8a6f7908026859146 : SUCCESS in 54m 07s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e69bfce34ae44106bd444a5a4f0a72dd : SUCCESS in 1h 34m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/5a9dabc1e18c496596f38ddc88f12d1e : SUCCESS in 1h 40m 56s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/089b649930d84b3792bddc21ecdd0bc0 : SUCCESS in 21m 52s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a46a9ba72d08400988927d07b46edb4c : SUCCESS in 19m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ffe9c8946cb64d6d808f56432d514624 : SUCCESS in 2h 34m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd9676d11c814c64a1a279916667cb9a : SUCCESS in 1h 06m 53s","accounts_in_message":[],"_revision_number":19},{"id":"6f86f9e6bc6989cca13d003592028a1f2d47d7ba","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 06:23:36.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/19/check/EMC_VxFlexOS_NOVA/6b00a3f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":19},{"id":"a66a063e1abaf6caceb834e74625d1176ddfa12b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-06 09:40:24.000000000","message":"Patch Set 19:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-19 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":19},{"id":"80680ede274c2f0e6e97efd5d8527a2721afc8ff","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-06 10:17:19.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"d7cc9176844603429c16cc46c84ca14281d6e477","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 10:22:56.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"2e1d6f7e93771913e071979508c881a2e61c6101","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-06 10:40:24.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"1840b49150a5ddac5960fc64ab58cf8555dbc259","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-06 10:44:33.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"35a0f685f9663e1d391b30bdc3b737648b7dc758","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 10:49:40.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"4117eec24ef0ed514edfa3be6c596dbb02fad9b7","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 12:21:54.000000000","message":"Patch Set 21:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/21/check/check-dsvm-tempest-vz7-exe-minimal/689c40a : FAILURE in 13m 43s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/21/check/check-dsvm-tempest-vz7-exe-multinode/7b72184 : FAILURE in 13m 48s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":21},{"id":"deed0534ca5bd90f328182eb679f5c4c8dd809c4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-06 13:42:12.000000000","message":"Patch Set 21:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36737 : FAILURE in 1h 06m 51s","accounts_in_message":[],"_revision_number":21},{"id":"f34933dd3c949ab8ab90dd69f57af0d20bcf97b2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-06 15:01:45.000000000","message":"Patch Set 21:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/21/check/tempest-dsvm-full-xenial/ee5ced8/ : SUCCESS in 2h 06m 48s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/21/check/tempest-dsvm-full-xenial-py3/1c22288/ : SUCCESS in 1h 31m 34s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/21/check/grenade-dsvm-xenial/4352aa2/ : SUCCESS in 2h 16m 47s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"b6213db7f182195e4c7e6923e523523c310bc640","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-06 19:45:06.000000000","message":"Patch Set 21:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/21 : SUCCESS in 2h 44m 33s","accounts_in_message":[],"_revision_number":21},{"id":"01733491d02aa951c103163caf4e9b2f1c3316cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-06 22:04:04.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6f6d40f8368c4bc1b583e08c19f6862e : SUCCESS in 1h 03m 46s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4027b33ccc6e4cb48740a4403500c704 : SUCCESS in 1h 32m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4376527cb6244d2eab7e469e38a2f055 : SUCCESS in 16m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d958e39e2ea45a29a78223053db8108 : SUCCESS in 14m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f21489af47374f4488c22698ab8f0024 : SUCCESS in 9m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/137660dd640d416d957c9cbe1d13e3e6 : SUCCESS in 12m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81c330a056454416af5c858d8e0e9a1e : SUCCESS in 15m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5e088d331c814be0ac7df8c245ebc29c : SUCCESS in 13m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b07bba9322442df8f26105ec53a8049 : SUCCESS in 9m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/637aa65ff2a640df9c2689c8c3c39a2a : SUCCESS in 56m 33s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2a70d873951d48b8b583f40b0e51683a : SUCCESS in 1h 16m 55s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2ca043a33504eb3b5a01766ef4d8494 : SUCCESS in 1h 16m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8516d70cc834a46b987b4e1931e6b17 : SUCCESS in 1h 00m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b1f5250e15ac44a0a7ebb719ee891db8 : SUCCESS in 1h 01m 43s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/7741b72e8cd346b9a96753ab930357cc : SUCCESS in 50m 47s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a20cb67a88064d938a7bdbedd25a3bd9 : SUCCESS in 1h 41m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5a2295425e341deb6aa32ca668ff726 : SUCCESS in 1h 33m 01s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/e7ca8d2ff7f048e0842496e3260b3e96 : SUCCESS in 23m 08s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cae4e1da130f45f8aa183ceda0098ea2 : SUCCESS in 16m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54237438f01b4c4f89ab4a09b975b71d : SUCCESS in 2h 27m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/078fb60936344a3fb922326750ee5796 : SUCCESS in 1h 10m 36s","accounts_in_message":[],"_revision_number":21},{"id":"bd4de00e22b4eb049f03d4e2b7acfd0f9ca9210a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-06 22:05:18.000000000","message":"Patch Set 21:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/21/check/EMC_VxFlexOS_NOVA/56ac665/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":21},{"id":"828f8890d013c9dcbad6ee84f2942f10a11cba3b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-06 23:00:04.000000000","message":"Patch Set 21: Code-Review-1\n\n(2 comments)\n\nSoft -1 to add some more unit tests.","accounts_in_message":[],"_revision_number":21},{"id":"6d36c36237883b40d5e5c5735416827ceac8c685","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-07 02:37:02.000000000","message":"Patch Set 21:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/21/check-vote/ext-nova-zuul/1b1705d : SUCCESS in 1h 14m 50s","accounts_in_message":[],"_revision_number":21},{"id":"2ca4f4273f7634be8cae3da2da1ea86f03495ae2","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-07 03:05:59.000000000","message":"Patch Set 21:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-21 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":21},{"id":"68db4722f164f7126a1d5e57e3f7c84093b442da","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-09-08 12:48:34.000000000","message":"Patch Set 21:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/21/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/48cdfb3 : SUCCESS in 1h 53m 18s","accounts_in_message":[],"_revision_number":21},{"id":"0582085630a7bb90f061ad423de25608f41b03ba","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-09 05:45:02.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"f81c22c4066c485f91b85e569211aba34fefc72b","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-09 05:46:07.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"d04530da930bdd0002fadb70d4a038d84813404e","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-09 05:50:14.000000000","message":"Patch Set 22:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"2aeb5fccc1229af8c819f9a7d38aa36cba3072a3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-09 06:57:32.000000000","message":"Patch Set 22:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/22/check-vote/ext-nova-zuul/75543c0 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":22},{"id":"d0276074b7552b6fbb241aa480f70bfe954067e2","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-09 07:08:22.000000000","message":"Patch Set 22:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36850 : FAILURE in 1h 10m 01s","accounts_in_message":[],"_revision_number":22},{"id":"8a479595fe30066023e9a0a0b947887a7185e128","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-09 07:48:51.000000000","message":"Patch Set 22:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/22/check/check-dsvm-tempest-vz7-exe-minimal/73f2e0d : FAILURE in 18m 08s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/22/check/check-dsvm-tempest-vz7-exe-multinode/da9f625 : FAILURE in 13m 56s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":22},{"id":"f39d770dd4d5f6492a63ac8836e3fb70625f8dc9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-09 08:11:53.000000000","message":"Patch Set 22:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/22/check/tempest-dsvm-full-xenial/6321491/ : SUCCESS in 1h 43m 36s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/22/check/tempest-dsvm-full-xenial-py3/bb38666/ : SUCCESS in 1h 57m 03s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/22/check/grenade-dsvm-xenial/e23dc48/ : SUCCESS in 1h 45m 22s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"22cc7cf6ef9db35f28b316ec4a04db08c4838067","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-09 08:24:48.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/22 : SUCCESS in 2h 29m 54s","accounts_in_message":[],"_revision_number":22},{"id":"ea7f278d46e2fa772fee4ebdaf3c3dd7852ea99d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-09 08:30:59.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/78006f5f85bb4c5dbfec8e94077bca77 : SUCCESS in 1h 04m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a10ff4e0ba3f44399d882d7be77b9dde : SUCCESS in 1h 32m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bde3c7be78c415a8d3956c94f5962a1 : SUCCESS in 15m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d67d3e228b54c2cbe60a914870bd228 : SUCCESS in 14m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b404fdb9920b43e2b26c522c83ad2b53 : SUCCESS in 10m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d56b996d2906433baca57c3571f8ddb1 : SUCCESS in 16m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2254b441af1146bc97af77570030e4bd : SUCCESS in 11m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/caca4791975846709970bd111570d75d : SUCCESS in 14m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/965f1a94d59a4749a380a32b69b92a04 : SUCCESS in 8m 52s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fc9f5f86e2c94d30a4eaaf47915bcb2a : SUCCESS in 1h 03m 11s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e60b853c30ae4e9598d9a8af7166c23a : FAILURE in 1h 20m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff91c048eedc4ff1bebf42b6badd3af5 : SUCCESS in 1h 14m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8ebdd892f04403b935292dc8f941731 : SUCCESS in 1h 26m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d26e4c958ac24a6b86878b304b13634d : SUCCESS in 51m 29s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fcc54c5822a54c599f658dc71673585e : SUCCESS in 1h 01m 28s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c2a6ef5b2619438a85a220fa6310cf36 : SUCCESS in 1h 30m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/3cb6ac86382f4f7b91dcebd2f68f149f : SUCCESS in 1h 24m 48s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/4b614e98dee0495491095f8344c55029 : SUCCESS in 21m 17s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b167d0933e940938373ec754b1085f2 : SUCCESS in 18m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72fc6f70fcd542af90dc0ef0dea861ab : SUCCESS in 2h 35m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8543e07376a54fcc864624b6451ccb02 : SUCCESS in 1h 06m 41s","accounts_in_message":[],"_revision_number":22},{"id":"4f6a135958b9ae2e5ac5e519329581f0d6e37c7e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-09 08:32:13.000000000","message":"Patch Set 22:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/22/check/EMC_VxFlexOS_NOVA/2af11af/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":22},{"id":"d9325304e401378ffc1816ce824032cd7f6e1cc1","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-09 14:48:21.000000000","message":"Patch Set 22:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/53/678453/22/check-nova/Nova-ML2-Sriov/45595b3 : SUCCESS in 1h 26m 51s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/53/678453/22/check-nova/Nova-MACVTAP-ML2-Sriov/6797b64 : FAILURE in 2h 03m 27s (non-voting)\n- NVMe http://13.74.249.42/53/678453/22/check-nova/NVMe/4ce21b2 : SUCCESS in 1h 07m 05s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":22},{"id":"77d14887397b524fb541c6d55c1856690a8fd5aa","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-09 21:47:08.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"c6c8a68081623bc110c0f8b240d7ff083ec8d93f","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-09-09 22:41:07.000000000","message":"Patch Set 22:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/22/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/94a9a01 : SUCCESS in 1h 49m 51s","accounts_in_message":[],"_revision_number":22},{"id":"7d7b9d024f7bc186c0a26cffd5975e01b4e10908","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-10 06:13:27.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"e23b18ef71fb8180fd0b69555f8fefbeefae86e1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 06:16:27.000000000","message":"Patch Set 23:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"71fa24ba7a705e94a6d0298c4de3b22349841cec","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-10 06:50:13.000000000","message":"Patch Set 23:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/23/check/pmem-tempest-plugin-filtered/8a17e9b/ : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":23},{"id":"af25437211d8a8ad3283ec05c5d480d16f9ca20f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-10 07:22:43.000000000","message":"Patch Set 23:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36930 : SUCCESS in 57m 55s","accounts_in_message":[],"_revision_number":23},{"id":"124d2e307b9dbd1cf01e343f4bbe011326ac6518","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-10 08:19:27.000000000","message":"Patch Set 23:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/23/check-vote/ext-nova-zuul/dd4ca90 : SUCCESS in 1h 11m 37s","accounts_in_message":[],"_revision_number":23},{"id":"7c96a8913cbaf541c5e026e79331b0ef1e55c883","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-10 08:47:28.000000000","message":"Patch Set 23:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/23/check/tempest-dsvm-full-xenial/0dcee79/ : SUCCESS in 2h 23m 34s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/23/check/tempest-dsvm-full-xenial-py3/bcedfdd/ : SUCCESS in 2h 22m 09s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/23/check/grenade-dsvm-xenial/43a61a2/ : SUCCESS in 1h 57m 35s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"a7ad8e9a7f35ec6242c75564ad6162bfcc093630","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-10 09:17:51.000000000","message":"Patch Set 23:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/23 : SUCCESS in 2h 47m 09s","accounts_in_message":[],"_revision_number":23},{"id":"e909537a9942a0e4f4ebb1f70c7e5872fea4f607","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 09:57:09.000000000","message":"Patch Set 23:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/23/check/check-dsvm-tempest-vz7-exe-minimal/e9c7b08 : FAILURE in 18m 08s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/23/check/check-dsvm-tempest-vz7-exe-multinode/da966a3 : FAILURE in 16m 16s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":23},{"id":"1c930d4f1148c0f8ec116ce21c89567cf81372b8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 12:55:24.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3c907d16c0ca43f980ac6e287093c2b5 : SUCCESS in 1h 02m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/27e8c616eba54c1f8208649e043ade78 : SUCCESS in 1h 28m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/599081f860b2478ebab471042140b5fd : SUCCESS in 18m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b62884364a774fba9fd341d7959889bb : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/291ac01397004aeb9b59458b0c733603 : SUCCESS in 9m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/60bd2a1e4481476d90ecd7602cdf56b2 : SUCCESS in 16m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6eb0f07453be4e61910d89a36dcca8e5 : SUCCESS in 12m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6b06810982b74f089ebcf78e5dacecc9 : SUCCESS in 12m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e09b0da3297487188cb35a6e534d1b8 : SUCCESS in 7m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/be3d413a50c8491cb1743887e39e2b46 : SUCCESS in 46m 27s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d36997c3d5544cae9b79bae5fe8a9868 : SUCCESS in 1h 18m 11s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf73bbd423a4488db830a603decd83eb : SUCCESS in 1h 08m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4cda5579148744dbac7402e74e2de0ee : SUCCESS in 1h 04m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a1f166b507f249e7b8e189863d87aa92 : SUCCESS in 51m 28s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/05216ae425ec4d029747c3a71ba448b1 : SUCCESS in 45m 30s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ff757aafba9d44deaf71b524edba595b : SUCCESS in 1h 33m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/5221b9e205fe4bd3a8d54eb537f3ef8f : SUCCESS in 1h 26m 09s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/7db140cc55c0489291ecf0a8ec0a968e : SUCCESS in 19m 43s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8b52aa569dec46818c068d76834b79a5 : SUCCESS in 17m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a3acc194fe04f999375ac3edac39393 : SUCCESS in 2h 29m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6f141d0ab0c4d4f8be0d7674a22bafa : SUCCESS in 1h 08m 19s","accounts_in_message":[],"_revision_number":23},{"id":"ae029fbd5eaaa68f8e7c556bf53867c8429d2f02","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 12:56:38.000000000","message":"Patch Set 23:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/23/check/EMC_VxFlexOS_NOVA/ca1af2f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":23},{"id":"163981d2ab8f29530c435988a7569d666d59e6be","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-10 14:16:26.000000000","message":"Patch Set 23:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-23 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":23},{"id":"a9aa0d955eccfc4658c2c4a426eb9608b0aaa099","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-10 14:25:18.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"590a986530a96f1b7779c13e3ab39b4d328288b6","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 14:28:54.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"913204380ea54b4ffc776ae7f2d756fc3923d88b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-10 15:11:22.000000000","message":"Patch Set 24:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/24/check/pmem-tempest-plugin-filtered/403ac9c/ : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":24},{"id":"898ceb98afd0323a95060d35663487854703a052","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-10 15:28:54.000000000","message":"Patch Set 24: Code-Review-1\n\n(17 comments)\n\nLoads of small stuff but combined I think they\u0027re -1 worthy, tbh. You can ignore the nits if you want","accounts_in_message":[],"_revision_number":24},{"id":"bee93320c776122be708327d4afc557f4796ab71","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-10 18:59:30.000000000","message":"Patch Set 24:\n\n(2 comments)","accounts_in_message":[],"_revision_number":24},{"id":"c17ee4f913896843ca52bd7748994337532d581a","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-10 19:14:50.000000000","message":"Patch Set 24:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36973 : FAILURE in 2h 08m 32s","accounts_in_message":[],"_revision_number":24},{"id":"87268221c182d0ae6b21d4749afe4e91767025b6","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 20:42:34.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/24/check/check-dsvm-tempest-vz7-exe-minimal/5de53cd : FAILURE in 18m 53s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/24/check/check-dsvm-tempest-vz7-exe-multinode/6687b80 : FAILURE in 18m 28s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"999cba669a352412d9d80451e36a8d0eca95a706","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-10 21:44:13.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/24 : SUCCESS in 2h 31m 11s","accounts_in_message":[],"_revision_number":24},{"id":"d9085a9c272a025b86d91125a7eeb7baf4c7a8fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-10 22:45:53.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/700da16ce5c8464bb9c983f25b712e92 : SUCCESS in 59m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dd26aaf9c3f84ed6b09ad46eb3059503 : SUCCESS in 1h 54m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c1cb63885ff4b9fa8508ad3c81add60 : SUCCESS in 21m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c688f8186ee48438d5daa61d7a8a976 : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c963d4ac163f4c4a897ba7e57b3f60dc : SUCCESS in 10m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4536d17fbfb5444b941de1b847d0f73a : SUCCESS in 15m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4207c54080c54b1a815b2088043edab6 : FAILURE in 14m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/90dc8245af50405890e6b00afa822287 : SUCCESS in 17m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/353b83e94ff14936b619d6d36e846ee7 : SUCCESS in 8m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d1801322ee9741e5800119d51a4d457b : SUCCESS in 54m 34s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/66735818024749908732c19a692e69cb : SUCCESS in 1h 15m 21s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc3ecd5ade5e42b69b82e5c12fc17796 : SUCCESS in 1h 14m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3903208f73a461b8b34ffc5e0c8f48a : SUCCESS in 1h 20m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0af4acff4ee34e858060c80b9b689ef0 : SUCCESS in 54m 11s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/bd2f71a37cea467cb7bfe118438cda18 : SUCCESS in 1h 04m 25s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9bd842ab79834e968d086df7475e4af0 : SUCCESS in 1h 33m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/f0102baaf16c41a9a264ecc047dda650 : SUCCESS in 1h 14m 51s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/c1c713dd1b4c4533ac4e94eb08699bf3 : SUCCESS in 20m 20s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1cf1b1a2af244601b5704c901bf0c2f3 : SUCCESS in 19m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f44ddf4204c4432ac9334147ee504b3 : SUCCESS in 2h 33m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/879d9c1994564ab4aed30387b4fff861 : SUCCESS in 1h 24m 24s","accounts_in_message":[],"_revision_number":24},{"id":"9b7a875052ba6bfe3eca949b0098a7fac90e34a5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-10 23:09:25.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"3701857b213c117f7b4d74e82cfbd2d3c1f1ff59","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-10 23:11:20.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"9c037fa613d8d8896dfa7e4f773f7414a1924763","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-10 23:37:04.000000000","message":"Patch Set 24:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/24/check-vote/ext-nova-zuul/898d814 : SUCCESS in 1h 06m 52s","accounts_in_message":[],"_revision_number":24},{"id":"88f05f2c732609cf43771d74ec2818315ba757ad","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-10 23:48:05.000000000","message":"Patch Set 24:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/24/check/tempest-dsvm-full-xenial/27bbca1/ : FAILURE in 1h 58m 22s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/24/check/tempest-dsvm-full-xenial-py3/0e38aa4/ : SUCCESS in 1h 34m 42s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/24/check/grenade-dsvm-xenial/da41560/ : SUCCESS in 1h 54m 10s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d0814dd3462fa16a50909341676c25856693c76d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-10 23:51:23.000000000","message":"Patch Set 24:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/24/check/pmem-tempest-plugin-filtered/3e20faa/ : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":24},{"id":"8093306c36e1ff2e95cb13df60dfa5a6431628aa","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-11 00:24:48.000000000","message":"Patch Set 24:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-36999 : FAILURE in 1h 13m 04s","accounts_in_message":[],"_revision_number":24},{"id":"a9fdd92ab3a962db0f42322302a6044df1e75279","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 02:37:35.000000000","message":"Patch Set 24:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97438bc67726446dbe1b89065d90152b : SUCCESS in 52m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/234c72cb3f5a477ba714fd44c9711349 : SUCCESS in 1h 43m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76d4088b8d674fd5a209c268940d16be : SUCCESS in 17m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec5005d98ab94b919214a8e97eac8939 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e283c091bd074d50987d5ec23ad97f14 : SUCCESS in 9m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04f4ac62bc954fa7a54243e064f91ca3 : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7da0c488fca5424ca3eeefb6d8aa1183 : SUCCESS in 17m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7413496e8ba947bfae776c37c5ae27d6 : SUCCESS in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05340f93d5b047c0a9e11a99ce43f6d8 : SUCCESS in 6m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/533fd82843444724afa8e8bc243d5f43 : SUCCESS in 51m 50s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6e4e1c2217d6457cbaede7438e7bb5a4 : SUCCESS in 1h 10m 48s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cd2d5db0e13341a4b0e15738040d2042 : SUCCESS in 1h 07m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ab9e18b425fd4697aa975f4e7478316c : FAILURE in 50m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1528e0d138474b86b4937a1d37502258 : SUCCESS in 52m 19s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/0fe59b9ffd464ec7875e02eb8844545a : SUCCESS in 58m 54s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f48392279ac4426ea6f5ad74a8c5d25e : SUCCESS in 1h 22m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/eb4f889df86d463d8507f40f0b74685d : SUCCESS in 1h 15m 51s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/898b21306204421b839677279802dd5a : SUCCESS in 20m 21s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d8992bd1972e40ac9b45a0e9acdc8df6 : SUCCESS in 17m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/889c4d76d47a4d899e07d54386dffd37 : SUCCESS in 2h 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0c3c60879db43748eb599708908ffb0 : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":24},{"id":"c24916e0be71a46815cde046b1ea9a7412287b68","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 02:44:46.000000000","message":"Patch Set 24:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/24/check/check-dsvm-tempest-vz7-exe-minimal/2950bc9 : FAILURE in 11m 26s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/24/check/check-dsvm-tempest-vz7-exe-multinode/9d2102b : FAILURE in 11m 20s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":24},{"id":"50824dc2c6a95798ff7aa3c6a9868c487b8f6b12","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-11 04:07:06.000000000","message":"Patch Set 24:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/24 : SUCCESS in 2h 26m 34s","accounts_in_message":[],"_revision_number":24},{"id":"14e28c6e82d3ebac27a696f0ce1c21e59ce089e9","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 08:01:21.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"727d9fe4d3ccf64c9d986127f57d47e5e45f6f7d","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 08:05:20.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"d083fa960afc5c0caa2d6e0df55f1803f87727be","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 08:13:34.000000000","message":"Patch Set 24:\n\n(11 comments)","accounts_in_message":[],"_revision_number":24},{"id":"68aae7417241666eb7193999c92c87a3c7d50bd5","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 08:28:00.000000000","message":"Patch Set 26:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"23d2e009cb2f22b3f7057dd78c0a221182bbd785","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-09-11 08:32:40.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"e0cbee7e15961678cc4ac6a29016bdc864270988","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 08:46:24.000000000","message":"Patch Set 26:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":26},{"id":"061e5c5a521f08e02072bff4fcb363e7f052b1b0","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-11 08:48:10.000000000","message":"Patch Set 26:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/26/check/pmem-tempest-plugin-filtered/8444a59/ : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":26},{"id":"cdc057282b156fbdac4248199774ff9f530a6440","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-11 09:04:19.000000000","message":"Patch Set 26:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":26},{"id":"c25935240981ac01b2fbc25cf16ab98dd4fcb6ca","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 09:51:20.000000000","message":"Patch Set 26: Code-Review+2\n\n(4 comments)\n\nI was +2 before, minor updates lgtm. The couple of remaining tweaks can be done in a fup.\n\nMerge conflict will need to be resolved.","accounts_in_message":[],"_revision_number":26},{"id":"a0559fb189a47e80faf3282e377a4f7c88a185c5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-11 10:35:25.000000000","message":"Patch Set 26: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"6bf075130a08b98f945031a05912b35203010e22","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-09-11 10:35:27.000000000","message":"Patch Set 26:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"1ea980656b76b01dfbec5caa21abb88cace6956b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-11 10:35:31.000000000","message":"Patch Set 26:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":26},{"id":"b36db205fe3b0d99341eaa48cfc0885aa340d5d5","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-11 10:35:38.000000000","message":"Patch Set 26:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov Nova-ML2-Sriov : SKIPPED (non-voting)\n- Nova-MACVTAP-ML2-Sriov Nova-MACVTAP-ML2-Sriov : SKIPPED (non-voting)\n- NVMe NVMe : SKIPPED (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":26},{"id":"e7865d8c20c9d269076086203929e2292f964734","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 11:38:19.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"7b134e09c830d7f7b06f205611072d4d76fe4187","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 11:38:30.000000000","message":"Patch Set 26:\n\n(5 comments)","accounts_in_message":[],"_revision_number":26},{"id":"0c2cb387b9639ab6bbb36dee9eecadec7709d220","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 11:40:22.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"31bafb2625390c2c93cc61486d0af238a5cbba95","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-11 12:12:16.000000000","message":"Patch Set 27:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/27/check/pmem-tempest-plugin-filtered/0912552/ : SUCCESS in 31m 25s","accounts_in_message":[],"_revision_number":27},{"id":"e0b13335837134bc8254ef4da395e5ef8c057a8f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 12:15:08.000000000","message":"Patch Set 27:\n\n(3 comments)","accounts_in_message":[],"_revision_number":27},{"id":"7391b704a5fde815128e38521980b0ad885a5a9c","author":{"_account_id":6962,"name":"Kashyap Chamarthy","email":"kchamart@redhat.com","username":"kashyapc"},"date":"2019-09-11 12:29:22.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"1a9f2ed94b4de4ebca8862fd00cc1339752a6e6a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 12:52:01.000000000","message":"Patch Set 27:\n\n(1 comment)","accounts_in_message":[],"_revision_number":27},{"id":"a035198429d7af173f47e0159f8dafce2e32fe4c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 12:52:51.000000000","message":"Patch Set 27: Code-Review+2\n\nDocstring mismatch not block-worthy, fix if you respin, or in a fup.","accounts_in_message":[],"_revision_number":27},{"id":"327ce9c77fad6ec971041f079fe3ed5f509b1d10","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-11 13:17:05.000000000","message":"Patch Set 27:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37047 : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":27},{"id":"71bdfe0fd57bf9d8012090ae195bd05e705148c5","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 13:27:13.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"548d260a0f6b4e15b0ac7cb23090170099b470eb","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 13:29:21.000000000","message":"Patch Set 28:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"50e6635cc3c38d6f3bd348845d5fd7f29db68287","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 13:30:27.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"581a324fe3d994d330fe878ebb3df4862f4946af","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-11 14:00:53.000000000","message":"Patch Set 28:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/28/check/pmem-tempest-plugin-filtered/7528398/ : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":28},{"id":"378c0f67e22164fc649ee92b7cbb88b7d1471a43","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 15:33:23.000000000","message":"Patch Set 28:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/28/check/check-dsvm-tempest-vz7-exe-minimal/37d411d : FAILURE in 17m 41s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/28/check/check-dsvm-tempest-vz7-exe-multinode/0d2f8ff : FAILURE in 11m 07s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":28},{"id":"a17f06c11e6b47e46696c197ee0e4fc2555b91a0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-11 15:58:16.000000000","message":"Patch Set 28:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37055 : FAILURE in 2h 06m 12s","accounts_in_message":[],"_revision_number":28},{"id":"43c5484b1287cd3d4d14d523f1da0389dd9a93fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-11 16:50:17.000000000","message":"Patch Set 28: Code-Review-1\n\n(5 comments)\n\nAnother object question and an unhandled exception. The latter can be fixed in a follow-up, but the latter needs to be done here if we\u0027re doing it","accounts_in_message":[],"_revision_number":28},{"id":"ee3329a0b61fe05604447cf0caa2f14da8874427","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-09-11 16:51:31.000000000","message":"Patch Set 28:\n\nBuild succeeded.\n\n- Nova-ML2-Sriov http://13.74.249.42/53/678453/28/check-nova/Nova-ML2-Sriov/237830b : FAILURE in 22m 43s (non-voting)\n- Nova-MACVTAP-ML2-Sriov http://13.74.249.42/53/678453/28/check-nova/Nova-MACVTAP-ML2-Sriov/589bfbd : FAILURE in 2h 04m 11s (non-voting)\n- NVMe http://13.74.249.42/53/678453/28/check-nova/NVMe/d05a3cd : FAILURE in 11m 07s (non-voting)\n\nTo re-run the job post \u0027recheck nova-mlnx\u0027 comment. For more information visit https://wiki.openstack.org/wiki/ThirdPartySystems/Mellanox_CI","accounts_in_message":[],"_revision_number":28},{"id":"03c86284ece616b6d18e08c4cc2a8cc058ef4d4e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 17:08:04.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"a1ec86537895a584397b86e68182f20232c72074","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 17:08:10.000000000","message":"Patch Set 28: -Code-Review","accounts_in_message":[],"_revision_number":28},{"id":"266e5092de4e23f917d8ccd4d1877d1ec37d0647","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 17:39:48.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"01c52a564ed4b1539cd96570db7ea4d1840eda17","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 18:07:04.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"9a835c32fd929a6e28713061f6cec893ae7d3f11","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-11 18:32:54.000000000","message":"Patch Set 27:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/27 : SUCCESS in 2h 46m 41s","accounts_in_message":[],"_revision_number":27},{"id":"09f5255b5af822ed759f8e931f20eb1852b49eca","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 18:40:43.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"0413d2acfc2c201271f2bb6a1c5df5924dcea479","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 18:52:20.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"12295d7b374a9c0cc1929145ddc2d893698e4bf3","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 18:57:01.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"a238e4ce391e27da935104ff676221d899c99fc9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 19:00:39.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"a4be3137761704bb49f4ebccedc2d6a7003f2ada","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-11 19:03:31.000000000","message":"Topic changed from virtual-persistent-memory-nonuma-v10 to bp/virtual-persistent-memory","accounts_in_message":[],"_revision_number":29},{"id":"9245478c8a35c725edad806c512a461aea3c3f53","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-11 19:18:51.000000000","message":"Patch Set 29: Code-Review+2","accounts_in_message":[],"_revision_number":29},{"id":"f7e54f43dc20ccc656d12e31f8adde817ab8e8df","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-11 21:26:28.000000000","message":"Patch Set 29:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37092 : SUCCESS in 1h 54m 07s","accounts_in_message":[],"_revision_number":29},{"id":"3dde0772c372ff72dc621a40788c68f8e094fcf0","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-11 22:08:44.000000000","message":"Patch Set 29:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/29 : FAILURE in 44m 33s","accounts_in_message":[],"_revision_number":29},{"id":"278495316f3fed6637ecf54336ebac494a93d0b3","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-11 22:42:12.000000000","message":"Patch Set 29:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/29/check/check-dsvm-tempest-vz7-exe-minimal/e6378d1 : FAILURE in 10m 27s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/29/check/check-dsvm-tempest-vz7-exe-multinode/048f2b2 : FAILURE in 11m 15s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":29},{"id":"1c5cb07a97f1732980f28e1c4484e26a586e1c59","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-12 01:43:58.000000000","message":"Patch Set 29:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/29/check-vote/ext-nova-zuul/ebba31a : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":29},{"id":"b645af5551677689e2502d5feb0f54fcb64f2815","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 02:12:19.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e6dae7585e624ec29b67c93637c56581 : SUCCESS in 1h 18m 20s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3fd194c70e4d41fbaf2f1cd01a1a8ea8 : SUCCESS in 1h 24m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa989f358cc42c4af14ea74b2fd2610 : SUCCESS in 16m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/500faf1938474c41990852f8c7e37c45 : SUCCESS in 16m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e01eb63fde419aaa7ce9555f13a6b1 : SUCCESS in 12m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3224eb6b11a14a0cb397de92eba57c6a : FAILURE in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8e8b29eb7f64fdd9168f83f88bb0eb9 : SUCCESS in 16m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f841adef90d54a24b1fd81129a19913a : SUCCESS in 15m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11197f9137b46918f7b4e24e14d1aa8 : SUCCESS in 7m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b105e7f1be7f4bcdbab23ed279d24cab : SUCCESS in 59m 37s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5eadb5c09b604d73bf345dc3e799b190 : SUCCESS in 1h 28m 57s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/99f0f37a4edb48eaab232b54800c88d9 : SUCCESS in 1h 23m 38s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a58724e580494c609e18599524dad5a7 : FAILURE in 1h 00m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/18cc7462e396492fb2ef52d72e94a936 : SUCCESS in 56m 37s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fde0bbba5c554538894a0781efbdb49c : SUCCESS in 46m 23s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/982d87cf53104992980de0bbf936bb2c : SUCCESS in 1h 32m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/b321b45398c44f019b431dabb87ad05d : SUCCESS in 1h 25m 28s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/912d8a61afd24c09bc667a06a18c6862 : FAILURE in 3m 49s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be87b37667494d299bfa0e07ead77c14 : SUCCESS in 19m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b47758adf1d46e8b88a682ea3ecabbd : SUCCESS in 2h 15m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27e20af403594f37aab8f9a3088a8109 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":29},{"id":"9042993c822e2ac98f0f977288ede7c984becb34","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-12 03:12:42.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"f7ba4c26b09f061dcaf721382626520344b686ec","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 03:19:26.000000000","message":"Patch Set 30:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"4f9208d70f117b3084d4b74c2fc0bdbad2e45ecf","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-12 03:32:04.000000000","message":"Patch Set 30:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/30/check/pmem-tempest-plugin-filtered/a86f209/ : FAILURE in 13m 18s","accounts_in_message":[],"_revision_number":30},{"id":"10d9ea5dd8cbacaacafd138a4617b77209300b85","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-12 03:44:01.000000000","message":"Patch Set 30:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37144 : FAILURE in 12m 39s","accounts_in_message":[],"_revision_number":30},{"id":"0f0778748a63ad6872bdfb72ff9f729f19d544eb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-12 03:57:49.000000000","message":"Patch Set 30:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/30 : FAILURE in 28m 13s","accounts_in_message":[],"_revision_number":30},{"id":"83261c5954da517125494a66095257f82efbab5c","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-12 06:28:05.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"cb838eba25010cba9c65a078c4e5ff769fe8c456","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 06:32:58.000000000","message":"Patch Set 31:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"66be4ad118592a78a395e260741c091e3f27241a","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-12 07:02:26.000000000","message":"Patch Set 31:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/31/check/pmem-tempest-plugin-filtered/b3b661b/ : FAILURE in 30m 03s","accounts_in_message":[],"_revision_number":31},{"id":"99bde8213420e67b7f4b36c8b312df09ea2222cc","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-12 07:14:54.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"b082b3cc8261398db7ddc7667b8969470247cca1","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 07:19:49.000000000","message":"Patch Set 32:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"4ddf8d212e4298075c46949c5ac1b6eda9980754","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-12 07:49:28.000000000","message":"Patch Set 32:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/32/check/pmem-tempest-plugin-filtered/6cc3b65/ : FAILURE in 30m 04s","accounts_in_message":[],"_revision_number":32},{"id":"a27a5387d078e4f74823384f9915df635cde7e07","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-12 08:57:12.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.","accounts_in_message":[],"_revision_number":33},{"id":"fa866c06cfbe058938a08d8fa1e490465709b6f2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 09:02:06.000000000","message":"Patch Set 33:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"6de51885f718739268be871cf6b2918f8f2608cb","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-12 09:33:15.000000000","message":"Patch Set 33:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/33/check/pmem-tempest-plugin-filtered/a9cbf58/ : FAILURE in 32m 01s","accounts_in_message":[],"_revision_number":33},{"id":"1e79fdbd89653318767e37818be9f19d9c325767","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-12 10:34:55.000000000","message":"Patch Set 33: Code-Review+2\n\nPS29..30 trivial manual rebase","accounts_in_message":[],"_revision_number":33},{"id":"d965881261ee5374b95b43f88459cba8187e3b98","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-12 11:14:19.000000000","message":"Patch Set 33:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37188 : SUCCESS in 1h 16m 23s","accounts_in_message":[],"_revision_number":33},{"id":"b6741b96616251902d3e1a4d002063dcdfe2a147","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 11:58:59.000000000","message":"Patch Set 33:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/33/check/check-dsvm-tempest-vz7-exe-minimal/1ddfcf1 : FAILURE in 12m 07s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/33/check/check-dsvm-tempest-vz7-exe-multinode/61b1400 : FAILURE in 12m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":33},{"id":"a2d6009bc99ace98c2ac15b4238a33fd1584444a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-12 12:52:43.000000000","message":"Patch Set 33:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/33 : SUCCESS in 2h 45m 25s","accounts_in_message":[],"_revision_number":33},{"id":"f693c3121f6448f4aab302a3d6b86091427ebab3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-12 15:59:40.000000000","message":"Patch Set 33: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"47b85195cb6f2e9c9972c30711f17cc515a466b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-12 18:32:09.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9d4e4c5e7a9b47c1b1b55a97535efb47 : SUCCESS in 1h 06m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a329ac5632de45f2beaf08c58507094b : SUCCESS in 1h 37m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ddc40922058a402e86bd28b7a1468e67 : SUCCESS in 15m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b38b5c351204d58aa16919a594b7092 : SUCCESS in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fea06089aef47918e551aafa1370c10 : SUCCESS in 11m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/86c423bc4ec2406d87d6c48c0b2791ec : SUCCESS in 13m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe3027996552415da0770dd09ddb88a8 : SUCCESS in 13m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34864a5efd1e4f47a5fcfd4d139bf927 : SUCCESS in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faafffc2c3b54294b9ef84e272b17410 : SUCCESS in 7m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3ed557856cf24f3d9fc131cbee227c64 : SUCCESS in 1h 04m 55s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2dc9f4587b464b099c56dd4ce694fef7 : SUCCESS in 1h 15m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1dc14a54d4cd48ed8cb88ee321814eed : SUCCESS in 1h 13m 29s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e41f6a6f09834dd090123ba5305dde8b : SUCCESS in 1h 10m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ebb338c3a9034b33a2a3669b66782359 : SUCCESS in 59m 01s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1eb4439ce5d34a038954e9b0781c5068 : SUCCESS in 1h 02m 20s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a935ebb581984b7ca3f410be29892dde : SUCCESS in 1h 25m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/907ac7e32ad341d0893c6a325ec3f29d : TIMED_OUT in 2h 09m 33s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/fffce7dd5abf408fb7019b2b5e8d1f4b : SUCCESS in 21m 33s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5712ecb0985f4130a1b10615599beb60 : SUCCESS in 17m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ba27a360d334335bcd885a730f7ec33 : SUCCESS in 2h 24m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e13a592c79e54c41968e2049f8f369de : SUCCESS in 1h 10m 32s","accounts_in_message":[],"_revision_number":33},{"id":"8ca97b0a7e4664ffaaa17e6daef8bdf7d1847a5f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-12 18:42:48.000000000","message":"Patch Set 33:\n\nrecheck","accounts_in_message":[],"_revision_number":33},{"id":"16c22de3ce1485794bfa19d9e5b3b93b3a6c6422","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 18:46:02.000000000","message":"Patch Set 33:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"0e34e05cf9ffd84eff25b3ee1631f005347026ae","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-12 19:12:59.000000000","message":"Patch Set 33:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/33/check/pmem-tempest-plugin-filtered/1bb9500/ : FAILURE in 28m 21s","accounts_in_message":[],"_revision_number":33},{"id":"24c47b8d00b76ac19fffe54897754e82bac89efc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-12 20:21:16.000000000","message":"Patch Set 33:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37251 : FAILURE in 1h 25m 27s","accounts_in_message":[],"_revision_number":33},{"id":"a7adf0d1104314b72af0ebe398e58dc9d371a70c","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-12 21:00:20.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"29d4c372e179c26d30f6a4f88809c901b0d1665b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-12 21:02:43.000000000","message":"Patch Set 34: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"fb690d87ef22a7f49b8c4a49e59ac09602c713b9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-12 21:02:52.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"918fea2fb61cafc9cee2a9b53bd57269ccf90feb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-12 21:04:39.000000000","message":"Patch Set 34: Code-Review+2 Workflow+1\n\nMakes sense. The joys of a blocked up gate, eh?","accounts_in_message":[],"_revision_number":34},{"id":"6913bf3ae44b0ed70e48aba5aed96cdc0d203742","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-12 21:35:22.000000000","message":"Patch Set 34:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/34/check/pmem-tempest-plugin-filtered/a9dedc1/ : FAILURE in 32m 19s","accounts_in_message":[],"_revision_number":34},{"id":"c975ef1766979289caca2f2462a4395dc743d2fb","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-12 22:22:41.000000000","message":"Patch Set 34:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37255 : FAILURE in 1h 07m 38s","accounts_in_message":[],"_revision_number":34},{"id":"6c62190a4fa0026d1e1fd605b5d9045497015c04","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 00:00:56.000000000","message":"Patch Set 34:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/34/check/check-dsvm-tempest-vz7-exe-minimal/56a2f5f : FAILURE in 11m 25s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/34/check/check-dsvm-tempest-vz7-exe-multinode/f1eaf8d : FAILURE in 12m 05s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":34},{"id":"d52525160d711980193237bae4771661cd2bdadc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 06:14:40.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c8a93448e1d645e5b023be2cab35715e : SUCCESS in 1h 02m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2e0583e6c3764705b0cce57daa855e15 : SUCCESS in 1h 34m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb4923a1b9e442b6bc7160966f72bf50 : SUCCESS in 17m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/185109f1e9234ccbae5cde3adae2fab4 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf22e0517903404e84f943cdbf2a9134 : SUCCESS in 11m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12c160539e444e2c9d8db3c053fd84e9 : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc76135a63bf40ca8c54d0dc6c7f1fed : SUCCESS in 14m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/730d1d623ba74797ba5634f5ffdb69b0 : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f6afc91d294ee78150c1c83262c006 : SUCCESS in 6m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/25e1cb9636c8438fa00571921a550f54 : SUCCESS in 59m 15s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/d6f7ad31d7e3484596c6957ff1612963 : SUCCESS in 1h 28m 02s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ed5a4d896ff4cf686b1c819e4b7502a : SUCCESS in 1h 21m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/965bbeb1bef342c2810187cb34306f1c : SUCCESS in 1h 14m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9133788048b34548b4ec3f4239c20565 : SUCCESS in 51m 02s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/75591750576542239c52a8c5902a614e : SUCCESS in 51m 44s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1002d98a14124f98900b3fa8721cb6b2 : SUCCESS in 1h 36m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/debd30c8cddc4b7e81635a67b060c80b : SUCCESS in 1h 24m 03s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/bfaf4209ce6e4d25b3dcbbc01f988cf5 : SUCCESS in 18m 36s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3edf860cecd847629bf8626c697ed8de : SUCCESS in 17m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4dfef59ea89b453e9cae71f305b01d54 : SUCCESS in 2h 21m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e473163a6b7e486b83bb22bc8b5cb6a8 : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":34},{"id":"33edf2e78f53f75207c4ca5b29bc79adb2caab0a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 06:15:55.000000000","message":"Patch Set 34:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/34/check/EMC_VxFlexOS_NOVA/c41f648/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":34},{"id":"0ae88e04deefdfb8ca790333214fd1d02bc22d78","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-13 06:35:31.000000000","message":"Patch Set 34:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/34 : FAILURE in 2h 54m 32s","accounts_in_message":[],"_revision_number":34},{"id":"3aa52e2bb0c7c5e30eb5c9b061d0115e17719301","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-13 08:42:58.000000000","message":"Patch Set 34:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/34/check/tempest-dsvm-full-xenial/a12a9a2/ : SUCCESS in 2h 14m 31s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/34/check/tempest-dsvm-full-xenial-py3/7253f65/ : SUCCESS in 1h 55m 07s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/34/check/grenade-dsvm-xenial/9a5e573/ : SUCCESS in 1h 44m 57s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"149b44c1bc960bdaa102295ed4910b7e871d925d","author":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"date":"2019-09-13 08:55:20.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"bef07704501d38fd692e1f5f9fcc1a3fb281d4e2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 09:02:39.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"380209a5c291edb93e9676d66c62e1caa7b035d4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-13 09:20:39.000000000","message":"Patch Set 35: Workflow+1","accounts_in_message":[],"_revision_number":35},{"id":"a4dd999be7b32b5b2a5080fa33624ea4155640a1","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-13 09:34:01.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/35/check/pmem-tempest-plugin-filtered/c2781ea/ : FAILURE in 28m 56s","accounts_in_message":[],"_revision_number":35},{"id":"0b79de203b5ce53449569b3b9dc4283a3306fd20","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-13 09:45:27.000000000","message":"Topic changed from bp/cpu-resources to bp/virtual-persistent-memory","accounts_in_message":[],"_revision_number":35},{"id":"05d4e1796e6f18191443524da7089d1943da765b","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-13 10:05:40.000000000","message":"Patch Set 34:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-34 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":34},{"id":"c1efccd0a3e0dfe057bd9ce33453749ec9fa9d73","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 11:46:49.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-minimal/3d271c3 : FAILURE in 17m 13s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-multinode/6daf531 : FAILURE in 13m 31s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"faaef2aff8cadb11f379b06f954cbdeb0c12f8e3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-13 12:03:46.000000000","message":"Patch Set 35:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37310 : SUCCESS in 1h 39m 17s","accounts_in_message":[],"_revision_number":35},{"id":"bd786114b9abdaf3d68296912ba08c2ed6ddff75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-13 13:48:58.000000000","message":"Patch Set 35:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial/04fbe43/ : FAILURE in 2h 12m 30s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial-py3/dc4b850/ : SUCCESS in 2h 19m 06s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/grenade-dsvm-xenial/73a4c38/ : SUCCESS in 2h 09m 59s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"52ee3ba1df4db1c6606f7903993c739bd51f91dd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-13 14:52:09.000000000","message":"Patch Set 35:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/35 : FAILURE in 16m 03s","accounts_in_message":[],"_revision_number":35},{"id":"464c537c51cd8c66a16efa0a2f2b5aece16ed9cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-13 19:06:49.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/170ee8e801be46edb0ba019555ddfdad : SUCCESS in 1h 03m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b45b04ac688c44c48250cedf240a090e : SUCCESS in 1h 35m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc5e66ecc88b4d3482c460484b3360e6 : SUCCESS in 18m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f35a4fdbab2845e690ccab4978bb3f3e : FAILURE in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20789cf6ad054e6fa9c8afee9f665cfe : SUCCESS in 11m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c75b9b7e0da4c4181b9d0c59232209a : SUCCESS in 15m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01c7bfae3e294d098671576ebdb3c7b3 : SUCCESS in 11m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ae9f17edd424feeaf1e3384b44c886b : SUCCESS in 14m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f162a993a1d24d91812e754db622f952 : SUCCESS in 6m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b486f705df8245fa932c236e8a91973b : FAILURE in 44m 57s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/9aeaf6a602964302953c902c0af8da9f : SUCCESS in 1h 20m 15s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6a4411cb39224b709b68652de6a1fe09 : SUCCESS in 1h 12m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e0734b98b994dc38e2480138f9dbfb5 : SUCCESS in 1h 17m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/948a4f2249ec4257b9009e6ce7b74a97 : SUCCESS in 54m 39s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b4765d1b5c434e479fc86a0e1fe66959 : SUCCESS in 55m 18s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b586057862cd4f18afcc381a7d1e5610 : SUCCESS in 1h 27m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/51d99546d35a46ff989d19b0c6adbd31 : SUCCESS in 1h 22m 20s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/cbdf85ae0f4241ae90f7e905bd972c65 : SUCCESS in 22m 22s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b75a4a0c40a4befa2ac6e8248018503 : SUCCESS in 21m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0947fc513526418ab9cfb940e0bec606 : SUCCESS in 2h 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46ae9e6cc3994c86b7e937df7dc20df8 : SUCCESS in 1h 22m 19s","accounts_in_message":[],"_revision_number":35},{"id":"91994838c673d18eb4c127d95f55f9dc95852268","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-13 21:24:46.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/35/check-vote/ext-nova-zuul/510b3b5 : FAILURE in 1h 37m 17s","accounts_in_message":[],"_revision_number":35},{"id":"b83e2d1b687076a8278f09ad83860124acb90402","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-09-13 22:19:41.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"30bd7daf973c8afc1a434d98a67508ed92be86a4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 22:24:29.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"2ce655792cd47ef302cf386be0f0abb68e1e2620","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-13 22:51:36.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/35/check/pmem-tempest-plugin-filtered/73913e8/ : FAILURE in 28m 17s","accounts_in_message":[],"_revision_number":35},{"id":"ee9df4a94db04331546d3f23a54e376844641e32","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-13 23:31:53.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-minimal/8f044c9 : FAILURE in 16m 03s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-multinode/4eee6ad : FAILURE in 11m 13s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"ee1d144a9943bf754bf68f95f355d1bb9091fca0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-14 00:49:55.000000000","message":"Patch Set 35:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37349 : FAILURE in 1h 50m 16s","accounts_in_message":[],"_revision_number":35},{"id":"decdea783f93dc23e953d2fb7c9f282505ad1626","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-14 01:20:09.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/35 : SUCCESS in 2h 55m 07s","accounts_in_message":[],"_revision_number":35},{"id":"29c17c105f2b00aad31a0b11de10b10a9f305fa0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-14 03:48:16.000000000","message":"Patch Set 35:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial/457f3c8/ : SUCCESS in 2h 42m 02s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial-py3/27c1324/ : SUCCESS in 2h 05m 16s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/grenade-dsvm-xenial/4cd2fa8/ : SUCCESS in 2h 30m 44s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"3ebc8202dd981ce41d15d52af10377022c5c3d27","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 04:54:53.000000000","message":"Patch Set 35:\n\nTesting succeeded on ubuntu-xenial-s390x. For rechecking only on the ubuntu-xenail-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x http://sng01.objectstorage.softlayer.net/v1/AUTH_1940ea10-6e82-4501-b2f9-eb236510e575/ibmzkvmci/production/678453/35/check-tempest-dsvm-neutron-full-ubuntu-xenial-s390x/7fce14d : SUCCESS in 1h 49m 49s","accounts_in_message":[],"_revision_number":35},{"id":"4958d963d6a91f36e9948814603b4e76ea6db94b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 05:20:47.000000000","message":"Patch Set 35:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7d9801b11fb4415cbbc80b63887d95bb : FAILURE in 1h 03m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ab80a30bedff4deaa062d80bddd586bb : SUCCESS in 1h 25m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be143a1131224ea895d6f7ea5ee714b3 : SUCCESS in 17m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d0a4101b33042cea1a5e3334cde03fa : SUCCESS in 16m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dbdb192826040f2bfdbd1429401f0cf : SUCCESS in 11m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/93a2d70cccc7470eaa11cf04e3e27aa0 : SUCCESS in 15m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1065182a8107467a9ad7ecd871f42674 : SUCCESS in 14m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e9ebd6f208824e158ba8adf17e6e96fd : SUCCESS in 13m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fcc7ce6a334410a8f0244b99461d386 : SUCCESS in 6m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a756774ae9914b4ba2975528edbbda74 : SUCCESS in 52m 08s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/8bd9eeddb8bd4dafa0b2675898d117ec : SUCCESS in 1h 06m 05s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ccb59e134a84203b67750c9c75aa18a : SUCCESS in 1h 17m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3acfada7f114edbb512373757946ae1 : SUCCESS in 1h 19m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/59831ad28cef43d5a5dc961ebe121389 : SUCCESS in 52m 23s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d55469349f584896bf360aa282861bc3 : SUCCESS in 1h 00m 29s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/38b05e7c93a94cbe9abc0e1f18decc4f : SUCCESS in 1h 28m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/6997a5b786d64fbd882a8e0e71e47c74 : SUCCESS in 1h 18m 33s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/f98c2b29988b44dca9551150306e4c04 : SUCCESS in 20m 10s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e353e8200bf4b91a5b328deb9aae5ea : SUCCESS in 22m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a89f493ba524895b750db4e5361f2fc : SUCCESS in 2h 24m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/874c08db7ef54f26a72d397e8fc9dd84 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":35},{"id":"3830f566441488439c394e97fa933c7daf46820f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-14 05:36:09.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"6c43e16ec96786fbb8f932a5d0ee72680586dad7","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 05:40:18.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"7101d9a293f8f4b2be385a8740c4b8915ea4ac90","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-14 06:07:43.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/35/check/pmem-tempest-plugin-filtered/1dc3fe5/ : FAILURE in 28m 08s","accounts_in_message":[],"_revision_number":35},{"id":"696206ec613b60141b5603a392987575e2471daf","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 06:14:52.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-minimal/9d5c43d : FAILURE in 16m 42s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-multinode/26518a2 : FAILURE in 18m 57s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"c621a71bcbdd6798f1f245e2117e3c158a240a5b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-14 07:04:42.000000000","message":"Patch Set 35:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37361 : FAILURE in 1h 14m 43s","accounts_in_message":[],"_revision_number":35},{"id":"b5381205ac9cc0be70d5ec09fa488dc4eff90dee","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-14 08:10:34.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/35 : SUCCESS in 2h 28m 05s","accounts_in_message":[],"_revision_number":35},{"id":"756ae2883af52566aef6fe6f404558adbd7eb527","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 08:30:54.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/054b7fa83d684f11aa583c1193d5f695 : SUCCESS in 1h 51m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/afe5de1ad577400b81ff5590590a26f9 : SUCCESS in 1h 41m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bdf366e782d4989a9fa609b8531339b : SUCCESS in 30m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f534a5456db4240a24ab110475c0cb0 : SUCCESS in 15m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dae097a36364b3fa3ee981e2a018a90 : SUCCESS in 11m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5d089b14bd848b9bb0452348fee0e5c : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4c5bbbffaef41649b24797b084c0fc8 : SUCCESS in 13m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b859ba5f239043f1a3cb1d5b273f2218 : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6588efe2554aeaaf5783fc7ed5e88c : SUCCESS in 6m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/42af6daff4484e55b101c3b67a33233c : SUCCESS in 1h 02m 10s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/345d51766ebb44b3a4b44745db94152e : SUCCESS in 1h 11m 35s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ead0bb36651244b4bda7d8ff8220a2c0 : SUCCESS in 2h 18m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9a6130854c3e4b7fb401b9247dae78f4 : SUCCESS in 1h 30m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bae0d344efc64db89cc85d280e27456f : SUCCESS in 1h 51m 13s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b8bcfce533d84952bf784373528ef669 : SUCCESS in 52m 24s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e51d6d1084664d58adda3f326db14273 : SUCCESS in 1h 51m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/1cd569ea3cfa42d1b4269a5c4ca011e3 : SUCCESS in 1h 39m 32s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/969bfa57191a40a597dc60ae8b82e945 : SUCCESS in 17m 50s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/72cceec92e444e87bd6545fe7988558a : SUCCESS in 23m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c1c9a17dc8e341bda5ae073456c402bf : SUCCESS in 2h 48m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7ef929b8d6d4a1ea38df472ccd76f28 : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":35},{"id":"f0255d4049520cb20384673af228a2434d3a9e53","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 08:32:06.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/35/check/EMC_VxFlexOS_NOVA/1b311a0/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":35},{"id":"ca0f2634690f98502e7ed15386ed0304bd99c747","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-14 08:44:41.000000000","message":"Patch Set 35:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial/39d1e9d/ : SUCCESS in 2h 47m 13s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial-py3/99fdb62/ : SUCCESS in 2h 10m 59s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/grenade-dsvm-xenial/b69dcdd/ : SUCCESS in 1h 53m 04s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"e7b67d5de987b3c8d8a0d06f30ac151aa9982f68","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-14 10:56:49.000000000","message":"Patch Set 35:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-35 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":35},{"id":"77293c1c27c0d8f0c9e03300919c02424615040f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-14 12:00:10.000000000","message":"Patch Set 35:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/35/check-vote/ext-nova-zuul/c8baba7 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":35},{"id":"bf3e1158ba85470624625a2e76d81c3641351227","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 12:03:14.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"4f26a95511f2dd2bdd011e75830f2bf5af8d4d18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 16:27:40.000000000","message":"Patch Set 35: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":35},{"id":"d96ac7d55c9e074e55186b4ab7c46d7940f13a29","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-14 17:01:48.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"d65b7a35d816577f8f2ba8dd88a45fe979c2e254","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 18:23:33.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"46bfa4d85fdc3ae4c7a7955494c1989fd4d4533a","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-14 19:24:29.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/35/check/pmem-tempest-plugin-filtered/4be98a2/ : FAILURE in 27m 42s","accounts_in_message":[],"_revision_number":35},{"id":"f629dcf2f7cda9b136fa5da9e6c99c6180ff8f58","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-14 20:25:49.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4ef4f7679a754fd88b5ff1df1d48926f : SUCCESS in 1h 01m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2a3428b0778b4531954249052980bb39 : SUCCESS in 1h 23m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16cbcce8104941479ce488a5f840a036 : SUCCESS in 17m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31f8b944c9fc4d69bc2a840e5e507eb2 : SUCCESS in 15m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/445339af1a29459bb4ebfd5a1150e667 : SUCCESS in 10m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f1f1c37a84142979dd4e84ceb94d672 : SUCCESS in 13m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42e3a13dd7214c118936b937523b6fc2 : SUCCESS in 12m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b17ea4740d345969055cfc16aa438e7 : SUCCESS in 12m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c95533865be48e1b1b37857344ad618 : SUCCESS in 10m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/81186fe4a29a4f98aede3c5389047ec1 : SUCCESS in 55m 06s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0d838d0c33ca41539ebca57c9af9038a : SUCCESS in 1h 05m 50s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b6508bf14d84d06b27beb581bdf60bd : SUCCESS in 1h 15m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1b6d4c2e593340dd86fd55c60020fe0a : SUCCESS in 1h 14m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f47571494e5448ddb01a8a18e3b5e6fe : SUCCESS in 54m 42s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4a5166bc203945089ba3d26364dcc3ca : SUCCESS in 53m 14s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/eaed707013b24cc49ca2edb493e74987 : SUCCESS in 1h 40m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/570c8e4bf06b457c97290c907bfd009b : SUCCESS in 1h 30m 05s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/cde1d0d8e27a4b3884561dab129cf05f : SUCCESS in 18m 57s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56ead85539f64db29f215e4f9a935d21 : SUCCESS in 20m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f5a247137de42f48deb05bc51c040e4 : SUCCESS in 2h 29m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5519a47ebb0a4bbd8f457ecd8d9cebaf : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":35},{"id":"d7d1d44b1e2946eff6d2e1cdb96028134713bf58","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 20:26:59.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/35/check/EMC_VxFlexOS_NOVA/4f528f1/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":35},{"id":"2c81c596486637f07ae8dea68aa8dea61d1b93a4","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-14 22:13:01.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-minimal/2851a26 : FAILURE in 11m 17s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-multinode/7f38530 : FAILURE in 17m 25s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"b870d6a17bca471369622b20457172ac0c58ff3d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-14 22:24:23.000000000","message":"Patch Set 35:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37396 : SUCCESS in 1h 14m 15s","accounts_in_message":[],"_revision_number":35},{"id":"90443e2ae11821d837ea76e698eb32a7d67be02b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-15 00:11:49.000000000","message":"Patch Set 35:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/678453/35 : SUCCESS in 2h 26m 11s","accounts_in_message":[],"_revision_number":35},{"id":"93eead84e45e1b3eb2986f9c23eb13c9ebe975a2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-15 00:27:58.000000000","message":"Patch Set 35:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/35/check-vote/ext-nova-zuul/2455f50 : SUCCESS in 1h 01m 22s","accounts_in_message":[],"_revision_number":35},{"id":"c17a83d7f45d02a22db173e909ccca266dd88a52","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-15 00:46:29.000000000","message":"Patch Set 35:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-35 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":35},{"id":"8a93b639130a4bb74890e69c15855c8f1734dd74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 03:00:01.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"d34651ef4b4b2194d1711e07f0c8518f590686ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 06:34:04.000000000","message":"Patch Set 35: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":35},{"id":"a32af08bf5e57326e2652daac023f569fad26749","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-09-15 06:50:36.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"6f2a488dc1de2932040185dea9eebb77819713b0","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-15 06:51:31.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"7bace69ff9984496c2994ae3808a708622eb090e","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-15 07:21:44.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/35/check/pmem-tempest-plugin-filtered/924167e/ : FAILURE in 28m 27s","accounts_in_message":[],"_revision_number":35},{"id":"ac239acc81c3c890f63afb479c85c23bd6b38fd2","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-15 07:38:06.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-minimal/1f23300 : FAILURE in 17m 45s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-multinode/956bd82 : FAILURE in 12m 10s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"cbb16874f55511e49ed621fa8eb266356ff802df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-15 07:56:20.000000000","message":"Patch Set 35:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/35/check-vote/ext-nova-zuul/995ed9b : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":35},{"id":"bf7e71eaf09938ec5695cd562a427f3e5e6be32c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-15 08:04:00.000000000","message":"Patch Set 35:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial/35b0a23/ : SUCCESS in 1h 46m 21s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial-py3/256beed/ : SUCCESS in 2h 17m 55s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/grenade-dsvm-xenial/a8989a0/ : FAILURE in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"a78ae312a3bf83d7e73bc009339b03e7be718029","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-15 08:04:34.000000000","message":"Patch Set 35:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37410 : FAILURE in 1h 12m 26s","accounts_in_message":[],"_revision_number":35},{"id":"fef4bfa3a64e4c7f0c0963841a6bf7698698cacc","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-15 09:18:23.000000000","message":"Patch Set 35:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/35 : FAILURE in 2h 26m 21s","accounts_in_message":[],"_revision_number":35},{"id":"376a9d39c7b9b37dc233742856de8e04511afb3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 09:40:04.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c0dfaec92bbd4ee5b604bd8511f44d02 : SUCCESS in 1h 09m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/abed98ccb03c46fcb0fabb7d2273796d : SUCCESS in 1h 27m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab4ca3cc98bc48ef95eae674833f3bf8 : SUCCESS in 16m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2192e35212f472dbe14462ab4c367e6 : SUCCESS in 13m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d633928d394de3ae86ba7fc39f3896 : SUCCESS in 9m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/286827932e2a4edfaa5a5c977e75bac0 : SUCCESS in 12m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46b60b3559694ae1a815fe132013648f : SUCCESS in 10m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eaa49e98dd944f69932e2be45c191c89 : SUCCESS in 11m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66ab67a627764f558bac621a42f6a67f : SUCCESS in 9m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0aa63565b27f47f294266f0c23860f96 : SUCCESS in 58m 24s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5d2e1bdc155c4374b468bc3933402413 : SUCCESS in 1h 30m 47s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/db32a88fb6e441f3a6dc183f2eb2b7fd : SUCCESS in 1h 16m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f7e885eb9fa14ec1941b84432a8526aa : SUCCESS in 1h 12m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c87cf033a5714e61b28efc7b1b719ce6 : SUCCESS in 58m 03s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4ab6d9fd80444c04a52cf3b3cbd5da05 : SUCCESS in 56m 39s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ad5dd769c32749719faf2be45e70385a : SUCCESS in 1h 46m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f31a492d5284456abc0239d2447fda6 : SUCCESS in 1h 26m 44s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/f9a3bb8f85ca452fa0c9efbd35b29f7c : SUCCESS in 20m 15s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c9649c8602946f280713b6611481bfe : SUCCESS in 18m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c6df451a9c2405fa08920a704ad2cc1 : SUCCESS in 2h 20m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38380dedc45042eca68739acbc4e59f9 : SUCCESS in 1h 03m 03s","accounts_in_message":[],"_revision_number":35},{"id":"d285216eb718bcd4f171db998f156fce1ed31940","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 09:40:34.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"5474892628464b675b5beaa931b43f66ca467cd0","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-15 09:43:02.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/35/check/EMC_VxFlexOS_NOVA/1ce5af8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":35},{"id":"3c5cc0d5e039544d4b126a91a42e627ebe0068f3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-15 10:59:21.000000000","message":"Patch Set 35:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-35 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":35},{"id":"636e632a7892b28aeda51a433fbce632c64a63dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 14:55:45.000000000","message":"Patch Set 35: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":35},{"id":"d04fb8954dc64e2fd4e57a26069edd8fa81a994f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-09-15 16:27:33.000000000","message":"Patch Set 35:\n\nrecheck","accounts_in_message":[],"_revision_number":35},{"id":"b315d4bb028b847a192a188651c55cb55a677875","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-09-15 16:38:37.000000000","message":"Patch Set 35:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"805b67c488b4466bda103ca313928b65360dfaae","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-09-15 17:33:54.000000000","message":"Patch Set 35:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/678453/35/check/pmem-tempest-plugin-filtered/f7f68af/ : FAILURE in 33m 27s","accounts_in_message":[],"_revision_number":35},{"id":"cae2085b88262cb09d4cac95c66f8cfc9ff68b4b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-09-15 18:48:18.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/678453/35/check-vote/ext-nova-zuul/9ff86ae : FAILURE in 17m 20s","accounts_in_message":[],"_revision_number":35},{"id":"6fbb6eb15fc8a8464461445b78a1d27b5cf9f2b0","author":{"_account_id":16898,"name":"Virtuozzo CI","email":"virtuozzo6-ci@virtuozzo.com","username":"virtuozzo6-ci","tags":["SERVICE_USER"]},"date":"2019-09-15 18:49:33.000000000","message":"Patch Set 35:\n\nBuild failed\n\n- check-dsvm-tempest-vz7-exe-minimal http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-minimal/d352d27 : FAILURE in 18m 00s\n- check-dsvm-tempest-vz7-exe-multinode http://openstack-3rd-party-virtuozzo-ci-logs.virtuozzo.com/53/678453/35/check/check-dsvm-tempest-vz7-exe-multinode/2cdfb14 : FAILURE in 20m 49s\n\nFor information, see https://wiki.openstack.org/wiki/ThirdPartySystems/Virtuozzo_CI Make the comment \u0027run-Virtuozzo CI\u0027 to recheck","accounts_in_message":[],"_revision_number":35},{"id":"edcc5cdf3278416a245c382c656ef800a51a6ace","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 18:55:24.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/33dd70092abc460c850fc239f8382a52 : SUCCESS in 59m 25s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/32edec62fded476494565011b74ea11d : SUCCESS in 1h 25m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1c00d61dc3f49d9b606c3dfa86fe4e0 : SUCCESS in 23m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/239079a383464a50b256bb7fb378b1f5 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0fe5e2c0b474d8a86216a4996afd608 : SUCCESS in 13m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4395b3e3d8af4c9aae210de2a043e8a2 : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fe2e01f5e3e4ca6b1df374a747c5fee : SUCCESS in 19m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/487d0edb440840aaa6d9ab9d6468bfa7 : SUCCESS in 14m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357201bcb3ee40f2938bccced709034e : SUCCESS in 9m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/69c13bba573c49bf84a668696ad45107 : SUCCESS in 50m 26s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/57ae51b918cc48d9823f5a378c28937a : SUCCESS in 1h 28m 39s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d921a019939041c9baeda966e0c645b3 : SUCCESS in 1h 09m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d751cdb6115b43139ff3bf1c15460923 : SUCCESS in 1h 08m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5188c362fc244ed6a6f763753004edcd : SUCCESS in 50m 53s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/684342110c744c4db603f0d389e63bda : SUCCESS in 58m 31s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8553c7ea07cf4b66a01b7085502997a5 : SUCCESS in 1h 39m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/36a70af94f4942cea0cd053604d23ca2 : SUCCESS in 1h 21m 14s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/3beeb68e103746159211bd02ccdfb318 : SUCCESS in 18m 35s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51995a83c2014302bb39e9db51bdb84c : SUCCESS in 19m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0a6df8ef6436472e977f752a8120504c : SUCCESS in 2h 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f7b14cb6e8941fd889dd2e7a7a2ae19 : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":35},{"id":"30e9ae9b630d85d6ffe0573e2fb2df9e1d6596a0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 18:55:52.000000000","message":"Patch Set 35: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":35},{"id":"31e426090084bf278f42caa86c51f9b7c91e7f3b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-09-15 18:56:03.000000000","message":"Patch Set 35:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/678453/35/check/EMC_VxFlexOS_NOVA/8cad094/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":35},{"id":"b5cf808d8af856b8768c6110c8da4670658dc7ea","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-09-15 19:17:58.000000000","message":"Patch Set 35:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/678453/35 : FAILURE in 2h 31m 13s","accounts_in_message":[],"_revision_number":35},{"id":"ce532c1c10db3850526a224ce12e5736d8877b07","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:22:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":35},{"id":"4605bf01cd30513a756a7c1f45d16582ba1998c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:22:49.000000000","message":"Patch Set 35: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9b784d8ba8354877989a68aa7107a21e : SUCCESS in 1h 12m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c06315f9406349aba36fde586cac47dc : SUCCESS in 1h 33m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62148c111d8f41d38c5e1fe124c628b6 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc3e49df7e3f45f4b11b6423c3f0ac0a : SUCCESS in 12m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ec8a97628d84eb1854dd558789353c6 : SUCCESS in 15m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dffc377973a34e108cdb628ac69a7065 : SUCCESS in 17m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e454528c445478c82ace89941ccd485 : SUCCESS in 18m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a7436d57124ab8ab883a445db66774 : SUCCESS in 8m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/10104d0a4b6745f69be3383ad909d827 : SUCCESS in 1h 20m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3285747d7c754ca48c17e8d169e888ac : SUCCESS in 52m 28s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/dc4d03d008a6457da40f7fd274ebe75e : SUCCESS in 23m 11s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4826208725c94c4798f0603097f4f53e : SUCCESS in 16m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6b3eb00bad524ca5bc937da0f03ad886 : SUCCESS in 1h 52m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/db4bae53f8c74105879f0c8ec594747e : SUCCESS in 1h 31m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc9d8ff8a34d41e09fedfde358b58df9 : SUCCESS in 2h 23m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5aaf8ebb01541348f2bb522b2bda00e : SUCCESS in 1h 16m 17s","accounts_in_message":[],"_revision_number":35},{"id":"d544ddd7413b7d1b8d9826535146e98c675e22cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-15 21:25:29.000000000","message":"Patch Set 35:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b7e0d933ff47f9ac428709ffe9dc76 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":35},{"id":"fc67604621bd6df38ebd9352cf2f7fa28c546965","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-09-15 21:33:25.000000000","message":"Patch Set 35:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-678453-35 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":35},{"id":"1a0037e94c3f3c3dbd289a16c17ba2784a051b97","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-09-16 10:47:48.000000000","message":"Patch Set 35:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial/f7321fe/ : SUCCESS in 2h 01m 50s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/tempest-dsvm-full-xenial-py3/d66250d/ : SUCCESS in 1h 33m 43s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/678453/35/check/grenade-dsvm-xenial/2d729d1/ : SUCCESS in 1h 24m 14s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"01e6a562df64ec494da5a8b8e9ba65fc49cf0e87","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-09-17 05:45:49.000000000","message":"Patch Set 35:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-37447 : FAILURE in 2h 11m 58s","accounts_in_message":[],"_revision_number":35}],"current_revision_number":35,"current_revision":"3978b99d8f37643a3b32f6b7ead9274c454095a1","revisions":{"0c38880bc0025bc635f134b19398914dcc9221ec":{"kind":"REWORK","_number":1,"created":"2019-08-26 03:25:40.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/1"}}},"commit":{"parents":[{"commit":"132ddc6cc4693c8a524007606976c42f2aa90c13","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/132ddc6cc4693c8a524007606976c42f2aa90c13"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-26 03:25:22.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c38880bc0025bc635f134b19398914dcc9221ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c38880bc0025bc635f134b19398914dcc9221ec"}]},"branch":"refs/heads/master"},"3fc53af00a49c2646f7777c821cbef3588cb2bb2":{"kind":"NO_CHANGE","_number":2,"created":"2019-08-26 06:45:20.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/2"}}},"commit":{"parents":[{"commit":"6ab149d5206a52de945f5488295412897fe0b5a5","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6ab149d5206a52de945f5488295412897fe0b5a5"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-26 06:44:01.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fc53af00a49c2646f7777c821cbef3588cb2bb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3fc53af00a49c2646f7777c821cbef3588cb2bb2"}]},"branch":"refs/heads/master"},"7cdf2ed65e039ea65b23b6fa0fafa5e15faed9f1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-08-26 12:39:40.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/3"}}},"commit":{"parents":[{"commit":"e1a5f3c119c12a4e9d51691efc599577f0ce0fd6","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1a5f3c119c12a4e9d51691efc599577f0ce0fd6"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-26 12:39:25.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7cdf2ed65e039ea65b23b6fa0fafa5e15faed9f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7cdf2ed65e039ea65b23b6fa0fafa5e15faed9f1"}]},"branch":"refs/heads/master"},"0691125e7525f7dcb99f656f1f51df5b63efb3d9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-08-27 09:44:23.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/4"}}},"commit":{"parents":[{"commit":"65d7fc96b8a564619b991b642d516dfa8c0c5530","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65d7fc96b8a564619b991b642d516dfa8c0c5530"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-27 09:44:08.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0691125e7525f7dcb99f656f1f51df5b63efb3d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0691125e7525f7dcb99f656f1f51df5b63efb3d9"}]},"branch":"refs/heads/master"},"150e6630bc97fc25b1d27fa07df232e2149cd70e":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-08-28 05:45:09.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/5"}}},"commit":{"parents":[{"commit":"1abdf7731f96443254adf6106b315416283c40cc","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1abdf7731f96443254adf6106b315416283c40cc"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-28 05:44:48.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/150e6630bc97fc25b1d27fa07df232e2149cd70e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/150e6630bc97fc25b1d27fa07df232e2149cd70e"}]},"branch":"refs/heads/master"},"35327e12e6d6c4da47a1528c77c0b49df2455b6e":{"kind":"REWORK","_number":6,"created":"2019-08-28 10:45:47.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/6"}}},"commit":{"parents":[{"commit":"0c931e56c023ae6dd1f084cbd7df265ecccd9c1e","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c931e56c023ae6dd1f084cbd7df265ecccd9c1e"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-28 10:05:59.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35327e12e6d6c4da47a1528c77c0b49df2455b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35327e12e6d6c4da47a1528c77c0b49df2455b6e"}]},"branch":"refs/heads/master"},"c746d2b0e735fd7a7767089312966b53e3ef6d2c":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-29 03:21:36.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/7"}}},"commit":{"parents":[{"commit":"fc0d1b1258d42a20044504a7ee6a1791183e3dd2","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fc0d1b1258d42a20044504a7ee6a1791183e3dd2"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-29 03:18:20.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c746d2b0e735fd7a7767089312966b53e3ef6d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c746d2b0e735fd7a7767089312966b53e3ef6d2c"}]},"branch":"refs/heads/master"},"35a5fd36d30df4fec268d19a06f6e5097fb1b634":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-30 09:22:49.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/8"}}},"commit":{"parents":[{"commit":"96c195f0ffa583c4ff6352fb75f617a552b08a9d","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/96c195f0ffa583c4ff6352fb75f617a552b08a9d"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-30 09:21:54.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35a5fd36d30df4fec268d19a06f6e5097fb1b634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/35a5fd36d30df4fec268d19a06f6e5097fb1b634"}]},"branch":"refs/heads/master"},"4c0c94f1dd0d1b5468541cfa2ab9ef9a84251aa8":{"kind":"NO_CHANGE","_number":9,"created":"2019-08-30 09:26:04.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/9"}}},"commit":{"parents":[{"commit":"ee2a07b8b0cec57b32daf39811223b90ed7edab8","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee2a07b8b0cec57b32daf39811223b90ed7edab8"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-30 09:25:42.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c0c94f1dd0d1b5468541cfa2ab9ef9a84251aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4c0c94f1dd0d1b5468541cfa2ab9ef9a84251aa8"}]},"branch":"refs/heads/master"},"c15656e1bf7876e408a67b82d81f254044123731":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-08-30 10:53:24.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/10"}}},"commit":{"parents":[{"commit":"3263c8915f740ac3f311e743ddac63151055114e","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3263c8915f740ac3f311e743ddac63151055114e"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-30 10:48:44.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c15656e1bf7876e408a67b82d81f254044123731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c15656e1bf7876e408a67b82d81f254044123731"}]},"branch":"refs/heads/master"},"1ac11feeac61fe422b9289f7ada9b9aafc5e864b":{"kind":"REWORK","_number":11,"created":"2019-09-02 07:10:24.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/11"}}},"commit":{"parents":[{"commit":"c88aa44bc75c05894759446bfe83163e919cb031","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c88aa44bc75c05894759446bfe83163e919cb031"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-08-24 12:20:59.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 07:07:07.000000000","tz":0},"subject":"libvirt: Enable driver configuring PMEM namespaces","message":"libvirt: Enable driver configuring PMEM namespaces\n\nAdd one configuration option CONF.libvirt.pmem_namespaces:\n\"$LABEL:$NSNAME[|$NSNAME][,$LABEL:$NSNAME[|$NSNAME]]\"\n\ne.g. \"128G:ns0|ns1|ns2|ns3,262144MB:ns4|ns5,MEDIUM:ns6|ns7\"\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and config file.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ac11feeac61fe422b9289f7ada9b9aafc5e864b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ac11feeac61fe422b9289f7ada9b9aafc5e864b"}]},"branch":"refs/heads/master"},"d10b0018447fecd2dcfc038b627b52d2504d61e6":{"kind":"REWORK","_number":12,"created":"2019-09-02 13:00:22.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/12"}}},"commit":{"parents":[{"commit":"a04e92b81d40bba5cfa081b415411479b0138d3b","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a04e92b81d40bba5cfa081b415411479b0138d3b"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 12:58:09.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d10b0018447fecd2dcfc038b627b52d2504d61e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d10b0018447fecd2dcfc038b627b52d2504d61e6"}]},"branch":"refs/heads/master"},"da907f17e6cb50c2c6e8e93c725f4d1f21b6b7a9":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2019-09-03 06:21:41.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/13"}}},"commit":{"parents":[{"commit":"6650f2ee650f904d60edb15505eea7edb766a92b","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6650f2ee650f904d60edb15505eea7edb766a92b"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-03 06:21:27.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da907f17e6cb50c2c6e8e93c725f4d1f21b6b7a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/da907f17e6cb50c2c6e8e93c725f4d1f21b6b7a9"}]},"branch":"refs/heads/master"},"67512691ec4fbcb2e153b1f9845271f5efd9e943":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2019-09-04 12:49:50.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/14"}}},"commit":{"parents":[{"commit":"87cc71d58b2c458be853440546548fb1f6687a20","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87cc71d58b2c458be853440546548fb1f6687a20"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-04 11:40:59.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/67512691ec4fbcb2e153b1f9845271f5efd9e943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/67512691ec4fbcb2e153b1f9845271f5efd9e943"}]},"branch":"refs/heads/master"},"d1944a69ce0849c594e3450386e7800d327df094":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2019-09-04 14:58:12.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/15"}}},"commit":{"parents":[{"commit":"72ead1e17360836862e2566ed7f679102961c28b","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/72ead1e17360836862e2566ed7f679102961c28b"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-04 14:56:54.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1944a69ce0849c594e3450386e7800d327df094"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d1944a69ce0849c594e3450386e7800d327df094"}]},"branch":"refs/heads/master"},"75c37299007d40377d626e52e8a2d089c6639f52":{"kind":"NO_CHANGE","_number":16,"created":"2019-09-04 15:04:19.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/16"}}},"commit":{"parents":[{"commit":"e3eaed24d084693034e5c10c0de155bd8df36442","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3eaed24d084693034e5c10c0de155bd8df36442"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-04 15:04:05.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75c37299007d40377d626e52e8a2d089c6639f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75c37299007d40377d626e52e8a2d089c6639f52"}]},"branch":"refs/heads/master"},"796b1e87189a487817abdba1115a1e584c42f3f1":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2019-09-04 15:12:21.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/17"}}},"commit":{"parents":[{"commit":"cdc5f556173cdb1fe738af7e95569bbf31a117b6","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cdc5f556173cdb1fe738af7e95569bbf31a117b6"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-04 15:12:05.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/796b1e87189a487817abdba1115a1e584c42f3f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/796b1e87189a487817abdba1115a1e584c42f3f1"}]},"branch":"refs/heads/master"},"0d3fb42d8aadee5a6a0e155ec14352aa48b7c162":{"kind":"REWORK","_number":18,"created":"2019-09-05 10:29:58.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/18"}}},"commit":{"parents":[{"commit":"eb1366773471f004d06c51487fd670e1364bf2e6","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb1366773471f004d06c51487fd670e1364bf2e6"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-05 10:29:14.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d3fb42d8aadee5a6a0e155ec14352aa48b7c162"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d3fb42d8aadee5a6a0e155ec14352aa48b7c162"}]},"branch":"refs/heads/master"},"ceb63f35d2b7ebbdb4eb393f3b44eba015dbe2ec":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2019-09-05 11:39:10.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/19"}}},"commit":{"parents":[{"commit":"5a3ccc43a203759ba3e958a8d4f0513cbb6bc544","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5a3ccc43a203759ba3e958a8d4f0513cbb6bc544"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-05 11:38:49.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ceb63f35d2b7ebbdb4eb393f3b44eba015dbe2ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ceb63f35d2b7ebbdb4eb393f3b44eba015dbe2ec"}]},"branch":"refs/heads/master"},"9d241087998a1a1826fa64af01f427f159007a81":{"kind":"REWORK","_number":20,"created":"2019-09-06 10:17:19.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/20"}}},"commit":{"parents":[{"commit":"3c5ea26ebe69a2a94e0f447e7ce6d120a81d7cd4","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c5ea26ebe69a2a94e0f447e7ce6d120a81d7cd4"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-06 09:58:15.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d241087998a1a1826fa64af01f427f159007a81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9d241087998a1a1826fa64af01f427f159007a81"}]},"branch":"refs/heads/master"},"4f4e492f664f36323acdc9ce75a58a1f594fa117":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2019-09-06 10:44:33.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/21"}}},"commit":{"parents":[{"commit":"e7d1221ffa382a93cada01cd006965781edec2af","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7d1221ffa382a93cada01cd006965781edec2af"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-06 10:38:29.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f4e492f664f36323acdc9ce75a58a1f594fa117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4f4e492f664f36323acdc9ce75a58a1f594fa117"}]},"branch":"refs/heads/master"},"f60fe5b22524d686cf9d6731f05ef2f3fa6aa835":{"kind":"REWORK","_number":22,"created":"2019-09-09 05:45:02.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/22"}}},"commit":{"parents":[{"commit":"34317d58cee9597773f2a54c0634784e7cfd3e01","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/34317d58cee9597773f2a54c0634784e7cfd3e01"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-09 03:08:29.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f60fe5b22524d686cf9d6731f05ef2f3fa6aa835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f60fe5b22524d686cf9d6731f05ef2f3fa6aa835"}]},"branch":"refs/heads/master"},"ca51c3fa65e7d09d4aec42acc9c5b975e1a8345d":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2019-09-10 06:13:27.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/23"}}},"commit":{"parents":[{"commit":"0dc8a1b449672feb1422900e8a6ec801ee8a5aa1","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0dc8a1b449672feb1422900e8a6ec801ee8a5aa1"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-10 05:52:30.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca51c3fa65e7d09d4aec42acc9c5b975e1a8345d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca51c3fa65e7d09d4aec42acc9c5b975e1a8345d"}]},"branch":"refs/heads/master"},"6d533cd38d77bf2ef8f5e25787b89795888cfc34":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2019-09-10 14:25:18.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/24"}}},"commit":{"parents":[{"commit":"7368314db9cca88c2d1aa3c36703a5282840b805","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7368314db9cca88c2d1aa3c36703a5282840b805"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-02 09:30:53.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-10 14:25:03.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d533cd38d77bf2ef8f5e25787b89795888cfc34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d533cd38d77bf2ef8f5e25787b89795888cfc34"}]},"branch":"refs/heads/master"},"e1ab19bf0d6ab415463d5ce8029cde18856e026f":{"kind":"REWORK","_number":25,"created":"2019-09-11 08:01:21.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/25"}}},"commit":{"parents":[{"commit":"427088d4a3601ac2d14a579a08dc36bb34c1203f","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/427088d4a3601ac2d14a579a08dc36bb34c1203f"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 07:57:52.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1ab19bf0d6ab415463d5ce8029cde18856e026f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e1ab19bf0d6ab415463d5ce8029cde18856e026f"}]},"branch":"refs/heads/master"},"f76307412bd371b1c6314cf0ea276726335e39b1":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2019-09-11 08:05:20.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/26"}}},"commit":{"parents":[{"commit":"c74dfa61a03b04800c5fb992cad0d0638d484b67","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c74dfa61a03b04800c5fb992cad0d0638d484b67"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 08:05:03.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f76307412bd371b1c6314cf0ea276726335e39b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f76307412bd371b1c6314cf0ea276726335e39b1"}]},"branch":"refs/heads/master"},"ab4be1ec699bc86a7012a1cc863cc8b86b58340d":{"kind":"REWORK","_number":27,"created":"2019-09-11 11:38:19.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/27"}}},"commit":{"parents":[{"commit":"3d9b11dca1884c2c8e41f00ad3fed97fb1468fbe","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d9b11dca1884c2c8e41f00ad3fed97fb1468fbe"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 11:38:03.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab4be1ec699bc86a7012a1cc863cc8b86b58340d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab4be1ec699bc86a7012a1cc863cc8b86b58340d"}]},"branch":"refs/heads/master"},"113dff6c0b45ea1ee68ea1fb501b23c4f253038f":{"kind":"REWORK","_number":28,"created":"2019-09-11 13:27:13.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/28"}}},"commit":{"parents":[{"commit":"3d9b11dca1884c2c8e41f00ad3fed97fb1468fbe","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d9b11dca1884c2c8e41f00ad3fed97fb1468fbe"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 13:07:33.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/113dff6c0b45ea1ee68ea1fb501b23c4f253038f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/113dff6c0b45ea1ee68ea1fb501b23c4f253038f"}]},"branch":"refs/heads/master"},"555363c27d99746cb96cfc06e3790ada0986ad4f":{"kind":"REWORK","_number":29,"created":"2019-09-11 18:52:20.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/29"}}},"commit":{"parents":[{"commit":"b58347172b1edfd7d25d0f8e0efa7abc2ddd93a8","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b58347172b1edfd7d25d0f8e0efa7abc2ddd93a8"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 18:44:15.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/555363c27d99746cb96cfc06e3790ada0986ad4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/555363c27d99746cb96cfc06e3790ada0986ad4f"}]},"branch":"refs/heads/master"},"bf1546d476c28675d66fe4f7f1735c727c283c51":{"kind":"REWORK","_number":30,"created":"2019-09-12 03:12:42.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/30"}}},"commit":{"parents":[{"commit":"2e5edb87ff0bbf496a56d293fac2d49b102351b3","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2e5edb87ff0bbf496a56d293fac2d49b102351b3"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-12 02:26:38.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf1546d476c28675d66fe4f7f1735c727c283c51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf1546d476c28675d66fe4f7f1735c727c283c51"}]},"branch":"refs/heads/master"},"b49bfd88f1346b17baf0269e2b4f8e82b8761d9a":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2019-09-12 06:28:05.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/31"}}},"commit":{"parents":[{"commit":"17d4f019cb715b1a774c58ce1a4af04c6bc78f4d","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/17d4f019cb715b1a774c58ce1a4af04c6bc78f4d"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-12 06:22:31.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b49bfd88f1346b17baf0269e2b4f8e82b8761d9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b49bfd88f1346b17baf0269e2b4f8e82b8761d9a"}]},"branch":"refs/heads/master"},"e2b2c5175fab964f05d005fa8eb0fd9051cf650d":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2019-09-12 07:14:54.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/32"}}},"commit":{"parents":[{"commit":"2955368d4e185e620325d4eeb1a93510c9dbd53a","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2955368d4e185e620325d4eeb1a93510c9dbd53a"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-12 07:13:24.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e2b2c5175fab964f05d005fa8eb0fd9051cf650d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e2b2c5175fab964f05d005fa8eb0fd9051cf650d"}]},"branch":"refs/heads/master"},"0eb722767992727fcc67b453703238672cb7f044":{"kind":"NO_CHANGE","_number":33,"created":"2019-09-12 08:57:12.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/33"}}},"commit":{"parents":[{"commit":"ce14f7a6ba0f979a18aed285600fd41162618e31","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce14f7a6ba0f979a18aed285600fd41162618e31"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-12 08:56:44.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0eb722767992727fcc67b453703238672cb7f044"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0eb722767992727fcc67b453703238672cb7f044"}]},"branch":"refs/heads/master"},"bf004ccd6e14c780a19764fe6a7be6f49ccf3d95":{"kind":"REWORK","_number":34,"created":"2019-09-12 21:00:20.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/53/678453/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/34"}}},"commit":{"parents":[{"commit":"ce14f7a6ba0f979a18aed285600fd41162618e31","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ce14f7a6ba0f979a18aed285600fd41162618e31"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-09-12 20:59:39.000000000","tz":-300},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf004ccd6e14c780a19764fe6a7be6f49ccf3d95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf004ccd6e14c780a19764fe6a7be6f49ccf3d95"}]},"branch":"refs/heads/master"},"3978b99d8f37643a3b32f6b7ead9274c454095a1":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2019-09-13 08:55:20.000000000","uploader":{"_account_id":23598,"name":"Zhong Luyao","email":"luyao.zhong@intel.com","username":"ZhongLuyao"},"ref":"refs/changes/53/678453/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/678453/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/678453/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/678453/35"}}},"commit":{"parents":[{"commit":"04231112d96d80bf85892e8717cf290681ba9a46","subject":"Claim resources in resource tracker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04231112d96d80bf85892e8717cf290681ba9a46"}]}],"author":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-11 03:50:21.000000000","tz":0},"committer":{"name":"LuyaoZhong","email":"luyao.zhong@intel.com","date":"2019-09-13 08:50:35.000000000","tz":0},"subject":"libvirt: Enable driver discovering PMEM namespaces","message":"libvirt: Enable driver discovering PMEM namespaces\n\nAdd related functions to enable libvirt driver to get PMEM\nnamespaces(VPMEMs) info from host and configuration.\n\nnote: PMEM namespaces configuration is not supported yet,\n      so libvirt driver will not get any PMEM namespaces\n      from host.\n\nChange-Id: Idd49b0c70caedfcd42420ffa2ac926a6087d406e\nPartially-Implements: blueprint virtual-persistent-memory\nCo-Authored-By: He Jie Xu \u003chejie.xu@intel.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3978b99d8f37643a3b32f6b7ead9274c454095a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3978b99d8f37643a3b32f6b7ead9274c454095a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
