)]}'
{"id":"openstack%2Fnova~68658","triplet_id":"openstack%2Fnova~master~I78aa2f4243899db4f4941e77014a7e18e27fc63e","project":"openstack/nova","branch":"master","topic":"1251590/libvirt-Fix-root-disk-leak-in-live-mig","hashtags":[],"change_id":"I78aa2f4243899db4f4941e77014a7e18e27fc63e","subject":"libvirt: Fix root disk leak in live mig","status":"MERGED","created":"2014-01-23 15:12:58.000000000","updated":"2014-01-25 12:18:57.000000000","submitted":"2014-01-25 12:18:53.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"204c454965b1aa7eb2f4cf18245422512dea4840","_number":68658,"virtual_id_number":68658,"owner":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":2,"date":"2014-01-25 12:18:53.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":1,"date":"2014-01-24 16:49:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"all":[{"value":2,"date":"2014-01-24 18:03:00.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":2,"date":"2014-01-24 18:52:09.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"all":[{"value":0,"date":"2014-01-24 18:03:00.000000000","_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":1,"date":"2014-01-24 23:17:07.000000000","_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":1,"date":"2014-01-24 18:52:09.000000000","_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"value":0,"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"value":0,"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"value":0,"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"value":0,"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},{"value":0,"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-01-23 16:31:55.000000000","updated_by":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"reviewer":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"state":"REVIEWER"},{"updated":"2014-01-24 16:09:27.000000000","updated_by":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"reviewer":{"_account_id":7730,"name":"Sahid Orentino Ferdjaoui","email":"sahid.ferdjaoui@industrialdiscipline.com","username":"sahid"},"state":"REVIEWER"},{"updated":"2014-01-24 16:49:03.000000000","updated_by":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-01-24 18:03:00.000000000","updated_by":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"reviewer":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"state":"REVIEWER"},{"updated":"2014-01-24 18:52:09.000000000","updated_by":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"reviewer":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"state":"REVIEWER"},{"updated":"2014-01-24 23:17:07.000000000","updated_by":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"reviewer":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"state":"REVIEWER"},{"updated":"2014-01-25 12:18:53.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"031b721a0fc03a878000b14e577ecaa24937bafc","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-01-23 15:14:09.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\napproved in advance on launchpad, but let\u0027s wait for check to pass before approving","accounts_in_message":[],"_revision_number":1},{"id":"28b54eee2c5319d8c811c9a10454c3074bf7a89a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-23 15:32:15.000000000","message":"Patch Set 1: Works for me\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/67/67824/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/474609f/131007_devstack_export.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_devstack_150/fdf1992/datasets_devstack_150.log : No migrations run in 4m 41s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_user_001/68ceb0b/user_001.log : SUCCESS in 17m 10s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/4444bfa/131007_devstack_export.log : SUCCESS in 5m 25s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_150/e8a14c3/datasets_devstack_150.log : No migrations run in 5m 23s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_001/0155d35/user_001.log : SUCCESS in 18m 44s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_user_002/a2a9ba3/user_002.log : SUCCESS in 9m 14s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_002/a45ac41/user_002.log : SUCCESS in 8m 31s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"a588e44e59df05b8f1af631dbf1e4c86bad46a6b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-23 16:15:22.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/6d8a828 : SUCCESS in 7m 14s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/ada8a87/doc/build/html/ : SUCCESS in 6m 52s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/40cc956 : SUCCESS in 17m 02s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/856dab1 : SUCCESS in 13m 49s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/11842bd : FAILURE in 1h 02m 00s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/dcd44de : SUCCESS in 55m 21s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/593c219 : SUCCESS in 31m 50s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/6c1689c : SUCCESS in 14m 36s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/9f7874b : SUCCESS in 12m 50s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/d5b562c : SUCCESS in 32m 38s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/b2155e6 : SUCCESS in 9m 55s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/7388e26 : SUCCESS in 15m 02s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"ac8b6cd335a5a894b82e50f9656b9c4a5eb5f422","author":{"_account_id":9311,"name":"Tristan Cacqueray","email":"tdecacqu@redhat.com","username":"tristanC"},"date":"2014-01-23 16:31:55.000000000","message":"Patch Set 1:\n\nrecheck bug 1270702","accounts_in_message":[],"_revision_number":1},{"id":"950b613af7601a67c496da3f1bc5022908ab0a40","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-23 16:50:42.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/0bc0aa5/131007_devstack_export.log : SUCCESS in 5m 04s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_devstack_150/7b5a096/datasets_devstack_150.log : No migrations run in 4m 57s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_user_001/ab88c0a/user_001.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/da82dd8/131007_devstack_export.log : SUCCESS in 5m 36s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_150/cfdcf9c/datasets_devstack_150.log : No migrations run in 4m 31s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_001/1d4d141/user_001.log : SUCCESS in 18m 23s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/62/62896/11/check/gate-real-db-upgrade_nova_mysql_user_002/82654af/user_002.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_002/9a04b2a/user_002.log : SUCCESS in 8m 17s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"b24e5f093838fec999c509fbb0c9dc697217581c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-23 17:41:49.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/bd5c7af : SUCCESS in 7m 36s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/6c3468c/doc/build/html/ : SUCCESS in 7m 49s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/106c390 : SUCCESS in 19m 44s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/7fe17eb : SUCCESS in 15m 56s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/c877560 : FAILURE in 1h 00m 35s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/28bc051 : SUCCESS in 1h 00m 58s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/49ae011 : SUCCESS in 33m 19s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/eae5adc : SUCCESS in 12m 02s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/9e7fc9c : SUCCESS in 12m 24s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/74bb3fb : SUCCESS in 37m 25s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/fbb8b8e : SUCCESS in 10m 43s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/e906c15 : SUCCESS in 15m 15s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"cd7263d06b6f80829e4de5f8d62529ad99088082","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2014-01-23 17:42:23.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\n(One of the advance approvers)","accounts_in_message":[],"_revision_number":1},{"id":"d7f13c96c00e4d85c438e8cc0aab5e5050cb9b14","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-01-23 18:50:22.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"d2a646b8983ae47044dc37e0e38f5a3aa04aa13e","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-23 19:08:31.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/4636282/131007_devstack_export.log : SUCCESS in 5m 59s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_devstack_150/91cb3d8/datasets_devstack_150.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_user_001/29437c1/user_001.log : SUCCESS in 17m 45s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/0a37460/131007_devstack_export.log : SUCCESS in 5m 28s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_150/967d03d/datasets_devstack_150.log : No migrations run in 4m 35s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_001/9db672d/user_001.log : SUCCESS in 17m 22s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/62/62896/11/check/gate-real-db-upgrade_nova_mysql_user_002/82654af/user_002.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_002/86f9f81/user_002.log : SUCCESS in 8m 52s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"0ebcb5a4cf793e9bbfc7ed7f060af9101fbddb12","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-23 20:00:23.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/e051902 : SUCCESS in 10m 53s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/533aee0/doc/build/html/ : SUCCESS in 7m 01s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/3db725b : SUCCESS in 15m 39s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/c05755b : SUCCESS in 12m 24s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/ef0227a : SUCCESS in 57m 00s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/d9ff808 : SUCCESS in 57m 05s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/a46f614 : FAILURE in 33m 09s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/cfe7f77 : SUCCESS in 15m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/b1309bc : SUCCESS in 14m 39s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/6ccdab6 : SUCCESS in 34m 52s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/d0c43d0 : SUCCESS in 10m 51s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/a333862 : SUCCESS in 16m 55s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"ef838f7fadc9a3e40a56fa4f09a56c557d835867","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-01-23 20:13:03.000000000","message":"Patch Set 1:\n\nI noticed tempest failed, I think you hit bug(s):\n\n- https://bugs.launchpad.net/bugs/1268732\n- https://bugs.launchpad.net/bugs/1269940\n- https://bugs.launchpad.net/bugs/1210483\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1268732\n\nFor a code review which has been approved but failed to merge,\nyou can reverify by leaving a comment like this:\n\n    reverify bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"62d9a4e069c244c8c5dbc179da4ff076b3f5bc81","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-01-23 21:23:49.000000000","message":"Patch Set 1:\n\nrecheck bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"6534779b0c22e82d0a834fc439e42438bbb01304","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-23 22:05:44.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/67/67824/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/474609f/131007_devstack_export.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_devstack_150/91cb3d8/datasets_devstack_150.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_user_001/ab88c0a/user_001.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/055c37a/131007_devstack_export.log : SUCCESS in 5m 38s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_150/6cfd86e/datasets_devstack_150.log : No migrations run in 4m 20s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_001/338cfd0/user_001.log : SUCCESS in 16m 53s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/62/62896/11/check/gate-real-db-upgrade_nova_mysql_user_002/82654af/user_002.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_002/534229f/user_002.log : SUCCESS in 7m 58s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"707c033cb50de02d9c734c9a403e7ba7c0eecdfa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-23 22:43:43.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/ef21624 : SUCCESS in 7m 42s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/a27a4bc/doc/build/html/ : SUCCESS in 8m 08s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/4da3ae1 : SUCCESS in 22m 35s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/7dc7325 : SUCCESS in 14m 50s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/fc4b783 : FAILURE in 1h 05m 08s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/e27155b : SUCCESS in 59m 24s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/bf172b0 : SUCCESS in 30m 59s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/2e4b3cb : SUCCESS in 14m 42s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/19a3127 : SUCCESS in 12m 59s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/9176c98 : SUCCESS in 38m 11s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/3546007 : SUCCESS in 12m 23s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/5a0c0a2 : SUCCESS in 14m 58s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"a10544c5bfaf43835a33b9376d70b1acc88a11a5","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-01-23 23:11:38.000000000","message":"Patch Set 1:\n\nI noticed tempest failed, I think you hit bug(s):\n\n- https://bugs.launchpad.net/bugs/1268732\n- https://bugs.launchpad.net/bugs/1269940\n- https://bugs.launchpad.net/bugs/1210483\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1268732\n\nFor a code review which has been approved but failed to merge,\nyou can reverify by leaving a comment like this:\n\n    reverify bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"3370cbf5b7098502bf4f2e8c7b0be6d4a4d14e06","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-01-23 23:39:41.000000000","message":"Patch Set 1:\n\nrecheck bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"061d558a59a206ce8691beb320f848c7343457a7","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-23 23:57:19.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/3b2b1be/131007_devstack_export.log : SUCCESS in 5m 29s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_devstack_150/bb27680/datasets_devstack_150.log : No migrations run in 4m 39s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_user_001/a5d45d7/user_001.log : SUCCESS in 16m 43s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/3333854/131007_devstack_export.log : SUCCESS in 5m 23s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_150/0d24046/datasets_devstack_150.log : No migrations run in 4m 35s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_001/7280fb3/user_001.log : SUCCESS in 17m 13s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_mysql_user_002/0c91b7f/user_002.log : SUCCESS in 8m 45s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_002/8eb21bf/user_002.log : SUCCESS in 7m 56s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"bee1f9fc9a4bf93d1468ae3cda566679299daf62","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-24 01:33:11.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/a20821b : SUCCESS in 8m 01s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/60ea3f6/doc/build/html/ : SUCCESS in 7m 35s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/33fe33a : SUCCESS in 16m 05s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/0b41dcf : SUCCESS in 14m 55s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/834e591 : FAILURE in 58m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/e382fcc : SUCCESS in 1h 02m 58s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/0c760fc : SUCCESS in 33m 04s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/5901aea : SUCCESS in 12m 13s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/6993015 : SUCCESS in 14m 36s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/2210155 : SUCCESS in 34m 11s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/dce3a11 : SUCCESS in 11m 13s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/438e8b6 : SUCCESS in 14m 43s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"d2491285702d3464c4e028b2b51271b6f12b0760","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-01-24 10:14:00.000000000","message":"Patch Set 1:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":1},{"id":"10f10aa6405d40d97655b027ee113a65a15af46a","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-24 10:41:44.000000000","message":"Patch Set 1:\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/67/67824/1/check/gate-real-db-upgrade_nova_mysql_devstack_131007/474609f/131007_devstack_export.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_devstack_150/91cb3d8/datasets_devstack_150.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_user_001/ab88c0a/user_001.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_131007/b58f4d3/131007_devstack_export.log : SUCCESS in 5m 22s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_devstack_150/970ef84/datasets_devstack_150.log : No migrations run in 4m 39s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_001/2a8aa35/user_001.log : SUCCESS in 18m 04s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/62/62896/11/check/gate-real-db-upgrade_nova_mysql_user_002/82654af/user_002.log : No migrations run in 0s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/1/check/gate-real-db-upgrade_nova_percona_user_002/3cb9711/user_002.log : SUCCESS in 8m 28s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"4a3f94c85c99e3d581379a803fdb571cba278843","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-24 11:14:29.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/50a995d : FAILURE in 7m 59s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/cefee8c/doc/build/html/ : SUCCESS in 7m 28s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/14855f5 : SUCCESS in 15m 32s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/4c397ed : SUCCESS in 14m 47s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/9eb54dc : SUCCESS in 55m 24s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/196c611 : SUCCESS in 54m 02s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/c275f54 : SUCCESS in 29m 52s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/b3cd89c : SUCCESS in 11m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/6f48b31 : SUCCESS in 15m 27s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/eba8cda : SUCCESS in 34m 17s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/54f1fff : SUCCESS in 12m 40s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/793e8f2 : FAILURE in 14m 27s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"108a41ed0295952fa79a56fce0933e52b83d0366","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-01-24 11:14:50.000000000","message":"Patch Set 1:\n\nI noticed tempest failed, I think you hit bug(s):\n\n- https://bugs.launchpad.net/bugs/1268732\n- https://bugs.launchpad.net/bugs/1269940\n- https://bugs.launchpad.net/bugs/1210483\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1268732\n\nFor a code review which has been approved but failed to merge,\nyou can reverify by leaving a comment like this:\n\n    reverify bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"ed858a9302d47a0d146534e5c297014b36e7d23d","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-01-24 13:25:42.000000000","message":"Patch Set 1:\n\nrecheck bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"3100f417c311041d6c1879ab12a24bb96ea79d8c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-24 14:33:24.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/1/check/gate-nova-pep8/9c81c0c : FAILURE in 7m 34s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/1/check/gate-nova-docs/c4def9b/doc/build/html/ : SUCCESS in 7m 48s\n- gate-nova-python26 http://logs.openstack.org/58/68658/1/check/gate-nova-python26/29560b4 : SUCCESS in 15m 23s\n- gate-nova-python27 http://logs.openstack.org/58/68658/1/check/gate-nova-python27/c63e72d : SUCCESS in 14m 37s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-full/6bed97e : SUCCESS in 52m 48s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-postgres-full/f5fb9d5 : SUCCESS in 58m 49s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/1/check/check-tempest-dsvm-neutron/37958bf : SUCCESS in 29m 57s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-large-ops/d9fcc57 : SUCCESS in 14m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/1/check/gate-tempest-dsvm-neutron-large-ops/5d41e72 : SUCCESS in 15m 10s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/1/check/check-grenade-dsvm/4cb8d07 : SUCCESS in 31m 18s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/1/check/gate-devstack-dsvm-cells/aad91f6 : SUCCESS in 10m 54s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/1/check/gate-nova-pylint/0c61cba : FAILURE in 14m 44s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"fbe66804bff83a513168c39950ea5930e8c8293e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-01-24 14:33:39.000000000","message":"Patch Set 1:\n\nI noticed tempest failed, I think you hit bug(s):\n\n- https://bugs.launchpad.net/bugs/1268732\n- https://bugs.launchpad.net/bugs/1269940\n- https://bugs.launchpad.net/bugs/1210483\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1268732\n\nFor a code review which has been approved but failed to merge,\nyou can reverify by leaving a comment like this:\n\n    reverify bug 1268732","accounts_in_message":[],"_revision_number":1},{"id":"d937f492a39d81ca8f4075824ea8e07b4a8f64e1","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-01-24 16:08:26.000000000","message":"Patch Set 1: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"9b3cda1cc8a53200ba851fc989135d698bf40f4f","author":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"date":"2014-01-24 16:28:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2cbc5943e1d96644a4e645ac6d4e40775779fe39","author":{"_account_id":9578,"name":"DB Datasets CI","email":"turbo-hipster@lists.rcbops.com","username":"turbo-hipster","tags":["SERVICE_USER"]},"date":"2014-01-24 16:49:03.000000000","message":"Patch Set 2: Works for me\n\nDatabase migration testing successful.\n\n- gate-real-db-upgrade_nova_mysql_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/2/check/gate-real-db-upgrade_nova_mysql_devstack_131007/33573ed/131007_devstack_export.log : SUCCESS in 5m 31s\n- gate-real-db-upgrade_nova_mysql_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/2/check/gate-real-db-upgrade_nova_mysql_devstack_150/21e4f1a/datasets_devstack_150.log : No migrations run in 4m 23s (non-voting)\n- gate-real-db-upgrade_nova_mysql_user_001 http://www.rcbops.com/turbo_hipster/results/62/62896/12/check/gate-real-db-upgrade_nova_mysql_user_001/ab88c0a/user_001.log : SUCCESS in 0s\n- gate-real-db-upgrade_nova_percona_devstack_131007 http://www.rcbops.com/turbo_hipster/results/68/68658/2/check/gate-real-db-upgrade_nova_percona_devstack_131007/670b1d9/131007_devstack_export.log : SUCCESS in 5m 14s\n- gate-real-db-upgrade_nova_percona_devstack_150 http://www.rcbops.com/turbo_hipster/results/68/68658/2/check/gate-real-db-upgrade_nova_percona_devstack_150/8cb3d6b/datasets_devstack_150.log : No migrations run in 5m 07s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_001 http://www.rcbops.com/turbo_hipster/results/68/68658/2/check/gate-real-db-upgrade_nova_percona_user_001/59f80d4/user_001.log : SUCCESS in 19m 39s\n- gate-real-db-upgrade_nova_mysql_user_002 http://www.rcbops.com/turbo_hipster/results/68/68658/2/check/gate-real-db-upgrade_nova_mysql_user_002/08cdd5e/user_002.log : SUCCESS in 8m 39s (non-voting)\n- gate-real-db-upgrade_nova_percona_user_002 http://www.rcbops.com/turbo_hipster/results/67/67855/12/check/gate-real-db-upgrade_nova_percona_user_002/bcd1bca/user_002.log : SUCCESS in 0s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"b836f748928be6ca46d59107576aa80ea21bc931","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-24 17:39:24.000000000","message":"Patch Set 2: Works for me\n\nBuild succeeded.\n\n- gate-nova-pep8 http://logs.openstack.org/58/68658/2/check/gate-nova-pep8/a1159c2 : SUCCESS in 7m 34s\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/2/check/gate-nova-docs/da4fde0/doc/build/html/ : SUCCESS in 7m 24s\n- gate-nova-python26 http://logs.openstack.org/58/68658/2/check/gate-nova-python26/ed805ee : SUCCESS in 20m 42s\n- gate-nova-python27 http://logs.openstack.org/58/68658/2/check/gate-nova-python27/925b48b : SUCCESS in 15m 21s\n- check-tempest-dsvm-full http://logs.openstack.org/58/68658/2/check/check-tempest-dsvm-full/a57d102 : SUCCESS in 59m 24s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/2/check/check-tempest-dsvm-postgres-full/082aae4 : SUCCESS in 52m 59s\n- check-tempest-dsvm-neutron http://logs.openstack.org/58/68658/2/check/check-tempest-dsvm-neutron/2c5f254 : SUCCESS in 32m 56s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/2/check/gate-tempest-dsvm-large-ops/9924862 : SUCCESS in 16m 18s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/2/check/gate-tempest-dsvm-neutron-large-ops/a6d574c : SUCCESS in 13m 45s\n- check-grenade-dsvm http://logs.openstack.org/58/68658/2/check/check-grenade-dsvm/556131d : SUCCESS in 31m 29s\n- gate-devstack-dsvm-cells http://logs.openstack.org/58/68658/2/check/gate-devstack-dsvm-cells/831f9e8 : SUCCESS in 12m 15s (non-voting)\n- gate-nova-pylint http://logs.openstack.org/58/68658/2/check/gate-nova-pylint/6caed83 : SUCCESS in 15m 46s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"da2e36321f5871f15e9bcfff33c1e48f112f4a8a","author":{"_account_id":5441,"name":"Andrew Laski","email":"andrew@lascii.com","username":"alaski"},"date":"2014-01-24 18:03:00.000000000","message":"Patch Set 2: Looks good to me (core reviewer)\n\nRetaining the pre-approval.  But there are two new code paths and only one is tested, and it would be nice if both were.","accounts_in_message":[],"_revision_number":2},{"id":"4c032578588cbea472308207368988a457c13e84","author":{"_account_id":1561,"name":"Russell Bryant","email":"rbryant@redhat.com","username":"russellb"},"date":"2014-01-24 18:52:09.000000000","message":"Patch Set 2: Looks good to me (core reviewer); Approved\n\n","accounts_in_message":[],"_revision_number":2},{"id":"57ff4f25aa4ce66d2506c75ffae1496112afdbaa","author":{"_account_id":1,"name":"James E. Blair","email":"jim@acmegating.com","username":"corvus"},"date":"2014-01-24 23:17:07.000000000","message":"Patch Set 2: Approved\n\nReapproving due to Zuul bug","accounts_in_message":[],"_revision_number":2},{"id":"776c622c369b94ea08b33c17f9f46733a3f5dcd8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-24 23:18:13.000000000","message":"Patch Set 2: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"ea8605ac081eb09602b15c0ad7075f713597971d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-25 12:18:53.000000000","message":"Patch Set 2: Verified\n\nBuild succeeded.\n\n- gate-nova-docs http://docs-draft.openstack.org/58/68658/2/gate/gate-nova-docs/60f902f/doc/build/html/ : SUCCESS in 7m 09s\n- gate-nova-pep8 http://logs.openstack.org/58/68658/2/gate/gate-nova-pep8/f718936 : SUCCESS in 7m 23s\n- gate-nova-python26 http://logs.openstack.org/58/68658/2/gate/gate-nova-python26/ffd5cf9 : SUCCESS in 20m 29s\n- gate-nova-python27 http://logs.openstack.org/58/68658/2/gate/gate-nova-python27/8a9882b : SUCCESS in 14m 49s\n- gate-tempest-dsvm-full http://logs.openstack.org/58/68658/2/gate/gate-tempest-dsvm-full/f607bb9 : SUCCESS in 49m 51s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/58/68658/2/gate/gate-tempest-dsvm-postgres-full/15c464f : SUCCESS in 48m 34s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/58/68658/2/gate/gate-tempest-dsvm-neutron/a9e5761 : SUCCESS in 30m 28s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/58/68658/2/gate/gate-tempest-dsvm-large-ops/f1870f7 : SUCCESS in 14m 24s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/58/68658/2/gate/gate-tempest-dsvm-neutron-large-ops/6f347a0 : SUCCESS in 14m 27s\n- gate-grenade-dsvm http://logs.openstack.org/58/68658/2/gate/gate-grenade-dsvm/e17d67c : SUCCESS in 36m 59s\n","accounts_in_message":[],"_revision_number":2},{"id":"8a9ff179883ce541d4b54bf8998990cbe704e207","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-01-25 12:18:57.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b0d36683fe064b32cbef013e1c0c46bd018ab9a1","revisions":{"69f2a1aadfc35873377f0b492b5013b5e7e74c7f":{"kind":"REWORK","_number":1,"created":"2014-01-23 15:12:58.000000000","uploader":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"ref":"refs/changes/58/68658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/68658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/68658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/68658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/68658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/68658/1"}}},"commit":{"parents":[{"commit":"80efcae3a3a46a1c2960bbc0460d0e42112b60a0","subject":"Merge \"Fix rbd backend not working for none admin ceph user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/80efcae3a3a46a1c2960bbc0460d0e42112b60a0"}]}],"author":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2013-12-10 16:43:17.000000000","tz":60},"committer":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2014-01-23 09:26:10.000000000","tz":60},"subject":"libvirt: Fix root disk leak in live mig","message":"libvirt: Fix root disk leak in live mig\n\nThis patch makes sure that i_create_images_and_backing method of the\nlibvirt driver (called in several places, but most problematic one is\nthe call in the pre_live_migration method) creates all the files the\ninstance needs that are not present.\n\nPrioir to this patch - the method would only attempt to download the\nimage, and if it did so with the path of the ephemeral drives, it could\nexpose the image to other users as an ephemeral devices. See the related\nbug for more detaiis.\n\nAfter this patch - we properly distinguish between image, ephemeral and\nswap files, and make sure that the imagebackend does the correct thing.\n\nCloses-bug: #1251590\n\nCo-authored-by: Loganathan Parthipan \u003cparthipan@hp.com\u003e\n\nChange-Id: I78aa2f4243899db4f4941e77014a7e18e27fc63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69f2a1aadfc35873377f0b492b5013b5e7e74c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69f2a1aadfc35873377f0b492b5013b5e7e74c7f"}]},"branch":"refs/heads/master"},"b0d36683fe064b32cbef013e1c0c46bd018ab9a1":{"kind":"REWORK","_number":2,"created":"2014-01-24 16:28:57.000000000","uploader":{"_account_id":5511,"name":"Nikola Dipanov","email":"ndipanov@redhat.com","username":"ndipanov"},"ref":"refs/changes/58/68658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/68658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/68658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/68658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/68658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/68658/2"}}},"commit":{"parents":[{"commit":"f579affb5058cdfef9ce606d62cf64345b87cf55","subject":"Merge \"Create common method for MTU treatment\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f579affb5058cdfef9ce606d62cf64345b87cf55"}]}],"author":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2013-12-10 16:43:17.000000000","tz":60},"committer":{"name":"Nikola Dipanov","email":"ndipanov@redhat.com","date":"2014-01-24 16:28:27.000000000","tz":60},"subject":"libvirt: Fix root disk leak in live mig","message":"libvirt: Fix root disk leak in live mig\n\nThis patch makes sure that i_create_images_and_backing method of the\nlibvirt driver (called in several places, but most problematic one is\nthe call in the pre_live_migration method) creates all the files the\ninstance needs that are not present.\n\nPrioir to this patch - the method would only attempt to download the\nimage, and if it did so with the path of the ephemeral drives, it could\nexpose the image to other users as an ephemeral devices. See the related\nbug for more detaiis.\n\nAfter this patch - we properly distinguish between image, ephemeral and\nswap files, and make sure that the imagebackend does the correct thing.\n\nCloses-bug: #1251590\n\nCo-authored-by: Loganathan Parthipan \u003cparthipan@hp.com\u003e\n\nChange-Id: I78aa2f4243899db4f4941e77014a7e18e27fc63e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0d36683fe064b32cbef013e1c0c46bd018ab9a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0d36683fe064b32cbef013e1c0c46bd018ab9a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
