)]}'
{"id":"openstack%2Fnova~686835","triplet_id":"openstack%2Fnova~master~Ibd51d6fc2d4c9fd142ba17f2f87214bb62f72440","project":"openstack/nova","branch":"master","topic":"bug/1846777","hashtags":[],"change_id":"Ibd51d6fc2d4c9fd142ba17f2f87214bb62f72440","subject":"Cache image GETs for multi-create/multi-BDM requests","status":"MERGED","created":"2019-10-04 20:39:08.000000000","updated":"2019-10-10 09:10:51.000000000","submitted":"2019-10-09 23:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686835-1570663959752-0f5a3271","meta_rev_id":"cb2e8898ab941fc10b2af9acee9caf14218c4c92","_number":686835,"virtual_id_number":686835,"owner":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-10-09 20:00:23.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2019-10-09 16:33:55.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-09 23:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-10 09:10:51.000000000","post_submit":true,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-09 22:27:11.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-10-09 15:58:13.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-09 16:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2019-10-09 15:57:18.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-09 16:09:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-05 03:13:32.000000000","updated_by":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-08 17:50:51.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-10-08 21:46:53.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 03:04:40.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-10-09 10:38:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 15:58:13.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 16:09:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-09 16:33:55.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-10-09 20:00:23.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 22:27:11.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-09 23:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-10 09:10:51.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7858f3ec0ad2e2eb34c1b6f1d9bfafa2fd5a0a17","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-04 20:39:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c82f2f65798c4636a3f1f6556ae3963bc0a649a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-04 20:39:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66c6c92512829f0a32604c8f8f78659c2def6aec","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-04 21:10:50.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/35/686835/1/check/pmem-tempest-plugin-filtered/22ea0e0/ : SUCCESS in 30m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9ea24344ae899dbeb9ab1b9450e9a33148907bb4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-04 21:42:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2af1eee34f12540bd31654e9c7e74e56cf39a2a4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-04 21:42:34.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19fff878015fec80855fa4d2954825c67740c81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-05 03:10:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/997f424484264f1d911eced1b6960fad : SUCCESS in 1h 05m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/14810fd94a9642b39c3ba460d035b2be : SUCCESS in 1h 20m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/339d46e60f5b47d9bab9fd4ee35585a6 : SUCCESS in 15m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5565fd624fa4d8598d344ce3f17f0ff : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d79f6ab29f8a46bca413157d7ac62093 : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/78a39bbe2a454104badb7047aca7a5fd : SUCCESS in 12m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28ef87d673534128aaf3d4aceccefe19 : SUCCESS in 11m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e484552148bd4daabf659ae0deacaa26 : SUCCESS in 13m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ce87247e46740c7b6b84130701c7c5d : SUCCESS in 10m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8d5953f8fd004341bf44a0c52365054d : SUCCESS in 55m 16s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/743d9e704f574353aad7ab43c5bb1698 : SUCCESS in 1h 12m 11s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2920d65d68594d5eba359e3b7b919901 : SUCCESS in 1h 09m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c897b28d93604867bfc0a3bf7701541c : SUCCESS in 1h 12m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/54976777a47c4ac588d2fdd156af1c4a : SUCCESS in 57m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c2a6e260e7684ab69d23d051efeb3ce9 : SUCCESS in 1h 30m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/d2c8c54aabd84ec7b874b9b1b2057c1a : SUCCESS in 1h 16m 18s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/95999e1a79cf4319831d5ac23d18314d : FAILURE in 22m 28s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bea4f9c50024baeb16d9004ec125545 : SUCCESS in 16m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6a06f762ac94fafa35db1eb8e3d73e1 : SUCCESS in 2h 17m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1cd8bebd24834cb3a3fc142250826750 : SUCCESS in 1h 08m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ba2eee74b1e07da8318cc4781399214ba7a8c6e3","author":{"_account_id":16128,"name":"IBM PowerVM CI","email":"powervmci@linux.vnet.ibm.com","username":"powervmci","tags":["SERVICE_USER"]},"date":"2019-10-05 03:13:32.000000000","message":"Patch Set 1:\n\nBuild failed. Comment \u0027powervm: recheck\u0027 to recheck.\n For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_PowerVM_CI\n\n- nova-out-of-tree-pvm http://184.172.12.213/35/686835/1/check/nova-out-of-tree-pvm/a2d8e97 : FAILURE in 2h 40m 08s\n- nova-in-tree-pvm http://184.172.12.213/35/686835/1/check/nova-in-tree-pvm/9693fc2 : FAILURE in 2h 05m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5c4cb5ea7a7e57599c6ac2bd9243cfd34198df4f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-05 04:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/686835/1 : SUCCESS in 2h 25m 21s","accounts_in_message":[],"_revision_number":1},{"id":"d1d777c3855bad9d0b48ddba4afe45cfe5aa0ea4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-05 12:44:59.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/1/check/tempest-dsvm-full-xenial/7d8eb57/ : SUCCESS in 2h 19m 37s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/1/check/tempest-dsvm-full-xenial-py3/b49385a/ : SUCCESS in 2h 05m 26s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/1/check/grenade-dsvm-xenial/88b0b04/ : SUCCESS in 1h 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c2f89fb2f3b96890d74a012780b126cba582b06","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-07 14:01:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d31b9a8bd2904790f23d4ab376ff5151c3ad47a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-07 14:02:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84be03bc4febcaedc028c079e3871d94c81dbc1a","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-07 14:33:30.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/35/686835/2/check/pmem-tempest-plugin-filtered/f1e3d74/ : SUCCESS in 30m 16s","accounts_in_message":[],"_revision_number":2},{"id":"9662dbe49879e0e5057844efe4a9efbfe5db63d9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-07 15:48:36.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/2/check/tempest-dsvm-full-xenial/a5271fb/ : SUCCESS in 1h 43m 19s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/2/check/tempest-dsvm-full-xenial-py3/77357a0/ : SUCCESS in 1h 36m 43s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/2/check/grenade-dsvm-xenial/ebf752e/ : SUCCESS in 1h 42m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"992fb6f851c733f446c5d6cdfd19e402bd6ee4e4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-07 15:55:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c71ae074c49251fa2c04cd581b150a5adeb72bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 16:24:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ee8bc9c1f23541fdb08bef04e40e5d7a : SUCCESS in 1h 02m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5b9eba174b964d96836b1689eccbe8e1 : SUCCESS in 1h 37m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34ad45cb8f6648a39423b70282531cf3 : SUCCESS in 19m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7731b5e6802f418d82060cbfac351b11 : SUCCESS in 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63e721a1fe3a459590e5c5bbd6c35f2a : SUCCESS in 10m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b96e83017fb04d9db69bf45e61c02015 : SUCCESS in 12m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31f459a60a044b10bc29e5c9683ee781 : SUCCESS in 11m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9de01a90b1f04fa4b50c128df860ca73 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c5b20b3f72418aa6cd72e401444543 : SUCCESS in 8m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7183963d9aec4753b023e33b620aa778 : SUCCESS in 53m 50s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/2aaca48bce3146d3b80d63f481d41792 : SUCCESS in 1h 13m 23s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/03481a6e2da640c290a93adc175e049b : SUCCESS in 1h 18m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e6d1b4bfeac4578bbf0e77f840e780d : SUCCESS in 1h 08m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7c043d9b69f34d9c86cdd58a5128da04 : SUCCESS in 48m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/568febdced524ea6958371459626ff4a : SUCCESS in 1h 46m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/b1b0891f1d494008a143d6bc14fe6510 : SUCCESS in 1h 19m 34s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/3f21bd82d4f34022b4e2dbf908e23dbb : SUCCESS in 17m 09s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6d87a63583a47f68a82abb2d2890e3a : SUCCESS in 16m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/49efcafbb9434fe0b94651af6b44c2c0 : SUCCESS in 2h 16m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3e7ac9c3af4d7893dad9a122c9bf04 : SUCCESS in 1h 16m 02s","accounts_in_message":[],"_revision_number":2},{"id":"6c4fb12c3bed24e72d4718966ab22a5ce69a14bc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-07 16:25:36.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/35/686835/2/check/EMC_VxFlexOS_NOVA/82781cb/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"f28f64f88a85b28fd234e6ed7d6f60f32115d05b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-07 16:50:54.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/686835/2 : SUCCESS in 2h 47m 53s","accounts_in_message":[],"_revision_number":2},{"id":"8f943b65de948916a70cb7e74589eb751784bbcd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-07 17:50:36.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-686835-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"861c3f23e16ac52dea9e9c7c364defc5eaa2e072","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-10-08 08:12:14.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38026 : FAILURE in 1h 44m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8f18af4d405cde208a24a6712acac2bd2fe711e9","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 15:52:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ffd3e3d9dc9e9963810b08af5991a535ab6dd3ac","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 15:55:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"48e1fdd8eba19c02f0834587dd3f8923590a38b6","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 16:57:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f98c172b303861d69bf3e39ebcdc1360646ac747","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 17:39:33.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b97d7a3e1694393c9f1f3557b630d36283ae19bc","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 17:40:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"901eef26630b2dc0b8d8ffdb5753fbbc24383205","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 17:50:51.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)\n\nNice.","accounts_in_message":[],"_revision_number":3},{"id":"3a76e89d11b41a5841cae06019f59bf408651a11","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-08 18:19:53.000000000","message":"Patch Set 3:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/35/686835/3/check/pmem-tempest-plugin-filtered/24a0cc7/ : SUCCESS in 37m 46s","accounts_in_message":[],"_revision_number":3},{"id":"18e69bc1e0915aa5d0b642b4e9288ed97e97e20a","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 20:42:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"baf3be51b1f56812c3cc12147800dfc908aa6689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-08 21:46:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/724d73e0973d4068ae8d4a9eaccdb380 : SUCCESS in 57m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d224baf31b7949eda0ba872032d6921c : SUCCESS in 1h 26m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db2a2184ad6c48da931b1732047bae79 : SUCCESS in 16m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9731b0c01368439aa52bd4cee19d9bb5 : SUCCESS in 16m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/273f07c7911949b989e62b7eb7c7e21f : SUCCESS in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70a4e8447d5f49248ae4510e2a39c4c9 : SUCCESS in 12m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cca723c59694c93a3ae46aba70da33f : SUCCESS in 15m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ff36e424b1cd4a38b26be9470610aec6 : SUCCESS in 17m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15cfdaabe0b541ef9b4daf8b3895b870 : SUCCESS in 8m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/57804bc452784e2daeb1902f6c1e4c1d : SUCCESS in 54m 29s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0f0f34e03bd345b0b16ee1f07825b5b5 : SUCCESS in 1h 18m 33s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b7fc6c9b9de0419a95c81fa6109df218 : SUCCESS in 1h 09m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/64cf953ce6be4676bf8b910e268727a2 : SUCCESS in 1h 20m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9ad6c65d807f4984bcda53c8d7b8d78d : SUCCESS in 56m 37s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0b4f94c7ad274caaad3a0bd951ea8cd3 : SUCCESS in 1h 29m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/4963803cbd9f4b9e87ebba828ebc7b3d : SUCCESS in 1h 17m 47s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/6bf9a536613447c5b3d62e59e92429b0 : SUCCESS in 17m 53s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8ac47a534824a5fb5e902eca0953835 : SUCCESS in 16m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/08a964a0fcf94b0f85f8a27111e1fc3c : SUCCESS in 2h 18m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/174b22ddbed8433fa3d05650d55de474 : SUCCESS in 1h 08m 45s","accounts_in_message":[],"_revision_number":3},{"id":"0a7d93c402e44fb1b5aff90817a4c243694f2c01","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 21:46:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/35/686835/3/check/EMC_VxFlexOS_NOVA/87ae98e/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"62f313b02ee7211eb534a449fc1fb76aa951813f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-10-09 03:04:40.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38108 : FAILURE in 50m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9a8e5c1c60b09ad769c490febdfe7ef2ee4c6455","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-09 06:49:16.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/686835/3 : SUCCESS in 2h 39m 41s","accounts_in_message":[],"_revision_number":3},{"id":"17955188da062524e766a0a3dfc5f1c965f0899a","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-09 10:38:32.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-35-686835-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"b00fb4039a93571e79b81fd3307fac854d1569c9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-09 12:10:40.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/35/686835/3/check-vote/ext-nova-zuul/038ead0 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":3},{"id":"45c511ce9e5bc01709c15b8e39fef8439d8c06e3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-09 13:53:52.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/3/check/tempest-dsvm-full-xenial/f727797/ : SUCCESS in 2h 17m 11s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/3/check/tempest-dsvm-full-xenial-py3/bd9812e/ : SUCCESS in 2h 19m 28s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/3/check/grenade-dsvm-xenial/1cba869/ : SUCCESS in 2h 22m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c98b2f2b46e1234ef6a269bfd11f9b6e5894ca9e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-09 15:57:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"500647ab5b839140beea94db994aa9d87c136665","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-09 15:57:55.000000000","message":"Topic changed from bug/1694844 to bug/1846777","accounts_in_message":[],"_revision_number":4},{"id":"ea37dc5e0e200091243991b7acd0d8d510ce1d75","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-09 15:58:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"119b536f1021b32d2732a3f8789166acaa5871fe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-09 16:09:04.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nSane","accounts_in_message":[],"_revision_number":4},{"id":"404169945551f3795d0b87ca26004e85e0bd1732","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-09 16:33:55.000000000","message":"Patch Set 4:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/35/686835/4/check/pmem-tempest-plugin-filtered/f22ddaa/ : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":4},{"id":"3dc87deb019a53c7ba77513c391375987f55d65d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 19:41:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/90032acb6f6642719d8b25fd3f8bbf97 : SUCCESS in 1h 07m 19s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/080ff14277fa40cdae80796ad41a800e : SUCCESS in 1h 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0025c199c8b743b7a349df2a0adcc872 : SUCCESS in 17m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/996ed89ea56245479f9afc4ef8982675 : SUCCESS in 14m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f366594a0e7d4820906930d791ca39d9 : SUCCESS in 12m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9127fe7dabac45ef9ddd1931278888e9 : SUCCESS in 15m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ada56cae32b94c558bba550328d7d686 : SUCCESS in 14m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7271bb5d68544eea5565965ef0e55bf : SUCCESS in 16m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a19d8317d849cdbf704ecbb53a8517 : SUCCESS in 9m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3511370316ba4adebd4654ec6b8fcf91 : SUCCESS in 59m 17s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/74ac1598480d4e818a4912f0a1b95e80 : SUCCESS in 1h 33m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2b1d4142b77343ad9707becf367b1312 : SUCCESS in 1h 25m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ce54a2c2644040e9b495b8ca9d845d2a : SUCCESS in 1h 14m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/649fc487f6e1471e9a25ae94bbe17ada : SUCCESS in 1h 05m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5b42944b2bd942eb8861977402cca8d8 : SUCCESS in 1h 39m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/e879eabeb123458da7686fa4a0f3b0e9 : SUCCESS in 1h 18m 48s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/9aed696c60964af8a742fe6392be1166 : SUCCESS in 17m 02s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1510b67595cc4b508cff97e5bd125f02 : SUCCESS in 19m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/636efae8424646738a583fdd64b30fc5 : SUCCESS in 2h 18m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22c3e3be8b8541299a3c17dc909505fc : SUCCESS in 1h 14m 14s","accounts_in_message":[],"_revision_number":4},{"id":"ab11a05fed68fb4394be21704862e123b6fdbedb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 19:42:36.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2b37bb0881b0219705c3f26851e73765f2c1f753","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-09 20:00:23.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/4/check/tempest-dsvm-full-xenial/b25285c/ : FAILURE in 2m 17s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/4/check/tempest-dsvm-full-xenial-py3/5a85f6b/ : FAILURE in 2m 15s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/35/686835/4/check/grenade-dsvm-xenial/c9f1afa/ : FAILURE in 2m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5c3e3187929180cc602f1e8a7b8a916b65b0119e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-09 22:27:11.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/686835/4 : SUCCESS in 2h 30m 41s","accounts_in_message":[],"_revision_number":4},{"id":"275b0ec60c80fb1c23067ebf3cc8c9b92eac63e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 23:32:39.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9c74f46ac79146e2bb5796599194bbec : SUCCESS in 1h 07m 30s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e572a47df6d34a3eb7935ce77fb5509c : SUCCESS in 1h 31m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f38f4e05fc34815a101560779d13b8a : SUCCESS in 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30bf6d2c7ba84c19bacc9ab3f3f9d23f : SUCCESS in 12m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2ebd02d466a6474d9d5da121e3fac2a8 : SUCCESS in 18m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a69e8ac4f5564f9c9861e5655f1dd6c0 : SUCCESS in 18m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/30a821690be74bc68e146ac34f1a2e06 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbce745ce0134a2092ad029e04cdb0d9 : SUCCESS in 10m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/14bbe38bcb0d40cca268faae9bd410c7 : SUCCESS in 1h 40m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3f008a6c5f824d819f9389c3a42a2dd5 : SUCCESS in 1h 02m 36s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/64de341dda3b44b48b21628ab81faf1b : SUCCESS in 20m 37s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1fdfaf075cb54842b5bc881d83911792 : SUCCESS in 14m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/920551b8c8f84c0d9ea071570fe3d85f : SUCCESS in 1h 44m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/6a9ac06caf324e48939304ec85372985 : SUCCESS in 1h 33m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d832dd049edd4779811267b447469152 : SUCCESS in 2h 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cd70b042c61494b9487e9933db28c3d : SUCCESS in 1h 07m 11s","accounts_in_message":[],"_revision_number":4},{"id":"9fabb647ab53b3b481fb91229937f3d742a56ae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 23:32:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ec2282e95f696bfdade3fea6b30024c4d7f259d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 23:34:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea917a61dff4216a7bacafec2c847aa : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":4},{"id":"d18e3c7918267bcea6e772dffaad62d00d49a85f","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-10-10 00:28:17.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38145 : FAILURE in 3h 37m 12s","accounts_in_message":[],"_revision_number":4},{"id":"cb2e8898ab941fc10b2af9acee9caf14218c4c92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-10 09:10:51.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/35/686835/4/check-vote/ext-nova-zuul/d13b6e6 : FAILURE in 34m 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2d84a564fbb528734776d8432a31a4bb4b40af71","revisions":{"115fc9481825552dd9e7416c9cee02be7708074d":{"kind":"REWORK","_number":1,"created":"2019-10-04 20:39:08.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/35/686835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/686835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/686835/1"}}},"commit":{"parents":[{"commit":"1bd3f1d9136cd8bb74de9727e478289e99aab237","subject":"Add boot from volume functional test with a huge request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1bd3f1d9136cd8bb74de9727e478289e99aab237"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-04 20:32:33.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-04 20:32:33.000000000","tz":-240},"subject":"Cache image GETs for multi-create/multi-BDM requests","message":"Cache image GETs for multi-create/multi-BDM requests\n\nThis builds on change I1233fbbaebd1750f572432bf74759b94408a17e3 which\nadded a functional test to create 10 servers with 255 BDMs each using\nthe same image. That scenario results in 2551 calls to the image service\nto get the same image. While an unrealistic scenario, it shows how\ninefficient this can be even in a simple multi-create request where\neach server has one or more BDMs that are using one or more of the\nsame images.\n\nThis change adds a simple cache that is used while processing the list\nof instances to create and the list of BDMs per instance that are using\na source_type of \u0027image\u0027 to validate the image. This drops the image GET\ncalls in the functional test from 2551 to 2.\n\nThe functional test verifies the cache is being used properly. Existing\nunit tests are updated for the method signature changes. Note that some\nAPI controller tests were mocking a bit too deep and ignoring all of the\nparameters to the _validate_bdm mock so rather than continue that pattern\nthe assertions are just changed to assert_called_once() since the tests\nclearly don\u0027t care about the actual parameters passed to that method.\n\nFinally, surprisingly the functional test is not noticeably faster after\nthe image cache is added which means there is an opportunity for further\nprofiling of this scenario to weed out other inefficiencies.\n\nChange-Id: Ibd51d6fc2d4c9fd142ba17f2f87214bb62f72440\nCloses-Bug: #1846777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/115fc9481825552dd9e7416c9cee02be7708074d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/115fc9481825552dd9e7416c9cee02be7708074d"}]},"branch":"refs/heads/master"},"59ba804fa0a9e8cbbd4d3a2bb8052871bf502419":{"kind":"REWORK","_number":2,"created":"2019-10-07 14:01:18.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/35/686835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/686835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/686835/2"}}},"commit":{"parents":[{"commit":"90bc173268c32fd90f5f7013c3fcdb90dd8d22a1","subject":"Add boot from volume functional test with a huge request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90bc173268c32fd90f5f7013c3fcdb90dd8d22a1"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-04 20:32:33.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-07 14:00:39.000000000","tz":-240},"subject":"Cache image GETs for multi-create/multi-BDM requests","message":"Cache image GETs for multi-create/multi-BDM requests\n\nThis builds on change I1233fbbaebd1750f572432bf74759b94408a17e3 which\nadded a functional test to create 10 servers with 255 BDMs each using\nthe same image. That scenario results in 2551 calls to the image service\nto get the same image. While an unrealistic scenario, it shows how\ninefficient this can be even in a simple multi-create request where\neach server has one or more BDMs that are using one or more of the\nsame images.\n\nThis change adds a simple cache that is used while processing the list\nof instances to create and the list of BDMs per instance that are using\na source_type of \u0027image\u0027 to validate the image. This drops the image GET\ncalls in the functional test from 2551 to 2.\n\nThe functional test verifies the cache is being used properly. Existing\nunit tests are updated for the method signature changes. Note that some\nAPI controller tests were mocking a bit too deep and ignoring all of the\nparameters to the _validate_bdm mock so rather than continue that pattern\nthe assertions are just changed to assert_called_once() since the tests\nclearly don\u0027t care about the actual parameters passed to that method.\n\nChange-Id: Ibd51d6fc2d4c9fd142ba17f2f87214bb62f72440\nCloses-Bug: #1846777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59ba804fa0a9e8cbbd4d3a2bb8052871bf502419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59ba804fa0a9e8cbbd4d3a2bb8052871bf502419"}]},"branch":"refs/heads/master"},"6c2172f2473e87c053679305cbef6f4d4a068c97":{"kind":"REWORK","_number":3,"created":"2019-10-08 17:39:33.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/35/686835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/686835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/686835/3"}}},"commit":{"parents":[{"commit":"a8d7d58400a02010eab083b8b5ff91d37a1607f2","subject":"Add boot from volume functional test with a huge request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8d7d58400a02010eab083b8b5ff91d37a1607f2"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-04 20:32:33.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-08 17:38:21.000000000","tz":-240},"subject":"Cache image GETs for multi-create/multi-BDM requests","message":"Cache image GETs for multi-create/multi-BDM requests\n\nThis builds on change I1233fbbaebd1750f572432bf74759b94408a17e3 which\nadded a functional test to create 10 servers with 255 BDMs each using\nthe two images. That scenario results in 2551 calls to the image service\nto get the images. While an unrealistic scenario, it shows how\ninefficient this can be even in a simple multi-create request where\neach server has one or more BDMs that are using one or more of the\nsame images.\n\nThis change adds a simple cache that is used while processing the list\nof instances to create and the list of BDMs per instance that are using\na source_type of \u0027image\u0027 to validate the image. This drops the image GET\ncalls in the functional test from 2551 to 3.\n\nThe functional test verifies the cache is being used properly. Existing\nunit tests are updated for the method signature changes. Note that some\nAPI controller tests were mocking a bit too deep and ignoring all of the\nparameters to the _validate_bdm mock so rather than continue that pattern\nthe assertions are just changed to assert_called_once() since the tests\nclearly don\u0027t care about the actual parameters passed to that method.\n\nChange-Id: Ibd51d6fc2d4c9fd142ba17f2f87214bb62f72440\nCloses-Bug: #1846777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c2172f2473e87c053679305cbef6f4d4a068c97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c2172f2473e87c053679305cbef6f4d4a068c97"}]},"branch":"refs/heads/master"},"2d84a564fbb528734776d8432a31a4bb4b40af71":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-10-09 15:57:18.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/35/686835/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/35/686835/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/35/686835/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/35/686835/4"}}},"commit":{"parents":[{"commit":"123357e35b47697681a29ca672121323c0074722","subject":"Add boot from volume functional test with a huge request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/123357e35b47697681a29ca672121323c0074722"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-04 20:32:33.000000000","tz":-240},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-10-09 15:52:33.000000000","tz":-240},"subject":"Cache image GETs for multi-create/multi-BDM requests","message":"Cache image GETs for multi-create/multi-BDM requests\n\nThis builds on change I1233fbbaebd1750f572432bf74759b94408a17e3 which\nadded a functional test to create 10 servers with 255 BDMs each using\nthe two images. That scenario results in 2551 calls to the image service\nto get the images. While an unrealistic scenario, it shows how\ninefficient this can be even in a simple multi-create request where\neach server has one or more BDMs that are using one or more of the\nsame images.\n\nThis change adds a simple cache that is used while processing the list\nof instances to create and the list of BDMs per instance that are using\na source_type of \u0027image\u0027 to validate the image. This drops the image GET\ncalls in the functional test from 2551 to 3.\n\nThe functional test verifies the cache is being used properly. Existing\nunit tests are updated for the method signature changes. Note that some\nAPI controller tests were mocking a bit too deep and ignoring all of the\nparameters to the _validate_bdm mock so rather than continue that pattern\nthe assertions are just changed to assert_called_once() since the tests\nclearly don\u0027t care about the actual parameters passed to that method.\n\nChange-Id: Ibd51d6fc2d4c9fd142ba17f2f87214bb62f72440\nCloses-Bug: #1846777\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d84a564fbb528734776d8432a31a4bb4b40af71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2d84a564fbb528734776d8432a31a4bb4b40af71"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
