)]}'
{"id":"openstack%2Fnova~687373","triplet_id":"openstack%2Fnova~master~I9e2be5423cc0821a628db7a68ad52bbd91264acd","project":"openstack/nova","branch":"master","topic":"kill-noauth2","hashtags":[],"change_id":"I9e2be5423cc0821a628db7a68ad52bbd91264acd","subject":"Deprecate [api]auth_strategy and noauth2","status":"MERGED","created":"2019-10-08 18:09:16.000000000","updated":"2019-10-14 21:04:02.000000000","submitted":"2019-10-14 20:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687373-1571086568030-9904e72e","meta_rev_id":"145873525a3cb5b62c94278b01a42a20b6b3b4e7","_number":687373,"virtual_id_number":687373,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-09 20:33:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2019-10-10 02:57:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2019-10-09 20:48:00.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 20:56:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-10 11:52:55.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-10 01:23:07.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-10-09 20:16:44.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-14 14:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-10-12 17:27:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-10 17:54:02.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-10-13 15:13:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-10-14 09:18:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":1,"date":"2019-10-14 03:23:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-14 14:16:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-09 01:21:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 11:57:15.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 20:16:44.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 20:33:39.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-09 20:48:00.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-10-10 01:23:07.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-10-10 02:57:25.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-10-10 11:52:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-10 17:54:02.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-10-12 17:27:22.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-10-13 15:13:29.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-10-14 03:23:19.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-10-14 09:18:42.000000000","updated_by":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"reviewer":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"state":"REVIEWER"},{"updated":"2019-10-14 14:16:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-10-14 20:56:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5f693c07699db81ecfdbb1b6b92e40dd63392c97","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 18:09:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9e9218d0c2443fb56e5f79669732c4a692ec40b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 18:10:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbbf761b6a48e8352079ee8bb8d4dbb39fc10cfe","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 18:23:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"76d8f23304c15ac823761be44fb07b2c54c350b5","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-08 18:40:10.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/73/687373/1/check/pmem-tempest-plugin-filtered/a65bce2/ : SUCCESS in 29m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a28ee6b16531ce9425432638f1bde7a5c9c11bf9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 20:33:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1bdf15ad4cd8e9f83987c6d66ebf5d2b6a5022d2","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 20:33:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"59952debf2562ae6abf1823b02b7e2671eafaeab","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 20:35:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01797ad436d0cfcc2fa8e00706b5486b85fbd390","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-08 20:40:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nRelease note brosef!","accounts_in_message":[],"_revision_number":2},{"id":"b348ac4ba09317916a62d54619266e3f067267a9","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 20:46:54.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)\n\n\u003e Release note brosef!\n\nD\u0027oh, yah.","accounts_in_message":[],"_revision_number":2},{"id":"b14a0b0768f5a2e325f2817afb221d249501220c","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-08 21:04:51.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/73/687373/2/check/pmem-tempest-plugin-filtered/5f86f70/ : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9b2001a33c7fad15a432803ee65fe67637aa647f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 21:09:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7160379a6d6c6b54cdb8df5f727571e5868574e5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-08 22:43:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"922bcedd7dd885acd3094418a79647efff4b52a4","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-08 22:44:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a464c96165e8182611e385a6a86a386ae8f438ec","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-08 23:32:31.000000000","message":"Patch Set 3:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/73/687373/3/check/pmem-tempest-plugin-filtered/801ccc8/ : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":3},{"id":"57322d4e521d760a047ad318f178ef5781120f65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 01:20:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/851a49cb71a1483ab6cdc9f212919a6e : SUCCESS in 59m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/637b8a6e909642cc9bddb1479545b058 : SUCCESS in 1h 25m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b8cd89d64ee4b72a9489bac30516757 : SUCCESS in 14m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fae4266a3e64dc5a2a21d5b934e3ead : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f9fca5c912e406db1853f1e5205978d : SUCCESS in 10m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b676e1a90a7740d5a46ce252af05600e : SUCCESS in 19m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d39d0d70d352421d88892d09b1f5cc8c : SUCCESS in 18m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9922288ccdc644d8a4c76e46dff0644a : SUCCESS in 18m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a82f6ae8380641bf9152e91f70df4eeb : SUCCESS in 12m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b8e79a4f1534d75b75b7f1486558eb8 : SUCCESS in 42m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ce8b6690f2924306b6ecff14592fe1e1 : SUCCESS in 55m 27s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0761fdb028da43c6836469b68783435a : SUCCESS in 1h 18m 55s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b89e241011a43de89ed246e0e923141 : SUCCESS in 1h 19m 59s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3a4cc427e6a943beacaee9f2ec84870f : SUCCESS in 1h 23m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2abcee1be36c43d98c113e0676b91d6b : SUCCESS in 53m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4c512aae8b7e481abd3058c7efb3d9fc : SUCCESS in 1h 30m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/f071d14c2f4e494eb202d3ed0652b907 : SUCCESS in 1h 32m 15s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/45d61eb2fb634ce49cde7512637dd3c9 : SUCCESS in 21m 43s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aadbc957e94c4679b7c827df917ee4a1 : SUCCESS in 20m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5ea388a983944e2daefece54ac0c8e58 : SUCCESS in 2h 30m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09bb06c667d246278eab5c9838eba764 : SUCCESS in 1h 07m 28s","accounts_in_message":[],"_revision_number":3},{"id":"838047bc86f4206b03b9bc4aa18d840b0516e876","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-09 01:21:38.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/73/687373/3/check/EMC_VxFlexOS_NOVA/51f0e6d/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"474d4ad42747106a11b3511969cea883119a5fb4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-10-09 05:15:25.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38118 : FAILURE in 1h 45m 21s","accounts_in_message":[],"_revision_number":3},{"id":"c35b3b806ee1943477d134aee604ba9f4dc45b65","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-09 09:34:24.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/687373/3 : FAILURE in 2h 36m 30s","accounts_in_message":[],"_revision_number":3},{"id":"0a75f8290b26104b11dd794db6c247a66be24e08","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-09 11:57:15.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-73-687373-3 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":3},{"id":"6f5600951e07c3e227e28310935991987d3eff75","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-09 14:10:30.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/687373/3/check/tempest-dsvm-full-xenial/fe29d21/ : SUCCESS in 2h 03m 50s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/687373/3/check/tempest-dsvm-full-xenial-py3/3a72768/ : SUCCESS in 2h 36m 52s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/687373/3/check/grenade-dsvm-xenial/2c61864/ : FAILURE in 2h 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f92aca3093297333552eaca82580257eeedcefc3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-09 14:51:29.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/73/687373/3/check-vote/ext-nova-zuul/d64f06a : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":3},{"id":"8c32a31f6f823c1a9179e2453928d6d100a64d47","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-09 15:41:46.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nA few nits which I\u0027d let go but soft -1 because we can add deprecation comments to the api-paste.ini file itself.","accounts_in_message":[],"_revision_number":3},{"id":"82cb21496b031643aa8b3c87e1a99fe182f38cef","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-09 20:16:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6a7e1bacfe009ec712cd73ffbbd2254bcc64b89e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-09 20:16:32.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a1a1e1100eaba4349183b15e0096ab190a042a8a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-09 20:16:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3142fd1ff6b9069b12c15820a452032fcf0d1071","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-10-09 20:33:39.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/687373/4/check/tempest-dsvm-full-xenial/28e47b9/ : FAILURE in 2m 16s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/687373/4/check/tempest-dsvm-full-xenial-py3/bf6e27d/ : FAILURE in 2m 16s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/687373/4/check/grenade-dsvm-xenial/9f6de4a/ : FAILURE in 2m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"503a5ea64870f722e2896dd809911f2cf67f09cf","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-09 20:48:00.000000000","message":"Patch Set 4:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/73/687373/4/check/pmem-tempest-plugin-filtered/07ff94e/ : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":4},{"id":"330c4e70f305bf4c6a79f5c7f8efad5dbec1c206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 22:28:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2642ac06c5c44300861b9c01f60665f6 : SUCCESS in 53m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c4516b8bd9bd4aabb312f571d8685e70 : SUCCESS in 1h 30m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d20550392f248ed9ff7fd8617bb316a : SUCCESS in 15m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d384e74916f4d01a9f2493ebb5821c6 : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f99fbe76365b4d4884c1a814f03de3ba : SUCCESS in 12m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a710e758d94c472898dc26702bda0d48 : SUCCESS in 14m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad2ab7a856ee4e3ba996bdd058df551c : SUCCESS in 15m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/68d640392932483385dfdaa53ffe2946 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a364c86a3ff4ed48708355b9b7f25be : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00f20a0a028a4ac2abef534a647aeac1 : SUCCESS in 52m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/adeefe9a3a614e1cb6c5207a4fa18baf : SUCCESS in 56m 10s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f1e8cc00f16b4842bd00b9938af64481 : SUCCESS in 1h 26m 39s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f62434ea99a48bcac1b5db04836310e : SUCCESS in 1h 21m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1c843cce9d2422d9966d82fd5247b5b : SUCCESS in 1h 24m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/baaf7c0da7b7443cab228bbb17a17f31 : SUCCESS in 1h 03m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/321dd7de219a4e73bc94dc484c095d28 : SUCCESS in 1h 31m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/bbe5ec3e5551433db6810709e97d47d4 : SUCCESS in 1h 32m 14s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/62a99065f08046a59e27161cb837c029 : SUCCESS in 20m 27s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/90524e21f4604370956502bc6c50bc9e : SUCCESS in 17m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/719d3be16b19470d94e9ae557b2e64da : SUCCESS in 2h 09m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/384843c3ee994b589df272b6cf87e2ed : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":4},{"id":"3496ebaafd7378d8f11cea9f42f17704cadf23c5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-10 01:23:07.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/687373/4 : SUCCESS in 2h 55m 55s","accounts_in_message":[],"_revision_number":4},{"id":"b3153c8e77cb4043ea8e601530eb9d8bd2420200","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-10-10 02:57:25.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38153 : FAILURE in 3h 08m 57s","accounts_in_message":[],"_revision_number":4},{"id":"23ccaca1b7d517aa5ba4690b6c313bd54fe239d3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-10 11:52:55.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/73/687373/4/check-vote/ext-nova-zuul/c55f22f : FAILURE in 40m 24s","accounts_in_message":[],"_revision_number":4},{"id":"73f10864317d4dac39d9c82c3582968deb45ecb4","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-10-10 17:54:02.000000000","message":"Patch Set 4: Code-Review+2\n\nLooks OK to me.","accounts_in_message":[],"_revision_number":4},{"id":"f77de6fd2145f138d694d79b3683fe1d2ea63abc","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-12 17:27:22.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"54814bf5f4a74778cd51e752f8abe4512cd98d81","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-10-13 15:13:29.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"cd6763de8fb713636c7fdfd2b8a4ae9c4a6566ca","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-10-14 03:23:19.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2665157e6639c4f9b751a8e64850a7a040765600","author":{"_account_id":17499,"name":"Filippo Inzaghi","email":"zhao.daqing@99cloud.net","username":"qingszhao"},"date":"2019-10-14 09:18:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"eaec9dbe4cabd784f8330993867d2ff44419da4d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-10-14 14:16:34.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n\\o/","accounts_in_message":[],"_revision_number":4},{"id":"169f7d10b7110d3d397cb16db3da5ed1effce582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 14:16:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e634b9833fcc2273bb6903dc1acc2f368ae05ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:56:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5e16f01d231848679ce29028299b03a0 : SUCCESS in 54m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/51bab216575c42f885e23164868b5296 : SUCCESS in 1h 25m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d8d05ea766d4f42bc167d7932d8bb53 : SUCCESS in 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ac38ac5cc1472294716bdb3812cfca : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0ccffe87d7b942308749626b08f78d9f : SUCCESS in 16m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01a2d874342946cabc5431697e9859a8 : SUCCESS in 18m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40d865c4cb7f43e1b65cdf95dfc6b78e : SUCCESS in 18m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc9ae2cf41e644d89ea572e93add55ed : SUCCESS in 13m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04826afe22ca4c9397953505a8b3bb60 : SUCCESS in 42m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c5ecb870665e465a9fda571d592444ce : SUCCESS in 1h 16m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3682777f2f45454b8244f8d070f1dee9 : SUCCESS in 51m 28s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/0a3b36e0cb1d46488dbf8d0e85652a9b : SUCCESS in 22m 14s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7ca273da68334e71822d0e91f6ca5d22 : SUCCESS in 15m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e5381e831cb541b7adc98b75e0ea2888 : SUCCESS in 1h 54m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/42f6dda10aba447789788d560f323428 : SUCCESS in 1h 30m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9084258c4c774f7b8e903978576a86c5 : SUCCESS in 2h 46m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01feaf8a9225472aa56347f51b7053ca : SUCCESS in 1h 09m 42s","accounts_in_message":[],"_revision_number":4},{"id":"4ea2cee9bcc3b6f5837bc434e2c838416ee3276f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 20:56:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b3ecef8360243028e58d4ae2a5477d91f342a614","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 21:04:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d842ca9005843f4bc4ea01154602534 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5f980562c294cda866e7f7dc574cd4f : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"18de63deaab792a490e987663d96b6025309b862","revisions":{"2920fa5d230c13657db221b7e7047daca1ba4bbf":{"kind":"REWORK","_number":1,"created":"2019-10-08 18:09:16.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/73/687373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/687373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/687373/1"}}},"commit":{"parents":[{"commit":"aebf8d36b8cfc62cc7102c9365215f5c7ba67a18","subject":"Merge \"trivial: Make it obvious where we\u0027re getting our names from\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aebf8d36b8cfc62cc7102c9365215f5c7ba67a18"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 18:02:29.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 18:02:29.000000000","tz":-300},"subject":"Deprecate [api]auth_strategy","message":"Deprecate [api]auth_strategy\n\n[api]auth_strategy defaults to `keystone`. The only other choice is\n`noauth2`, which activates noauth paste pipelines, which go through\nNoAuthMiddleware, which is crusty and bogus. It is used in our\nfunctional tests to avoid having to fixture out keystone, but should not\nbe used in real deployments, ever. Deprecate the option for removal,\nwhereupon we\u0027ll just hardcode to `keystone`. At that time, we also need\nto move the middleware under the nova.tests package -- or find a way to\nget rid of it entirely by stubbing out keystone if that\u0027s relatively\neasy.\n\nChange-Id: I9e2be5423cc0821a628db7a68ad52bbd91264acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2920fa5d230c13657db221b7e7047daca1ba4bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2920fa5d230c13657db221b7e7047daca1ba4bbf"}]},"branch":"refs/heads/master"},"1ef8e728594cb08c087d64b5b12d09bcd7a9431f":{"kind":"REWORK","_number":2,"created":"2019-10-08 20:33:03.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/73/687373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/687373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/687373/2"}}},"commit":{"parents":[{"commit":"aebf8d36b8cfc62cc7102c9365215f5c7ba67a18","subject":"Merge \"trivial: Make it obvious where we\u0027re getting our names from\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aebf8d36b8cfc62cc7102c9365215f5c7ba67a18"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 18:02:29.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 19:31:04.000000000","tz":-300},"subject":"Deprecate [api]auth_strategy and noauth2","message":"Deprecate [api]auth_strategy and noauth2\n\n[api]auth_strategy defaults to `keystone`. The only other choice is\n`noauth2`, which activates noauth paste pipelines, which go through\nNoAuthMiddleware, which is crusty and bogus. It is used in our\nfunctional tests to avoid having to fixture out keystone, but should not\nbe used in real deployments, ever. Deprecate the option for removal, and\nadd a deprecation warning in the paste pipeline if it is used.\n\nWhen we remove the option, we could just hardcode to `keystone`. At that\ntime, we also need to move the middleware under the nova.tests package\n-- or find a way to get rid of it entirely by instead stubbing out\nkeystone in tests if that\u0027s relatively easy.\n\nChange-Id: I9e2be5423cc0821a628db7a68ad52bbd91264acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ef8e728594cb08c087d64b5b12d09bcd7a9431f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1ef8e728594cb08c087d64b5b12d09bcd7a9431f"}]},"branch":"refs/heads/master"},"b472efff62788b0629ef1e828e294b2de3f84743":{"kind":"REWORK","_number":3,"created":"2019-10-08 22:43:34.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/73/687373/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/687373/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/687373/3"}}},"commit":{"parents":[{"commit":"aebf8d36b8cfc62cc7102c9365215f5c7ba67a18","subject":"Merge \"trivial: Make it obvious where we\u0027re getting our names from\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aebf8d36b8cfc62cc7102c9365215f5c7ba67a18"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 18:02:29.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 21:03:16.000000000","tz":-300},"subject":"Deprecate [api]auth_strategy and noauth2","message":"Deprecate [api]auth_strategy and noauth2\n\n[api]auth_strategy defaults to `keystone`. The only other choice is\n`noauth2`, which activates noauth paste pipelines, which go through\nNoAuthMiddleware, which is crusty and bogus. It is used in our\nfunctional tests to avoid having to fixture out keystone, but should not\nbe used in real deployments, ever. Deprecate the option for removal, and\nadd a deprecation warning in the paste pipeline if it is used.\n\nWhen we remove the option, we could just hardcode to `keystone`. At that\ntime, we also need to move the middleware under the nova.tests package\n-- or find a way to get rid of it entirely by instead stubbing out\nkeystone in tests if that\u0027s relatively easy.\n\nChange-Id: I9e2be5423cc0821a628db7a68ad52bbd91264acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b472efff62788b0629ef1e828e294b2de3f84743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b472efff62788b0629ef1e828e294b2de3f84743"}]},"branch":"refs/heads/master"},"18de63deaab792a490e987663d96b6025309b862":{"kind":"REWORK","_number":4,"created":"2019-10-09 20:16:17.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/73/687373/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/687373/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/687373/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/687373/4"}}},"commit":{"parents":[{"commit":"aebf8d36b8cfc62cc7102c9365215f5c7ba67a18","subject":"Merge \"trivial: Make it obvious where we\u0027re getting our names from\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aebf8d36b8cfc62cc7102c9365215f5c7ba67a18"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-08 18:02:29.000000000","tz":-300},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-10-09 20:16:07.000000000","tz":-300},"subject":"Deprecate [api]auth_strategy and noauth2","message":"Deprecate [api]auth_strategy and noauth2\n\n[api]auth_strategy defaults to `keystone`. The only other choice is\n`noauth2`, which activates noauth paste pipelines, which go through\nNoAuthMiddleware, which is crusty and bogus. It is used in our\nfunctional tests to avoid having to fixture out keystone, but should not\nbe used in real deployments, ever. Deprecate the option for removal, and\nadd a deprecation warning in the paste pipeline if it is used.\n\nWhen we remove the option, we could just hardcode to `keystone`. At that\ntime, we also need to move the middleware under the nova.tests package\n-- or find a way to get rid of it entirely by instead stubbing out\nkeystone in tests if that\u0027s relatively easy.\n\nChange-Id: I9e2be5423cc0821a628db7a68ad52bbd91264acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18de63deaab792a490e987663d96b6025309b862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18de63deaab792a490e987663d96b6025309b862"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
