)]}'
{"id":"openstack%2Fnova~688600","triplet_id":"openstack%2Fnova~master~I4d37d6b093e56f85a5a5de8d81b6bb0411a8f73c","project":"openstack/nova","branch":"master","topic":"trival/fields","hashtags":[],"change_id":"I4d37d6b093e56f85a5a5de8d81b6bb0411a8f73c","subject":"cleanup to objects.fields","status":"MERGED","created":"2019-10-15 02:05:00.000000000","updated":"2019-11-07 07:58:43.000000000","submitted":"2019-11-07 07:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688600-1573113268783-91d4b73c","meta_rev_id":"4cf830caae0be9db14cd640d690ba56bdd0e9528","_number":688600,"virtual_id_number":688600,"owner":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-11-07 03:43:49.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-07 06:16:20.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-07 05:55:54.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-11-07 05:23:17.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"date":"2019-11-07 06:23:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-07 03:26:35.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-11-07 04:35:03.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2019-11-07 03:57:01.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-07 07:54:28.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-30 17:52:25.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2019-11-06 21:25:07.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-30 17:51:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2019-11-06 21:25:07.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-30 17:51:39.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-10-30 17:52:25.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2019-11-06 21:25:07.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-11-07 03:26:35.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-07 03:43:49.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-07 03:57:01.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-11-07 04:35:03.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2019-11-07 05:23:17.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-07 05:55:54.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-11-07 06:16:20.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-07 06:23:44.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-11-07 07:54:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f940c0203c91720ea7f5ce43694d4c5fbb930721","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2019-10-15 02:05:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"355d7c4fed307980c4a36ec83a6d7a3e1f9754e9","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 02:37:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2f6d2e0192758aaa8f05d14d511cd1ba6033607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 04:19:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7f4b028231ad4b64a368919790bf862a : SUCCESS in 55m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2383915dabee4d17a446d0ed4b378b66 : SUCCESS in 1h 15m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6459a59be0ba451bb5aa67f1a27258ae : SUCCESS in 15m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18ab1bb37dda4da388e065f4a8e9eda9 : SUCCESS in 14m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c683f560c140b28487883395f39cd7 : SUCCESS in 9m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8c31cfce7e6f4f6181c8e86a32bed6ca : SUCCESS in 15m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/240cde9c2cc74ee593509634eceea1ca : SUCCESS in 10m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5752bf6aa1043459aa18cde8e9368a2 : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d570c91b9e480690caf43ebbc19f01 : SUCCESS in 9m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5c463d25ab284a05a7190a8ca5a958ed : SUCCESS in 53m 43s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/abe8a6ec569c45a38e12110876e19f5d : SUCCESS in 1h 10m 33s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/03de00ec17ff4d2586a7fa2f2e0240fe : SUCCESS in 1h 09m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/39084aee80124ffea57c0843d1235b68 : SUCCESS in 1h 14m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/279da3427ae848e7b1282656917d8827 : SUCCESS in 49m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/765903896a104219bccfc80a01d71ea0 : SUCCESS in 1h 30m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/0933cb18c12549c0979669a804b46362 : SUCCESS in 1h 13m 58s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/42a26e001ca44864b1b37bbf945cd5f3 : SUCCESS in 18m 38s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20101430dd834b06bd8fc8770ccc9ec9 : SUCCESS in 17m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/236efc4bf5cd438982ca3d5e37b20d4b : SUCCESS in 2h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06c9ce44ec7348999e7acdff339fd65d : SUCCESS in 59m 08s","accounts_in_message":[],"_revision_number":1},{"id":"884503383c8412b23cbeac1f10a2b85ce067b0ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-10-15 04:20:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/00/688600/1/check/EMC_VxFlexOS_NOVA/88aae6e/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"d72a6af5ccc2844f9ee53d75a1a7d575dc05f4f7","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-10-15 10:50:22.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-688600-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"f2e2a7f40f5caf7bc0f56c13892525760764e178","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-10-15 16:21:30.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/00/688600/1/check/pmem-tempest-plugin-filtered/66e7423/ : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":1},{"id":"01dbccd358d64b43b601397c44e452865b34c406","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-10-15 19:13:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/688600/1 : FAILURE in 3h 56m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2ecf07bb024b0175a65db2ed98d9c9d7147c0acf","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-10-16 04:16:33.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/00/688600/1/check-vote/ext-nova-zuul/0b150a9 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ca750b72c7f8254b250a83171f14322761f69597","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-10-30 17:52:25.000000000","message":"Patch Set 1: Code-Review+2\n\nThis seems reasonable.\n\nNot sure why we have the redefinitions in the first place, but since we do, we ought to use them. (E.g. so we don\u0027t get nasty surprises if we someday decide to actually subclass AutoTypedField.)","accounts_in_message":[],"_revision_number":1},{"id":"355728efaba31f0f60e86a7118427cda98be67de","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-11-06 21:25:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)\n\nsure","accounts_in_message":[],"_revision_number":1},{"id":"76589e71581814c4ab861f0c75804f34fa1953ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 21:25:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7935e21c8efdec2856820219a4441226b023fc81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 23:00:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/abd9f258950c4898b4d88b4056c791a9 : SUCCESS in 59m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b6b4f3999267427cacc619d6a45f54f6 : FAILURE in 1h 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e945125c9e0343bda0680b4620a48d1a : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b62b261b03ea49a5b911bbf1e72be2d0 : SUCCESS in 10m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c9df1c991f64bd28c80f7411df3537c : SUCCESS in 12m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/afdeff20538a41db88a0b9dd967c88a5 : SUCCESS in 12m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af83dbc08b0e49e790e9ac1fff159e3d : SUCCESS in 12m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdfc5b358294123991d282dc3b0f605 : SUCCESS in 9m 22s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e62a6c0b69fa4820a57b1d0aeef16017 : SUCCESS in 1h 17m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f6483830e0a2469caecb1ffbdf09ae51 : SUCCESS in 45m 36s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/c25173b7d775409f808702d61f6b0f85 : SUCCESS in 16m 09s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/972ba9c532d44d0697bd32448cca352e : SUCCESS in 19m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7309aa42f25d4a8d89f65688ce79a428 : SUCCESS in 1h 15m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/0e48c7be2014498182a56cbf84d6afca : SUCCESS in 1h 32m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a6d68d65cca4accbf979bac3c67d895 : SUCCESS in 1h 03m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7ab61a71c46e9a14cc30db197a381a2f2bbb4141","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-06 23:00:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"51f18d4cd77ef8fe7f1463af90e2e0f8118c353d","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-06 23:01:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d7f57aaa4b659156198380c312c416b705eb2640","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-06 23:18:26.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/00/688600/1/check-vote/ext-nova-zuul/c301498 : FAILURE in 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"7de44e0fa04a12dfd9a6f0cfd6c0fe89ed840393","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-11-06 23:29:24.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/00/688600/1/check/pmem-tempest-plugin-filtered/4defb68/ : SUCCESS in 27m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e5c16ac3069f79e4297d941b25a071ef414f14a3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-11-07 00:22:22.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38894 : SUCCESS in 1h 21m 22s","accounts_in_message":[],"_revision_number":1},{"id":"96a037b9666a4084546bcd65ddd291348f258e37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 01:03:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2aea3f74f86540699524b17aa30327fa : SUCCESS in 59m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2c91be27c7c147ef9b38f5b79ea48a60 : SUCCESS in 1h 12m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ad7f498aae04bbf8f9b5fd9ff07e518 : SUCCESS in 17m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddb8cce073924b70b095b57c7c55eb56 : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60a5031862948cda0c09074db37e621 : SUCCESS in 9m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00195fcdbe0b4cd9bf74cc455c364c57 : SUCCESS in 12m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb110b4327e34b2ba951b41049b638d0 : SUCCESS in 12m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af4111b7f3be4b86abbae8c4b2935d3c : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f78184fc69406fa5795cda5f62a552 : SUCCESS in 9m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/089ecc211cf24144a6416f49b7ea61f4 : SUCCESS in 59m 30s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9b1f3d91388a423d991a08aae24d62aa : SUCCESS in 1h 09m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae2b646daa3d439a96698e6e265a6360 : SUCCESS in 1h 52m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ab19f2507dc464d889a3debc2ade160 : SUCCESS in 1h 15m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c4dc15c81224491bad3a1541945116e8 : SUCCESS in 50m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6ff217eb94e243d0a64813baa164281d : SUCCESS in 1h 23m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/3923b41550df41d3a0dd41194486370b : SUCCESS in 1h 26m 11s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/e003870e7b204a52a7b85c4c4e3dadf6 : SUCCESS in 20m 17s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb35db20d2914ab389065e1fd20d2eea : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76b609d7958a439d8e9dd85064479abb : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ea61946a9c382c9f34c59dea46a6cc1f3bd94c5e","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-11-07 01:04:21.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/00/688600/1/check/EMC_VxFlexOS_NOVA/95910a3/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"7fc5038aaece37a332ea50764aa9db2a62a87977","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 01:04:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fad6ba3c69f7b6655732ec6af59b6e7dee15c736","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-11-07 01:24:29.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/688600/1 : SUCCESS in 2h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"28514f76474cde9a3ad5f2dd3b2e9ecb5c4e32c9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-11-07 02:01:30.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/00/688600/1/check/tempest-dsvm-full-xenial/a845aa2/ : FAILURE in 2h 59m 07s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/00/688600/1/check/tempest-dsvm-full-xenial-py3/31b0827/ : SUCCESS in 2h 51m 22s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/00/688600/1/check/grenade-dsvm-xenial/d154be7/ : FAILURE in 2h 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7972d1a08a5f50387294c5d60744fac9bab1f2e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-11-07 02:41:54.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-688600-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"d9274afefd8258be306c13544988e835f710803b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 03:20:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d8a2cdba92aa4840bec168ee457045bd : SUCCESS in 58m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b59c307703dc408f960e097d851171da : SUCCESS in 1h 26m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae8b14676725492bab8d13b0e4619dd8 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6125955fbbd4bcd8b0e6d5bb4a2f09f : SUCCESS in 10m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/30f7a11e723a4108bf9a3f2bac800190 : SUCCESS in 13m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6edf172aa2374108b3edad6d94d46a63 : SUCCESS in 17m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/248d0bac551640b49d2895aebac815c6 : SUCCESS in 12m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28fe7c5040604c57a2e135667b4fc36e : SUCCESS in 11m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7144f6686b664793864dfbf246a7b0e9 : SUCCESS in 1h 11m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/32c71f029c0849f99fb65484d49c978c : SUCCESS in 49m 49s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/32b28ae809054c4b8ecca4d8e605b922 : FAILURE in 24m 25s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1090be53cdb2417eb9348d3212a7c153 : SUCCESS in 31m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/72fcfd54b5c14a458410856f97db959f : SUCCESS in 1h 23m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/b663bdbd50ed427c99cb54be97df9d19 : FAILURE in 1h 38m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6348bde50d0d4bffbb28ef3a28687900 : SUCCESS in 1h 00m 58s","accounts_in_message":[],"_revision_number":1},{"id":"24d4b4cfb8e936ef12e7892d527d02409d094c00","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-11-07 03:26:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b4ba56706523def37a137138d25ee844a323926b","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2019-11-07 03:26:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85781b87b597fa55234d113d0c7ad2096e1d3f06","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-07 03:43:49.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/00/688600/1/check-vote/ext-nova-zuul/b3fe613 : FAILURE in 17m 00s","accounts_in_message":[],"_revision_number":1},{"id":"13b15567387a46cdc00caff757e45c812d693095","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-11-07 03:57:01.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/00/688600/1/check/pmem-tempest-plugin-filtered/c3ce619/ : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":1},{"id":"da800b6f8ad23de15ea847c6f0692ef5c2e6c543","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-11-07 04:35:03.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-38897 : FAILURE in 1h 08m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4578fd4343feb3c6f935e0e055cc65ab06314605","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 05:22:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/13684a6349264f8c9ae577614596a5cf : SUCCESS in 51m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/089ba1bef7af4e7185e19ec7bf0e1ac1 : SUCCESS in 1h 31m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/853e431833084211a54d0fb58838ce25 : SUCCESS in 17m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdb918949a7d4316a3b351036c3566bf : SUCCESS in 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a452733f1227408f8cde4f0c9cfee8d5 : SUCCESS in 10m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6becb8cf875454f8997a6cd2004d41c : SUCCESS in 14m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f76b491eda44164a55abadf61ed6f87 : SUCCESS in 12m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9ada028c46cd45ec8aa9e944d12b0479 : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bed4a962bfaa4075b8a8f790ac607fa8 : SUCCESS in 10m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f2f8481e65cf46e186f5f15df3a3b6ce : SUCCESS in 51m 08s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/68549f16591441b3b375872b80db45bf : SUCCESS in 55m 20s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/4093b73d2e9440b891b02138d6a96c66 : SUCCESS in 1h 06m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6fc18e0dda2144e6a802b429a94aa091 : SUCCESS in 1h 08m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/682e1152cacb499381f0534b96e43992 : SUCCESS in 46m 53s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4096bc4938f1461ba15b8dd807d6b220 : SUCCESS in 1h 16m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/2bbabdcf5bff4cfd8459ed0eb59c72f7 : SUCCESS in 1h 48m 38s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/24daa25f9a084b57a7214ec55f8ce817 : SUCCESS in 17m 04s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8664ead3b5a478c9a65b6c209e18a1e : SUCCESS in 17m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d97b70374c35411dbce5da2476af45a1 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a4bd306a9e2ba92de60573567b93f7c01dd42535","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 05:23:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e4789e8d9c3fa968e7a19e7c380a6c94ee867d8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-11-07 05:23:17.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/00/688600/1/check/EMC_VxFlexOS_NOVA/694f2b0/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"4117421633ead305f1dde2977deb8ad5d0af970d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-11-07 05:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/688600/1 : SUCCESS in 2h 29m 01s","accounts_in_message":[],"_revision_number":1},{"id":"df1ec03584f1d47eb14e966ba1800416dd0aee56","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-11-07 06:16:20.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-00-688600-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"c6abd2fe2679efeac6a053c0cb836ff7cbac167a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-11-07 06:23:44.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/00/688600/1/check/tempest-dsvm-full-xenial/26f345f/ : SUCCESS in 2h 49m 10s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/00/688600/1/check/tempest-dsvm-full-xenial-py3/6e4cbe0/ : SUCCESS in 2h 47m 11s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/00/688600/1/check/grenade-dsvm-xenial/27ab623/ : FAILURE in 2h 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adebc46f8b2381b314f029ee13d0024221176a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 07:54:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1601874b1758420797a3cb52c728d287 : SUCCESS in 57m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8f671347d9d84c95bc0cc17787d57f8d : SUCCESS in 1h 28m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6114d44870946f59f22e3aab5e0cf16 : SUCCESS in 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bab70b6557044e785805cfa06e8e69a : SUCCESS in 9m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb2ac952714f4174b40c877521be473d : SUCCESS in 13m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3a3998e835d464e818f53d4d5e8ff7d : SUCCESS in 21m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ae4bab4e2e94350bf583c429c051201 : SUCCESS in 23m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/395a2dc894994fcd855d0eab8e9931cf : SUCCESS in 9m 51s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0dc96e3b9e64d73b9fe09e05b14759d : SUCCESS in 1h 33m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/65795427985c46958bc7abdcb70578fc : SUCCESS in 1h 01m 10s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/37195846bda04fff9245fe41a3110880 : SUCCESS in 20m 30s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c3dc6df349884db6ba267737a9355a48 : SUCCESS in 14m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/824b165abfcf4d9eaae5d51717810184 : SUCCESS in 1h 36m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/eb903feb9ad943398ee79fc2af0b4132 : SUCCESS in 1h 33m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b972ac021234aa19cfdde77ce757522 : SUCCESS in 1h 01m 38s","accounts_in_message":[],"_revision_number":1},{"id":"64150e4c218af97205bbcb1a94185f692405f0ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 07:54:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2dda4068ae31aaeb8a0a78e612c760f2eac47b20","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-07 07:58:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8324c8c184a145658669b32c9b00720e : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8d4e191dbfaad3fc6d6330f601bd4548be7e73c","revisions":{"b8d4e191dbfaad3fc6d6330f601bd4548be7e73c":{"kind":"REWORK","_number":1,"created":"2019-10-15 02:05:00.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/00/688600/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/00/688600/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/688600/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/688600/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/00/688600/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/00/688600/1"}}},"commit":{"parents":[{"commit":"6b60cae01966a6a799eec5e8a04a9a661eda7a12","subject":"Merge \"setup.cfg: Cleanup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b60cae01966a6a799eec5e8a04a9a661eda7a12"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2019-10-15 09:53:21.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2019-10-15 10:07:29.000000000","tz":480},"subject":"cleanup to objects.fields","message":"cleanup to objects.fields\n\nSince \u0027AutoTypedField\u0027 has been redefined in scope of objects.fields,\nwe should use local definitions.\nThis is better for code consistence by using same coding style.\n\nChange-Id: I4d37d6b093e56f85a5a5de8d81b6bb0411a8f73c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8d4e191dbfaad3fc6d6330f601bd4548be7e73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b8d4e191dbfaad3fc6d6330f601bd4548be7e73c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
