)]}'
{"id":"openstack%2Fnova~696746","triplet_id":"openstack%2Fnova~master~Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488","project":"openstack/nova","branch":"master","topic":"bp/remove-nova-network-ussuri","hashtags":[],"change_id":"Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488","subject":"nova-net: Make the security group API a module","status":"MERGED","created":"2019-11-29 17:19:06.000000000","updated":"2020-01-23 16:34:25.000000000","submitted":"2020-01-23 16:30:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696746-1579797023489-090dce9f","meta_rev_id":"bfdaf3db8309e4798524215c9555fc24b08692c6","_number":696746,"virtual_id_number":696746,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 14:37:49.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 15:48:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2020-01-23 12:15:27.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 16:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 15:11:40.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-01-23 14:51:51.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":2,"date":"2020-01-23 12:10:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2020-01-23 12:10:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2020-01-23 12:10:21.000000000","_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-23 12:10:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 15:39:41.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2020-01-06 16:10:28.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-08 17:40:27.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-01-15 18:19:09.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-20 14:37:31.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-01-21 18:30:33.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-01-21 19:53:15.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-23 12:15:27.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-01-23 14:37:49.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-23 14:51:51.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-01-23 15:11:40.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-23 15:48:41.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-23 16:30:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"04b7f3f311ba1509e2555c84c653666b8270e628","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-29 17:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96c26d5c32d121f27c83c08038b5bce726d0fc79","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-29 17:20:33.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"a55caad9f4362f54afbe6991c4e80e0c9d4a6b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-29 18:48:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/015b9d626bf74eda8223bc2aea69fe5e : FAILURE in 1h 00m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e4adf658e791486a95ee681a30107611 : FAILURE in 31m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee4ace5fc7534bfeb3ed1a76ab4c578b : FAILURE in 12m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3ef5c9577fc419bae088e34cfc5dd03 : FAILURE in 12m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2dcc0ff390146c7b915f1aa583d3351 : FAILURE in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/098c8360d90d49e9998422e5e4565956 : FAILURE in 11m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c2141b515f44ec2b0a37b30f62a4788 : FAILURE in 13m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef91c8a9d822415aba9326703b6befec : SUCCESS in 11m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4c5df9cd5a2f489cbd9f811a1bbf21dd : FAILURE in 28m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a20fec8ea16745799c71a925ef8766f0 : FAILURE in 24m 53s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a10e1cbc60a94ed488d01932cab74ef7 : FAILURE in 24m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9996618b45a4da5b2dcf6781c6980ce : FAILURE in 55m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a0942c4bd83e4c119d06bc96dcb4499d : FAILURE in 27m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3fc865c2324d4e028e6f0d3f48ec4724 : FAILURE in 34m 15s\n- nova-next https://zuul.opendev.org/t/openstack/build/2563a6b2e0d04465b99ca812ece4cbbb : POST_FAILURE in 33m 06s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/915e8d25c9354f13be4bfc3c937a2080 : FAILURE in 10m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ce40bbc3f3c49f6839bb03750ef54b1 : FAILURE in 24m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ad48f31a2b7244ac2b12e0d2c1a0e2a6378f3723","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-11-29 19:05:25.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/2/check/pmem-tempest-plugin-filtered/581eeed/ : FAILURE in 19m 33s","accounts_in_message":[],"_revision_number":2},{"id":"5f3e8036663118e46d0be2e8cfb4f794399b2193","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-11-29 22:57:39.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39648 : FAILURE in 29m 20s","accounts_in_message":[],"_revision_number":2},{"id":"4b24c560b44902ad697ae58dc831995a982cb061","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-30 00:30:36.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/2/check-vote/ext-nova-zuul/e9c6d10 : FAILURE in 0s","accounts_in_message":[],"_revision_number":2},{"id":"a5c20a3936e2e07e1447cf13f85f9dacb31f06e5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-11-30 03:36:37.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696746/2 : FAILURE in 4h 14m 37s","accounts_in_message":[],"_revision_number":2},{"id":"d17b7e367db3033cf11d7e90a9e1421eaa896be1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-30 11:41:49.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"68c071ee1d89ca484bfac6398bc3ba6ee55e54f1","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-11-30 12:32:20.000000000","message":"Patch Set 3:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/3/check/pmem-tempest-plugin-filtered/52a9620/ : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":3},{"id":"949362da686f239da0dce4aef75458e997181fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 13:52:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b4bb6bd5500c4102835418e0072d289a : SUCCESS in 48m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/37a7303d1c3c49dea815650a37d970cf : SUCCESS in 1h 22m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3bd639eecaa438781e657ef1137d67a : FAILURE in 12m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9cc422b752ef46849fe54f2624248af3 : FAILURE in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4673951eda714412a42dc373aef9d727 : FAILURE in 9m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73eb3fc19ec4402788b91c78d3c9c4d3 : FAILURE in 13m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/170241e2167841418b9aec89796a7820 : FAILURE in 13m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0072b7a53e954982ae6901b8c4f087ef : SUCCESS in 10m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f78687d4049846be9bc9b5f0ae60e031 : SUCCESS in 1h 24m 37s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1d744461960241e288aef6e7ee12f431 : SUCCESS in 1h 02m 54s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/749ed02573c041098319bad72c46a5d9 : SUCCESS in 1h 08m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3645958acdf8463e8f2881410a708ccb : SUCCESS in 1h 20m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e1ccd5ec606f4f059ec6cbf34d2dd9e1 : SUCCESS in 1h 22m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bdd80dc3aa354b6194f90aed741e52f2 : SUCCESS in 1h 28m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/e6f987b1846143248e293555efacb242 : SUCCESS in 1h 50m 08s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c856cc4279ed41cb9f1dc12350b27075 : SUCCESS in 17m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3514c289ef95457499072ae2cf8a2a9f : SUCCESS in 1h 20m 49s","accounts_in_message":[],"_revision_number":3},{"id":"47e3c04d04860728f300cb356ab80fceb7c278bb","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-11-30 14:50:03.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39666 : FAILURE in 26m 47s","accounts_in_message":[],"_revision_number":3},{"id":"f12e0de09dce253246c2ffec375fddf25ef5fff6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-30 16:04:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"6e5cbec4ff61333d51d4984fad84d7f662ecdd76","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-11-30 16:36:02.000000000","message":"Patch Set 4:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/4/check/pmem-tempest-plugin-filtered/44ca0f8/ : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":4},{"id":"9eb9525b1a33acb07b0e6b0bea3896f753707413","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-11-30 17:41:12.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39670 : FAILURE in 26m 42s","accounts_in_message":[],"_revision_number":4},{"id":"fb8e47782dc6200d0eca1048714f0ba5e693a404","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 17:57:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d2f7d3db181e4373a6761bdd15783c47 : SUCCESS in 48m 53s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/658cf271fd18499cb7b00382b944dc96 : FAILURE in 1h 17m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd8f1254ccdc4898abbf3b8dfa938148 : FAILURE in 16m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad3d623d290840d1809c90a9a3c7c244 : FAILURE in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1689e6d7bc2b482b854bac82518a342a : SUCCESS in 9m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0ede52c8b724a32ad7b73ded2f36dbc : FAILURE in 11m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4560fb540e0842c7adb9fdd10f7a23fc : FAILURE in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d60fdbfe14ad425399edab1f80bb633b : SUCCESS in 10m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/be072d821b814fa0a1b9811ff0eb226f : SUCCESS in 1h 02m 45s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d74767bad28e48dc85d6a57ebff4dd9f : SUCCESS in 1h 09m 40s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f723481981fd4ca48966baf13a92e085 : SUCCESS in 1h 12m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0359683a3ea84dc89bc538cf52de9f64 : SUCCESS in 1h 32m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/da1878538c3c4c57b042863d19e67ec5 : SUCCESS in 42m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b4ee2c2854e94c2a9229c8ef49168379 : SUCCESS in 1h 16m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/b01d73e5e9ce4de28d624bb73069d052 : SUCCESS in 1h 43m 40s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6d3b6715975a4e39a7f0f1a18934d190 : SUCCESS in 17m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bde4963311a343e799097dc166d274ca : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":4},{"id":"2b100efa68374d3001f89eb04bd25c8a4fb0a268","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-11-30 18:13:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3ead32142e37d85de5c01de4035f7831bd139116","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-11-30 18:42:04.000000000","message":"Patch Set 5:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/5/check/pmem-tempest-plugin-filtered/3c732ad/ : SUCCESS in 24m 57s","accounts_in_message":[],"_revision_number":5},{"id":"f2865598f7e8f56203b034573f6a8ab172a511fe","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-11-30 19:15:05.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39672 : FAILURE in 26m 36s","accounts_in_message":[],"_revision_number":5},{"id":"85f92329ec1bfc2a86242e24d64c58fba7ef9ed3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-30 19:52:24.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4b086cd6443d43e4bd6f062ab960c011 : SUCCESS in 50m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d2458bdd12d14ad58cb9e05e51414820 : SUCCESS in 1h 13m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba3ad26fe4944beac6c73327ab67255 : SUCCESS in 17m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9993cfc446474d24a0694d057184ffb2 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8139ecd3fdf949a6b9c1a0eebae7cb13 : SUCCESS in 9m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/366926234eaf478487714e77dd3fd0ab : SUCCESS in 18m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0d2d54fdf9a64ec899b74a5ec2c11b27 : SUCCESS in 19m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03bc3d9ae834400093446a9db3f0fe66 : SUCCESS in 8m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/50cdbad194e44ce4b549a37f8c666767 : SUCCESS in 49m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b3c40e64b5e843919ff4c8225f17466d : SUCCESS in 1h 03m 30s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/f9823b120c0c47a5b7038ad7c17af7cc : SUCCESS in 1h 07m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/df88075f45a34ccd93c8aabbcd221098 : SUCCESS in 1h 19m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dbf2b120a70145908c3833e856b23bf7 : SUCCESS in 45m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/25932b2952794034b6a9f51f4e7551d2 : SUCCESS in 1h 18m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/a4a15285706649a590238b7f3902449f : SUCCESS in 1h 25m 44s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51264116aafd4fe9947cf50a81d2e45e : SUCCESS in 14m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e84259ad64d4c509e575e60b2603443 : SUCCESS in 1h 08m 16s","accounts_in_message":[],"_revision_number":5},{"id":"dd7d115e4c69f5f012ba0124e0248289ba50b324","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-11-30 20:02:16.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/5/check/EMC_VxFlexOS_NOVA/3949558/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":5},{"id":"0e27a62f313aee01f88649ca6327175e6a0874c4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-11-30 20:17:30.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/5/check-vote/ext-nova-zuul/2142692 : FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"0c84d6b4bba6a11228b8fd1ebf6f51b716f9d2ab","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-11-30 20:48:24.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-46-696746-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"a11c00ef5691819b68015eddc671a9cb39a5ddae","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-11-30 20:54:08.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/5 : SUCCESS in 2h 17m 29s","accounts_in_message":[],"_revision_number":5},{"id":"d535cffc218cd9184dde9668e8a35f9cb5210597","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-01 02:55:34.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/5/check/tempest-dsvm-full-xenial/1f09eb2/ : FAILURE in 2h 09m 47s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/5/check/tempest-dsvm-full-xenial-py3/d8c04fd/ : SUCCESS in 2h 52m 50s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/5/check/grenade-dsvm-xenial/ef251e1/ : FAILURE in 2h 55m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"390622bda6bd1a8cca898b53d6bbe7f66a08ead6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-03 18:31:13.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"8cc3f73651e5477afc89ccea524302b6c0c1cf50","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-03 19:03:12.000000000","message":"Patch Set 6:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/6/check/pmem-tempest-plugin-filtered/128f2e4/ : SUCCESS in 24m 42s","accounts_in_message":[],"_revision_number":6},{"id":"5a01a014045d822cacbcc16b7a010beb55f73f82","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-03 19:50:23.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/6/check-vote/ext-nova-zuul/c556155 : FAILURE in 0s","accounts_in_message":[],"_revision_number":6},{"id":"736f452490abf985a22b535a782ac01a51e38a22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-03 20:13:55.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/97e31d9bf702441aa293696c14ec6501 : SUCCESS in 57m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/46ca852d90074e1c9e2e6b022bc0cc23 : SUCCESS in 1h 18m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a6f8e1fc9db41b8ae3e0a95ea7fc113 : SUCCESS in 15m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/93dcfac112354d39b4a38c2fd5a93177 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd0d4d6428642f19ae1cbe524d4b6d9 : SUCCESS in 11m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f993ae84d17347a399ad9daf109b24bd : SUCCESS in 14m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bfa80397277247cea2eb1466291ea183 : SUCCESS in 15m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb86cf7b640e4d54830862be75bff10a : SUCCESS in 12m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/142ac40a9919463d837f5cb0f2efa70a : SUCCESS in 49m 57s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/295885c5f3994c6eb3b92319ef7127af : SUCCESS in 56m 51s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/5f60f73a4072453d87a15e4447abb8a7 : SUCCESS in 1h 07m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6ed431dff50c4a0ba60d3ccfc0141918 : SUCCESS in 1h 10m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/060e814446a44d62a592a3a06a17da56 : SUCCESS in 45m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f9398419d0f642bda02b5a4e3e1fd41b : SUCCESS in 1h 17m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/4efb1c5e8a3b44eba0d464cda3308653 : SUCCESS in 1h 30m 21s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07ecd11271f547f283e6541c27698b90 : SUCCESS in 19m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8dcd1cfb8c8242fa87d0f049a3785e7d : SUCCESS in 57m 05s","accounts_in_message":[],"_revision_number":6},{"id":"3086a4afacc128616083bb1cf385e2f98fc26d16","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-12-03 20:21:45.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/6/check/EMC_VxFlexOS_NOVA/4e2d8c8/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":6},{"id":"61879e8f6ff0e385c601da707f259b66ecac80fb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-03 21:45:05.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/6 : SUCCESS in 2h 19m 22s","accounts_in_message":[],"_revision_number":6},{"id":"cd2f848613c050da298ee27ad3c4f19e4cfcfdf8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-12-03 21:59:07.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-46-696746-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"d706ee4858f0aded8fdba6c20ff0a36dd06afce0","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-03 22:22:21.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39741 : FAILURE in 37m 38s","accounts_in_message":[],"_revision_number":6},{"id":"cb184c0c40d09e6a75cd59fd613d42db0317942a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 22:42:08.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/6/check/tempest-dsvm-full-xenial/77823a8/ : FAILURE in 1h 54m 42s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/6/check/tempest-dsvm-full-xenial-py3/a5943b1/ : SUCCESS in 2h 49m 30s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/6/check/grenade-dsvm-xenial/2555975/ : FAILURE in 2h 57m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ce17af14b12d43fb1712b26a6ff3fa5df4a72038","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-06 19:21:25.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"08f22c7cd9c8c5266dcfb244b89daa4dbee2ece1","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-06 20:25:34.000000000","message":"Patch Set 7:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/7/check/pmem-tempest-plugin-filtered/567fabd/ : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":7},{"id":"f20c24f5f7faf35f5aad9870009c8247b5f2ede1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-06 21:23:28.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/7/check-vote/ext-nova-zuul/d9f2f91 : FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"725600962af007caddc430725d182a990bc413ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:28:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8995a2fbe7e44099b63a8eea3a9f41a9 : FAILURE in 39m 34s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e620baf0d18a48919b72ca7967a52dcb : SUCCESS in 1h 28m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dcca51f3ffc41ff9531dcd8092c0445 : SUCCESS in 17m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ccd3515d1324a34a5be67d0cae9a2f9 : SUCCESS in 22m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6bb447c0e0a49959be9eaf39080f5a9 : SUCCESS in 10m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/498c14399ba94b3085eadaf78859bdbc : SUCCESS in 15m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5615a64fbb63420cb9c781f376bc914f : SUCCESS in 23m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5e263cd7bd04dadb3178d7f1867711a : SUCCESS in 10m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7b84506858ed49e89b2f04b9b188a114 : FAILURE in 1h 11m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fbacad460c8b4e9bb9de2d9f1a8daabe : SUCCESS in 1h 06m 15s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ec69317b631249ce9d463c5a8f123d6f : SUCCESS in 1h 23m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/69e33d1a56fe4629b7d2d527c74d04c8 : SUCCESS in 1h 29m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc4a11e4f0c44b2c8d1f6ecac025c939 : SUCCESS in 58m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e57d47b81f5442849cde2c7b0606005d : SUCCESS in 1h 52m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/5622ed9d4b0f40c892ac2daec1f2d45c : SUCCESS in 1h 31m 15s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3b5880ecbff6490e9ba8a60bbef9e08d : SUCCESS in 17m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/618017da5d644dc2af058ad58f6c5d02 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":7},{"id":"cccb5fddb0b69dea55fb284e8f9fb7a621f49ce6","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-06 22:00:55.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39911 : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":7},{"id":"5c4b4e61af2029028f1b4b9d65800b1690df98c5","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-07 00:46:36.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/7/check/tempest-dsvm-full-xenial/7d49df3/ : FAILURE in 1h 25m 36s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/7/check/tempest-dsvm-full-xenial-py3/3f4b646/ : FAILURE in 1h 31m 00s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/7/check/grenade-dsvm-xenial/4cb3e39/ : FAILURE in 2h 55m 54s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8bd0f7838a43384bf7d229f0c38cbc7d54ee0282","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-07 03:22:28.000000000","message":"Patch Set 7:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/7 : SUCCESS in 2h 26m 14s","accounts_in_message":[],"_revision_number":7},{"id":"3a6f7bc7726f1cb6c230870bf71c0875f530da1a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-08 00:15:35.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"1a9764776752c239538734f16b28215471898fa4","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-08 00:56:35.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/7/check-vote/ext-nova-zuul/2ee34a5 : FAILURE in 0s","accounts_in_message":[],"_revision_number":7},{"id":"3c4ccb36b92ff09f2227eda6e4995de77713f6af","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-08 01:15:56.000000000","message":"Patch Set 7:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39927 : FAILURE in 6m 31s","accounts_in_message":[],"_revision_number":7},{"id":"0806ce52324a7e0e54f5b83b17362042eec0cc99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 02:21:08.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6010183dc1b248ec9159f3c23989fea9 : SUCCESS in 1h 04m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/10f35f698aed4bde92a214458930feee : SUCCESS in 1h 24m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0ffa12882b248bda819a57b43abed93 : SUCCESS in 17m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6c6eca598504dca90e24f5743b18c89 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d65e25b5494d72ab03426a39715bb1 : SUCCESS in 12m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8b6d1c88e6145ab9b918926f6000abb : SUCCESS in 13m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73fae2a1bea6455c89460eece06c4d56 : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc66daed1ef4091a91cad6637b22d19 : SUCCESS in 9m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c46cddd4ebc14ae9852cffe61f331ba1 : FAILURE in 1h 05m 03s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1d99fa1ca1194278b4e015e5748125a4 : FAILURE in 1h 18m 55s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/48aa56b3a92a43528ec7b2169c090b21 : SUCCESS in 1h 15m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e447a8f65cf3499c886221d0d17802e1 : FAILURE in 1h 33m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e8682ae8ea024675b90b54620e9fa75f : SUCCESS in 48m 25s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2390a1bf7deb4e458765218427fa123d : SUCCESS in 1h 12m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/d93be2e9952e4aadb61c344beed9fbe0 : SUCCESS in 1h 57m 35s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/93ff37e9354d45b39f855d6cd593e306 : SUCCESS in 15m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa7ce08998ad42ea917bfb9b06eaa39f : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":7},{"id":"e1feee0ad854214f2e87fb6ac2e38805fa4914e3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-08 03:05:38.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696746/7 : FAILURE in 2h 22m 30s","accounts_in_message":[],"_revision_number":7},{"id":"6549ca6ef3f591b2351943c1f000bde703b63f60","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-08 04:19:59.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/7/check/tempest-dsvm-full-xenial/1926ba2/ : FAILURE in 1h 27m 11s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/7/check/tempest-dsvm-full-xenial-py3/684799e/ : FAILURE in 1h 29m 09s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/7/check/grenade-dsvm-xenial/e94b221/ : FAILURE in 2h 56m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d2f850142e1daa599970f3f8ef7db3b1b4db65b5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-09 15:39:34.000000000","message":"Patch Set 7:\n\n(1 comment)\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"72a17497e219f0ac6fdbb7a1ca141cbfa8995856","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-10 11:06:58.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"e3398a60d3184bc2a5afeea170bf282f22dc7d4b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-10 11:26:57.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"cd053028acfa7771699257d288d75dfe1c660149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 13:50:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/d1b95d7ec4264dd7a5757eff9e0e9838 : SUCCESS in 1h 06m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9ea20cc388874b1aa5e9842a22744633 : SUCCESS in 1h 19m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b62471b0eccd486c87741a6dc8450d36 : SUCCESS in 14m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c63a6c6af9345cdb9885735860ad7e4 : SUCCESS in 23m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65a52895500f482ab2dfed71941b554b : SUCCESS in 10m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa980167a8694ccb8e6c2eb56852e196 : SUCCESS in 14m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dc8561bcc414a4d8e0ba7f7f9e67ba9 : SUCCESS in 15m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b984d9dff614cba83e1402ea41eea3a : SUCCESS in 9m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6399794881da4efc9adf4d40c2a41914 : FAILURE in 59m 42s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1a5504fb376a45e7936cb73879d92a55 : SUCCESS in 1h 01m 31s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6d1d49106bfc46c48fda90975fb3f497 : SUCCESS in 1h 15m 16s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3799eb32945a409f9f0e047f759db30b : SUCCESS in 1h 25m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ce9f22afcf4e471ca6ef90acb125c2da : SUCCESS in 50m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/94550eaef8ed4da5afa91c41982ba721 : SUCCESS in 1h 31m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/19c49f36198146568bd118cc0fb0f9e3 : SUCCESS in 1h 55m 09s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/060f0d29ecee45e8b8b1340ef80a7525 : SUCCESS in 17m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/213faebc56f14b9595e2b55569550e8e : SUCCESS in 1h 01m 05s","accounts_in_message":[],"_revision_number":8},{"id":"6ee2fba99040b3bbe6146ca0b1e5033b1fe83204","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-12-10 13:58:30.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/8/check/EMC_VxFlexOS_NOVA/123089a/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":8},{"id":"454d17c4b709fc64138ee986601737458b46eadf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-12-10 15:14:08.000000000","message":"Patch Set 8:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-46-696746-8 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":8},{"id":"5a3c1d13a588b6433e6ad1cf818f1c3c0bbbf596","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-10 18:25:27.000000000","message":"Patch Set 8:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/8 : SUCCESS in 2h 33m 08s","accounts_in_message":[],"_revision_number":8},{"id":"7ee74d3e3addcc408145836efe60f5e70390aa70","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-10 22:05:23.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/8/check-vote/ext-nova-zuul/bcc0a7f : FAILURE in 0s","accounts_in_message":[],"_revision_number":8},{"id":"3cc14f74dce96bfc3c4afd1fb042cd3613147015","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-11 15:29:03.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/8/check/tempest-dsvm-full-xenial/de83275/ : FAILURE in 1h 25m 05s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/8/check/tempest-dsvm-full-xenial-py3/4cb8e47/ : FAILURE in 1h 35m 33s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/8/check/grenade-dsvm-xenial/ad4c0c1/ : FAILURE in 2h 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9ab2fe0b2f2c9ea01ff31bb432a9018d5f3a42c5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-12 04:49:06.000000000","message":"Patch Set 8:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40013 : FAILURE in 12s","accounts_in_message":[],"_revision_number":8},{"id":"86e9293916ed386eb07d70f6f099dae09019419b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-12 10:29:02.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"aa2258b169a19c3f99224f7164082ae3ae259fe2","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-12 10:51:35.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"517e9431c3a75095e9d4968474c704b75edd8724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:55:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/abfa2e0ea15a4fc893a6be9c539cc374 : SUCCESS in 1h 07m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d7926b68c9244f86b7869c270a61294d : SUCCESS in 1h 24m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b42c7f5eee0421aa60f28a28e939f30 : SUCCESS in 23m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cfef77afe6a54e3983951f2fa4609e0f : SUCCESS in 16m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f223dd654224a48babf47bdd4fc7da6 : SUCCESS in 10m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0dc9d52f15c748c790314f9095a98375 : SUCCESS in 13m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bfb3a678e814f4891940cfe1b54f0d0 : SUCCESS in 16m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b253765e9d584653b8111133a41856f8 : SUCCESS in 11m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/273bd9e6ef8e41bc8161eb96de02b0ed : SUCCESS in 1h 03m 18s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ed17095468f74b0c8e75f0322001fe44 : SUCCESS in 1h 12m 29s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/72f9aa9e38b2434ea0ce03bf7bce269a : SUCCESS in 1h 22m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cecdd1cdcfa4a88a5a1823a0d79d33a : SUCCESS in 1h 15m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3bfc34408f7844d29e24f93c8e9af1ce : SUCCESS in 47m 57s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/87e9c2516dfe4798bae04b0f228e618e : SUCCESS in 1h 51m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/e6dc923b53f249ac994ae3588d20b76b : SUCCESS in 1h 32m 54s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ea9806f75d0b4b8e99ca5ee029f883fe : SUCCESS in 19m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa64a0f889c0478e95972177e10c2c73 : SUCCESS in 1h 07m 32s","accounts_in_message":[],"_revision_number":9},{"id":"14ad45fcc8698c1e25b02edcfdc3bca1478f6eb9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2019-12-12 13:06:24.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/9/check/EMC_VxFlexOS_NOVA/4c51e9c/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"332144d1e4d48b96c5fbcf07db68f8c35cad16ed","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-12 13:23:54.000000000","message":"Patch Set 9:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/9/check-vote/ext-nova-zuul/186dc62 : FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"de6d420b9c30086b6659b9d4c0fbf4525a6a836f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2019-12-12 14:36:43.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-46-696746-9 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":9},{"id":"11bc920a88193ce3bbc27684418f2190a340cb72","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-12 15:57:11.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40090 : FAILURE in 1h 49m 39s","accounts_in_message":[],"_revision_number":9},{"id":"fbca932e031610f6db29d1e230c6fbbf1783f138","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-12 18:07:59.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/9 : SUCCESS in 2h 41m 11s","accounts_in_message":[],"_revision_number":9},{"id":"c4804febf71a75b389b7b9f1c89686f8a8f7a2d0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-12 22:38:44.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/9/check/tempest-dsvm-full-xenial/7d3629b/ : FAILURE in 1h 31m 49s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/9/check/tempest-dsvm-full-xenial-py3/66b6d4a/ : FAILURE in 1h 32m 09s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/9/check/grenade-dsvm-xenial/adcb810/ : FAILURE in 2h 55m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"08b06c54335494fb349752c1bf03e8408424cd4a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-16 10:14:44.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"e3d38e6adcfd4329f2c662558c46a5e95d2f8d6a","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-16 10:34:44.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"8dcaa502ecc14c65cf6799cb2a5dabd700d8916d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-16 12:34:18.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/10/check-vote/ext-nova-zuul/0be70cb : FAILURE in 14m 20s","accounts_in_message":[],"_revision_number":10},{"id":"40386d0599f4b9f5c40b6ce34afdebda8656da5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 13:46:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/558da4365fef47eb95bbd34543d4edcf : SUCCESS in 57m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4cc1e8ce05cd4b96887b2498b4cb49aa : FAILURE in 1h 41m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a22738c00d4ba98b4d02ab91a6a6cd : SUCCESS in 19m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/131a60684fd54714b8c2410866d2cc85 : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e3b55b311744348b9cf6321e6b8509 : SUCCESS in 11m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa936217950744c7a78c46c737d16b5e : SUCCESS in 13m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/313edb44bda24b6da1f256d16446119c : SUCCESS in 16m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f501b52f160b4e968fb9e6efe08b7156 : SUCCESS in 10m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/33762496c4f64a4694d450169a65922f : SUCCESS in 1h 02m 13s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/701f1a36262a49058d002d7c466f2783 : SUCCESS in 1h 18m 47s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/86e8d2ad83464d2db77c2a392bbee0ea : SUCCESS in 1h 29m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f20781219a84bb199b69ba35d6d79b0 : SUCCESS in 1h 12m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1aa20a9bba7046acb7ae52931b1a7a65 : SUCCESS in 1h 08m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a90c4100a4c54331ae01a8cec5505887 : SUCCESS in 1h 43m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/2c9cc08ed22e4d09a6878186673983c7 : SUCCESS in 1h 30m 45s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f406dc610d5f48e3b93b64fd82c4695f : SUCCESS in 25m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3fb47d46e4b14c53a6a84ad602606bf6 : SUCCESS in 1h 07m 50s","accounts_in_message":[],"_revision_number":10},{"id":"20b0dc9fb2e76780c99f8c4dae638cc0cfa58741","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-16 14:02:04.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40262 : FAILURE in 1h 02m 36s","accounts_in_message":[],"_revision_number":10},{"id":"b32c7ea969e10ec14c06aef3349f963c8f19ee92","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-16 16:09:23.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/10/check/tempest-dsvm-full-xenial/6d45b9b/ : FAILURE in 1h 32m 02s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/10/check/tempest-dsvm-full-xenial-py3/ea1f5b2/ : FAILURE in 1h 31m 33s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/10/check/grenade-dsvm-xenial/f76e481/ : FAILURE in 2h 57m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ff4270596a59549baa58bb4d5fd704b9e5d78a37","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-16 16:34:02.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"3c50b6c02d573dcf79018eb7da0d5be6d326f5ee","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-16 16:40:36.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"89ff5deb532d3431d528fe3677176741d22126a2","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-16 17:52:07.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/10 : SUCCESS in 2h 54m 40s","accounts_in_message":[],"_revision_number":10},{"id":"d7a6c20bb0cceb1624c6e4fc2022e94c52d0f18f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 18:48:29.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/2108310f4a1740aaafc7ac071bf1f1e7 : FAILURE in 10m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ccaa9c21dd8c4385af68a9c7a8167d0e : SUCCESS in 1h 32m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d03f5256471a43b99a4fe450bd8f2841 : SUCCESS in 17m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f8b94c3299a4599a3b2f0d52312d5fb : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13af01a9af5743e5b42dcdf651a043a3 : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d606fa7681f4ac7bedf2975d8bb49d5 : SUCCESS in 14m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b30276015ef945abb25241caac53e9aa : SUCCESS in 14m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae3d35cc69b4988b4951cca1f9f1491 : SUCCESS in 11m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/48ecd8dabd6d43c4a3294196ea710b19 : SUCCESS in 56m 36s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3b80033d02ba46b0b7081a335e0d6f4d : SUCCESS in 1h 05m 36s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/290c226b6ac9418d8f9bb47a856f1fd3 : SUCCESS in 1h 26m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8add5f327ddd4d3bb87049d4f13ce77b : SUCCESS in 1h 40m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5b4b07b7de814928bd7fa8422b97daff : SUCCESS in 55m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0eb103740624477fac4e9ae688caf206 : SUCCESS in 1h 54m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/1becb2f44f1a4ba486c923d5e3f3c317 : SUCCESS in 1h 35m 17s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f3d0145587b743d08cb9cb6a60a02731 : SUCCESS in 17m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7ecb7ffa2de487f8af1e2a51f7c3098 : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":10},{"id":"0a2217bc7bb0aef5f9dc57e54d02fee18c1666d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-16 20:38:56.000000000","message":"Patch Set 10:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/10/check-vote/ext-nova-zuul/93d221a : FAILURE in 0s","accounts_in_message":[],"_revision_number":10},{"id":"f35bf25d3e06dfb2f7d25ed41e34db5a6b7f1935","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-16 21:06:28.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40281 : FAILURE in 1h 15m 59s","accounts_in_message":[],"_revision_number":10},{"id":"2208b9db1e360e465b096dcb91c46f02cab4ad4c","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-17 05:15:18.000000000","message":"Patch Set 10:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/10/check/tempest-dsvm-full-xenial/f83c06d/ : FAILURE in 1h 28m 33s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/10/check/tempest-dsvm-full-xenial-py3/656dc42/ : FAILURE in 1h 28m 37s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/10/check/grenade-dsvm-xenial/81393bb/ : FAILURE in 2h 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"969cc05e655ec7e5bef37904e585cb8d1211a6b3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-24 12:07:40.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"e4ef283538f690c2c63dd49b2a07d66a30c6b2aa","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-24 13:01:29.000000000","message":"Patch Set 11:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/11/check/pmem-tempest-plugin-filtered/5582557/ : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":11},{"id":"3ddc477a18a6769a23910c25826cba90cb5505e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-24 14:28:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8886d3fef98642258f73cf93ca86433a : SUCCESS in 1h 12m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f27a553068ce4e06a75ea3d2b60ccf1d : SUCCESS in 1h 39m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6a1dac2fe44c24be4a10b6b4af64bc : FAILURE in 21m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/323e89dd171044999ecdde5e72913cf2 : FAILURE in 14m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d9a052d522455fbd5cbbc5f88352e4 : SUCCESS in 11m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2717d3d206c4399a6510031c7d1acab : FAILURE in 12m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/73c5ec39808a4fd5ab71da11e30c3e81 : FAILURE in 20m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/923fe5dd89e14a268d34fb763a792aca : SUCCESS in 10m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/110b03b5ee83451ea06862f837a52426 : SUCCESS in 54m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ef9b6518a2844d1b39490d05874ca8b : SUCCESS in 1h 06m 01s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6922e19441164f1d95eb98672a0975d5 : SUCCESS in 1h 22m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1dd0fb5145be4d919c8fec8d972c1fd5 : SUCCESS in 1h 32m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cdf3a1e7fec44b228237b8b280d8481b : SUCCESS in 48m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/27605927c7e146e7b6743a20eb7a7ef6 : SUCCESS in 1h 37m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/9f5eb83809964f3785f48991a1fc59aa : SUCCESS in 1h 35m 08s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc3d0f528ba7447ea024bdf01779f8f2 : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/029dd877c53a408ea15a5ef2c6656fd9 : SUCCESS in 1h 01m 33s","accounts_in_message":[],"_revision_number":11},{"id":"1c90abde506ba6b0f83963b4d7f5fec4408af459","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2019-12-24 15:45:47.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/11/check-vote/ext-nova-zuul/55b9d28 : SUCCESS in 46m 25s","accounts_in_message":[],"_revision_number":11},{"id":"dbc4a0361ad5806a6f9e9a632a640733da5191f7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-24 17:10:23.000000000","message":"Patch Set 11:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/11 : SUCCESS in 2h 33m 33s","accounts_in_message":[],"_revision_number":11},{"id":"1d8a4216212327801da1a5975ac9102e53718627","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-26 02:30:25.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40473 : FAILURE in 1m 41s","accounts_in_message":[],"_revision_number":11},{"id":"effb849212afa37e4f27051ce574a2437e325d4e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-06 13:02:49.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"95f9c0ca4d5353a15013ae8df7cc82f7603fd35c","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-06 13:44:00.000000000","message":"Patch Set 12:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/12/check/pmem-tempest-plugin-filtered/facd65e/ : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":12},{"id":"14b724c6a721cb1e8cb5906f3fc7693dd26d54cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-06 14:31:17.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"817bb5bf8cc8e310210b01026c0bfca1baf455b4","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-06 15:26:20.000000000","message":"Patch Set 13:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/13/check/pmem-tempest-plugin-filtered/833586b/ : SUCCESS in 23m 47s","accounts_in_message":[],"_revision_number":13},{"id":"6e27fc81febdeffdec3df68da49a2454c518da38","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-01-06 16:10:28.000000000","message":"Patch Set 13:\n\nBuild Successful \n\n* Nova-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/46/696746/13/Nova-MACVTAP-ML2-Sriov : SUCCESS","accounts_in_message":[],"_revision_number":13},{"id":"04834e0b0e5fe25499b773c0988aeea217637dae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:21:12.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/cacc74c71b354017ba3626c8e66f957d : FAILURE in 1h 11m 05s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/335870a0cdd24814b2c52f0ad0338afd : FAILURE in 1h 22m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/239791614ae94be7bd10144317de58b5 : FAILURE in 20m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df100efffe9845d986701bfcdb06279b : FAILURE in 24m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2628bd2e4564e4d8bcdec4fc3168ded : SUCCESS in 11m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a1ae7b1c18c4d0f9b763d6da9467edd : FAILURE in 19m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/631a821a3d5c408ea223b09019f61878 : FAILURE in 15m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2fd3ebf318d4285864ca5beb1b182c7 : SUCCESS in 12m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee302e04ce194d0286cd617c350f3fce : SUCCESS in 56m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0109bc7baf2d4d7e9d60ee7a908c8c0d : FAILURE in 1h 10m 43s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/361516b439f74f369cda5063106b8636 : FAILURE in 1h 12m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/58a16e653f694ff2bd7b5bd06af1717a : SUCCESS in 1h 15m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d2a4d49f17964d7fae5e1e0540d64ea8 : SUCCESS in 51m 18s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/52ddd57c1207445a997edbd4d50d37b9 : FAILURE in 1h 40m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/9d94b3268a744932a0277321c480b58d : FAILURE in 1h 37m 12s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/65a81ec6576e40e98c0efeb49fed7395 : SUCCESS in 17m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77d8900322e941fda0605dbc09ac6172 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":13},{"id":"abb5bd400d6976bc67ab47b9c8307379cd39a532","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-06 17:48:47.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"2866b1ad33f873c79c81bd862ff3253751bdcb0b","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-01-06 17:49:30.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"9c3f9ae8f3021f555cfcfd394d2466e740e829b3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-06 17:49:31.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"f7d5131ee81ac356ddcb23f5add7b8d2c21e3b52","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-06 17:50:10.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"51d43632e8081b60b10eddadaa4a45f1e6f7db9a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-06 17:50:13.000000000","message":"Patch Set 13:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":13},{"id":"c1c1db7333f96163f7c6daeeb35fd830f50f947c","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-06 18:26:13.000000000","message":"Patch Set 14:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/14/check/pmem-tempest-plugin-filtered/1afa547/ : SUCCESS in 26m 36s","accounts_in_message":[],"_revision_number":14},{"id":"4f34f386398e72cec72787a3e2abccac2834dd71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 19:53:30.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bf8938a31b8643e488b80b47f29309ae : SUCCESS in 1h 03m 16s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f79afef92c8243f6a09d38afc4405a80 : SUCCESS in 1h 23m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72e5ae9f483e45f4b738391f45250286 : SUCCESS in 18m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d4e3d59d5b074bb8911f0d5379aaf9e2 : SUCCESS in 22m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c98f7df11bd04c45940f58bad8a032ef : SUCCESS in 12m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df2de4e9fa084cc3be5b6841c94f808f : SUCCESS in 15m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/84a774b5d2684c5c90efa03e69ae7893 : SUCCESS in 15m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3431d0b1d04e33a511995757897b8e : SUCCESS in 12m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/71aaa8ef4e5443a09031a617cec3b1e0 : FAILURE in 1h 24m 02s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/30146a4d3bcb489195355d16c1447293 : SUCCESS in 1h 36m 23s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/22794569e808470a90f195aa24f4119f : SUCCESS in 1h 43m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ebb34d7cdf640e3b8125ef47a35b1e5 : FAILURE in 1h 07m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/923d759340704ea9a8ce56b5b11980c1 : SUCCESS in 53m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/006e6b7d4eed46b784bdb9584c838f41 : SUCCESS in 1h 24m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/8042799f70e543e19c6939cf18295ac5 : SUCCESS in 1h 30m 10s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/853ce1957616496a97b592037fd22bee : SUCCESS in 18m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/952b940c2cfb4f93be9e1419a14b2a0c : SUCCESS in 1h 00m 54s","accounts_in_message":[],"_revision_number":14},{"id":"49fc9db6a9751f893da99116c08a30f3f65924da","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-06 21:24:40.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/14 : SUCCESS in 2h 41m 26s","accounts_in_message":[],"_revision_number":14},{"id":"170e337bcc70bcfc16cb27836edf9c8ad4081c00","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-06 23:46:46.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/14/check-vote/ext-nova-zuul/86187a4 : FAILURE in 1h 12m 23s","accounts_in_message":[],"_revision_number":14},{"id":"3d0ba77ebfb3bf1df034b83b6c1a0b2c875db3de","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-07 04:19:04.000000000","message":"Patch Set 14:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/14/check/tempest-dsvm-full-xenial/6163f33/ : FAILURE in 1h 26m 39s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/14/check/tempest-dsvm-full-xenial-py3/e431725/ : FAILURE in 1h 29m 54s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/14/check/grenade-dsvm-xenial/1dd64ca/ : FAILURE in 2h 55m 22s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"7841abdf2cf50bd5ed08cdc9c55ed852a57d2e55","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-01-08 10:04:25.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40560 : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":14},{"id":"ec6fa2953adc6d6492117a6a6579f43f709b9814","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-08 13:57:16.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"a8f088dc6695c92bbfb2b7b814f7c9702e97d8b6","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-08 14:39:08.000000000","message":"Patch Set 15:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/15/check/pmem-tempest-plugin-filtered/f9dab11/ : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":15},{"id":"58422db25b8b2bc2a6e76b718e32d01739d9b80c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-08 16:25:09.000000000","message":"Patch Set 15:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/15/check-vote/ext-nova-zuul/81f3f48 : FAILURE in 31m 30s","accounts_in_message":[],"_revision_number":15},{"id":"50246817a31d177a0edcf89485c84d4dda00c90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 16:25:43.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/01fa60a07e544c49af973aa1ef8b95a6 : SUCCESS in 1h 00m 18s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8d0bab2c52a2492aa70ee396262f18d0 : SUCCESS in 1h 23m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f14a89c6b14e406d8535ab439ec3fc39 : SUCCESS in 15m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a73c82a6fffb43f2875d0bba92dcd0bd : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6af6388510d84da5b9b84550776a7b60 : SUCCESS in 10m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f3778ad85ed4df895ff4e2d36041201 : SUCCESS in 14m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/da889035a03e4a2793df42a72cd19b9e : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf24c2c313614f6c95227cb8e6cc878a : SUCCESS in 10m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/55618e1f95f047ca8563bae86f14f2e5 : SUCCESS in 55m 40s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/342c32b722144de69c01a05662304913 : SUCCESS in 1h 04m 35s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1a43be249eae4efa907830740d377812 : SUCCESS in 1h 19m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/837eaf127665494dba5c6ddfb56e6edb : SUCCESS in 1h 13m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/22478a305893426d99851f9a36a08228 : SUCCESS in 1h 05m 46s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/28a2fa1dd2ca4d8f8dd96952ede7eb32 : SUCCESS in 1h 26m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/3b31b447b46048f78407b82334e5ef95 : SUCCESS in 1h 55m 56s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2109ad1797594cfd8d75579457c55fb0 : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06c09a2acc0442b59bbcf02276efb511 : SUCCESS in 1h 03m 33s","accounts_in_message":[],"_revision_number":15},{"id":"9871d037a7903ce52c2329c11801ade2931c8f07","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-01-08 16:33:34.000000000","message":"Patch Set 15:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/15/check/EMC_VxFlexOS_NOVA/ddf3c7a/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":15},{"id":"ccceb9f1c41f867c8541a72b264681dec8c206da","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-08 16:55:03.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696746/15 : FAILURE in 21m 54s","accounts_in_message":[],"_revision_number":15},{"id":"3d565df1a8bd37de446bff6fd6d669a1b800b528","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-01-08 17:40:27.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40605 : SUCCESS in 1h 09m 10s","accounts_in_message":[],"_revision_number":15},{"id":"7116754771aac3b6ae7726c9ed8e59bf5e7b9e31","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-01-08 17:58:28.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-46-696746-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"ac5afbdcb8605395d266ea26c3939fda1983ce67","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-08 18:21:29.000000000","message":"Patch Set 15:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/15/check/tempest-dsvm-full-xenial/539f94a/ : FAILURE in 1h 27m 48s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/15/check/tempest-dsvm-full-xenial-py3/f6a64da/ : FAILURE in 1h 32m 11s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/15/check/grenade-dsvm-xenial/97eeed5/ : FAILURE in 2h 54m 43s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"70a23519f85286fc6e68cbec056d3a93aaa2a377","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 10:54:52.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"0da1cf229e2523304baa100c35cf1d5aacd4c617","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 10:58:16.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"7440a243d2b4c85dd5ca40b95abfa12189e49493","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-15 11:28:46.000000000","message":"Patch Set 17:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/17/check/pmem-tempest-plugin-filtered/b2ed61f/ : SUCCESS in 24m 58s","accounts_in_message":[],"_revision_number":17},{"id":"cecffe63039cbba2c4bc73834244408f6a057836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 13:47:15.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8d616369bd474da9a70e408a988a6061 : SUCCESS in 1h 06m 54s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/895340f111de46afa559b225ae17a378 : SUCCESS in 1h 11m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9ba8ad5224f4a26b5a6da009c33a5b8 : FAILURE in 13m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d618471075804d3d8beff1a7c2acf5a0 : FAILURE in 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d143c6c8f1a41c9b348f275748d6b81 : SUCCESS in 11m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67b9ca2398ac45efbc57880b5127c853 : FAILURE in 12m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/70e66d9ed96645e5805455baa603249d : FAILURE in 14m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d66d9b111984975a89a13ba39276773 : SUCCESS in 9m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ac9992a22a414fb39daef473d6a5c250 : SUCCESS in 1h 06m 50s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/28e6b07b20134c148918fcc1be9c2cae : SUCCESS in 1h 09m 14s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c27c79d84edc471e835a02801d113969 : SUCCESS in 1h 32m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0df8b6384494f0cb25b592bf1b69ca9 : SUCCESS in 1h 28m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7e8352f50cba4b03b485d19a012f9dfa : SUCCESS in 51m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/dede596cd5ec4701936277795d609ea5 : SUCCESS in 1h 20m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/51edd94307bd4a6ea7767ae75c0ff1af : SUCCESS in 2h 01m 58s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac65aa6fca6a4e779880c23a4d5b11ea : SUCCESS in 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0b324033cd24b129f13918a3ccdf540 : SUCCESS in 1h 05m 53s","accounts_in_message":[],"_revision_number":17},{"id":"298c4d532a6640653b80b2ea211e79a4c2c5789a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-15 14:22:22.000000000","message":"Patch Set 17:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/17 : SUCCESS in 2h 34m 27s","accounts_in_message":[],"_revision_number":17},{"id":"f720934031a206377ce1aef8af86d9ef03512191","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-15 14:58:32.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"a2c7560c04c53eeb4b88f99290b47c255194f146","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-15 15:23:16.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696746/18 : FAILURE in 20m 10s","accounts_in_message":[],"_revision_number":18},{"id":"d345aca40537ae835ab89768032149fc7ad54fed","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-15 15:28:25.000000000","message":"Patch Set 18:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/18/check/pmem-tempest-plugin-filtered/6d59fa8/ : SUCCESS in 26m 28s","accounts_in_message":[],"_revision_number":18},{"id":"f277e5b36a175bde8021b2ba21c3beb8982dff34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 17:31:44.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1de3ad097241416dbfeb336682f18ab6 : SUCCESS in 1h 05m 22s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/393344e12bf14bb793cbf444958311e6 : SUCCESS in 1h 28m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87988b1fb4794ac59564324f5a1fa60a : SUCCESS in 15m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d0e2037a36443b6bc5ce27822d74a6f : SUCCESS in 17m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ebee40382fb47ae9307a0baa316ec70 : SUCCESS in 12m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cf1aad057ba4c8bb34aa5b06857671c : SUCCESS in 16m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1e3814323c74cc29676ca50cd78a31e : SUCCESS in 15m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e057d1fc482640299860e7576cceb802 : SUCCESS in 12m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/afa81776138b40a594b33b8a8e0e3d1a : SUCCESS in 1h 11m 41s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/51953e48910a435290bdd2cdc3cd365b : SUCCESS in 1h 05m 27s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ad07f5c8017c4e5597cf79b220e07349 : SUCCESS in 1h 20m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5429b7c5493043b3bd6ec9333d529eab : SUCCESS in 1h 21m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1b030ab9746c4f4e8eac552c49401590 : SUCCESS in 59m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/01c3644c1b924b5a9aeab382054443d0 : SUCCESS in 1h 31m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/cad8f0ddca9b4278bc93c5136a652959 : SUCCESS in 1h 28m 57s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7938aa0ff23a46a491ed369b89a0f46c : SUCCESS in 15m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5b13a364328f4a548b5b652121554941 : SUCCESS in 1h 06m 22s","accounts_in_message":[],"_revision_number":18},{"id":"4b212a808e266727112e32c208b04fe3a8a8064b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-01-15 17:32:48.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/18/check/EMC_VxFlexOS_NOVA/e3fea40/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":18},{"id":"ddf29487e9bee6a7e62aea5133d5ff7331fe0e04","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-01-15 18:19:09.000000000","message":"Patch Set 18:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-46-696746-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"d167b9e0f294a9d47262e49f0bbb422cc7956a9b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-15 18:40:44.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial/8b1f77e/ : FAILURE in 1h 24m 23s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial-py3/42a8893/ : FAILURE in 1h 30m 25s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/grenade-dsvm-xenial/a23990c/ : FAILURE in 2h 59m 04s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3a4a9b3fd52071af3330b6f2889dea2c0787adc0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-01-20 14:37:31.000000000","message":"Patch Set 18: Code-Review+2\n\n(2 comments)\n\nLGTM","accounts_in_message":[],"_revision_number":18},{"id":"777077523d58ede2612e3a4b69d35fd7e23d7ac3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-21 18:30:33.000000000","message":"Patch Set 18: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"4313842aed8fb39228b9c71ff6b07f0c8d72e289","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-21 19:53:15.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"44b61d6ab286b28165b36f7e6720cf0f0868a569","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 19:53:40.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"265fecd0bff042b4986bcdd14bcf1956a447564c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 22:19:04.000000000","message":"Patch Set 18: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f884a30b9abe4103942c59a9b02843c7 : SUCCESS in 1h 04m 55s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2cf5fc355c9d43aba83aec3efb706c86 : SUCCESS in 1h 28m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc10e0235f1247c898ba00f21a413be0 : SUCCESS in 20m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869c8a44743d43d3a8e594a7219aa354 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c06e5c511c784c0abdf7b257d67c07ca : SUCCESS in 16m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5b4dc78f855b430a92c487d9db527cc0 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c12700ce8ac04a28af92eadaf0fcfd04 : SUCCESS in 10m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/38247ea8dd044b179f146f7e0e8a97c2 : SUCCESS in 1h 12m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/95dd54fd086844f58d343e075c0846b7 : FAILURE in 15m 26s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3c8748fe74644094a4df0f6a5fb46535 : SUCCESS in 16m 42s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9bd61e96cc2d4d879028827a24144635 : SUCCESS in 1h 19m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/81924392cc434d28b82eb9f4da6a6538 : SUCCESS in 1h 19m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4e0250966874293a03962d7951a9bad : SUCCESS in 1h 41m 52s","accounts_in_message":[],"_revision_number":18},{"id":"b1a448c4d35e1f0419cc83b8392ab17030bc2d02","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-21 23:13:08.000000000","message":"Patch Set 18:\n\nrecheck PyYAML","accounts_in_message":[],"_revision_number":18},{"id":"f27cc5f17bae938d7b2e5237eee7af6dc086b720","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-21 23:24:45.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/46/696746/18/check/pmem-tempest-plugin-filtered/81ca482/ : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":18},{"id":"9f667b2352937a738a88c9f1a9241bfbe441824f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-21 23:42:56.000000000","message":"Patch Set 18:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696746/18 : FAILURE in 28m 54s","accounts_in_message":[],"_revision_number":18},{"id":"58cc4b29612aeeaa47ddbd69f2c5b1e6805583fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:23:45.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/08aadfab025c4527b0af611fd666574c : SUCCESS in 1h 05m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/236e328480b5472384fec550ebb74de9 : SUCCESS in 1h 24m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a36795728cc45bcae223d4842ec09d8 : SUCCESS in 17m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/836ed0a762894f4683e53a48d0675382 : SUCCESS in 15m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5e2637bf6d44d9bb97d2bf85f674b6 : SUCCESS in 10m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de9ae732061248808b666c53b34f15f2 : SUCCESS in 12m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/55f6dfa375ca44a096870617db609d25 : SUCCESS in 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef35ddd5780144e3847fa7d7ae82d954 : SUCCESS in 10m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eec926fa5cb346d5b2066dd84d75240b : SUCCESS in 55m 27s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8990679a9b264918954cc35ab440575d : SUCCESS in 1h 10m 23s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/71b9333d11184016b1cf8d1d60ddbdef : SUCCESS in 1h 09m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/acf3176824df4301a513f43568944628 : SUCCESS in 1h 14m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fe8d7c6ffdb64db79b8374385ac7988a : FAILURE in 19m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ecebf7bc0468484baba5a55fb18a22a7 : SUCCESS in 1h 49m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/3db8f56e525040eea0c5dfc6771ea08b : SUCCESS in 2h 06m 20s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6f4ddc372ec431eac382699e0d181f2 : SUCCESS in 20m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d7072c599f34699b0d334e57955056b : SUCCESS in 1h 16m 15s","accounts_in_message":[],"_revision_number":18},{"id":"388db6a0919c2771d6fe4a0ea894f71ef83e1f16","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-22 01:56:53.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial/c4fa028/ : FAILURE in 41m 38s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial-py3/c0aa9b6/ : FAILURE in 50m 54s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/grenade-dsvm-xenial/824d600/ : FAILURE in 2h 38m 53s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"053f221239158224b535eb0c313660ca218fc1d8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-22 02:50:57.000000000","message":"Patch Set 18:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/18/check-vote/ext-nova-zuul/3b75503 : FAILURE in 38m 20s","accounts_in_message":[],"_revision_number":18},{"id":"21b8c577a6c3385279b551b8bc541e7da31bdebd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-01-22 20:53:32.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"84bda5e0f8742d703e980f22a2eef8dfae6e48c9","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-22 20:56:16.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered finger://d0fe4b8af32b/46f3c064367943de849ad279c8e9fdb9 : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":18},{"id":"0ed15af8db401e9ae8bf28f3c895fcb2b7163ba8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 23:10:57.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b2d45631c2b645acbcde86399350879e : FAILURE in 1h 27m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/32bd336a6d584ea1ae3bfd2c49513e65 : SUCCESS in 1h 21m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/344b83544f394992ba74ee02b51c7c60 : SUCCESS in 30m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59d359bfc2924dd5ab88042acfc049cc : SUCCESS in 31m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79bdf5f7504d49a5ba5ac85d319b17e9 : SUCCESS in 12m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27738b36756a4b73aacaf02861cbb302 : SUCCESS in 16m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e1a9df0d754415e893769cc72a35259 : SUCCESS in 27m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6f4582c952f4067972e8a72e17abfa4 : SUCCESS in 12m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2bc5ff8e6105425a9c6ca894fb10d2e4 : SUCCESS in 1h 07m 45s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/998077f31011463892d46f7484a49a77 : SUCCESS in 59m 05s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/205e6c2749d8464cbe0b9a3f864c097e : SUCCESS in 1h 37m 43s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e47bdbe0de34d5586b7463173df35b0 : SUCCESS in 1h 42m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bc736441306245b8b4688b4ce3713101 : SUCCESS in 1h 16m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/326cc2dbb05e4a89b37cd7eb5a392d93 : SUCCESS in 1h 51m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/3ef34293aba941a5b9e25dacdde58d16 : SUCCESS in 2h 11m 03s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0972e894a94b4492b6c73d6ada2f3a92 : SUCCESS in 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36dc89750af44e5bbe0e216014f78ff8 : SUCCESS in 1h 24m 39s","accounts_in_message":[],"_revision_number":18},{"id":"d39f8d4017e5311cfff044e6cf2907d655a6b01b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-22 23:24:39.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/18 : SUCCESS in 2h 30m 05s","accounts_in_message":[],"_revision_number":18},{"id":"35b9e596e37670527ea7006175722bdf6c87302d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-22 23:38:03.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial/21d61f4/ : FAILURE in 45m 31s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial-py3/edf57e3/ : FAILURE in 47m 55s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/grenade-dsvm-xenial/ba84c0a/ : FAILURE in 2h 42m 32s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b29918d936945d2df6ddd8b035abf6ed7d951bf5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-22 23:51:52.000000000","message":"Patch Set 18:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/18/check-vote/ext-nova-zuul/e4a1962 : FAILURE in 28m 57s","accounts_in_message":[],"_revision_number":18},{"id":"43895f780328b33f8b547d5a4754630643f68041","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-23 00:01:38.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"e91b5cefe601bf721ac8bd7732636f25980c407e","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-23 00:05:31.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered finger://d0fe4b8af32b/0bff7275852940ada1a5d3076eb8ae45 : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":18},{"id":"166d8f96a7d8a481b292592062b069537cef9acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 01:57:08.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/bfb9b5085a0c4dd9b9b1b9c984da679d : SUCCESS in 1h 15m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/181d5fa8d8fe45e089d8fd976870336b : SUCCESS in 1h 41m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b15f3dab8d0440284f9ee18a839bb17 : SUCCESS in 20m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe2f0d020abc4faf9c8ae1a4df02ebc9 : SUCCESS in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a5234d606546ffad96681a9f6957f0 : SUCCESS in 10m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0a56daf3a84ee4bca6a7823cc15606 : SUCCESS in 13m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87f78cfc35504a54b80619c86b10450b : SUCCESS in 18m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0613097acbe4b1cb5fde5485c2a83f4 : SUCCESS in 9m 55s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c3391d13387b463abeaa8efc1048fdb2 : SUCCESS in 53m 03s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3176793e91964c408b98aa4165aeef06 : SUCCESS in 1h 10m 16s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/22dd206b4d79484a8360b7984dfe8343 : SUCCESS in 1h 25m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/18c9d01a92494f4592be01f9a357af83 : SUCCESS in 1h 23m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/66b76333273d4668840ec2f92707adbe : FAILURE in 1h 03m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b1842750ec294a50bee84b5159b8e27f : SUCCESS in 1h 30m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/d5f7b30edba44d5e86bc0432a055016b : SUCCESS in 1h 49m 39s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e58caf26f29b49c69a20ed8a1bd83ac1 : SUCCESS in 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edd520b46908412cb13d0e12e1980953 : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":18},{"id":"5b920e8365ce8eb8acc336770d1be8f61351e968","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-23 02:36:39.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/18 : SUCCESS in 2h 34m 02s","accounts_in_message":[],"_revision_number":18},{"id":"17eb47afb2b33ba7f13ed1e801e5841457a058db","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-23 02:48:34.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial/aedb567/ : FAILURE in 41m 38s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/tempest-dsvm-full-xenial-py3/6d987fa/ : FAILURE in 44m 58s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/18/check/grenade-dsvm-xenial/093a679/ : FAILURE in 2h 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"1a1c7fa9c256f33af01b96e25e2ab193f7e9b805","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-23 06:40:26.000000000","message":"Patch Set 18:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/18/check-vote/ext-nova-zuul/7d6e90d : FAILURE in 30m 01s","accounts_in_message":[],"_revision_number":18},{"id":"61b8b82d047a9b82cf5d268dba10aafcca6d55b2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-23 12:10:21.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"a9379b39be902db39b78cf7396b7bb3efa92ef63","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-23 12:10:56.000000000","message":"Patch Set 19: Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"16d1c1ba411ffbbed5d4a3683d93fa41a482b9e3","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-01-23 12:15:27.000000000","message":"Patch Set 19:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered finger://d0fe4b8af32b/a6f12ad8cf2c48e598be754296b3be8b : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":19},{"id":"f1397fb5baa22cd903cd4f1f12183f976eedd6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 14:37:19.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7b3149ddfd8146b186f2c6cb67dbaa7f : SUCCESS in 1h 01m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a823066dbd25470ca9f6144062b6bb6b : SUCCESS in 1h 15m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b6f0d948bb045fca240d4df10ff24cf : SUCCESS in 21m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf0acbdbe2ed4e5699bd220a241d6f9f : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9ce72f27fa4f1997384dbda227d613 : SUCCESS in 12m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6f02bcaa7d041a0927451025a13ba52 : SUCCESS in 15m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19bfad2c1971466d9163a5ba0b96f788 : SUCCESS in 16m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f542f5bf724cea835e2c7238850a82 : SUCCESS in 10m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9a302beb2df74b09a540f49423757275 : SUCCESS in 57m 04s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3828852f2a3d43e49cc4d0591d2203f7 : SUCCESS in 1h 03m 12s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2f56987c43034f659e428aa3de0e6042 : SUCCESS in 1h 27m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e53ab2517594f4f89b158a81079c2d7 : SUCCESS in 1h 39m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/57f8a4eaa9b44f13bd54900775b5eb2a : SUCCESS in 52m 28s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/111443f21db24eb4aa1bf53b6108f24c : SUCCESS in 1h 24m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/a9d52278519f401cab8b9232563d90dc : SUCCESS in 1h 31m 55s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f17873d8300d48ddaf6e269453be02ea : SUCCESS in 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6597f69e11264cd1ac74a188c822def3 : SUCCESS in 1h 17m 37s","accounts_in_message":[],"_revision_number":19},{"id":"3474721225206ad7f45ad92c457ce47a4a75fa5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 14:37:48.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"ae714021141321f5b1e806eaf54a5b18ae123f8f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-01-23 14:37:49.000000000","message":"Patch Set 19:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/46/696746/19/check/EMC_VxFlexOS_NOVA/da79e5d/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":19},{"id":"03341483b494b64fb9a3822d06112213c8fd3836","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-01-23 14:51:51.000000000","message":"Patch Set 19:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696746/19 : SUCCESS in 2h 39m 04s","accounts_in_message":[],"_revision_number":19},{"id":"71ea7c15818df4e69ddfc8fb75f387e5cb2ee72b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-01-23 15:11:40.000000000","message":"Patch Set 19:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/46/696746/19/check-vote/ext-nova-zuul/d2ae3d3 : FAILURE in 35m 57s","accounts_in_message":[],"_revision_number":19},{"id":"4115a5ddefe96958c416edda4df46d51bcadfe1a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-01-23 15:48:41.000000000","message":"Patch Set 19:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/19/check/tempest-dsvm-full-xenial/596ec23/ : FAILURE in 43m 04s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/19/check/tempest-dsvm-full-xenial-py3/0524d22/ : FAILURE in 48m 47s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/46/696746/19/check/grenade-dsvm-xenial/1a4dd72/ : FAILURE in 2h 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"3592f3bf728eb8105785b3fc27cf57f6c5aec22d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 16:30:23.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/86b349057c1643e1ab96c7feaa22b408 : SUCCESS in 1h 13m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d1ad5c99d95843c9a6fff0ff215848b7 : SUCCESS in 1h 43m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3688c7dace214bc7ba90307bef54ef4e : SUCCESS in 17m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccd527a627044375abe64a321c6708ee : SUCCESS in 9m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a679b66f994d4d819a4f60cc64bfe093 : SUCCESS in 14m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf48c9b210914a70a6d7a37c3d0b9fdb : SUCCESS in 16m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a58fa465b894264a984878bfcee2e72 : SUCCESS in 13m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f93bc8adf781470aa10d934826b4c29f : SUCCESS in 1h 22m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d86ec9d32450448f89fe8244309e097b : SUCCESS in 59m 48s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7d1839eb8e247d586f754a8e4f1dc78 : SUCCESS in 17m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3ac2c0bf1fd94042bbd3a6b009a699a7 : SUCCESS in 1h 41m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/5fe003a8caad49e5b28d83c910be864d : SUCCESS in 1h 28m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0df7e4f7ec8b445ab32302452c2bd204 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":19},{"id":"4263dc0f78e8ea135f84dad094603896c1132b3d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 16:30:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":19},{"id":"68cb9e639cf1e654f8e8584b82592ba2f0003a54","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 16:34:25.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b31a4b0ef44915a8a666d27be5e573 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"110a683486b750488ee147b16a17451750af4c5e","revisions":{"9f5b3e8f140678acaba53f08e71abc17add774c6":{"kind":"REWORK","_number":1,"created":"2019-11-29 17:19:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/1"}}},"commit":{"parents":[{"commit":"04b23d3e6b894022062769a0ba53d6d828102f9a","subject":"Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04b23d3e6b894022062769a0ba53d6d828102f9a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 17:18:39.000000000","tz":0},"subject":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027","message":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027\n\nAnother layer of complexity removed.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f5b3e8f140678acaba53f08e71abc17add774c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9f5b3e8f140678acaba53f08e71abc17add774c6"}]},"branch":"refs/heads/master"},"409f35d9baf6e77ddccb9e5b3836d802cd9c4089":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-29 17:20:33.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/2"}}},"commit":{"parents":[{"commit":"8ac05e8b2c98adf1b14640eddfb85d32f5247a14","subject":"Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ac05e8b2c98adf1b14640eddfb85d32f5247a14"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 17:20:03.000000000","tz":0},"subject":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027","message":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027\n\nAnother layer of complexity removed.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/409f35d9baf6e77ddccb9e5b3836d802cd9c4089"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/409f35d9baf6e77ddccb9e5b3836d802cd9c4089"}]},"branch":"refs/heads/master"},"4a931d751e52bfe69b7cf3179a8167ed3a2bcf1e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-30 11:41:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/3"}}},"commit":{"parents":[{"commit":"0173106905548526ac36729f8efcff6f25600f42","subject":"Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0173106905548526ac36729f8efcff6f25600f42"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-30 11:41:23.000000000","tz":0},"subject":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027","message":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027\n\nAnother layer of complexity removed.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a931d751e52bfe69b7cf3179a8167ed3a2bcf1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4a931d751e52bfe69b7cf3179a8167ed3a2bcf1e"}]},"branch":"refs/heads/master"},"670b89d69eb1d7fe846047ac2bb9a6a03b124e3e":{"kind":"REWORK","_number":4,"created":"2019-11-30 16:04:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/4"}}},"commit":{"parents":[{"commit":"03a89d7a4e27d587b6883edc483fd446d18b0e31","subject":"Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03a89d7a4e27d587b6883edc483fd446d18b0e31"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-30 16:03:58.000000000","tz":0},"subject":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027","message":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027\n\nAnother layer of complexity removed.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/670b89d69eb1d7fe846047ac2bb9a6a03b124e3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/670b89d69eb1d7fe846047ac2bb9a6a03b124e3e"}]},"branch":"refs/heads/master"},"6a3d3ad7d1b8f886453777be923df60a73f7e79a":{"kind":"REWORK","_number":5,"created":"2019-11-30 18:13:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/5"}}},"commit":{"parents":[{"commit":"03a89d7a4e27d587b6883edc483fd446d18b0e31","subject":"Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/03a89d7a4e27d587b6883edc483fd446d18b0e31"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-30 18:11:11.000000000","tz":0},"subject":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027","message":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027\n\nAnother layer of complexity removed.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a3d3ad7d1b8f886453777be923df60a73f7e79a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6a3d3ad7d1b8f886453777be923df60a73f7e79a"}]},"branch":"refs/heads/master"},"5ea61d45fccef3dbe87ef02dcabbb04d1be77900":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-03 18:31:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/6"}}},"commit":{"parents":[{"commit":"176a0993ba36259ca9b009e7c3ccd275073a357f","subject":"Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/176a0993ba36259ca9b009e7c3ccd275073a357f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-03 13:07:52.000000000","tz":0},"subject":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027","message":"Rename \u0027nova.network.security_group.neutron_driver\u0027 -\u003e \u0027nova.network.security_group\u0027\n\nAnother layer of complexity removed.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ea61d45fccef3dbe87ef02dcabbb04d1be77900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ea61d45fccef3dbe87ef02dcabbb04d1be77900"}]},"branch":"refs/heads/master"},"033506cb24567f56a5c8408f7f9df34ccd0adb99":{"kind":"REWORK","_number":7,"created":"2019-12-06 19:21:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/7"}}},"commit":{"parents":[{"commit":"e14ee09676b141a43177cb12ccc2e2a64a498854","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e14ee09676b141a43177cb12ccc2e2a64a498854"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-06 19:21:03.000000000","tz":0},"subject":"nova-net: Make the security group API as module","message":"nova-net: Make the security group API as module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/033506cb24567f56a5c8408f7f9df34ccd0adb99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/033506cb24567f56a5c8408f7f9df34ccd0adb99"}]},"branch":"refs/heads/master"},"33862e1ac95fe519570e4cf8165db2d3d9ff6a5e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-12-10 11:06:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/8"}}},"commit":{"parents":[{"commit":"02d5929a3e0528239ae9c898a1790377d1cccf70","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02d5929a3e0528239ae9c898a1790377d1cccf70"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-10 11:06:16.000000000","tz":0},"subject":"nova-net: Make the security group API as module","message":"nova-net: Make the security group API as module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33862e1ac95fe519570e4cf8165db2d3d9ff6a5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/33862e1ac95fe519570e4cf8165db2d3d9ff6a5e"}]},"branch":"refs/heads/master"},"c0586129fc1d133379c419c1313b60fe763e0dfe":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2019-12-12 10:29:02.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/9"}}},"commit":{"parents":[{"commit":"18a9c75ae82fcd801136ce32d200fb210cf2f884","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18a9c75ae82fcd801136ce32d200fb210cf2f884"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-12 10:27:54.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0586129fc1d133379c419c1313b60fe763e0dfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c0586129fc1d133379c419c1313b60fe763e0dfe"}]},"branch":"refs/heads/master"},"55d1b7b4204ab76d352a43c496d85d3bea2f7a3f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2019-12-16 10:14:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/10"}}},"commit":{"parents":[{"commit":"823ccddb9e5e118917904ad691df7715e256061a","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/823ccddb9e5e118917904ad691df7715e256061a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-16 10:10:19.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55d1b7b4204ab76d352a43c496d85d3bea2f7a3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/55d1b7b4204ab76d352a43c496d85d3bea2f7a3f"}]},"branch":"refs/heads/master"},"dc57e978ae0bb50ba53356979eb5a0ceef4e048d":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2019-12-24 12:07:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/11"}}},"commit":{"parents":[{"commit":"d5a0492faf40c03c642832a9cbfd9020fd346614","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d5a0492faf40c03c642832a9cbfd9020fd346614"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-12-24 11:10:45.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc57e978ae0bb50ba53356979eb5a0ceef4e048d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dc57e978ae0bb50ba53356979eb5a0ceef4e048d"}]},"branch":"refs/heads/master"},"04670e5a347595bc239069490bc2135ce09abc35":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2020-01-06 13:02:49.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/12"}}},"commit":{"parents":[{"commit":"2491ce79e1a12756788fc4ccd40a288e84699578","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2491ce79e1a12756788fc4ccd40a288e84699578"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-06 12:50:36.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04670e5a347595bc239069490bc2135ce09abc35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/04670e5a347595bc239069490bc2135ce09abc35"}]},"branch":"refs/heads/master"},"65680934b49ab199b1ac69ac91b5784b5c1d2eae":{"kind":"REWORK","_number":13,"created":"2020-01-06 14:31:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/13"}}},"commit":{"parents":[{"commit":"ca2efbe293725090aaf1dba1115a3965399d68ea","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ca2efbe293725090aaf1dba1115a3965399d68ea"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-06 14:30:43.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65680934b49ab199b1ac69ac91b5784b5c1d2eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/65680934b49ab199b1ac69ac91b5784b5c1d2eae"}]},"branch":"refs/heads/master"},"0324b41ea8bf08146934f17a1d02fc1cc701f47e":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2020-01-06 17:48:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/14"}}},"commit":{"parents":[{"commit":"f58be593ed50d57e006c2d7dee9a893d37cce3bf","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f58be593ed50d57e006c2d7dee9a893d37cce3bf"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-06 17:47:46.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0324b41ea8bf08146934f17a1d02fc1cc701f47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0324b41ea8bf08146934f17a1d02fc1cc701f47e"}]},"branch":"refs/heads/master"},"23b5e7ea0ef55adcd300d85c49680d89adffd1cb":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2020-01-08 13:57:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/15"}}},"commit":{"parents":[{"commit":"f3fab5667f375ee69a01c735893ec1f69ca00de5","subject":"nova-net: Rename \u0027nova.network.neutronv2\u0027 -\u003e \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f3fab5667f375ee69a01c735893ec1f69ca00de5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-08 13:54:25.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23b5e7ea0ef55adcd300d85c49680d89adffd1cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/23b5e7ea0ef55adcd300d85c49680d89adffd1cb"}]},"branch":"refs/heads/master"},"f8d9810a763ae6e8f538283ee1a3e8dc69d1e671":{"kind":"REWORK","_number":16,"created":"2020-01-15 10:54:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/16"}}},"commit":{"parents":[{"commit":"977810d378cc56ae63b5e90bcbc6bd2bd933938c","subject":"nova-net: Remove layer of indirection in \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/977810d378cc56ae63b5e90bcbc6bd2bd933938c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-15 10:52:27.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8d9810a763ae6e8f538283ee1a3e8dc69d1e671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f8d9810a763ae6e8f538283ee1a3e8dc69d1e671"}]},"branch":"refs/heads/master"},"cfdc7f09658f932de9222dce5fbec6ad1c6fd943":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2020-01-15 10:58:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/17"}}},"commit":{"parents":[{"commit":"90af23d0858eb44f87d4e1f4f8b81518ae5e86ce","subject":"nova-net: Remove layer of indirection in \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/90af23d0858eb44f87d4e1f4f8b81518ae5e86ce"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-15 10:57:43.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfdc7f09658f932de9222dce5fbec6ad1c6fd943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cfdc7f09658f932de9222dce5fbec6ad1c6fd943"}]},"branch":"refs/heads/master"},"4e72eb7d8414be165c50b85fccbb771b52769994":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2020-01-15 14:58:32.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/18"}}},"commit":{"parents":[{"commit":"fadeedcdeaa949ae85938e78e2f34e75b9383721","subject":"nova-net: Remove layer of indirection in \u0027nova.network\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fadeedcdeaa949ae85938e78e2f34e75b9383721"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-15 14:57:53.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e72eb7d8414be165c50b85fccbb771b52769994"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4e72eb7d8414be165c50b85fccbb771b52769994"}]},"branch":"refs/heads/master"},"110a683486b750488ee147b16a17451750af4c5e":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2020-01-23 12:10:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/696746/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/46/696746/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/46/696746/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/46/696746/19"}}},"commit":{"parents":[{"commit":"cc0e1950a0679321a207011f327f79a147cf3758","subject":"Merge \"requirements: Limit hacking to one minor version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cc0e1950a0679321a207011f327f79a147cf3758"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2019-11-29 11:17:06.000000000","tz":0},"subject":"nova-net: Make the security group API a module","message":"nova-net: Make the security group API a module\n\nWe\u0027re wrestling with multiple imports for this thing and have introduced\na cache to avoid having to load the thing repeatedly. However, Python\nalready has a way to ensure this doesn\u0027t happen: the use of a module.\nGiven that we don\u0027t have any state, we can straight up drop the class\nand just call functions directly. Along the way, we drop the\n\u0027ensure_default\u0027 function, which is a no-op for neutron and switch all\nthe mocks over, where necessary.\n\nChange-Id: Ia8dbe8ba61ec6d1b8498918a53a103a6eff4d488\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110a683486b750488ee147b16a17451750af4c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/110a683486b750488ee147b16a17451750af4c5e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
