)]}'
{"id":"openstack%2Fnova~696851","triplet_id":"openstack%2Fnova~master~I1734d4b8e4d7f226f203a3cf81ec5c2d87baa107","project":"openstack/nova","branch":"master","topic":"bug/1699054","attention_set":{"28889":{"account":{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},"last_update":"2022-08-16 08:33:44.000000000","reason":"Someone else replied on the change"},"26286":{"account":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"last_update":"2022-08-16 08:33:44.000000000","reason":"Someone else replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I1734d4b8e4d7f226f203a3cf81ec5c2d87baa107","subject":"WIP:Drop unreliable host_aggregate_map","status":"NEW","created":"2019-12-02 10:11:40.000000000","updated":"2022-08-16 10:45:34.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ac1d78a658153bc5fd974b795ca2c45991e62da6","_number":696851,"virtual_id_number":696851,"owner":{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-08-16 09:56:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 10:37:45.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-12-02 11:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-02 12:40:19.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-02 13:12:05.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-03 07:09:57.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2022-08-16 08:34:47.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"0141d398bdf7a265b9bab5b8f3915a84591a14e6","author":{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},"date":"2019-12-02 10:11:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a99f49c4c14c74a1047d6bf4cf9003e007733246","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-02 10:37:45.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/51/696851/1/check/pmem-tempest-plugin-filtered/1abdec9/ : SUCCESS in 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"58a24aa8804f5da3381f3d0ac43e08a4734630b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 11:55:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/dfd72a9d15ba43c48cfcc99f98816e8f : SUCCESS in 1h 06m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f6957ce87d9e4c1a8b1b76162907e686 : FAILURE in 1h 12m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d828ac7cd8d4c8095248f5b1890d5df : FAILURE in 14m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c219dc8b05545aa994f043abba3cbb4 : FAILURE in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31bd758832d4f758e7e30ff37bb2a9d : SUCCESS in 10m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1d4afbfc1e9413e9af27145fe7eb647 : FAILURE in 14m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6680486b37164ab3a78a30168e85c2b4 : FAILURE in 14m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcfd8edb9844b95be13c7e17e41551e : SUCCESS in 9m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6480f3ec34e143b1ab31cd48f222cef5 : SUCCESS in 52m 03s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe45201da45f42c5ade209e257399965 : SUCCESS in 57m 23s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/649cde8fd4ac41908e48771c128a0efe : SUCCESS in 1h 16m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a81e5b06c4654a8292503b6603f276c8 : SUCCESS in 47m 07s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/633209a2c08b4c968e718e30c0ef7926 : SUCCESS in 1h 12m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/0d28d80afc544da293a7638b24b84e8e : SUCCESS in 1h 35m 09s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4d7359491c2546448138c21b7ff71aa9 : SUCCESS in 17m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38203c0c234340aca8d5bfaa390d7f81 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"20b67872d093162df98dc663a166cbd4e70be8da","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-02 12:40:19.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/696851/1 : SUCCESS in 2h 27m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b1f5dda4a14b141523b3d1bae7c85de8a4d0ee14","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-02 13:12:05.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/51/696851/1/check/tempest-dsvm-full-xenial/bde4324/ : FAILURE in 1h 58m 21s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/51/696851/1/check/tempest-dsvm-full-xenial-py3/41adcea/ : SUCCESS in 2h 50m 59s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/51/696851/1/check/grenade-dsvm-xenial/2d7f623/ : FAILURE in 2h 58m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13fe8a49391d96f9fb9c90ecdb26f53347076ab1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2019-12-03 07:09:57.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-39707 : FAILURE in 2h 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f3d6f9279bda9e4d7a5a2572cfbac567aae122d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"date":"2022-08-16 08:15:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"76933e40754c1ba84272d288760824acfa6809ca","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-08-16 08:33:44.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696851/2/iscsi : FAILURE in 16m 47s","accounts_in_message":[],"_revision_number":2},{"id":"0216fae9684bec285e86a5cfe47a9add099d8d19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-08-16 08:34:47.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/51/696851/2/openstack-sriov-nova-master-macvtap : ABORTED\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/51/696851/2/openstack-sriov-nova-master-direct : ABORTED","accounts_in_message":[],"_revision_number":2},{"id":"5b4c251aaf8ff6804fd04a3f285213c322668c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 08:49:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c83b047447c45bb9ee3770caf3613cb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/501f56300d36466ea370030f90b56873 : FAILURE in 31m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/441587558833405fbf90edc756f430d2 : FAILURE in 16m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4cb23db04e1730838d981802cde1d5ba3222251","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-08-16 09:50:43.000000000","message":"Patch Set 2:\n\nBuild Successful \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/51/696851/2/openstack-sriov-nova-master-macvtap : SUCCESS\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/51/696851/2/openstack-sriov-nova-master-direct : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"0bcf3e69b11860286440e333b3b2c51570569a44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 09:56:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/913a72c9476e4a5abc1e9c13a0b47092\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/95985be773874b75bb5a1415c83ea50c : SUCCESS in 1h 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ad99f05552a4b40a1813f6022449efb : SUCCESS in 51m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6dfcd01bef942f8bea3ce32a90af340 : FAILURE in 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b7ffa9c38ec4747bf8752cb5de60690 : SUCCESS in 12m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8651cf4e9d464e93aa3316294340ffdd : FAILURE in 13m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/498cfb649c0d4fc785b0babc3531eabe : FAILURE in 13m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00a333c5eeac49d3acf13405208b2d63 : FAILURE in 12m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a7e66910fe64cc1bd2628c2ef45b731 : SUCCESS in 10m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ca072dcf73754eab93a638d3becbab61 : SUCCESS in 1h 07m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b58c6c46d204769a44ea4bac45b95d1 : SUCCESS in 1h 03m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0635b614a0c54f19b87ec10650231cc2 : SUCCESS in 54m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/04ea5baeb72643b38175b1228838ee6f : SUCCESS in 57m 38s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5aae95810d0f41808473fb3bcb9b5847 : SUCCESS in 1h 37m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/1efd1802d73342e085b4c2a5ad3b5281 : SUCCESS in 1h 38m 34s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/a0e87f57bb944708a89ae6790f6277d3 : SUCCESS in 55m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/83c323307b5543ef98efe8479465c296 : SUCCESS in 6m 09s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d4b1afa7b4c546e3b54da910086b25f8 : SUCCESS in 21m 00s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5f59855bcb948e191c9901151daed11 : SUCCESS in 19m 45s\n- nova-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6a431f5a4d84dc791184938671bc740 : SUCCESS in 21m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/518bfbaae6324a1d963ec5577c5e092e : SUCCESS in 1h 13m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c226215d28440ec9d798449bd817077 : SUCCESS in 55m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b00fbc79a29941a8b084837fdfefc9a0 : SUCCESS in 25m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/46595dddc0d147848b72e7d3452d5b34 : SUCCESS in 43m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9cab920b67911cf3c4f209d35ff034387aeb707","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-08-16 10:10:35.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/51/696851/2/check/tempest-dsvm-full-focal-py3/089c052/ : SUCCESS in 1h 51m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ac1d78a658153bc5fd974b795ca2c45991e62da6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-08-16 10:45:34.000000000","message":"Patch Set 2:\n\nBuild Failed \n\n* openstack-sriov-nova-master-macvtap http://13.74.249.42/refs/changes/51/696851/2/openstack-sriov-nova-master-macvtap : FAILURE\n\n* openstack-sriov-nova-master-direct http://13.74.249.42/refs/changes/51/696851/2/openstack-sriov-nova-master-direct : SUCCESS","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1a6d055bd4ca219ac45ff0c7f554208277d41e52","revisions":{"3dfa37e7cd86bb1928feb1b9579f163c7a3da09e":{"kind":"REWORK","_number":1,"created":"2019-12-02 10:11:40.000000000","uploader":{"_account_id":28889,"name":"Guo Jingyu","email":"guojy8993@163.com","username":"pandatt"},"ref":"refs/changes/51/696851/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/696851/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/696851/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/696851/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/696851/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/696851/1"}}},"commit":{"parents":[{"commit":"56eac06561bc843dcf7497e8b7ac3027f62544c1","subject":"Merge \"hyperv: Remove vestigial nova-network support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/56eac06561bc843dcf7497e8b7ac3027f62544c1"}]}],"author":{"name":"pandatt","email":"guojy8993@163.com","date":"2019-12-02 10:02:26.000000000","tz":480},"committer":{"name":"pandatt","email":"guojy8993@163.com","date":"2019-12-02 10:02:26.000000000","tz":480},"subject":"WIP:Drop unreliable host_aggregate_map","message":"WIP:Drop unreliable host_aggregate_map\n\nChange-Id: I1734d4b8e4d7f226f203a3cf81ec5c2d87baa107\nRelated-bug: #1699054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3dfa37e7cd86bb1928feb1b9579f163c7a3da09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3dfa37e7cd86bb1928feb1b9579f163c7a3da09e"}]},"branch":"refs/heads/master"},"1a6d055bd4ca219ac45ff0c7f554208277d41e52":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-08-16 08:15:52.000000000","uploader":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"ref":"refs/changes/51/696851/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/696851/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/696851/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/696851/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/696851/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/696851/2"}}},"commit":{"parents":[{"commit":"627af67f5e631e8e1b23aa1527517a6cd6abd462","subject":"Merge \"objects: Don\u0027t use generic \u0027Field\u0027 container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/627af67f5e631e8e1b23aa1527517a6cd6abd462"}]}],"author":{"name":"pandatt","email":"guojy8993@163.com","date":"2019-12-02 10:02:26.000000000","tz":480},"committer":{"name":"huanhongda","email":"hongda.xun@easystack.cn","date":"2022-08-16 08:15:52.000000000","tz":0},"subject":"WIP:Drop unreliable host_aggregate_map","message":"WIP:Drop unreliable host_aggregate_map\n\nChange-Id: I1734d4b8e4d7f226f203a3cf81ec5c2d87baa107\nRelated-bug: #1699054\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a6d055bd4ca219ac45ff0c7f554208277d41e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1a6d055bd4ca219ac45ff0c7f554208277d41e52"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
