)]}'
{"id":"openstack%2Fnova~696953","triplet_id":"openstack%2Fnova~stable%2Ftrain~I17e0e45117a3312c11d6c7f2762dd416b6067979","project":"openstack/nova","branch":"stable/train","topic":"bug/1844296","hashtags":[],"change_id":"I17e0e45117a3312c11d6c7f2762dd416b6067979","subject":"compute: Use long_rpc_timeout in reserve_block_device_name","status":"MERGED","created":"2019-12-02 20:13:46.000000000","updated":"2019-12-06 01:23:49.000000000","submitted":"2019-12-06 01:22:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696953-1575595333053-a858fa3e","meta_rev_id":"b027166a1f16d79bc3b626eb1d50df0008ef1663","_number":696953,"virtual_id_number":696953,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-02 22:41:53.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2019-12-03 02:19:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2019-12-02 20:40:07.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 01:22:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2019-12-04 16:02:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-05 16:59:18.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":2,"date":"2019-12-05 17:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":1,"date":"2019-12-05 17:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-02 20:40:07.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-12-02 22:41:53.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-03 02:19:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 16:02:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2019-12-05 16:59:18.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-05 17:03:17.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2019-12-06 01:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31189dc5c7dc2dd16d1ba538f40dc4beea6303e0","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-02 20:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6084ab1cb75be2fc21b12a73cbdb6cdb7e9dc7ed","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-02 20:40:07.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/696953/1/check/pmem-tempest-plugin-filtered/a53ded2/ : SUCCESS in 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9b54d2e0b84ffc7fd7771f89e8f8383756992ca6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-02 22:41:53.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/696953/1 : FAILURE in 2h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1e3769cd07a98c9d7474eefd3ea0cdf605814b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 22:50:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8701aebf6a5943abada0716f3c960e65 : SUCCESS in 1h 14m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4a31f043c85e4f4db98b20fb76499665 : SUCCESS in 1h 26m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e328f5022a1d4442b53ee4fd0c3f45ce : SUCCESS in 25m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cb8dc670ef74fb6a251925ed6a410ec : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89bd023b5c1844b282b9d72f0f0a93d7 : SUCCESS in 11m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e18d641b86204e8a8ab3c38bc500b3fe : SUCCESS in 23m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/172f10faec0e43ed8e21ade9b0377a51 : SUCCESS in 10m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ced22f240d9f4fbf8c0b78a51256b867 : SUCCESS in 22m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a56196e55cde47718c77a0aff7ffc00d : SUCCESS in 18m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3837b8a6995b4de183821d96e55d0213 : SUCCESS in 55m 22s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/0230fc9176754c8eb4967a21a773393a : SUCCESS in 1h 14m 23s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fad8bdc036a94617a6603495684a057e : SUCCESS in 1h 13m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b1e353e862604b389cb7cbf03c620598 : SUCCESS in 1h 18m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fd9f377b652a4d35b259d0d3abd93192 : SUCCESS in 53m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6c89fedb0ccf479ab4eba2b859e6b603 : SUCCESS in 1h 54m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/67c73e8e8de34f8fb9258686ac8cadc0 : SUCCESS in 1h 32m 21s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/16da2ffa32184ebdbbd54bac98f3fcb1 : SUCCESS in 26m 03s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f402f8aacd8541a7b67d94804bc4493c : SUCCESS in 25m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df6f6c6c9c9a460da9e9c48533d47c7d : SUCCESS in 2h 26m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9f4306244c149c7af323eafe4b0ccdb : SUCCESS in 1h 02m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4bfcc4793c935e9742a1f0167b22a3b22dcb0f30","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-03 02:19:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/696953/1/check/tempest-dsvm-full-xenial/3372ae8/ : SUCCESS in 2h 52m 03s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/696953/1/check/tempest-dsvm-full-xenial-py3/58274a9/ : SUCCESS in 2h 58m 09s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/696953/1/check/grenade-dsvm-xenial/b688f55/ : FAILURE in 2h 55m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5143cfaac9e20937a393ba08d092a1d681482be8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2019-12-04 16:02:40.000000000","message":"Patch Set 1: Code-Review+1\n\nThe backport itself looks good, the small difference/conflict is noted correctly. Otherwise I saw some uncertainties regarding the patch on master.","accounts_in_message":[],"_revision_number":1},{"id":"4acd05062aaf2b82f7eeab87bf121d9a4ac4e6e8","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-05 16:55:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b4953ffee2e5315db7e8e969ce5893da6058f1b","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-05 16:59:18.000000000","message":"Patch Set 1: Code-Review+2\n\nI\u0027ll just +2 since the cherry-pick, though not the merged patch set on master, is the same fix and you\u0027ve already resolved conflicts here in train and on stein and I don\u0027t want to make you redo those just for that git hash line.\n\nAs for validity, there was quite a bit of discussion on this related change https://review.opendev.org/#/c/682594/ and the summary was that change didn\u0027t have much chance of fixing the bug if it occurs (the MessagingTimeout) and the proper fix is this one - extending the timeout on the call to the compute service. So yeah I\u0027m OK with this.","accounts_in_message":[],"_revision_number":1},{"id":"e8de104cc2bedb1a5b4fa5764043a49201a3c59e","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-05 17:01:24.000000000","message":"Topic set to bug/1844296","accounts_in_message":[],"_revision_number":1},{"id":"2406c3b7e8cc5cf140f1f429d3588e2e263c0e0d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2019-12-05 17:03:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"aff83158567df22141d03420338dfc93ef57be3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 17:03:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af7c719dc21a033777f49cff579a523984832595","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:22:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/7a53bf868fef48e6a1d4a1039ab8e223 : SUCCESS in 58m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fabb8ed1c68348bda4b971790a6fb6da : SUCCESS in 1h 26m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1acdbca1eb934907b08804fe7bbf89f2 : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/464d2e762e0548bc8e51657af8ef3f43 : SUCCESS in 11m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6393e5dba240496fbaf30b8d79969cab : SUCCESS in 14m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9f38c3421f4b8bbf5efc66a13ef9b9 : SUCCESS in 15m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56eb74153cf94889916abbbb45593c63 : SUCCESS in 18m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/505d6ef6b010453d8aa14309cc15c1c0 : SUCCESS in 10m 37s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7f396a5c92c45ce81ca79fba704ea57 : SUCCESS in 1h 08m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/eeb23a3977634f619994b06bbe110826 : SUCCESS in 46m 47s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/a500327a227a4de6a7be6cef9cdb9411 : SUCCESS in 18m 17s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6ab8eef3bc6410482581e4db74ef704 : SUCCESS in 17m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bf1e7c6e1eed48b99fc973cb9dc79fcb : SUCCESS in 1h 22m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/a6b6c97760de4de49eca3eb1c6d2c895 : SUCCESS in 1h 18m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8c24b5b2ff7d4893a6d27f14bdd813d3 : SUCCESS in 2h 10m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d873ac0495e24130ad386b21d4e42b1b : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d692f07ce409e7cb22a12d803c0c0093dcbde98d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:22:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b027166a1f16d79bc3b626eb1d50df0008ef1663","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 01:23:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae074bfcba18430496628a57e4787f4f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9a1bc71050b41753259b96998a2a4b4dc8ecd79","revisions":{"f9a1bc71050b41753259b96998a2a4b4dc8ecd79":{"kind":"REWORK","_number":1,"created":"2019-12-02 20:13:46.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/53/696953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/696953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/696953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/696953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/696953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/696953/1"}}},"commit":{"parents":[{"commit":"155267516200e8b077f428037cf58a265f8ccdf7","subject":"Merge \"Reset instance to current vm_state if rolling back in resize_instance\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/155267516200e8b077f428037cf58a265f8ccdf7"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-11-08 12:26:01.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-02 20:09:55.000000000","tz":0},"subject":"compute: Use long_rpc_timeout in reserve_block_device_name","message":"compute: Use long_rpc_timeout in reserve_block_device_name\n\nGiven the instance.uuid lock taken on the remote compute, calls to\nreserve_block_device_name can take a large amount of time to complete\nwhen attaching multiple volumes. To help avoid timeouts during such\nattempts this change switches to using the long_rpc_timeout for the\noverall timeout for each call.\n\nConflicts:\n        nova/conf/rpc.py\n\nNOTE(lyarwood): Conflicts due to the following not being present in\nstable/train. I9115ef6df59844cd6e702f19ba38ffbf9f8b35d3,\nI518ae675b7a67da64a5796e57e87860f0c3ef0db and\nIf373fedb8d2e0dfc46b8ac5b018f8216aa5c643c.\n\nRelated-Bug: #1844296\nChange-Id: I17e0e45117a3312c11d6c7f2762dd416b6067979\n(cherry picked from commit 648c05f7bee025087c2f9d8e2f9cda6e2c13e13f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9a1bc71050b41753259b96998a2a4b4dc8ecd79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9a1bc71050b41753259b96998a2a4b4dc8ecd79"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
