)]}'
{"id":"openstack%2Fnova~699115","triplet_id":"openstack%2Fnova~stable%2Ftrain~I0ef4b954b7f4ae65b6c0f96580c5f9472a2b873c","project":"openstack/nova","branch":"stable/train","topic":"bug/1756118","hashtags":[],"change_id":"I0ef4b954b7f4ae65b6c0f96580c5f9472a2b873c","subject":"Skip cpu comparison on AArch64","status":"ABANDONED","created":"2019-12-15 11:12:23.000000000","updated":"2020-03-09 20:21:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d47930e57356d93f893ee5a5d9657e7c7cc3a2e6","_number":699115,"virtual_id_number":699115,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2019-12-15 13:42:08.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2019-12-15 11:26:10.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2019-12-15 14:10:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"date":"2019-12-15 11:14:03.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-15 13:36:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":-1,"date":"2020-01-21 12:15:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-09 16:44:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":1,"date":"2020-01-21 09:10:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-01-21 12:18:40.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-15 11:14:03.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-12-15 11:26:10.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-15 13:36:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-15 13:42:08.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-15 14:10:13.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-09 16:44:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-01-21 09:10:10.000000000","updated_by":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"reviewer":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"state":"REVIEWER"},{"updated":"2020-01-21 12:15:23.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"c7bf8f486a30e2734df3c70c707273c1bed671f1","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-15 11:12:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4033658d88d92a95760650ff67d8a9ce6b8af78e","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-15 11:14:03.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"37d7b029065feb30a8bea2e63de486cbf176fa07","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2019-12-15 11:23:00.000000000","message":"Topic set to bug/1756118","accounts_in_message":[],"_revision_number":1},{"id":"49c54ed165eeaef6094003c8d26cd84154479d8d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2019-12-15 11:26:10.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* Nova-MACVTAP-ML2-Sriov http://13.74.249.42/refs/changes/15/699115/1/Nova-MACVTAP-ML2-Sriov : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b8fa5cfe0d73c49f960de7d623f4292a014567e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 13:36:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ced8277a73aa483bb7eda7a2aa810b24 : SUCCESS in 49m 04s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/28670a188a1646dfa63eeb110a65bf1b : SUCCESS in 1h 15m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d37a08bcd404d8cb8298adb7573b830 : SUCCESS in 21m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7ed07031b8349bf9765571354b10588 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986e7e93e80149b79fe2ad4e328ec668 : SUCCESS in 9m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/57d18aa3c4e34687988862bb2183f5b4 : SUCCESS in 13m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f334bdbf38e5456f903f368e413a4a6c : SUCCESS in 11m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1f17e0f3cbba42048bc2ac519fd65b79 : SUCCESS in 17m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4741cf62b964329a1dda105b7e094a9 : SUCCESS in 13m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/84d9e9847fb04643a3f829ed271c9eb6 : SUCCESS in 53m 04s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3eafd8d395804d5f9823c69a50c5525d : RETRY_LIMIT in 16m 06s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9f16fa7d78134fa295e12bf6fde7b3c6 : SUCCESS in 1h 03m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/71a6e4333fda41c0ae35ce222619b879 : SUCCESS in 1h 06m 14s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/917ec02e2a924983a0a9ce994c8c2405 : SUCCESS in 49m 02s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4f2ccc3f9d614823ab77b506f0622f7d : SUCCESS in 45m 59s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b0d0aa77399b43eb9bd038ab8ee34311 : SUCCESS in 1h 32m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/4eac949fa98e486d883dbb5ebbb13089 : SUCCESS in 1h 12m 18s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/9da1da53b77d421cb8d7066acddbe427 : SUCCESS in 17m 28s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/05f17cd206904c5b888046233da92624 : SUCCESS in 19m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/761062ae4a6241abb61a86b0edf37add : SUCCESS in 2h 09m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b18f862878248068f2c083702714b62 : SUCCESS in 1h 01m 19s","accounts_in_message":[],"_revision_number":1},{"id":"7b9d94587f92d25ce8ce516d032889d132036a48","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-15 13:42:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/699115/1 : SUCCESS in 2h 27m 34s","accounts_in_message":[],"_revision_number":1},{"id":"099c6eebd30a490e788750437496c8f0e6a04783","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-15 14:10:13.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/15/699115/1/check/tempest-dsvm-full-xenial/8b468bd/ : SUCCESS in 2h 53m 17s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/15/699115/1/check/tempest-dsvm-full-xenial-py3/fcf5deb/ : SUCCESS in 2h 54m 10s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/15/699115/1/check/grenade-dsvm-xenial/73f40a5/ : FAILURE in 2h 56m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"127c14b3699c86aff95718a2285e8a7b6631520c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-01-09 16:44:09.000000000","message":"Patch Set 1: Code-Review+2\n\nClean cherry-pick, merged to master. According to bug report this needs to be backported until queens, so looks ok to me.","accounts_in_message":[],"_revision_number":1},{"id":"55ce445c3f184815760dba0c8c0d26cf116d7bb7","author":{"_account_id":22076,"name":"Kevin Zhao","email":"kevin.zhao@linaro.org","username":"KevinZhao"},"date":"2020-01-21 09:10:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b18c9e43e545c1cade0896f08b69404304cefb4","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-21 12:15:23.000000000","message":"Patch Set 1: Code-Review-1\n\ni dont agree with the premiss that if you use host passthough then checking compatiablity does not matter.\n\nwe want the live migration to be prevented before we try to do the migration to prevent doing as much work as possibel if its going to fail. when using host-passtough it is more important to do this check as it is more likely that it will fail. i dont think this should be backported and i would sugges we should revert this on master.","accounts_in_message":[],"_revision_number":1},{"id":"f8ac53168233530ff0d7f70e92b15627f5869759","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-01-21 12:18:40.000000000","message":"Patch Set 1: Workflow-1\n\n\u003e i dont agree with the premiss that if you use host passthough then\n \u003e checking compatiablity does not matter.\n \u003e \n \u003e we want the live migration to be prevented before we try to do the\n \u003e migration to prevent doing as much work as possibel if its going to\n \u003e fail. when using host-passtough it is more important to do this\n \u003e check as it is more likely that it will fail. i dont think this\n \u003e should be backported and i would sugges we should revert this on\n \u003e master.\n\nCan you propose the revert and give that justification there? I\u0027ll -W this until that\u0027s resolved one way or another.","accounts_in_message":[],"_revision_number":1},{"id":"9b10422a4f8fc99ebb510eb6f1add2bef5113c64","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-21 12:55:12.000000000","message":"Patch Set 1:\n\nhttps://review.opendev.org/#/c/703596/1 i was just doing that\n\nat a minium i think we should be adding a workaround config option to opt into this behavior so that when libvirt finally fixes this we dont have to revert the code but i also think we can do at least a partial check unless for compatibility as we can compare the cpu feature flags.\n\nlooking at the output of the domain capablities api we do not get back alot of useful info but we can at least check the gic version compatiblity and likely the machine type.\n\n\u003cdomainCapabilities\u003e\n  \u003cpath\u003e/usr/local/bin/qemu-system-aarch64\u003c/path\u003e\n  \u003cdomain\u003eqemu\u003c/domain\u003e\n  \u003cmachine\u003evirt-4.1\u003c/machine\u003e\n  \u003carch\u003eaarch64\u003c/arch\u003e\n  \u003cvcpu max\u003d\u0027512\u0027/\u003e\n  \u003cos supported\u003d\u0027yes\u0027\u003e\n    \u003cloader supported\u003d\u0027yes\u0027\u003e\n      \u003cenum name\u003d\u0027type\u0027\u003e\n        \u003cvalue\u003erom\u003c/value\u003e\n        \u003cvalue\u003epflash\u003c/value\u003e\n      \u003c/enum\u003e\n      \u003cenum name\u003d\u0027readonly\u0027\u003e\n        \u003cvalue\u003eyes\u003c/value\u003e\n        \u003cvalue\u003eno\u003c/value\u003e\n      \u003c/enum\u003e\n    \u003c/loader\u003e\n  \u003c/os\u003e\n  \u003ccpu\u003e\n    \u003cmode name\u003d\u0027host-passthrough\u0027 supported\u003d\u0027no\u0027/\u003e\n    \u003cmode name\u003d\u0027host-model\u0027 supported\u003d\u0027no\u0027/\u003e\n    \u003cmode name\u003d\u0027custom\u0027 supported\u003d\u0027yes\u0027\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa262\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270-a0\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm1136\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a15\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa260\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm1136-r2\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa261\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa255\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a72\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-m33\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm926\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-r5f\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm11mpcore\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa250\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003eti925t\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a57\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003esa1110\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm1176\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a53\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003esa1100\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270-c5\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a9\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a7\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-a8\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270-c0\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm1026\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270-b1\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-m3\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003emax\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-m4\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270-b0\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003earm946\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-m0\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003ecortex-r5\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270-a1\u003c/model\u003e\n      \u003cmodel usable\u003d\u0027unknown\u0027\u003epxa270\u003c/model\u003e\n    \u003c/mode\u003e\n  \u003c/cpu\u003e\n  \u003cdevices\u003e\n    \u003cdisk supported\u003d\u0027yes\u0027\u003e\n      \u003cenum name\u003d\u0027diskDevice\u0027\u003e\n        \u003cvalue\u003edisk\u003c/value\u003e\n        \u003cvalue\u003ecdrom\u003c/value\u003e\n        \u003cvalue\u003efloppy\u003c/value\u003e\n        \u003cvalue\u003elun\u003c/value\u003e\n      \u003c/enum\u003e\n      \u003cenum name\u003d\u0027bus\u0027\u003e\n        \u003cvalue\u003efdc\u003c/value\u003e\n        \u003cvalue\u003escsi\u003c/value\u003e\n        \u003cvalue\u003evirtio\u003c/value\u003e\n        \u003cvalue\u003eusb\u003c/value\u003e\n        \u003cvalue\u003esata\u003c/value\u003e\n      \u003c/enum\u003e\n    \u003c/disk\u003e\n    \u003cgraphics supported\u003d\u0027yes\u0027\u003e\n      \u003cenum name\u003d\u0027type\u0027\u003e\n        \u003cvalue\u003esdl\u003c/value\u003e\n        \u003cvalue\u003evnc\u003c/value\u003e\n        \u003cvalue\u003espice\u003c/value\u003e\n      \u003c/enum\u003e\n    \u003c/graphics\u003e\n    \u003cvideo supported\u003d\u0027yes\u0027\u003e\n      \u003cenum name\u003d\u0027modelType\u0027\u003e\n        \u003cvalue\u003evga\u003c/value\u003e\n        \u003cvalue\u003ecirrus\u003c/value\u003e\n        \u003cvalue\u003evmvga\u003c/value\u003e\n        \u003cvalue\u003evirtio\u003c/value\u003e\n      \u003c/enum\u003e\n    \u003c/video\u003e\n    \u003chostdev supported\u003d\u0027yes\u0027\u003e\n      \u003cenum name\u003d\u0027mode\u0027\u003e\n        \u003cvalue\u003esubsystem\u003c/value\u003e\n      \u003c/enum\u003e\n      \u003cenum name\u003d\u0027startupPolicy\u0027\u003e\n        \u003cvalue\u003edefault\u003c/value\u003e\n        \u003cvalue\u003emandatory\u003c/value\u003e\n        \u003cvalue\u003erequisite\u003c/value\u003e\n        \u003cvalue\u003eoptional\u003c/value\u003e\n      \u003c/enum\u003e\n      \u003cenum name\u003d\u0027subsysType\u0027\u003e\n        \u003cvalue\u003eusb\u003c/value\u003e\n        \u003cvalue\u003epci\u003c/value\u003e\n        \u003cvalue\u003escsi\u003c/value\u003e\n      \u003c/enum\u003e\n      \u003cenum name\u003d\u0027capsType\u0027/\u003e\n      \u003cenum name\u003d\u0027pciBackend\u0027/\u003e\n    \u003c/hostdev\u003e\n  \u003c/devices\u003e\n  \u003cfeatures\u003e\n    \u003cgic supported\u003d\u0027yes\u0027\u003e\n      \u003cenum name\u003d\u0027version\u0027\u003e\n        \u003cvalue\u003e2\u003c/value\u003e\n        \u003cvalue\u003e3\u003c/value\u003e\n      \u003c/enum\u003e\n    \u003c/gic\u003e\n  \u003c/features\u003e\n\u003c/domainCapabilities\u003e\n\nif we dont think that is worth added custom checking for fine but i still think this shoudl at least have a workaround config option to contol.","accounts_in_message":[],"_revision_number":1},{"id":"e8ba7415f34f920f433784602570ade360292e85","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-01-21 13:07:48.000000000","message":"Patch Set 1:\n\nby the way as noted in the bug we can actully get more detailed info form lscpu or /proc/cpuinfo which include the cpu flags and the cpu model names. although lscpu does need to be compiled with that support as it was added in 2018","accounts_in_message":[],"_revision_number":1},{"id":"d47930e57356d93f893ee5a5d9657e7c7cc3a2e6","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-09 20:21:00.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ef88647974c91d500edf09d1c57bfeac552a1085","revisions":{"ef88647974c91d500edf09d1c57bfeac552a1085":{"kind":"REWORK","_number":1,"created":"2019-12-15 11:12:23.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/15/699115/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/699115/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/699115/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/699115/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/699115/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/699115/1"}}},"commit":{"parents":[{"commit":"3c62efc5f0754dfed79eedb4dce62ae83d9e9aeb","subject":"Merge \"Do not update root_device_name during guest config\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3c62efc5f0754dfed79eedb4dce62ae83d9e9aeb"}]}],"author":{"name":"Kevin Zhao","email":"kevinzs2048@gmail.com","date":"2018-08-08 09:04:52.000000000","tz":480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2019-12-15 11:10:56.000000000","tz":0},"subject":"Skip cpu comparison on AArch64","message":"Skip cpu comparison on AArch64\n\nHost CPU compatibility check does not make sense on AArch64,\nthis patch skips CPU comparison.\n\nCloses-bug: #1756118\n\nChange-Id: I0ef4b954b7f4ae65b6c0f96580c5f9472a2b873c\nSigned-off-by: Kevin Zhao \u003ckevinzs2048@gmail.com\u003e\n(cherry picked from commit 4bb54ae86981d186e41dcc325d171ab951beb7b6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef88647974c91d500edf09d1c57bfeac552a1085"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ef88647974c91d500edf09d1c57bfeac552a1085"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
