)]}'
{"id":"openstack%2Fnova~699924","triplet_id":"openstack%2Fnova~stable%2Fstein~Ifd4387a02b3103e1258e146e63c73be1ad10030c","project":"openstack/nova","branch":"stable/stein","hashtags":[],"change_id":"Ifd4387a02b3103e1258e146e63c73be1ad10030c","subject":"Zuul v3: use devstack-plugin-nfs-tempest-full","status":"MERGED","created":"2019-12-19 07:41:16.000000000","updated":"2019-12-20 01:15:37.000000000","submitted":"2019-12-20 01:12:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699924-1576804346944-b2b5e395","meta_rev_id":"3304708ba47286b393cdedb3776703549df7c631","_number":699924,"virtual_id_number":699924,"owner":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-19 23:37:35.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2019-12-19 23:31:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2019-12-19 21:13:54.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"date":"2019-12-19 20:26:28.000000000","_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 01:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-19 14:26:23.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-19 14:26:23.000000000","_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 14:26:23.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2019-12-19 21:13:54.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2019-12-19 23:31:00.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-19 23:37:35.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2019-12-20 01:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"764abc5bf3407e1799196c1a2724c2be0d5c1d3b","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-19 07:41:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52673dc7801b803cfd47f639ff0d934b248dd6df","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-19 07:43:04.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"b5548044871461393da88c2fbd2b42849705e3e4","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-19 09:40:28.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"67fa1b72d3f47642233232d86b8b5b1d15d134ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:56:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5c64a6b60b174d6da41a28ecd0b544e9 : SUCCESS in 50m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74b38c1b8d474586b84844e7805b4e61 : SUCCESS in 1h 19m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaed846914bd44d9b14979b1f6d7a4b3 : SUCCESS in 19m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/065f49644f3344ac9f5c1a32ef8dbfc5 : SUCCESS in 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f94589d403456db1e68ead2f58f87e : SUCCESS in 9m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6037d3194423478b9e012e1703577d0a : SUCCESS in 14m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8974e02249ff40bc8a9844481364b9aa : SUCCESS in 20m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e1d1966a6cd49a8a9507d76371eb833 : SUCCESS in 12m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f0c05edaa7e54347b35344af3e08f08e : SUCCESS in 16m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b61d40b563744e0db097632b1d6ff4b1 : SUCCESS in 7m 26s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1b6ee489133a4503976487a4d53f7039 : SUCCESS in 48m 07s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/c0033159e67d4de281f05d8fc9b74a7a : SUCCESS in 1h 03m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/885a0f6288204aff82851a2108a1452a : SUCCESS in 1h 06m 20s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/2c15a5ba803e4724b4b62789845c241d : SUCCESS in 59m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5706e5eaf86349da924fbc73c5fcefda : SUCCESS in 45m 48s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/52e1cd490c774e1d9e2a43eb7265e24f : SUCCESS in 52m 52s (non-voting)\n- nova-next https://zuul.opendev.org/t/openstack/build/421ee54cff784fc38022a78627664476 : SUCCESS in 1h 08m 48s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/acbdc57de59b4be3a5d4a3dc426bbe1c : SUCCESS in 22m 31s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/f95ee26c744241af987d68f59d89e81d : SUCCESS in 17m 31s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/9e4dddaf697149aabec3d0f308dca2e3 : FAILURE in 1h 41m 32s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fee66e368e3f4d7492afbe10459e1005 : SUCCESS in 2h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"4f53778109ce23bdde5752698246b43ca2e16575","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-19 10:16:40.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/699924/1 : SUCCESS in 2h 34m 09s","accounts_in_message":[],"_revision_number":1},{"id":"af1925b996a2304b46e546c38c245adf3e61dac0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-19 10:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/1/check/tempest-dsvm-full-xenial/1ef714f/ : SUCCESS in 2h 36m 57s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/1/check/tempest-dsvm-full-xenial-py3/3000219/ : SUCCESS in 2h 49m 53s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/1/check/grenade-dsvm-xenial/1649c50/ : FAILURE in 2h 55m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"839feafb87b98aa0fac531684e3ee793faa84f22","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 12:25:08.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- ironic-tempest-bfv https://zuul.opendev.org/t/openstack/build/bd5123e9e2054a4f824670a174936aa4 : SUCCESS in 36m 01s\n- ironic-tempest-ipa-wholedisk-direct-tinyipa-multinode https://zuul.opendev.org/t/openstack/build/aa309afd2e844e19b499da89b41a0dad : FAILURE in 1h 28m 58s\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/42fb349a75fd44c985416a661b6c39a5 : SUCCESS in 38m 25s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/12b0838218814a15bde29000f107c463 : SUCCESS in 1h 13m 54s\n- legacy-grenade-dsvm-neutron-multinode-zero-downtime https://zuul.opendev.org/t/openstack/build/a742ce25364e489a8a508d09ea8190bf : FAILURE in 57m 11s\n- devstack-plugin-nfs-tempest-full https://zuul.opendev.org/t/openstack/build/df1fe3f237ad4d27981401254a4aa61b : SUCCESS in 1h 41m 36s\n- tempest-full-py3-opensuse15 https://zuul.opendev.org/t/openstack/build/109a0227a57a4d05a4d58787c6df9812 : SUCCESS in 1h 23m 11s (non-voting)\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/65c2ed1ff59c43c5bf608c86f9ad2f54 : SUCCESS in 1h 21m 22s\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/eda14dcab8954d608b58b1976fd57340 : SUCCESS in 52m 24s\n- nova-tempest-full-oslo.versionedobjects https://zuul.opendev.org/t/openstack/build/96c5413ceb0d4dddb4d4d8b9cedaec54 : SUCCESS in 1h 41m 19s\n- legacy-tempest-dsvm-nova-libvirt-kvm-apr https://zuul.opendev.org/t/openstack/build/3947f17b7f39417a980b3370d343f28f : FAILURE in 9m 37s\n- legacy-tempest-dsvm-neutron-dvr-multinode-full https://zuul.opendev.org/t/openstack/build/de2862058e7b4906bf7c3d9ebdc3756a : SUCCESS in 1h 30m 08s\n- neutron-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/6ee9bde816a345499342d9fdf351500a : SUCCESS in 2h 31m 45s (non-voting)\n- os-vif-ovs https://zuul.opendev.org/t/openstack/build/bf675c038dfa4fc29c48568ed2002892 : SUCCESS in 1h 22m 36s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58cce5ce85d64f27a41709aa510e27cd : SUCCESS in 16m 45s\n- nova-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/7922b038f0334208abeaf3a207fae638 : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"82277bd4c7ad355906abb19997f18f1e584e6ef1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-19 14:25:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d93cd05c684a1a8663cad066bc3d1dc01fa45e1","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-19 14:26:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nI re-did the cherry-pick from stable/train so it\u0027s clear these are going in order. I see the nfs job ran in the experimental queue and since this is for a job in the experimental queue it\u0027s low risk so I\u0027m going to fast approve.","accounts_in_message":[],"_revision_number":2},{"id":"ffcdfde2382c93604509189cc7e85dd9527b8942","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-19 14:27:29.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"e32984a3e54afced3087700029c4d92b93f23149","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-19 16:36:32.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/699924/2 : SUCCESS in 2h 09m 31s","accounts_in_message":[],"_revision_number":2},{"id":"887a67958a574d433f83695e25ce08c40fc07599","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:16:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e5c81399fddb4fb5b3ba4877f8f1e58e : SUCCESS in 1h 13m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f737fc1693554bf4b3e890c00aafe406 : SUCCESS in 1h 20m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6a89ffffeae4f73afeb98ccf22f5b5c : SUCCESS in 15m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1010c6a547b447ca95cae232a13aa4f5 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808040cdbde942bab5c5beb52a883569 : SUCCESS in 12m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/163d6a7ece79488f870463f1edb5a9de : SUCCESS in 15m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0231d6d93cf24a809cad37873c66693d : SUCCESS in 15m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dc9febc3efb436891b24be800585c37 : SUCCESS in 13m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/52ddb72af930447a8c736efb161f39ee : SUCCESS in 12m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c83f54b02cc4d72b3afc1d373e5b542 : SUCCESS in 8m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dfaa54ab9f6e4efa92fef1a2076394f8 : SUCCESS in 48m 17s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e36bbefb11064204bae3a19f3ecab1f1 : SUCCESS in 1h 16m 02s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/633d3cc3033348faa4b718a157a83f8e : SUCCESS in 1h 21m 24s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/dd17a29296aa4888a4b97a87e792e264 : SUCCESS in 1h 02m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/de216c708eee40e6b6a78724aae7df78 : SUCCESS in 50m 41s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/56fa77ffa3eb42bcbc8258b711bc1c1e : SUCCESS in 48m 28s (non-voting)\n- nova-next https://zuul.opendev.org/t/openstack/build/faf0364d145c4e18abb222ebb6693843 : SUCCESS in 1h 16m 37s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/ed0d56b6e80f4f588b302cd14b30c79f : SUCCESS in 19m 27s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/fe4e833034794addb5415478b5ada141 : SUCCESS in 27m 30s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/35b1d4ddeaeb47cc9531e10cb0be2df9 : SUCCESS in 1h 40m 11s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/35a82276ce9a4f75b8bfc6b050fd5ade : SUCCESS in 2h 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"b490b4b470d553a968b1b045943ac2c0b5110d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 17:17:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3eb5fbf318e1aac989d387ba4ba3be2ae0005c86","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-19 17:24:10.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/2/check/tempest-dsvm-full-xenial/b8f60bb/ : SUCCESS in 2h 39m 47s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/2/check/tempest-dsvm-full-xenial-py3/f0262ba/ : SUCCESS in 2h 50m 29s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/2/check/grenade-dsvm-xenial/13cd01f/ : FAILURE in 2h 56m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bda9b04d7d31a4789c849edd5e5d0f91de67602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 20:24:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ebde44eceade4ea4a0d79fe225a2f8b2 : SUCCESS in 1h 03m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fb5de8d33564fa39533166e785c5aee : SUCCESS in 1h 31m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/979f07f5623048e0921263c07e2f7591 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c1385d86d44598832da8acdeef1675 : SUCCESS in 10m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84bfd2310a2440dbb2e2704fd130ebf2 : SUCCESS in 12m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/465c9f492ed84ec994c5cae146465611 : SUCCESS in 19m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0031fef7f18046fd874f420a4483bd38 : SUCCESS in 11m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1dd273c87e1d42ecba3e237ca07021f2 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44b0742b08b14051af98804b230a953d : SUCCESS in 8m 12s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/c6826df67c96420b9ad3dabfff533068 : FAILURE in 13m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1d893b4b913845a9b565f86e585a3884 : SUCCESS in 48m 26s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/ca39e73aa8dc462bb447f57123aef224 : SUCCESS in 20m 15s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/aead8e2ea8ec4cac8b26afbf36c25a20 : SUCCESS in 17m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/614852e64db9495a8bf061afc9dcf9f1 : SUCCESS in 1h 13m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b63ea1c4e7a4212a48e4d1ccd98cfb0 : SUCCESS in 2h 12m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c30b9948763d76269da17de16ea95a054a24cb81","author":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"date":"2019-12-19 20:26:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"077e6685c46b33190d60528ae48c7a39cd15d1d8","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2019-12-19 20:26:28.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f45d9c863995e88c0c60c30873315f6e1b178241","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2019-12-19 21:13:54.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/24/699924/2/check/pmem-tempest-plugin-filtered/da0a1cd/ : FAILURE in 40m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7f18fb506ece17def47d4c6fd1aa24e297f48448","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 22:56:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5fd27ab2d5104f52a38a7bfd039cdae9 : SUCCESS in 57m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb46098e63324005849a5f931c52f413 : SUCCESS in 1h 31m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bfaf78f77104878b278a83749fa944f : SUCCESS in 19m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04789e03a9c9415fa3d15c42de54c985 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4a16990a1a4b229d6e27904008562c : SUCCESS in 12m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9ebe5a6aae04f9da938bcafe829ac58 : SUCCESS in 13m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e3aad391e38644d5b5661b0e58feaf50 : SUCCESS in 15m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c2336931904aecb7af48c89cae6e1f : SUCCESS in 14m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2951f97961cc4f3f9ff684b6f6173b9b : SUCCESS in 16m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b78e512a71f94396b2491194916aaedf : SUCCESS in 7m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/939fd66c0b6745aaab7e76e19d05ff57 : SUCCESS in 51m 09s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/825104bcf7f146a1b3a27c2ca7ed3ac1 : SUCCESS in 58m 14s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3c54fc51dc5f44a898f96fb2556fa3a2 : SUCCESS in 1h 12m 47s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/7e58149cbc5147b2bb0c8b0b4a37f366 : SUCCESS in 1h 00m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bec02dadd09a4f538fff7a9080629f1c : SUCCESS in 56m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2ef466264a16453188f955332afa5926 : SUCCESS in 1h 05m 07s (non-voting)\n- nova-next https://zuul.opendev.org/t/openstack/build/affaec212d97462499e237a934dbb4f2 : SUCCESS in 1h 28m 01s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/fae75904d766405f9c1fb3a6239fc614 : SUCCESS in 20m 44s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4e8f0524043f4e299aa37badbbf292cf : SUCCESS in 19m 39s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/fd5e7b4e03194522bcb0dbf7d0882237 : SUCCESS in 1h 33m 43s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c7a92734ce24b3995935f8468d2939a : SUCCESS in 2h 17m 12s","accounts_in_message":[],"_revision_number":2},{"id":"5642d462c1172f92b5dbef34b7e48685c0336877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 22:56:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"753b88b36c91039fd0196286632dabfcadf8dc08","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2019-12-19 23:31:00.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm: recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/2/check/tempest-dsvm-full-xenial/af93a5a/ : SUCCESS in 2h 40m 02s\n- tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/2/check/tempest-dsvm-full-xenial-py3/9952a05/ : FAILURE in 3h 01m 23s (non-voting)\n- grenade-dsvm-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/699924/2/check/grenade-dsvm-xenial/4caccde/ : FAILURE in 2h 56m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"541e0f9f513e6af59866da39320efedf307362d3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2019-12-19 23:37:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/699924/2 : SUCCESS in 2h 07m 47s","accounts_in_message":[],"_revision_number":2},{"id":"a539fdead6e4fe073829a476cef78dbb49f2f41b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:12:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/6517df0b1575439bb761e01761d44ef6 : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68d0e0d472744353b41d0d2987ef9acd : SUCCESS in 1h 12m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9ed2b3104a84762a0c2b27eac33c3ca : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cf0314deecd4dc484fb831cd71983ce : SUCCESS in 9m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5b47e0d2346d40fd88d1b8c8521565d2 : SUCCESS in 15m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/30d33df4b4b8477095f7b06e7dab4ab6 : SUCCESS in 14m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efa28bb5b3be4fd79bd0b3958f73898b : SUCCESS in 13m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/04bf5a5baa704aa5b148ff5133fa36b7 : SUCCESS in 11m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55e45eb35503459aa823dcf3c231aa91 : SUCCESS in 6m 49s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/426f74d9ddcf4380a1998d92f4d1ce2d : SUCCESS in 1h 06m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/50aaae83bef141d4a5b22e1963d7ff6c : SUCCESS in 49m 51s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/1ba28edfac8249af9b1f48a9638ae37d : SUCCESS in 20m 10s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0b88283a3c574637ac127ec1d886ec48 : SUCCESS in 19m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/021ecf9877794b4da9ff35df5a7675a1 : SUCCESS in 1h 13m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15555639e5304701bd635705ed51f333 : SUCCESS in 2h 09m 57s","accounts_in_message":[],"_revision_number":2},{"id":"5b47eff584b1ab2ebee5f352168014e55ba14c06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:12:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3304708ba47286b393cdedb3776703549df7c631","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 01:15:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a682a788b32d472ca7d93152eea42f1f : SUCCESS in 1m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"99aa8ebc12949f9bba76f22e877b07d02791bf5b","revisions":{"a97032ac126b16e427efd2bd39e92fcf9888934e":{"kind":"REWORK","_number":1,"created":"2019-12-19 07:41:16.000000000","uploader":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"ref":"refs/changes/24/699924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/699924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/699924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/699924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/699924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/699924/1"}}},"commit":{"parents":[{"commit":"a3bba2b3202e5957bab387bab4e41568c597aeb0","subject":"Merge \"compute: Use long_rpc_timeout in reserve_block_device_name\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3bba2b3202e5957bab387bab4e41568c597aeb0"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-17 23:28:15.000000000","tz":60},"committer":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-19 07:41:03.000000000","tz":60},"subject":"Zuul v3: use devstack-plugin-nfs-tempest-full","message":"Zuul v3: use devstack-plugin-nfs-tempest-full\n\n... and replace its legacy ancestor.\n\nChange-Id: Ifd4387a02b3103e1258e146e63c73be1ad10030c\n(cherry picked from commit e7e39b8c2e20f5d7b5e70020f0e42541dc772e68)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a97032ac126b16e427efd2bd39e92fcf9888934e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a97032ac126b16e427efd2bd39e92fcf9888934e"}]},"branch":"refs/heads/stable/stein"},"99aa8ebc12949f9bba76f22e877b07d02791bf5b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-12-19 14:25:43.000000000","uploader":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"ref":"refs/changes/24/699924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/699924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/699924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/699924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/699924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/699924/2"}}},"commit":{"parents":[{"commit":"a3bba2b3202e5957bab387bab4e41568c597aeb0","subject":"Merge \"compute: Use long_rpc_timeout in reserve_block_device_name\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3bba2b3202e5957bab387bab4e41568c597aeb0"}]}],"author":{"name":"Luigi Toscano","email":"ltoscano@redhat.com","date":"2019-12-17 23:28:15.000000000","tz":60},"committer":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-12-19 14:25:43.000000000","tz":0},"subject":"Zuul v3: use devstack-plugin-nfs-tempest-full","message":"Zuul v3: use devstack-plugin-nfs-tempest-full\n\n... and replace its legacy ancestor.\n\nChange-Id: Ifd4387a02b3103e1258e146e63c73be1ad10030c\n(cherry picked from commit e7e39b8c2e20f5d7b5e70020f0e42541dc772e68)\n(cherry picked from commit e82e1704caa1c2baea29f05e8d426337e8de7a3c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/99aa8ebc12949f9bba76f22e877b07d02791bf5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/99aa8ebc12949f9bba76f22e877b07d02791bf5b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
