)]}'
{"id":"openstack%2Fnova~707040","triplet_id":"openstack%2Fnova~master~If9d7fafa7d49b49879b6f1b27a62292ab522c4d2","project":"openstack/nova","branch":"master","topic":"bp/policy-defaults-refresh","hashtags":[],"change_id":"If9d7fafa7d49b49879b6f1b27a62292ab522c4d2","subject":"Introduce scope_types in os-console-output","status":"MERGED","created":"2020-02-11 02:36:23.000000000","updated":"2020-03-07 19:57:38.000000000","submitted":"2020-03-06 13:29:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707040-1583501376295-a42b5d9c","meta_rev_id":"c51d940d7aa73c5166fe14c998f630943f840fe9","_number":707040,"virtual_id_number":707040,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-06 03:53:31.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-06 02:21:12.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-03-07 19:57:38.000000000","post_submit":true,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-06 03:32:57.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-03-06 13:04:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-06 03:32:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2020-03-06 01:55:57.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 13:29:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-06 10:52:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2020-03-06 09:54:08.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-06 10:52:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 01:55:57.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-06 02:21:12.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-06 03:32:39.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-03-06 03:32:57.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-06 03:53:31.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-06 09:54:08.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2020-03-06 10:52:40.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-06 13:04:27.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-06 13:29:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-07 19:57:38.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e1f511f7100b37b69ad9f5896e7cc30b738bcf1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-11 02:36:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6e288dc9ec13c22388ee43e09a082283176fa31","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-02-11 02:39:19.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered finger://d0fe4b8af32b/d3963f89bc0248ceb2ea0c4984a25ccb : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":1},{"id":"25454ae9380ec5f187bc420f3e8d5b716314beda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 04:15:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b5271ac19cb349d7926954b4b7681dda : FAILURE in 43m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ba2e646004654abc9ee531fa7429cb44 : SUCCESS in 1h 25m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a137ef84213478e91e000053260c5d3 : SUCCESS in 18m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f9919556e3d049b29fb8bb7647fb0bc5 : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1460bd92b449bc827afda06633e966 : SUCCESS in 9m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e333b6b10584795b97fd6254b481203 : SUCCESS in 12m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b49eff9b0c8840269b4bee536a6f787f : SUCCESS in 15m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d6830fdb0854fca84a49bb880757215 : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9673df82469433c8f0b63d1a0687d10 : SUCCESS in 10m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7ac336ce38024d989feb679991ad0665 : SUCCESS in 1h 02m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/db51b79d2b4246a9977a464ce09464ab : SUCCESS in 1h 00m 03s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/93ff435b96214e46922ad58e7f7149ba : SUCCESS in 1h 21m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a57824ef96b349f399dea60e1a37d8cb : SUCCESS in 47m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ee9b5609b3a7423f95df03c6b1ccd103 : SUCCESS in 1h 24m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/9e422cc0fcc148aa822e5236538136d9 : SUCCESS in 1h 30m 02s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/005dd8b6c3f1488c9634f78580982dcb : SUCCESS in 19m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74c75e130fea4b9b9d98d357113072ad : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c395223bba811e61a1aca9814bc0a96c80163da3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-02-11 05:11:03.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/707040/1 : SUCCESS in 2h 32m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b106c06f94b3c25b399bebeca3c476dfbfee506e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-11 10:22:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/tempest-dsvm-full-bionic/772f9f2/ : SUCCESS in 2h 07m 00s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/tempest-dsvm-full-bionic-py3/5628c48/ : SUCCESS in 2h 07m 14s (non-voting)\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/grenade-dsvm-bionic/5162034/ : SUCCESS in 3h 06m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2ba2dc5647bd30c76b18304534c7e972eb2ec3c9","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-02-11 12:44:42.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40948 : FAILURE in 4h 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"998d66807915178c5e431006320e3f9ad8fcd6dd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-11 22:27:41.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eb5e0a3d8c6e72007532ab8aa512cdeed2ddbd12","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-02-11 22:31:37.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered finger://d0fe4b8af32b/b8691c37ea0e4d50abb3a9e999a33949 : RETRY_LIMIT in 7s","accounts_in_message":[],"_revision_number":1},{"id":"254301b8c5641169ad1ecdcaa96ec8dd598932c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 00:57:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e4f7be11c51446c38614d1be8aead835 : SUCCESS in 1h 08m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ddbed38c52b443f6a34f08e9a886d4ea : SUCCESS in 1h 28m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55024d96d7764e1695dca4343124a2e0 : SUCCESS in 17m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/563bf13625db4faaad3c64c148296563 : SUCCESS in 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42f7b09d38c74624a0ed5b2a30f0d7fc : SUCCESS in 9m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a3bc8563daa40cbbce4a0e8835c3fba : SUCCESS in 14m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a26a55d5fc4a4196a339a8cce5968dc0 : SUCCESS in 16m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cab33212319e4db0a7d06eb83ab942fa : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5441ffaee4284a1a8c7986fb9a11cbcd : SUCCESS in 14m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d679e0b13ae849bb9ad80edebe252116 : SUCCESS in 57m 28s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe713826c4a04e0ba1a4594b58ffabbd : SUCCESS in 1h 11m 39s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c20fd0df3eb74b6da305b435cd80040b : SUCCESS in 1h 26m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/455b8ac0edac4e1e945d3b519757c1cc : SUCCESS in 52m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0f946ce9bf1540b7a86b38ed9b4742dd : SUCCESS in 1h 25m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/7194486139e343b5b07b039914454add : SUCCESS in 1h 32m 52s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a82bc5d929044d7ab7cf8df80a14c20e : SUCCESS in 18m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/437b0650c0764512937c211c388f99f6 : FAILURE in 1h 25m 30s","accounts_in_message":[],"_revision_number":1},{"id":"59c8a6b9138f7fbd5bf9445bcc27f531d7f0f8df","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-02-12 00:57:46.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/707040/1 : FAILURE in 2h 28m 19s","accounts_in_message":[],"_revision_number":1},{"id":"098bed44e1cecebcf0b536528206076e66a84953","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-02-12 01:37:13.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40972 : FAILURE in 3h 06m 30s","accounts_in_message":[],"_revision_number":1},{"id":"bcbbe997136112376149ed695156e5b0f77abd9a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-12 01:42:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/tempest-dsvm-full-bionic/1eabd7e/ : SUCCESS in 2h 08m 41s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/tempest-dsvm-full-bionic-py3/33e3530/ : SUCCESS in 2h 05m 25s (non-voting)\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/grenade-dsvm-bionic/224cb2e/ : SUCCESS in 3h 08m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15f796cf3929ed096aceb66cbd4cbb028dc93a25","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-15 15:16:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f341ef8a6b6385ec3b82447bc5d6ee480245f7c7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-02-15 15:31:29.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41077 : FAILURE in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"58da61f84502970e1cc639ac44eb416c421e3de1","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-02-15 15:45:37.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/40/707040/1/check/pmem-tempest-plugin-filtered/e3016ab/ : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":1},{"id":"25d7da1162931b4da99db85dc8baef06eb8e7316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-15 16:47:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c28c83d0371440a2990816b15e927622 : SUCCESS in 1h 01m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b02e6920ef2146b192eabb4bf359be3f : SUCCESS in 1h 29m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08c7708844af4036b6ff0f12a589cbe4 : SUCCESS in 16m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/482e6fcf62d24205ad4b4cf89eedfcea : SUCCESS in 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44ca2c30cdba4e43adf127e85bb9eab2 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4398493a58d847948f80547cceac572b : SUCCESS in 12m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37fa2a1ea4e64aeb881440fb01e5aad2 : SUCCESS in 13m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c1da3be1ab244d1832f3f55b8889c02 : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2c779238cb424ea2125393175395e3 : SUCCESS in 11m 20s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3ba1d380a6224664a1ea33c1b3e00aba : SUCCESS in 16m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c05c858b6bf78000900daf3da0f8d1833f256346","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-15 18:38:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/tempest-dsvm-full-bionic/3524bb4/ : SUCCESS in 1h 41m 36s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/tempest-dsvm-full-bionic-py3/63a011c/ : SUCCESS in 1h 38m 06s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/1/check/grenade-dsvm-bionic/7ba969c/ : SUCCESS in 3h 18m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"454904ae19a8addbb9d6b19c77695e90491e1af5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-15 19:09:31.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-40-707040-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"6b430652847d1a97d2dfbf4206bf64da16e67d4a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-20 11:26:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d402c817c377d2d4e47513171e518d9388bbf8a4","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-03-02 11:18:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"123aa4333ad05de4aff25481df7e3b7fbeb19e11","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-03-02 11:20:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"905462d32f1aa45283cbfabff7ce01f9d470ab6d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-06 01:27:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"ec1cd6ea98ffe57a9dd20e65f834740bff49233e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-06 01:28:34.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"562ec794eded6f9aa7a331607911fe4559116eee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-06 01:28:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3c139f6db28f5db1e61b56f9504ce3b323fa2a1f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-06 01:29:36.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"7745897deb55f1c522a94bab7ac9bd5f704f8131","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-06 01:55:57.000000000","message":"Patch Set 4:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/40/707040/4/check/pmem-tempest-plugin-filtered/7a3bf00/ : SUCCESS in 24m 41s","accounts_in_message":[],"_revision_number":4},{"id":"a6f7c3544b318a6c27481d82833fb8c4166a8e35","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-06 02:21:12.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/707040/4 : FAILURE in 25m 30s","accounts_in_message":[],"_revision_number":4},{"id":"d749c39c0fdbef0362f2e07139c7943046bf4eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 03:31:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/53e0a5ce5dd34c199a75b8d27252f5e4 : SUCCESS in 56m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d936b4ac468848879ac72a7466ccbf62 : SUCCESS in 1h 34m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/492d1f3040fd462c8b34fd90f02ff80f : SUCCESS in 15m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90bc273fa46c4d30963b8d6d0391352f : SUCCESS in 14m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f5825c95b34dd0819536594c5fd148 : SUCCESS in 7m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1594740b3a404b259b3f2d5ab27e450a : SUCCESS in 12m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3444ae9ac50744da8b4a1021ea3d8946 : SUCCESS in 14m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d5394dc55254898a25aae67f1bf1e7a : SUCCESS in 13m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e98bc790fec430583105e1f550442a1 : SUCCESS in 11m 29s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1d76b026f552498e906b7cdd8a1498cd : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":4},{"id":"2eb6f98b9c0c75e9a889a285a8d5a31197a53915","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-06 03:32:39.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41520 : FAILURE in 1h 17m 12s","accounts_in_message":[],"_revision_number":4},{"id":"70a35196c8da1b06974438381dea93e3282c8043","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-06 03:32:57.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/40/707040/4/check/EMC_VxFlexOS_NOVA/2423f7d/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"20d4b5a79cb1619053ad920a2f799cf19b85bf08","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-06 03:53:31.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-40-707040-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"5d28aa929de71bc253b8803e880d35215a246f1e","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-03-06 09:54:08.000000000","message":"Patch Set 4: Code-Review+2\n\nI would be tempted to do the test in the previous patch in this one, but I am OK with splitting it out too.","accounts_in_message":[],"_revision_number":4},{"id":"4effd07c655dff89e42abae06e9b50f76b4bf577","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-03-06 09:54:27.000000000","message":"Patch Set 4:\n\nignore me, this is perfect.","accounts_in_message":[],"_revision_number":4},{"id":"da688e3b5238ca3d71a9c50ed4b11cfab3c38bf0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-06 10:52:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"146bf0a267a90d419447eb047201dd09683acefb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 10:53:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dd0f6c6a4f155435f81dd077b06d7fd48fc7cafb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-06 13:04:27.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/4/check/tempest-dsvm-full-bionic/19ff6cd/ : SUCCESS in 1h 46m 10s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/4/check/tempest-dsvm-full-bionic-py3/2d9b7df/ : SUCCESS in 1h 42m 15s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/40/707040/4/check/grenade-dsvm-bionic/c2b3ae0/ : SUCCESS in 3h 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e66b746028851410f46cd6bd0d1211bb5f00ae42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 13:29:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3fb2647b5fce4cc09eca7b246a6b7b33 : SUCCESS in 1h 02m 37s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5b1558425121498c88be48463952fbf3 : SUCCESS in 1h 18m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2241983365b41c7ae5771a4a6b7f94a : SUCCESS in 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6401440094d41be81a0cb542aa6432d : SUCCESS in 9m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a9536d1555d43e0abc9a3e857c42736 : SUCCESS in 17m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e1cb5e5efff4da49f5381bd633fe535 : SUCCESS in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b14626369f74430f82c7b8286a6a6c64 : SUCCESS in 10m 28s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7dd5c6eb1d054bd08d4e26cba1110149 : SUCCESS in 17m 08s","accounts_in_message":[],"_revision_number":4},{"id":"887819974f258526e4877aee4912cc9e5adccedb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 13:29:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c1ec58839c32291a8ed7369db7b71fd62180ea4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 13:32:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b367835b7c4c67a2386b99be091ffe : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4},{"id":"9817739955f1edf79c00c886c7678af7dbcfdd2e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-07 19:57:38.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/707040/4/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/b45318f : RETRY_LIMIT","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5b97df72371bfe878e62a7575017c468c4c1f599","revisions":{"285642d05f41967120db4cab01f54a75b4921588":{"kind":"REWORK","_number":1,"created":"2020-02-11 02:36:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/707040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/707040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/707040/1"}}},"commit":{"parents":[{"commit":"f9cb4c5b11d5efdb92543fa9e6bc52b76887299d","subject":"Add test coverage of existing console_output policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f9cb4c5b11d5efdb92543fa9e6bc52b76887299d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-11 02:35:23.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-11 02:35:23.000000000","tz":-360},"subject":"Introduce scope_types in os-console-output","message":"Introduce scope_types in os-console-output\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-console-output API policies\nas \u0027system\u0027 and \u0027project\u0027.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: If9d7fafa7d49b49879b6f1b27a62292ab522c4d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/285642d05f41967120db4cab01f54a75b4921588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/285642d05f41967120db4cab01f54a75b4921588"}]},"branch":"refs/heads/master"},"a8d02ab01d37067d5e3bccffce0a6fe9481abcae":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-06 01:27:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/707040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/707040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/707040/2"}}},"commit":{"parents":[{"commit":"ab8bf49d163bca7c75ee15305f80c436956964ef","subject":"Add test coverage of existing console_output policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab8bf49d163bca7c75ee15305f80c436956964ef"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-11 02:35:23.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-06 01:27:00.000000000","tz":0},"subject":"Introduce scope_types in os-console-output","message":"Introduce scope_types in os-console-output\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-console-output API policies\nas \u0027system\u0027 and \u0027project\u0027.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: If9d7fafa7d49b49879b6f1b27a62292ab522c4d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8d02ab01d37067d5e3bccffce0a6fe9481abcae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a8d02ab01d37067d5e3bccffce0a6fe9481abcae"}]},"branch":"refs/heads/master"},"cb8bea416b4539981f3f3282b9dc76cb063d0ebe":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-06 01:28:34.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/707040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/707040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/707040/3"}}},"commit":{"parents":[{"commit":"b0d5357b29720ee66a7cf619ea31370a99b410df","subject":"Add test coverage of existing console_output policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0d5357b29720ee66a7cf619ea31370a99b410df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-11 02:35:23.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-06 01:28:34.000000000","tz":0},"subject":"Introduce scope_types in os-console-output","message":"Introduce scope_types in os-console-output\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-console-output API policies\nas \u0027system\u0027 and \u0027project\u0027.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: If9d7fafa7d49b49879b6f1b27a62292ab522c4d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb8bea416b4539981f3f3282b9dc76cb063d0ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb8bea416b4539981f3f3282b9dc76cb063d0ebe"}]},"branch":"refs/heads/master"},"5b97df72371bfe878e62a7575017c468c4c1f599":{"kind":"REWORK","_number":4,"created":"2020-03-06 01:29:36.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/40/707040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/40/707040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/40/707040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/40/707040/4"}}},"commit":{"parents":[{"commit":"b0d5357b29720ee66a7cf619ea31370a99b410df","subject":"Add test coverage of existing console_output policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0d5357b29720ee66a7cf619ea31370a99b410df"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-02-11 02:35:23.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-06 01:29:33.000000000","tz":0},"subject":"Introduce scope_types in os-console-output","message":"Introduce scope_types in os-console-output\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-console-output API policies\nas \u0027system\u0027 and \u0027project\u0027.\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: If9d7fafa7d49b49879b6f1b27a62292ab522c4d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b97df72371bfe878e62a7575017c468c4c1f599"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b97df72371bfe878e62a7575017c468c4c1f599"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
