)]}'
{"id":"openstack%2Fnova~707224","triplet_id":"openstack%2Fnova~master~I61c1a47559645c41089747cc81270848b58b68f9","project":"openstack/nova","branch":"master","topic":"bp/flavor-extra-spec-validators","hashtags":[],"change_id":"I61c1a47559645c41089747cc81270848b58b68f9","subject":"trivial: Remove FakeScheduler","status":"MERGED","created":"2020-02-11 18:13:11.000000000","updated":"2020-02-26 12:12:51.000000000","submitted":"2020-02-26 12:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707224-1582718865066-3cc46f25","meta_rev_id":"8b81b1872239b554faf1c5f7054499bfd46961d2","_number":707224,"virtual_id_number":707224,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-18 16:46:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-19 00:47:00.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-18 19:25:57.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-02-18 23:48:38.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-18 20:35:01.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-02-25 01:11:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"date":"2020-02-18 16:42:23.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-26 12:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":2,"date":"2020-02-26 10:07:22.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":2,"date":"2020-02-20 13:43:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":1,"date":"2020-02-26 10:07:22.000000000","_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 16:42:23.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-02-18 16:46:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 19:25:57.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-02-18 20:35:01.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 23:48:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-19 00:47:00.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-20 13:43:54.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-02-25 01:11:09.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-02-26 10:07:22.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-02-26 12:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cc72d606527ce8d9028882977464980cd4e1970a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-11 18:13:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7277597ad57156328344c3c5a2e5fa97490c1924","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-11 18:14:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"319a93da87862cfbee791a0036b1b98ae7cae956","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-02-11 18:21:00.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered finger://d0fe4b8af32b/b9757354bf1d4691a27bea9040cd7a19 : RETRY_LIMIT in 8s","accounts_in_message":[],"_revision_number":1},{"id":"f6a440b0cf79ef752d69a2501d62a8ac8260bc80","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-02-11 20:46:01.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/707224/1 : FAILURE in 2h 25m 57s","accounts_in_message":[],"_revision_number":1},{"id":"a78c85b469fb5e914301cbc3b7d92dc6e9d90ce5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-02-11 21:53:26.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-40969 : FAILURE in 3h 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9d38fc2c7f85e36bd9a898927d646ad1f5dce086","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-11 23:21:51.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/707224/1/check/tempest-dsvm-full-bionic/fd99580/ : SUCCESS in 2h 09m 13s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/707224/1/check/tempest-dsvm-full-bionic-py3/5e96361/ : SUCCESS in 2h 05m 03s (non-voting)\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/707224/1/check/grenade-dsvm-bionic/b0adf2d/ : SUCCESS in 3h 09m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"50855132ef73f76252b990f98920f916f1973d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 23:59:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/531411ea2c594260ab6de57403ccdd89 : SUCCESS in 6m 54s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fcfcf4b97e564d07b533798aa0ec31e0 : SUCCESS in 1h 01m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/18c34d81284b4b6289d96fc6c60b16b2 : SUCCESS in 1h 21m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a77ec0729ba24f10a828b9ff23f7dc90 : SUCCESS in 18m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3095029fb8ea45bf99423aed095f8d00 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d37ef55fb146219c6be8f40520082d : SUCCESS in 12m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16879c0474334758b19b95c3f14673b5 : SUCCESS in 15m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9f68273b0ee43848c11a5dcb608ecd8 : SUCCESS in 16m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3744b34ef3e4b67ac78da5d67af15aa : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2ec5dc345442aa9af180eeced6201d : SUCCESS in 14m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/22e7ea33b01b447f97481f0b0ae882c3 : SUCCESS in 53m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f0c2b24e6484491e92f4876a345e02a2 : SUCCESS in 1h 16m 27s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/61460198cf6449728b421509d42cb83a : SUCCESS in 1h 17m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/70a31eae544e41329075afdbd452a451 : FAILURE in 1h 01m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9da9bf2638e2473f8e46c768fd157222 : SUCCESS in 1h 17m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/7f698f26202c4a70b03c42a664cb59cd : SUCCESS in 1h 29m 27s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35d1c1ed048f4aeea033e363a771a75a : SUCCESS in 18m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1785c512400f479db72cf82ed26e592b : SUCCESS in 1h 03m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1aee2dcde527ca36edb2f90d85f2dc1ded4669ac","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-12 10:05:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"18e6b5a6036213a3bca5a30ae4f92a763bd24577","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-18 16:17:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e125ed2553adcde7108da9044033f53baa650452","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-02-18 16:42:23.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/24/707224/2/check/pmem-tempest-plugin-filtered/f81ea29/ : SUCCESS in 23m 18s","accounts_in_message":[],"_revision_number":2},{"id":"05b3fd996a87cb8b643a384d7c6b30444c3da7ef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-02-18 16:46:56.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/24/707224/2/check-vote/ext-nova-zuul/5e8e7cf : FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f4914aac3542f742f2f57aed24952e711fcd4859","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-02-18 19:25:57.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/707224/2 : SUCCESS in 3h 07m 09s","accounts_in_message":[],"_revision_number":2},{"id":"6049106f11723c6dd7e14b1b65ad8d82a86a319e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-02-18 20:35:01.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/4b2e23d7e66f492f807f4bfacdd51781 : POST_FAILURE in 1h 59m 59s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/17e1f3361ba84371a0cb771c7fc22d89 : POST_FAILURE in 1h 37m 47s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/24/707224/2/check/grenade-dsvm-bionic/0f004b2/ : SUCCESS in 3h 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"600d12e357b620a66d801cf869024130670b9c5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 23:48:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9d2593a8f6294abeb98f7232ba3f0267 : SUCCESS in 4m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/429404c5fe554bf4a6af7b6f0f692e0a : SUCCESS in 1h 20m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0585a4d6587b4ad7b1a277cc61fec7fa : SUCCESS in 1h 33m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a555e3aa102641f9ac9acb83a1c8afcb : SUCCESS in 17m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8cf960f1b154a6aaa611cf6265ced49 : SUCCESS in 20m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f4f9f442604c2c89d36a3e391e262f : SUCCESS in 9m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eec4ce3141374333bc9549f9a1b68480 : SUCCESS in 16m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c755dea83be841c7b9f1e8a7c84469c4 : SUCCESS in 15m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/194382224b034124bc431fac11fc08f7 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bf5562ccfde4357b2e0fcd4254ec538 : SUCCESS in 14m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/03d456a6bfec42d680cd2ede7c4084cc : SUCCESS in 58m 50s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f5538de6e1784846885de8ce62694cd3 : SUCCESS in 1h 10m 59s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6a11a462a0a4e20b50c52768fdb42cf : SUCCESS in 1h 28m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b0a114e1fc2d4b049f3f1aee507898ce : SUCCESS in 56m 57s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0289551f2d824e398b55648e51c7915d : SUCCESS in 1h 39m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/29f7b6719a4447949e513aa4b0c0bf4f : SUCCESS in 1h 37m 47s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/33a6e2e1cd5444998db25e1c460bd7e9 : SUCCESS in 20m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/246c8b85005549e686d8899b806c9471 : SUCCESS in 1h 04m 01s","accounts_in_message":[],"_revision_number":2},{"id":"8e4f94fe0deab0f7450a57826922c97660aeeb08","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-02-18 23:48:38.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/24/707224/2/check/EMC_VxFlexOS_NOVA/25f9832/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"6a2e89b50778e2192d27e38cfb814cacd1135215","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-02-19 00:47:00.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-24-707224-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"fcf4346f57e5cc5f732ef0beb95c5a6526c59ab9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-20 13:43:54.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a770f01419cf8108e79773924f842ad173fc25a7","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-02-25 01:11:09.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41119 : SUCCESS in 2h 38m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9dcb2b73f18f7b88cefffa2cc3a283961f29a9e4","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-02-26 10:07:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5748edd2c3240931420bfb9396d65be977129004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 10:07:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5fc4e2f97c1e277ae37c6d29534aca11253f0467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:07:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7fe81424ca424791a35d1f6d8e6e06fe : SUCCESS in 4m 53s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b8b0b0787d0543228de8861e55cc68ec : SUCCESS in 1h 10m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/621f9f3302864721811b8d4c1fe1ea9b : SUCCESS in 1h 13m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b42cc78535ea4195be6e897d46d8c787 : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0046be82613b4b63b3b908e26b46abf3 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17c52841d0e446e4adaa9ea8434e51f3 : SUCCESS in 14m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4baf53cedf62429693c0e673a5b2e643 : SUCCESS in 15m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19a88a05f8d844ddb21cc4d67679f48a : SUCCESS in 11m 27s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fac1061e4a404cd7b8d95c1a0f15224f : SUCCESS in 1h 16m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e5ad580504174518a5564ca9ef95dbc3 : SUCCESS in 51m 17s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b61b3c89326c40d5b77f5b77d90c8669 : SUCCESS in 17m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/acec27587eff4b29b413964b1fe39db2 : SUCCESS in 1h 37m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/7d1631c77ba64e39921d297463577bde : SUCCESS in 1h 42m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52b4c71b65714743abb618a70ccebf1c : SUCCESS in 1h 12m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c196a0fb395621ca6624e39ec6320002dad7034b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:07:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d0f8b6226e5baee03e07410bcdd656fec3583173","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 12:12:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e2a72d4dd65433fa49a6d11ba54c83b : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a0a88d7c0c4d7470ef86f40b153b3ec606b80537","revisions":{"5ab47ae34ee5a0e9cc293ca1cafda3b13db9af7d":{"kind":"REWORK","_number":1,"created":"2020-02-11 18:13:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/707224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/707224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/707224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/707224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/707224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/707224/1"}}},"commit":{"parents":[{"commit":"1fcd74730d343b7cee12a0a50ea537dc4ff87f65","subject":"Merge \"Fix instance.hidden migration and querying\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1fcd74730d343b7cee12a0a50ea537dc4ff87f65"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-11 17:16:14.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-11 17:43:49.000000000","tz":0},"subject":"trivial: Remove FakeScheduler","message":"trivial: Remove FakeScheduler\n\nWe don\u0027t need this and it\u0027s blocking us from removing pluggable\nscheduler drivers. Remove it.\n\nChange-Id: I61c1a47559645c41089747cc81270848b58b68f9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ab47ae34ee5a0e9cc293ca1cafda3b13db9af7d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5ab47ae34ee5a0e9cc293ca1cafda3b13db9af7d"}]},"branch":"refs/heads/master"},"a0a88d7c0c4d7470ef86f40b153b3ec606b80537":{"kind":"REWORK","_number":2,"created":"2020-02-18 16:17:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/24/707224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/24/707224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/707224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/707224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/24/707224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/24/707224/2"}}},"commit":{"parents":[{"commit":"e69dbfa0d34d6b3f51282ac0ab51cdab3c2115e4","subject":"Merge \"Recalculate \u0027RequestSpec.numa_topology\u0027 on resize\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e69dbfa0d34d6b3f51282ac0ab51cdab3c2115e4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-11 17:16:14.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-18 15:13:45.000000000","tz":0},"subject":"trivial: Remove FakeScheduler","message":"trivial: Remove FakeScheduler\n\nWe don\u0027t need this and it\u0027s blocking us from removing pluggable\nscheduler drivers. Remove it.\n\nChange-Id: I61c1a47559645c41089747cc81270848b58b68f9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0a88d7c0c4d7470ef86f40b153b3ec606b80537"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a0a88d7c0c4d7470ef86f40b153b3ec606b80537"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
