)]}'
{"id":"openstack%2Fnova~709166","triplet_id":"openstack%2Fnova~stable%2Ftrain~I7be32e4fc2e69f805535e0a437931516f491e5cb","project":"openstack/nova","branch":"stable/train","topic":"bug/1862633","hashtags":[],"change_id":"I7be32e4fc2e69f805535e0a437931516f491e5cb","subject":"Reproduce bug 1862633","status":"MERGED","created":"2020-02-21 16:02:53.000000000","updated":"2020-03-06 14:13:11.000000000","submitted":"2020-03-06 09:33:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709166-1583487180594-1d1e633c","meta_rev_id":"f0c49411397128b684464535f23e1de175c952e5","_number":709166,"virtual_id_number":709166,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-06 09:29:39.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-03-06 09:02:28.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-06 09:33:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2020-03-06 08:54:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2020-03-06 08:54:57.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-03-06 08:55:04.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-06 08:54:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-03-06 08:55:04.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-03-06 09:02:28.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-06 09:29:39.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-06 09:33:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5b982a06e2fe16b05ebe0553d9643fe1cd72afc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-02-21 16:02:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bf6d8d47b0863d3aa6815be69a95f9bbc8364da","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-02-21 16:28:17.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/66/709166/1/check/pmem-tempest-plugin-filtered/e643b57/ : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9f9c32cf415dd8c181b6f3ca9eea2d17f210b304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:47:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abab46f6b364492894ab26e5fdc7e3cb : SUCCESS in 19m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/190bebb93701476191f8cbc5693f9105 : SUCCESS in 17m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5399dc19e94002a63f43e53e9e53b3 : SUCCESS in 13m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d87e40cfc4d43b4b1ede55bf92081b5 : SUCCESS in 16m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f82cac6cec7a4767b86055ac30ebfb36 : SUCCESS in 15m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/062b21fd4bf64157844431e80225b292 : SUCCESS in 18m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa225cd6c02a4884b60b38974a365983 : SUCCESS in 14m 24s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/768f123a8d6d4842b4950e79c0569f82 : SUCCESS in 23m 04s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c8359968ce444ec89e1f45cdb5ca74f2 : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":1},{"id":"643ee824ad2f824deac9850d86bcc8ab4badc494","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-02-25 20:39:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08e2c0761231afda3c7d5b425b6fece07afbdf9b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-05 12:12:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bf76e7f76360f9b7235cf57918a2d818e25c4f9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-05 14:00:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c36e1c8b35c78d8c4875cf46169a11700cae4e61","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-03-06 08:36:57.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"591c35bb3b929dac917a3513c630926fc61c36fd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-06 08:54:39.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks for the update, I think this is good like this.","accounts_in_message":[],"_revision_number":2},{"id":"2c05557d303ed9f22873732238aba92899fe5ae0","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-06 08:54:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nLGTM now, thanks!","accounts_in_message":[],"_revision_number":2},{"id":"be605f63c80d1282f56c5cd419051ced52cdbf64","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-06 08:55:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"30b2917f14b8b739038bb5ffbad619930fc986ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:01:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f766b8918e04a66b8837ab950493079 : SUCCESS in 20m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5b7483fcb1944edb38644f3fbcfc1be : SUCCESS in 18m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2598b00c60254bd9b78c3543005fd15a : SUCCESS in 11m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b973dfb617f4873b4eb5f13ddc29370 : SUCCESS in 16m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5b0af9886641a7992481ca00ac752d : SUCCESS in 15m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bade8c52d1e4a58baa2553f8b0bf12c : SUCCESS in 14m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdc2580e27054a97a4aa4cf25d641b2d : SUCCESS in 9m 32s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/d5500c08fa834f8f90a8fff4185e9c71 : SUCCESS in 22m 27s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25e5d444e1ef454da83e0c4cd974072f : SUCCESS in 19m 05s","accounts_in_message":[],"_revision_number":2},{"id":"b48ca0bec531a3670a4f3dfd173731b41b3aee29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:01:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a46a295a1b13d6605fae2d53872378a174d1936","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-06 09:02:28.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/66/709166/2/check/pmem-tempest-plugin-filtered/fb9578a/ : SUCCESS in 24m 09s","accounts_in_message":[],"_revision_number":2},{"id":"b120ef97bdbb80f41fb0601a65478e121bb1d045","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-06 09:29:39.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/709166/2 : FAILURE in 51m 40s","accounts_in_message":[],"_revision_number":2},{"id":"77b439b7ab84a475b14fd0ad994ee455fb325b4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:33:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3422b7608cf46e1ada90464c164a2a5 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f895dfcef0804629b4d447ce76427925 : SUCCESS in 11m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59ec39b5d1184ca284f7c4be62ef05de : SUCCESS in 13m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99677aa6c48c4e2b974a674930563d60 : SUCCESS in 13m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c42969f7f7d4a99ad5fe8dc7487f2d0 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8aaa02d5aca4997a934d16b3f010297 : SUCCESS in 9m 21s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/335c644e8c4f43f29383ec485687149e : SUCCESS in 19m 25s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4a3d2b25a57743eb83eef88013af59fd : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6a807a30411d33737e3debff25d635fca41df8cd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:33:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"82772be2c4f94848b8bd6052911f4a51f341d77f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 09:35:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44587825ab142b399dd712621b194c4 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":2},{"id":"d20fc7c2414683e785c718256d4d9177b9b6dbb2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-03-06 14:13:11.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit a85359cfca4671acc0f45d0f23e30100e9d85866","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd1bfc13d7e2c418afc409871ab56da454a1334d","revisions":{"31df2b6b921f5121741b8aa9b695b42860c77cb3":{"kind":"REWORK","_number":1,"created":"2020-02-21 16:02:53.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/66/709166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/709166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/709166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/709166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/709166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/709166/1"}}},"commit":{"parents":[{"commit":"8363905a6a6d5c8b2488619bdf807c5dc17b2842","subject":"Fix instance.hidden migration and querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8363905a6a6d5c8b2488619bdf807c5dc17b2842"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-02-10 14:23:02.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-02-21 15:51:22.000000000","tz":60},"subject":"Reproduce bug 1862633","message":"Reproduce bug 1862633\n\nIf port update fails during unshelve of an offloaded server then\nplacement allocation on the target host is leaked.\n\nChanges in test_bug_1862633.py is due to\nI84c58de90dad6d86271767363aef90ddac0f1730 is missing from train.\n\nChange-Id: I7be32e4fc2e69f805535e0a437931516f491e5cb\nRelated-Bug: #1862633\n(cherry picked from commit c33ebdafbd633578a0a4b6f1b118c756510acea6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/31df2b6b921f5121741b8aa9b695b42860c77cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/31df2b6b921f5121741b8aa9b695b42860c77cb3"}]},"branch":"refs/heads/stable/train"},"bd1bfc13d7e2c418afc409871ab56da454a1334d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-03-06 08:36:57.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/66/709166/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/66/709166/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/709166/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/709166/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/66/709166/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/66/709166/2"}}},"commit":{"parents":[{"commit":"8363905a6a6d5c8b2488619bdf807c5dc17b2842","subject":"Fix instance.hidden migration and querying","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8363905a6a6d5c8b2488619bdf807c5dc17b2842"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-02-10 14:23:02.000000000","tz":60},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-03-06 08:36:54.000000000","tz":0},"subject":"Reproduce bug 1862633","message":"Reproduce bug 1862633\n\nIf port update fails during unshelve of an offloaded server then\nplacement allocation on the target host is leaked.\n\nChanges in test_bug_1862633.py is due to\nI84c58de90dad6d86271767363aef90ddac0f1730 and \nI8c96b337f32148f8f5899c9b87af331b1fa41424 are missing from train.\n\nChange-Id: I7be32e4fc2e69f805535e0a437931516f491e5cb\nRelated-Bug: #1862633\n(cherry picked from commit c33ebdafbd633578a0a4b6f1b118c756510acea6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd1bfc13d7e2c418afc409871ab56da454a1334d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd1bfc13d7e2c418afc409871ab56da454a1334d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
