)]}'
{"id":"openstack%2Fnova~713353","triplet_id":"openstack%2Fnova~master~Ide92f4b8e646585c562a7808fe261ec2761c8dd7","project":"openstack/nova","branch":"master","topic":"bp/use-pcpu-and-vcpu-in-one-instance","hashtags":[],"change_id":"Ide92f4b8e646585c562a7808fe261ec2761c8dd7","subject":"\u0027cpu_pinning_requested\u0027 property could be derived directly from cpu_policy","status":"ABANDONED","created":"2020-03-17 00:55:27.000000000","updated":"2020-03-31 12:23:11.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c739c2dcbcd179810d35724253852e1524b5bc5a","_number":713353,"virtual_id_number":713353,"owner":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-03-27 15:45:04.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-03-26 19:53:32.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"date":"2020-03-26 22:56:21.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"date":"2020-03-26 17:25:43.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-26 20:05:50.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"date":"2020-03-26 16:53:42.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-03-26 17:24:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-26 16:25:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 01:01:35.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-03-24 14:31:54.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 16:58:45.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-26 16:53:42.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-26 17:24:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 17:25:43.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 19:53:32.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 20:05:50.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-03-26 22:56:21.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-27 15:45:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5ee804087821260cb626800b6bc98e6a18e51e56","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-17 00:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a4381d5193e1720dfc2c5d35b4c4ba4c03ed2b2","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-17 00:57:37.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":1},{"id":"958c3b87e76b09d1c5aede61139410c9013cdf1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 01:24:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e21550eca344ceb67bec4c8f8d4f36 : SUCCESS in 16m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/925941916b824334aa95e9d881ac9f44 : SUCCESS in 19m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b60e302730a84531b0a000635358fdee : SUCCESS in 10m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88c25dc9c1924965b1c81cc2bb6206ba : SUCCESS in 17m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be7d6f6b035e40c89b72206b9ff9ce64 : SUCCESS in 18m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4777bfb331b34d8f8a604e8fe8b59b74 : SUCCESS in 17m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ec028da7a14a87898e5797ba03b180 : SUCCESS in 13m 46s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d4526e166114e18a88b4be9ddbfb25f : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":1},{"id":"232c1e392ac5889f88b0509283562a4ff48d65cf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-17 01:26:08.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/713353/1/check/EMC_VxFlexOS_NOVA/8ab4ef7/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"cb778661cb4e1b0a05b0e193b60b5cde972c6115","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-17 02:43:26.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-713353-1 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":1},{"id":"11ef8488a4f7e0fe0b2b2f27f574b6e35465ab31","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-17 02:44:01.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41828 : SUCCESS in 1h 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"03bf093498f6e32ebf353c8e9f221586612f7769","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-17 03:36:39.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/713353/1 : SUCCESS in 2h 35m 27s","accounts_in_message":[],"_revision_number":1},{"id":"27923ae3807af5d4e13b4ea9e4891e8e6ae09577","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-18 01:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/713353/1/check-vote/ext-nova-zuul/ebf3f05 : SUCCESS in 1h 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8297fcac70ea8bd691c6ecf8c32acef3f2ff7cd9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-18 17:27:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nGood idea. One style nit inline, if you\u0027ve a chance to respin","accounts_in_message":[],"_revision_number":1},{"id":"b61a9910f8be68fc4e58f312156bbbc5875366ab","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-19 07:59:47.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n@Stephen, Thanks for review.","accounts_in_message":[],"_revision_number":1},{"id":"21ddedf7d977d2621a949196ccb274b512e06fe8","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-24 14:30:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4c5b520b6c6079b0a03a730f690b0edb007f6335","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-24 14:31:54.000000000","message":"Patch Set 2:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":2},{"id":"d97ebb9351be001de8d9d08323ab414b658216bb","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-24 14:56:47.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/713353/2/check/pmem-tempest-plugin-filtered/90246c9/ : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":2},{"id":"0c034628a3f764a53aa722468065f05b428d21f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 15:11:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/327b2988ba7c4efab99656cd9ac656f7 : SUCCESS in 16m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e8ab45d7fd447ba81a00e9358a7b474 : SUCCESS in 21m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b00650e23a8a48fd8ce36d2ea338ad8e : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bd2ba2d969443d3bcb1814773537447 : SUCCESS in 17m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8725437e91a14ad2bdf3e1f79315fe20 : SUCCESS in 15m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eef386161bd94e239d5e43cc00154f2a : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5968b5edc18d400ea0a3124ab429514f : SUCCESS in 10m 47s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/26fcc37453634c57b76ab5f2ded1d78e : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":2},{"id":"93363ff08f22f5226a50e1aeb4137cb7cde5c95c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 15:12:54.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/713353/2/check/EMC_VxFlexOS_NOVA/8e387ab/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"0139e08c45f59c4bcc3dd35bb3016b9954f3438d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-24 16:58:45.000000000","message":"Patch Set 2: Code-Review+2\n\nThis could go at the front of the series since it\u0027s mostly independent","accounts_in_message":[],"_revision_number":2},{"id":"87253f5301523cbbb8b6a48ea919197f6b91bd48","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-24 17:20:39.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-713353-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"f185e5816202863e6893455128eebef1b35be372","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-24 17:29:29.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42117 : SUCCESS in 1h 50m 28s","accounts_in_message":[],"_revision_number":2},{"id":"e6793157aedd7c22b04606966c374384f36879db","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-24 19:15:34.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/713353/2 : SUCCESS in 2h 25m 48s","accounts_in_message":[],"_revision_number":2},{"id":"2d76f664819db2ed1f20d8768b6ffad599bb2ccd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-24 23:58:51.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/713353/2/check-vote/ext-nova-zuul/df8831d : FAILURE in 37m 31s","accounts_in_message":[],"_revision_number":2},{"id":"bee977e021bccf3934c2874bd7e58d6531491beb","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-26 16:25:31.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"335c4ea4ded8a85e2408a4cde1bd75bee774d462","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-26 16:53:42.000000000","message":"Patch Set 3:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/53/713353/3/check/pmem-tempest-plugin-filtered/dcd70d0/ : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":3},{"id":"46963ec24de8fe3233bbd95a883373334e9124fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 17:24:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11a2c6926a8c403c88be1e870801a273 : SUCCESS in 24m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/500cd29c4fa14c2eb25eff457d5b80b5 : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93c07b834e6c4d44a3d1a4f9b49a5235 : SUCCESS in 16m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ebb0551afa24a399acbe5ccc3b58ac0 : SUCCESS in 16m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b632dc8a48af4dfdbdc2c303dd52a892 : SUCCESS in 17m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66d88b3e1a244ac0ae03636d0379aaaf : SUCCESS in 15m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffed7fa1f654224a105ff1520a77a8e : SUCCESS in 12m 06s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/38e731de52f946df9e1b4958b02479c1 : SUCCESS in 21m 23s","accounts_in_message":[],"_revision_number":3},{"id":"a7cfdbad8b4c9bbfffed0dc1b7edb8d723cf6e7a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-26 17:25:43.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/53/713353/3/check/EMC_VxFlexOS_NOVA/da7a271/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"c09b9e3cd17169be1b7dd890b3752dc738d79601","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-26 19:53:32.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-713353-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"13e1ac029773433f287eb7fb1cbe74fa258a1b6c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-26 20:05:50.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42297 : SUCCESS in 1h 33m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9fb06e2a6e24b49c9c3d6825d8533f63965ad895","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-26 22:56:21.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/713353/3 : FAILURE in 2h 58m 17s","accounts_in_message":[],"_revision_number":3},{"id":"1fa5ea36e730c3f6a1f759c48ad4ae9e10f73cfb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-27 15:45:04.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/713353/3/check-vote/ext-nova-zuul/d561885 : SUCCESS in 1h 14m 34s","accounts_in_message":[],"_revision_number":3},{"id":"3ad1d61f749abbef0818787fa2df1049b9613095","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-31 12:23:11.000000000","message":"Abandoned\n\nThere is another patch for the same purpose with more consideration.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dce31cd9e7e0a965c76360e3309bcffee3224bcc","revisions":{"5fcaaff2d08be274000bd7e9148dacd7dbff09d2":{"kind":"REWORK","_number":1,"created":"2020-03-17 00:55:27.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/53/713353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/713353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/713353/1"}}},"commit":{"parents":[{"commit":"3d5c26c101dd1dacdd486217614b80af534ff454","subject":"scheduler: Introduce \u0027dedicated\u0027 and \u0027sharing\u0027 CPUs for InstanceNUMACell","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3d5c26c101dd1dacdd486217614b80af534ff454"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-16 16:29:16.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-17 00:25:35.000000000","tz":480},"subject":"Derive \u0027cpu_pinning_requested\u0027 property from cpu_policy","message":"Derive \u0027cpu_pinning_requested\u0027 property from cpu_policy\n\nDerive \u0027cpu_pinning_requested\u0027 property from cpu_policy assigned instead\nof being assigned by a mock.\n\nChange-Id: Ide92f4b8e646585c562a7808fe261ec2761c8dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5fcaaff2d08be274000bd7e9148dacd7dbff09d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5fcaaff2d08be274000bd7e9148dacd7dbff09d2"}]},"branch":"refs/heads/master"},"e002f4c374f56fb30e42310d2b4fc57c53696890":{"kind":"REWORK","_number":2,"created":"2020-03-24 14:30:00.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/53/713353/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/713353/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/713353/2"}}},"commit":{"parents":[{"commit":"5b3a6bdc76c9e4c9777e7a4bc22f41fe7fd7f897","subject":"libvirt: support to create instance with dedicated and shared CPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b3a6bdc76c9e4c9777e7a4bc22f41fe7fd7f897"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-16 16:29:16.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-24 13:52:41.000000000","tz":480},"subject":"\u0027cpu_pinning_requested\u0027 property could be derived directly from cpu_policy","message":"\u0027cpu_pinning_requested\u0027 property could be derived directly from cpu_policy\n\n\u0027cpu_pinning_requested\u0027 property could be derived from the\n\u0027dedicated\u0027 cpu_policy. No need to mock the \u0027cpu_pininng_requested\u0027\nand return with \u0027True\u0027 intentionally.\n\nChange-Id: Ide92f4b8e646585c562a7808fe261ec2761c8dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e002f4c374f56fb30e42310d2b4fc57c53696890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e002f4c374f56fb30e42310d2b4fc57c53696890"}]},"branch":"refs/heads/master"},"dce31cd9e7e0a965c76360e3309bcffee3224bcc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-26 16:25:31.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/53/713353/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/713353/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/713353/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/713353/3"}}},"commit":{"parents":[{"commit":"a3d472f502bfa7e2ac53173ce9f0af3e4cf9a95b","subject":"libvirt: support to create instance with dedicated and shared CPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a3d472f502bfa7e2ac53173ce9f0af3e4cf9a95b"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-16 16:29:16.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-26 16:22:17.000000000","tz":480},"subject":"\u0027cpu_pinning_requested\u0027 property could be derived directly from cpu_policy","message":"\u0027cpu_pinning_requested\u0027 property could be derived directly from cpu_policy\n\n\u0027cpu_pinning_requested\u0027 property could be derived from the\n\u0027dedicated\u0027 cpu_policy. No need to mock the \u0027cpu_pininng_requested\u0027\nand return with \u0027True\u0027 intentionally.\n\nChange-Id: Ide92f4b8e646585c562a7808fe261ec2761c8dd7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dce31cd9e7e0a965c76360e3309bcffee3224bcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dce31cd9e7e0a965c76360e3309bcffee3224bcc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
