)]}'
{"id":"openstack%2Fnova~713354","triplet_id":"openstack%2Fnova~master~Ide2eab6dfb0d2100d0f5501ed2a4070b9201320f","project":"openstack/nova","branch":"master","topic":"bp/use-pcpu-and-vcpu-in-one-instance","hashtags":[],"change_id":"Ide2eab6dfb0d2100d0f5501ed2a4070b9201320f","subject":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance","status":"ABANDONED","created":"2020-03-17 00:55:27.000000000","updated":"2020-03-31 12:27:06.000000000","total_comment_count":46,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f70bbd91b388cf27b182c5d8d5c033a02113aa48","_number":713354,"virtual_id_number":713354,"owner":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"date":"2020-03-27 08:04:57.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2020-03-26 19:35:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"date":"2020-03-26 16:47:13.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-26 18:12:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-03-27 16:21:10.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"date":"2020-03-26 21:10:31.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-17 01:01:46.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-03-19 01:51:41.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-03-24 14:07:27.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-03-24 16:27:01.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-24 17:07:00.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-24 18:48:58.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-25 21:18:27.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 16:47:13.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-26 18:12:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-26 19:35:23.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-03-26 21:10:31.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-27 08:04:57.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-27 16:21:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c43e47cf719791948fc53ea6afb474cb349148c","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-17 00:55:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3848b27cda2d5dfc10d12297f32abc29cf4bf8e4","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-17 00:57:50.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":1},{"id":"12759ed0cedc7feb4be11df99d1be0ed25c05645","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-17 01:31:32.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/713354/1 : FAILURE in 28m 39s","accounts_in_message":[],"_revision_number":1},{"id":"8fe9a9778a68d2e3d127f6d7ae940a1c9f2977a5","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-17 01:36:59.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41829 : FAILURE in 33m 04s","accounts_in_message":[],"_revision_number":1},{"id":"57ad22a582ee44d367ccc84c783267729ef65ff8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 02:07:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ac3f84520c1147bdaca3a4aa557c8924 : FAILURE in 1h 00m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/40b3db372ad94899b2c04fcc100bd802 : FAILURE in 26m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7bb3eec90a1a487c81138617f8f515fe : FAILURE in 4m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9581a842deb459481a4a34b60429e4f : FAILURE in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e4007c4894c4db8a22773e9cbc84d5f : FAILURE in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eed7f9b407d14bebb68d2fe62bcc6755 : FAILURE in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b869ecc649de456196aeefeb2a6d80ce : FAILURE in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7bf8e89f68b4d20bc9251e47010b503 : FAILURE in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e174a52f0f6345c18951f860df1e2b52 : FAILURE in 6m 27s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2d2f5cd814184cb99a8f4fe34b683422 : FAILURE in 24m 48s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8ebc17678ad44bcd9f0b4d78d1f46df0 : FAILURE in 27m 01s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/493c82697e9f47d7aedad1fa7d13d832 : FAILURE in 1h 04m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b2f3a9fefb7145de8829846074f95526 : FAILURE in 28m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/23369d27d72a477d9a2e7270c12a34e6 : FAILURE in 27m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/562073dcd3a44c2a869e1aa8986d814d : POST_FAILURE in 28m 39s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5316c21f24d54de38fed8c012fba91ad : FAILURE in 5m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6eafca5e0e54ebc9408c20163fee3f1 : FAILURE in 24m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70d5964f020b40e7a550ce37b5151239 : FAILURE in 28m 10s","accounts_in_message":[],"_revision_number":1},{"id":"325dfb441e613a5322de58e057891f6073a2bcd9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-17 04:25:08.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/1/check/tempest-dsvm-full-bionic/3c1625a/ : FAILURE in 1h 02m 58s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/1/check/tempest-dsvm-full-bionic-py3/c3be90e/ : FAILURE in 54m 08s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/1/check/grenade-dsvm-bionic/f59c094/ : SUCCESS in 3h 05m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0be00f6402f1a3a91db0c3beff40512fec6d17c8","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-17 06:17:00.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713354/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/074eb5c : FAILURE in 49m 26s","accounts_in_message":[],"_revision_number":1},{"id":"9e6df241aef4f474a7c967dfa49d629caa392352","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-18 01:37:48.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/54/713354/1/check-vote/ext-nova-zuul/fc28e69 : FAILURE in 24m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ede39c8d7c53c5a190aec6b3c3f79685a35ad6ea","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-18 17:27:45.000000000","message":"Patch Set 1:\n\nWow, I don\u0027t think I\u0027ve ever seen a CI so unhappy :)","accounts_in_message":[],"_revision_number":1},{"id":"8f92c28798cdbf69d3e173caf97196d8caf16502","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-18 18:01:38.000000000","message":"Patch Set 1: Code-Review-1\n\n(16 comments)\n\nGood start. Lots of comments here. Let me know if any of them aren\u0027t clear","accounts_in_message":[],"_revision_number":1},{"id":"aef8439e86e2c314a3069ea670fe23484efe9ed8","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-03-19 01:51:41.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d16ba804de9c5bdc5e551f5e92fa0fb8883d0937","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-03-19 02:45:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef293c730f5a081bb0a8e2fcc2b0d30870db02d1","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-19 11:30:49.000000000","message":"Patch Set 1:\n\n(15 comments)\n\n@Stephen, Thanks for the review, and addressed your comments in lines. Please check.","accounts_in_message":[],"_revision_number":1},{"id":"6a0cd91440273c6677e6bd40d2a06455539da01c","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-19 11:31:52.000000000","message":"Patch Set 1:\n\n@Alex, Thanks for your review. Will follow your comment.","accounts_in_message":[],"_revision_number":1},{"id":"1be8676b8a3e778b8c415e9916e9a6b8ffba276d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-19 14:18:47.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5607ecba0ecdfff4552f0d33f9371add6e7908bc","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-20 11:33:10.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHi Stephen, thanks for review, please see my reply inline.","accounts_in_message":[],"_revision_number":1},{"id":"a6f84863853b1e37eeb395901726122d54887bea","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-23 13:12:42.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n@stephen, about the adding of InstanceNUMATopology.cpu_policy property, I have some new opinion, please check my reply in inner dicussion block.","accounts_in_message":[],"_revision_number":1},{"id":"e4d8e4b600ada93e252658fbddac97a309c81d43","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2020-03-24 14:07:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nGood catch Alex!","accounts_in_message":[],"_revision_number":1},{"id":"88274bddccadf2400a12ad0dde1d0f79c109dda4","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-24 14:30:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"598c204ac0b429d97956edcc006bfdb894002687","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-24 14:33:40.000000000","message":"Patch Set 2:\n\nUnable to freeze job graph: Decryption failed.","accounts_in_message":[],"_revision_number":2},{"id":"36c8f250106ff55a794e600af6633111873458b2","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-24 14:58:34.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/54/713354/2/check/pmem-tempest-plugin-filtered/cba075b/ : SUCCESS in 26m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4f757a2eb15965b24aaf7eb1c4c58e17e0cba768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-24 16:23:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4620520b76f1432298a4d8e474a1d97c : SUCCESS in 1h 06m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2cd130b170494cebaa12503a3f6a3187 : SUCCESS in 1h 28m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b71412d711f49528fe2674dc21ef2bd : SUCCESS in 18m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c975a4bcb834245a4536d8f967f03da : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4be6a996ed3043af9a6d91ec1ca6bfe3 : SUCCESS in 10m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dbea8ad153584c35a6a7ea86a5117687 : SUCCESS in 14m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6999c6fb01d54f98849ac9f1a1ada76a : SUCCESS in 15m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/352da1ec8e8149de8159cc11fa2d98bd : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9751c020c3a344edbb8b6ebf044ccb87 : SUCCESS in 14m 08s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/be38b872017749b6bf0595285f9739ef : SUCCESS in 1h 04m 52s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0f616765ec3341e49b4c0d0f1889d0f3 : SUCCESS in 1h 01m 50s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec25a9cb66094eafb93b923cbb84ac99 : SUCCESS in 1h 19m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/78bd03509bd54689a684c3bc4b6f5f13 : SUCCESS in 50m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0218b14a57e44a1aae4aad078d895836 : SUCCESS in 1h 33m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/3c1e906280ea4da49749d67ed852509c : SUCCESS in 1h 35m 48s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/01124ee5a8c04590a773a7218efa227b : SUCCESS in 20m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/471e6d999a544527aaa65f92f70620c9 : SUCCESS in 59m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c146c060887846099d9721985f4713d2 : SUCCESS in 59m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d46c3af9c075f005c4c8ef70eae0865f7d9331dc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 16:27:01.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/54/713354/2/check/EMC_VxFlexOS_NOVA/89ad606/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"4f1565c827f072db43c657edc9fc8d2d432a2b89","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-24 17:07:00.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nNeed to review this fully again, but dansmith\u0027s comments would appear to still apply here","accounts_in_message":[],"_revision_number":2},{"id":"6514d02c6639a1aac3906426a317024424b83d8c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-24 17:26:41.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42118 : SUCCESS in 1h 40m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5b471f3512b38eef58ef9467d86114c2906c1368","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-24 18:48:58.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-54-713354-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"a76c80a66be95a3928514caa24aa7257b373743b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-24 19:39:39.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/713354/2 : SUCCESS in 2h 49m 51s","accounts_in_message":[],"_revision_number":2},{"id":"721626bd2014aa34405c08f9f340580da3d0c555","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-24 19:50:40.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713354/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/5d63b57 : SUCCESS in 1h 41m 16s","accounts_in_message":[],"_revision_number":2},{"id":"3c678e25d4849e0fec3846e3adc3382c154ce7f3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-24 20:23:30.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/2/check/tempest-dsvm-full-bionic/c951898/ : SUCCESS in 1h 49m 56s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/2/check/tempest-dsvm-full-bionic-py3/72ba6b6/ : SUCCESS in 1h 53m 47s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/2/check/grenade-dsvm-bionic/a48e713/ : SUCCESS in 3h 04m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b407e0c2ecb69c1d5d137ca862ac4b3721fa9d1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-25 01:58:26.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/54/713354/2/check-vote/ext-nova-zuul/6eea723 : SUCCESS in 1h 41m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b02c95d5387e59c20f81f58547178725636de809","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-25 05:08:32.000000000","message":"Patch Set 1:\n\n(3 comments)\n\n@smith, My understanding is I need to bump the version of object that used the modified \u0027CPUAllocationPolicy\u0027, and I still need to update the MD5 code of objects such as \u0027InstanceNUMACell\u0027 right?","accounts_in_message":[],"_revision_number":1},{"id":"afd96cc98c740e0ca843369856f0ee4bd06569b0","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-25 09:36:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc23fcf5a0d5e4a0b0814626130698a3f9c324e2","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-25 09:37:50.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"9bc8d416a5c7a56f33340e6e1fe6bcbe93693e86","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-25 09:47:34.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/713354/3 : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":3},{"id":"02ce4dc09c94dae3f59a3637dad0e975a69edec4","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-25 11:05:48.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42193 : SUCCESS in 1h 18m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ea05e072cc0840d78a71b38cdf50572fdf60adce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 11:13:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/94970efd67da46e2b3bd21fd4d8cebe3 : SUCCESS in 1h 02m 21s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7eb68edc2e1e4470ba3075a23f0a04b9 : SUCCESS in 1h 33m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a02e4fda6061498d843626e0bc6fb45a : SUCCESS in 23m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fbfa2d73a6454d79a3c4cc550bf61130 : SUCCESS in 20m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5c622c0a0f74571b0ea472d07d30148 : SUCCESS in 10m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4753bb2a8e1d47cc978130caaca8c78b : SUCCESS in 20m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b9252225019144e1bbf823d688efd394 : SUCCESS in 15m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b36c00c74d444206b649c3963e6f2828 : SUCCESS in 20m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd26807569147efbd2030cf017aa425 : SUCCESS in 12m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dae79424c83f4bf89516cef1acfe8f2c : SUCCESS in 54m 19s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/255fcbd14f284e3eb1aa5cc3950a622f : SUCCESS in 55m 37s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c9e80c6ead2640dfa1f4f2d35e11286b : SUCCESS in 1h 08m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9003a7f2a3ba4205aed47bff42b3d956 : SUCCESS in 48m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6cb43d2f41164d9a8a9506b8b916464d : SUCCESS in 1h 15m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/7dc7e1d7a0864e8ab0d6a63a3c84b596 : SUCCESS in 1h 23m 23s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25f4c4180c684fd6a68c573de3b2c752 : FAILURE in 15m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c18c6dab6c2403084ef883828070d4b : SUCCESS in 58m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a01b1d8e0f3f498a8e5d44bc4fc2e59f : SUCCESS in 50m 46s","accounts_in_message":[],"_revision_number":3},{"id":"ae25e5e9adb86ff90e73294e77a9994bf97bc4fd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-25 21:18:27.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/3/check/tempest-dsvm-full-bionic/8ba00a2/ : SUCCESS in 1h 49m 23s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/3/check/tempest-dsvm-full-bionic-py3/b22410f/ : SUCCESS in 2h 01m 32s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/713354/3/check/grenade-dsvm-bionic/88ffefd/ : POST_FAILURE in 3h 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d86f37613de2d76dd8e7f3c75c88f290660687e3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-25 22:59:59.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713354/3/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/065bbcf : SUCCESS in 1h 36m 58s","accounts_in_message":[],"_revision_number":3},{"id":"09e66a1ca2d0b953417bbf74df3d8605e90c3cc3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-26 02:20:29.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/54/713354/3/check-vote/ext-nova-zuul/5088c6a : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b3d1d0cb634bcecee3b1764b48dfe8110784d65a","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-26 16:25:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"83f943449d131431bd98ff09076702caecc2a956","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-26 16:47:13.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/54/713354/4/check/pmem-tempest-plugin-filtered/7389ac6/ : FAILURE in 17m 11s","accounts_in_message":[],"_revision_number":4},{"id":"91b3eb9f009252fd31e1681cfc0a85c998c472a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-26 18:12:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/48d67bae938e4b95b59eabdcdb73f780 : FAILURE in 55m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/76922f1a2baa4da6abf408eb41440182 : FAILURE in 24m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3375de2320743ed9875a19200e3ad1c : FAILURE in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/080b3672d05f4c88bc6d3cbbf0d750da : FAILURE in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5fadcc8e0ac435fb11af14eabc71fb5 : FAILURE in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b9e38c1e95d49f1a92a9af193f692e0 : FAILURE in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86a5836cb2b845d4a1d73294e47e9181 : FAILURE in 8m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4cb8882840f48ebbccd4e4adc93e482 : FAILURE in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5064383d12354a57abad204bb9fd81c3 : FAILURE in 10m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1664fc0550004e73a05d8686709ba871 : FAILURE in 27m 20s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ea786820fcf546b4be21b13b45119d0a : FAILURE in 27m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5b3cf52a3b1490bb933812d856172df : FAILURE in 1h 06m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f5a9a1c9e1f34e159d7e2424df660cb7 : FAILURE in 31m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5d61b6372fc54d21b6fa908de48a2603 : FAILURE in 35m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/463b79fcf1e8447996662fcacba3c5d8 : POST_FAILURE in 33m 26s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6578dd7a335b4d83b3e786ef8e7ef680 : FAILURE in 9m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77afc0e73760454c802ab81afed69a6b : FAILURE in 28m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5728f6021ca74126aabcb4ac1a940e2c : FAILURE in 32m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b91079d92ed86d75995d4a84fef252771941a90d","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-26 19:35:23.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42300 : FAILURE in 51m 02s","accounts_in_message":[],"_revision_number":4},{"id":"a8d65087cacb5ddc6bfdbf6b8fc2bb2b736f1f56","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-26 21:10:31.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/713354/4 : FAILURE in 29m 47s","accounts_in_message":[],"_revision_number":4},{"id":"e6f4a91b905fb4ef0765c151c5c62e074bd61c6b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-27 08:04:57.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713354/4/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/4d7c6c5 : FAILURE in 50m 49s","accounts_in_message":[],"_revision_number":4},{"id":"86cb2984f253cf4fc77ca603461c3d894aefdd7a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-27 16:21:10.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/54/713354/4/check-vote/ext-nova-zuul/5ae7b44 : FAILURE in 26m 14s","accounts_in_message":[],"_revision_number":4},{"id":"f42173400932d42a7398e1e37c2ede4527cf89f0","author":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"date":"2020-03-31 12:27:06.000000000","message":"Abandoned\n\nThis patch is splited into a number of small patches within same bp.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"87f30de7c8bd2d3923ba3052935751ca4037ed12","revisions":{"7b7c688ff58e8c3ee94fdf94be3151743d5c6d57":{"kind":"REWORK","_number":1,"created":"2020-03-17 00:55:27.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/54/713354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/713354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/713354/1"}}},"commit":{"parents":[{"commit":"5fcaaff2d08be274000bd7e9148dacd7dbff09d2","subject":"Derive \u0027cpu_pinning_requested\u0027 property from cpu_policy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5fcaaff2d08be274000bd7e9148dacd7dbff09d2"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-16 16:08:53.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-17 00:37:36.000000000","tz":480},"subject":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance","message":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance\n\nCurrently we have \u0027DEDICATED\u0027 and \u0027SHARED\u0027 policies for instance CPU\nallocation from host CPUs. \u0027MIXED\u0027 is another CPU allcation policy\nfor the instance obtaining some \u0027dedicated\u0027 CPUs, that\u0027s pinning to\nparticular host CPU and used exclusively by the instance, and some\n\u0027sharing\u0027 host CPUs, which will be sharing with other instances.\n\nAlso introduce the interface of flavor extra spec \u0027hw:pinned_cpus\u0027\nfor specifiying the bitmask of \u0027dedicated\u0027 instance CPUs.\n\nChange-Id: Ide2eab6dfb0d2100d0f5501ed2a4070b9201320f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b7c688ff58e8c3ee94fdf94be3151743d5c6d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7b7c688ff58e8c3ee94fdf94be3151743d5c6d57"}]},"branch":"refs/heads/master"},"bd89162b336eaab228dc5c374f308410a5e77904":{"kind":"REWORK","_number":2,"created":"2020-03-24 14:30:00.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/54/713354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/713354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/713354/2"}}},"commit":{"parents":[{"commit":"4cfd4c5944569b316b01ae99cf76951977ebc8f2","subject":"Refactor the code in checking available host CPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4cfd4c5944569b316b01ae99cf76951977ebc8f2"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-24 14:05:47.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-24 14:07:13.000000000","tz":480},"subject":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance","message":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance\n\nCurrently we have \u0027DEDICATED\u0027 and \u0027SHARED\u0027 policies for instance CPU\nallocation from host CPUs. \u0027MIXED\u0027 is another CPU allcation policy\nfor the instance obtaining some \u0027dedicated\u0027 CPUs, that\u0027s pinning to\nparticular host CPU and used exclusively by the instance, and some\n\u0027sharing\u0027 host CPUs, which will be sharing with other instances.\n\nAlso introduce the interface of flavor extra spec \u0027hw:dedicated_cpus\u0027\nfor specifiying the bitmask of \u0027dedicated\u0027 instance CPUs.\n\nChange-Id: Ide2eab6dfb0d2100d0f5501ed2a4070b9201320f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd89162b336eaab228dc5c374f308410a5e77904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd89162b336eaab228dc5c374f308410a5e77904"}]},"branch":"refs/heads/master"},"a6d582f94e4bd479cf48b8316f18f3935199a000":{"kind":"REWORK","_number":3,"created":"2020-03-25 09:36:57.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/54/713354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/713354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/713354/3"}}},"commit":{"parents":[{"commit":"896547884916b9dad2e7f832cb1eb740ff6d7126","subject":"Refactor the code in checking available host CPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/896547884916b9dad2e7f832cb1eb740ff6d7126"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-24 14:05:47.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-25 09:36:25.000000000","tz":480},"subject":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance","message":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance\n\nCurrently we have \u0027DEDICATED\u0027 and \u0027SHARED\u0027 policies for instance CPU\nallocation from host CPUs. \u0027MIXED\u0027 is another CPU allcation policy\nfor the instance obtaining some \u0027dedicated\u0027 CPUs, that\u0027s pinning to\nparticular host CPU and used exclusively by the instance, and some\n\u0027sharing\u0027 host CPUs, which will be sharing with other instances.\n\nAlso introduce the interface of flavor extra spec \u0027hw:dedicated_cpus\u0027\nfor specifiying the bitmask of \u0027dedicated\u0027 instance CPUs.\n\nChange-Id: Ide2eab6dfb0d2100d0f5501ed2a4070b9201320f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6d582f94e4bd479cf48b8316f18f3935199a000"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a6d582f94e4bd479cf48b8316f18f3935199a000"}]},"branch":"refs/heads/master"},"87f30de7c8bd2d3923ba3052935751ca4037ed12":{"kind":"REWORK","_number":4,"created":"2020-03-26 16:25:31.000000000","uploader":{"_account_id":30209,"name":"Huaqiang","email":"huaqiang.wang@intel.com","username":"Huaqiang.Wang"},"ref":"refs/changes/54/713354/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/713354/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/713354/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/713354/4"}}},"commit":{"parents":[{"commit":"ced4cbdf2da004612fc83a3d7afaf3f2980608d8","subject":"Refactor the code in checking available host CPUs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ced4cbdf2da004612fc83a3d7afaf3f2980608d8"}]}],"author":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-24 14:05:47.000000000","tz":480},"committer":{"name":"Wang Huaqiang","email":"huaqiang.wang@intel.com","date":"2020-03-26 16:24:58.000000000","tz":480},"subject":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance","message":"Introduce \u0027MIXED\u0027 CPU allocation policy for instance\n\nCurrently we have \u0027DEDICATED\u0027 and \u0027SHARED\u0027 policies for instance CPU\nallocation from host CPUs. \u0027MIXED\u0027 is another CPU allcation policy\nfor the instance obtaining some \u0027dedicated\u0027 CPUs, that\u0027s pinning to\nparticular host CPU and used exclusively by the instance, and some\n\u0027sharing\u0027 host CPUs, which will be sharing with other instances.\n\nAlso introduce the interface of flavor extra spec \u0027hw:dedicated_cpus\u0027\nfor specifiying the bitmask of \u0027dedicated\u0027 instance CPUs.\n\nChange-Id: Ide2eab6dfb0d2100d0f5501ed2a4070b9201320f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87f30de7c8bd2d3923ba3052935751ca4037ed12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/87f30de7c8bd2d3923ba3052935751ca4037ed12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
