)]}'
{"id":"openstack%2Fnova~713559","triplet_id":"openstack%2Fnova~master~Id3066c198b616fd5a82ccacde49776f1ec6cf221","project":"openstack/nova","branch":"master","topic":"bp/policy-defaults-refresh","hashtags":[],"change_id":"Id3066c198b616fd5a82ccacde49776f1ec6cf221","subject":"Introduce scope_types in os-flavor-access","status":"MERGED","created":"2020-03-18 00:46:16.000000000","updated":"2020-03-20 17:25:58.000000000","submitted":"2020-03-20 17:15:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713559-1584724559736-eaa21e4c","meta_rev_id":"cdede370ae37054f14aa1d7892547d7da34c4d1f","_number":713559,"virtual_id_number":713559,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-03-20 14:22:53.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-20 14:37:47.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-18 02:40:43.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-18 03:27:52.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-03-20 13:20:28.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-03-20 15:41:53.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-20 12:47:27.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2020-03-20 11:52:27.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-20 17:15:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-18 18:11:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-20 09:14:52.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-20 09:14:52.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-18 02:40:43.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-18 03:27:52.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-18 18:11:29.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-03-20 09:14:52.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2020-03-20 11:52:27.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-20 12:47:27.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-03-20 13:20:28.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-20 14:22:53.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-20 14:37:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-20 15:41:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-20 17:15:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8fc6425defcfa3007ae99dfd31bbe71321e74591","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-18 00:46:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c849ce3f3cdeae593643a891e1dd6c30aacbe95f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-18 00:46:48.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c2fe2fa2032208bc51d13540ce47d9960320d5c5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-18 00:48:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f06e77252e4b4dbd6723374653dd9058acc3593f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-18 00:49:06.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"b0b662c9fbe67733bbe598f430c40cb5048459f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-18 00:53:53.000000000","message":"Topic changed from bug/1867840 to bp/policy-defaults-refresh","accounts_in_message":[],"_revision_number":2},{"id":"daf7e392de2fc0b3b63d37552b3637a115cf460b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-18 01:05:56.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"32742957f48b0cf9959581d20d88a0a0d9468df1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-03-18 01:06:14.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"5ecbeac56a9d927e8bcc81140b600514d231278f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-18 01:06:30.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"f4324a29c2f91d8ade764f7f23df5bbe744092d8","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-18 01:06:53.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"cc8c9c426a30d7dfa453064ddaa4e96c56e29642","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-18 02:34:34.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41882 : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":4},{"id":"b1e99859aa7c6950114dfad6be55d98dc95321f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-18 02:40:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/4dc8f73069c4419dac54de3bd44221e4 : SUCCESS in 1h 05m 15s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6abba18531a4400e8afd6a8b7629abac : SUCCESS in 1h 30m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c93dbe54c58b475bb8817fc70d9b6518 : SUCCESS in 15m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/824271a552cd47ecba4069e19789fad6 : SUCCESS in 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40190c28646140c7b1e072e08760ef5c : SUCCESS in 7m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/900045ae02624c4d918b2ab7ae1acdfc : SUCCESS in 14m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fd0ba240a6849a59348a803fa464cee : SUCCESS in 14m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01333096278a4d7882f9ecca01c39bda : SUCCESS in 16m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab41eb73c5c64302be64996d9ec8315e : SUCCESS in 9m 58s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff5b793d8a6e4a188e35d25eacee2fc6 : SUCCESS in 15m 53s","accounts_in_message":[],"_revision_number":4},{"id":"0345411179f2bc0ceed35a010ce894ef6b742bec","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 02:40:43.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713559/4/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/0d2bc8c : FAILURE in 1h 33m 53s","accounts_in_message":[],"_revision_number":4},{"id":"7b1be23914f6678ed08085a28c6062cc118078ff","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-18 02:41:18.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/59/713559/4/check/EMC_VxFlexOS_NOVA/5b047af/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"cdc92b589da4661eccea2667177acbcbf9110ee5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-18 03:27:52.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/713559/4 : SUCCESS in 2h 19m 36s","accounts_in_message":[],"_revision_number":4},{"id":"03a7ba42871dd41feb6a6947cc272fe5c67dc7a8","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-18 03:38:09.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-59-713559-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"03d48c31b94e0146c80ff93f3962a1bf514eb1c1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-18 04:14:28.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/59/713559/4/check/tempest-dsvm-full-bionic/66a4884/ : SUCCESS in 1h 39m 19s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/59/713559/4/check/tempest-dsvm-full-bionic-py3/47048b9/ : SUCCESS in 1h 23m 42s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/59/713559/4/check/grenade-dsvm-bionic/b2ca9f5/ : SUCCESS in 3h 02m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a38d1d1bd385637a1f305e101b70bfea1cc23d76","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-18 18:11:29.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1e73fe45135071159a38d6b01c72215f0f8ebf3f","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-03-19 17:54:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8183c16dd6fed1c2fe1ebd6807f83df915e1121b","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2020-03-20 09:14:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"00bdcd429b3aaeead05a68f0bd6034deb965744f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 09:15:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"af94e9a64d094c331737e4b362229d62371d4b83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 10:36:44.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/69a01481f7b34382a5002805a0f679b6 : FAILURE in 44m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a883e3118d5c461a9ac65d33ab5768d3 : SUCCESS in 1h 19m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d90dd0e228ad41d0b06e726ededbf9c5 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046c70f4b27f465dbccbfb7550860a2a : SUCCESS in 7m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9acaea9168e94e7ebeca9444f5452ade : SUCCESS in 12m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3374ecdf77c74ae1a11ee9dc39aac686 : SUCCESS in 13m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ea7b4a6e8854906b9e51dee787f6892 : SUCCESS in 11m 39s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/13cd63c0a2194e6bbe826c07bb9c46a6 : SUCCESS in 14m 49s","accounts_in_message":[],"_revision_number":4},{"id":"bd4d7ca817b8d027ea2127a7ac192263fc780de8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-03-20 11:51:55.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"2bd58dfa168d9fb44db2c30b30632d5f666120b0","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-20 11:52:27.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"b3b28bc04b3f2ce1715e22c5ab22c3b2e13917bc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-20 12:47:27.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-41994 : SUCCESS in 55m 04s","accounts_in_message":[],"_revision_number":4},{"id":"91e2ee326c953d97d41759ca0b4bd9f2ce34beb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 13:19:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/13fdd53fb227449f949d12b5d29fd5eb : SUCCESS in 1h 07m 58s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/432d53da5350438989d3ec3acbc342fc : SUCCESS in 1h 21m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/046e9d4e1be74c24849b997a3b255bea : SUCCESS in 17m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3d117a8db1f942b9b105ad7c7531fc9c : SUCCESS in 15m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e641013a33d4c71a233832823f47775 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b0b0d4c44c74680bd9d7d04b5256105 : SUCCESS in 14m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5d1414ddb3404f689bbcc6fa85093482 : SUCCESS in 15m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cdfee6466bd4afaa6cee70cf273fad3 : SUCCESS in 13m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f3cb9122490411594e2b0bd6c5ccca5 : SUCCESS in 10m 30s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f396b70fb36142a6870132cb1dc99f4e : SUCCESS in 16m 06s","accounts_in_message":[],"_revision_number":4},{"id":"f1b3477310aa962ab65adf7e556fcbcd0bffc78f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-03-20 13:20:28.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/59/713559/4/check/EMC_VxFlexOS_NOVA/571440a/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"23f4c67b92d104032efaeadc605c1e76f6cb6e73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 13:20:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"757f97461467fc5726be2d8ce069a5e63eacc9d3","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-03-20 14:22:53.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-59-713559-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"e6c0fe6ef1934a2c6515521debe3e4a21341a5dc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-03-20 14:37:47.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/59/713559/4/check-vote/ext-nova-zuul/fd65885 : FAILURE in 1h 18m 43s","accounts_in_message":[],"_revision_number":4},{"id":"44453dec006b1c5687cd83869ac0e9dcea024341","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-20 15:41:53.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/59/713559/4/check/tempest-dsvm-full-bionic/f1e100e/ : SUCCESS in 1h 45m 32s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/59/713559/4/check/tempest-dsvm-full-bionic-py3/939f512/ : SUCCESS in 1h 46m 17s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/59/713559/4/check/grenade-dsvm-bionic/a6f858b/ : SUCCESS in 3h 16m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d9d0f4567d6ed5e84c3097b5ddd3af8a8a33a558","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:15:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1f27ab131a18edad6267d12880e08a02277eedbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:15:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f2649816df594254a766a07c70fb2eb6 : SUCCESS in 1h 14m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9e0b7f5b8dd3498295b21361d7671327 : SUCCESS in 1h 34m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/89e97f1b9ab343aba00f652d28dbd969 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89850ff02aa54f45baec066ee82b82f6 : SUCCESS in 11m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d388b86717549aa8e1d91603e76f230 : SUCCESS in 14m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac34964ce9fd4165a5075652f7d24119 : SUCCESS in 16m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdab4931f5c4031b7d811f68250dab2 : SUCCESS in 16m 09s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/70ee299f430445439804a607c1c16f8b : SUCCESS in 21m 27s","accounts_in_message":[],"_revision_number":4},{"id":"cdede370ae37054f14aa1d7892547d7da34c4d1f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 17:25:58.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b9a191cedd4ab4a5e9407a6059380f : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"741c0f60bde282c5cb7b94f4574c3f6be697a207","revisions":{"828ced9f2e98ae0dba2b71a975a0810dec2332eb":{"kind":"REWORK","_number":1,"created":"2020-03-18 00:46:16.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/713559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/713559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/713559/1"}}},"commit":{"parents":[{"commit":"673806dd67591c8ae0f6fab9c832057d0672ff3a","subject":"Add test coverage of existing flavor_access policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/673806dd67591c8ae0f6fab9c832057d0672ff3a"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-17 19:42:45.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-17 19:42:45.000000000","tz":-300},"subject":"Introduce scope_types in os-flavor-access","message":"Introduce scope_types in os-flavor-access\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-flavor-access API policies\nas \u0027system\u0027. List flavor access policy is allowed for everyone even\nit is default as admin_or_owner we can correct it to admin only\nwhile intriducing the new defaults. Details: Bug#1867840\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Id3066c198b616fd5a82ccacde49776f1ec6cf221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/828ced9f2e98ae0dba2b71a975a0810dec2332eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/828ced9f2e98ae0dba2b71a975a0810dec2332eb"}]},"branch":"refs/heads/master"},"7098b32056002e8ee4c2ceacc86c9d21910bfd8b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-03-18 00:48:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/713559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/713559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/713559/2"}}},"commit":{"parents":[{"commit":"8ae48db559d8e11b00ae0f15644782a9e04c6955","subject":"Add test coverage of existing flavor_access policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8ae48db559d8e11b00ae0f15644782a9e04c6955"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-17 19:42:45.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-18 00:48:20.000000000","tz":0},"subject":"Introduce scope_types in os-flavor-access","message":"Introduce scope_types in os-flavor-access\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-flavor-access API policies\nas \u0027system\u0027. List flavor access policy is allowed for everyone even\nit is default as admin_or_owner we can correct it to admin only\nwhile intriducing the new defaults. Details: Bug#1867840\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Id3066c198b616fd5a82ccacde49776f1ec6cf221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7098b32056002e8ee4c2ceacc86c9d21910bfd8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7098b32056002e8ee4c2ceacc86c9d21910bfd8b"}]},"branch":"refs/heads/master"},"2aafb26907737608d9ab9681a5db4fe665275831":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-18 01:05:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/713559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/713559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/713559/3"}}},"commit":{"parents":[{"commit":"3393d998203be0cfb1e09f17316177b73343cfdc","subject":"Add test coverage of existing flavor_access policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3393d998203be0cfb1e09f17316177b73343cfdc"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-17 19:42:45.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-18 01:05:56.000000000","tz":0},"subject":"Introduce scope_types in os-flavor-access","message":"Introduce scope_types in os-flavor-access\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-flavor-access API policies\nas \u0027system\u0027. List flavor access policy is allowed for everyone even\nit is default as admin_or_owner we can correct it to admin only\nwhile intriducing the new defaults. Details: Bug#1867840\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Id3066c198b616fd5a82ccacde49776f1ec6cf221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2aafb26907737608d9ab9681a5db4fe665275831"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2aafb26907737608d9ab9681a5db4fe665275831"}]},"branch":"refs/heads/master"},"741c0f60bde282c5cb7b94f4574c3f6be697a207":{"kind":"REWORK","_number":4,"created":"2020-03-18 01:06:14.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/59/713559/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/59/713559/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/59/713559/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/59/713559/4"}}},"commit":{"parents":[{"commit":"3393d998203be0cfb1e09f17316177b73343cfdc","subject":"Add test coverage of existing flavor_access policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3393d998203be0cfb1e09f17316177b73343cfdc"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-17 19:42:45.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-03-18 01:06:11.000000000","tz":0},"subject":"Introduce scope_types in os-flavor-access","message":"Introduce scope_types in os-flavor-access\n\noslo.policy introduced the scope_type feature which can\ncontrol the access level at system-level and project-level.\n - https://docs.openstack.org/oslo.policy/latest/user/usage.html#setting-scope\n - http://specs.openstack.org/openstack/keystone-specs/specs/keystone/queens/system-scope.html\n\nAppropriate scope_type for nova case:\n- https://specs.openstack.org/openstack/nova-specs/specs/ussuri/approved/policy-defaults-refresh.html#scope\n\nThis commit introduce scope_type for os-flavor-access API policies\nas \u0027system\u0027. List flavor access policy is allowed for everyone even\nit is default as admin_or_owner we can correct it to admin only\nwhile intriducing the new defaults. Details: Bug#1867840\n\nAlso adds the test case with scope_type enabled and verify we\npass and fail the policy check with expected context.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Id3066c198b616fd5a82ccacde49776f1ec6cf221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/741c0f60bde282c5cb7b94f4574c3f6be697a207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/741c0f60bde282c5cb7b94f4574c3f6be697a207"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
