)]}'
{"id":"openstack%2Fnova~713962","triplet_id":"openstack%2Fnova~stable%2Fstein~I8af2ad741ca08c3d88efb9aa817c4d1470491a23","project":"openstack/nova","branch":"stable/stein","topic":"bug/1813789","hashtags":[],"change_id":"I8af2ad741ca08c3d88efb9aa817c4d1470491a23","subject":"nova-live-migration: Ensure subnode is fenced during evacuation testing","status":"MERGED","created":"2020-03-19 18:43:39.000000000","updated":"2020-03-25 21:43:56.000000000","submitted":"2020-03-25 21:41:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"713962-1585172481963-d1a6a7d1","meta_rev_id":"39d383eb31ebedc0076fd5e5a727006b9a86c534","_number":713962,"virtual_id_number":713962,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-03-23 10:33:34.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-03-23 10:48:19.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-03-23 10:49:24.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-25 21:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2020-03-25 17:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-25 14:32:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2020-03-25 17:22:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-19 23:55:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-23 10:33:34.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-03-23 10:48:19.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-23 10:49:24.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-25 14:32:30.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-03-25 17:22:40.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-03-25 21:41:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5391a7b1dd38472103798e49346f3fa50a823a01","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-19 18:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"218d10b4ed9650858cbfb77b0425abd33fe69d20","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-19 18:44:14.000000000","message":"Patch Set 1:\n\nUnable to freeze job graph: \u0027NoneType\u0027 object has no attribute \u0027decrypt\u0027","accounts_in_message":[],"_revision_number":1},{"id":"5055322874d25d60829242a0c98015d0adf23a16","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-19 18:45:59.000000000","message":"Topic set to bug/1813789","accounts_in_message":[],"_revision_number":1},{"id":"93f041487bd06ec71b6212503b542cb26eddd3a5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-19 18:54:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/713962/1 : FAILURE in 9m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b3b0e160796121fbc5e4ba9c8b338aa9009406d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:19:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/79782f74c71d45cf90e4f9618eae381e : SUCCESS in 59m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/882d02b6a92149b482646572f5b677c0 : SUCCESS in 1h 21m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82bd5ce966c842a1917a0896f357e4fd : SUCCESS in 17m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a05299bd0644a98ab6377bbb52b5215 : SUCCESS in 12m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf802dce5e44c37ace0579a4fe29373 : SUCCESS in 11m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b68325b26d040d19a760bb03ab7c903 : SUCCESS in 13m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/42110ea2d8464ae29d99a9b6e0133380 : SUCCESS in 13m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/045f9e0657894c7a9ed296a5cb6e50cc : SUCCESS in 13m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6468138e3ec4af986a261026f809103 : SUCCESS in 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3b71bd2ca841d592910d341e094003 : SUCCESS in 7m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/aeb2dafdb1734c6b8b0b9ae6cde2a042 : SUCCESS in 49m 01s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/233674476428412cab106298fa3aac2b : SUCCESS in 1h 14m 35s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c52f7498bd294da1b7137c426a43e0cc : SUCCESS in 1h 02m 04s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/65814a6c434c4f158a7b53e3dd30bbb7 : SUCCESS in 1h 15m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73bc38f0250f4a539eaacaac18da836d : SUCCESS in 47m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/76487d4c39a848bca81fe0bd720751d4 : SUCCESS in 1h 16m 21s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/66c5aabf0e4a456199194c19b81f3baf : SUCCESS in 20m 41s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7200e2673d5e4da5ba278a5924f4469d : SUCCESS in 22m 56s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/bc2b5845a5b34ab487c6e612a444e365 : SUCCESS in 1h 51m 22s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c12409a51a954d3a9e6fa1a2f5208771 : SUCCESS in 2h 28m 08s","accounts_in_message":[],"_revision_number":1},{"id":"16e1d00eed85d9e0a11c66629c9f38099b645644","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-19 22:03:11.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713962/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/3f932bd : FAILURE in 24m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5de0d46ebdccdf8d1647ce11f7ca8cab1d783770","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-03-19 23:55:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/62/713962/1/check/tempest-dsvm-full-bionic/426e45f/ : TIMED_OUT in 2h 47m 50s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/62/713962/1/check/tempest-dsvm-full-bionic-py3/8564fa1/ : FAILURE in 1h 34m 13s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/62/713962/1/check/grenade-dsvm-bionic/dd70134/ : POST_FAILURE in 2h 37m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db4ea64b10c7c6c3f0e64b3e3e1691f27d55d351","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-20 09:58:49.000000000","message":"Patch Set 1: Workflow-1\n\nWill need to merge Ia7c317e373e4037495d379d06eda19a71412d409 into this.","accounts_in_message":[],"_revision_number":1},{"id":"7885d9b12e0d7069a2cef6aa2c4fdced6e81a7d3","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-03-23 10:22:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40b930c35b880d246e93949c2ce6b67b71b81552","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-23 10:33:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/713962/2 : FAILURE in 9m 45s","accounts_in_message":[],"_revision_number":2},{"id":"770d19827d8237a3cd597ac42bae2a6ca0e477db","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-03-23 10:48:19.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/713962/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/b19442a : FAILURE in 24m 48s","accounts_in_message":[],"_revision_number":2},{"id":"8319e9fc555b778102863b1fe33d26b0f60dc5e9","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-23 10:49:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/62/713962/2/check/pmem-tempest-plugin-filtered/466ada2/ : FAILURE in 25m 00s","accounts_in_message":[],"_revision_number":2},{"id":"08c4aa1ef372526c518842c53b19764316b69304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 12:47:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/62b68cca4a784a7fa39c1e1f9ed20968 : SUCCESS in 52m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70c72cad247845c4974c216d8a90f01a : SUCCESS in 1h 21m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/985f47144fe0482aab858614fd65dcee : SUCCESS in 17m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d5a30512b4b24cf28eeeb62c4b94434f : SUCCESS in 12m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4109e8c429148bf907f390f655aab38 : SUCCESS in 9m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/359cb2bedef445ff89b0e6c14964a184 : SUCCESS in 13m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/960af672f2d648d3bba31270f04601a1 : SUCCESS in 13m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b561a6c55900400aa7268eb7ba5e0200 : SUCCESS in 14m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/48b56c91a73e4fa3bb785a7f43f06372 : SUCCESS in 12m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f9aa9695e7404e9f4cabacfecb6524 : SUCCESS in 7m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c16aaf84500c4092983902c5dd21be4c : SUCCESS in 49m 24s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/343e47b1a7234e9c84beb5f9c1353bc1 : SUCCESS in 57m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a476e8c757f04b009913b0f32e773515 : SUCCESS in 1h 08m 49s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/0ef8109424d641f68ce8f6e6a66f8231 : SUCCESS in 1h 04m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9c9a6e87f1fd4e5883537238134e6a01 : SUCCESS in 48m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/4adb5af4e0f64a09adfa560b9e1194bf : SUCCESS in 1h 25m 58s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/a977d7a6a8ef451098995a9b3da544a9 : SUCCESS in 19m 12s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e9730ec1058f4c3da49aa5b48638826e : SUCCESS in 19m 45s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c3aec7a896864742981614c61d294cfd : SUCCESS in 1h 56m 00s (non-voting)\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/81b271516d104e1d955775c5d67574c5 : SUCCESS in 2h 16m 42s","accounts_in_message":[],"_revision_number":2},{"id":"398a097c0794adf3a97f757c146c139ea5d0b66a","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-24 22:06:15.000000000","message":"Patch Set 2: Code-Review+1\n\nClean cherry pick from the train change. Must wait for merge of the train change","accounts_in_message":[],"_revision_number":2},{"id":"e0c8e2b038d10d382f626a51952aec63d23b3838","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-03-25 14:32:30.000000000","message":"Patch Set 2: Code-Review+2\n\nClean cherry pick from Train, where the patch has merged. LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"e7db78214b018ccc14efde524384d5bdb8a59bcd","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-03-25 17:22:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nTrain change has merged, upgrading to +2/+W","accounts_in_message":[],"_revision_number":2},{"id":"4f7b6ce4c621e9df2485ccd97e5fe205983b0827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 17:23:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0b2c3611d12913d2dcfcf79429d9bb5180fb9151","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:41:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"036dde1570d7d71fe05a5a4a8a71bf9e03187f85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:41:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/53db166d00a9461a9bdcee65e04652f2 : SUCCESS in 1h 05m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d22edfe2a5654a5e85f228ee38fd5ecb : SUCCESS in 1h 23m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc822e2ea25848cead82c1cd142c1d33 : SUCCESS in 13m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12965380119d40518705d2bda0e70667 : SUCCESS in 11m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8999d2cc4c764b3b8f52a7ef138aa3ec : SUCCESS in 14m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36310f4803fc4190bc60c8ef0ea363db : SUCCESS in 19m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e663363d352a422e80e006ee927b72c9 : SUCCESS in 14m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/279a7c2a41c34c7592cd8d0dbdad07be : SUCCESS in 16m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8a28021df004140b3d36413e8b3d523 : SUCCESS in 8m 08s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/940b465adf3f4c749029855a34530f32 : SUCCESS in 1h 12m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/90c1b18480934c4fa0f4f7f1863f09c9 : SUCCESS in 55m 49s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/4c0b097904894c76ae342e92f847dad0 : SUCCESS in 20m 38s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/90545c366a804db7b50cde74d7e95c6a : SUCCESS in 19m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/3de76bf0000744238834218f18d97a09 : SUCCESS in 1h 20m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b2b88ab4da94a438efb316511de5a59 : SUCCESS in 2h 17m 42s","accounts_in_message":[],"_revision_number":2},{"id":"39d383eb31ebedc0076fd5e5a727006b9a86c534","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-25 21:43:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28738c05f6e34748a28d83d9b48c7b2d : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e7cd8b49e21852b5fafec7215d40e758cde92a35","revisions":{"75e13e539567e782baf389d268a1083f11cf5976":{"kind":"REWORK","_number":1,"created":"2020-03-19 18:43:39.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/62/713962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/62/713962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/713962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/713962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/713962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/62/713962/1"}}},"commit":{"parents":[{"commit":"373c4ffde2053c7ff11bd38339b88d144cd442f2","subject":"nova-live-migration: Wait for n-cpu services to come up after configuring Ceph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/373c4ffde2053c7ff11bd38339b88d144cd442f2"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-03-18 15:17:27.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-03-19 18:43:08.000000000","tz":0},"subject":"nova-live-migration: Ensure subnode is fenced during evacuation testing","message":"nova-live-migration: Ensure subnode is fenced during evacuation testing\n\nAs stated in the forced-down API [1]:\n\n\u003e Setting a service forced down without completely fencing it will\n\u003e likely result in the corruption of VMs on that host.\n\nPreviously only the libvirtd service was stopped on the subnode prior to\ncalling this API, allowing n-cpu, q-agt and the underlying guest domains\nto continue running on the host.\n\nThis change now ensures all devstack services are stopped on the subnode\nand all active domains destroyed.\n\nIt is hoped that this will resolve bug #1813789 where evacuations have\ntimed out due to VIF plugging issues on the new destination host.\n\n[1] https://docs.openstack.org/api-ref/compute/?expanded\u003dupdate-forced-down-detail#update-forced-down\n\nRelated-Bug: #1813789\nChange-Id: I8af2ad741ca08c3d88efb9aa817c4d1470491a23\n(cherry picked from commit b097959c1cbc9af1d90c7502286bc3e20972201f)\n(cherry picked from commit e15d3483f1248b1bba3259f537b7ebf58dada226)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75e13e539567e782baf389d268a1083f11cf5976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75e13e539567e782baf389d268a1083f11cf5976"}]},"branch":"refs/heads/stable/stein"},"e7cd8b49e21852b5fafec7215d40e758cde92a35":{"kind":"REWORK","_number":2,"created":"2020-03-23 10:22:59.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/62/713962/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/62/713962/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/713962/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/713962/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/62/713962/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/62/713962/2"}}},"commit":{"parents":[{"commit":"373c4ffde2053c7ff11bd38339b88d144cd442f2","subject":"nova-live-migration: Wait for n-cpu services to come up after configuring Ceph","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/373c4ffde2053c7ff11bd38339b88d144cd442f2"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-03-18 15:17:27.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-03-23 10:22:33.000000000","tz":0},"subject":"nova-live-migration: Ensure subnode is fenced during evacuation testing","message":"nova-live-migration: Ensure subnode is fenced during evacuation testing\n\nAs stated in the forced-down API [1]:\n\n\u003e Setting a service forced down without completely fencing it will\n\u003e likely result in the corruption of VMs on that host.\n\nPreviously only the libvirtd service was stopped on the subnode prior to\ncalling this API, allowing n-cpu, q-agt and the underlying guest domains\nto continue running on the host.\n\nThis change now ensures all devstack services are stopped on the subnode\nand all active domains destroyed.\n\nIt is hoped that this will resolve bug #1813789 where evacuations have\ntimed out due to VIF plugging issues on the new destination host.\n\n[1] https://docs.openstack.org/api-ref/compute/?expanded\u003dupdate-forced-down-detail#update-forced-down\n\nNOTE(lyarwood): The following change is squashed here to allow both to\npass the gate without encoutering additional failures.\n\nnova-live-migration: Only stop n-cpu and q-agt during evacuation testing\n\nI8af2ad741ca08c3d88efb9aa817c4d1470491a23 started to correctly fence the\nsubnode ahead of evacuation testing but missed that c-vol and g-api\nwhere also running on the host. As a result the BFV evacuation test will\nfail if the volume being used is created on the c-vol backend hosted on\nthe subnode.\n\nThis change now avoids this by limiting the services stopped ahead of\nthe evacuation on the subnode to n-cpu and q-agt.\n\nChange-Id: Ia7c317e373e4037495d379d06eda19a71412d409\nCloses-Bug: #1868234\n(cherry picked from commit 1e16b3184d4e298c454ede7c56040f6d70276a0c)\n\nRelated-Bug: #1813789\nChange-Id: I8af2ad741ca08c3d88efb9aa817c4d1470491a23\n(cherry picked from commit b097959c1cbc9af1d90c7502286bc3e20972201f)\n(cherry picked from commit 91d410b92f86524dffb781c7a49ecb3ec8ac8a56)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7cd8b49e21852b5fafec7215d40e758cde92a35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e7cd8b49e21852b5fafec7215d40e758cde92a35"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
