)]}'
{"id":"openstack%2Fnova~716275","triplet_id":"openstack%2Fnova~master~Ibc939bcec7cf911ec33c159c4a8ecd6ac587a544","project":"openstack/nova","branch":"master","topic":"bug/1869877","hashtags":[],"change_id":"Ibc939bcec7cf911ec33c159c4a8ecd6ac587a544","subject":"[WIP] Respect multiple segments in network","status":"NEW","created":"2020-03-31 12:29:21.000000000","updated":"2020-04-13 08:05:48.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7386b9198ae8e0c8c5ef41d74b08cb44bc2e3d26","_number":716275,"virtual_id_number":716275,"owner":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-01 18:36:42.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-04-13 08:05:48.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"date":"2020-03-31 20:44:19.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"tag":"autogenerated:jenkins-gerrit-trigger","date":"2020-04-01 09:01:06.000000000","_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-01 00:24:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"date":"2020-03-31 18:24:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"date":"2020-03-31 12:39:20.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-03-31 15:00:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-03-31 12:38:06.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-31 12:39:20.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-03-31 15:00:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-03-31 18:24:39.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-03-31 20:44:19.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-01 00:24:53.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 09:01:06.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-01 18:36:42.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-13 08:05:48.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ceccb848a227d8b2950739ababc7234bc23639a","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:29:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"769758e423802c56954394c3d2fd6baa15d46298","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:29:38.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7402d8e70f3cba063fd6e9f1bb1b5138f2e7faaf","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:30:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f021af4e37427d34bf5768a9ce427b62981e3faa","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:30:47.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"c7d5d2e0a18c2dcf1e3563485dd00e27e013d187","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:31:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4d11ba00ef7e6d73dde12f6d082ccebdd243e75b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-31 12:34:56.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"8e508e0d10a59f0d5aecddcfa0ac2d2df5a2c2c9","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:38:00.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"f02b3d09a6c77a586feee3fab0053cc1c1afc531","author":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"date":"2020-03-31 12:38:06.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"0586bf2fc9c1e21362d7921c3c91d76a8ede2973","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-03-31 12:39:20.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"2e670f13beb8d0430bbd694815ac25ef7e74e263","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-31 14:26:25.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-macvtap : FAILURE\n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-direct : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ea68735d6275c4aa4d490bf39c5aa6326605f257","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 15:00:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e712a876359044d4bec51c868d3eedc5 : SUCCESS in 1h 06m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a15c32f234724a0cbb67e52f59372c92 : SUCCESS in 1h 33m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/887e578d15504f4684125a599cf7a2c2 : FAILURE in 15m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/afffbf4f01ef4aaab8aeb3d3175c235e : FAILURE in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec0727b95134467b93ea5e22cccfe13 : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12674cd855164200b6a061d9f50a09df : FAILURE in 14m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33dbb12b4d9f4bc2a163ac96096ca2f7 : FAILURE in 16m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41f5d45072864b11839e7bf0866eeae8 : FAILURE in 16m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d00e62537d9e47de86bae5051797a994 : SUCCESS in 15m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d269769b36149a18c2a3bb49c7d2157 : SUCCESS in 59m 52s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3dc7376dbc734001ab4bc8c83407eadf : SUCCESS in 1h 00m 29s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/886951a513014f9da63f6ec53bf42f7b : SUCCESS in 1h 07m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/23b53bdda0834b2da7450faa48c03c3a : SUCCESS in 1h 15m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/caf2872940824b21bad42a964b051a32 : SUCCESS in 53m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b35719c9b80e4e248b2d52efbde147dc : SUCCESS in 1h 36m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/7222153c050d48b0beea9aa3fe1225d8 : SUCCESS in 1h 41m 16s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e6ea693f090b4263b38370e46ca90c83 : SUCCESS in 17m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f672eea4f644aaf93b74f8aa62088a6 : SUCCESS in 1h 08m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6976ab3cc8bb4d98b3be2da510134f3b : SUCCESS in 46m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cdd7c70ee05ce947b4d7906c0966f7d641d3592c","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-03-31 18:24:39.000000000","message":"Patch Set 3:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42559 : SUCCESS in 2h 13m 40s","accounts_in_message":[],"_revision_number":3},{"id":"7ff6dff8f243c2744b79317668d9075d327cc318","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-03-31 20:44:19.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/716275/3 : FAILURE in 3h 03m 20s","accounts_in_message":[],"_revision_number":3},{"id":"eb17c54194e8ea0786793d29994a571efdd4fc14","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-03-31 22:10:46.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"df9755ef9435d09fc6a43316966765e47194e264","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-01 00:24:53.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/716275/3/check/tempest-dsvm-full-bionic/dea64a5/ : SUCCESS in 2h 03m 01s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/716275/3/check/tempest-dsvm-full-bionic-py3/2464ef2/ : SUCCESS in 2h 17m 12s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/716275/3/check/grenade-dsvm-bionic/ecea928/ : SUCCESS in 3h 07m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"de2d866eb2823355d75aed30e07dedb714e7e349","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-01 02:28:15.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"9f99c83cba8c7edd2d6a81df3b2410cd7a9ac31d","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-04-01 07:54:19.000000000","message":"Patch Set 3:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/716275/3/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/2c2d9b9 : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":3},{"id":"fc67bcb5b6a5e2147a4793441785d7e24ea0d484","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-01 09:01:06.000000000","message":"Patch Set 3:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/716275/3/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"ce68d8cc1678c15d5bb9dda09c7da1c8cb62d13a","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-01 18:36:42.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/75/716275/3/check-vote/ext-nova-zuul/dbaee7c : FAILURE in 1h 29m 56s","accounts_in_message":[],"_revision_number":3},{"id":"00df44e80f1b479c258fbf223cf20bf62b656b41","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-04-13 08:05:48.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6696c1b01ec40dec1b97298646978f8cd6e4684f","revisions":{"e28007d6d9487d864888b0c9f8e989419aac30ff":{"kind":"REWORK","_number":1,"created":"2020-03-31 12:29:21.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/75/716275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/716275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/716275/1"}}},"commit":{"parents":[{"commit":"eb6bd04e4c27c70b5239dbe7c17607b37f4e87dd","subject":"Merge \"Enable and use COMPUTE_ACCELERATORS trait.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb6bd04e4c27c70b5239dbe7c17607b37f4e87dd"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-31 10:42:10.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-31 12:28:38.000000000","tz":0},"subject":"[WIP] Respect multiple segments in network","message":"[WIP] Respect multiple segments in network\n\nEach network could have multiple segments defined, but\nall segments within the network should be using the same\nnetwork type.\nProvider networks, like vlan or flat, could have multiple segments\ndefined and different configuration of \u0027provider:physical_network\u0027,\nthat means different physical_network. We should follow those\ndifferences.\n\nIn addition tunneled type networks (geneve, vxlan) could also have\nmultiple segments.\n\nThis patch modifies the behavior to respect multiple segments\nfor all types of networks.\n\nChange-Id: Ibc939bcec7cf911ec33c159c4a8ecd6ac587a544\nRelated-Bug: 1869877\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e28007d6d9487d864888b0c9f8e989419aac30ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e28007d6d9487d864888b0c9f8e989419aac30ff"}]},"branch":"refs/heads/master"},"f7640df162afa7e62edc3b4e85509e0b8c4b63c5":{"kind":"REWORK","_number":2,"created":"2020-03-31 12:30:36.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/75/716275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/716275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/716275/2"}}},"commit":{"parents":[{"commit":"eb6bd04e4c27c70b5239dbe7c17607b37f4e87dd","subject":"Merge \"Enable and use COMPUTE_ACCELERATORS trait.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eb6bd04e4c27c70b5239dbe7c17607b37f4e87dd"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-31 10:42:10.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-31 12:29:57.000000000","tz":0},"subject":"[WIP] Respect multiple segments in network","message":"[WIP] Respect multiple segments in network\n\nEach network could have multiple segments defined, but\nall segments within the network should be using the same\nnetwork type.\nProvider networks, like vlan or flat, could have multiple segments\ndefined and different configuration of \u0027provider:physical_network\u0027,\nthat means different physical_network. We should follow those\ndifferences.\n\nIn addition tunneled type networks (geneve, vxlan) could also have\nmultiple segments.\n\nThis patch modifies the behavior to respect multiple segments\nfor all types of networks.\n\nChange-Id: Ibc939bcec7cf911ec33c159c4a8ecd6ac587a544\nRelated-Bug: 1869877\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7640df162afa7e62edc3b4e85509e0b8c4b63c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7640df162afa7e62edc3b4e85509e0b8c4b63c5"}]},"branch":"refs/heads/master"},"6696c1b01ec40dec1b97298646978f8cd6e4684f":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-03-31 12:38:00.000000000","uploader":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"},"ref":"refs/changes/75/716275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/716275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/716275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/716275/3"}}},"commit":{"parents":[{"commit":"5c119a21ea7013d7e7c06541b6256d0d6e70015c","subject":"Merge \"Add new default roles in os-agents policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5c119a21ea7013d7e7c06541b6256d0d6e70015c"}]}],"author":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-31 10:42:10.000000000","tz":0},"committer":{"name":"Maciej Józefczyk","email":"mjozefcz@redhat.com","date":"2020-03-31 12:38:00.000000000","tz":0},"subject":"[WIP] Respect multiple segments in network","message":"[WIP] Respect multiple segments in network\n\nEach network could have multiple segments defined, but\nall segments within the network should be using the same\nnetwork type.\nProvider networks, like vlan or flat, could have multiple segments\ndefined and different configuration of \u0027provider:physical_network\u0027,\nthat means different physical_network. We should follow those\ndifferences.\n\nIn addition tunneled type networks (geneve, vxlan) could also have\nmultiple segments.\n\nThis patch modifies the behavior to respect multiple segments\nfor all types of networks.\n\nChange-Id: Ibc939bcec7cf911ec33c159c4a8ecd6ac587a544\nRelated-Bug: 1869877\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6696c1b01ec40dec1b97298646978f8cd6e4684f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6696c1b01ec40dec1b97298646978f8cd6e4684f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":24791,"name":"Maciej Jozefczyk","email":"jeicam.pl@gmail.com","username":"maciej.jozefczyk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
