)]}'
{"id":"openstack%2Fnova~719163","triplet_id":"openstack%2Fnova~master~Ide8732be6c047ad1b141b89df676783b2fa2f25a","project":"openstack/nova","branch":"master","topic":"bug/1872385","hashtags":[],"change_id":"Ide8732be6c047ad1b141b89df676783b2fa2f25a","subject":"Delete resource provider in tree by top-down traversable order","status":"MERGED","created":"2020-04-11 10:29:01.000000000","updated":"2020-08-25 11:41:22.000000000","submitted":"2020-06-26 19:40:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":64,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719163-1593200440537-0147cf3e","meta_rev_id":"c9814cac5f0919fedbd730aefba71414b6bc6fff","_number":719163,"virtual_id_number":719163,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-06-23 08:50:13.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-23 16:00:33.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-23 09:48:21.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"date":"2020-06-23 12:34:48.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 19:40:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2020-06-23 12:44:56.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-23 16:26:08.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-23 14:18:30.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-06-23 11:49:37.000000000","_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-26 17:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-23 14:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-26 17:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"date":"2020-06-26 08:14:29.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 05:30:20.000000000","updated_by":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"reviewer":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"state":"REVIEWER"},{"updated":"2020-05-18 10:10:09.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-19 04:14:00.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-06-19 12:30:53.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-06-22 13:41:44.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-06-23 06:21:22.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-06-23 08:50:13.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 09:48:21.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 12:34:48.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-06-23 12:44:56.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 14:09:37.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-06-23 14:18:30.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-06-23 16:00:33.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-23 16:26:08.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-26 08:14:29.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2020-06-26 17:07:56.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-06-26 19:40:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ff2f5c1f8800a104975109bdd54d1c091de9ead","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-04-11 10:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9554257ebe14644be694f2394e6eb37572c88bca","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-11 11:17:02.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/1/check/pmem-tempest-plugin-filtered/e7478ff/ : SUCCESS in 21m 48s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/1/check/pmem-tempest-plugin-filtered-multi-node/8778866/ : SUCCESS in 46m 32s","accounts_in_message":[],"_revision_number":1},{"id":"5018f389782709d7c444e75bc632723407b5ffe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-11 12:09:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/62ef5df9ba1b488d8bb84413efce59ea : SUCCESS in 1h 12m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/87978b790d624b5894c51c3d841f034e : SUCCESS in 1h 10m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02c59282c79948c78057a4aca89cfdbb : SUCCESS in 18m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c03109ced1bc4328ac2aa81739743d64 : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ff660c2b614714bf5da57abc08c3a1 : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/610bbdc9d96f401ba1221827525022ab : SUCCESS in 11m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95a116e6620043419377969041af8873 : SUCCESS in 13m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99fdf6dcaffc4124bb4756b3d21cf424 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23df457afc54999aaf290ca62f015d8 : SUCCESS in 9m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4fe4a36decd445dd8027f04dc5071f60 : SUCCESS in 52m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d14caf80fd5f424f89c0d579c0eff222 : SUCCESS in 1h 05m 02s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b687139fdd5e4c7892e172e52c0adcf3 : SUCCESS in 1h 15m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ee6684de0bad4d7e86d8933f6072c650 : SUCCESS in 43m 32s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/add1d2f73bd94d2dbbefb9218387fe3a : SUCCESS in 1h 11m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/ab37700fc36d4ac1bd1fd5176c1dc503 : FAILURE in 1h 37m 56s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22a783abe15b48caabd2c7ffa62c4ed7 : SUCCESS in 17m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a49d4080a66f4e54ab541a26c9746d3f : SUCCESS in 59m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69e1609397a44d01a6f61016968c1d34 : SUCCESS in 53m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0c94f6ee47114a6f99a08c1de81b77b5 : SUCCESS in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b0967dd4394ab51359b3ed9be1393142d2982d76","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-11 12:51:08.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/1 : SUCCESS in 2h 21m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b958d6903a8525ae07c77a66eb9d5078487a9a4b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-11 17:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/1/check/tempest-dsvm-full-bionic/772fb36/ : SUCCESS in 1h 28m 00s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/1/check/tempest-dsvm-full-bionic-py3/12a910d/ : SUCCESS in 1h 30m 11s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/1/check/grenade-dsvm-bionic/e61e5c6/ : SUCCESS in 3h 07m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"10999e6c158f3a154f1aefc2a08123dc5ab58a2d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-13 01:36:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nThanks catch this.","accounts_in_message":[],"_revision_number":1},{"id":"16247930fcef731cac9c7e3fa857639c98271c4c","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-04-13 03:00:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"10599b361a5dbc5dd9136dd1048aebb6e11e847c","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-13 04:26:56.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/2/check/pmem-tempest-plugin-filtered/9e377f0/ : SUCCESS in 23m 28s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/2/check/pmem-tempest-plugin-filtered-multi-node/dd69397/ : SUCCESS in 42m 39s","accounts_in_message":[],"_revision_number":2},{"id":"324b744e240df36566f1a7d8d72832125aa337f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 04:49:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/13cf3788930c41329a095e8d5c2ee1c0 : SUCCESS in 1h 13m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6b9e3fb747594fc2b586d39ba0935609 : SUCCESS in 1h 22m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/003a5bb7c3a34f7689fc2d2afb6d067e : SUCCESS in 14m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4dc696a2537f4f0fb44cf08cbac25d4a : SUCCESS in 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a1c1b09f7ec498689569c76800d85f8 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cbb35af173841d2bd482ced4b3ce7d0 : SUCCESS in 11m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7b8fe3cc56f14d469cc586c20fdddd72 : SUCCESS in 12m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fb0ae931f2d4d79b129c43c4e509efe : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86cafc421d2b49fd86b4ae6da5d64d26 : FAILURE in 8m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c28d18c350fe4da9895e373be6848ffe : SUCCESS in 59m 03s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ec532216a27946bab11f2b313e1179b1 : SUCCESS in 1h 03m 13s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8ff5b20783144ee7b13cf8084579958d : SUCCESS in 1h 32m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bef0b2f3a5f24efbb7da30dc873f0158 : SUCCESS in 50m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e94698ff654c4f9b8318dab853d00674 : SUCCESS in 1h 34m 04s\n- nova-next https://zuul.opendev.org/t/openstack/build/19cabd4579254f7abbe3670081301857 : SUCCESS in 1h 30m 37s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f9f2f1e9edd466b9a5c08d64c030c0a : SUCCESS in 15m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/51c7695a3e4d40288c64e516a8b6bc37 : SUCCESS in 54m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f0a2621c7244889ac09bdb100b86755 : SUCCESS in 53m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ff242d817c6f4099a55d348f5dc80ef8 : SUCCESS in 33m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58e32105f76323354e39956874de72c09ff4fc9f","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-04-13 05:33:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0dc08eea9e6098d51a7fc723ff264b5a55327f75","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-13 05:34:25.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/2 : SUCCESS in 2h 32m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f0e73a500560712246ec90fe92f3361c67618985","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-13 06:18:54.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/2/check/pmem-tempest-plugin-filtered/e94facd/ : SUCCESS in 23m 13s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/2/check/pmem-tempest-plugin-filtered-multi-node/6852f55/ : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":2},{"id":"33657e5cae6a909558405121525c6e570c6fc2c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-13 06:30:52.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/2/check/tempest-dsvm-full-bionic/19eab48/ : SUCCESS in 1h 37m 47s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/2/check/tempest-dsvm-full-bionic-py3/27656f6/ : SUCCESS in 1h 41m 52s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/2/check/grenade-dsvm-bionic/2b16092/ : POST_FAILURE in 3h 07m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d3c31c2fa0ecde73d5e59ae7b2eb3db23bf6970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 07:52:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/81d1757e65bd4cea8c19ec564b178b54 : SUCCESS in 1h 11m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4aa7de8bc57c40c38a1275306b9c925e : SUCCESS in 1h 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ecd3dff8da64d67ab0816f9d071fd76 : SUCCESS in 15m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/48096895ff64421eb477415e31ad3368 : SUCCESS in 15m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7291cf080a6403f9ea01845c2e5bad8 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/385e9e65a9dc42e687a10cfbcdeb2590 : SUCCESS in 13m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e277c0d6d92e41e1b9570fe614b7bb3d : SUCCESS in 14m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26a498142d0b4137bd5b8056196476cf : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d6edf7d7024f9cad07161198da8cb9 : FAILURE in 8m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/65b8bcadb8e24dc1a292b959d2535006 : TIMED_OUT in 1h 34m 22s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9aa6cd65be1d41a2b5de6e7dcc081f43 : SUCCESS in 1h 56m 57s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0c26e1a3018142579089cf585b000776 : FAILURE in 1h 40m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a4161e628f234a5db66043434c0b6485 : SUCCESS in 1h 39m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1b47414c6c1c42f0a9b853ef1da022cb : SUCCESS in 1h 27m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/a4a7fd2609934d7c890b448208bde6b1 : FAILURE in 2h 15m 06s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7a0c12a35ed41f89cc70f50af0a683e : FAILURE in 25m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7910728e81144ea9a4a290eb6fa443e7 : SUCCESS in 1h 46m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf5e4ce5124c461596760bbe3246d03d : SUCCESS in 1h 00m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/89c763bc009343599d73f56e6a77ef31 : SUCCESS in 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a589d5e7be3a89250e8b268970857e2f33bf1c92","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-04-13 08:17:03.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ebcc152c14f2bf9ae522ac1a68a0d864550cf389","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-04-14 09:25:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b5c279205800a6233fe54d635fabdb26314fad49","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-14 10:24:59.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/2/check/pmem-tempest-plugin-filtered/fdd0626/ : SUCCESS in 27m 38s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/2/check/pmem-tempest-plugin-filtered-multi-node/821c56b/ : SUCCESS in 54m 23s","accounts_in_message":[],"_revision_number":2},{"id":"833bbd29d2fb3f480cf33411363b7e0560e7efac","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-14 11:47:30.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/2 : SUCCESS in 2h 20m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5b6bc4e7019ab565df3b7b112a42ce887628b42d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-14 13:29:49.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/2/check-vote/ext-nova-zuul/562233f : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f61baac9e1d32dd5ae06b7444a33c7c860792a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 17:58:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/1768f1d49b3045fe83bd49174ff7ea77 : SUCCESS in 1h 12m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7d855cc7eebd4fefa720747c02ecef56 : SUCCESS in 1h 32m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a5137cfb4524a598d59c8c56e65609d : SUCCESS in 17m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c435732c832b40faa44f2847404f938a : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d69f2effb8a4868a7a4098e4940e266 : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2add287edd9e443596d9ca419b9702ea : SUCCESS in 16m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22402f41e8b246449079560a9b2d8206 : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b87fc6590809479d89d89c9f737674c6 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d05d09ad071429084c6e4d51f5121f8 : SUCCESS in 14m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1b5ffae2db7b4758920f776df503a5cd : SUCCESS in 1h 05m 30s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bebc8bad543b4209b7c29c2f7e9cb98d : SUCCESS in 1h 03m 05s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/538220a14a5647b5a5b5ae55b63c3b9b : SUCCESS in 1h 34m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b7b5a8eea0874e6b9890339f2be2e19b : SUCCESS in 1h 03m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2e0c30a17eab47789aa3045c91469f53 : SUCCESS in 1h 38m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/d14ad26d21704fc191e180a224b0b8cb : SUCCESS in 1h 28m 19s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/89f46044367244ea8678d87bf18cdca3 : SUCCESS in 22m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c597ff045e274352a1aac1bee35ef79b : SUCCESS in 1h 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e9e534dc3d248c9abc25bd4e0353d26 : SUCCESS in 1h 06m 32s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fd11a32bcc75450390b54f40d82f8083 : SUCCESS in 41m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2ac4d3e048808c70d4bb46aa5c2e1689148d242c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-14 17:59:15.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/63/719163/2/check/EMC_VxFlexOS_NOVA/37801ee/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"ee4e73de22391a81d144ded13adebd445185402e","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-14 18:09:58.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-719163-2 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":2},{"id":"8439052016ec030573d8747315515742c8a5fa76","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-16 12:49:50.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6436d6781d7374489486882e1d405494547031a2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-17 09:15:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"131592508bfcac78a9b0cdf9018c0315295a5f24","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-17 09:20:35.000000000","message":"Patch Set 2:\n\n(1 comment)\n\n@gibi, do we need to call cyborclinet to clean the rp? But that need to consider the user whether deploy the Cyborg in their cloud.","accounts_in_message":[],"_revision_number":2},{"id":"2084469e8748130457bc36c33c7884480ef2a87a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-17 09:21:16.000000000","message":"Topic changed from bug/delete-sub-resource-provider to bug/1872385","accounts_in_message":[],"_revision_number":2},{"id":"ac0d49c42efc2a02961de31650445ef339361702","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-17 14:40:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a716d30d5ee5a4d9c463cb67471f5f0506a2ec9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-17 15:04:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a4739b0dfe08b0586701b8f407c2879e704b4bd8","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-13 02:27:52.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"9f4be968939bbf1f3710b8d972f2f44decaddf33","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 02:28:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2d22a5e4dc6161a78ecf0a1d6546c775efe3312e","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-13 02:59:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"41420ba54d1d80f5110556c927b0fd1a9f5d63de","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 02:59:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1b923d2bf3e7b1b278d4a188778c132494a8495d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-13 03:51:58.000000000","message":"Patch Set 4:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/4/check/pmem-tempest-plugin-filtered/871f763/ : SUCCESS in 25m 59s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/4/check/pmem-tempest-plugin-filtered-multi-node/401df2c/ : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":4},{"id":"cc9d7f26942c2f0a49b2c3fff51e69ff5449e186","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-13 04:10:44.000000000","message":"Patch Set 4:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42982 : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":4},{"id":"112dcc3320a1cf9a20c984bd5d8a59c9471fb7d6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-13 04:24:03.000000000","message":"Patch Set 4:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/9ec1281861f449cdb6aaedfc7e211821 : POST_FAILURE in 16m 33s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/cec2276609a649539bf630bc5e07c49a : POST_FAILURE in 13m 16s\n- grenade-dsvm-bionic finger://zuul-v3-executor.novalocal/74d774ddfa864d7396be24f7807a7740 : POST_FAILURE in 1h 06m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"116e8427527e350a9afee34898b7f1fdbe5ff4f0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 04:33:10.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/4/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/5e4521f : FAILURE in 1h 31m 46s","accounts_in_message":[],"_revision_number":4},{"id":"40dc1c47a82b820abb81d0ba24845a2d61d9d491","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-13 05:35:07.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/4 : FAILURE in 2h 34m 48s","accounts_in_message":[],"_revision_number":4},{"id":"c2ad1a7b05ff06926478348e7f507a4208f584a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 05:46:49.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a634b0fff8e549f8a649eee20c728909 : SUCCESS in 58m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9f4b72b3797d495bb9a60abe5f604441 : SUCCESS in 1h 19m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6443f4fa78944afb8bc703a53633f4ac : FAILURE in 15m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7569c506fbc74102b14d80db134c0d09 : FAILURE in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bba678c542940468704c99ab899c1b9 : FAILURE in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc2223b95f043d5b9004c2c5654c7c9 : FAILURE in 14m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feee6c22ce89426a98f2ec18835be7ab : FAILURE in 14m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2d784b2b69432e9913706e565e9aa8 : SUCCESS in 10m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5b1daaabddd942d080a4dcd59b4cf415 : SUCCESS in 54m 01s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/15d5de96e98143d98d4b0c7ec078df5e : SUCCESS in 58m 04s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4e1b951399404afc94f4e0d012c3989e : SUCCESS in 1h 10m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e6ffeba58c1244e89e8f53480dd476a6 : SUCCESS in 47m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a0b074d50d8f427b98ba799681dcbaf2 : SUCCESS in 1h 16m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/2300d2f0eb994e3e8c6894b18f20a50c : SUCCESS in 2h 10m 47s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a0bd82a0af924517ba5dfc927fe0d99a : FAILURE in 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/977d96cfe3794de4a4a531029323c422 : SUCCESS in 1h 03m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0434e8bdd3c14babbf3909b1de78005d : SUCCESS in 54m 58s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6487fbae89d240468237bb764790ce86 : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2d0ad50bf06882cd642937784c1042500bf36e42","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-13 06:12:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1bb38b00c35c275a5f58f078f3bfc3848eb1d763","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 06:12:46.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fe763c531826f41291294e4acd979e5b9be53179","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-13 07:53:45.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/5/check/pmem-tempest-plugin-filtered/ba85e44/ : SUCCESS in 29m 15s\n- pmem-tempest-plugin-filtered-multi-node pmem-tempest-plugin-filtered-multi-node : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":5},{"id":"86bd325509a9cdaca89489bef85c62204deb3d1a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-13 08:41:03.000000000","message":"Patch Set 5:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/c473af7da53245c8b80c32b9555f5a0e : POST_FAILURE in 16m 21s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/7c8c59432f9b483984446c3d0bbbd4bc : POST_FAILURE in 17m 01s\n- grenade-dsvm-bionic finger://zuul-v3-executor.novalocal/61fc19b9ef3742d3991759a1a11a1d16 : POST_FAILURE in 1h 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c78b41df3ca63d028be74e344040b534f72fff9e","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-13 09:01:19.000000000","message":"Patch Set 5:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42997 : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":5},{"id":"175edeb04fdf20ac78fa4d80fdf108e803637c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 09:12:13.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b23834d53ff64aca9b7e3020608914c8 : SUCCESS in 1h 07m 11s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8f088c2e44e8430786a9198289bbc841 : SUCCESS in 1h 25m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f46df259d0640f2a12422b7ff8dcfb3 : FAILURE in 19m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8e2589763b524241a41bc6d448c147e0 : FAILURE in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac4274766d0b45afb01a2f6ec5ac3d48 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d9b0a7cc829b494695a11fddda32bc72 : FAILURE in 13m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1450b767348144d7bba46f16b421ac82 : FAILURE in 14m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a965c81bce75466396107c5af936202c : SUCCESS in 13m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5d33dd98eed24ba48a4d041028da5e94 : SUCCESS in 55m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/09e3268f2b1a41038050dcc16cc120f3 : SUCCESS in 1h 10m 28s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d23e9b116fc5437db9638bf3eaf5c450 : SUCCESS in 1h 22m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dbd9e9a0123b486589a56641cbb16c63 : SUCCESS in 52m 40s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/aab598cecc764fb8969dfbda8ad792cc : FAILURE in 1h 16m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/746d6ae0e3ef4a5b89f626679f9f233a : SUCCESS in 1h 44m 25s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/caad3ccb67a3420295961b94d7d57cee : FAILURE in 15m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3e239ae032e426f9c322a6bafd67c42 : SUCCESS in 1h 04m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ecf419cb59f4feb9792c21e4f3155b9 : SUCCESS in 51m 06s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6e37e58a532e4210b66ad904b4f03ec3 : SUCCESS in 30m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"32c457aa4a5907cb8ea10c869a4cc00c051624e7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-13 10:33:40.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/5 : FAILURE in 2h 38m 21s","accounts_in_message":[],"_revision_number":5},{"id":"9e20bb7760686ea2d370895ae16ddae9722c1da0","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-13 11:55:34.000000000","message":"Patch Set 5:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/5/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/4a6e320 : FAILURE in 1h 39m 03s","accounts_in_message":[],"_revision_number":5},{"id":"ed09815211357bad5fff25bd50623994e8cc852b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-14 00:46:44.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bfeb695389bf199a80b71809344a5317c74a0e1a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-14 00:47:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"99766d3580b01f687c1394d3780142a601ad73af","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-14 01:33:34.000000000","message":"Patch Set 6:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/6/check/pmem-tempest-plugin-filtered/09c0443/ : SUCCESS in 23m 32s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/6/check/pmem-tempest-plugin-filtered-multi-node/9cd98fe/ : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":6},{"id":"1be97fb81a496429d8761f55ce2aeb4fb47810a8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-14 01:47:55.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43027 : SUCCESS in 1h 00m 35s","accounts_in_message":[],"_revision_number":6},{"id":"8b998757c899099efe3d23802c80f5e98d839050","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-14 02:14:03.000000000","message":"Patch Set 6:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/6/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/a5fceaa : SUCCESS in 1h 26m 44s","accounts_in_message":[],"_revision_number":6},{"id":"a3a8d07078bd057abb830b46ff2b0d19d92b203b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 02:18:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c5820ff3abb4bceb01a3a6297c1fd72 : SUCCESS in 1h 04m 47s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/848b64e63dec479f9ef930bcd3008ab3 : SUCCESS in 1h 18m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e7bd4dfdd3945f6a1119a94c0fdb529 : SUCCESS in 19m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cb28593bf8e40048561ae017e9cc95a : SUCCESS in 15m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8a8bdcfd1646568b63c472f89c3713 : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb960dfdbcfc43bab69a237215d8cac6 : SUCCESS in 12m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af68a7be8fd1475fbf57bf90db9c46c7 : SUCCESS in 12m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cee37a6e92d94e359a9997abd9831d10 : SUCCESS in 10m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e418e3a673a741ca8afd2a8914d4f742 : SUCCESS in 54m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8dd424ea65194a20bb4cd5676ce1bc92 : SUCCESS in 1h 05m 06s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f92fab12609845aa9a2b7062b0cbee3b : SUCCESS in 1h 09m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9a63f4c4639d40c693c0466885ba529c : SUCCESS in 50m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7e57c17155fc4e49b4474b44a4a97c2f : SUCCESS in 1h 21m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/81c53c8e0c854e0db292d738e1781061 : SUCCESS in 1h 26m 11s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f0271d05ae214d6fa4db1e68fd825d38 : FAILURE in 18m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/123cbf19bd5b4218bafca3e04fe9fb4b : SUCCESS in 59m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46add2f1e1f64cfe9fccb94834e81a69 : SUCCESS in 57m 35s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e91503d1536247ea888e85763ddbcbd2 : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e7a1bbf46f7981439fd1b9ae10eaeccfdb7b3fdf","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-14 02:37:35.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"d6b5e3f86fa285eb3c40ac252202787ed80c8f35","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-14 02:38:07.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"25ceba260e93cedfdfd5e00b419c8d35fc7329fa","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-14 03:06:17.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/6 : SUCCESS in 2h 18m 26s","accounts_in_message":[],"_revision_number":6},{"id":"80c35aa30f7c47889b00996ec1e2e0acba78ca8a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-14 03:15:13.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/6/check/tempest-dsvm-full-bionic/4754142/ : SUCCESS in 1h 32m 02s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/6/check/tempest-dsvm-full-bionic-py3/8e8e4d6/ : SUCCESS in 1h 14m 13s\n- grenade-dsvm-bionic finger://zuul-v3-executor.novalocal/d6f158421fcb49c58d0a470c3cc8c842 : POST_FAILURE in 1h 07m 55s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"de6cfff15d30c0b7d91837d8af68ffcaa9ce2fca","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-14 03:29:27.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/6/check/pmem-tempest-plugin-filtered/51a661d/ : SUCCESS in 26m 50s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/6/check/pmem-tempest-plugin-filtered-multi-node/5c04bd1/ : FAILURE in 50m 32s","accounts_in_message":[],"_revision_number":6},{"id":"618dc66dc3dd23a58626e65e9d368072b2b668cd","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-14 03:51:01.000000000","message":"Patch Set 6:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43030 : FAILURE in 1h 12m 57s","accounts_in_message":[],"_revision_number":6},{"id":"3973793dfefb72bd2d56bb960466e4eda8d7e4e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 05:02:14.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/21d0a157a5de47c98f0347b4cf19f1d5 : SUCCESS in 1h 04m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8cf265fa5fe14fd5b95a05f0d2dc6d5e : SUCCESS in 1h 21m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d04ab155c404642ab2453bbfdebd701 : SUCCESS in 22m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/285d02667faa48f4af93bf920af237f4 : FAILURE in 36m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d82b8e64c6234f95a274a8e20fb64269 : SUCCESS in 16m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ea5d34a5a474aeba5c8241549ab818b : SUCCESS in 39m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c47fa9ccb863470ababb1be1d705e439 : FAILURE in 39m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c37529f4856b4cc984b0c8898f25c65e : FAILURE in 27m 09s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/63a245394dd84a738a699d5a310221a9 : TIMED_OUT in 1h 34m 06s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3108707b47ac4a2a8aebbc34de40d328 : SUCCESS in 1h 50m 43s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6f409baec9f44e0f83e979557afb1b3c : SUCCESS in 2h 02m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3654d3f7934646beb1fff9546f381752 : SUCCESS in 1h 36m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/177939275406428388595dfd94c26f3a : SUCCESS in 1h 37m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/cb58a43c10804942a369ea704f4a329f : FAILURE in 2h 21m 42s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0e7040af0c54ae19442f2acb76321ca : FAILURE in 22m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d510318ef3f740619c4ea9d1af3db8a2 : SUCCESS in 1h 08m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f01fa771f2f043f08a32259d9c98af6e : SUCCESS in 1h 29m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c036e94f9daf48ec8d189564bfc5336f : SUCCESS in 38m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ea1892d617f45c874e6d978bc1724a2ebe51b077","author":{"_account_id":5754,"name":"Alex Xu","email":"hejie.xu@intel.com","username":"xuhj"},"date":"2020-05-14 05:30:20.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6be1fb46e920171887d8186c5820e7a7d7271475","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-15 22:28:58.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- ext-nova-zuul : EXCEPTION","accounts_in_message":[],"_revision_number":6},{"id":"fb1cc7021e5dca921f9581e9ac0e8f48435c0762","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-18 04:36:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ac3bf075365422aeb8bb24cf6879f45dae528d20","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 04:37:21.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9a5194998f0b464ac2d59083363e76731c0e4127","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 04:48:14.000000000","message":"Patch Set 7:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/7/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/8a1da6b : FAILURE in 10m 53s","accounts_in_message":[],"_revision_number":7},{"id":"54137aa7296d1b19998bd494abb584429ef00602","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-18 05:02:30.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/7 : FAILURE in 24m 46s","accounts_in_message":[],"_revision_number":7},{"id":"b5688b24790fc779338124e7c06aac9b3831186f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-18 05:20:12.000000000","message":"Patch Set 7:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/7/check/pmem-tempest-plugin-filtered/63f93cb/ : SUCCESS in 23m 00s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/7/check/pmem-tempest-plugin-filtered-multi-node/abc6a13/ : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":7},{"id":"c1d10cc442aee9259ae1f3192487f541a63557e6","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-18 05:52:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2b4aadbdbdf388ba112baa2746dad0bc7064a3bd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-18 06:11:00.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/7/check/tempest-dsvm-full-bionic/28aae40/ : FAILURE in 13m 16s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/7/check/tempest-dsvm-full-bionic-py3/8dd1f7d/ : FAILURE in 12m 23s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/7/check/grenade-dsvm-bionic/50f5c55/ : POST_FAILURE in 1h 30m 30s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8c425637c1aa059ac9a597b3715583f7bc17f6ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:40:53.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e34aceac2a054f29ba8944d4c80c6ba2 : SUCCESS in 53m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/050541df6e474b60892a1bec33f360e5 : SUCCESS in 1h 12m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73d9f718765e4d299834b5b9e010e508 : FAILURE in 15m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce1294a6b82f45728047909f3e847d40 : FAILURE in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/244c062c6a6645539603954bd1dfd0bc : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e8894a5c2e14a5d91769f6a97209f52 : FAILURE in 14m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00ae86a622d9446c91d7e0f2f178519e : FAILURE in 17m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b74dcd939344cb68a4228b6dbc6afa8 : SUCCESS in 10m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/211cce51e7234bd6897330382787e31d : SUCCESS in 1h 04m 34s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/beca58564c2b4735b0d88771f9dc7bff : SUCCESS in 57m 53s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/30122e3c8e86477abbdc7d7aa6c4d320 : SUCCESS in 1h 36m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c4386e91312b4b59866f7e4a21e46bdc : SUCCESS in 54m 41s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ee015935767d43eb99e8bf4a0681f17d : SUCCESS in 1h 46m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/78cddd807e10462eb54a80fcd2fb6636 : SUCCESS in 2h 02m 04s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1c21d526d09142da8e82a9b63a467930 : FAILURE in 24m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1d957307f8c4265b308174a46256db1 : SUCCESS in 55m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70c85b440aa64d2caac9cfcf2b200d2c : SUCCESS in 56m 20s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2745fc3509484431b73a865693249a88 : SUCCESS in 26m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a94098a44a2f24fe41f4760009dcbd2bed215cad","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-18 07:19:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"67742ccc0206f5cabeb0755206369e755ac58a0a","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 07:20:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7322c20db365d1392e4be68da4166d669a8cd151","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 07:31:27.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/8/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/5792045 : FAILURE in 11m 03s","accounts_in_message":[],"_revision_number":8},{"id":"4ba49c9d65698a64eeb50d7943a3c6145a7852c1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-18 07:47:58.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/8 : FAILURE in 27m 13s","accounts_in_message":[],"_revision_number":8},{"id":"2a2808f479106c498ced2db152582f469384d510","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-18 08:48:01.000000000","message":"Patch Set 8:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/8/check/tempest-dsvm-full-bionic/1f52a96/ : FAILURE in 12m 31s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/8/check/tempest-dsvm-full-bionic-py3/f1e27c2/ : FAILURE in 14m 24s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/8/check/grenade-dsvm-bionic/9ac49fd/ : POST_FAILURE in 1h 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"451643b61a45a96dc4df15ac4486e11c45a6c65b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 08:53:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/434b8cd809184bf7bd299724ac3dd9a6 : SUCCESS in 1h 15m 01s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a6af0082ae8442d4b2388c4b314f1ec4 : SUCCESS in 1h 13m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/984da2f52a9f4fdba2f3c09914729b06 : SUCCESS in 17m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2715ff545e5f4f679bbd67d2a8a007e2 : SUCCESS in 16m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f96e16ce9ef46abb32d23061057b832 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed299657f4a423e8008a365a0b966c5 : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32850a826a4b4fd0b20baf4b63ffe6ac : SUCCESS in 15m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c467082a144a59a9787e96de10521e : SUCCESS in 10m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d94908c132eb43b3bf240966dc00d177 : SUCCESS in 49m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0922155729134a0c85f53994a2587dc0 : SUCCESS in 1h 14m 07s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/34cd2c78da5743fe88d0b01f4f5b64ff : FAILURE in 1h 00m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b490d7d3dac04c119ca9a4f0ee28c12a : FAILURE in 56m 39s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e9b7d6a905904d51b65b48a4ab4686ad : SUCCESS in 1h 31m 14s\n- nova-next https://zuul.opendev.org/t/openstack/build/00cb405fa222447b9738418bfe55b3d0 : SUCCESS in 1h 27m 54s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58c406bee1154c6ab3af43a303b3514f : FAILURE in 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2d7aae1db1d478fbf8ca06ab853e6be : SUCCESS in 1h 06m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33d088f4a31d4e619f5d2ef02ff8af24 : SUCCESS in 1h 01m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5db5c35bd8e84e98a5eed5028a7414a6 : SUCCESS in 29m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9083aa99822f29e06d7151154be217aa9ca3eb06","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-18 09:40:02.000000000","message":"Patch Set 8:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/8/check/pmem-tempest-plugin-filtered/a37e381/ : SUCCESS in 27m 21s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/8/check/pmem-tempest-plugin-filtered-multi-node/da47189/ : SUCCESS in 47m 28s","accounts_in_message":[],"_revision_number":8},{"id":"05ab3d2088a5e6e1a0d54f9f38d00c6068e5c8b5","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-18 10:09:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"f5ff807fb7ac394f20fceff749cced0660010a18","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 10:10:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"540b98e983c7b26c833e5ee764211f2ed6ce34d9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 10:21:03.000000000","message":"Patch Set 9:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/9/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/8df2a93 : FAILURE in 10m 52s","accounts_in_message":[],"_revision_number":9},{"id":"a8b7b18ecd2b4951ad635610ca8c6886df8f9970","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:54:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ecab44a093e14018a17e24fd169be99c : SUCCESS in 1h 00m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/58bd08bcc9e94df797ee3781f81c0ca4 : SUCCESS in 1h 29m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1615aa8d4d1a4aebba25a51affa7b7f4 : SUCCESS in 16m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37d07cc6421d4483a1c6c6c255a91b28 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560d2b13259c4d3d85edcaa26c101ddc : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c85e945fb5047f78af70dda238303c9 : SUCCESS in 13m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a511352c0894f0a94a8acf9c9263b20 : SUCCESS in 13m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2846a1d0fcf54bc8879a6d4f92f6034b : SUCCESS in 10m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d211a2827664424792cca9be7164b9cd : SUCCESS in 48m 18s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/f9bd59c54fdc48f1a1a87cf571924150 : SUCCESS in 1h 06m 31s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/47121c3fe6994f5e853d06ca90f38910 : SUCCESS in 1h 23m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bd6f055ce3f64419a7e89e4cb2e1368f : SUCCESS in 1h 06m 58s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ad204a054cd44692b4974e5d1104a1cf : SUCCESS in 1h 21m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/a6de2534aa084ae5b89bf49728af54f2 : SUCCESS in 1h 38m 42s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b41e9900965840af8bce35608515bc39 : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15078eeb8b904ccdae901020643bb6af : SUCCESS in 1h 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56a34ad010d54d2980620e103488db40 : SUCCESS in 58m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9c5f8244ba9f4e0dbcb72c87aad3d87d : SUCCESS in 29m 30s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"6cf79103c1428006d31913a7467e7e453660e564","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-05-18 11:55:32.000000000","message":"Patch Set 9:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/63/719163/9/check/EMC_VxFlexOS_NOVA/013289f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":9},{"id":"5bb5b5d14213b56a371258be510a2f325b8fd6f4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-18 12:08:02.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/9/check/tempest-dsvm-full-bionic/7e49405/ : FAILURE in 20m 12s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/9/check/tempest-dsvm-full-bionic-py3/5f6a5d6/ : FAILURE in 17m 49s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/9/check/grenade-dsvm-bionic/e0d3d18/ : POST_FAILURE in 1h 24m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a7b61ab1b1814819a160f6c7628f7dbf46dbb770","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-18 12:31:11.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/9 : SUCCESS in 2h 20m 34s","accounts_in_message":[],"_revision_number":9},{"id":"86ede203cd3819398b34890260df246f4e182b8f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-05-18 13:01:19.000000000","message":"Patch Set 9:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-719163-9 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":9},{"id":"71c6ca53e11541d7577cdf14325a74d40581c80f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-18 18:11:45.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/63/719163/9/check/pmem-tempest-plugin-filtered/14be301/ : RETRY_LIMIT in 2m 00s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/9/check/pmem-tempest-plugin-filtered-multi-node/eda0dc1/ : RETRY_LIMIT in 2m 04s","accounts_in_message":[],"_revision_number":9},{"id":"5f1af296b40d8b11f4a471d47eaa9739aba5bc62","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-20 15:18:00.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43135 : FAILURE in 2h 53m 12s","accounts_in_message":[],"_revision_number":9},{"id":"8da8af2f8592f7986d65e0ddb39614c830d1bd3e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-12 11:04:06.000000000","message":"Patch Set 9: Code-Review-1\n\n(18 comments)\n\nI have couple of test related comments inline.","accounts_in_message":[],"_revision_number":9},{"id":"8bd1995946c23971a5b67711714a1856e0f5e8a1","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-17 09:35:34.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"417c134558ee6467d46dbe8cae6bd16596653baf","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-17 09:54:16.000000000","message":"Patch Set 9:\n\n(15 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c8aed9d8e3ab8c2f7a6f700cd1cdbc5f7e5c1722","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-17 10:33:10.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/10/check-vote/ext-nova-zuul/7c4d4ae : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":10},{"id":"0f834b628962b7250a45694870f3f1c0043c1f16","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-17 10:58:17.000000000","message":"Patch Set 10:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/10/check/pmem-tempest-plugin-filtered-multi-node/b93aed3/ : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":10},{"id":"f449e95c5b20ab29a8f500a3723a080bc0ca5f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 11:24:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3650c0bc7829455187cc02d0734d0d0b : SUCCESS in 1h 06m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/324fbf4656c3422dac34387b9a86d3e2 : SUCCESS in 1h 12m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/991d729fa0ba474997a79bf452fe25e7 : SUCCESS in 21m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6752809d2f314980a3f0d72f0c0d13e2 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b07baaa95e3d4dbe9d90a22162bf4444 : SUCCESS in 16m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/919364c398da4d03aa03f2382179e497 : SUCCESS in 17m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/911581eae41d49909791b82e009dab02 : SUCCESS in 16m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ca0f678ce84aa7a7703b77dd0c002b : SUCCESS in 12m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9f862ade8a614ba5bed962b3ba487099 : SUCCESS in 56m 41s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0f70708bd4b54589a39b9b1839968fe3 : SUCCESS in 1h 00m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/46c8f59bf09149429cb91d22e7840534 : SUCCESS in 1h 12m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/92480a4fc7d44fb1a767f6a476379c37 : SUCCESS in 38m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ffeb2e68fe0a4287955a68a4d1f2c1e0 : SUCCESS in 1h 27m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/3893b958ba5c4d198d93b1973e14545d : SUCCESS in 1h 42m 51s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/46e8b544d0ff4bca85346a0f41fbe8fd : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27514124a35c44b08250c8dad02c55b7 : SUCCESS in 46m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25bb4af973234a00a18843565c0a1645 : SUCCESS in 1h 08m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6d2a6fb7fdeb4d219c379e3d2facb42e : SUCCESS in 29m 25s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/2401d013679e422884706d8df377ae31 : SUCCESS in 50m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d3dc350cbc00ba48926cea833f7161f39187aefc","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-17 11:25:18.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/63/719163/10/check/EMC_VxFlexOS_NOVA/0739dd9/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":10},{"id":"8cde2adf6cb958b6e78031bfe48136847e3619d4","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-17 11:47:53.000000000","message":"Patch Set 10:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-719163-10 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":10},{"id":"72a5630645333f725f68af9866fc180a6a06e582","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-17 11:48:34.000000000","message":"Patch Set 10:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43518 : SUCCESS in 1h 56m 21s","accounts_in_message":[],"_revision_number":10},{"id":"b0fc6bb2645ebf926ac2f9779ac2424f1fc4c705","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-17 13:40:41.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/10 : SUCCESS in 2h 39m 41s","accounts_in_message":[],"_revision_number":10},{"id":"67c332ea387b7255b528e047da8e5b6c775898ac","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-17 13:58:02.000000000","message":"Patch Set 10:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/10/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/4ace10c : FAILURE in 1h 37m 41s","accounts_in_message":[],"_revision_number":10},{"id":"242e749dd44e09928749262ebc237e63c0f2f1f1","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-18 08:39:30.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"769f9d4012d8f5b41dfb5fcb34b1a20090757e3f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-18 09:26:12.000000000","message":"Patch Set 11:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/11/check/pmem-tempest-plugin-filtered-multi-node/9d5b0d4/ : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":11},{"id":"9bc530a88d1d5e9ed878ab626a352eaa3c643a55","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-18 09:35:02.000000000","message":"Patch Set 11:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43535 : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":11},{"id":"a9c1614a59d657a64441117d30985200904438c8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-18 09:39:39.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/11/check-vote/ext-nova-zuul/daf40c8 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":11},{"id":"7f89e1f783469dfd546fb2a72420a1e5f198e9df","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-18 09:54:39.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"a563cac9b2d9fbf2673562a0a21bd46d56c91bd3","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-18 10:40:11.000000000","message":"Patch Set 12:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/12/check/pmem-tempest-plugin-filtered-multi-node/ab58d63/ : SUCCESS in 43m 58s","accounts_in_message":[],"_revision_number":12},{"id":"a7862f4d81d26ec414ec32be9000a9c81a9a1445","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-18 10:53:34.000000000","message":"Patch Set 12:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/12/check-vote/ext-nova-zuul/605231c : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":12},{"id":"111d8f460e1a43ce94f4caded2f9395ac9ec36f8","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-18 10:56:03.000000000","message":"Patch Set 12:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43537 : SUCCESS in 1h 00m 43s","accounts_in_message":[],"_revision_number":12},{"id":"156d70f2e1de766a8a7ef4b8b5095afb54060c40","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-18 11:20:31.000000000","message":"Patch Set 12:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/12/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/ae8b87b : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":12},{"id":"43f6071964642fd3b563bdadf63145fdce4809f4","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-18 12:40:50.000000000","message":"Patch Set 12:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/12 : FAILURE in 2h 45m 07s","accounts_in_message":[],"_revision_number":12},{"id":"fda572204d2167941912cb87e9c07cefe7db13e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:43:40.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc2d6c7e2b0f481c92fbefbfeb35e194 : SUCCESS in 1h 03m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c7f2cbdcab79401aa42da82cfdb22032 : SUCCESS in 1h 31m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/200331350ba74174ace31e2223cfc21b : SUCCESS in 24m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d12205f58be746ba900523b4439953ff : SUCCESS in 22m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5dee61229be44f14b3d386ed8fd3c966 : SUCCESS in 11m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50f8537304994e1ca250c14796f513ca : SUCCESS in 18m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3822315339284cc3a25c45b1e3c4525d : SUCCESS in 15m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fb0c10a3424964b0b28ad2795a5941 : SUCCESS in 10m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/62c027c75e5a4ff78f6230f008a65ba6 : SUCCESS in 51m 42s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/89e0558bfc6046eb8a7a4e09b60a8807 : SUCCESS in 59m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/06941bf4b3554b489b36451991281e15 : SUCCESS in 55m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3fdddd09963243afb2f749c65036b98c : SUCCESS in 55m 52s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/96da9d8fd22a4ab385d95b7ea9bca1c0 : SUCCESS in 1h 26m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/3bb7ce0c7df54a929c440bfb21b58053 : SUCCESS in 1h 31m 21s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bfa863161eb94563b12892ebf090120e : FAILURE in 25m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9400d7c84f5d4b289f4d4dbaf63504fd : SUCCESS in 1h 06m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78d409dab68340edb606b102d29918ab : SUCCESS in 58m 18s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3c7b018d73cf43ceb0feb250e9204dc8 : SUCCESS in 45m 23s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/34b72dfd3c3a486d9c019d32c7d36def : SUCCESS in 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"708691cd2c7beee141de7090e239c2d7197d59c3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-18 12:53:21.000000000","message":"Patch Set 12:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/12/check/tempest-dsvm-full-bionic/cfc36cd/ : SUCCESS in 1h 26m 56s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/12/check/tempest-dsvm-full-bionic-py3/265b6dd/ : SUCCESS in 1h 25m 01s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/12/check/grenade-dsvm-bionic/b370f61/ : SUCCESS in 1h 55m 47s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"4a0d11d6a667e0fa81d95103b2c07d40b278ee5b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-19 00:29:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"be6854e61708a4baeac07e22156d17a929723590","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-19 01:15:08.000000000","message":"Patch Set 13:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/13/check/pmem-tempest-plugin-filtered-multi-node/71bd47a/ : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":13},{"id":"d30f1ecef7c899dfa6cb4f1b07df3f347b4e0828","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-19 01:31:13.000000000","message":"Patch Set 13:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/13/check-vote/ext-nova-zuul/7a65a57 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":13},{"id":"fa1b018c0e47156d1cdf4d9ffc7561b7525129d3","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-19 01:40:55.000000000","message":"Patch Set 13:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43546 : SUCCESS in 1h 10m 25s","accounts_in_message":[],"_revision_number":13},{"id":"66b1d10ddb43081cf01875f5a365e1397801c068","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-19 01:56:44.000000000","message":"Patch Set 13:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/13/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/88ee337 : SUCCESS in 1h 26m 16s","accounts_in_message":[],"_revision_number":13},{"id":"7042e8b8c7865183a5269fac2afb1cbf88cee2d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 02:27:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b6588d2a74d47e48acac3f31cf4f942 : SUCCESS in 1h 02m 35s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/caa0f634b1fb423985722ebdff594b9f : SUCCESS in 1h 55m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d7c0e727397475fa494f178ce6a7931 : SUCCESS in 12m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f5003a59adb43a2903f9e0ee2640d32 : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc05173e028c49569d381307100dd45c : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14d5f1fa5472476c92850c4c9cf085ea : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/828a0363c4b24bd28974297968aa5b2e : FAILURE in 38m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/989cf51760794377a68f216a7b7d1558 : SUCCESS in 29m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/203815e1032148f38d91993a80d980f0 : TIMED_OUT in 1h 34m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d2e62f365c8f4208ba1beed55b391a12 : SUCCESS in 1h 48m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b84c5d5093b94191a6767656c3620973 : SUCCESS in 1h 54m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0565f075d25e4d149281cdad15c8adca : SUCCESS in 53m 14s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a16ede60b71a469aa05980abd3e94879 : SUCCESS in 1h 24m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/ed11f78e784d4338b219b5869a378740 : SUCCESS in 1h 26m 00s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da05ba79296e4d52951dd8c232382e50 : SUCCESS in 34m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8e5977a06c44453be75bc6b470fa323 : SUCCESS in 36m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eca398b937514c7091dea120cc14f6c7 : SUCCESS in 38m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5e7ef093648a4e5fbf3c8cd69bf149fb : SUCCESS in 28m 15s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/06c6ca97f55449be895b9a2413683a9f : TIMED_OUT in 1h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"aedd02a4004c4ff30b6eec0b0f3988f67c7263a5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-19 02:47:37.000000000","message":"Patch Set 13:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/13 : FAILURE in 2h 16m 43s","accounts_in_message":[],"_revision_number":13},{"id":"9d8a3b1f9c45e63b77b1b050cb7c7cba85f0ede9","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-19 03:21:27.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"aac5b5fd7cc31805e5b7d236692b459fb69c7bf1","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-19 03:21:46.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"930e4aa297d215d2f9c5b5a5c4a97cfcc6395643","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-19 04:08:26.000000000","message":"Patch Set 14:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/14/check/pmem-tempest-plugin-filtered-multi-node/c887898/ : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":14},{"id":"6fdfdeb4c845519ed47da0176ababe9632881881","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-19 04:21:37.000000000","message":"Patch Set 14:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43547 : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":14},{"id":"2447ab8507756726fdc57b4c5dfdc15d9a1c4bae","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-19 04:25:45.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/14/check-vote/ext-nova-zuul/8d85de4 : SUCCESS in 1h 02m 42s","accounts_in_message":[],"_revision_number":14},{"id":"8cd1fd1ec201a54893ea9dae744b13b974f9e1f9","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-19 04:46:26.000000000","message":"Patch Set 14:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/14/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/faa2bd9 : SUCCESS in 1h 24m 15s","accounts_in_message":[],"_revision_number":14},{"id":"028942432747d32cc3b81a6f9332c95ce9a8633c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 05:05:23.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/038155fc45504d639c00908ed80a861e : SUCCESS in 1h 04m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7308ea85440544e6b4b4665bcbe93449 : SUCCESS in 1h 21m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99b977c99fb345b6a66addd3d17a3dfb : SUCCESS in 17m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7b175534388470e9cc95c7534e60290 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a6b762dbb34c10aa17ef12628cfed3 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34c42557862b414395a544d53e7177b5 : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53dbe5fd557241a79311eccda9adc6e8 : SUCCESS in 14m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b03f42db4a9a46b3bbda01c09694340f : SUCCESS in 12m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e481f16148f34ecdb8f212382b92e1fc : SUCCESS in 54m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e5569b73c5ab49a9b74b471419eff76a : SUCCESS in 1h 07m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2fd6ed7141f410db217ac3feb195186 : SUCCESS in 1h 32m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/073c481b29f944bf820d7f45bb2becd2 : SUCCESS in 1h 08m 05s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f3341d3e354b48d1867ddd4ef30def59 : SUCCESS in 1h 26m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/a56e98b086ab41cfb7c4d9c2455bc96e : SUCCESS in 1h 41m 24s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/09083a8eb3374d0bb4682ca7228f0500 : FAILURE in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b614222316c74a67810c53c8c3020108 : SUCCESS in 50m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18702df3b4644c3aab23a4060a914aac : SUCCESS in 48m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0f4ef6973eb848499f65a9fdb47c424a : SUCCESS in 32m 26s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/697268aef40e465cba282b7d73c7f667 : SUCCESS in 37m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"193d62202862a8342b9d36918ff86d482bc571ea","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-19 05:38:39.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/14 : SUCCESS in 2h 16m 12s","accounts_in_message":[],"_revision_number":14},{"id":"9642f7f22f124c83beec8e79cf80ff288f0cc35d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-19 06:29:15.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"2e36bc9f3cf6a9803eed55f9b8f7aa7396621361","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-19 06:51:39.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/15 : FAILURE in 21m 23s","accounts_in_message":[],"_revision_number":15},{"id":"17404e0605bdbf6c788c7e3f4b4c980b2326e689","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-19 07:16:58.000000000","message":"Patch Set 15:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/15/check/pmem-tempest-plugin-filtered-multi-node/e8ee698/ : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":15},{"id":"ffcaea91a6209b72cd48df8d2359667e1e4dd5ee","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-19 07:33:09.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/15/check-vote/ext-nova-zuul/7a2f5ec : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":15},{"id":"b7e2b93b5437968c0999b2d68530b18c318567db","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-19 07:33:35.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43549 : SUCCESS in 1h 03m 37s","accounts_in_message":[],"_revision_number":15},{"id":"93e47a3f4b2887dd7796b2d89eb5cde245d08570","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-19 07:55:14.000000000","message":"Patch Set 15:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/15/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/590d055 : FAILURE in 1h 25m 13s","accounts_in_message":[],"_revision_number":15},{"id":"582d0c7b7f2458a742bcb30d8a41bf3475de0bd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:05:19.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e84d5a2c01d2422c84f9b19c566d1c08 : SUCCESS in 1h 15m 17s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0bc616c463444c0bb64b9662f9526ff1 : SUCCESS in 1h 44m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9a7572c3d564534b17d471b4000df1c : SUCCESS in 23m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6c28d5e80354d26bf2f89734bf95b4e : SUCCESS in 18m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79c27577f4884592bb0963c531ab2757 : SUCCESS in 13m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b76d23fcdd74ef2b540405fbf65fbc6 : SUCCESS in 18m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c45025ab68d4b618e7ac4e591993a00 : SUCCESS in 19m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06410ad0303478bb6e33f9d7c1344ca : SUCCESS in 16m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/10b96de77bde438eb44b5e868b3805f2 : SUCCESS in 1h 00m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2c397dd6bc2d494b9fb173ff900cfa33 : SUCCESS in 1h 07m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/59010ed6e1a74534a9945ea92897ceb8 : SUCCESS in 1h 19m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d4e94dff157e431baa1c546fdfa5d84b : SUCCESS in 1h 03m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/636d66f13eec4802bf330350bdbf5602 : SUCCESS in 1h 24m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/e5338379f1fc4804b4dc0b95143ec9b8 : FAILURE in 1h 53m 11s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c655a925d1c4f68be123acbee6a29e3 : SUCCESS in 16m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/134870c86cdb44bea6aee6bc2fb6d23f : SUCCESS in 1h 02m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84f9c24a94204a86884788d653266de9 : SUCCESS in 1h 00m 01s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f63fe89785e14cbcb9d01fc5442293f3 : SUCCESS in 29m 51s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/83f03305da9b47bbb29f6c565c3f28a4 : SUCCESS in 40m 09s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"991f8a516df6855b84b5f155edf1e7ef2e3cd083","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-19 09:14:11.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"815df12c5c206675dbc9b05da8a692b66fe1e06e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-19 09:34:10.000000000","message":"Patch Set 15:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/15 : FAILURE in 19m 14s","accounts_in_message":[],"_revision_number":15},{"id":"e973878d27531f1bd32b679780bc60d8fc9e52a2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-19 09:34:36.000000000","message":"Patch Set 15:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/15/check/tempest-dsvm-full-bionic/c69df7f/ : SUCCESS in 1h 34m 21s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/15/check/tempest-dsvm-full-bionic-py3/f0b8ffc/ : SUCCESS in 1h 29m 19s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/15/check/grenade-dsvm-bionic/2817567/ : SUCCESS in 1h 54m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"1b7972eb23dfacbb357dd585a667b06bbdad7363","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-19 09:58:03.000000000","message":"Patch Set 15:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/15/check/pmem-tempest-plugin-filtered-multi-node/3a9f184/ : SUCCESS in 42m 46s","accounts_in_message":[],"_revision_number":15},{"id":"24ebbbef5c275c488d4da1613b277608b185f7ef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-19 10:11:36.000000000","message":"Patch Set 15:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/15/check-vote/ext-nova-zuul/3020789 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":15},{"id":"0a5c2dfcfeb9f295e0df2c7db320b73301fec044","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-19 10:13:19.000000000","message":"Patch Set 15:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43550 : SUCCESS in 58m 37s","accounts_in_message":[],"_revision_number":15},{"id":"10c5cd9f6763327c9ba974987d7458ddddc7fec8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:52:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a91f5e4ec8904646aa51fcbdf26dc491 : SUCCESS in 1h 04m 27s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dfe464ed2d2a4b54af8b545bfc73388f : SUCCESS in 1h 09m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59b89f2d0fdb4360a93abdb9cfac103e : SUCCESS in 18m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d2b6cd657f6475bae43aa95d9d38ef4 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290a433b10a2421696a51f5e058d10e6 : SUCCESS in 11m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0316c83f155647edabb11a0aa3bb24b1 : SUCCESS in 14m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b036d9d40c2b4e30b0427701795a8362 : SUCCESS in 15m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0ecc7d93964b52b9d9d3ca72ce781d : SUCCESS in 13m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/846e3e1ffbdb49968711360402674dd3 : SUCCESS in 1h 06m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8d7a27142dc84956a38cda1d752dc325 : SUCCESS in 1h 07m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/016faaa0813e446db4ddd230c55f8900 : SUCCESS in 1h 09m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f8f428b84d764104b04dcb55143275b3 : SUCCESS in 59m 34s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ad739b6ddc674c3084e31fe3a399c776 : SUCCESS in 1h 32m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/38449e522fc9450fb99ec8ead6c4940b : SUCCESS in 1h 27m 31s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fe6daa2b15574f3a85a1b79fcd40fb4c : SUCCESS in 22m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/825e3d2fb87c487fb96170dc28397e73 : SUCCESS in 49m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10a321a7e42145de946506347398aeb1 : SUCCESS in 37m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/52b49c97aed7495199249c4b9ccc4e80 : SUCCESS in 27m 54s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/af503b2cc49542bdbbdcd12aeaf03ad1 : SUCCESS in 29m 22s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2c793c70a497232f556de8f786779b9d391f3eed","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-19 11:15:47.000000000","message":"Patch Set 15:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-719163-15 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":15},{"id":"7e9340fe9b7d70dce964bf2dd5e0f47e32065aea","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-19 11:30:05.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"3be7835ecf8462cf9bc9c01f5f8c46ca492f103f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-19 11:47:37.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/16 : FAILURE in 16m 35s","accounts_in_message":[],"_revision_number":16},{"id":"c8e74a9864ea7d58c2a8b6b7625ee15a96696ea0","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-19 12:14:40.000000000","message":"Patch Set 16:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/16/check/pmem-tempest-plugin-filtered-multi-node/15bed48/ : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":16},{"id":"15501ca28ab430b81c7a9ebb97507b0342a3ace6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-19 12:26:59.000000000","message":"Patch Set 16:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/16/check-vote/ext-nova-zuul/519a602 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":16},{"id":"04e603ecbb8e8bfacd2ba3d6d7b0cd0f5a803d45","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-06-19 12:30:53.000000000","message":"Patch Set 16:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43553 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":16},{"id":"7fd25c70d4f849cc0cc14e3ad869e0155ced9e11","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-19 13:00:30.000000000","message":"Patch Set 16:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/16/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/011ccb5 : SUCCESS in 1h 29m 46s","accounts_in_message":[],"_revision_number":16},{"id":"bbe56b08afbb880cf729fb0d5224bd22c330ebcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:02:57.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/681e939f6bac46348a68d3f5f21e53a6 : SUCCESS in 44m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6afcfb4f7fed4150ad06b61c50dd8205 : SUCCESS in 53m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a38d4a5a9b134096b1494ad3687b107d : SUCCESS in 13m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25ca638c5bc94dfe851774d09e4242d0 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c49a824e82ca441da24c9157aff6a168 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3043c6cda7f4ff8b78691eb2d66bd36 : SUCCESS in 10m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef7c92345fab42448919cf1a41a73474 : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35a4f50cc13406a9a4f444962f21911 : SUCCESS in 8m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/63432d3160f54078a50e8872697901ce : SUCCESS in 40m 56s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/06aed94f58624730bfcdfeaee949ba71 : SUCCESS in 48m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e2870408b0654c3fb101682777adf6d6 : SUCCESS in 1h 15m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1a3bd1c61f0d4d519010055e467a4c1e : SUCCESS in 49m 51s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6fec20efd711485ea0183a3282b5c6d8 : SUCCESS in 1h 25m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/d705a1c0054c417494a4657883534055 : SUCCESS in 1h 26m 47s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cbc6edd9d8084962b5134727d0886f88 : SUCCESS in 21m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dab36235b3848f1ad5c267126d6b19d : SUCCESS in 57m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5f50a2f955ef4d6f8500d90cecff75da : SUCCESS in 1h 03m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/503ddc41c2414ba38a30a4f34f13d2fc : SUCCESS in 31m 33s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/57b26d90025743cf93fcc282816177b0 : SUCCESS in 36m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8f4171d69640e4f4f02d72cdda619bf376547993","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-19 13:25:51.000000000","message":"Patch Set 16:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-719163-16 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":16},{"id":"d75772b61cc9e1ec61894d23f11adc8da0c06fe8","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-19 14:03:53.000000000","message":"Patch Set 16:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/16/check/tempest-dsvm-full-bionic/48a98f8/ : SUCCESS in 1h 35m 32s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/16/check/tempest-dsvm-full-bionic-py3/6530753/ : SUCCESS in 1h 18m 40s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/16/check/grenade-dsvm-bionic/5bbb437/ : SUCCESS in 1h 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a9200cc66b7e54f8a5450d8ca2e4d4815115ace2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-22 12:32:58.000000000","message":"Patch Set 16: Code-Review-1\n\n(6 comments)\n\nThanks for the cleanup. This patch starts looking good. I still left some suggestions inline. After your fix those I think I will +2.\n\nCheers,\ngibi","accounts_in_message":[],"_revision_number":16},{"id":"0cc42e5764f3b7feeb096b74e1fb20d52f440c5d","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-06-22 13:41:44.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nThanks for tackling this. Gibi, let me know if you have a hard time getting a second +2 from active cores.","accounts_in_message":[],"_revision_number":16},{"id":"ec1def79207de376d7cf960d2ca90cfb163cd6ae","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-06-23 06:21:22.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"c6ef3478cf3101a501eee10316afbae3c188f0ef","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-23 07:07:52.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"e9f84e58a340dde63150a99462fdcbd2e9755ac7","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-23 07:25:36.000000000","message":"Patch Set 16:\n\n(8 comments)","accounts_in_message":[],"_revision_number":16},{"id":"195d3cb04dc52672afbc43977ddb61fa89bb5714","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-23 07:25:47.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"588e481a2373eb2e80c9d4f6b8cc4628ab359ad8","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-23 07:49:04.000000000","message":"Patch Set 18:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/18/check/pmem-tempest-plugin-filtered-multi-node/ba878c1/ : FAILURE in 21m 52s","accounts_in_message":[],"_revision_number":18},{"id":"86ed68322295f31d74c6b36ab720d0bf61e4bfa8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-23 08:24:30.000000000","message":"Patch Set 18:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/18/check-vote/ext-nova-zuul/e85a99c : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":18},{"id":"0d7a26af7b631d597e72e4cec14eb928bcff6caa","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-06-23 08:50:13.000000000","message":"Patch Set 18:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/719163/18/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/3cdb4d6 : FAILURE in 1h 23m 48s","accounts_in_message":[],"_revision_number":18},{"id":"21f77b8773f169ebcd93a5507033cd3add378d4e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-06-23 09:48:21.000000000","message":"Patch Set 18:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/18/check/tempest-dsvm-full-bionic/66c6a44/ : SUCCESS in 1h 32m 42s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/8fd1f1957bfc47f2aa15ec5a077f6638 : POST_FAILURE in 1h 28m 10s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/63/719163/18/check/grenade-dsvm-bionic/2b67107/ : SUCCESS in 1h 56m 51s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b8669924c97b1cb5fce6003b99a22007d7613cf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 10:06:23.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fdd8daab90dc421f880e361277306601 : SUCCESS in 1h 07m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/88cb1ded4fac4455a56e8ea0b522286f : SUCCESS in 1h 33m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e33559355ef34d9ea550de55ed36a12b : SUCCESS in 25m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b0d90a67ff6434993e2cda545f194ad : FAILURE in 36m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c96e83fbdf074909a5c6f63311c33ade : SUCCESS in 14m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f18a256f6a4b4733b814663e19ca6ebd : FAILURE in 34m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b5218eb0f454b76b23eb0794ccafc82 : SUCCESS in 38m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9610ecced86241c8be75aeb66ab90a59 : SUCCESS in 13m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/21c967ec34114675acd7060682a5d971 : SUCCESS in 1h 19m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ee636bfbce96479e9a0757815a860965 : SUCCESS in 59m 50s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6366097995c4fdb8414e2046ac42e9e : SUCCESS in 1h 38m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f536bb0991424698a2832b3c4635d1ef : SUCCESS in 50m 30s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2f7b120a637246cfa5c2716d09ac252c : FAILURE in 2h 34m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/3dd2ff8e2da746b2a15c1e6d4ccf77f0 : SUCCESS in 1h 19m 00s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16bdf6897b87487f985a8b8aeba58d18 : SUCCESS in 22m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c78e3fe630294786978acf24db75e528 : SUCCESS in 53m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41f8d7c784b94abfb4efb7e26efe3a57 : SUCCESS in 57m 17s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bf8c4836988e47cabf4616b533b98bd0 : SUCCESS in 23m 16s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/186259e00ecc40db9ed58ad22ad7ebbb : SUCCESS in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"8bc7d723110ea641077057882ac0f856f523b059","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-23 10:11:37.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/18 : SUCCESS in 2h 44m 33s","accounts_in_message":[],"_revision_number":18},{"id":"7b6ef6c281aed081bf57f8aac6af74d5aa07a458","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-23 10:16:07.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"e8e55b1fbf59694afd7cc93f610514c1cce70039","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-06-23 11:49:37.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"4e4d59db440472727e3d156927363f9ba40bcfae","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-06-23 12:34:48.000000000","message":"Patch Set 18:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/63/719163/18/check/pmem-tempest-plugin-filtered-multi-node/59897d2/ : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":18},{"id":"c477938d4debf2df2b99dbc43d76d2271e22cbae","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-06-23 12:44:56.000000000","message":"Patch Set 18:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/63/719163/18/check-vote/ext-nova-zuul/79b1a33 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":18},{"id":"2625f97a2d17fea4569528afb573b22909470595","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-23 12:51:07.000000000","message":"Patch Set 16:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/719163/16 : FAILURE in 2h 34m 03s","accounts_in_message":[],"_revision_number":16},{"id":"994a4846f45e0f60e2445338fd3aa0a9d3d3be4f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-23 14:09:37.000000000","message":"Patch Set 18: Code-Review+2\n\nLooks good to me. Thank!","accounts_in_message":[],"_revision_number":18},{"id":"ba1bcd685dd9022b6ded455779b6ed239f12584b","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-06-23 14:10:35.000000000","message":"Patch Set 18:\n\n\u003e (1 comment)\n \u003e \n \u003e Thanks for tackling this. Gibi, let me know if you have a hard time\n \u003e getting a second +2 from active cores.\n\n@Eric: Ack, I will let you know.","accounts_in_message":[],"_revision_number":18},{"id":"583ea2e1834ca6edfb1fb4f3046b4e0852cf5b71","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-06-23 14:18:30.000000000","message":"Patch Set 18:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/719163/18 : SUCCESS in 2h 28m 04s","accounts_in_message":[],"_revision_number":18},{"id":"3a412f3751c4a5b41a83c89d233483d4ea6063d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 15:59:55.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/937922e929524157a49fce02ac571c42 : SUCCESS in 1h 02m 07s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/73689d85ab0148f680f17ed199615ec4 : SUCCESS in 1h 32m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3fedffa3527405db366a43bd9fd642d : SUCCESS in 20m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a69754aa4a5a48a4ad13fa1b501db7c5 : SUCCESS in 20m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/513d41f6874f4bfb960c1f86766dc22f : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e0bad79ba6843549e2a9ac923a1206a : SUCCESS in 19m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f3cbc3d3e9a44e8b61afa835cf69caf : SUCCESS in 19m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff65879840664679abc3dc0ccecf1f34 : SUCCESS in 21m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7b903e58bcff412290a1aa09b34f05be : SUCCESS in 1h 03m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/95ae2773205f4bfc9ef9ad88a3d4389a : SUCCESS in 55m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/20d1c935ca6640d5a1bc08185674a1e3 : SUCCESS in 1h 10m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c03d41232a0c4f7ca7ac96a9e9220707 : SUCCESS in 47m 44s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1cd5c0f35a8b4879b60169a6a45b4ffd : SUCCESS in 1h 28m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/92690d50cbf140c787af5973c84c7c95 : SUCCESS in 1h 39m 19s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cef3c699cb542eb9b76acaa0e1737d9 : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cdbcbc3d57748bf88c9d1013ed134e3 : SUCCESS in 1h 00m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca91c31c92ae4710a020ea031c1e47f3 : SUCCESS in 51m 36s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fea1897d992b470792ccdd08732308de : SUCCESS in 29m 23s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/a1345049a20a4a889bf4d5ef253dd2eb : SUCCESS in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"552b8887d8b39b6d28ee00083de09b6769eb9636","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-06-23 16:00:33.000000000","message":"Patch Set 18:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/63/719163/18/check/EMC_VxFlexOS_NOVA/118942f/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":18},{"id":"47ec21053d3456364e2d802553707802d7a8f551","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-06-23 16:26:08.000000000","message":"Patch Set 18:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-63-719163-18 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":18},{"id":"b18f15d0372ec4a66cae61390deddb6a43750f8e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-06-26 17:07:56.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"d26264aaad1831526dff3f6de51b2e1a887caae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 17:10:24.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"5897e468182e2044dbc5590cdd5a3aa12ef2a9cc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:40:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"6899332ad2f2d7e7da8b36cb2a3829329b3a5027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:40:40.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/418887fcd90b433dbc83b385a7004989 : SUCCESS in 58m 52s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/44b451f4f9a241b19bef5e8ee94d7566 : SUCCESS in 1h 15m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/105d550f64e846329c38086f3e0777ca : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e47c2f8b4e43538bdbbafb0701e6c8 : SUCCESS in 8m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9acebb04f5f540f08e7115f7f1fe10fe : SUCCESS in 22m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96a61a77ebc2402cb60703abdd2f0f64 : SUCCESS in 21m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d1fe818b6ae4d37b1502969c3a33d98 : SUCCESS in 17m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a434cb5b9b0457ea54d107e23585a89 : SUCCESS in 1h 18m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6de01c32a0314bac8c2da5566e7a203b : SUCCESS in 44m 27s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/24050003a5c241e799f84d172660426e : SUCCESS in 21m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e52a9faa8e1a4be287cd6dc08c523a58 : SUCCESS in 1h 11m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/0eacfe12acbd4a66877bb8c1ee71ce13 : SUCCESS in 1h 18m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dce72f89573549a49a5c4a5e85ea201e : SUCCESS in 51m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d93be8d390b4644b9bf2716ca3a87eb : SUCCESS in 1h 09m 41s","accounts_in_message":[],"_revision_number":18},{"id":"7b29813b04f3da681c477dc79c4bd15fa8ecee1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 19:42:57.000000000","message":"Patch Set 18:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dc5d515061f42ec89477123e9e101be : FAILURE in 1m 34s","accounts_in_message":[],"_revision_number":18},{"id":"c9814cac5f0919fedbd730aefba71414b6bc6fff","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-25 11:41:22.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"a45daafd3111a6804eb5adafe4860db122cb09b8","revisions":{"4820a9395eca1fa3fb06d81b6460c0e7131116a8":{"kind":"REWORK","_number":1,"created":"2020-04-11 10:29:01.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/1"}}},"commit":{"parents":[{"commit":"a40c69412510aede53a3f4bdb050b54cf010a633","subject":"Merge \"api: Introduce microverion 2.87 allowing boot from volume rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a40c69412510aede53a3f4bdb050b54cf010a633"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4820a9395eca1fa3fb06d81b6460c0e7131116a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4820a9395eca1fa3fb06d81b6460c0e7131116a8"}]},"branch":"refs/heads/master"},"d71f0fd03170f27f9f060767d3824edec35334ca":{"kind":"REWORK","_number":2,"created":"2020-04-13 03:00:40.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/2"}}},"commit":{"parents":[{"commit":"a40c69412510aede53a3f4bdb050b54cf010a633","subject":"Merge \"api: Introduce microverion 2.87 allowing boot from volume rescue\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a40c69412510aede53a3f4bdb050b54cf010a633"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-13 02:59:39.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d71f0fd03170f27f9f060767d3824edec35334ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d71f0fd03170f27f9f060767d3824edec35334ca"}]},"branch":"refs/heads/master"},"6b86e8a5a5a1c906d6b6f973440a644c2221d1b4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-13 02:27:52.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/3"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-13 02:11:24.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b86e8a5a5a1c906d6b6f973440a644c2221d1b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6b86e8a5a5a1c906d6b6f973440a644c2221d1b4"}]},"branch":"refs/heads/master"},"21c91891c65911d986f0a9a62b5a89c4c58b4bc0":{"kind":"REWORK","_number":4,"created":"2020-05-13 02:59:23.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/4"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-13 02:55:10.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/21c91891c65911d986f0a9a62b5a89c4c58b4bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/21c91891c65911d986f0a9a62b5a89c4c58b4bc0"}]},"branch":"refs/heads/master"},"8f58fba15ff4045774b640e10b62dd6ce8ee4ccb":{"kind":"REWORK","_number":5,"created":"2020-05-13 06:12:16.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/5"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-13 06:05:34.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f58fba15ff4045774b640e10b62dd6ce8ee4ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8f58fba15ff4045774b640e10b62dd6ce8ee4ccb"}]},"branch":"refs/heads/master"},"85fa922e45299f72079e93077e39b7929fbcf9c8":{"kind":"REWORK","_number":6,"created":"2020-05-14 00:46:44.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/6"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-14 00:45:14.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/85fa922e45299f72079e93077e39b7929fbcf9c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/85fa922e45299f72079e93077e39b7929fbcf9c8"}]},"branch":"refs/heads/master"},"cb72e51facb4b7f59c98e55bf1e1e80a5298f635":{"kind":"REWORK","_number":7,"created":"2020-05-18 04:36:49.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/7"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-18 04:35:40.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb72e51facb4b7f59c98e55bf1e1e80a5298f635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cb72e51facb4b7f59c98e55bf1e1e80a5298f635"}]},"branch":"refs/heads/master"},"4224c916fcaf2a15e112b41876b09bfd6fb7ae2d":{"kind":"REWORK","_number":8,"created":"2020-05-18 07:19:50.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/8"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-18 07:17:42.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4224c916fcaf2a15e112b41876b09bfd6fb7ae2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4224c916fcaf2a15e112b41876b09bfd6fb7ae2d"}]},"branch":"refs/heads/master"},"5f9923be897bb41b0bd95fd41499cf9fc0ad997d":{"kind":"REWORK","_number":9,"created":"2020-05-18 10:09:36.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/9"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-05-18 10:04:57.000000000","tz":480},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f9923be897bb41b0bd95fd41499cf9fc0ad997d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5f9923be897bb41b0bd95fd41499cf9fc0ad997d"}]},"branch":"refs/heads/master"},"585af254256cffb0a13b9f4780820152161ef23e":{"kind":"REWORK","_number":10,"created":"2020-06-17 09:35:34.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/10"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-17 08:02:11.000000000","tz":0},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/585af254256cffb0a13b9f4780820152161ef23e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/585af254256cffb0a13b9f4780820152161ef23e"}]},"branch":"refs/heads/master"},"958c5b63169a9a9b69634bdcffd8109c90008876":{"kind":"REWORK","_number":11,"created":"2020-06-18 08:39:30.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/11"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-18 08:38:35.000000000","tz":0},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/958c5b63169a9a9b69634bdcffd8109c90008876"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/958c5b63169a9a9b69634bdcffd8109c90008876"}]},"branch":"refs/heads/master"},"af38720dc7b0ec422a0394df776b4af69fac7a87":{"kind":"REWORK","_number":12,"created":"2020-06-18 09:54:39.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/12"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-18 09:54:10.000000000","tz":0},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af38720dc7b0ec422a0394df776b4af69fac7a87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af38720dc7b0ec422a0394df776b4af69fac7a87"}]},"branch":"refs/heads/master"},"446f6e6048c602364825f672652274627a002da7":{"kind":"REWORK","_number":13,"created":"2020-06-19 00:29:52.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/13"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-19 00:29:22.000000000","tz":0},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/446f6e6048c602364825f672652274627a002da7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/446f6e6048c602364825f672652274627a002da7"}]},"branch":"refs/heads/master"},"01dc145dab27bb756f0c059fbe5ec811aa0fa899":{"kind":"REWORK","_number":14,"created":"2020-06-19 03:21:27.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/14"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-19 03:20:55.000000000","tz":0},"subject":"delete sub resource provider when delete resource provider","message":"delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01dc145dab27bb756f0c059fbe5ec811aa0fa899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/01dc145dab27bb756f0c059fbe5ec811aa0fa899"}]},"branch":"refs/heads/master"},"1994aabb12774bfb64de00715192b70436961771":{"kind":"REWORK","_number":15,"created":"2020-06-19 06:29:15.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/15"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-19 05:53:44.000000000","tz":0},"subject":"Delete sub resource provider when delete resource provider","message":"Delete sub resource provider when delete resource provider\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1994aabb12774bfb64de00715192b70436961771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1994aabb12774bfb64de00715192b70436961771"}]},"branch":"refs/heads/master"},"0ae2a01e61b27ac049b4bb32b71a708ea660f9a8":{"kind":"REWORK","_number":16,"created":"2020-06-19 11:30:05.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/16"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-19 11:28:41.000000000","tz":0},"subject":"Delete resource provider in tree by top-down traversable order","message":"Delete resource provider in tree by top-down traversable order\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ae2a01e61b27ac049b4bb32b71a708ea660f9a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0ae2a01e61b27ac049b4bb32b71a708ea660f9a8"}]},"branch":"refs/heads/master"},"7c90188f26baab07155a1f2dd01a92c1f8cef6e9":{"kind":"REWORK","_number":17,"created":"2020-06-23 07:07:52.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/17"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-23 06:56:28.000000000","tz":0},"subject":"Delete resource provider in tree by top-down traversable order","message":"Delete resource provider in tree by top-down traversable order\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c90188f26baab07155a1f2dd01a92c1f8cef6e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7c90188f26baab07155a1f2dd01a92c1f8cef6e9"}]},"branch":"refs/heads/master"},"a45daafd3111a6804eb5adafe4860db122cb09b8":{"kind":"REWORK","_number":18,"created":"2020-06-23 07:25:47.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/63/719163/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/63/719163/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/63/719163/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/63/719163/18"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-04-11 10:28:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-06-23 07:20:29.000000000","tz":0},"subject":"Delete resource provider in tree by top-down traversable order","message":"Delete resource provider in tree by top-down traversable order\n\nWhen delete nova service by \"nova service-delete \u003cid\u003e\", this will delete the\ncompute node\u0027s resource provider in Placement. But if there are child resource\nprovider, Placement will throw exception(CannotDeleteParentResourceProvider)\nand don\u0027t delete the resource provider. When we add the host service again,\nPlacement cannot insert new compute node resource provider, and we cannot use\nthe compute node any more. So we should delete sub resource provider in tree\nwhen delete resource provider.\n\nModify unit test.\n\nChange-Id: Ide8732be6c047ad1b141b89df676783b2fa2f25a\nCloses-Bug: #1872385\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a45daafd3111a6804eb5adafe4860db122cb09b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a45daafd3111a6804eb5adafe4860db122cb09b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
