)]}'
{"id":"openstack%2Fnova~720104","triplet_id":"openstack%2Fnova~master~Ibfc1917d35ab9e7657b13ec32582891f06d88176","project":"openstack/nova","branch":"master","topic":"bp/policy-defaults-refresh","hashtags":[],"change_id":"Ibfc1917d35ab9e7657b13ec32582891f06d88176","subject":"Add test coverage of existing remaining servers policies","status":"MERGED","created":"2020-04-15 00:14:35.000000000","updated":"2020-04-20 18:34:02.000000000","submitted":"2020-04-20 18:34:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"720104-1587407642194-a663f06a","meta_rev_id":"8f462fa9aeafe06c905248966998115d92ab3d06","_number":720104,"virtual_id_number":720104,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-04-18 19:06:39.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-18 19:30:12.000000000","_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-18 20:28:00.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-04-18 18:25:40.000000000","_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-18 18:58:44.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"date":"2020-04-20 16:58:52.000000000","_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-20 18:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-20 16:48:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":2,"date":"2020-04-20 15:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-20 16:50:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-15 00:15:30.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-18 18:25:40.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-18 18:58:44.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-04-18 19:06:39.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-18 19:30:12.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-18 20:28:00.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-04-20 15:19:20.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-04-20 16:50:04.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-04-20 18:34:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6698cc4de4dacd3b5962c38f02a8d361b28dad70","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-15 00:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f28a5087f103b839ba4e3dc48fb0c47a74058862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 00:15:04.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"881bc07c906061a04e1cfe8b3ec491a44fd82ab8","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-15 00:15:16.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"c2dc40ca0772c237e13d08433cf85c22f0c2c609","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-15 00:15:30.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"5f152584aef580c6b83332bf88e94e3854307170","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-15 00:18:09.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"0940b94fea52aaa28b0f091abf5e7bcad23177a2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-15 00:45:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c123db4582b3816c54e52d2e14d29309d6fdc12d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 01:26:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c52b6489d294fa1b9ae05def05428c8 : SUCCESS in 14m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6307769c91241a1989ea38928c9c5c4 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3470ca61628414abac210ed26172408 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/015312628d1a4951a6ffcffd4d0110a6 : SUCCESS in 13m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c6d618aa11b0418d87443c25e743804a : SUCCESS in 13m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe8c3aeb32164e728e29b23c35c9f7ea : SUCCESS in 15m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced7eeedc3b5472baef25b8ebc317467 : SUCCESS in 11m 53s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1104d809f4f54f188ce1f4ca4d31d0b9 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":2},{"id":"bf6acd4a8a8017af245e0a1b12ca10cf4a357dd4","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-15 01:29:26.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/04/720104/2/check/EMC_VxFlexOS_NOVA/560af27/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":2},{"id":"4f6c058ffd47734e5caba9c4088be6c510c68d2f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-15 02:12:00.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/04/720104/2/check/pmem-tempest-plugin-filtered/d400c56/ : SUCCESS in 24m 35s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/04/720104/2/check/pmem-tempest-plugin-filtered-multi-node/8d813fa/ : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":2},{"id":"3b8782d599f8c2ab481228056ab08a54dde50f25","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-15 03:42:35.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/720104/2 : SUCCESS in 2h 43m 31s","accounts_in_message":[],"_revision_number":2},{"id":"1de89560ed507593912040fd467e86c05a4721dd","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-15 04:04:01.000000000","message":"Patch Set 2:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-04-720104-2 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":2},{"id":"fdacef2be0ab0858cfa5584a00518bb9b808183a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-15 15:54:32.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"9f1b457100b993933a5a8b92ca348ed885059192","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-15 16:55:16.000000000","message":"Patch Set 3:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/04/720104/3/check/pmem-tempest-plugin-filtered/b08398c/ : SUCCESS in 28m 31s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/04/720104/3/check/pmem-tempest-plugin-filtered-multi-node/866842c/ : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c1a33953b244e36a3caf64c206067e14623c4061","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-15 17:09:33.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/04/720104/3/check-vote/ext-nova-zuul/d8df04a : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":3},{"id":"1087d9cec239c5bce41c9f7305590dd14ee11c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 17:20:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cb8232290544879b5ad7e6f36bd0b31 : SUCCESS in 17m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e277fc8713a342d59e94d31ca67af2c8 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/678800d477e44df096197d3a68ee9cea : SUCCESS in 9m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/837a054888634eb9a9e91a7f421e4a57 : SUCCESS in 13m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c909f9db5e1841a3b772ad45d034c108 : SUCCESS in 14m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3f3b2a7e2eb4fa7914c69af106eba34 : SUCCESS in 14m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72436f8f748430b9feed92c49db37ee : SUCCESS in 15m 29s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f548c0d6c35e41f287eefbf74ccbb188 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d0acca02bb8b2465c27f8f668c7611df0c30e16c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-15 17:22:53.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/04/720104/3/check/EMC_VxFlexOS_NOVA/d6de5c2/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":3},{"id":"91970f6c017922fe2df2caac14987724b0883e5b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-15 18:24:09.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/720104/3 : SUCCESS in 2h 20m 56s","accounts_in_message":[],"_revision_number":3},{"id":"9a374b112309c0d39f1ee37f73eae96b609795cf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-15 19:09:50.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-04-720104-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"bfab1f51bb2f448ca2d9592c03ae1dd28c582d5b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-18 18:03:56.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"476d684bd7eac091f0767ba44cc46834fe372395","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-18 18:23:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08af078959fc425d9b82f1818138751a : SUCCESS in 17m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bfe902d7ec4a49818c8ddef98245aa53 : SUCCESS in 15m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b3a4a12b1984ea48b8c52954733a3b9 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58f5526750de4f4e90912ecbacae9174 : SUCCESS in 15m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/27715fc35035419a8484e81c3e197bba : SUCCESS in 14m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a305fa4ef43d4132880e6d4acf905cbe : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0da908f81ee4ff88eccb532d42f5d96 : SUCCESS in 10m 59s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/715801ced42a45e0a94967fb191dd30a : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":4},{"id":"bb3363c21e65a33f737423251a8ac3c9992d2c9f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-18 18:25:40.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/04/720104/4/check/EMC_VxFlexOS_NOVA/5415098/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":4},{"id":"68990e724fe95b87c4fe8011d89862d796b2693d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-18 18:58:44.000000000","message":"Patch Set 4:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/04/720104/4/check/pmem-tempest-plugin-filtered/bd5aa5b/ : SUCCESS in 23m 35s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/04/720104/4/check/pmem-tempest-plugin-filtered-multi-node/f16bca3/ : SUCCESS in 52m 21s","accounts_in_message":[],"_revision_number":4},{"id":"8f6d2d9ad5dcd1224aa831a9532a92e681058ae8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-18 19:06:39.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/04/720104/4/check-vote/ext-nova-zuul/faeabd6 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":4},{"id":"9aacb99b3304a61b10086f9e4fe72b40639c4bc1","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-18 19:30:12.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-04-720104-4 : SUCCESS \n\nhttps://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI","accounts_in_message":[],"_revision_number":4},{"id":"d16c6226bebc6d79e7583d2c69f98a48e71b7c88","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-18 20:28:00.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/720104/4 : SUCCESS in 2h 22m 11s","accounts_in_message":[],"_revision_number":4},{"id":"d06863cf93702abdbf5c46f4d449ecedd2c10621","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-04-20 15:19:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"fd909a6f2af5c4ac6ef45fad0007307352e0ef8c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-20 16:48:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4c8cf5c8228bd58c70831a45b700f59191cbc5c3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-04-20 16:50:04.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3dd3f3528f8947caf681cf2278faec6e979066d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 16:51:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"49a784ce4147572cbc1342beff45b5fcea716461","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-04-20 16:58:52.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d3a8533f7db9fc86fdd62fa99b641cbbb710fbcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 18:34:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ee122562c504fc68a18104c4f79818c : SUCCESS in 21m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bf2e31df1114761a10260995c0ff270 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e058955c1f484a4e8a4d8a6e2db15f94 : SUCCESS in 14m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ee2aac1960f544af9d48d28e928cb68e : SUCCESS in 16m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4320f515c71e43549230e3312d754966 : SUCCESS in 17m 34s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7d8cf5143fd24f108311dd737ee6cce3 : SUCCESS in 19m 58s","accounts_in_message":[],"_revision_number":4},{"id":"a3800891f1a06f1605c5cbc0d0fe25a44c219dfd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-20 18:34:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"95c74a26d40883faa053febc72657a8b8e6c5bd3","revisions":{"af64d9d308146cd1dfc55f6d9ea1136b8288a1fb":{"kind":"REWORK","_number":1,"created":"2020-04-15 00:14:35.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/720104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/720104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/720104/1"}}},"commit":{"parents":[{"commit":"05ec728ebbac263bb610fcf775a2fc57d0d12d73","subject":"Add new default roles in servers attributes policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/05ec728ebbac263bb610fcf775a2fc57d0d12d73"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-14 19:10:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-14 19:10:52.000000000","tz":-300},"subject":"Add test coverage of existing remaining servers policies","message":"Add test coverage of existing remaining servers policies\n\nCurrent tests do not have good test coverage of existing policies.\nEither tests for policies do not exist or if they exist then they\ndo not cover the actual negative and positive testing.\n\nFor Example, if any policy with default rule as admin only then\ntest should verify:\n- policy check pass with context having admin or server owner\n- policy check fail with context having non-admin and not server owner\n\nAs discussed in policy-defaults-refresh, to change the policies\nwith new default roles and scope_type, we need to have the enough\ntesting coverage of existing policy behavior.\n\nWhen we will add the scope_type in policies or new default roles,\nthen these test coverage will be extended to adopt the new changes\nand also make sure we do not break the existing behavior.\n\nThis commit covers the testing coverage of existing remaining server\npolicies.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Ibfc1917d35ab9e7657b13ec32582891f06d88176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af64d9d308146cd1dfc55f6d9ea1136b8288a1fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/af64d9d308146cd1dfc55f6d9ea1136b8288a1fb"}]},"branch":"refs/heads/master"},"4da82308ed6f03666faa068a611d367cd83e9f0a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-15 00:45:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/720104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/720104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/720104/2"}}},"commit":{"parents":[{"commit":"aac18338d1675168c0aed92b835d7d8fe943d9d6","subject":"Add new default roles in servers attributes policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aac18338d1675168c0aed92b835d7d8fe943d9d6"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-14 19:10:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-15 00:45:43.000000000","tz":0},"subject":"Add test coverage of existing remaining servers policies","message":"Add test coverage of existing remaining servers policies\n\nCurrent tests do not have good test coverage of existing policies.\nEither tests for policies do not exist or if they exist then they\ndo not cover the actual negative and positive testing.\n\nFor Example, if any policy with default rule as admin only then\ntest should verify:\n- policy check pass with context having admin or server owner\n- policy check fail with context having non-admin and not server owner\n\nAs discussed in policy-defaults-refresh, to change the policies\nwith new default roles and scope_type, we need to have the enough\ntesting coverage of existing policy behavior.\n\nWhen we will add the scope_type in policies or new default roles,\nthen these test coverage will be extended to adopt the new changes\nand also make sure we do not break the existing behavior.\n\nThis commit covers the testing coverage of existing remaining server\npolicies.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Ibfc1917d35ab9e7657b13ec32582891f06d88176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4da82308ed6f03666faa068a611d367cd83e9f0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4da82308ed6f03666faa068a611d367cd83e9f0a"}]},"branch":"refs/heads/master"},"a98a18db6e42b93c17a6a72da8aff4c8b077f528":{"kind":"REWORK","_number":3,"created":"2020-04-15 15:54:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/720104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/720104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/720104/3"}}},"commit":{"parents":[{"commit":"59116714546c0d17cbad2b33324bc0a992eb4a47","subject":"Add new default roles in servers attributes policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/59116714546c0d17cbad2b33324bc0a992eb4a47"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-14 19:10:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-15 15:54:32.000000000","tz":0},"subject":"Add test coverage of existing remaining servers policies","message":"Add test coverage of existing remaining servers policies\n\nCurrent tests do not have good test coverage of existing policies.\nEither tests for policies do not exist or if they exist then they\ndo not cover the actual negative and positive testing.\n\nFor Example, if any policy with default rule as admin only then\ntest should verify:\n- policy check pass with context having admin or server owner\n- policy check fail with context having non-admin and not server owner\n\nAs discussed in policy-defaults-refresh, to change the policies\nwith new default roles and scope_type, we need to have the enough\ntesting coverage of existing policy behavior.\n\nWhen we will add the scope_type in policies or new default roles,\nthen these test coverage will be extended to adopt the new changes\nand also make sure we do not break the existing behavior.\n\nThis commit covers the testing coverage of existing remaining server\npolicies.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Ibfc1917d35ab9e7657b13ec32582891f06d88176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a98a18db6e42b93c17a6a72da8aff4c8b077f528"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a98a18db6e42b93c17a6a72da8aff4c8b077f528"}]},"branch":"refs/heads/master"},"95c74a26d40883faa053febc72657a8b8e6c5bd3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-18 18:03:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/04/720104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/720104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/720104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/720104/4"}}},"commit":{"parents":[{"commit":"76d3457bd9fd7ec597fd0fdc6cff8fdbdcec8aca","subject":"Add new default roles in servers attributes policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/76d3457bd9fd7ec597fd0fdc6cff8fdbdcec8aca"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-14 19:10:52.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-04-18 18:03:56.000000000","tz":0},"subject":"Add test coverage of existing remaining servers policies","message":"Add test coverage of existing remaining servers policies\n\nCurrent tests do not have good test coverage of existing policies.\nEither tests for policies do not exist or if they exist then they\ndo not cover the actual negative and positive testing.\n\nFor Example, if any policy with default rule as admin only then\ntest should verify:\n- policy check pass with context having admin or server owner\n- policy check fail with context having non-admin and not server owner\n\nAs discussed in policy-defaults-refresh, to change the policies\nwith new default roles and scope_type, we need to have the enough\ntesting coverage of existing policy behavior.\n\nWhen we will add the scope_type in policies or new default roles,\nthen these test coverage will be extended to adopt the new changes\nand also make sure we do not break the existing behavior.\n\nThis commit covers the testing coverage of existing remaining server\npolicies.\n\nPartial implement blueprint policy-defaults-refresh\n\nChange-Id: Ibfc1917d35ab9e7657b13ec32582891f06d88176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95c74a26d40883faa053febc72657a8b8e6c5bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95c74a26d40883faa053febc72657a8b8e6c5bd3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
