)]}'
{"id":"openstack%2Fnova~724386","triplet_id":"openstack%2Fnova~master~I1f8c9a2595a489f3ba7a6fff2fb74ae246c9ff8f","project":"openstack/nova","branch":"master","topic":"bug/1815989","hashtags":[],"change_id":"I1f8c9a2595a489f3ba7a6fff2fb74ae246c9ff8f","subject":"add workaround to disable multiple port bindings","status":"ABANDONED","created":"2020-04-29 17:42:07.000000000","updated":"2020-11-23 21:00:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7c2afe648ee0d4c6a5086644a87c3e59112bcb3f","_number":724386,"virtual_id_number":724386,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-12 13:29:23.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-14 13:47:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-05-14 13:09:55.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"date":"2020-05-14 12:35:02.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-12 14:54:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-05-15 22:28:20.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"date":"2020-05-14 14:13:25.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"date":"2020-05-14 11:47:59.000000000","_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"date":"2020-05-14 11:48:42.000000000","_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"date":"2020-07-21 20:08:38.000000000","_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 19:20:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-29 20:29:03.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-29 23:26:25.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-07 16:42:56.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2020-05-11 09:38:22.000000000","updated_by":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"reviewer":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"state":"REVIEWER"},{"updated":"2020-05-12 13:29:23.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-12 14:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-14 11:48:42.000000000","updated_by":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-14 12:35:02.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-05-14 13:09:55.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-05-14 13:47:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-14 14:13:25.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-05-15 22:28:20.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-21 20:08:38.000000000","updated_by":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"reviewer":{"_account_id":16011,"name":"James Denton","email":"james.denton@outlook.com","username":"busterswt"},"state":"REVIEWER"}],"messages":[{"id":"0425c66998a42276245cbffccab6c4b1781bd50a","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-04-29 17:42:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e9890a1af24ddb6b86a224c45de5c76a7e61839","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-04-29 18:26:29.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/86/724386/1/check/pmem-tempest-plugin-filtered/82e6967/ : SUCCESS in 22m 34s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/86/724386/1/check/pmem-tempest-plugin-filtered-multi-node/9d1daa5/ : SUCCESS in 42m 12s","accounts_in_message":[],"_revision_number":1},{"id":"697ff84f3d582acea44a30872ed917ef37657c88","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-04-29 18:40:19.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/86/724386/1/check-vote/ext-nova-zuul/c0405f3 : FAILURE in 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"aba006921b53805f8f4a68c94a27eb44ad8f8a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 19:20:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5e89e9e4b0c4dcaa8f25462efb7a12d : SUCCESS in 1h 06m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f86876bb4b3b4228b09eacb2acfcf07c : SUCCESS in 1h 18m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/553e104537344341a29185eb822fc1d1 : SUCCESS in 28m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9377d2b581ea4b19bd8eef6e4efb6545 : SUCCESS in 25m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688be7afcbfd42e693f65b5fbdc54895 : SUCCESS in 12m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e6a0effcaad410c85f8c8504f268367 : SUCCESS in 25m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9cdc9c54b8b42ed8bf9dfc7ebf6a190 : SUCCESS in 24m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f55d073748641cb8590879ca27c077c : SUCCESS in 11m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1e957f15480540fbb9571c649b3c4be9 : SUCCESS in 59m 21s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c12f4ea696a5424e916bb6571c3f1191 : SUCCESS in 1h 01m 20s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/81678b39a1a34ebd80b2b4492bb6ac7c : SUCCESS in 1h 14m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f9031d1bc9534a09b2878f2308418f19 : SUCCESS in 1h 31m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0c469bc854784ba69244cf3c83302dc8 : SUCCESS in 52m 00s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1ca5d0f674c54c678ec5e67ac98d18cc : SUCCESS in 1h 28m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/e2cae37826b746b0802d5235c80fcaad : SUCCESS in 1h 30m 40s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/727fdbdb285847e093f5c15c4787fdde : SUCCESS in 19m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c0253389b5a4848a9328c778377d083 : SUCCESS in 1h 02m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bfb22a63b5b349ceb03cd565191a1959 : SUCCESS in 52m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e0ee6740e7ce4b318f40395f699680e0 : SUCCESS in 34m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83dc70528b530ac9264daaa52c7a635a1a35bd28","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-04-29 19:20:58.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_VxFlexOS_NOVA http://publiclogs.emc.com/86/724386/1/check/EMC_VxFlexOS_NOVA/99e6f88/EMC_VxFlexOS_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC VxFlexOS CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_VxFlexOS_CI","accounts_in_message":[],"_revision_number":1},{"id":"477df99090c93eefd6687ffed2e580a0ea139e33","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-04-29 20:07:33.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/724386/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/0d3f224 : FAILURE in 2h 24m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1dfd3e1a31e97ab6c63e248874fee7f52369cdd6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-04-29 20:13:04.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/724386/1 : FAILURE in 2h 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"44b83257de7142c76b6c77dfb197c114615a06c5","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-04-29 20:29:03.000000000","message":"Patch Set 1:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-86-724386-1 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":1},{"id":"2c4a8f95e25ef4a1d9d8ebf7f6420d3abce8abc2","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-04-29 21:16:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/86/724386/1/check/tempest-dsvm-full-bionic/b454482/ : SUCCESS in 1h 45m 59s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/86/724386/1/check/tempest-dsvm-full-bionic-py3/a32d66a/ : SUCCESS in 2h 08m 32s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/86/724386/1/check/grenade-dsvm-bionic/7e68989/ : SUCCESS in 2h 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2bbb7dfbbc6fc053c656aa56cedf198e6e11ee4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2020-04-29 23:26:25.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/86/724386/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/86/724386/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"59c8601ab780464ecd9719b9ae8153d951e39fa2","author":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"date":"2020-05-11 09:38:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"25948bfa17753f1c21f1bb6d0c4770d9300f08a9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-12 11:45:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61cba43cfc88a24ae47c2e7b1b81767343ab5638","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-12 11:45:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f17a1bc9f7fdf3450e79e420c93ae2b43c3f395e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-12 12:11:57.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/724386/2 : FAILURE in 25m 37s","accounts_in_message":[],"_revision_number":2},{"id":"f13bc7313db1d24bf63f3623decd435d6ee82702","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-12 12:41:29.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/86/724386/2/check/pmem-tempest-plugin-filtered/9b528b4/ : SUCCESS in 25m 39s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/86/724386/2/check/pmem-tempest-plugin-filtered-multi-node/f11779c/ : SUCCESS in 54m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3452b39d21e36ec480c4d24704dc5e0896e04ee7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-12 13:08:15.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/e4f746c6dcaa40e6a3628f32358d47e7 : POST_FAILURE in 20m 11s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/ecb468967c654c629a7fcdd4138221c5 : POST_FAILURE in 20m 26s\n- grenade-dsvm-bionic finger://zuul-v3-executor.novalocal/7c2330c1c3cb4acb9e940d72feabe73d : POST_FAILURE in 1h 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56172d8b7b905af5abceb65222bf4457340b1416","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-12 13:12:58.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-42961 : SUCCESS in 1h 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"f76cb1d85d331900ef76145a80cc001fa6342b6c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-05-12 13:29:23.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/724386/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/57d808e : FAILURE in 1h 43m 29s","accounts_in_message":[],"_revision_number":2},{"id":"8efa75da1c92d5832f16404b44a37fad159a944d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 14:54:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c00be205716452293cd5f5654b2588d : SUCCESS in 1h 08m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b6154e4a00f544af90289fe9e52a5d0f : SUCCESS in 1h 39m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8478b30158ec49a39c2f1ac4e14397ff : SUCCESS in 19m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b77f803b6d24e92905d084671dff8ee : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b240aa912d4468eaa76d8f72d732fbb : FAILURE in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac2deb9350249719859dba6fc75d0e0 : SUCCESS in 12m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8ae3537f18f4140ab6a40d0302556ba : SUCCESS in 12m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420816366175489eafae0da424fa89f0 : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a939dc976c2a4d2fbb97ad266bdf587c : TIMED_OUT in 1h 01m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fcabb2ac902c4c7d9ac13ad92a560eeb : SUCCESS in 1h 03m 36s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/42c55dd9c3184a5590bffb9780a328dd : SUCCESS in 1h 09m 32s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/19094a7f107c4046ba70efa25ac60d5e : SUCCESS in 1h 22m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8b887262887140b88080cc0d67f9ea47 : SUCCESS in 1h 26m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/64a3a9a47ec0418b90bc30585680dfb7 : SUCCESS in 55m 55s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bea6cac2b2f14eec9fcf0c88d4d0c048 : SUCCESS in 1h 49m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/9e190db4cf3a4ecbae2fc1940ef36053 : SUCCESS in 1h 40m 54s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7baeab344b314ef4a1b9c1c5daa44eea : SUCCESS in 17m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4b9647cf900427881fa28cf646e814a : SUCCESS in 58m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28b421dc099a42069e56db82cc4d9ace : SUCCESS in 50m 52s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2efdc32ddbaa4f98868abd41d7d8f728 : SUCCESS in 31m 43s (non-voting)\n\nWarning:\n  Comments left for invalid file nova/tests/unit/virt/vmwareapi/test_driver_api.py\n  Comments left for invalid file nova/tests/unit/compute/test_compute_mgr.py\n  Comments left for invalid file nova/api/openstack/compute/servers.py\n  Comments left for invalid file nova/db/sqlalchemy/migrate_repo/versions/330_enforce_mitaka_online_migrations.py","accounts_in_message":[],"_revision_number":2},{"id":"386f43c14fe7f67f231307e40320c5faa5f69a28","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-05-14 11:47:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"748937c6dff7c6bfb01cb3b7ff6777ac454d8f95","author":{"_account_id":16376,"name":"Intel NFV CI","email":"openstack-nfv-ci@intel.com","username":"intel-nfv-ci","tags":["SERVICE_USER"]},"date":"2020-05-14 11:48:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tempest-dsvm-intel-nfv-xenial tempest-dsvm-intel-nfv-xenial : SKIPPED (non-voting)\n- tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial tempest-dsvm-multinode-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)\n- tempest-dsvm-ovsdpdk-nfv-networking-xenial tempest-dsvm-ovsdpdk-nfv-networking-xenial : SKIPPED (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"780456878e5e4274f6950aef21cd26d09f849ea5","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-05-14 12:35:02.000000000","message":"Patch Set 2:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered http://52.27.155.124/86/724386/2/check/pmem-tempest-plugin-filtered/b28e42e/ : SUCCESS in 23m 41s\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/86/724386/2/check/pmem-tempest-plugin-filtered-multi-node/4be1115/ : SUCCESS in 45m 31s","accounts_in_message":[],"_revision_number":2},{"id":"814d024399cbdb8de9650cdcb2b378b002068575","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-05-14 13:09:55.000000000","message":"Patch Set 2:\n\nTesting completed on the zVM Driver CI system check-nova pipeline.  To recheck only the zVM driver plugins, submit a comment with only zvm: recheck in the comment. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-43042 : SUCCESS in 1h 21m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0a6b13a710d3ad9e8c0171af439f2abe02b27255","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:31:40.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2498cc3cefaf4b84b072b8d970ba9a31 : SUCCESS in 1h 00m 24s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cf62d2b590b940b49787b9249bffbc87 : SUCCESS in 1h 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee7ede0c0854c2895b931a1fe401da9 : SUCCESS in 18m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aafda879efd4c6f9e897436c01dee29 : SUCCESS in 16m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1971d803b1a74cb588e203eec558a74c : SUCCESS in 9m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1bd7add8fa25494e8b32c7b3f2e03a33 : SUCCESS in 15m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc80ad46bf08409f8d2067e5154a733e : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7bbf623b27942f0ae3b29c9dabced29 : FAILURE in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47daeb8e7e5a49a588369a3fd1cc316d : TIMED_OUT in 1h 02m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/51979ddc6a3847ebb8996a0cf9fdd3ed : SUCCESS in 58m 08s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a4787b70b64741aabfcd212f7b96d4d8 : SUCCESS in 1h 11m 18s (non-voting)\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4e472580eedb48adb63162d59b435200 : SUCCESS in 1h 22m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/120ea98463b2402191f3f908fc942cc6 : SUCCESS in 1h 26m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5e91cd0a0d6a4843b5ee742923f812e5 : SUCCESS in 49m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/32294e6071cf43449855fd33b6b19d69 : SUCCESS in 1h 26m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/1212cb20689844b5bb7fdcd586d54ded : SUCCESS in 1h 33m 34s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2fbb464373ee4f34932943f07b272f2d : SUCCESS in 16m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5942f82e99a49e39c1bc9a41cb6a6b7 : SUCCESS in 1h 06m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3cc5b5492339446d927ed0bf48266e76 : SUCCESS in 47m 11s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/97ef54e3ec3745b6b98d9ec7d3b09c10 : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"481be89e6dadcb3f75aecbab8018126b98c6732f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-05-14 13:47:09.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/86/724386/2/check/tempest-dsvm-full-bionic/e922245/ : SUCCESS in 1h 29m 54s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/86/724386/2/check/tempest-dsvm-full-bionic-py3/70d1322/ : SUCCESS in 1h 25m 27s\n- grenade-dsvm-bionic finger://zuul-v3-executor.novalocal/56d9e66c7ee34e048636d47253dc9932 : POST_FAILURE in 1h 06m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f867b492097bf88a6440702f7fe45a51cf5f267","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-05-14 14:13:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/724386/2 : FAILURE in 2h 24m 38s","accounts_in_message":[],"_revision_number":2},{"id":"a1ddc601fa0ee79611995b35a5600f244a6773b0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-05-15 22:28:20.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- ext-nova-zuul : EXCEPTION","accounts_in_message":[],"_revision_number":2},{"id":"7c2afe648ee0d4c6a5086644a87c3e59112bcb3f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-11-23 21:00:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bf674497c59ec93b713189483f443286100f663a","revisions":{"e53464a38f9549532f394601df4f516bae2394bd":{"kind":"REWORK","_number":1,"created":"2020-04-29 17:42:07.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/86/724386/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/86/724386/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/724386/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/724386/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/724386/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/86/724386/1"}}},"commit":{"parents":[{"commit":"730eeb5fe002c093a5b477e095c3374457820007","subject":"Merge \"Feature matrix: update AArch64 information\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/730eeb5fe002c093a5b477e095c3374457820007"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-04-29 17:31:45.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-04-29 17:34:45.000000000","tz":60},"subject":"[WIP] add workaround to disable multiple port bindings","message":"[WIP] add workaround to disable multiple port bindings\n\nThis change adds a workaround config option to allow\ndisabling the multiple port bindings workflow\nas a temporay mitigation to bug: #1815989\nwhere the RARP mac learning packets sent by qemu are lost\non migration resulting in ta loss of network connectivity\nuntil the vm sends a packet.\n\nChange-Id: I1f8c9a2595a489f3ba7a6fff2fb74ae246c9ff8f\nRelated-Bug: #185989\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e53464a38f9549532f394601df4f516bae2394bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e53464a38f9549532f394601df4f516bae2394bd"}]},"branch":"refs/heads/master"},"bf674497c59ec93b713189483f443286100f663a":{"kind":"REWORK","_number":2,"created":"2020-05-12 11:45:26.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/86/724386/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/86/724386/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/724386/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/724386/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/86/724386/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/86/724386/2"}}},"commit":{"parents":[{"commit":"1d2934445732bfa214b1e6f2772ef16b1f51b3f9","subject":"Merge \"Suppress remaining policy warnings in unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1d2934445732bfa214b1e6f2772ef16b1f51b3f9"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-04-29 17:31:45.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2020-05-12 11:44:38.000000000","tz":0},"subject":"add workaround to disable multiple port bindings","message":"add workaround to disable multiple port bindings\n\nThis change adds a workaround config option to allow\ndisabling the multiple port bindings workflow\nas a temporay mitigation to bug: #1815989\nwhere the RARP mac learning packets sent by qemu are lost\non migration resulting in ta loss of network connectivity\nuntil the vm sends a packet.\n\nChange-Id: I1f8c9a2595a489f3ba7a6fff2fb74ae246c9ff8f\nRelated-Bug: #185989\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf674497c59ec93b713189483f443286100f663a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bf674497c59ec93b713189483f443286100f663a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
