)]}'
{"id":"openstack%2Fnova~743208","triplet_id":"openstack%2Fnova~master~I8017f9cef6177bac6e60648d1c0c8badf1959822","project":"openstack/nova","branch":"master","topic":"bp/separate-resize-cold-migrate","hashtags":[],"change_id":"I8017f9cef6177bac6e60648d1c0c8badf1959822","subject":"api: Split \u0027nova.compute.API._resize\u0027","status":"ABANDONED","created":"2020-07-27 12:16:24.000000000","updated":"2025-01-28 16:47:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"be2cb2a80ad122237ad995770a287c6a1ad40ec3","_number":743208,"virtual_id_number":743208,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-28 10:19:10.000000000","_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"date":"2020-09-28 15:05:19.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-28 14:29:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-09-28 11:08:14.000000000","_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"date":"2020-09-28 13:44:02.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-28 13:42:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-31 23:39:21.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-05 04:07:26.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-09-28 10:19:10.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 11:08:14.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 13:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 13:44:02.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-28 14:29:19.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-28 15:05:19.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"}],"messages":[{"id":"3f89bcd0c697c1be5a3bee92638310ddab022694","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-27 12:16:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c128232077040d34ba297cb431e8fb76f0b4572c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-07-27 13:04:23.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743208/1 : FAILURE in 19m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6065175c6ff172b15635ad5cf9769c02ff621e2c","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-07-27 13:51:04.000000000","message":"Patch Set 1:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/08/743208/1/check/pmem-tempest-plugin-filtered-multi-node/05bc295/ : SUCCESS in 45m 30s","accounts_in_message":[],"_revision_number":1},{"id":"1c699ddab53624a81f3d703013d17d31387585b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 16:11:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8b710728a644fe3be4a40360c9be07e : SUCCESS in 1h 14m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/483dd73d7d1044349ba1ec4c11ea36b3 : SUCCESS in 1h 24m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b185cb266e344886a49a25b20b1e1b31 : FAILURE in 17m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e6d56d87f3eb418089d455a7f609caff : FAILURE in 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a9995c62e40488eb06fd24f7b7ccf13 : SUCCESS in 8m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d0d42b7f25c404cbfdee6116e9f1ce7 : FAILURE in 14m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6506930ae9a34f96b4f5e210107425b2 : FAILURE in 14m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef3af80ba3994380a493bd1222352565 : SUCCESS in 12m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f9570717268644459b759ccd1a14ddb0 : SUCCESS in 52m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4ab2da4687e470ba02f0d3c4d165198 : SUCCESS in 1h 00m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/eab5aa239291474a986b844934e70080 : SUCCESS in 1h 26m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5bc8a7dfe92c40a195140af50dfa920d : SUCCESS in 58m 13s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ccdd7d4bcff9434c9f481387695b694b : SUCCESS in 1h 17m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/470ebb59fec64598ad1dbc7424c414a8 : SUCCESS in 1h 43m 55s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4e6f1bcadd6d462c853bc8e8730a9740 : SUCCESS in 18m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/824851498fd64c8997d47ad1e8001411 : SUCCESS in 1h 10m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/573c7d80b5e14f438a01258d05ee8510 : SUCCESS in 58m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5348fb875b85430e9c436138e6170f62 : SUCCESS in 25m 56s (non-voting)\n- barbican-simple-crypto-devstack-tempest https://zuul.opendev.org/t/openstack/build/71b56d63dcb9402db381df011fa75f2f : SUCCESS in 38m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0869afcfc5ca842c71bfc317e4f58d61021e8b14","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-27 16:17:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/1/check/tempest-dsvm-full-bionic/555cef7/ : SUCCESS in 1h 56m 15s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/1/check/tempest-dsvm-full-bionic-py3/631ac90/ : SUCCESS in 2h 02m 25s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/1/check/grenade-dsvm-bionic/15fb940/ : SUCCESS in 2h 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e88003034a79aa7d030e88a04bda79e21ccf072","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-07-27 17:08:03.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/08/743208/1/check-vote/ext-nova-zuul/9b0ab80 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1c3c4a1d1d88d7b0ee3e5355931915a7b4637fb5","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-07-31 23:39:21.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/743208/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/f6f9110 : FAILURE in 1h 34m 24s","accounts_in_message":[],"_revision_number":1},{"id":"59b5a1ed4339831161cd5e217a7fc606ad67a555","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-08-05 04:07:26.000000000","message":"Patch Set 1:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-44092 : FAILURE in 3h 00m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b76de81da1622f14e0db454e780eea4803cc12ab","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-25 16:50:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9f987311464a355e3edccf7f3b22afb564947219","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-09-25 19:46:36.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/743208/2 : SUCCESS","accounts_in_message":[],"_revision_number":2},{"id":"74ffe8ffe8f94f6bf133ae9a23666ca10c994ee8","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-25 23:17:32.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/08/743208/2/check-vote/ext-nova-zuul/e5c8d76 : FAILURE in 55m 48s","accounts_in_message":[],"_revision_number":2},{"id":"808483150a775f9dd5b214465de103767bb9b437","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-25 23:31:26.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/08/743208/2/check/pmem-tempest-plugin-filtered-multi-node/47112f0/ : FAILURE in 46m 07s\n- devstack-nfv-numa http://52.27.155.124/08/743208/2/check/devstack-nfv-numa/5c31e57/ : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":2},{"id":"f9a5cbddb8d725a4ac5241da696c32c2272934ee","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-25 23:46:30.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743208/2 : FAILURE in 3h 15m 45s","accounts_in_message":[],"_revision_number":2},{"id":"0fc24ec3046da4dd631bd1eba4f557a541a14ec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-26 00:35:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/82f56fa9df8f4fc390636d61366fa96c : SUCCESS in 44m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d936521e37ef47aeaedeb0129a8c50f1 : FAILURE in 56m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/303983e1a34b48c7be7c9cf36956de48 : FAILURE in 13m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80f0ca7a2a244ecda14979d698c483ff : FAILURE in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b06f1eb1d84925b032004331b1eba0 : SUCCESS in 8m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b36a7ae15f1e4e91ba93e97615663458 : FAILURE in 12m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/896056f9f2b04027824ddec7be79b5b3 : FAILURE in 11m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643ba11e146c43e691cd533369e4fe1c : SUCCESS in 10m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c7b4687f71ff46eb8962e42b617353ec : SUCCESS in 1h 18m 38s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/138dbabe6de948c09173f31e7be69d86 : FAILURE in 1h 03m 43s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/be64ab97fe244623aca45c249793c34d : FAILURE in 1h 08m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0ebe3f705be74e219a6559a56e112016 : SUCCESS in 50m 00s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/02948d4dc3a64bba8867b05e46036103 : FAILURE in 49m 40s (non-voting)\n- nova-evacuate https://zuul.opendev.org/t/openstack/build/1e92bb7091144e5982c7e5b719ed9ec6 : ERROR Unable to find playbook /var/lib/zuul/builds/1e92bb7091144e5982c7e5b719ed9ec6/untrusted/project_5/opendev.org/openstack/nova/playbooks/nova-evacuate/run.yaml in 1m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8faad0cdf31e43f2ac052c79af105b56 : FAILURE in 1h 21m 05s\n- nova-next https://zuul.opendev.org/t/openstack/build/6f44b780a2c14cee9d24432820236abb : FAILURE in 1h 30m 35s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/75ee25526d9b4c93ab9e31f3f9e567f0 : FAILURE in 16m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1b6585b854c4b9caafa32d14156c1f7 : SUCCESS in 53m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18966debd5ce45a79e54cea70ed9b7f2 : SUCCESS in 48m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ad7965222f6d48ad80e3a2abf8035ea2 : SUCCESS in 25m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2ff626e03abf498d835553550c04dc69 : FAILURE in 37m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fffe492369e7169ef5c70eacd687855eee25904e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-26 03:33:52.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/2/check/tempest-dsvm-full-bionic/b5e0e81/ : FAILURE in 23m 07s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/2/check/tempest-dsvm-full-bionic-py3/5005fa1/ : FAILURE in 22m 57s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/2/check/grenade-dsvm-bionic/77e430d/ : POST_FAILURE in 1h 27m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3202d73096e01f6baa32e7a9c68ac9c52f5de66f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-28 10:05:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dc853ed2fc631f1d36ecf2242aa07908dfe3cca1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-28 10:19:10.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/08/743208/3/check-vote/ext-nova-zuul/f339121 : FAILURE in 0s","accounts_in_message":[],"_revision_number":3},{"id":"7101d10121b55719aa0989e4c7b6639c4194af82","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-09-28 11:08:14.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/743208/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"f057f216c8f0a6618843d2b279a0ad87c6adba2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-28 13:42:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/91d3a093b5b246eeb390d3b51484297f : SUCCESS in 55m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/736121e9f46d424b95dc50aee27f9ddb : SUCCESS in 1h 05m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77f94ebf40a3454d94592435be588db3 : FAILURE in 13m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/369f52884b5b4cdc8b1ca46dfbf06be7 : FAILURE in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc94bbbe92bf4667bac4c44dc98c7e21 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/955824630cb74f6181f30b0791f7ed60 : FAILURE in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4245d2eda0b436da9052978c64f00ae : FAILURE in 12m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa4b919f051b4056b9d06f2c81a836b4 : SUCCESS in 14m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ca8ae8ad4eb34a65ab961630c413aeb2 : SUCCESS in 1h 21m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c1ad75b77c846c8b914020ceffb205a : SUCCESS in 1h 05m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/856d38d7aef9496aad0d91f46cc500b0 : SUCCESS in 1h 04m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dfda92c90bcb43f3bda4540d3c793505 : SUCCESS in 37m 13s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/2403c22b56c144caa82f1f51e28c520d : SUCCESS in 44m 36s (non-voting)\n- nova-evacuate https://zuul.opendev.org/t/openstack/build/13c0e2ab00b840ec964bf3ef3da9185e : FAILURE in 59m 17s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7611991bce124e22a1889e569a1ed093 : FAILURE in 1h 47m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/377956194b514e9293a8395f925b2dad : SUCCESS in 2h 23m 48s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a7982d73f934eba87b73ee9f8bb5d01 : SUCCESS in 19m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc7d98b85296476084c8bc9010115625 : SUCCESS in 1h 07m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/438ee9aeab904d3dbc3b12328bbd180c : SUCCESS in 59m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/955bf9e482814fbd8bed35932f982386 : SUCCESS in 36m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7e3cb59b73d04352ac81f244ca48ca30 : FAILURE in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"768ef00c3f1c4e3d69e214b388e6667f4b4bf33b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-28 13:44:02.000000000","message":"Patch Set 3:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/08/743208/3/check/pmem-tempest-plugin-filtered-multi-node/3e2b7a7/ : SUCCESS in 54m 14s\n- devstack-nfv-numa http://52.27.155.124/08/743208/3/check/devstack-nfv-numa/6676551/ : SUCCESS in 50m 19s","accounts_in_message":[],"_revision_number":3},{"id":"1d6042340befdc1520f48ef85680851215b02be9","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-28 14:29:19.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/3/check/tempest-dsvm-full-bionic/841f5cc/ : FAILURE in 1h 23m 58s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/3/check/tempest-dsvm-full-bionic-py3/bb9f2c6/ : FAILURE in 1h 46m 37s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/08/743208/3/check/grenade-dsvm-bionic/f58cee9/ : POST_FAILURE in 1h 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"16802db58c58f7afa38580ff70a6fa2616638627","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-28 15:05:19.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/743208/3 : SUCCESS in 2h 51m 55s","accounts_in_message":[],"_revision_number":3},{"id":"be2cb2a80ad122237ad995770a287c6a1ad40ec3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-28 16:47:49.000000000","message":"Abandoned\n\nNeeds a rebase. Can be taken by someone else if needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"462ff4703fd09470740f9210062946f2572207e8","revisions":{"ee5a99a26fd6f0536517353a95e5736ff9275650":{"kind":"REWORK","_number":1,"created":"2020-07-27 12:16:24.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/743208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/743208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/743208/1"}}},"commit":{"parents":[{"commit":"4acc2b3872c73aa706188a575d8fc07bb6830bc3","subject":"api: Add \u0027nova.compute.API.migrate\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/4acc2b3872c73aa706188a575d8fc07bb6830bc3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-17 11:40:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-27 12:15:39.000000000","tz":60},"subject":"api: Split \u0027nova.compute.API._resize\u0027","message":"api: Split \u0027nova.compute.API._resize\u0027\n\nWe split the public \u0027resize\u0027 method into \u0027_resize\u0027 and \u0027_migrate\u0027 in\nchange If8ef9077f2cd99ed8d2fa8968b0016d7ff1bd2e7. Now that we\u0027ve fixed\nall the callers, we can work on splitting the inner function.\n\nChange-Id: I8017f9cef6177bac6e60648d1c0c8badf1959822\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee5a99a26fd6f0536517353a95e5736ff9275650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ee5a99a26fd6f0536517353a95e5736ff9275650"}]},"branch":"refs/heads/master"},"ac5ce65e75f2c4a6c444fba3a724f2da7e3e7857":{"kind":"REWORK","_number":2,"created":"2020-09-25 16:50:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/743208/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/743208/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/743208/2"}}},"commit":{"parents":[{"commit":"9074a53c1565e1d0592a0c459a497925d80ca39a","subject":"api: Add \u0027nova.compute.API.migrate\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9074a53c1565e1d0592a0c459a497925d80ca39a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-17 11:40:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-09-25 16:49:00.000000000","tz":60},"subject":"api: Split \u0027nova.compute.API._resize\u0027","message":"api: Split \u0027nova.compute.API._resize\u0027\n\nWe split the public \u0027resize\u0027 method into \u0027_resize\u0027 and \u0027_migrate\u0027 in\nchange If8ef9077f2cd99ed8d2fa8968b0016d7ff1bd2e7. Now that we\u0027ve fixed\nall the callers, we can work on splitting the inner function.\n\nChange-Id: I8017f9cef6177bac6e60648d1c0c8badf1959822\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac5ce65e75f2c4a6c444fba3a724f2da7e3e7857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ac5ce65e75f2c4a6c444fba3a724f2da7e3e7857"}]},"branch":"refs/heads/master"},"462ff4703fd09470740f9210062946f2572207e8":{"kind":"REWORK","_number":3,"created":"2020-09-28 10:05:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/08/743208/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/08/743208/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/08/743208/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/08/743208/3"}}},"commit":{"parents":[{"commit":"44b1a48d6f1a1250d7be4ca0dece52c4ac1e63f3","subject":"api: Add \u0027nova.compute.API.migrate\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/44b1a48d6f1a1250d7be4ca0dece52c4ac1e63f3"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-17 11:40:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-09-28 10:04:51.000000000","tz":60},"subject":"api: Split \u0027nova.compute.API._resize\u0027","message":"api: Split \u0027nova.compute.API._resize\u0027\n\nWe split the public \u0027resize\u0027 method into \u0027_resize\u0027 and \u0027_migrate\u0027 in\nchange If8ef9077f2cd99ed8d2fa8968b0016d7ff1bd2e7. Now that we\u0027ve fixed\nall the callers, we can work on splitting the inner function.\n\nChange-Id: I8017f9cef6177bac6e60648d1c0c8badf1959822\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462ff4703fd09470740f9210062946f2572207e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/462ff4703fd09470740f9210062946f2572207e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
