)]}'
{"id":"openstack%2Fnova~743268","triplet_id":"openstack%2Fnova~stable%2Frocky~I8a2726b39d0444de8c35480024078a97430f5d0c","project":"openstack/nova","branch":"stable/rocky","topic":"bug/1843708","hashtags":[],"change_id":"I8a2726b39d0444de8c35480024078a97430f5d0c","subject":"objects: Update keypairs when saving an instance","status":"MERGED","created":"2020-07-27 15:53:37.000000000","updated":"2020-08-10 17:44:58.000000000","submitted":"2020-08-10 17:44:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743268-1597081498951-ff0595dd","meta_rev_id":"edc2fcf673c34f29f5cc43da9747e1f154541802","_number":743268,"virtual_id_number":743268,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-01 21:30:58.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-04 13:15:21.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-08-04 16:36:25.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"date":"2020-08-10 15:27:47.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-08-04 13:20:16.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-10 17:44:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2020-08-05 15:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2020-08-10 15:13:48.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-08-10 15:13:48.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-01 21:30:58.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-04 13:15:21.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-08-04 13:20:16.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-08-04 16:36:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-05 15:56:03.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-08-10 15:13:48.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-08-10 17:44:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"859cd8a24638901e1ef91e143349f837f1fa8d92","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-27 15:53:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35ef1a8a5b5d49d51b66996ef0a62e6019ccc3d1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-07-27 16:04:09.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743268/1 : FAILURE in 9m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f5c92dce070be223d8e7b1f35078cba89369f1d2","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-07-27 16:11:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/68/743268/1/check/pmem-tempest-plugin-filtered-multi-node/a2190fa/ : RETRY_LIMIT in 4m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a1accbae461b3aa6c636e93017532ba18fc030a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 19:45:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e81f3b4218d548608822ce15d64ae26e : FAILURE in 11m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2d06c6d3a0564f71add8d919728ad1d0 : SUCCESS in 14m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4f10e3352dd54ea3aded4f5f7c1ee5c4 : SUCCESS in 13m 48s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3d548e58c60a46ceafa6dc01bd72ab83 : SUCCESS in 8m 27s\n- tempest-full https://zuul.opendev.org/t/openstack/build/86ef057cf5f642e5b7a8e52a2c1dd04d : SUCCESS in 1h 18m 10s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/998c26f2f7b3495ba14a8e4e93bb900a : SUCCESS in 1h 00m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac4b7a9226dd41b69f439a7550991ad0 : SUCCESS in 1h 22m 47s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a3e1139010c4498cbb1833617190d3a8 : FAILURE in 42m 46s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/6d30d1d006b44bd3af94d90ee3d4b9a1 : SUCCESS in 55m 54s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/40dcefa405fa437a8d549eb08cd2064e : SUCCESS in 1h 03m 37s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/155607c63fce4c2aa52e465bc903c693 : SUCCESS in 51m 48s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/c8dbde3712a64af78dff4d2f53c1f93d : SUCCESS in 1h 21m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/d71375bf2124430b824aa7bb5d95ee7e : SUCCESS in 1h 30m 13s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/616584119f094ef7b8bdee424df1c74c : SUCCESS in 2h 11m 56s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/40080f7c7a6644a198184b1ea8f94595 : FAILURE in 6m 35s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7e1e7478d0814644951c59b35a291be0 : FAILURE in 8m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49698f0ec2884a8cabde22bdf383927e : SUCCESS in 16m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/5613a9e75a774b1388f27766831ec769 : SUCCESS in 43m 44s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/beeb99d7dedf44eca072d77e627b2137 : FAILURE in 50m 05s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/c24dcde11a284c2aa97a66e11689888f : SUCCESS in 1h 32m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae2376afc0a95ff959a4a9d2f3e0adaf735f6e1f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-27 19:53:53.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/1/check/tempest-dsvm-full-bionic/7ef55b8/ : FAILURE in 24m 08s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/1/check/tempest-dsvm-full-bionic-py3/2eebb34/ : SUCCESS in 1h 35m 22s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/1/check/grenade-dsvm-bionic/af32ff5/ : POST_FAILURE in 45m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"feb216a12a9d21674734def55462f5ae3dcaa984","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-31 15:20:19.000000000","message":"Patch Set 1: Workflow-1\n\nFunctional test failures look real :(","accounts_in_message":[],"_revision_number":1},{"id":"40accb7014e494ddcc823e0a5fe34252531c3254","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-31 15:24:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63dbd77c0152e23904f70e7875607ccb3570a6fa","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-07-31 17:26:24.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/68/743268/2/check/pmem-tempest-plugin-filtered-multi-node/9709cd7/ : RETRY_LIMIT in 4m 31s","accounts_in_message":[],"_revision_number":2},{"id":"6b7e1654055e70cb9f8d80591b0288e4fa6d9b9a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-07-31 19:25:02.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743268/2 : FAILURE in 10m 07s","accounts_in_message":[],"_revision_number":2},{"id":"efc4a2b62e8c40319efaa0f85188687dfd1f7e23","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-31 22:11:38.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/2/check/tempest-dsvm-full-bionic/f579977/ : FAILURE in 28m 19s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/2/check/tempest-dsvm-full-bionic-py3/cfe2fa2/ : SUCCESS in 1h 20m 32s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/2/check/grenade-dsvm-bionic/37e2f97/ : POST_FAILURE in 38m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc7ef682c9ec01dded56af1d616a36cba559863f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 22:48:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e703d3a9b66433e8adb3e280fbfe70a : FAILURE in 11m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76384a2975db4543a453ce2ef961edf5 : SUCCESS in 13m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4394adae223045e0bc7a56c22f290791 : SUCCESS in 17m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ec648b2d3bff4aa6a49471cdb6f8379e : SUCCESS in 10m 17s\n- tempest-full https://zuul.opendev.org/t/openstack/build/611956fbd26d45db90c9e7ae19fabe54 : SUCCESS in 1h 23m 43s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/171c9f7ce47a48639b83d726bda91989 : SUCCESS in 36m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7f471a40ba746e8b1a090716a4d806c : SUCCESS in 1h 22m 25s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5fa12ab08e874a7d962ee108b43366e2 : SUCCESS in 55m 58s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/db6e187ddd164620975327dd03ba6c85 : SUCCESS in 39m 53s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/3e159ad7cc3c440ead306f332abd534c : SUCCESS in 1h 06m 05s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/050e0406a98b4092821fe72565ae2c53 : SUCCESS in 58m 30s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/c80baca2e1574e53a074f64b812a8d50 : SUCCESS in 37m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/52a5a434d1e443628f9bea65a88e4449 : SUCCESS in 1h 24m 27s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/d275518ef5674fd584874dc287c406d1 : FAILURE in 1h 24m 05s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/55a68f3825da40e5ae13a92b77947b36 : SUCCESS in 25m 47s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7e1e02d2fa9b4d70a1d07e0cd5a23213 : SUCCESS in 17m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/96330d4cabd943babbed3bdf01660b2a : SUCCESS in 13m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7f2d9e09b13488b9fc221bfd760be83 : SUCCESS in 52m 38s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/edfe99bf57f54e9aa75f1e614128fc9c : SUCCESS in 59m 00s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/656aec3ddd0b4d7ba4324205cc59f88c : SUCCESS in 1h 49m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97f1edd81e828e0e242c18989340e55a2894c74e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-01 21:30:58.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/743268/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/36fa967 : FAILURE in 25m 47s","accounts_in_message":[],"_revision_number":2},{"id":"220d36e62ba3620174012f9f931b698b95a7ebfc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-08-04 13:04:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"eb4d484d282310b1e30e0f7140935af29651e0b6","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-08-04 13:15:21.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743268/2 : FAILURE in 10m 03s","accounts_in_message":[],"_revision_number":2},{"id":"73fd9759be260b2f3e6cfdfd9c6961fb4c871a16","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-08-04 13:20:16.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/68/743268/2/check/pmem-tempest-plugin-filtered-multi-node/098f034/ : RETRY_LIMIT in 4m 02s","accounts_in_message":[],"_revision_number":2},{"id":"f1b70007533897eaaf88ad1f5eef3652c6556f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-04 15:28:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c379709c9a046a0a4596f00b1a8b034 : SUCCESS in 10m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7c5700a8064042bcb26d98385fab0623 : SUCCESS in 10m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f8170089a314a92b71e5f932e09ecc5 : SUCCESS in 11m 33s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/add76093eced42b2854c2b8b94cac8db : SUCCESS in 10m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/94af5868f4204f8a832ac77568572d9a : SUCCESS in 1h 41m 59s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/9baff2ee3047418dbed457103b1e9c53 : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fe12ff37bfe42b3b0152828b5e025a5 : SUCCESS in 1h 38m 00s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c32e3a07a91748c0b6c75711c952e684 : FAILURE in 59m 15s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/80b8b4a71add4eb095b475ba1622532f : SUCCESS in 1h 18m 50s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/2e54b780b554496d9f813f8dc5abbbdd : SUCCESS in 1h 12m 41s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e5156fce94bb42c287b2b8ffe0821d96 : SUCCESS in 58m 12s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/6a0344b695804d9294ac22345250055b : SUCCESS in 1h 17m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/be85e089b42f4d8e9b56d6f2cedf2244 : SUCCESS in 1h 31m 04s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/ee512603f709438dab38bfccf46a68e0 : SUCCESS in 2h 15m 03s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/b899e19d614e4739835150244ad23bfc : SUCCESS in 17m 10s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/8fb85509efb84c468cb691baf835af6d : SUCCESS in 19m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b25f70fdaf4d299d3518aec7ff8d0b : SUCCESS in 15m 26s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ae8d9a66472a4fdba7603515bde6e7df : SUCCESS in 1h 15m 56s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/3f16686e9eb34b4c86e4ecf0bf2e9cb0 : SUCCESS in 1h 10m 31s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/631b138dece3456085ae44c96cdd8765 : SUCCESS in 1h 48m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a76917176e3e9f25ff12b34a5e30430ecc674eb3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-04 16:36:25.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/2/check/tempest-dsvm-full-bionic/d29da18/ : FAILURE in 27m 06s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/2/check/tempest-dsvm-full-bionic-py3/e29dff7/ : SUCCESS in 1h 24m 52s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/68/743268/2/check/grenade-dsvm-bionic/8e6738c/ : POST_FAILURE in 42m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fcc1f92161bc6a83dd97b9546216426cfb62db1c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-08-05 15:56:03.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)\n\nThe backport looks OK to me. A small modification was needed, and is noted in commit message. Also the patch is merged in Stein.","accounts_in_message":[],"_revision_number":2},{"id":"5d41af6427684212e606241ed6ea296d1f30527c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-08-10 15:13:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"67494c88ed66177ea27cda94a9c935ca967d54be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 15:14:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"00c2cb2565c7219fad9d28fd3a95a1101e604f55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:44:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/921d1de5d75d424aa65a5878c1c7c928 : SUCCESS in 11m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8979d4ac16b84333b9e766830060fdc0 : SUCCESS in 15m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cfe0eefc9965452fb87c96aaf0e4e0a2 : SUCCESS in 14m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/95e17671bf1f44dc967fa521b42aa40d : SUCCESS in 9m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/27844177baf14608be179f3f8db6b989 : SUCCESS in 54m 52s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a1d4c261e34a4748885f936c341f26da : SUCCESS in 1h 04m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d426c25d78dc4bfe92546e40e934f036 : SUCCESS in 1h 20m 19s\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/3d263d6a7b2d46e3a5d79b8c28c5c8f1 : SUCCESS in 1h 20m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9f3070a56d0e4e929af92d6c898b7481 : SUCCESS in 46m 42s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/629cc972b7df4489b3cea2db7ed18fb5 : SUCCESS in 58m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/5b38d2cd66004c588efb235423a31337 : SUCCESS in 1h 24m 05s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/0254c1905a254b38b90087e60b012c60 : SUCCESS in 2h 21m 46s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/47eea7a5b3624f12b3e3aa2781c887df : SUCCESS in 17m 01s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/c6fc3b394c8244f3aa211cbec1b878c7 : SUCCESS in 17m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8463d6c7a68c4c4f9ff120346aaf78e8 : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":2},{"id":"3e9201894c32b77c8d164af0137ecc51ea7ef607","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-10 17:44:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa7a6939d5177c0dd8c9f5a7bf7975264d2f5a2a","revisions":{"02504f7a4d6ece341f2d2c74f54c6c0a6c3da608":{"kind":"REWORK","_number":1,"created":"2020-07-27 15:53:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/743268/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/743268/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/743268/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/743268/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/743268/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/743268/1"}}},"commit":{"parents":[{"commit":"e3093d42f46af810f316421a9b59eafe94039807","subject":"zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3093d42f46af810f316421a9b59eafe94039807"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-09-19 06:57:44.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-27 15:52:00.000000000","tz":60},"subject":"objects: Update keypairs when saving an instance","message":"objects: Update keypairs when saving an instance\n\nThe keypair of a server is updated when rebuilding the server with a\nkeypair. This function has been added since API microversion 2.54.\nHowever the \u0027keypairs\u0027 of the instance object is not saved when saving\nthe instance object currently.\n\nMake the instance object update the \u0027keypairs\u0027 field when saving the\ninstance object.\n\nChange-Id: I8a2726b39d0444de8c35480024078a97430f5d0c\nCloses-Bug: #1843708\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 086796021b189c3ac64805ed8f6bde833906d284)\n(cherry picked from commit aed86ee5d6289edf1baf9fe0b2a9e509031fdd25)\n(cherry picked from commit b971dc82cb524fe86284c95ec671e2bad1c2874f)\n(cherry picked from commit 0bc5a4ecb524a73aacb5d0dd2887799885bdbb14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02504f7a4d6ece341f2d2c74f54c6c0a6c3da608"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/02504f7a4d6ece341f2d2c74f54c6c0a6c3da608"}]},"branch":"refs/heads/stable/rocky"},"aa7a6939d5177c0dd8c9f5a7bf7975264d2f5a2a":{"kind":"REWORK","_number":2,"created":"2020-07-31 15:24:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/743268/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/68/743268/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/743268/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/743268/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/68/743268/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/68/743268/2"}}},"commit":{"parents":[{"commit":"e3093d42f46af810f316421a9b59eafe94039807","subject":"zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e3093d42f46af810f316421a9b59eafe94039807"}]}],"author":{"name":"Takashi NATSUME","email":"natsume.takashi@lab.ntt.co.jp","date":"2019-09-19 06:57:44.000000000","tz":540},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-31 15:22:35.000000000","tz":60},"subject":"objects: Update keypairs when saving an instance","message":"objects: Update keypairs when saving an instance\n\nThe keypair of a server is updated when rebuilding the server with a\nkeypair. This function has been added since API microversion 2.54.\nHowever the \u0027keypairs\u0027 of the instance object is not saved when saving\nthe instance object currently.\n\nMake the instance object update the \u0027keypairs\u0027 field when saving the\ninstance object.\n\nChanges:\n\tnova/tests/functional/regressions/test_bug_1843708.py\n\nNOTE(stephenfin): Changes are necessary since \u0027PlacementFixture\u0027 moved\nin stable/stein with change Idaed39629095f86d24a54334c699a26c218c6593\nbut it\u0027s still in the old place here.\n\nChange-Id: I8a2726b39d0444de8c35480024078a97430f5d0c\nCloses-Bug: #1843708\nCo-authored-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 086796021b189c3ac64805ed8f6bde833906d284)\n(cherry picked from commit aed86ee5d6289edf1baf9fe0b2a9e509031fdd25)\n(cherry picked from commit b971dc82cb524fe86284c95ec671e2bad1c2874f)\n(cherry picked from commit 0bc5a4ecb524a73aacb5d0dd2887799885bdbb14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa7a6939d5177c0dd8c9f5a7bf7975264d2f5a2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/aa7a6939d5177c0dd8c9f5a7bf7975264d2f5a2a"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
