)]}'
{"id":"openstack%2Fnova~743805","triplet_id":"openstack%2Fnova~stable%2Ftrain~Ic01603a91f6099f1068af0e955f3e1056021d673","project":"openstack/nova","branch":"stable/train","topic":"bug/1889257","hashtags":[],"change_id":"Ic01603a91f6099f1068af0e955f3e1056021d673","subject":"Handle multiple \u0027vcpusched\u0027 elements during live migrate","status":"MERGED","created":"2020-07-29 17:10:23.000000000","updated":"2020-08-01 13:29:54.000000000","submitted":"2020-07-31 02:38:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"743805-1596163123870-2eb1796f","meta_rev_id":"1bd84de321dc5b0cb92b91eb661dec54ee267e7c","_number":743805,"virtual_id_number":743805,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-30 19:18:25.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-08-01 13:29:54.000000000","post_submit":true,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-07-30 22:04:09.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-07-30 19:34:00.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 02:38:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-30 20:25:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-30 19:17:20.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-30 20:25:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-30 19:17:20.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-07-30 19:18:25.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-07-30 19:34:00.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-07-30 20:25:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-07-30 22:04:09.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-31 02:38:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-08-01 13:29:54.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dcdeb18474271af495fe29f4c9cf04f5f57e9c72","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-07-29 17:10:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b76ecba3c7e17dff1eb323fb9d76cc0ebe0ad7a0","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-07-29 17:12:43.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743805/1 : FAILURE in 1m 25s","accounts_in_message":[],"_revision_number":1},{"id":"ad5e67b739cce43933796dccdeae843bf7d116fa","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-07-29 17:29:35.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/05/743805/1/check/pmem-tempest-plugin-filtered-multi-node/6d20029/ : RETRY_LIMIT in 5m 22s","accounts_in_message":[],"_revision_number":1},{"id":"7b0b460eb9375ceaeafe7ad5241e6ce4b5e95ac1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-30 00:14:27.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/743805/1/check/tempest-dsvm-full-bionic/e7c1763/ : TIMED_OUT in 2h 37m 26s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/743805/1/check/tempest-dsvm-full-bionic-py3/47bfe1b/ : SUCCESS in 1h 52m 09s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/743805/1/check/grenade-dsvm-bionic/52531d5/ : POST_FAILURE in 1h 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcf26153c3c669a1ecc58adf79425d1bdcdb4627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-30 01:03:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/27ef325dd5234f94a326148b795b654f : SUCCESS in 1h 12m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5efe29d2baae40c0ab3aee49c9973385 : SUCCESS in 1h 35m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c58564cd44a42e2ba19907c1adf5087 : SUCCESS in 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/849d678c7c204bd8b5249f92f557e30c : SUCCESS in 19m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c18c18bf48d41c3b94035868ddb490e : FAILURE in 11m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/983e988db1234fff8af79a014c9ca14f : SUCCESS in 13m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/833c5a789a2a48c4aeaf47ac7dae9302 : SUCCESS in 11m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/124cd1021c7345d68b22f778dce4b94b : SUCCESS in 11m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5691b329f2b44d6ca0580a483d155af1 : SUCCESS in 10m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/71eeef0d4467488689dd785494bdac7d : SUCCESS in 1h 12m 06s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/eb46e49bdf06496d9d85ff982e0c5c98 : SUCCESS in 1h 05m 28s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b965fe816dcc496a8b1f3ad41eaf0518 : SUCCESS in 1h 07m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4af644c5d5a14447a172551d197c619b : SUCCESS in 58m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2f34eefadf884d819eb431a659e9e7fb : SUCCESS in 56m 51s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/49eb5d55b8404f469db4b2f6ce3452da : SUCCESS in 46m 17s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/37c2e2643f4b49f388e24efd40037da6 : SUCCESS in 1h 25m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/685a733e520d401c9a6941349369b553 : SUCCESS in 1h 11m 16s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/42c969f59a4740cda91fbce1e705a01a : SUCCESS in 23m 06s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dbc7956102b3414daf5e0b5df40e9824 : SUCCESS in 19m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6ff158cd32574b16992d20a091bf2b34 : SUCCESS in 2h 23m 17s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c341b6cfd1de4659944c874a9065a684 : SUCCESS in 1h 09m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a897205c63f46d495b849e51c323aca : SUCCESS in 1h 03m 49s","accounts_in_message":[],"_revision_number":1},{"id":"880bde2f914ed6d80b486ccd3061e099ef1980d5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-30 19:16:45.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"65148787c8243334e22905af73fd95af35d9ccee","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-30 19:17:20.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM now, merged in stable/ussuri, clean to stable/train.","accounts_in_message":[],"_revision_number":2},{"id":"4b35ca2632addbc4550653de1ede00ac27dff001","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-07-30 19:18:25.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/743805/2 : FAILURE in 45s","accounts_in_message":[],"_revision_number":2},{"id":"ce31b9330ad836690272cccf27fdc08ff4b36f97","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-07-30 19:34:00.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/05/743805/2/check/pmem-tempest-plugin-filtered-multi-node/92c1532/ : RETRY_LIMIT in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"8e945091c3ab8efb4c1d8d57b4418bdba4a652a9","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-07-30 20:25:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nClean cherry pick from Ussuri, where the patch has merged, indeed. This passed already (pep8 failed as ussuri patch was not yet merged), so I approve in advance.\n\nLee, I think, in this case the rebase was not needed, a recheck should have been enough. (The only case when a rebase is necessary is when patches are stacked and the patch\u0027s parent had another patchset meanwhile, i.e. the given parent hash is \u0027non-existent\u0027 anymore)","accounts_in_message":[],"_revision_number":2},{"id":"2e562ea0089ebe69a0aef975349840d303c6b397","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-07-30 20:52:03.000000000","message":"Patch Set 2:\n\n\u003e Clean cherry pick from Ussuri, where the patch has merged, indeed.\n \u003e This passed already (pep8 failed as ussuri patch was not yet\n \u003e merged), so I approve in advance.\n \u003e \n \u003e Lee, I think, in this case the rebase was not needed, a recheck\n \u003e should have been enough. (The only case when a rebase is necessary\n \u003e is when patches are stacked and the patch\u0027s parent had another\n \u003e patchset meanwhile, i.e. the given parent hash is \u0027non-existent\u0027\n \u003e anymore)\n\nYup indeed but there\u0027s no harm in rebasing if we need to recheck anyway IMHO.","accounts_in_message":[],"_revision_number":2},{"id":"c808085d479874f85862cc89d5e29e90160ffb70","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-07-30 22:04:09.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/743805/2/check/tempest-dsvm-full-bionic/7f500a7/ : TIMED_OUT in 2h 36m 35s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/743805/2/check/tempest-dsvm-full-bionic-py3/c985398/ : SUCCESS in 1h 33m 47s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/05/743805/2/check/grenade-dsvm-bionic/16cd11a/ : POST_FAILURE in 1h 18m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2c9b6aa73c1f2f19da5778355bb388e036fd82e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 00:24:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/28193039762946ffbb8bd3db45e568b2 : SUCCESS in 1h 21m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7b1b1ba1eb754c8ca935b71ae7f2338e : SUCCESS in 1h 17m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed5b95bfb9d248e2b1f5776146c6c17a : SUCCESS in 15m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caa0ec6ffb2649a9a238eae519fdbfda : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c75ad1aaf84e59a7302ec052a281bb : SUCCESS in 13m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d469a66e76a4d4aaa7417f72fde8fbc : SUCCESS in 15m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b068f1325b3460088fe7f8db8c886a5 : SUCCESS in 13m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14a912fa30364c5b99242b388eb71584 : SUCCESS in 10m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f3d4d465104ba1a2cf6efddf56451a : SUCCESS in 7m 56s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3af69d274c214f22afa070b0a5e5a0ab : SUCCESS in 50m 56s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/6df7033c2f944f4bbc86bbe25ad8b96e : SUCCESS in 1h 15m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e6d3e84b9804a89b94c0eb83867d7cf : SUCCESS in 1h 19m 49s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/05294fcf243440d9aca0a38c80acc968 : SUCCESS in 1h 09m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aaea6f6a08d548e8ab9dee9ff8843a4a : SUCCESS in 55m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/591f1e0458db4dfca6e4da3d9fbcbd7f : SUCCESS in 57m 06s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/db415f574ae643a9943efcbfad9d82fc : SUCCESS in 1h 18m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/abab81c8f7c940aaa787b8f51f802f63 : SUCCESS in 1h 28m 22s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/d0c84b320e1b4337bc7b1cda4a3cd903 : SUCCESS in 19m 53s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f6ca24a193b14100b1f80592ad7debc6 : SUCCESS in 18m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7b50d3a4f15c4075a23f3527c1c37d94 : SUCCESS in 1h 45m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/9db13b1e567c4aaf833649938528af9c : SUCCESS in 44m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9f8ab3050d140cca6c4a588736a7b37 : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":2},{"id":"453f53f0ef09c92920eae33bdbc2b0b1bffe2787","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 00:24:25.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9108208e4c2b7461ae868be257a4fa33994d1c90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 02:38:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/01e0c04b0de048e39fb5affc890f6e4b : SUCCESS in 59m 25s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a969b47845904f35b40cd6b950932379 : SUCCESS in 1h 12m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d48d07923b7a41cf9e304907d53dab74 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42e2dc698ffa4ac7852a667d6ed7bd5c : SUCCESS in 10m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de4589e4d8ce4a279d1634a052c30f21 : SUCCESS in 12m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd7dfd9468fe40178890dea40497aacc : SUCCESS in 13m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c177375391a47ac976e0b605e05ffd0 : SUCCESS in 15m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8480034a930414f9c54117ad33ab1dd : SUCCESS in 12m 20s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e00b2187fa804dc5a3d39354a447c495 : SUCCESS in 1h 16m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a7d0dadb3574242804bb0cc7f9cc365 : SUCCESS in 49m 32s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/abdb860e65184100adcb68a646c91c26 : SUCCESS in 19m 12s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/00b8aacbdfbd4d9c84f69def5c751cb9 : SUCCESS in 17m 03s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e90dcec6124443cf89a00c4555dd84a0 : SUCCESS in 1h 32m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/d63bc418bd4949878fda7b019792d983 : SUCCESS in 1h 35m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdb1f5f5318747a4a097f769b953b286 : SUCCESS in 2h 03m 11s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3b66d69e10194113b95cd51729f196c6 : SUCCESS in 54m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1187da681fe4a87b4094a141d80a2c1 : SUCCESS in 59m 13s","accounts_in_message":[],"_revision_number":2},{"id":"3371d82f3ceefd6ecd0eead989362805ac5a56fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 02:38:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d9b29da04aea8b46621c2d90b5d987b8b99d08c6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 02:40:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9518ff4acb4ec0a5b5f251739ef90f : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":2},{"id":"f140dcb3d976a76149cf1535f7c1538674522a86","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-01 13:29:54.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/743805/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/76bf213 : FAILURE in 27m 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"95bce095ed8489968e8dec2f0b6977f79bb8f98a","revisions":{"58a928681c49c2aa04a29d2ddb5ad4bf502f429d":{"kind":"REWORK","_number":1,"created":"2020-07-29 17:10:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/743805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/743805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/743805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/743805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/743805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/743805/1"}}},"commit":{"parents":[{"commit":"7e91134c2304e0226e0b05fcaf3dae2fc93c111b","subject":"Merge \"zuul: remove legacy-tempest-dsvm-neutron-dvr-multinode-full\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e91134c2304e0226e0b05fcaf3dae2fc93c111b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-28 15:22:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-29 17:07:43.000000000","tz":60},"subject":"Handle multiple \u0027vcpusched\u0027 elements during live migrate","message":"Handle multiple \u0027vcpusched\u0027 elements during live migrate\n\nWhen live migrating a pinned instance, we recalculate pinning\ninformation for the destination host and then update the instance\u0027s XML\nbefore spawning the instance there. As part of the pinning information\nrecalculation, we must also recalculate information for realtime cores,\nwhich are configured using the \u0027\u003cvcpusched\u003e\u0027 element. The\n\u0027nova.virt.libvirt.migration._update_numa_xml\u0027 function, which handles\nthis updating, was assuming there would only be one of these elements.\nThis is a reasonably sane assumption since this is all we create in the\n\u0027nova.virt.libvirt.LibvirtDriver._get_guest_numa_config\u0027 function used\nto generate the initial instance XML. However, a look at logs show that\nat least some (all?) versions of libvirt actually rewrite the XML we\u0027re\nproviding them. Compare what is returned from \u0027_get_guest_xml\u0027:\n\n  DEBUG nova.virt.libvirt.driver [...] [instance: ...] End _get_guest_xml xml\u003d\u003cdomain type\u003d\"kvm\"\u003e\n    ...\n    \u003ccputune\u003e\n      \u003cshares\u003e4096\u003c/shares\u003e\n      ...\n      \u003cvcpusched vcpus\u003d\"2-3\" scheduler\u003d\"fifo\" priority\u003d\"1\"/\u003e\n    \u003c/cputune\u003e\n    ...\n  \u003c/domain\u003e\n   {{(pid\u003d12600) _get_guest_xml /opt/stack/nova/nova/virt/libvirt/driver.py:6331}}\n\nto what is seen when we enter \u0027_update_numa_xml\u0027 (or via \u0027virsh dumpxml\u0027\nat any point after instance creation):\n\n  DEBUG nova.virt.libvirt.migration [-] _update_numa_xml input xml\u003d\u003cdomain type\u003d\"kvm\"\u003e\n    ...\n    \u003ccputune\u003e\n      \u003cshares\u003e4096\u003c/shares\u003e\n      ...\n      \u003cvcpusched vcpus\u003d\"2\" scheduler\u003d\"fifo\" priority\u003d\"1\"/\u003e\n      \u003cvcpusched vcpus\u003d\"3\" scheduler\u003d\"fifo\" priority\u003d\"1\"/\u003e\n    \u003c/cputune\n    ...\n  \u003c/domain\u003e\n   {{(pid\u003d12600) _update_numa_xml /opt/stack/nova/nova/virt/libvirt/migration.py:97}\n\nThe solution is simple: rather than trying to modify the existing XML,\nsimply scrap it and rebuild the elements from scratch. We should\nprobably do this for all elements, but that can/should be tackled\nseparately.\n\nChange-Id: Ic01603a91f6099f1068af0e955f3e1056021d673\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #1889257\n(cherry picked from commit b6aef1ec4f9848f85e1f367e560c2bdb703fa110)\n(cherry picked from commit ac05bc3b38684e47e35e06a1514866773d8f3609)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/58a928681c49c2aa04a29d2ddb5ad4bf502f429d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/58a928681c49c2aa04a29d2ddb5ad4bf502f429d"}]},"branch":"refs/heads/stable/train"},"95bce095ed8489968e8dec2f0b6977f79bb8f98a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-07-30 19:16:45.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/05/743805/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/05/743805/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/743805/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/743805/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/05/743805/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/05/743805/2"}}},"commit":{"parents":[{"commit":"b971dc82cb524fe86284c95ec671e2bad1c2874f","subject":"objects: Update keypairs when saving an instance","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b971dc82cb524fe86284c95ec671e2bad1c2874f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-07-28 15:22:39.000000000","tz":60},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-07-30 19:16:45.000000000","tz":0},"subject":"Handle multiple \u0027vcpusched\u0027 elements during live migrate","message":"Handle multiple \u0027vcpusched\u0027 elements during live migrate\n\nWhen live migrating a pinned instance, we recalculate pinning\ninformation for the destination host and then update the instance\u0027s XML\nbefore spawning the instance there. As part of the pinning information\nrecalculation, we must also recalculate information for realtime cores,\nwhich are configured using the \u0027\u003cvcpusched\u003e\u0027 element. The\n\u0027nova.virt.libvirt.migration._update_numa_xml\u0027 function, which handles\nthis updating, was assuming there would only be one of these elements.\nThis is a reasonably sane assumption since this is all we create in the\n\u0027nova.virt.libvirt.LibvirtDriver._get_guest_numa_config\u0027 function used\nto generate the initial instance XML. However, a look at logs show that\nat least some (all?) versions of libvirt actually rewrite the XML we\u0027re\nproviding them. Compare what is returned from \u0027_get_guest_xml\u0027:\n\n  DEBUG nova.virt.libvirt.driver [...] [instance: ...] End _get_guest_xml xml\u003d\u003cdomain type\u003d\"kvm\"\u003e\n    ...\n    \u003ccputune\u003e\n      \u003cshares\u003e4096\u003c/shares\u003e\n      ...\n      \u003cvcpusched vcpus\u003d\"2-3\" scheduler\u003d\"fifo\" priority\u003d\"1\"/\u003e\n    \u003c/cputune\u003e\n    ...\n  \u003c/domain\u003e\n   {{(pid\u003d12600) _get_guest_xml /opt/stack/nova/nova/virt/libvirt/driver.py:6331}}\n\nto what is seen when we enter \u0027_update_numa_xml\u0027 (or via \u0027virsh dumpxml\u0027\nat any point after instance creation):\n\n  DEBUG nova.virt.libvirt.migration [-] _update_numa_xml input xml\u003d\u003cdomain type\u003d\"kvm\"\u003e\n    ...\n    \u003ccputune\u003e\n      \u003cshares\u003e4096\u003c/shares\u003e\n      ...\n      \u003cvcpusched vcpus\u003d\"2\" scheduler\u003d\"fifo\" priority\u003d\"1\"/\u003e\n      \u003cvcpusched vcpus\u003d\"3\" scheduler\u003d\"fifo\" priority\u003d\"1\"/\u003e\n    \u003c/cputune\n    ...\n  \u003c/domain\u003e\n   {{(pid\u003d12600) _update_numa_xml /opt/stack/nova/nova/virt/libvirt/migration.py:97}\n\nThe solution is simple: rather than trying to modify the existing XML,\nsimply scrap it and rebuild the elements from scratch. We should\nprobably do this for all elements, but that can/should be tackled\nseparately.\n\nChange-Id: Ic01603a91f6099f1068af0e955f3e1056021d673\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #1889257\n(cherry picked from commit b6aef1ec4f9848f85e1f367e560c2bdb703fa110)\n(cherry picked from commit ac05bc3b38684e47e35e06a1514866773d8f3609)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95bce095ed8489968e8dec2f0b6977f79bb8f98a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/95bce095ed8489968e8dec2f0b6977f79bb8f98a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
