)]}'
{"id":"openstack%2Fnova~747610","triplet_id":"openstack%2Fnova~stable%2Ftrain~If9acc054100a6733f3659a15dd9fc2d462e84d64","project":"openstack/nova","branch":"stable/train","topic":"bug/1841363","hashtags":[],"change_id":"If9acc054100a6733f3659a15dd9fc2d462e84d64","subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","status":"MERGED","created":"2020-08-24 04:08:24.000000000","updated":"2020-09-17 23:37:03.000000000","submitted":"2020-09-17 23:35:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747610-1600385729135-079f7ff7","meta_rev_id":"406c7e8d0e93b98785b55df979c782c5f269cdc3","_number":747610,"virtual_id_number":747610,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-09-17 16:06:14.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-09-16 18:14:46.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"date":"2020-09-17 17:56:44.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-09-17 13:33:02.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-17 23:35:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-09-16 10:11:44.000000000","_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-15 15:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-11 11:25:52.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2020-09-15 15:53:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-24 04:36:01.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-11 11:25:52.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-09-15 15:53:09.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-09-16 18:14:46.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-17 13:33:02.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-17 16:06:14.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-17 17:56:44.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-17 23:35:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ee1501185c3b6527d478833bf2bcd754ccca709","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-08-24 04:08:24.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"419df928bd804888f2ccdd32fd8995de32f373d5","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-08-24 04:08:33.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 75ab585e6d68ed8b66ca988bc480aa4858ffd292","accounts_in_message":[],"_revision_number":1},{"id":"71b1380156dcf35e806f691ad5d071c28357f445","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-08-24 04:28:15.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/10/747610/1/check/pmem-tempest-plugin-filtered-multi-node/150f005/ : RETRY_LIMIT in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5784d03ca8856df1e5247555cfea784fdefdb9f7","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 04:36:01.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/747610/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/290e577 : FAILURE in 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8fa9ec700fe5addbd4aac38a15e6bb55ec8d158d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 06:35:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d2965896756a425584b5eca7670dfe2f : SUCCESS in 1h 00m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/70f9be1e17fb43a283095f6d697e7d7f : SUCCESS in 1h 19m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/864f1939a2424c06af8270fce3b64e04 : SUCCESS in 23m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e333af077359409a8d6cb2448bd99af1 : SUCCESS in 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c462632013ec4b75ae5fff24e30dd2ad : SUCCESS in 10m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e89d1a9ad9d74709bd863541b3f30e41 : SUCCESS in 20m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/298ccfc17b404d6fa4da075f98128e86 : SUCCESS in 13m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bc135c21bd9484c862351945a1d7108 : SUCCESS in 20m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31f69d5ebbe48758ae015d66b6297d0 : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ca7e5f9b7d4447983a9282bef5a27ef : SUCCESS in 11m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7d8398f36f3640168da53a0a3fa2105d : SUCCESS in 1h 15m 42s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/efe59e2adcd546909716977faaee8433 : SUCCESS in 1h 12m 46s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/ea7a7d9663ef4094b5911ac491434daf : SUCCESS in 1h 05m 53s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6e6371c2ba9144d0a1c33cfd8bc1aba8 : SUCCESS in 1h 11m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/df2a29e3aef841c6ad26cbeee2d8cdd5 : SUCCESS in 49m 49s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a87a281322f2410d9f91492703307ea0 : SUCCESS in 48m 42s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3a8cdd65949c4f1096e899dbf2560366 : FAILURE in 1h 15m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/5676a63896be494090c0c3a2b474544b : SUCCESS in 1h 14m 54s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/aa35f9f306df4e37bf3c159e5dcc6d67 : SUCCESS in 26m 14s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a4e0741ca452462c894ea53b3f1479cf : SUCCESS in 21m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7000d421a0d2488ba7b19249fa1dce72 : FAILURE in 2h 24m 47s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/ef288412178b43039f8d704d89eb8d39 : SUCCESS in 56m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273cea98cc084bb88232e9bfe366e81f : SUCCESS in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5266c9ce4fa3d2a8341ce95e7c423f0bc72b4dee","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-08-24 06:55:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/1 : FAILURE in 2h 45m 19s","accounts_in_message":[],"_revision_number":1},{"id":"500b13510e7186eb60f12540213919e0e31f6ba3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 06:55:16.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/1/check/tempest-dsvm-full-bionic/4e89bbc/ : TIMED_OUT in 2h 37m 05s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/1/check/tempest-dsvm-full-bionic-py3/9ee7dab/ : SUCCESS in 1h 23m 19s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/1/check/grenade-dsvm-bionic/4f45ba2/ : POST_FAILURE in 1h 41m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4bdba58637dbd12acd39f7745256e8171cdb0c56","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-08 16:16:37.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"bcc71a13152b43c30e09f6daa4f9d4615131204b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-08 16:20:53.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"bdf4afad439445e70d6968bb39b75bf0df7a02fd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-08 16:30:36.000000000","message":"Topic changed from bug/1841363-stable/ussuri-stable/train to bug/1841363","accounts_in_message":[],"_revision_number":3},{"id":"43421862cbe20c742dfbc459d89ad471af79f63b","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-08 16:53:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/10/747610/3/check/pmem-tempest-plugin-filtered-multi-node/2f72b08/ : RETRY_LIMIT in 4m 29s\n- devstack-networking-ovs-dpdk http://52.27.155.124/10/747610/3/check/devstack-networking-ovs-dpdk/8a93d44/ : RETRY_LIMIT in 2m 27s","accounts_in_message":[],"_revision_number":3},{"id":"26c4289ce9997aaede5cd331de6839624274f7fa","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-08 18:54:06.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/3 : FAILURE in 2h 32m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3624b88bda184d5c17419112e0ec4f99d67e5ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-08 21:32:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/061492644fb14d84a20db181995a24b0 : SUCCESS in 58m 45s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/611328439bc74bf88825ecaf4a900fa9 : SUCCESS in 1h 08m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba0d64dbe85a4066b76b06ba27c4c61c : SUCCESS in 17m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1aea7085c76b435abbdef0c95fef7bc8 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b1aefe1e4ef4e69b10d712ab6d36f19 : SUCCESS in 10m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9b100cb19746470db11993d1b08c4b34 : SUCCESS in 14m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/480a7ef1a49b4175aa99636bc705b9e2 : SUCCESS in 12m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f53e0d37cd4b44e0b054ef4b981227b3 : SUCCESS in 15m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfe92084bca41509d17110060f33f05 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6969ca152b946758e5f7746a8aa1549 : SUCCESS in 17m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e96da1d7d6384255bd0ba98fdc9bbf2a : SUCCESS in 1h 08m 47s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4a12cbf735d846dfb8a804aa67b35155 : SUCCESS in 1h 09m 25s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/345d120b5fc5425f85e1a173110addfc : SUCCESS in 1h 09m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd7083908a1b4b399d1902e4c9690cde : SUCCESS in 1h 08m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e02d52c512f4083a90eb4957c149ea4 : FAILURE in 59m 12s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/111f6f8c85a74ad6b049050da287d6dc : SUCCESS in 53m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5c4b13f7f259451e8587992fcecf5b50 : FAILURE in 1h 25m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/90339831e7ff41e7bef29ecf319bdc8f : SUCCESS in 1h 21m 29s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/b229ffb1c532406885dd4bc080032e6c : SUCCESS in 20m 00s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3922d769fad34cc19c3008e306b085c5 : SUCCESS in 19m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d66ed2296a1b4bd7b99f24a834e1bd39 : SUCCESS in 2h 12m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/26d5851f28d84791ab510c6ad0857995 : SUCCESS in 1h 10m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a86f72bb3a7247c9852cf6b20bb62f2d : SUCCESS in 59m 35s","accounts_in_message":[],"_revision_number":3},{"id":"3b6844cfda6bc1cd4bdc8e2acca99bf9a35e071a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-08 21:54:07.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic/e5ee2a2/ : TIMED_OUT in 2h 39m 57s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic-py3/23948ba/ : SUCCESS in 1h 40m 59s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/grenade-dsvm-bionic/ad15f72/ : POST_FAILURE in 1h 28m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"914c4a8f66d28f69db8e60785cf95eafa05024e5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-10 14:26:48.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec6fddbc48b2900adbcf61267b6a316cca3c5ff7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-10 15:33:29.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/3 : FAILURE in 31m 46s","accounts_in_message":[],"_revision_number":3},{"id":"5579c14dcefbacaa81fc6c95f0653945c8aa46de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 18:06:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/34f41d241d544e8dbdc879b4c7da0b65 : SUCCESS in 1h 04m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7d41855d2a4b475c8d62d4992254a7ff : SUCCESS in 1h 05m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b1c7274a3024136aecbcad1368cc96c : SUCCESS in 19m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a654ba1a8c4a4b75bd1ea96e2263c65b : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05678a2551e346f8a2f3076c150a16f5 : SUCCESS in 12m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e8912c7b0c9421abf26b5071f855748 : SUCCESS in 16m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7484ea4306041a885da30290a1f636e : SUCCESS in 13m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c284326ad9af465c96a2b2e01c5e452b : SUCCESS in 13m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65809ad49f794d08ad3e84dc9bec2e6a : SUCCESS in 9m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af8dfefbc06f4204a710599c3ec4b906 : SUCCESS in 15m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6ac55602625342feb8284b003ccd5e1f : SUCCESS in 1h 18m 48s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/998f22772e4e4085a57435007e21a68d : SUCCESS in 1h 10m 08s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c56b411046e84632a6c30028cb6bfa59 : SUCCESS in 50m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a2dbbc05b4434ec48e85dc9be7ca2efc : SUCCESS in 1h 12m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/063b3d197815436e9e69e297f4acec38 : SUCCESS in 55m 02s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/4e22d55fde01487d9d19d2ff6e44ebd5 : SUCCESS in 1h 03m 45s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f5749a54ef58406e8d9c800df9b609d2 : SUCCESS in 1h 25m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/ed8af2d1de1c4feba8af1b6e6ae1de07 : SUCCESS in 1h 30m 51s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/2493d4c2fddf4f4c83bee345c000dd7d : SUCCESS in 23m 31s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf04524a95a244fdba9aa7e08457fecb : SUCCESS in 23m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3bc9197a41084bc18a264371911d8b69 : SUCCESS in 1h 54m 04s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/586a359bdc3b4c889e6a19794bcf0fe5 : SUCCESS in 57m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8ba776039f2442b90e44d3408af0de5 : SUCCESS in 1h 04m 58s","accounts_in_message":[],"_revision_number":3},{"id":"f017abb1a9bcfe793e8b5e96f763f25f9d7121be","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-10 21:09:31.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic/6997cef/ : TIMED_OUT in 2h 37m 58s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic-py3/9bbacb5/ : SUCCESS in 1h 42m 48s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/grenade-dsvm-bionic/4e18ad3/ : POST_FAILURE in 1h 26m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7dd5414e506afaf32def609b4d093639d797ef7f","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-11 11:25:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"b1f2bd8d84162b60cc0fd16a5f6f2644bd140e0a","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-15 15:53:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nClean cherry pick from Ussuri, where the patch has merged. LGTM.\n\n(I only updated the commit message to have the correct hash in \"cherry picked\" line and then rebased the patch, otherwise the patch is the same clean cherry pick that Norman uploaded)","accounts_in_message":[],"_revision_number":3},{"id":"97501ed2489ec823caf36f25bfa1c61a13f6ba83","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 15:53:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"462e3973939334e9676cebd9a90ae9bf78725080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 18:05:57.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c413fd675d2c4a88a2f1d51b28842715 : SUCCESS in 1h 07m 33s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/312ca5abb477460c8013d0e9e5350364 : SUCCESS in 1h 27m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7c6c7766856482aa8209f5d4df3cff5 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54dad544c3394dbab101867203dc4cb6 : SUCCESS in 10m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50a9084b210a4e9b9f9bef2ba4b00db6 : SUCCESS in 13m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec873cbf36054bfbaeeb8b8f594b2b61 : SUCCESS in 13m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bbabc01cc4b849c58bba85e9e84fa1e3 : SUCCESS in 15m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93bd3ae592024c52961f5c3b65ed09e9 : TIMED_OUT in 32m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fd737fecddc432a9fba612d8fa60c63 : SUCCESS in 15m 15s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb726f6058574381ba4227a0773b83c8 : SUCCESS in 1h 10m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5c6eb93fe7e24d8fabf338c7b220c84b : FAILURE in 25m 55s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/873ff3f7c2d4496c98355556d31dc22e : SUCCESS in 19m 51s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9a23195e15a4441a5e7dd46ffb451d3 : SUCCESS in 18m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5e78e74f826441f69f7b3a1a40256f32 : FAILURE in 23m 13s\n- nova-next https://zuul.opendev.org/t/openstack/build/8a9b3a5b25b047db9eba8188ff714bd8 : SUCCESS in 1h 30m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f281389e20094b4086fa8fb10d6b09c9 : SUCCESS in 1h 57m 15s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3ff358507ef943aeb4e5a39a3f112f28 : FAILURE in 19m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecdfdd31e9bb4b12ac5c7ab6937e45e7 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":3},{"id":"8f9ff16cacb3993ccae0a45c6de6f46f0053692a","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-16 10:11:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0fd83e23f5df6730d8f8927fc14e5eaf0044734d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-16 10:38:53.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/10/747610/3/check/pmem-tempest-plugin-filtered-multi-node/d1fc410/ : RETRY_LIMIT in 4m 23s\n- devstack-networking-ovs-dpdk http://52.27.155.124/10/747610/3/check/devstack-networking-ovs-dpdk/97f3786/ : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a37931c2d51fe12e606bfe2bcea94cea2d32f2ff","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-16 10:58:55.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/3 : FAILURE in 46m 21s","accounts_in_message":[],"_revision_number":3},{"id":"b80b1c0cfdf76233f23c5ac97777018d9c39af37","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-16 11:47:41.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic finger://zuul-v3-executor.novalocal/b53b2ba7a23542ec9ed981db83d40d06 : DISK_FULL in 1h 07m 17s\n- tempest-dsvm-full-bionic-py3 finger://zuul-v3-executor.novalocal/16a6dfd8e70548e89a5a2d8b265ca0c5 : DISK_FULL in 42m 14s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/grenade-dsvm-bionic/f4956fd/ : POST_FAILURE in 1h 20m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e74384660fdaae107e85a6e6d94d7f5426a9cb45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 12:23:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f525977742194f2b9e3bf801eacc4e5c : SUCCESS in 1h 17m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/48c08ea6af9945e8b7c67ba3802879a1 : SUCCESS in 1h 13m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc58fb636e0e4b09b3064908cda8ab4c : SUCCESS in 19m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/574faa19104b4bc48c6da171b704833e : SUCCESS in 12m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c632a0ad6924581babaeb69ecbb83b4 : SUCCESS in 9m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b38fbb891394a79a2302f78ea5a1978 : SUCCESS in 18m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9eb0dea5fbeb425796e7d335f994404e : SUCCESS in 15m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f43c33e267f4b94bc249b399a2f7a5d : SUCCESS in 13m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbbbef838b964206a695cefa6fa0dbb1 : SUCCESS in 16m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1e82ef765ed49b391a8136384241b5b : SUCCESS in 13m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a9451170fa684b87b509b3b201cb0645 : SUCCESS in 1h 15m 47s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7a16c43f51254936901f1a782f3e66b2 : TIMED_OUT in 2h 05m 26s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe00fb0a91624d498f0dcb6c63dec3ab : SUCCESS in 1h 02m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0233a5b248164b748a6c307be9165326 : SUCCESS in 1h 11m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d6748fcc02fd40b2ad92e2085e478dab : SUCCESS in 50m 31s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8531cf0bc5664927ba05e5d22aed7bbb : SUCCESS in 1h 06m 27s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9eb3c5e496f94d0aae8d87ab11cb0df8 : SUCCESS in 1h 23m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/3cfc7126fa3148c991acd611441d9cb3 : SUCCESS in 1h 14m 14s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/af9789d6450a42098f505223299f86f5 : SUCCESS in 20m 14s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b83c2257d045411e893f3692d54ecd64 : SUCCESS in 18m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75901e2029bc422fa75ce1302b6dcf39 : SUCCESS in 1h 53m 14s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b469b4d9aad840b3a1423081879d6690 : SUCCESS in 1h 00m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a997db6d6cc4aa5bca9fb2654410995 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":3},{"id":"8acd6cc29e0c0df4f01d7ba2218a32b62a927636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 12:23:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1dcaf59c5b491542b53504289b7d50c5f01f4880","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 15:29:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9fd0f4044232472387100c471e004221 : SUCCESS in 1h 00m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7da06e9f8ae84b82b4a394c62f1de473 : SUCCESS in 1h 22m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51aa1b40dbe44c268f393fcf4c284615 : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ba67677d47473db9a24a38e7a4e5c9 : SUCCESS in 13m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c44650ff5f3a47ce8f777074e4faa46b : SUCCESS in 17m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2439f55491ab4ae392e6b58429c55d80 : SUCCESS in 15m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/329bc1f39455463ab4fe321b08fe5e0e : SUCCESS in 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fea1c85ef8034af1bc099daeed067f65 : SUCCESS in 23m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c358ccf33e64473fa88534980372592f : SUCCESS in 13m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7f1e1bafc20f4ce4a38333cec1cdf9e1 : SUCCESS in 1h 09m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e54040da85d24ac291d9254fe2ec6dcf : SUCCESS in 54m 05s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/e3152cffa0f2483f84823593f3c1e195 : SUCCESS in 21m 50s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cb9ce596f5545bd9cce6730f35eb00e : SUCCESS in 17m 35s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4589e968ebc84e3696dc6f4ccc697107 : SUCCESS in 1h 22m 00s\n- nova-next https://zuul.opendev.org/t/openstack/build/f65eb91ae98c4baf9a952dd0761f9478 : FAILURE in 1h 29m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/009d2aa0f07f431ca3f59dfdfc3279ac : SUCCESS in 2h 04m 37s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/816239e1a5f34520a66b73403be62256 : SUCCESS in 1h 42m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/89def030c683447fbab9c238bb581f09 : SUCCESS in 54m 29s","accounts_in_message":[],"_revision_number":3},{"id":"12269849ca341be45ba99e56857a4467fd159705","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-16 18:14:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4d2a67afad7b431f42acfd25c92d0c08b8a6a26f","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-16 19:24:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/10/747610/3/check/pmem-tempest-plugin-filtered-multi-node/cc10989/ : RETRY_LIMIT in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"9869f7a6527fd9a6abeaf89180a7d4bd26febf42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:14:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9bd88dd6cf854c1abe482dfc188e0939 : SUCCESS in 51m 43s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3d1d4f63016244eb89fdeb8e23158732 : SUCCESS in 1h 23m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dacc124dd4844e59965ef9753583e1f1 : SUCCESS in 23m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e2631d68fd44fa991394f91b4b0306f : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e441d8185df4804b10f797e9ad4b6dc : SUCCESS in 14m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7b95af7e0a54813a150587f297aedcb : SUCCESS in 21m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abbfa86899314cabaa715bd9e566f590 : SUCCESS in 14m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6531751595794d0b8870b23c7fb59dc8 : SUCCESS in 14m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a0b4f2522764e499f9d87221bd36384 : SUCCESS in 32m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aaa350bf5b49422290028e71e63c2cad : SUCCESS in 14m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0d3f8b1b555b4552a6dc4e91a8458c95 : SUCCESS in 1h 05m 33s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/fa55bf661c5b4d79b7898cdac6986d3a : SUCCESS in 1h 06m 13s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/109b8d678f68422eab6721072f69d858 : SUCCESS in 1h 17m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1ab07d4df84540cc847ea164f02653f5 : SUCCESS in 1h 05m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/464d2a13517c4dbd9b57ad8192d1a6d6 : SUCCESS in 52m 13s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2d07c0d78dfc4cf1badf63494887ca24 : SUCCESS in 44m 33s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2cdf2caadf0e4285afdb7eeda05d8e98 : SUCCESS in 1h 14m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/b16f98fb129747a4b158db15c6dd5ac1 : SUCCESS in 1h 10m 14s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/677c6b4e56b04958bf54b136792a273c : SUCCESS in 20m 56s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c52d01ad5ac94e5daae50b3601052ae4 : SUCCESS in 16m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/57c23d82418946e7ba5e1f8687136282 : SUCCESS in 1h 56m 21s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/69f8424f8ebe4762911b3f090a2ae800 : SUCCESS in 59m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a6049bf29b04f40bef5c5ca102f3c2d : SUCCESS in 50m 51s","accounts_in_message":[],"_revision_number":3},{"id":"a109204761ed29a90b931b4bdac75cbb4437ed9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 20:20:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1ce70fb1c02a2bd48a9ac589d706c8332f86561e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-16 20:54:07.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/3 : FAILURE in 2h 38m 36s","accounts_in_message":[],"_revision_number":3},{"id":"05860adb1782f4e161c360774fc670d35706a9ab","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-16 22:03:18.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic/b52b604/ : TIMED_OUT in 2h 41m 45s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic-py3/574ab9d/ : SUCCESS in 1h 40m 10s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/grenade-dsvm-bionic/85ad9d7/ : POST_FAILURE in 1h 07m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8d7ecba923afc5c745e7c7d67592bc869d868fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 00:28:01.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bfd44d88b1e64c7eb69e18919e2a8a37 : SUCCESS in 43m 40s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5f18d87b698748f49b43cf02e6591d21 : SUCCESS in 1h 16m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83ad73be435b4531a31c940d03203009 : SUCCESS in 21m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/950f7efe844e4e09906ac4eced69d94e : SUCCESS in 10m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b36c6d72ab02435ea69949c8a6746842 : SUCCESS in 19m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aafdcec7528640c18e7606c9b0e8756b : SUCCESS in 12m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ac361ae87d024bb48aef3604cddcab40 : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50d5002efdaf44a9a7c48d5b8662cd2f : TIMED_OUT in 31m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e35b459366f4ec9bb55b626cceb88cd : SUCCESS in 15m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb07b758751c4329bb8ae2f4beb5a684 : SUCCESS in 1h 11m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e0f4862cdc97486da12d11ac844407ed : SUCCESS in 54m 20s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/737876e7ae2a47eb92a23649219d6e51 : SUCCESS in 19m 54s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd9db2c7dbf44f6298bf8a0ab96d4f7f : SUCCESS in 20m 43s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4409a88bd8794824af77f9f9e0f8e2f1 : SUCCESS in 1h 22m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/37f1847afad6426f8910edb4d0b2bf04 : SUCCESS in 1h 26m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/087d049c651341e6857ec54599633d0f : SUCCESS in 2h 09m 01s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/a288f22b343f45098ff0b579160806b3 : SUCCESS in 1h 05m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a3f61ea43404d74be0729b6bcc17874 : SUCCESS in 1h 09m 26s","accounts_in_message":[],"_revision_number":3},{"id":"da9e83196d70fcff03ab3d7a68a25b1ece3481e2","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-17 07:58:35.000000000","message":"Patch Set 3:\n\nrecheck - tox-docs job timeout","accounts_in_message":[],"_revision_number":3},{"id":"faee883bb83184887fc9a291181275438e7bb426","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-17 08:15:33.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/10/747610/3/check/pmem-tempest-plugin-filtered-multi-node/bb707d4/ : RETRY_LIMIT in 4m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e7c3646f0adcdfdc343a101ca4946f5410c3b251","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-17 10:48:10.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/3 : FAILURE in 2h 48m 50s","accounts_in_message":[],"_revision_number":3},{"id":"595a7bf0429dd88572f9c1702866f272baab59f6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-17 11:01:04.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic/7b37a93/ : TIMED_OUT in 2h 39m 20s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic-py3/a07624c/ : SUCCESS in 1h 27m 26s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/grenade-dsvm-bionic/139148c/ : POST_FAILURE in 1h 29m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a3f6013bd4abc39b568620147f965ca01a8260b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 13:00:37.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4f2aa5dde7d455ebd1d134a356b3c78 : SUCCESS in 1h 00m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1cbbdf9d8b4a4e148cd02a3e4f177c54 : SUCCESS in 1h 14m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba428e7ecb694f1ea34ed646b7368713 : SUCCESS in 15m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f7a9a4ac8ac409ebcf2a44bfeacfb84 : SUCCESS in 15m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/309ca886748f4b449c2f6cd251e76aa8 : SUCCESS in 12m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/25b3cfe9a1644290added5f0b377e3c5 : SUCCESS in 15m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3cb6073d53d4fecbd6ad469c336db4b : SUCCESS in 13m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/938c67ab571a49d380258ca9c06c71e0 : SUCCESS in 19m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738d967c5ee9400596a8ec4aafd94ff1 : SUCCESS in 15m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53e31415419a4ed19ccb1ba914220db9 : SUCCESS in 30m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1b579bdcbdea4f448df82a953cee6479 : TIMED_OUT in 1h 35m 16s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/5f19ae9e147f4d6ab0ab2be583ad2ba7 : TIMED_OUT in 2h 01m 49s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d59543061c241108274b1fae3764fce : SUCCESS in 1h 09m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6eed888ef23849b788a9a5527feacf83 : FAILURE in 53m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2426f511cf4a461085fe18d655413ccf : SUCCESS in 55m 36s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8b289c92303149fa92c4af806a4bb8f4 : SUCCESS in 55m 02s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4ca2b9412537454ea0715cb00589d730 : SUCCESS in 1h 30m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/a87301017acb4b178012b2984ead4c1d : SUCCESS in 1h 16m 41s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/3732c4a6e2004e7692849409bab1ebf3 : SUCCESS in 19m 56s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22b3be3f72c94bc6869d22b1f13e6689 : SUCCESS in 19m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2655bfb3a704c85a56215be19afbeff : SUCCESS in 2h 10m 45s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/f1a79a54d3d741dea0be5f045297ca5b : SUCCESS in 1h 01m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0048681f1a24f7890998305f988e1b8 : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":3},{"id":"375864ff35dd9a99e3fc4ca205c4bc24adb1501e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-17 13:14:58.000000000","message":"Patch Set 3:\n\nrecheck - socket.timeout","accounts_in_message":[],"_revision_number":3},{"id":"082ae16634c4ff9734cdd2978e0c67609e7cb949","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-17 13:33:02.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/10/747610/3/check/pmem-tempest-plugin-filtered-multi-node/322d7b3/ : RETRY_LIMIT in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"ea6e25b166ea7025436e446e375e542155523bb3","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-17 16:06:14.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747610/3 : FAILURE in 2h 50m 29s","accounts_in_message":[],"_revision_number":3},{"id":"4ff59174690b3d149e040ca392b43c562e61e006","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-17 17:56:44.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic/5118292/ : TIMED_OUT in 2h 37m 55s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/tempest-dsvm-full-bionic-py3/94e9718/ : SUCCESS in 1h 40m 36s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/747610/3/check/grenade-dsvm-bionic/d5824c9/ : POST_FAILURE in 33m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"14374f10dbd9c9d6c1fdbf8e712cb27c3fe9c533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:29:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/160dd97dd15d49e2b89ac06de89da3af : SUCCESS in 1h 03m 31s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b6a4ddcfa4e84104bceb262bc430c19f : SUCCESS in 1h 33m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/379e4a7a35e54c0e8ac58183343d82c4 : SUCCESS in 16m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7411ab4def0d4a7bb9a69aa6fda8ccd9 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbdb35005c4a4ea39f20e28b341d7723 : SUCCESS in 15m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/975596b01fe94aa6ac26a82efc52297c : SUCCESS in 21m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34c4df8f0294574b789cf2dd671cc38 : SUCCESS in 13m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e7ea776e063f43288186d4fcc41d5813 : SUCCESS in 15m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0696429bbb14cdf866e819f6d054ff8 : SUCCESS in 17m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fa2578fde0a45baa0056ff5eb4132ae : SUCCESS in 20m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/66f7dbf733244b57b02bb7ff5fca4d34 : FAILURE in 1h 32m 41s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/4aa536e96f9b4ae48cceb030308986c7 : SUCCESS in 1h 11m 54s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a92a0b934bd44cfc9f28447a5245badd : SUCCESS in 1h 12m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a28fa3d8412b4a0eae99357a01ba4efd : SUCCESS in 1h 14m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b926985383d141b6bcf6b3fa534c534a : SUCCESS in 49m 40s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/849df27a4c604a5892376d0d7cd6830f : SUCCESS in 52m 25s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/59c7a316071d437e94b6e436feca9b02 : SUCCESS in 1h 23m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/7bca0d6f843e4a3a8931fbe6399c7ee5 : SUCCESS in 1h 09m 40s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/106cc6662a5248489e2759e8bf635481 : SUCCESS in 23m 18s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ba2ef8f293834054838ba10ef4df61e7 : SUCCESS in 14m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9098a25473f0460992f2d32b2878c80d : SUCCESS in 1h 49m 38s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/80484216f69c4d36b24a796f398294e1 : SUCCESS in 53m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f0c962367434295952810027159268d : SUCCESS in 53m 19s","accounts_in_message":[],"_revision_number":3},{"id":"4aa114c401c292f8acdc927fa6f5ab2d8a861049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 19:30:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a63593dbdb2817ef3fca494684499728bd320a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 23:35:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f936378288d4863b57497797814a2fa : SUCCESS in 50m 50s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/31a52d23f0524bcf9b92d1f76250d985 : SUCCESS in 1h 18m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/68fb99e677df45478f64224ce17265a6 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6eccb10b5fb455bac29e0b584e2c669 : SUCCESS in 16m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b606ade829f94bf5be4903d7ef281400 : SUCCESS in 12m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4cc053055bf04549b744f7b3e8bf0f0e : SUCCESS in 11m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/32c876fa44f64fabb5107fafcb637723 : SUCCESS in 13m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd779ea68537439bb388fbf8ad14f2fa : SUCCESS in 16m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c108b00086ae4d42ab8ebb25350ede2d : SUCCESS in 18m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6be27274eb6e48b6ae73b169a88ad97a : SUCCESS in 1h 23m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/527fc7ac96d2455793d5d138a2d39f7a : SUCCESS in 56m 34s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/025b700204b74af1a9fe81d7b02385cc : SUCCESS in 19m 36s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5866e1d9557246c1bccacac63b5c15b4 : SUCCESS in 20m 04s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/70bcadd501e2452a8c138ed9662da2c9 : SUCCESS in 1h 22m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/5c25526ea56a4341836153a8f522faf3 : SUCCESS in 1h 12m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdc215bdb9c440cf8ab7501250a8fed5 : SUCCESS in 2h 24m 31s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/8405ee004d844d04bdefd628456c4f52 : SUCCESS in 59m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ed9f4b3327245c09a0a2f52e1e3926b : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":3},{"id":"1fba8367ad0387d26154a96a130bdbcb2ed0c077","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 23:35:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"406c7e8d0e93b98785b55df979c782c5f269cdc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 23:37:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c48f70b12acf4e3fbb803dcf5a287787 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0abfc74f70d4fd2871115881a5234ee : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d92fe4f3e69db51b7bbad52a1fe1740e37bade9a","revisions":{"c3feac11299129de202a81b1c9e0ab29114d32e8":{"kind":"REWORK","_number":1,"created":"2020-08-24 04:08:24.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/10/747610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/747610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/747610/1"}}},"commit":{"parents":[{"commit":"2442344fa1afd8d0a4c58b4df35f1b896129a0a1","subject":"catch libvirt exception when nodedev not found.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2442344fa1afd8d0a4c58b4df35f1b896129a0a1"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-09-17 11:08:59.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-08-24 04:08:24.000000000","tz":0},"subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","message":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available\n\nBecause of the libvirt issue[1], there is a bug[2] that if we set cache mode\nwhose write semantic is not O_DIRECT (.i.e unsafe, writeback or writethrough),\nthere will be a problem with the volume drivers\n(.i.e nova.virt.libvirt.volume.LibvirtISCSIVolumeDriver,\nnova.virt.libvirt.volume.LibvirtNFSVolumeDriver and so on), which designate\nnative io explicitly.\n\nThat problem will generate a libvirt xml for the instance,\nwhose content contains\n\n```\n...\n\u003cdisk ... \u003e\n  \u003cdriver ... cache\u003d\u0027unsafe/writeback/writethrough\u0027 io\u003d\u0027native\u0027 /\u003e\n\u003c/disk\u003e\n...\n```\nIn turn, it will fail to start the instance or attach the disk.\n\n\u003e When qemu is configured with a block device that has aio\u003dnative set, but\n\u003e the cache mode doesn\u0027t use O_DIRECT (i.e. isn\u0027t cache\u003dnone/directsync or any\n\u003e unnamed mode with explicit cache.direct\u003don), then the raw-posix block driver\n\u003e for local files and block devices will silently fall back to aio\u003dthreads.\n\u003e The blockdev-add interface rejects such combinations, but qemu can\u0027t\n\u003e change the existing legacy interfaces that libvirt uses today.\n\n[1]: https://github.com/libvirt/libvirt/commit/058384003db776c580d0e5a3016a6384e8eb7b92\n[2]: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1086704\n\nCloses-Bug: #1841363\nChange-Id: If9acc054100a6733f3659a15dd9fc2d462e84d64\n(cherry picked from commit af2405e1181d70cdf60bcd0e40b3e80f2db2e3a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3feac11299129de202a81b1c9e0ab29114d32e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c3feac11299129de202a81b1c9e0ab29114d32e8"}]},"branch":"refs/heads/stable/train"},"1c75b519e0e7daf6330c7a754ed2a82e7583d0e6":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-08 16:16:37.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/747610/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/747610/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/747610/2"}}},"commit":{"parents":[{"commit":"2442344fa1afd8d0a4c58b4df35f1b896129a0a1","subject":"catch libvirt exception when nodedev not found.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2442344fa1afd8d0a4c58b4df35f1b896129a0a1"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-09-17 11:08:59.000000000","tz":480},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-09-08 16:12:27.000000000","tz":120},"subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","message":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available\n\nBecause of the libvirt issue[1], there is a bug[2] that if we set cache mode\nwhose write semantic is not O_DIRECT (.i.e unsafe, writeback or writethrough),\nthere will be a problem with the volume drivers\n(.i.e nova.virt.libvirt.volume.LibvirtISCSIVolumeDriver,\nnova.virt.libvirt.volume.LibvirtNFSVolumeDriver and so on), which designate\nnative io explicitly.\n\nThat problem will generate a libvirt xml for the instance,\nwhose content contains\n\n```\n...\n\u003cdisk ... \u003e\n  \u003cdriver ... cache\u003d\u0027unsafe/writeback/writethrough\u0027 io\u003d\u0027native\u0027 /\u003e\n\u003c/disk\u003e\n...\n```\nIn turn, it will fail to start the instance or attach the disk.\n\n\u003e When qemu is configured with a block device that has aio\u003dnative set, but\n\u003e the cache mode doesn\u0027t use O_DIRECT (i.e. isn\u0027t cache\u003dnone/directsync or any\n\u003e unnamed mode with explicit cache.direct\u003don), then the raw-posix block driver\n\u003e for local files and block devices will silently fall back to aio\u003dthreads.\n\u003e The blockdev-add interface rejects such combinations, but qemu can\u0027t\n\u003e change the existing legacy interfaces that libvirt uses today.\n\n[1]: https://github.com/libvirt/libvirt/commit/058384003db776c580d0e5a3016a6384e8eb7b92\n[2]: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1086704\n\nCloses-Bug: #1841363\nChange-Id: If9acc054100a6733f3659a15dd9fc2d462e84d64\n(cherry picked from commit af2405e1181d70cdf60bcd0e40b3e80f2db2e3a6)\n(cherry picked from commit 0bd58921a1fcaffcc4fac25f63434c9cab93b061)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c75b519e0e7daf6330c7a754ed2a82e7583d0e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1c75b519e0e7daf6330c7a754ed2a82e7583d0e6"}]},"branch":"refs/heads/stable/train"},"d92fe4f3e69db51b7bbad52a1fe1740e37bade9a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-08 16:20:53.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/747610/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/747610/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/747610/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/747610/3"}}},"commit":{"parents":[{"commit":"5cebdb7b718bf7765de9257b49448868ba673af5","subject":"Merge \"Add checks for volume status when rebuilding\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5cebdb7b718bf7765de9257b49448868ba673af5"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-09-17 11:08:59.000000000","tz":480},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-09-08 16:19:19.000000000","tz":120},"subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","message":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available\n\nBecause of the libvirt issue[1], there is a bug[2] that if we set cache mode\nwhose write semantic is not O_DIRECT (.i.e unsafe, writeback or writethrough),\nthere will be a problem with the volume drivers\n(.i.e nova.virt.libvirt.volume.LibvirtISCSIVolumeDriver,\nnova.virt.libvirt.volume.LibvirtNFSVolumeDriver and so on), which designate\nnative io explicitly.\n\nThat problem will generate a libvirt xml for the instance,\nwhose content contains\n\n```\n...\n\u003cdisk ... \u003e\n  \u003cdriver ... cache\u003d\u0027unsafe/writeback/writethrough\u0027 io\u003d\u0027native\u0027 /\u003e\n\u003c/disk\u003e\n...\n```\nIn turn, it will fail to start the instance or attach the disk.\n\n\u003e When qemu is configured with a block device that has aio\u003dnative set, but\n\u003e the cache mode doesn\u0027t use O_DIRECT (i.e. isn\u0027t cache\u003dnone/directsync or any\n\u003e unnamed mode with explicit cache.direct\u003don), then the raw-posix block driver\n\u003e for local files and block devices will silently fall back to aio\u003dthreads.\n\u003e The blockdev-add interface rejects such combinations, but qemu can\u0027t\n\u003e change the existing legacy interfaces that libvirt uses today.\n\n[1]: https://github.com/libvirt/libvirt/commit/058384003db776c580d0e5a3016a6384e8eb7b92\n[2]: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1086704\n\nCloses-Bug: #1841363\nChange-Id: If9acc054100a6733f3659a15dd9fc2d462e84d64\n(cherry picked from commit af2405e1181d70cdf60bcd0e40b3e80f2db2e3a6)\n(cherry picked from commit 0bd58921a1fcaffcc4fac25f63434c9cab93b061)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d92fe4f3e69db51b7bbad52a1fe1740e37bade9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d92fe4f3e69db51b7bbad52a1fe1740e37bade9a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
