)]}'
{"id":"openstack%2Fnova~747612","triplet_id":"openstack%2Fnova~stable%2Frocky~If9acc054100a6733f3659a15dd9fc2d462e84d64","project":"openstack/nova","branch":"stable/rocky","topic":"bug/1841363","hashtags":[],"change_id":"If9acc054100a6733f3659a15dd9fc2d462e84d64","subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","status":"MERGED","created":"2020-08-24 04:08:42.000000000","updated":"2021-06-17 07:20:45.000000000","submitted":"2021-06-17 07:18:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747612-bug/1841363","meta_rev_id":"600510f66fdc4f936c58d85a8e048b88a863d529","_number":747612,"virtual_id_number":747612,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-12-01 20:04:05.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2021-06-10 15:48:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2020-09-25 01:01:07.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 07:18:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2021-06-17 02:24:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2020-12-01 19:53:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2021-06-17 02:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 11:26:55.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-09-11 13:26:25.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 00:37:04.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-25 01:01:07.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-25 01:50:46.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-25 02:19:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 19:53:38.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-06-17 02:24:09.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"}],"messages":[{"id":"99256f7139e3ae1bc638cee09a3362a707be0167","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-08-24 04:08:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/stein.","accounts_in_message":[],"_revision_number":1},{"id":"8f059fa3f035532f634b7f4ded9ab7c82922f049","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-08-24 04:23:41.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747612/1 : FAILURE in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d903f0f68bb9e9bb06adbd74e689ed62f7ecf6fe","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-08-24 04:26:53.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/12/747612/1/check/pmem-tempest-plugin-filtered-multi-node/b79f32b/ : RETRY_LIMIT in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"faad0cf9962de5f6fc75bc9e82c093e0123ac76e","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-24 04:36:29.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/747612/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/36b8e1c : FAILURE in 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"7b18d53aa1d6b4d4e06609f9f9d6f116a87070c0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-24 05:43:31.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/1/check/tempest-dsvm-full-bionic/667b299/ : FAILURE in 21m 50s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/1/check/tempest-dsvm-full-bionic-py3/7bf8138/ : SUCCESS in 1h 25m 20s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/1/check/grenade-dsvm-bionic/0253ce3/ : POST_FAILURE in 43m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1fde88d6e43165e55afd4fd1c63d30ba9e82706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-24 05:52:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e81672e93c48a7abed583263792fe0 : SUCCESS in 10m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2209e18b52214d26a41d414ded739ddb : SUCCESS in 19m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a004b275d8304767b61a438500ba2b2b : SUCCESS in 12m 54s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9e6250d028284107b85aa49aaa7d469d : SUCCESS in 11m 12s\n- tempest-full https://zuul.opendev.org/t/openstack/build/143049f781d94eb6935b8a11ab4c9451 : SUCCESS in 1h 18m 25s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8b338a766e2f4d66a733516ad5d43788 : SUCCESS in 53m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83b1338bcbbe4bc499ff67b2c4c7a95f : SUCCESS in 1h 22m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fa0c118b20147cf8ebf9471936850b2 : SUCCESS in 11m 04s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/557f10359e134954bcfe44c2e83525ee : SUCCESS in 33m 26s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/ec75895012314d69b4a316deb18f54fa : SUCCESS in 56m 54s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/1c0fcba8dc7d43b8b66fd3b0c9bc8efd : SUCCESS in 1h 07m 38s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e3b6105d155c4b6383710208f303ad28 : SUCCESS in 38m 32s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1c8f7883615b4a0fb579cb1269c5d83c : SUCCESS in 49m 00s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/491bb6e8cab246068a897f2eeb3f2ac6 : SUCCESS in 1h 02m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/4a6ca485ced64b11a755315cf9e32fe0 : SUCCESS in 1h 21m 41s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a3305418456e45c4a3fc6160e38e7659 : SUCCESS in 1h 40m 38s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/6c465d2f6bcf4dc390dc4ef6e01e4759 : SUCCESS in 16m 51s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/57b5cc6bfe8e4d8dae6b729a562e9c1d : SUCCESS in 14m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f41e91c7251f45648f9a93f5d0be5558 : SUCCESS in 13m 01s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1272bb2fe8414caa8452c91dd6cf29fd : SUCCESS in 52m 02s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/04a1b5d25d09432cbfea417e1b574c79 : SUCCESS in 1h 06m 15s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/7610672d9ad94485b64c66d94d45b7f0 : SUCCESS in 1h 18m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"60628074bcf50f0534e2cb60c2dc0b3c0669f4d0","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-11 11:26:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"514ec41f0a01913ce93d0c86a49829a4ee8f641a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-11 11:52:34.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747612/2 : FAILURE in 24m 01s","accounts_in_message":[],"_revision_number":2},{"id":"64dbd645ae4f4d469465a5d5276e24b4fbb93e09","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 12:15:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/12/747612/2/check/pmem-tempest-plugin-filtered-multi-node/6a40003/ : RETRY_LIMIT in 4m 40s\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpcx6c7nhw/a431d85c2c554d578038efcdbe7140b9/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":2},{"id":"b9b0365efb400208a9759ce30893680abfa8a78c","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-09-11 13:26:25.000000000","message":"Patch Set 2:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/747612/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/e8a2f3d : FAILURE in 27m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9b1f10af9c53fea148ab0741fbc1d65c41eae48c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 14:28:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35259f0cc8404fc6af399b8f8305141f : FAILURE in 12m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6c5cfb4bf2c4aab87a10755128c1e8e : SUCCESS in 11m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/eaabc7a977794bc49da31161a9e77aad : SUCCESS in 16m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/8d9bd558b92e4ebbb193b8d9e7fc8a10 : SUCCESS in 10m 56s\n- tempest-full https://zuul.opendev.org/t/openstack/build/210dadafa3974f2e8ebecb1a59fd5894 : SUCCESS in 1h 30m 20s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/91f465c85e504c60a677d02184fa4bfa : SUCCESS in 1h 01m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d145b1bdb76345c5b2fe116fd8983fa3 : SUCCESS in 1h 26m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd5d19adbed34a9cbc02d308c79c40a3 : SUCCESS in 11m 12s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b4b84a56997d4251acda428e17db22f1 : SUCCESS in 41m 33s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/e6cd7e725415454b933cfff626b6e39e : SUCCESS in 1h 01m 33s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/d039ba2276d141e28a969b72ff743b6a : SUCCESS in 1h 09m 22s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1831b0078858485ab3741a5e667fb946 : SUCCESS in 46m 11s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/947fc76de6fe4dd786afa079cd036b0a : SUCCESS in 50m 54s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/8dbd950041774e098b8ac5645f5e1d07 : SUCCESS in 1h 03m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/3671528f624c41988883b4542c7bc8bf : SUCCESS in 1h 51m 28s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/daa13e0b0ec345198a8d1468651c1db9 : SUCCESS in 2h 05m 03s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/19cf5481d3734b87bd6a92e5d963b2d5 : SUCCESS in 12m 23s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/4dffbb4cd84b4f92963af996006be279 : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a46c5242adea455b918758e391d8762a : SUCCESS in 10m 27s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8eb2eacbd864beea6d0c834a5340b65 : SUCCESS in 1h 18m 11s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1a3be989919e4cef88e9d6841595b281 : SUCCESS in 58m 20s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/86d319e45ffb4fad92ae788371ca9cd1 : SUCCESS in 1h 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b27d37811703aed386a1183e9fb0efd519ddf8af","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-11 17:50:38.000000000","message":"Patch Set 2:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/2/check/tempest-dsvm-full-bionic/6b23994/ : FAILURE in 26m 47s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/2/check/tempest-dsvm-full-bionic-py3/0d55bf6/ : SUCCESS in 1h 23m 44s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/2/check/grenade-dsvm-bionic/ded840b/ : POST_FAILURE in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a210fedb8c8f0e265cfb175fdb60f7b7606c7af8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-17 14:08:53.000000000","message":"Topic changed from bug/1841363-stable/ussuri-stable/train-stable/stein-stable/rocky to bug/1841363","accounts_in_message":[],"_revision_number":2},{"id":"4d7213ba569025dafa35563c7f074bac655920ac","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2020-09-25 00:26:52.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2ba9de772670fca0f1a13b1e836fdeb07e457ade","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-25 00:37:04.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747612/3 : FAILURE in 9m 21s","accounts_in_message":[],"_revision_number":3},{"id":"07915964543fbaae5cd5504ed42804d604164111","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-25 01:01:07.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/12/747612/3/check/pmem-tempest-plugin-filtered-multi-node/080d212/ : POST_FAILURE in 23m 04s\n- devstack-nfv-numa http://52.27.155.124/12/747612/3/check/devstack-nfv-numa/cdf1666/ : POST_FAILURE in 32m 43s","accounts_in_message":[],"_revision_number":3},{"id":"d8754bcf865d8b27b23cac3d53751ef1ba6af730","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-25 01:50:46.000000000","message":"Patch Set 3:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/3/check/tempest-dsvm-full-bionic/a68b85d/ : FAILURE in 18m 47s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/3/check/tempest-dsvm-full-bionic-py3/aa21fa4/ : SUCCESS in 1h 18m 13s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/3/check/grenade-dsvm-bionic/3d54763/ : POST_FAILURE in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"22a68b577976645459c728147788966e032d0153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 02:19:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61627e4f4d04c9bb1bdaa886785404e : FAILURE in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8725959367cd462abf06c96b15124738 : SUCCESS in 13m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/38997bdeb7c540b39e416cf462539d10 : SUCCESS in 14m 39s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/729faac9ace245dcbc7d5b977f0d8a4c : SUCCESS in 9m 37s\n- tempest-full https://zuul.opendev.org/t/openstack/build/1dc2d3b5e6694ae48958d6f05bc172f7 : SUCCESS in 1h 29m 19s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/5beb6e4b9e6d4ccea5bc0de9cdbc1a3b : SUCCESS in 46m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/90465709faac4429bf1956e8d354064c : SUCCESS in 1h 19m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c0d05a6bbfc4fa7a737a62cb41fe118 : SUCCESS in 10m 59s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f7cd23cced49417f9f1d0641f54d4651 : SUCCESS in 41m 05s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/cac8e091bbcc4b1cbc705e7f4f562f4f : SUCCESS in 1h 06m 47s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/4c27aa1dfda64abea84f2c00d29537be : SUCCESS in 1h 34m 35s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/071be2e397264e56ad79b95fd5e82662 : SUCCESS in 1h 19m 50s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/fcc07a08f42f4c4daedf15332bacbff8 : SUCCESS in 55m 01s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/55e323ac05614fab8ccce5e56444a437 : SUCCESS in 1h 16m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/3814508ab54146cfbee57e03f7583dc3 : SUCCESS in 1h 22m 35s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/4e1cd5473ab442f6828a15c5b6314182 : SUCCESS in 1h 50m 24s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/2cdfdcd2fe44417cb8b4af4ee085e943 : SUCCESS in 17m 31s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/7e3a636d88c94069b209327da19cbaf1 : SUCCESS in 18m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/13d46cb33a0e4f4ca316a15819e5f6d4 : SUCCESS in 12m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9af2667d2f484227a0e67cae81cf5846 : SUCCESS in 1h 01m 32s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/caa247e5ec764190a3471cafb8bb0278 : SUCCESS in 1h 11m 11s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ad5a0cc294914a61a6f5f3d40f432c5b : SUCCESS in 1h 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f9b74e17b3fe96bd175b156c1e927a1f5d932091","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-01 19:53:38.000000000","message":"Patch Set 3: Code-Review+2\n\nAlready backported and merged in Stein, clean cherry pick. LGTM.","accounts_in_message":[],"_revision_number":3},{"id":"7ca2222d02870ccc39143e5e0d20dde504db7680","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-12-01 19:54:09.000000000","message":"Patch Set 3:\n\nrecheck - merged in Stein, so pep8 (cherry-pick-check) should pass now.","accounts_in_message":[],"_revision_number":3},{"id":"bf054676f75003481e62a328052176e3ea12e3a5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-12-01 20:04:05.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/747612/3 : FAILURE in 8m 51s","accounts_in_message":[],"_revision_number":3},{"id":"dd59f54183fabe1d3f971325153708f1be4fbd38","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-01 22:28:22.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/3/check/tempest-dsvm-full-bionic-py3/eab8ede/ : SUCCESS in 1h 34m 03s","accounts_in_message":[],"_revision_number":3},{"id":"eb5f14a92bbd4b4a0d9723445be970ac76b95356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-02 04:06:18.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a63ad4ad3f0a47e186d790df4d02ca3b : SUCCESS in 10m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9f46d91a8b646259e3844f662a9776a : SUCCESS in 13m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/94ebe457faf24af4b9bbba5599a175e6 : SUCCESS in 15m 12s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/eea1d093890d47cb943956a99ae26b77 : SUCCESS in 9m 38s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0f7408ef5c884d6c80c00c707047796c : SUCCESS in 1h 32m 48s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/04f56c7b5e01424980a05ec7d78548c5 : SUCCESS in 1h 03m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a68970d896a404baab4cef5a5bad408 : SUCCESS in 1h 21m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ebf1d3d3ca44f19aa58eea8e4ba31ed : SUCCESS in 21m 09s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/08a27c8a908a4329a2018e23e1ca3ea0 : SUCCESS in 51m 01s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/96685340738d4f659dfc02926e957ee5 : FAILURE in 55m 04s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/38a21cd0d2e142f9b9d7749646a6e1c6 : SUCCESS in 1h 51m 01s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ff444ff59e7444048e8740b3482b3603 : SUCCESS in 43m 09s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/da5d202f04574312b546a8f8ddb0807d : SUCCESS in 55m 26s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/d687dcd8e0d74e3993d3cc1b644c5846 : SUCCESS in 54m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/1205e4fb77b5448da29c46266ebd6583 : FAILURE in 1h 18m 15s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/7f4a11b40c3d438ebc01e2ac2ebf4bc0 : SUCCESS in 1h 58m 26s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/e4e97bda91a84e60a3957436e52b5cb7 : SUCCESS in 19m 36s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/99fa23522a7549eb9d54c66f41468d27 : SUCCESS in 18m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/304b91c83c9c45c7a313dd269556a585 : SUCCESS in 14m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d11b4ba07a554321a1bfc8d96fc2424b : SUCCESS in 1h 00m 21s\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/a7797aa34c394b1a81a565f9e1d3f900 : SUCCESS in 1h 04m 05s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d52ba813618148e6b3673e2f6fc4f9ca : SUCCESS in 1h 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b20e909bedbd5f68fabedf67be457372ce703950","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-06-10 14:03:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"95fe94c1aacd5e6c958cdff32f46a8530ecebdd4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-06-10 15:48:39.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/12/747612/3/check/tempest-dsvm-full-bionic-py3/e114122/ : SUCCESS in 1h 34m 55s","accounts_in_message":[],"_revision_number":3},{"id":"13d32a952b5e37d2dbb3406600b1e9678068cbd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:20:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e29c99fde8c24ce78f8216dbd2771e09 : SUCCESS in 12m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31d2e45e363f4a828c42eee8b84e7fe0 : SUCCESS in 11m 47s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6d8aa9e5022944a2b89ed85c1e05a27c : SUCCESS in 20m 49s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e95af880e9ac4825a7c56731dd1ace2e : SUCCESS in 9m 11s\n- tempest-full https://zuul.opendev.org/t/openstack/build/f941dfd1141b4ac5b7bb3b2661590dae : SUCCESS in 1h 22m 23s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/21051ae433de4555ac3b339856ae9bd0 : SUCCESS in 48m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2502c3277db44b58967a7da90c2db24f : SUCCESS in 1h 23m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887e177ea5d74828a70bae02c36291b5 : SUCCESS in 19m 27s\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f4d033263b064e22a7ef65822d03fea6 : SUCCESS in 51m 03s (non-voting)\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/59447e8b72604f269b9504216553a80d : SUCCESS in 1h 12m 17s\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/dea4b1761ff047a4921b7e5b81911687 : SUCCESS in 55m 09s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/40894b31422047b2902ddbfed8dc7de0 : SUCCESS in 51m 20s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/2cd9caa16dc34549a1d02dd038b192e8 : SUCCESS in 1h 03m 30s (non-voting)\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/c60c6c3d784e4c1bb205b22b9dadf949 : SUCCESS in 1h 10m 09s\n- nova-next https://zuul.opendev.org/t/openstack/build/6696f9d792094b6fa79389a780d720e5 : SUCCESS in 1h 32m 38s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/38472a64191f4a5ca4cea1fefea773ce : SUCCESS in 2h 10m 19s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/cf2b3a4a890f415785bbff9523007b86 : SUCCESS in 15m 33s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/874da96d85e9469884a760c12914c1b9 : SUCCESS in 18m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/647f01020bce4b06be8475e24c2d366e : FAILURE in 5m 41s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a3445f9d3d94ba8a7cb52bb4c591518 : SUCCESS in 1h 10m 30s\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/ef9e17a514e44c55b7d1ffed9ac9082f : SUCCESS in 1h 51m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b4f2da4202dbd6057fa3f09e7ce751b1a2035a7e","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2021-06-17 02:24:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nClean backport from stable/stein of useful bug fix, LGTM","accounts_in_message":[],"_revision_number":3},{"id":"c4ea5b8d9ab75881bf358a0e7a72648c4bf35064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 02:24:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c532a2bd29f00456f384a409dc5046e8dac79a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:18:08.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5b9cf1935f64d4b9af8e09c4601218d : SUCCESS in 10m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/44c539a7e14341729ef460d033b89dec : SUCCESS in 13m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d5713af00f62498bb0e46b2226b4059f : SUCCESS in 14m 03s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/74b3485df6e44f6fb797613ccbbfbd2e : SUCCESS in 10m 18s\n- tempest-full https://zuul.opendev.org/t/openstack/build/6c1097d35deb4e4996ccc6cb4acdd622 : SUCCESS in 1h 15m 42s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/1bea27cc24fb47d2872fb2ef26627d8a : SUCCESS in 1h 29m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4118b35aa77f4757919ed7fa92bcc8cd : SUCCESS in 1h 13m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72ed9f5be1bf4005981ca48a95827c91 : SUCCESS in 21m 56s\n- nova-cells-v1 https://zuul.opendev.org/t/openstack/build/b63800890e014b79b83fd103baa1ebcb : SUCCESS in 57m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c8ccff9bdf924b8a839acf88015c20a9 : SUCCESS in 45m 37s\n- nova-multiattach https://zuul.opendev.org/t/openstack/build/98fad32218d24eb397afa39641dde292 : SUCCESS in 1h 49m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/fe1cb0ae3d44438eadc3d4870ff4edf8 : SUCCESS in 1h 18m 09s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a17d907a306c42b2bfe1af10e8a1cde1 : SUCCESS in 2h 57m 02s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/511c4476283f4316b0121a94db55495b : SUCCESS in 16m 34s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/ae29c48a9d434be6b23953245ba32098 : SUCCESS in 17m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ca92dd6d2c94b33b9552c2226f6811d : FAILURE in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb9a258b5a955fb12fd126722babfd578a7baa77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:18:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"600510f66fdc4f936c58d85a8e048b88a863d529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 07:20:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50f8a3eefaac49c0adfb0224fd4521a0 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6e6440fe97c07489ab6b70029a22fa0b9b70bc02","revisions":{"a922df13382b76ef52829242f70c228b226c13ce":{"kind":"REWORK","_number":1,"created":"2020-08-24 04:08:42.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/12/747612/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/747612/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/747612/1"}}},"commit":{"parents":[{"commit":"69f740644bac092cee20fb1d112a837e7e14aeb5","subject":"Merge \"FUP for in-place numa rebuild\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/69f740644bac092cee20fb1d112a837e7e14aeb5"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-09-17 11:08:59.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-08-24 04:08:42.000000000","tz":0},"subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","message":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available\n\nBecause of the libvirt issue[1], there is a bug[2] that if we set cache mode\nwhose write semantic is not O_DIRECT (.i.e unsafe, writeback or writethrough),\nthere will be a problem with the volume drivers\n(.i.e nova.virt.libvirt.volume.LibvirtISCSIVolumeDriver,\nnova.virt.libvirt.volume.LibvirtNFSVolumeDriver and so on), which designate\nnative io explicitly.\n\nThat problem will generate a libvirt xml for the instance,\nwhose content contains\n\n```\n...\n\u003cdisk ... \u003e\n  \u003cdriver ... cache\u003d\u0027unsafe/writeback/writethrough\u0027 io\u003d\u0027native\u0027 /\u003e\n\u003c/disk\u003e\n...\n```\nIn turn, it will fail to start the instance or attach the disk.\n\n\u003e When qemu is configured with a block device that has aio\u003dnative set, but\n\u003e the cache mode doesn\u0027t use O_DIRECT (i.e. isn\u0027t cache\u003dnone/directsync or any\n\u003e unnamed mode with explicit cache.direct\u003don), then the raw-posix block driver\n\u003e for local files and block devices will silently fall back to aio\u003dthreads.\n\u003e The blockdev-add interface rejects such combinations, but qemu can\u0027t\n\u003e change the existing legacy interfaces that libvirt uses today.\n\n[1]: https://github.com/libvirt/libvirt/commit/058384003db776c580d0e5a3016a6384e8eb7b92\n[2]: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1086704\n\nCloses-Bug: #1841363\nChange-Id: If9acc054100a6733f3659a15dd9fc2d462e84d64\n(cherry picked from commit af2405e1181d70cdf60bcd0e40b3e80f2db2e3a6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a922df13382b76ef52829242f70c228b226c13ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a922df13382b76ef52829242f70c228b226c13ce"}]},"branch":"refs/heads/stable/rocky"},"eea24fa9e6b14927e00f6167e011e216a0677eb5":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-09-11 11:26:55.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/12/747612/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/747612/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/747612/2"}}},"commit":{"parents":[{"commit":"0d1fd02b301bbc25c75cb2476b24f3be5d7cda77","subject":"Merge \"libvirt: Provide VIR_MIGRATE_PARAM_PERSIST_XML during live migration\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0d1fd02b301bbc25c75cb2476b24f3be5d7cda77"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-09-17 11:08:59.000000000","tz":480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-11 11:26:46.000000000","tz":60},"subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","message":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available\n\nBecause of the libvirt issue[1], there is a bug[2] that if we set cache mode\nwhose write semantic is not O_DIRECT (.i.e unsafe, writeback or writethrough),\nthere will be a problem with the volume drivers\n(.i.e nova.virt.libvirt.volume.LibvirtISCSIVolumeDriver,\nnova.virt.libvirt.volume.LibvirtNFSVolumeDriver and so on), which designate\nnative io explicitly.\n\nThat problem will generate a libvirt xml for the instance,\nwhose content contains\n\n```\n...\n\u003cdisk ... \u003e\n  \u003cdriver ... cache\u003d\u0027unsafe/writeback/writethrough\u0027 io\u003d\u0027native\u0027 /\u003e\n\u003c/disk\u003e\n...\n```\nIn turn, it will fail to start the instance or attach the disk.\n\n\u003e When qemu is configured with a block device that has aio\u003dnative set, but\n\u003e the cache mode doesn\u0027t use O_DIRECT (i.e. isn\u0027t cache\u003dnone/directsync or any\n\u003e unnamed mode with explicit cache.direct\u003don), then the raw-posix block driver\n\u003e for local files and block devices will silently fall back to aio\u003dthreads.\n\u003e The blockdev-add interface rejects such combinations, but qemu can\u0027t\n\u003e change the existing legacy interfaces that libvirt uses today.\n\n[1]: https://github.com/libvirt/libvirt/commit/058384003db776c580d0e5a3016a6384e8eb7b92\n[2]: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1086704\n\nCloses-Bug: #1841363\nChange-Id: If9acc054100a6733f3659a15dd9fc2d462e84d64\n(cherry picked from commit af2405e1181d70cdf60bcd0e40b3e80f2db2e3a6)\n(cherry picked from commit 0bd58921a1fcaffcc4fac25f63434c9cab93b061)\n(cherry picked from commit d92fe4f3e69db51b7bbad52a1fe1740e37bade9a)\n(cherry picked from commit cc2f45ebb03326c757abbfaef4de3e2c1fa4ead2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eea24fa9e6b14927e00f6167e011e216a0677eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/eea24fa9e6b14927e00f6167e011e216a0677eb5"}]},"branch":"refs/heads/stable/rocky"},"6e6440fe97c07489ab6b70029a22fa0b9b70bc02":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-25 00:26:52.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/12/747612/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/12/747612/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/12/747612/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/12/747612/3"}}},"commit":{"parents":[{"commit":"f10d5bdbfa0a3d5deb633ccd43833d9abb2218d9","subject":"Merge \"libvirt: Do not reference VIR_ERR_DEVICE_MISSING when libvirt is \u003c v4.1.0\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f10d5bdbfa0a3d5deb633ccd43833d9abb2218d9"}]}],"author":{"name":"Arthur Dayne","email":"huyueheng@inspur.com","date":"2019-09-17 11:08:59.000000000","tz":480},"committer":{"name":"norman shen","email":"yshxxsjt715@gmail.com","date":"2020-09-25 00:26:52.000000000","tz":0},"subject":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available","message":"libvirt:driver:Disallow AIO\u003dnative when \u0027O_DIRECT\u0027 is not available\n\nBecause of the libvirt issue[1], there is a bug[2] that if we set cache mode\nwhose write semantic is not O_DIRECT (.i.e unsafe, writeback or writethrough),\nthere will be a problem with the volume drivers\n(.i.e nova.virt.libvirt.volume.LibvirtISCSIVolumeDriver,\nnova.virt.libvirt.volume.LibvirtNFSVolumeDriver and so on), which designate\nnative io explicitly.\n\nThat problem will generate a libvirt xml for the instance,\nwhose content contains\n\n```\n...\n\u003cdisk ... \u003e\n  \u003cdriver ... cache\u003d\u0027unsafe/writeback/writethrough\u0027 io\u003d\u0027native\u0027 /\u003e\n\u003c/disk\u003e\n...\n```\nIn turn, it will fail to start the instance or attach the disk.\n\n\u003e When qemu is configured with a block device that has aio\u003dnative set, but\n\u003e the cache mode doesn\u0027t use O_DIRECT (i.e. isn\u0027t cache\u003dnone/directsync or any\n\u003e unnamed mode with explicit cache.direct\u003don), then the raw-posix block driver\n\u003e for local files and block devices will silently fall back to aio\u003dthreads.\n\u003e The blockdev-add interface rejects such combinations, but qemu can\u0027t\n\u003e change the existing legacy interfaces that libvirt uses today.\n\n[1]: https://github.com/libvirt/libvirt/commit/058384003db776c580d0e5a3016a6384e8eb7b92\n[2]: https://bugzilla.redhat.com/show_bug.cgi?id\u003d1086704\n\nCloses-Bug: #1841363\nChange-Id: If9acc054100a6733f3659a15dd9fc2d462e84d64\n(cherry picked from commit af2405e1181d70cdf60bcd0e40b3e80f2db2e3a6)\n(cherry picked from commit 0bd58921a1fcaffcc4fac25f63434c9cab93b061)\n(cherry picked from commit d92fe4f3e69db51b7bbad52a1fe1740e37bade9a)\n(cherry picked from commit cc2f45ebb03326c757abbfaef4de3e2c1fa4ead2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e6440fe97c07489ab6b70029a22fa0b9b70bc02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6e6440fe97c07489ab6b70029a22fa0b9b70bc02"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
