)]}'
{"id":"openstack%2Fnova~748453","triplet_id":"openstack%2Fnova~master~If469762b22d687151198468f0291821cebdf26b2","project":"openstack/nova","branch":"master","topic":"bug/1893221","hashtags":[],"change_id":"If469762b22d687151198468f0291821cebdf26b2","subject":"Make PCI claim NUMA aware during live migration","status":"MERGED","created":"2020-08-27 14:56:37.000000000","updated":"2021-01-20 17:53:49.000000000","submitted":"2021-01-20 17:53:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"748453-bug/1893221","meta_rev_id":"2de218618c58b8250e1fca18229acdfd1132f20d","_number":748453,"virtual_id_number":748453,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-20 17:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-20 17:53:49.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-20 17:53:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2021-01-20 12:17:46.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":1,"date":"2021-01-20 17:53:49.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-01-20 12:17:46.000000000","_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"value":0,"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},{"value":0,"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"value":0,"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"value":0,"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-30 15:45:23.000000000","updated_by":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"reviewer":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"state":"REVIEWER"},{"updated":"2020-09-03 14:27:47.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2020-09-03 14:41:16.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-09-07 10:43:56.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-10 19:54:10.000000000","updated_by":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-14 16:40:31.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-09-14 16:43:58.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-14 17:31:27.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-10-13 20:19:03.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-11-16 12:16:23.000000000","updated_by":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-16 14:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 11:38:58.000000000","updated_by":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"reviewer":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"state":"REVIEWER"},{"updated":"2020-11-18 11:39:18.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-11-18 12:27:27.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-11-18 12:45:30.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-18 14:34:25.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-19 14:59:26.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2021-01-20 12:17:46.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"}],"messages":[{"id":"05a7c700de41dbacf9c5c37761bb2d7bca8d35bf","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-27 14:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a445fc36165d8d4e75ab74b5d9f5464ffe295b69","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-27 15:23:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\ni need to think about this. i think this is basiclaly what i suggested e.g. just pass teh toplogy down which is the correct thing to do.\n\nhowever if we generate teh dest toplogy seperate form claiming the pci devices we may fail to claim the devices.\n\nfor example the new toplogy that was generated might fit but the pci device might not be on the same nodes. so we might fail even though a different instance toplogy could have fit on the same host.\n\nthis might be the best we can do for now but i think this will need a big fixme or todo to ensure the numa topology on the dest takes account of the pci devices and other resouces.\n\ni can try and take a look but if nothing else there is likely a race here where the topology was vaild when we generateed but is not valid based on teh sriov numa affinity policy when we claim the pci devices.\n\nthis is still better then we do today so we might want to merge this anyway but its still not fully correct unless this is invoked under at a lock that prodcect resouce tracker assignment while both operations are done.","accounts_in_message":[],"_revision_number":1},{"id":"5befe06db10042840c96cdb3b962ac0a3e05fedb","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-08-27 15:30:00.000000000","message":"Patch Set 1:\n\n* pci-test http://52.27.155.124/pci/748453/1 : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"86619fbb399abb092651ca91d26713c121b5a202","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-08-27 15:52:50.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/1 : FAILURE in 53m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9e728c976aceca28c5b9a699a9220b4d95983f3b","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-27 16:32:50.000000000","message":"Patch Set 1:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/748453/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/427c9e8 : SUCCESS in 1h 35m 22s","accounts_in_message":[],"_revision_number":1},{"id":"aeb28ef44b502d902af3461f5499f01c01d1024a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-27 18:21:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1ab51cca20414bdbb622601cef944971 : SUCCESS in 1h 03m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/57b11253966f44afa7911acdeda180b2 : SUCCESS in 1h 41m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8db28151e076445da58c2fa7ea2d99ad : FAILURE in 15m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fb12da9f511046a982ad448c4ec47a2f : FAILURE in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd0a72b1e5d4ac085596802eae7ce06 : SUCCESS in 9m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/482e341d72ca4a6a921a14277dcb47fd : FAILURE in 14m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/febd3a2d0d3e44188aebed16fd037eec : FAILURE in 11m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac5dd4bdec0f43e4a21b937333307c49 : SUCCESS in 12m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4df8be34aa594ceeb2b84c9279c54d02 : SUCCESS in 1h 24m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fabbf17325bf46d7b07b08fcf0d3680d : SUCCESS in 1h 01m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d0b4baae07b14fed8671e26cc3af0897 : FAILURE in 39m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b8d49696c3454c40b830c361b78e8efe : SUCCESS in 50m 19s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5de2e57f8f7a4a8386cc499c63697f77 : FAILURE in 1h 52m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/ee2c6d7649904a81b3df483a1d7f5bed : FAILURE in 1h 55m 04s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d732f2a1afd048aebe6d95049095f49e : SUCCESS in 19m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b03f72d1f1e494c9fbc2ce51e1aa29c : SUCCESS in 55m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6f10b26f29a430caa1d2c57acd2adcc : SUCCESS in 1h 09m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/229b48b688234a9796cdaed7a1aa431c : SUCCESS in 26m 05s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/7d3ac53a7b66470d848f268672379d9e : SUCCESS in 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff6d13dddb53d788e00b3519784c9a14b5ff5c26","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-27 20:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/1/check/tempest-dsvm-full-bionic/4e3cdc8/ : SUCCESS in 1h 35m 05s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/1/check/tempest-dsvm-full-bionic-py3/7028106/ : SUCCESS in 1h 25m 29s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/1/check/grenade-dsvm-bionic/df15bf8/ : POST_FAILURE in 1h 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"94ace1bbf880ca0a2d68bcf2638f6de1b387bc8e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-28 11:40:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a50c62ab456da0e438c42071f317c4a306ca4300","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-28 11:40:59.000000000","message":"Patch Set 1:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e i need to think about this. i think this is basiclaly what i\n \u003e suggested e.g. just pass teh toplogy down which is the correct\n \u003e thing to do.\n \u003e \n \u003e however if we generate teh dest toplogy seperate form claiming the\n \u003e pci devices we may fail to claim the devices.\n\nthe code uses the existing MoveClaim logic that considers PCI as well.\n\n \u003e \n \u003e for example the new toplogy that was generated might fit but the\n \u003e pci device might not be on the same nodes. so we might fail even\n \u003e though a different instance toplogy could have fit on the same\n \u003e host.\n \u003e \n \u003e this might be the best we can do for now but i think this will need\n \u003e a big fixme or todo to ensure the numa topology on the dest takes\n \u003e account of the pci devices and other resouces.\n \u003e \n \u003e i can try and take a look but if nothing else there is likely a\n \u003e race here where the topology was vaild when we generateed but is\n \u003e not valid based on teh sriov numa affinity policy when we claim the\n \u003e pci devices.\n \u003e \n \u003e this is still better then we do today so we might want to merge\n \u003e this anyway but its still not fully correct unless this is invoked\n \u003e under at a lock that prodcect resouce tracker assignment while both\n \u003e operations are done.","accounts_in_message":[],"_revision_number":1},{"id":"0d22f3af9c110bf1d958a1e7ff8e049fd3c74511","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-08-28 12:35:50.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/2 : FAILURE in 49m 25s","accounts_in_message":[],"_revision_number":2},{"id":"484ce3b679d57faa38967ab18c070e7ddbe80827","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-08-28 13:23:28.000000000","message":"Patch Set 2:\n\n* pci-test http://52.27.155.124/pci/748453/2 : FAILURE","accounts_in_message":[],"_revision_number":2},{"id":"600de029ece70e4aa2925a92940448b2ba4dc2bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 13:50:29.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e46ea0597792457cb2356aa7ee666e63 : SUCCESS in 1h 00m 13s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0fcd734e7290415da88ca004e6b2f882 : SUCCESS in 1h 17m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b47bbfc1c1e845c19563c7b832f0888e : SUCCESS in 17m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/613aa687b2214b66b5f304a3c79823ac : SUCCESS in 13m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/255d8ac98e7f439084e0b829c0c09d57 : FAILURE in 8m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/deac98cf291f404296ddb838b5825a9f : SUCCESS in 13m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6e75a8f316543679fed3a6e596d2f5c : SUCCESS in 10m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edfbe7d6b2514913b13f345c95c272d6 : SUCCESS in 10m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b5653b01ba6343309972142e60bcdc13 : SUCCESS in 1h 26m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4587f6be8b28445293b67ef1239cd96c : SUCCESS in 1h 01m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0b88c38711a438bbc612eeb7f43418e : SUCCESS in 1h 20m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cb9d9049081e4e93be10b8c8659f0c6d : SUCCESS in 56m 04s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/d913db91725a47a5bfb8f3845eec1f0a : SUCCESS in 47m 18s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2f567e9637e44e99b32b710618620f64 : SUCCESS in 1h 17m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/8d5cbcef0cfc4492965527a2c1ff7f8b : SUCCESS in 1h 26m 13s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fbf3a72064b84746b07823f4b596869b : SUCCESS in 26m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d997398bbd0043c2a38a74e19fd82912 : SUCCESS in 59m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca71b36060c442aba47ad332a2ca9c6d : SUCCESS in 1h 01m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/bb26e218fce34413aa7e98f9101a05e4 : SUCCESS in 32m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6ab46e2e6b9247778cb437010d27f6a8 : SUCCESS in 28m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e2ef280b315fa3d69448759f31278679de0e5ae","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 14:14:40.000000000","message":"Patch Set 2:\n\nTesting succeeded on ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/748453/2/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/7a6a85c : SUCCESS in 1h 38m 06s","accounts_in_message":[],"_revision_number":2},{"id":"492df98193b68c9957a461a61bfa37cfb876f8d0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-08-28 16:55:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"adb50cbb84bf72687ff5536cd192cf0969fe7e23","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-08-28 17:28:07.000000000","message":"Patch Set 3:\n\n* pci-test http://52.27.155.124/pci/748453/3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"12ac98764afffc43bc157961b2c7411c524c5c27","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-08-28 18:00:20.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/3 : FAILURE in 1h 01m 37s","accounts_in_message":[],"_revision_number":3},{"id":"4ad18cafae111ca755653d907552d97c97a600fa","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-08-28 18:54:59.000000000","message":"Patch Set 3:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/748453/3/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/36d98e3 : FAILURE in 1h 31m 08s","accounts_in_message":[],"_revision_number":3},{"id":"2509321714656720d5c12fdafead63ae6d9f6ce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 20:31:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4c1ba479e551497fa712477002513ad4 : SUCCESS in 1h 02m 02s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4664dae77a574cdfa588b09463f44a48 : SUCCESS in 1h 36m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fec7d99f9274cdd90231e171b43621c : SUCCESS in 20m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caa3b5d83b104be786d7e9d8f2838335 : SUCCESS in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5374cfe0201943b7845f02a047387fbc : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2e44aa532604f5e9008c93191b7eefd : SUCCESS in 12m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82f0e88989ae4632b8565fd349fa078c : SUCCESS in 16m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481904983d914d5d85e6e61ade8c6627 : SUCCESS in 12m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/bdfb0f0d962742bbb618fe131ff4d448 : SUCCESS in 1h 10m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a1929bf7b2c42eebfe6cd997ce6aa94 : SUCCESS in 42m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec4d2dfcb29743c0af86d62d1cb6b92e : SUCCESS in 1h 27m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c2eb1d05cbda4b83a3326a3da6537d96 : SUCCESS in 57m 04s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/17892d737c594286b19b7f5cb2636629 : SUCCESS in 41m 14s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1c306cfeec2743698cc5dd8beeb2f3c2 : SUCCESS in 1h 28m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/663ef1cbc8d5435194192bb177546a9e : SUCCESS in 1h 02m 57s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/784f364007df4fc1847b90f950f97cdb : SUCCESS in 14m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2587ba981b047eb9afad15079287bff : SUCCESS in 52m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb2c5da5f78c4839a3e2715139319d45 : SUCCESS in 56m 25s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/aec694e0d5d4471c86e51c2d99bc9279 : SUCCESS in 38m 16s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fa82965d5d9544da937c251f269d75f7 : SUCCESS in 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d20370591ac8b377016a19663f473632f425e397","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-08-28 20:56:44.000000000","message":"Patch Set 3:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-748453-3 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":3},{"id":"998738c0523413afef55c80cd965ed0208fd7459","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-08-29 04:54:51.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/3/check/tempest-dsvm-full-bionic/e5f3ed5/ : SUCCESS in 1h 39m 35s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/3/check/tempest-dsvm-full-bionic-py3/79714a1/ : SUCCESS in 1h 29m 18s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/3/check/grenade-dsvm-bionic/f9c5624/ : POST_FAILURE in 1h 47m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2e3da20d79540f1cbaa3afc89a69a1e7e4c9706","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-29 12:53:51.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- ext-nova-zuul : EXCEPTION","accounts_in_message":[],"_revision_number":3},{"id":"48f365d43ad17e58795ec5b0a3d0f11788117059","author":{"_account_id":32071,"name":"Hui Jiang","email":"jhui@vmware.com","username":"jhui"},"date":"2020-08-30 15:45:23.000000000","message":"Patch Set 3:\n\nvmware-recheck-patch","accounts_in_message":[],"_revision_number":3},{"id":"a93fb45bb342311e225924ca5703247ffbc59ea0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-08-30 16:47:40.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/3/check-vote/ext-nova-zuul/cc8a466 : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":3},{"id":"2559d4fdeb3cb6d6014f0492857e37d3e59a44f9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-08-31 18:04:15.000000000","message":"Patch Set 3: Code-Review+1\n\n(4 comments)\n\ni think this is likely the best we can do in victoria","accounts_in_message":[],"_revision_number":3},{"id":"bc0de2fbc0f9c5e2d5e18ae62f40973b4330b4e5","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-01 15:23:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"49cda0be84fc5b1c30a28d0a978b697c1306d95e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-01 15:25:07.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e67b134e2f4d6a76e1a22fb8ad9725cd4018f1d3","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-09-01 15:56:47.000000000","message":"Patch Set 4:\n\n* pci-test http://52.27.155.124/pci/748453/4 : FAILURE","accounts_in_message":[],"_revision_number":4},{"id":"2e46cff8f84f31e17c9384f097a03966d5d793f3","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-09-01 16:32:10.000000000","message":"Patch Set 4:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/748453/4/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/60a893a : FAILURE in 13m 25s","accounts_in_message":[],"_revision_number":4},{"id":"86c84caec100d952bcecdb48345cd07869f2435c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-01 18:49:45.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/4 : FAILURE in 1h 00m 02s","accounts_in_message":[],"_revision_number":4},{"id":"147477225ca23b708da3dd8bf01a50b434b1daaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:05:14.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5ee16e489f0443c5b146cb2ad3e85c3b : SUCCESS in 1h 10m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/eff45e93a99e474d9dc5fc5dbe36408e : SUCCESS in 1h 14m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47b816f724dc453ea6748e1d89cc17aa : SUCCESS in 17m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5f4d90dd1654f36a789e2c9bc981cd2 : SUCCESS in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc968f071814acabd3eec2704a09e37 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc97616807a646eb8ed073020567a205 : SUCCESS in 12m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1db47a89abed49a397c94a3cc67ad6fe : SUCCESS in 13m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/970ee04fb8dc48c2add36f65af17be3f : SUCCESS in 9m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/85abf60b679c4f8da6b54fbed8815170 : SUCCESS in 1h 45m 49s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a533f76366d429f83eb71b42e17b056 : SUCCESS in 59m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f49af3eae587452785e2add8a08205b6 : SUCCESS in 1h 12m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7a57b8e83d4844cbbbc8fd4a1a9f3b90 : SUCCESS in 46m 50s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/82e94748187e4c7181ccf52ffea69e9b : SUCCESS in 46m 04s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d7c76cffd0f645279edb0402743713d5 : SUCCESS in 1h 25m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/f3e68b5720444e7f8f9a5401a6ff1ecd : SUCCESS in 1h 35m 48s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/59dba8ae35784ab29f97700ed052b817 : SUCCESS in 20m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0baf8c387f44889948aeb6673cfb597 : SUCCESS in 57m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c67e5c2bc36540aca1811e5e215798a2 : SUCCESS in 59m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/93ee70e862524ec593d44278816135d8 : SUCCESS in 31m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/4958d400a85a49eeb70f032c846ecf01 : SUCCESS in 57m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"739c570b39e8339d4615f9cdc57843d748e00bbf","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-01 20:29:06.000000000","message":"Patch Set 4:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-748453-4 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":4},{"id":"62a30f51ee72fba1dbdc870ea706c76c78192e5f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-01 23:50:57.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/4/check/tempest-dsvm-full-bionic/d2e4d16/ : SUCCESS in 2h 03m 34s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/4/check/tempest-dsvm-full-bionic-py3/bedca97/ : SUCCESS in 1h 19m 33s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/4/check/grenade-dsvm-bionic/d9e0bf7/ : POST_FAILURE in 1h 46m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ebecd419e3e5861c8ffe045d6dc59111545f92f6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-02 00:07:19.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/4/check-vote/ext-nova-zuul/7e6069a : FAILURE in 17m 22s","accounts_in_message":[],"_revision_number":4},{"id":"4de8961bf7fc437a48b519eda069ce5bae992970","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-02 08:24:22.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"805fb2ffdad477f6be5b24efd4bd6ccbd34d8a47","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-09-02 08:37:04.000000000","message":"Patch Set 5:\n\n* pci-test http://52.27.155.124/pci/748453/5 : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"61b723421b951847342200ac43a0e4ffb3a709dd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-02 09:25:20.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/5 : FAILURE in 55m 14s","accounts_in_message":[],"_revision_number":5},{"id":"3d047c9455aa3a11140deeda378c1f77bbc5f3d3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-02 10:26:59.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/5/check-vote/ext-nova-zuul/7452ff1 : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":5},{"id":"374d998ff4f23c16aab569a98ff70196a02ceb94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 13:28:23.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ea4bff57627b48808a17d83355d63086 : SUCCESS in 1h 02m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/dfd14fbef9574b96a95be71bd2f3047e : SUCCESS in 1h 14m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1177d87312724a3081b836c6dfd27d1c : SUCCESS in 17m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a6f6441ab0cb4b9992a633e115dc0f5d : SUCCESS in 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df8924e12cde4d1a9995c05626956fc1 : SUCCESS in 9m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea44065dcf6246cca66b2c71ac82170f : SUCCESS in 15m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12ff147fa9954c96b3d839e7f7c46b96 : SUCCESS in 14m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c64aa3ce6754731b9b6825a2314543c : SUCCESS in 15m 23s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0936dfda1a37415fbe1917b0188a9370 : SUCCESS in 1h 12m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/35b334a3ddb6486fb42aff500f580996 : SUCCESS in 1h 06m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e93f80eef02421cb1ffed6d0fc5f126 : SUCCESS in 1h 15m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6c2c3ffdab4449549c6639d8c60d2e41 : SUCCESS in 48m 28s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/7810b9fb30944914adf2b87d8fcb2405 : SUCCESS in 32m 09s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6b45f68ffe414935a0145d4b6412822c : TIMED_OUT in 3h 14m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/5b62102712c546b1a3c0942ef07aa381 : SUCCESS in 1h 35m 55s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9fa112d414714042abcd6e54d934ac45 : SUCCESS in 19m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48564003b02b4948a14f4da3f9f52621 : SUCCESS in 1h 04m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de9f06ef25264a29b60fd087db7e9111 : SUCCESS in 54m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc92cef2b98b43dda04de499e75bb4ed : SUCCESS in 28m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2cfccf29378e4d78aef55e86ea25a29e : SUCCESS in 45m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"08099c00f57f7881d772f94a5e33b1ebdffc4297","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-02 18:56:50.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/5/check/tempest-dsvm-full-bionic/c46c2c3/ : SUCCESS in 1h 40m 01s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/5/check/tempest-dsvm-full-bionic-py3/116dd65/ : SUCCESS in 1h 41m 38s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/5/check/grenade-dsvm-bionic/b14d910/ : POST_FAILURE in 1h 53m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e16ec83269c1c718c4d1b324911877b7b4cc912f","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-03 07:40:04.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"4d62450b7882bba196c85dbc9715b8f5fed8c50c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-03 08:07:02.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/5/check-vote/ext-nova-zuul/3693f96 : FAILURE in 24m 29s","accounts_in_message":[],"_revision_number":5},{"id":"82735b6a6131a1f2734acafd6cc1ac47b5827608","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-03 08:30:44.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/53/748453/5/check/pmem-tempest-plugin-filtered-multi-node/a59ed15/ : SUCCESS in 48m 43s\n- devstack-networking-ovs-dpdk http://52.27.155.124/53/748453/5/check/devstack-networking-ovs-dpdk/6376beb/ : RETRY_LIMIT in 1m 52s","accounts_in_message":[],"_revision_number":5},{"id":"fe3d039426dfd247d04900a429212b55b5949869","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-03 08:49:01.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/5 : FAILURE in 1h 05m 57s","accounts_in_message":[],"_revision_number":5},{"id":"91e099000ab3b90dc431be5a493a6aa9853ab7e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-03 09:27:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/06f6fbdc441c4f689cb8e6a530504c10 : SUCCESS in 1h 11m 38s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c3471f4018fa4ddab70de44c3252d4e3 : SUCCESS in 1h 31m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02ff4fed00a14008b2cd6034d2e7889f : SUCCESS in 18m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/accc51290ab54052ac6e87c6dd6e69ad : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0500bbf9b9bc4f3fafe3fc57217d2e90 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66a1f7b5b36a4d51affc2224b1aab0b3 : SUCCESS in 13m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aba6b22d81b0490481ee56b8e5275e6d : SUCCESS in 13m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698e3c5af90d41d78820f939d7fe28ed : SUCCESS in 11m 22s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e71acccdf2eb42cc9850b019deb58bbb : SUCCESS in 1h 13m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b37c532abf6e4f6eb7778a6f7705e36c : SUCCESS in 1h 04m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8836590846764de58db98b8cae7432f5 : SUCCESS in 1h 14m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/57327cb3a91743918df9d4a4093bb30b : SUCCESS in 51m 25s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/eb6a32cd9c534dc1bed1ab91cd1f6298 : FAILURE in 1h 06m 05s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3f24a17690f64f7897dedf2edab457a9 : SUCCESS in 1h 22m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/0dda7b0c7c0544f1a9864ce2fd832760 : SUCCESS in 1h 00m 24s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f73ee4ea64ab42a8adc2f877076251f0 : SUCCESS in 18m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c97b6d06f99b462ebbcfede7efe824d1 : SUCCESS in 52m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92c75925244643529cbd65680783ff93 : SUCCESS in 57m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/20b9b839933341b6a808232adc6da1e7 : SUCCESS in 28m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e1a703a8badd47af90c5250479d11b20 : SUCCESS in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"42cbdff3ebd367fc750ff55ac28472057e43d12c","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-03 09:29:07.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/53/748453/5/check/EMC_PowerFlex_NOVA/bb16ea8/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"8c4ef499ec0f46b4b767a500f239d2167c7d774f","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-03 09:47:45.000000000","message":"Patch Set 5:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-748453-5 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":5},{"id":"37cfd31add4e7e352c788f62e30d4dc28f6b53f7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-03 11:56:44.000000000","message":"Patch Set 5:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/5/check/tempest-dsvm-full-bionic/b1e06d4/ : SUCCESS in 1h 33m 41s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/5/check/tempest-dsvm-full-bionic-py3/5782c0c/ : SUCCESS in 1h 40m 22s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/5/check/grenade-dsvm-bionic/02bd959/ : POST_FAILURE in 1h 37m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"773d2b70ba7c30c2e10dd0ddd8787faa7979c28b","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-09-03 14:27:47.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)\n\nI still disagree on the approach (see TODO note inline), and normally I\u0027d have asked for functional tests, but given that https://review.opendev.org/#/c/749360/1/nova/tests/functional/libvirt/test_pci_sriov_servers.py exists and the amount of work and conflicts associated with it, I think we can move forward with this patch, merge the functional tests later, and *then* I\u0027d like to unify all the live migration stuff under the MoveClaim (unless Sean gets core by that time and -2s me ;)","accounts_in_message":[],"_revision_number":5},{"id":"91b28ab01bd9cac2c05338ebcaf8354b94249577","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2020-09-03 14:28:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9553fc14094639596310bab93951349d7f311be5","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2020-09-03 14:41:16.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"53dc0d3758bf67969fbab8fbdb83ccba9aef1c3e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-04 16:53:54.000000000","message":"Patch Set 5: Code-Review-1\n\nI\u0027m not certain this is wrong, but I can\u0027t reproduce this issue in functional tests with [1]. In that test, I create PCI devices on NUMA node 0 for the \u0027test_compute0\u0027 host, and on NUMA node 1 for the \u0027test_compute0\u0027 host. If I configure \u0027[compute] cpu_dedicated_set\u0027 so that all CPUs on NUMA node 1 are disabled, the live migration fails. This failure occurs because of the check at [2]. That being the case, how would one cause this bug? Is there just a potential for a race here, or is it even a bug?\n\n[1] https://review.opendev.org/#/c/749360\n[2] https://github.com/openstack/nova/blob/fa374dcc88/nova/compute/claims.py#L138-L149","accounts_in_message":[],"_revision_number":5},{"id":"eb075000a12292e470c1724183eaafcd7c058bda","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-07 08:54:43.000000000","message":"Patch Set 5:\n\n(2 comments)\n\n\u003e I\u0027m not certain this is wrong, but I can\u0027t reproduce this issue in\n \u003e functional tests with [1]. In that test, I create PCI devices on\n \u003e NUMA node 0 for the \u0027test_compute0\u0027 host, and on NUMA node 1 for\n \u003e the \u0027test_compute0\u0027 host. If I configure \u0027[compute]\n \u003e cpu_dedicated_set\u0027 so that all CPUs on NUMA node 1 are disabled,\n \u003e the live migration fails. This failure occurs because of the check\n \u003e at [2]. That being the case, how would one cause this bug? Is there\n \u003e just a potential for a race here, or is it even a bug?\n \u003e \n \u003e [1] https://review.opendev.org/#/c/749360\n \u003e [2] https://github.com/openstack/nova/blob/fa374dcc88/nova/compute/claims.py#L138-L149\n\nI moved this patch out of the SRIOV attach series so it is not block that effort. I have to look at the functional test for this to understand more why passing None does not cause any problem.","accounts_in_message":[],"_revision_number":5},{"id":"3f737148f376510f12c3327ee2b85b53052dacf1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-07 09:09:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"130625b5dd19dd75460c5119e103d1d9e87f720d","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-07 10:03:20.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/53/748453/6/check/pmem-tempest-plugin-filtered-multi-node/de2aca1/ : SUCCESS in 51m 01s\n- devstack-networking-ovs-dpdk http://52.27.155.124/53/748453/6/check/devstack-networking-ovs-dpdk/ae5cc0a/ : RETRY_LIMIT in 2m 24s","accounts_in_message":[],"_revision_number":6},{"id":"42cf16169b859011823f8e28c680b6fc7e86b206","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-07 10:13:32.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/6 : FAILURE in 58m 47s","accounts_in_message":[],"_revision_number":6},{"id":"5262d268ec1e8b4639fb42554959f46863e176f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-07 10:41:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/98bd7b6654ad41229b41ae8f04a399a3 : SUCCESS in 59m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/be66404ab5ef442889273edd69a55cc0 : SUCCESS in 1h 20m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f3d89085f54512bb827eb426dee2a2 : SUCCESS in 18m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e050771b7f4b4da8b83844c2ec832ec8 : SUCCESS in 13m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a24571927144ffbac272ef1cfea39d : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28dc8d20a9044a3a9a0099e3e8d056bf : SUCCESS in 14m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3835e13ef00c44d0a29208101c8c1573 : SUCCESS in 13m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd5c72c9bd74166b4bf302eb90bd5d8 : SUCCESS in 10m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee53270827b74f6f8cfbbdc08f862b8d : SUCCESS in 1h 11m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b7807fa3d6b467da0961289f2e9e70d : SUCCESS in 1h 03m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a4dd4b30129545718ff935f54d6f76b6 : SUCCESS in 1h 10m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/95e82dc69831438d9ff47ff68f4d7ff3 : SUCCESS in 54m 08s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/fe73c3c45d044b7ab7a7173843bb1500 : SUCCESS in 41m 44s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e2b19fef9da2496485806090f3ecb07b : SUCCESS in 1h 13m 57s\n- nova-next https://zuul.opendev.org/t/openstack/build/fbe431e1c3fa476b97698a99a37b0e2e : SUCCESS in 1h 29m 22s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04e00f0bc57f44f68ac393c4cf473b95 : SUCCESS in 17m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5d5ad698a4f3492ab07b78b87bbd3c7e : SUCCESS in 53m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6801fdb1fc094b9fb417cef4512f5878 : SUCCESS in 54m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0b4746af34624f91bc3f2ffdb2a172c5 : SUCCESS in 29m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5d6a1c2cf801406790202b26440da4ac : SUCCESS in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"858b5625af63d6d03ed717e465a8254f2bd41fea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-09-07 10:43:56.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/53/748453/6/check/EMC_PowerFlex_NOVA/847f846/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"90572d3655d4c4d6599d7cac3fa167b6d8f97b1d","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-07 11:02:13.000000000","message":"Patch Set 6:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-748453-6 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":6},{"id":"038083d28d0a5904b102e9f03adf8eb4ce1b7458","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-07 11:39:59.000000000","message":"Patch Set 6:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/6/check/tempest-dsvm-full-bionic/f6bf87f/ : SUCCESS in 1h 23m 15s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/6/check/tempest-dsvm-full-bionic-py3/d3a0a9e/ : SUCCESS in 1h 26m 43s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/6/check/grenade-dsvm-bionic/061f8bc/ : POST_FAILURE in 1h 35m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a3a116f3bf316986c254bd553683487dc629a0f1","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-08 07:34:50.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"ef0116fd659d6f2be0027ac0506289435e0355d4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-09 09:44:11.000000000","message":"Patch Set 6:\n\n\u003e I\u0027m not certain this is wrong, but I can\u0027t reproduce this issue in\n \u003e functional tests with [1]. In that test, I create PCI devices on\n \u003e NUMA node 0 for the \u0027test_compute0\u0027 host, and on NUMA node 1 for\n \u003e the \u0027test_compute0\u0027 host. If I configure \u0027[compute]\n \u003e cpu_dedicated_set\u0027 so that all CPUs on NUMA node 1 are disabled,\n \u003e the live migration fails. This failure occurs because of the check\n \u003e at [2]. That being the case, how would one cause this bug? Is there\n \u003e just a potential for a race here, or is it even a bug?\n \u003e \n \u003e [1] https://review.opendev.org/#/c/749360\n \u003e [2] https://github.com/openstack/nova/blob/fa374dcc88/nova/compute/claims.py#L138-L149\n\nI dig a bit and found the following:\n\nThe NUMA and PCI filtering works during the scheduling and enforces NUMA locality. So a destination host that has CPU on NUMA0 and PCI on NUMA1 is filtered out. I still assume that the NUMA and PCI claim would not work propely without this change. But to reproduce that we would need to create a race situation where the target host has the config:\n* two CPUs are only available on NUMA0\n* one PCI is available on NUMA0 and NUMA1 each.\n\u003d\u003e two instances are scheduled in parallel and the scheduler accepts the target host for both request as NUMA0 is OK but only one of the claims can succeed on the target NUMA0 so the second instance should fail the claim as PCI is only available on NUMA1.\n\nAt the moment I don\u0027t see how this can be reproduced in the functional env.","accounts_in_message":[],"_revision_number":6},{"id":"51f66fc3d4bcb9ac534d464e61e692107f3d3c33","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-09 13:46:49.000000000","message":"Patch Set 6:\n\n\u003e At the moment I don\u0027t see how this can be reproduced in the\n \u003e functional env.\n\nSo I spent time to make a unit test but that is also a pretty deep cut. I would need to call the high level check_can_live_migrate_destination() to trigger both NUMA and PCI claim, with a ResourceTracker that has valid NUMA and PCI data. I stopped after I did not find any good examples setting up a ResourceTracker like that in the unit test.\n\nI\u0027m dropping the Workflow -1. If somebody has a clever way to reproduce it in a test then please take over the patch.","accounts_in_message":[],"_revision_number":6},{"id":"30060363d17527f4e13258f2bf8c45a66e6c53ea","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-09 13:46:54.000000000","message":"Patch Set 6: -Workflow","accounts_in_message":[],"_revision_number":6},{"id":"d8faba1716b927dec695e62dd1ebaa0913cd0651","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-09-09 14:13:52.000000000","message":"Patch Set 6:\n\nso im not really sure what the problem with proving this is required is.\n\nregardelss of if we have other parts of the code that enforce numa affinity here we are claiming pci device without passing the numa toplogy of the instance. so even if other part of the code has ensure that a device existed that could fulfil the numa requirements previously the code that is doing the claim is not aware of the numa requirements so it can claim devices that dont fulfil them.\n\nthat said we can proably combine this with the other work on converging the code paths.","accounts_in_message":[],"_revision_number":6},{"id":"eafbc1ce8c4b6b19479c72a7d5f12637d0f9c1c9","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-10 08:06:08.000000000","message":"Patch Set 6:\n\n\u003e so im not really sure what the problem with proving this is\n \u003e required is.\n \u003e \n \u003e regardelss of if we have other parts of the code that enforce numa\n \u003e affinity here we are claiming pci device without passing the numa\n \u003e toplogy of the instance. so even if other part of the code has\n \u003e ensure that a device existed that could fulfil the numa\n \u003e requirements previously the code that is doing the claim is not\n \u003e aware of the numa requirements so it can claim devices that dont\n \u003e fulfil them.\n \u003e \n \u003e that said we can proably combine this with the other work on\n \u003e converging the code paths.\n\nI think the only missing piece is _ to prove with a test_ that the missing NUMA enforcement causes cross NUMA node resource allocation.","accounts_in_message":[],"_revision_number":6},{"id":"1d787364a1183b3cacea56e639cbd27cc3459e96","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-10 13:21:04.000000000","message":"Patch Set 6: Code-Review+2\n\nOkay, as lovely as it would be to prove this out with a functional test, it seems too difficult. Given we have at least tried to solve the issue, I\u0027m happy with merging this as-is","accounts_in_message":[],"_revision_number":6},{"id":"134f96973632bf4e167aecc391a6f17267412e26","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-10 17:45:39.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"9921eb855f356aed0074a980aeae2a1ce05950d5","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-10 18:08:45.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node pmem-tempest-plugin-filtered-multi-node : NODE_FAILURE in 0s\n- devstack-networking-ovs-dpdk http://52.27.155.124/53/748453/7/check/devstack-networking-ovs-dpdk/6614d3d/ : RETRY_LIMIT in 1m 57s","accounts_in_message":[],"_revision_number":7},{"id":"249f1a81944aae440d5966b6b6343df11aa8addd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-10 18:37:03.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/7 : FAILURE in 49m 31s","accounts_in_message":[],"_revision_number":7},{"id":"eb6222d79c4e16eaeddd581f0519aa9ae1194f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-10 19:26:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/07c7fa3ae5bb46e8885264fdc9d6623b : SUCCESS in 59m 42s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/98ae8871a6dd4af1b7d4e45ccf8f9eb4 : SUCCESS in 1h 12m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/135907b7fd584129a8e19514e2ef1495 : SUCCESS in 20m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/807e4ba6d5bc450db67b5d075b20334d : SUCCESS in 17m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72addbf8f964e2ebc54491c9cb04b75 : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f617c67a436842acb5173dfd46e1727a : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38b10e65f7447e8a0907e03d9fa2d20 : SUCCESS in 18m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a40c4ea5d143c7a405b5ad4e533dd4 : SUCCESS in 22m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c362bfa25be64547b5f14b86aef70024 : SUCCESS in 1h 11m 40s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/700d3cd08e3c4cd29525ebe6a21b87a0 : SUCCESS in 1h 03m 01s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/03f6ee47575148018a05e1ef3adfcbb1 : SUCCESS in 1h 16m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a90e3034b6e44eb08346eea9845ca0d3 : SUCCESS in 49m 11s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/5142b1f32bcc4591a496eaff6f88b22b : SUCCESS in 42m 48s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f623d48265754cde84a15555c0c9782d : SUCCESS in 1h 18m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/30b95ad147cf462490344165bd47fe1a : SUCCESS in 1h 31m 36s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/61655ca4b75249b0b6e0f9aab2f52cbe : SUCCESS in 17m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82ddb80e05934b3484c4f62c4b962300 : SUCCESS in 59m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a694c89ce45a409599e8cd930a7cf666 : SUCCESS in 54m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/450b65cb5a7b4735bcd4c1d49b478459 : SUCCESS in 27m 39s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a643e07daef24c428dfc7eea3c9b9076 : SUCCESS in 40m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8ff26dcbb10dcddd650bd9d924cdfde318c6bcba","author":{"_account_id":14384,"name":"Quobyte CI","email":"openstack-ci-external@quobyte.com","username":"quobyteci","tags":["SERVICE_USER"]},"date":"2020-09-10 19:54:10.000000000","message":"Patch Set 7:\n\n* nova-quobyteci-dsvm-volume http://78.46.57.153:8081/refs-changes-53-748453-7 : FAILURE \n\nSee https://wiki.openstack.org/wiki/ThirdPartySystems/Quobyte_CI for rechecking and info.","accounts_in_message":[],"_revision_number":7},{"id":"1c77d9819c837a654a181598ee4ce645c7241fa1","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-09-10 22:46:35.000000000","message":"Patch Set 7:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/748453/7/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/b711cd2 : FAILURE in 1h 28m 09s","accounts_in_message":[],"_revision_number":7},{"id":"5ec2732768236025a43b53d41bc8cad6b133681b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-10 23:37:34.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/7/check-vote/ext-nova-zuul/3634e6e : SUCCESS in 57m 23s","accounts_in_message":[],"_revision_number":7},{"id":"fe10ea074807391d5613e5827cfae08e278b7614","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-10 23:39:23.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/7/check/tempest-dsvm-full-bionic/e6388c3/ : SUCCESS in 1h 41m 24s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/7/check/tempest-dsvm-full-bionic-py3/e92b1d0/ : SUCCESS in 1h 39m 35s\n- grenade-dsvm-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/7/check/grenade-dsvm-bionic/f12cf91/ : POST_FAILURE in 1h 40m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"67519ec1736fb972774574f254f4a567dfb85914","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-09-14 15:09:24.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"d2582f4e6bdad90a8feb510ec0844e1ce19ae8a9","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-14 16:00:52.000000000","message":"Patch Set 8:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/8 : FAILURE in 50m 14s","accounts_in_message":[],"_revision_number":8},{"id":"0d4f8371790d752d66a4b8750a922c491e3866a5","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-14 16:02:50.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/53/748453/8/check/pmem-tempest-plugin-filtered-multi-node/7a39b8b/ : SUCCESS in 52m 11s\n- devstack-networking-ovs-dpdk http://52.27.155.124/53/748453/8/check/devstack-networking-ovs-dpdk/cede232/ : RETRY_LIMIT in 2m 33s","accounts_in_message":[],"_revision_number":8},{"id":"bfa59bd2465fef5675790f623d4fc2f98f0c31cc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-09-14 16:40:31.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"91a906f9ce423cf8caad604aec23471c35805c07","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-09-14 16:43:58.000000000","message":"Patch Set 8:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/748453/8/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/ec44f46 : FAILURE in 1h 34m 06s","accounts_in_message":[],"_revision_number":8},{"id":"30d564f341171cfadd9244375680d89df5b33438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 17:04:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6846a52b01a6432d9c0bd3b5980d621a : SUCCESS in 57m 10s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7be93b12b3d446baaaa08124c30a7dbc : SUCCESS in 1h 17m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5c9abcaf6cf544e49a66c422cfdf4db8 : SUCCESS in 16m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cee388be403440939198fee13624638d : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5eeff64938ed4eeca0f8b4a173402cec : SUCCESS in 8m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abe0a74e88ca4facaef3f44d7972a916 : SUCCESS in 13m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7792c4c0bd0940f1b226383eec33c302 : SUCCESS in 17m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aebae9e951484f00a63277739eea0283 : SUCCESS in 19m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e56a252e468d4bf9a822eec8b91244ac : SUCCESS in 1h 19m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4277ece086eb4ceaa386195e49af7d7a : SUCCESS in 1h 17m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d33af9230cf045a5a093c327296f9d66 : SUCCESS in 1h 09m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4b589252ac3943e2939dc6b077930bf6 : SUCCESS in 36m 45s\n- nova-multinode-live-migration https://zuul.opendev.org/t/openstack/build/885f6e354110446191f26f889fb04f92 : SUCCESS in 50m 55s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/15d598f796424e4ea87f573cb1771831 : SUCCESS in 1h 24m 45s\n- nova-next https://zuul.opendev.org/t/openstack/build/8c6619436da44c12b84b6e6fbceb0abb : SUCCESS in 1h 48m 20s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a107f6519224f28a1a620b37dfd47a9 : SUCCESS in 17m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b6675f9c204fcb9f47c5e47afde98d : SUCCESS in 1h 04m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4d9ba67904e46ecbaee5bf07e4a7243 : SUCCESS in 1h 06m 14s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/891daaf6ab0c4b8081c64b992127ed57 : SUCCESS in 33m 18s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/49da453917ca4097976a9779a3af68e6 : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7d8264bd6d3443be9d26a47aa6cd0dba06c3b138","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2020-09-14 17:31:27.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"32b4390731757ced49ccbfa071b32ec60e317011","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-09-14 17:53:12.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/8/check-vote/ext-nova-zuul/a096c05 : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":8},{"id":"6deb4fa2c34177157320d89dd00e085103c2a649","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-14 18:13:13.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/8/check/tempest-dsvm-full-bionic/f533eb1/ : SUCCESS in 1h 38m 43s\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/8/check/tempest-dsvm-full-bionic-py3/f2f2e96/ : SUCCESS in 1h 30m 31s\n- grenade-dsvm-bionic finger://zuul-v3-executor.novalocal/f967e6c1e26b4d89ac8c4f2fb6848f3b : RETRY_LIMIT in 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"966afcec9eae93aaae386f19849f33b4c056a0dd","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-10-13 20:19:03.000000000","message":"Patch Set 8:\n\nThe launchpad bug content for this and the test coverage on this patch are sparse IMHO. I see that functional tests have been discussed but we\u0027re thinking of working on them later, and that we don\u0027t yet understand why the existing functional tests around SRIOV live migration pass with None as the destination topology?\n\nMaybe it\u0027s just me but I think it would be worth the effort to figure out the functional tests first and be more certain there are not further issues beyond this one wrt PCI device live migration. Was there at least a local test done for this that gives the confidence that this will solve the problem and not cause other problems?","accounts_in_message":[],"_revision_number":8},{"id":"58e2a29804f72d463236d0699a2589f7514733a6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-10-16 11:41:45.000000000","message":"Patch Set 8:\n\n\u003e The launchpad bug content for this and the test coverage on this\n \u003e patch are sparse IMHO. I see that functional tests have been\n \u003e discussed but we\u0027re thinking of working on them later, and that we\n \u003e don\u0027t yet understand why the existing functional tests around SRIOV\n \u003e live migration pass with None as the destination topology?\n \u003e \n \u003e Maybe it\u0027s just me but I think it would be worth the effort to\n \u003e figure out the functional tests first and be more certain there are\n \u003e not further issues beyond this one wrt PCI device live migration.\n \u003e Was there at least a local test done for this that gives the\n \u003e confidence that this will solve the problem and not cause other\n \u003e problems?\n\nPlease see my analysis about functional test on PS9 from 9th of Sept. We would need to produce a pretty convoluted race condition scenario to trigger the fault. I tried but failed and gave up. \n\nThis whole issue is surfaced during reading the code and realizing that not passing some data to the pci_tracker is obviously wrong. I\u0027ve never seen the actual fault happen in a deployment. So maybe the race is so rare that nobody hit it before. So I\u0027m fine keeping this fix un-merged until we have more evidence.","accounts_in_message":[],"_revision_number":8},{"id":"8a6c12ccb5d89451a521d53aa1adf17a2b4404c2","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-16 10:23:48.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"4a325f31ef54620bdfe6b2a4b096f01287eaffc1","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-11-16 10:24:55.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-None : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"e9ffb153c11cc64e133888a13a2fb3991a75ded9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-16 11:15:40.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/9/check-vote/ext-nova-zuul/2206274 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":9},{"id":"3860ce20f81073d8b7ab9d420edcf3d23ffe9582","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-11-16 11:48:41.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/53/748453/9/check/pmem-tempest-plugin-filtered-multi-node/f4ea7af/ : SUCCESS in 43m 39s\n- devstack-nfv-numa devstack-nfv-numa : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":9},{"id":"46a3f13bd62c2914151ad4097b49661c57984f59","author":{"_account_id":15751,"name":"Intel PCI CI","email":"pci-ci@intel.com","username":"intelpcici","tags":["SERVICE_USER"]},"date":"2020-11-16 12:16:23.000000000","message":"Patch Set 9:\n\n* pci-test http://52.27.155.124/pci/748453/9 : SUCCESS","accounts_in_message":[],"_revision_number":9},{"id":"15c709643a44ac336ba6947a855193946775ea18","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-11-16 13:08:10.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/9 : FAILURE in 2h 41m 38s","accounts_in_message":[],"_revision_number":9},{"id":"ca96297b64df93d61fed5036e8d8efc38e981ad8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-16 14:23:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ccd350b81204624b7dc16f0d7524df3 : SUCCESS in 1h 11m 14s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ccbffd6e84264df88d1c98ecd13f84c6 : SUCCESS in 1h 33m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8f5280603104196a55341a3f93a7773 : SUCCESS in 23m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5274f95140d846c7b96143b4e6c86aa1 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f45bcf81bd462ba67bcb1a66a39f03 : SUCCESS in 9m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e47c188fe7e9478bbb4b03240a864470 : FAILURE in 24m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c564dca1930472d93f585269d30b95b : SUCCESS in 14m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5aa5494770104ccbb1dfbdb6553ac12e : SUCCESS in 26m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fb8f9a896543f19ddf80373a7c1c9c : SUCCESS in 11m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/628b69d9a4a84faf97b75da5cdecfd34 : SUCCESS in 1h 22m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2aa1ad202244e91a7fec74898d77ffb : SUCCESS in 1h 08m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d85f84f3f2044022b04ec0c36f7e3c6f : SUCCESS in 1h 18m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bd3a59a91de4438eae29948f7ede2c8e : POST_FAILURE in 1h 04m 27s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/4e3fc0161ba742d1a81d6050d452835f : SUCCESS in 1h 25m 39s\n- nova-next https://zuul.opendev.org/t/openstack/build/06b966be91124feba1cc9560042a4238 : SUCCESS in 1h 39m 49s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8e6706fd803846bfb6da23b271e4fff3 : SUCCESS in 19m 29s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92be6850e1ad47f98bbe22af1f780cc2 : SUCCESS in 19m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6c87296ff114fa4a6be203b244c277f : SUCCESS in 1h 03m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7ea9508dcda4a10befe62cd1d5d8632 : SUCCESS in 1h 01m 31s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/6450e81840bd479aa0c51bbaabbafb5a : SUCCESS in 37m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6145565cbbdf4bee944bc6cdfbdc434d : SUCCESS in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3c8833844d83cc54449e3b51a8b8c1ef4e21dd1b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-16 14:25:40.000000000","message":"Patch Set 9:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/9/check/tempest-dsvm-full-focal-py3/969d3a3/ : SUCCESS in 2h 23m 40s\n- grenade-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/9/check/grenade-py3/e267124/ : POST_FAILURE in 1h 09m 17s","accounts_in_message":[],"_revision_number":9},{"id":"38544e8fa2615197f9740c78c8ad9828b8f67b6c","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-18 11:38:06.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"52443eac0eeb0c413617a76d73ccd044a91c6ebc","author":{"_account_id":23498,"name":"IBM zVM CI","email":"zvmosci@us.ibm.com","username":"zvmosci"},"date":"2020-11-18 11:38:58.000000000","message":"Patch Set 9:\n\nTesting completed on the zVM Driver CI system check-nova pipeline and failed.  To recheck only the zVM driver plugins, submit a comment with only  zvm: recheck in the comment.. Contact information: zvmosci@us.ibm.com. For information see https://wiki.openstack.org/wiki/ZVMDriver.\n\n- check-nova-master http://extbasicopstackcilog01.podc.sl.edst.ibm.com/test_logs/jenkins-check-nova-master-None : NOT_REGISTERED","accounts_in_message":[],"_revision_number":9},{"id":"a3df4a7b9e9bd9e68acd7810ee1372e78417efc4","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-11-18 11:39:18.000000000","message":"Patch Set 9:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/9 : FAILURE in 25s","accounts_in_message":[],"_revision_number":9},{"id":"9ed9c92fd7e2dae0c45b05cbf06b25ce48508ada","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-11-18 12:27:27.000000000","message":"Patch Set 9:\n\nBuild SUCCESSFUL (check pipeline).\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/53/748453/9/check/pmem-tempest-plugin-filtered-multi-node/1f441b4/ : SUCCESS in 46m 18s\n- devstack-nfv-numa http://52.27.155.124/53/748453/9/check/devstack-nfv-numa/0448e12/ : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":9},{"id":"da45396f62b3dd56dddaaf6d2c0578aae70c4c98","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-11-18 12:45:30.000000000","message":"Patch Set 9:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/53/748453/9/check-vote/ext-nova-zuul/04d8358 : SUCCESS in 1h 05m 58s","accounts_in_message":[],"_revision_number":9},{"id":"02df0183cbc801acfdc5c71c172132fee70809c6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-18 14:34:25.000000000","message":"Patch Set 9:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/9/check/tempest-dsvm-full-focal-py3/44c9a5a/ : SUCCESS in 2h 34m 30s","accounts_in_message":[],"_revision_number":9},{"id":"0fb7e424959ab3a85c02e8bf75980bf4063bff6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-18 19:02:20.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2b1f52d508a5440c84fec8c6a9ab469d : FAILURE in 43m 00s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1be05d91ae924fcaa9f1b89c05069678 : SUCCESS in 1h 22m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a05ceaf97ec24126b8f04908459b86cd : SUCCESS in 18m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ef0c8e3ece349838fadd7ffa1e8efc9 : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b289dfabb274b09a989eb8a2b7bb052 : SUCCESS in 7m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34882614d4354ba08f5629f21f3bcf2f : SUCCESS in 10m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c402dbb50bcc422d8912d551deb5afd1 : SUCCESS in 13m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c139383fc9e3456b960d01af98ee1ee4 : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c29a09118a146418b1c39be3875a034 : SUCCESS in 10m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ff4a663250342de9965665e08e034ca : SUCCESS in 1h 12m 37s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d617008f57da4a4d83e02733ada60d64 : SUCCESS in 1h 16m 13s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a9cb2e04123d4f94bbae3ec5a7b83ae1 : SUCCESS in 1h 23m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f99ca55ef6434a9da7d36f3113b7e63b : SUCCESS in 51m 36s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f4c6974337f847e58adc909906442cd6 : SUCCESS in 1h 27m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/9a02865ded0d4b2890cb21012c17cd3f : SUCCESS in 1h 51m 44s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7407a73202af495b8946bf2e6739b6b0 : SUCCESS in 16m 46s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b1706de406f4f55a051034185c23ca0 : SUCCESS in 19m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e7a5c63b7064be3b35c91182ea3a80b : SUCCESS in 1h 00m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80ad361da3cb4d3f85cc7dafd0a9f6d7 : SUCCESS in 52m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/707d2220faad4313b8eb07ee02eecd4e : SUCCESS in 28m 20s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dfc404b0007247bd8d68f2c660ffab24 : FAILURE in 52m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9d983c2f7e5be62ec861a9c9c1c7d72e2362dc65","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-11-19 14:59:26.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c74d4ba34553701c22b896238f2e367fb5876da3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2020-11-24 11:54:14.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"a59b44cb1dd4e87897a454887e2d0769e3062c26","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-11-24 14:12:01.000000000","message":"Patch Set 10:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/53/748453/10/check/tempest-dsvm-full-focal-py3/d94a504/ : SUCCESS in 1h 49m 58s","accounts_in_message":[],"_revision_number":10},{"id":"6c41103837f15d6fbedf977df4c76c2a8b2c4ac7","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-11-24 14:55:51.000000000","message":"Patch Set 10:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/748453/10 : FAILURE in 2h 59m 49s","accounts_in_message":[],"_revision_number":10},{"id":"36c33f09a17b89ff6f3b6be735d61721387b066c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-24 16:19:58.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f564de3a803404f9909167a3c46cad0 : SUCCESS in 1h 03m 09s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3f011214e0e544a88662dd93c28cb497 : SUCCESS in 1h 18m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ef6b43672eb406c8ae73f0ef538663c : SUCCESS in 16m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0100a81bb0224c39bb8f2522486b05b1 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64008e6ffb5489e9c04ae8e8c21e4bb : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1951841266ff4545b2f33d94919b18e9 : SUCCESS in 12m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6698ddcbe9642ff9789e6f198ed123d : SUCCESS in 11m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de6cd47adc1c46d6b3e83cb84d59844b : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175dfc503a284de585e8dde3b86e6d8a : SUCCESS in 11m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/455b4113d35644d2b14998bd99181a48 : SUCCESS in 1h 19m 52s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b8a37bf1c1142eaba4be00ceafa3f12 : SUCCESS in 1h 17m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d21044c019e94e1da26bf5deb4d2a6c0 : SUCCESS in 1h 38m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a918722c563e4d30b36252b01ba780f9 : SUCCESS in 49m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/30a9b593928b44efb1bed4ed627953ba : SUCCESS in 1h 30m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/fbb56338ec1545cfa44d171cf3e82df9 : SUCCESS in 2h 53m 04s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/363df953b91842dd84fe0605d5ae4258 : SUCCESS in 13m 51s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42517f5d1c524288a77a5c562530230f : SUCCESS in 19m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0771793cd8a243c49fb573a9c32711b4 : SUCCESS in 1h 01m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dbb6044ea624b958a18c09cab6686af : SUCCESS in 56m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/92b53a1e6e99480584b276fa43270901 : SUCCESS in 26m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3d3b1503929b4c788158c7fe94ca888b : SUCCESS in 38m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5f3ed305deb1d2ee7bcb11a03835274b4e4bc84b","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-11-24 16:37:50.000000000","message":"Patch Set 10:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/53/748453/10/check/EMC_PowerFlex_NOVA/d402793/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":10},{"id":"5b9432b10eb6cf07031310d41f4456a3edc605df","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-01-20 12:17:46.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\n(1 comment)\n\nLooks like a good fix to me.","accounts_in_message":[],"_revision_number":10},{"id":"6d3e9da6237a0f4d17032ab3a42bf0fce50773ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 12:19:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2d437bc25ba5a1347944fcd8c0f5ba12428920df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 17:53:43.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba18947da67145d9be0dbb4c6d0e90eb : SUCCESS in 57m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d1929a26ff1546b68f1c21769db8e192 : SUCCESS in 1h 24m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9fea76fdc3449dfaeba8ace7d9d9571 : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19905c18259d4c7eb6906c00213bb4af : SUCCESS in 8m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f70fa3e1f9e842e0afed3cbd1c38eb88 : SUCCESS in 13m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d98ad63f1217499a9bb9f01b966c9394 : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2493b2d7bc94f818e166d378dfbb088 : SUCCESS in 12m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/17bf694287f24fef8276ac5bb5f2616a : SUCCESS in 1h 13m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4e4c2a74153c47df95d80514854882aa : SUCCESS in 1h 08m 56s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/458c58c169634df09c6af8d579430138 : SUCCESS in 17m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/46f89e112dd943e6b0cde4fdd102919c : SUCCESS in 1h 38m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/7401576a243d492e92110ea188c6777a : SUCCESS in 1h 27m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/293ef25810e8437ba005e3e4cb999a5c : SUCCESS in 1h 14m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bfb43b46a7f345228be37c378ff31837 : SUCCESS in 1h 19m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ef663241c974cd3adc8b5bd87d4438a : SUCCESS in 1h 08m 48s","accounts_in_message":[],"_revision_number":10},{"id":"195f289edbac7879cb35550fa0eb3fb9f4d9ba1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 17:53:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"1273c5ee0b18974d9837e9221fc9270429d428bf","revisions":{"d759ca3af6123e2444653bc0b43c1117abe30b49":{"kind":"REWORK","_number":1,"created":"2020-08-27 14:56:37.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/1"}}},"commit":{"parents":[{"commit":"d38d6966be906cb9bfe950e924b7abebb7a0e52e","subject":"Only unplug vif after the device is detached from libvirt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d38d6966be906cb9bfe950e924b7abebb7a0e52e"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:33:20.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d759ca3af6123e2444653bc0b43c1117abe30b49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d759ca3af6123e2444653bc0b43c1117abe30b49"}]},"branch":"refs/heads/master"},"b511751eb4168a9510640f8a134e2ab59fe5f27d":{"kind":"REWORK","_number":2,"created":"2020-08-28 11:40:11.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/2"}}},"commit":{"parents":[{"commit":"d38d6966be906cb9bfe950e924b7abebb7a0e52e","subject":"Only unplug vif after the device is detached from libvirt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d38d6966be906cb9bfe950e924b7abebb7a0e52e"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-28 11:37:35.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b511751eb4168a9510640f8a134e2ab59fe5f27d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b511751eb4168a9510640f8a134e2ab59fe5f27d"}]},"branch":"refs/heads/master"},"442981e16f9d7e331c2a3bf8ccdaa3d1ca79728c":{"kind":"REWORK","_number":3,"created":"2020-08-28 16:55:26.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/3"}}},"commit":{"parents":[{"commit":"d38d6966be906cb9bfe950e924b7abebb7a0e52e","subject":"Only unplug vif after the device is detached from libvirt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d38d6966be906cb9bfe950e924b7abebb7a0e52e"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-28 12:03:21.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/442981e16f9d7e331c2a3bf8ccdaa3d1ca79728c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/442981e16f9d7e331c2a3bf8ccdaa3d1ca79728c"}]},"branch":"refs/heads/master"},"fbc90b7e4f709493bf96bd88aff9f4bf4e5c221c":{"kind":"REWORK","_number":4,"created":"2020-09-01 15:23:41.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/4"}}},"commit":{"parents":[{"commit":"d38d6966be906cb9bfe950e924b7abebb7a0e52e","subject":"Only unplug vif after the device is detached from libvirt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/d38d6966be906cb9bfe950e924b7abebb7a0e52e"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-09-01 15:09:52.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbc90b7e4f709493bf96bd88aff9f4bf4e5c221c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fbc90b7e4f709493bf96bd88aff9f4bf4e5c221c"}]},"branch":"refs/heads/master"},"fdeae19875a7a068cedd5fdd3fcb817b1c21a1ec":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-09-02 08:24:22.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/5"}}},"commit":{"parents":[{"commit":"7e8a9786dc3d330b79970f398e73fa596dc9445a","subject":"Only unplug vif after the device is detached from libvirt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7e8a9786dc3d330b79970f398e73fa596dc9445a"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-09-01 15:25:37.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdeae19875a7a068cedd5fdd3fcb817b1c21a1ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdeae19875a7a068cedd5fdd3fcb817b1c21a1ec"}]},"branch":"refs/heads/master"},"cd619b874d03bd12bd32306c550c2b77db9e742f":{"kind":"REWORK","_number":6,"created":"2020-09-07 09:09:16.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/6"}}},"commit":{"parents":[{"commit":"a1169be58e15a1cfa7865667e2ca229cf22243b2","subject":"Support SRIOV interface attach and detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a1169be58e15a1cfa7865667e2ca229cf22243b2"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-09-07 09:06:44.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd619b874d03bd12bd32306c550c2b77db9e742f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/cd619b874d03bd12bd32306c550c2b77db9e742f"}]},"branch":"refs/heads/master"},"3512795656ae30619c5fa0eedaf14bf7488b05b3":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-09-10 17:45:39.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/748453/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/7"}}},"commit":{"parents":[{"commit":"1361ea5ad128e7048430612e01d97281fd094f05","subject":"Support SRIOV interface attach and detach","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1361ea5ad128e7048430612e01d97281fd094f05"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2020-09-10 17:44:56.000000000","tz":60},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3512795656ae30619c5fa0eedaf14bf7488b05b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3512795656ae30619c5fa0eedaf14bf7488b05b3"}]},"branch":"refs/heads/master"},"2ec67e284e2b34b0f7919f2354bb0d01f4451360":{"kind":"REWORK","_number":8,"created":"2020-09-14 15:09:24.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/8"}}},"commit":{"parents":[{"commit":"18bae88fd80c48a7587dc7a7662e786c32e3f6d6","subject":"Merge \"doc: mark the max microversion for victoria\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/18bae88fd80c48a7587dc7a7662e786c32e3f6d6"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-09-14 15:07:21.000000000","tz":120},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ec67e284e2b34b0f7919f2354bb0d01f4451360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2ec67e284e2b34b0f7919f2354bb0d01f4451360"}]},"branch":"refs/heads/master"},"b0fbb4c169648b00ecb3d9352ad9fc450f653efc":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2020-11-16 10:23:48.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/9"}}},"commit":{"parents":[{"commit":"43b8df3ae8c00819c955a61326435ea15fd8642c","subject":"Merge \"Remove compute service level check for qos ops\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/43b8df3ae8c00819c955a61326435ea15fd8642c"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-16 10:23:48.000000000","tz":0},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0fbb4c169648b00ecb3d9352ad9fc450f653efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b0fbb4c169648b00ecb3d9352ad9fc450f653efc"}]},"branch":"refs/heads/master"},"1273c5ee0b18974d9837e9221fc9270429d428bf":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2020-11-24 11:54:14.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/53/748453/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/53/748453/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/53/748453/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/53/748453/10"}}},"commit":{"parents":[{"commit":"e26a805d157a5c233273edd85c502a00d8198b70","subject":"Merge \"virt: Remove \u0027change_instance_metadata\u0027 API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e26a805d157a5c233273edd85c502a00d8198b70"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-08-27 13:12:50.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2020-11-24 11:54:14.000000000","tz":0},"subject":"Make PCI claim NUMA aware during live migration","message":"Make PCI claim NUMA aware during live migration\n\nNUMA aware live migration and SRIOV live migration was implemented as\ntwo separate feature. As a consequence the case when both SRIOV and NUMA\nis present in the instance was missed. When the PCI device is claimed on\nthe destination host the NUMA topology of the instance needs to be\npassed to the claim call.\n\nChange-Id: If469762b22d687151198468f0291821cebdf26b2\nCloses-Bug: #1893221\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1273c5ee0b18974d9837e9221fc9270429d428bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1273c5ee0b18974d9837e9221fc9270429d428bf"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
