)]}'
{"id":"openstack%2Fnova~749223","triplet_id":"openstack%2Fnova~stable%2Frocky~I369f82245465e96fc15d4fc71a79a8a71f6f2c6d","project":"openstack/nova","branch":"stable/rocky","topic":"bug/1889108","hashtags":[],"change_id":"I369f82245465e96fc15d4fc71a79a8a71f6f2c6d","subject":"Robustify attachment tracking in CinderFixtureNewAttachFlow","status":"MERGED","created":"2020-09-01 10:29:20.000000000","updated":"2020-09-19 14:20:07.000000000","submitted":"2020-09-19 14:20:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"749223-1600525207393-0e5654ba","meta_rev_id":"7ea65718c1370217fb678177de65d1892e4bdb8a","_number":749223,"virtual_id_number":749223,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"date":"2020-09-14 09:33:41.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"date":"2020-09-19 12:41:18.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"date":"2020-09-19 13:02:27.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-19 14:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":2,"date":"2020-09-18 21:39:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":2,"date":"2020-09-14 14:54:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"all":[{"value":1,"date":"2020-09-18 21:39:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"value":0,"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-01 10:29:20.000000000","updated_by":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"reviewer":{"_account_id":6873,"name":"Matt Riedemann","email":"mriedem.os@gmail.com","username":"mriedem"},"state":"REVIEWER"},{"updated":"2020-09-14 09:33:41.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-14 14:54:50.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-09-18 21:39:16.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-09-19 13:02:27.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-19 14:20:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d230600dc26dcd2b4fd97f58fdadb856114c583b","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-01 10:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5851fe860f5dc64462b851c57e05b7fc43a8f87f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-01 10:40:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/749223/1 : FAILURE in 9m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8f3f374ecf22ee8249e5643f6930a6bc287399fb","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-01 11:14:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"630a5d5e2e7c495a1740880a35f676e6493dc4b5","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-01 11:25:17.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/749223/2 : FAILURE in 8m 44s","accounts_in_message":[],"_revision_number":2},{"id":"c009e80f8d17c7579c85fcb66160281de7955e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 11:41:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a90984630cb44c4a8bf02508f1a1f3b4 : FAILURE in 10m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c41b33a554443f9aecaef9ff4ec70ff : SUCCESS in 13m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83b9eaef5f7a476cad92da088594c0f2 : SUCCESS in 14m 35s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d1be73e7cef840ea9dc99b93e5392dd4 : SUCCESS in 9m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/56b2ad01b25c4d4d97bbb3040fe43928 : FAILURE in 7m 30s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/602341344457409080e543270522bd4d : SUCCESS in 19m 33s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e6605c8670534ec9aa777e2036c8505e : SUCCESS in 16m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd9ce9171e2a4fc0bd1e8de163983f5b : SUCCESS in 14m 35s","accounts_in_message":[],"_revision_number":2},{"id":"015c9ab8fc8963cd3037a5dd0e9c0934cf6868c5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-01 19:40:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f93233b64b3992b3cc1e10f551409db99e58b88f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-01 19:53:20.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/749223/2 : FAILURE in 10m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9e3ee945f782709773cd6ea45b153e10258a8224","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-01 20:33:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b35c5db37c1148a6a1cb99a91ce6fe0e : FAILURE in 12m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1fde93f8829c4fe4bc4ed1da572fd188 : SUCCESS in 14m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7752defc69404a1dacd8cb4a0247c2cc : SUCCESS in 18m 22s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e13b9652d9d3415896e7f498970fcd00 : SUCCESS in 11m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/443330bf18024d1fbcd7d56a0a665568 : FAILURE in 11m 15s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/f14b4862907549b1a0a61a55ad7a5393 : SUCCESS in 17m 14s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/df73d6ecc7bc4d18b8c8ce156bf6cea2 : SUCCESS in 17m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d5dd0cbc2e541419e596b101158f8f5 : SUCCESS in 14m 55s","accounts_in_message":[],"_revision_number":2},{"id":"f2f70405439c629013f77a614c600c5647e98be4","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-02 19:12:13.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/23/749223/2/check/pmem-tempest-plugin-filtered-multi-node/d328b76/ : RETRY_LIMIT in 5m 12s\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmp0j1yzcsr/8e9a41b947534d3087cf272a1bb501c7/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":2},{"id":"d08fa56a233ddf5fa28206c47b8ca434ce4ae387","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-09 13:21:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9c84e8cc118c3a0a14925ef936e98992c2a3810a","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-09 13:33:53.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/749223/2 : FAILURE in 10m 18s","accounts_in_message":[],"_revision_number":2},{"id":"de2c9faf9cdd6130d64168fcef7150a373977a56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 14:39:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02ee464d1e264a0ab10379ad57a9104a : SUCCESS in 10m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a9531b3dac874fe38403c8a994af3b9d : SUCCESS in 14m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/87b97f07498c4082b1995f1ba5749b92 : SUCCESS in 11m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/049fbc9e6f3c422ea33a076ea174fa0b : SUCCESS in 10m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3ae0a9f9764541778498989b3d214057 : SUCCESS in 55m 04s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/7134831eae054cc8864cbcf865d7615c : SUCCESS in 18m 35s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/0397540bca9c4dff9d7ac69612696ccc : SUCCESS in 21m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c022ccb6f494986828e10a25b987469 : SUCCESS in 15m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d27cbb76a9eafb1e8fe7dd14e9b6e96a68b09c44","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-14 09:21:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dd85b6be3b10076abcabb3b393257a6578692889","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-14 09:33:41.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/749223/3 : FAILURE in 9m 43s","accounts_in_message":[],"_revision_number":3},{"id":"476919fccca9eec4bc82cf52b694787aefc28327","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-14 09:52:10.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/23/749223/3/check/pmem-tempest-plugin-filtered-multi-node/c21dce2/ : RETRY_LIMIT in 5m 12s\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpl_lcnf6d/7b78dd1feb5d4794b999dd30b37a0a85/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":3},{"id":"e22a60772655ad80f46751a7ace73929f896886e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 10:25:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/883c4e09036a4f0ab508ef34de527300 : SUCCESS in 11m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4e358bb5070402ab88a741184027f68 : SUCCESS in 13m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/568d19b8070f48358f0ba85583703ee5 : SUCCESS in 14m 14s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/80fe39c1a2034ebbac07dd4b985d6853 : SUCCESS in 12m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6790ce00a0fb45d2bbbb507929853cfc : SUCCESS in 51m 07s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/86bf82335f9845aea456d48bcb256573 : SUCCESS in 16m 23s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/e8b79cf347f44c4a9237b972bceb6821 : SUCCESS in 18m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/88c0e70b47f74deebe5760c52aee48ca : SUCCESS in 13m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9915012a55910fbe129c20dad683b1d685cb9f1f","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-14 14:54:50.000000000","message":"Patch Set 3: Code-Review+2\n\nClean cherry pick of a test only change that helps with the testing of bug/1889108. Already backported and merged to Stein.","accounts_in_message":[],"_revision_number":3},{"id":"0e217c2e932dad8e873f508198d5337c4f19f072","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-09-18 21:39:16.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nClean backport from stein, testing refactor to aid in a bug fix and ease future volume related bug fix backports, LGTM","accounts_in_message":[],"_revision_number":3},{"id":"24d092b2bc814543a8561d1b9e890a01d75a847e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 21:40:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"32598d506505dda5f749f8d22b85c4e27e491316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 23:03:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aef1c3e828a4fa3936295fcc437ec1c : SUCCESS in 11m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a778f69eee2a44ec8f5365ecbe90d829 : SUCCESS in 13m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1bd142a622054dd285c3af9a75c7198c : SUCCESS in 13m 47s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/74b8f4dad21e4fdfb8ec631439ff3978 : SUCCESS in 9m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7d49dd4b430e400191b2c8057dc3bfb9 : SUCCESS in 35m 51s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/8c55d0b50ffe4472bc5d1f2e6d9982ef : FAILURE in 15m 53s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/516f6620fd794c5e8322b22c85daefc6 : SUCCESS in 13m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cf8ca73093b4ff293122485aa09fc98 : SUCCESS in 19m 26s","accounts_in_message":[],"_revision_number":3},{"id":"417c21a47e91cf30b634429bf56e40edf1e68688","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-09-19 12:41:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ccc603ac5e9ec4bbdd34e98403bbc37e6b303333","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-19 13:02:27.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/23/749223/3/check/pmem-tempest-plugin-filtered-multi-node/db5155a/ : RETRY_LIMIT in 4m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f217e7afae9b63d232eddf14051fcf1f7ec3667b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 13:38:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b4d8a8959849359ad003f949a83dd4 : SUCCESS in 10m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/79789de6e33140f1bc660ab9e78d5483 : SUCCESS in 12m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ddf910f443a4ab0a0642d2e4085183c : SUCCESS in 12m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5cf96fb0aa1d41258c95d3bdc48309a2 : SUCCESS in 8m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/85d36de06a094fe3b197e921f6b29e4c : SUCCESS in 49m 45s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/657f3d4564b0413c8e86b0b835a212a5 : SUCCESS in 20m 39s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/cf36223452ec46968c4c59bcb8a18d2b : SUCCESS in 16m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5101796d42b480d994f6b38ac930d4d : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":3},{"id":"1dcfce2f4cb6aa08bdf9a9ffa894e303a243b7ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 13:38:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"539baa7aa76cf94ab82c843f2c1aea13ab2c00fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 14:20:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7e4c9de2bd4bda85680bebca676d53 : SUCCESS in 12m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3290542f154a4bb1b7e4a854d0027e97 : SUCCESS in 18m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e310cc71c7ea42c0804c9f537a89d8fb : SUCCESS in 19m 15s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/5271335f8fc34a39b4c33d44b969e154 : SUCCESS in 8m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e65a2e5bdfe74f948e717c12330e8208 : SUCCESS in 37m 03s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/737a21330dce4bf8aa8b6653ccf96542 : SUCCESS in 15m 47s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/eab94420f6cc45c1ba2b19652dff60b1 : SUCCESS in 14m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a214f3bdfd14a1c80e2c88838176714 : SUCCESS in 11m 54s","accounts_in_message":[],"_revision_number":3},{"id":"ccb82fc37e7a1da783a6732f7df563c301801e69","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-19 14:20:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c1b372fc9264be8f7feb23fc939cfbe254da6eec","revisions":{"ea3a254726e5887c51c92ecb5c89928b58020447":{"kind":"REWORK","_number":1,"created":"2020-09-01 10:29:20.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/23/749223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/749223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/749223/1"}}},"commit":{"parents":[{"commit":"c1fbb25408274a290976899213f6f69e46ef4f02","subject":"Improve CinderFixtureNewAttachFlow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1fbb25408274a290976899213f6f69e46ef4f02"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-13 18:42:25.000000000","tz":-240},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-01 10:24:44.000000000","tz":60},"subject":"Robustify attachment tracking in CinderFixtureNewAttachFlow","message":"Robustify attachment tracking in CinderFixtureNewAttachFlow\n\nCinder allows more than one attachment record on a non-multiattach\nvolume as long as only one attachment is \"connected\" (has a host\nconnector).\n\nWhen you create an empty (no connector) attachment for an in-use\nvolume, the volume status changes to \"attaching\". If you try to\nupdate the empty attachment before deleting the previous attachment,\nCinder will return a 400 like:\n\n  Invalid volume: Volume b2aba195-6570-40c4-82bb-46d3557fceeb status\n  must be available or downloading to reserve, but the current status\n  is attaching.\n\nThis change refactors the attachment tracking in the\nCinderFixtureNewAttachFlow fixture to be able to track the connector\nper attachment and if code tries to update an attachment on an\nalready connected volume it will fail.\n\nChange-Id: I369f82245465e96fc15d4fc71a79a8a71f6f2c6d\n(cherry picked from commit 1d3ca5a3a07fdfff0d61ac11c390dfd4acab23d7)\n(cherry picked from commit 1e1bbdc4cdd0104c214134ef964e8621eab3de9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea3a254726e5887c51c92ecb5c89928b58020447"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ea3a254726e5887c51c92ecb5c89928b58020447"}]},"branch":"refs/heads/stable/rocky"},"b6dc6e5d87bcc3bdb5418f1ed87ba5613d43f8ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-01 11:14:17.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/23/749223/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/749223/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/749223/2"}}},"commit":{"parents":[{"commit":"6d1503442154c15d6f3a4ef874890109dc9a7088","subject":"Improve CinderFixtureNewAttachFlow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6d1503442154c15d6f3a4ef874890109dc9a7088"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-13 18:42:25.000000000","tz":-240},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-01 11:14:00.000000000","tz":60},"subject":"Robustify attachment tracking in CinderFixtureNewAttachFlow","message":"Robustify attachment tracking in CinderFixtureNewAttachFlow\n\nCinder allows more than one attachment record on a non-multiattach\nvolume as long as only one attachment is \"connected\" (has a host\nconnector).\n\nWhen you create an empty (no connector) attachment for an in-use\nvolume, the volume status changes to \"attaching\". If you try to\nupdate the empty attachment before deleting the previous attachment,\nCinder will return a 400 like:\n\n  Invalid volume: Volume b2aba195-6570-40c4-82bb-46d3557fceeb status\n  must be available or downloading to reserve, but the current status\n  is attaching.\n\nThis change refactors the attachment tracking in the\nCinderFixtureNewAttachFlow fixture to be able to track the connector\nper attachment and if code tries to update an attachment on an\nalready connected volume it will fail.\n\nChange-Id: I369f82245465e96fc15d4fc71a79a8a71f6f2c6d\n(cherry picked from commit 1d3ca5a3a07fdfff0d61ac11c390dfd4acab23d7)\n(cherry picked from commit 1e1bbdc4cdd0104c214134ef964e8621eab3de9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6dc6e5d87bcc3bdb5418f1ed87ba5613d43f8ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6dc6e5d87bcc3bdb5418f1ed87ba5613d43f8ac"}]},"branch":"refs/heads/stable/rocky"},"c1b372fc9264be8f7feb23fc939cfbe254da6eec":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-14 09:21:54.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/23/749223/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/23/749223/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/23/749223/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/23/749223/3"}}},"commit":{"parents":[{"commit":"b84eb40a66ccd74b8862cfa8708c9d3eda68a3c4","subject":"Improve CinderFixtureNewAttachFlow","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b84eb40a66ccd74b8862cfa8708c9d3eda68a3c4"}]}],"author":{"name":"Matt Riedemann","email":"mriedem.os@gmail.com","date":"2019-05-13 18:42:25.000000000","tz":-240},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-09-14 09:20:11.000000000","tz":60},"subject":"Robustify attachment tracking in CinderFixtureNewAttachFlow","message":"Robustify attachment tracking in CinderFixtureNewAttachFlow\n\nCinder allows more than one attachment record on a non-multiattach\nvolume as long as only one attachment is \"connected\" (has a host\nconnector).\n\nWhen you create an empty (no connector) attachment for an in-use\nvolume, the volume status changes to \"attaching\". If you try to\nupdate the empty attachment before deleting the previous attachment,\nCinder will return a 400 like:\n\n  Invalid volume: Volume b2aba195-6570-40c4-82bb-46d3557fceeb status\n  must be available or downloading to reserve, but the current status\n  is attaching.\n\nThis change refactors the attachment tracking in the\nCinderFixtureNewAttachFlow fixture to be able to track the connector\nper attachment and if code tries to update an attachment on an\nalready connected volume it will fail.\n\nChange-Id: I369f82245465e96fc15d4fc71a79a8a71f6f2c6d\n(cherry picked from commit 1d3ca5a3a07fdfff0d61ac11c390dfd4acab23d7)\n(cherry picked from commit 1e1bbdc4cdd0104c214134ef964e8621eab3de9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1b372fc9264be8f7feb23fc939cfbe254da6eec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c1b372fc9264be8f7feb23fc939cfbe254da6eec"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
