)]}'
{"id":"openstack%2Fnova~751373","triplet_id":"openstack%2Fnova~stable%2Fpike~I8610af4f93f4d87001a8ff6687b42f126b94a6d5","project":"openstack/nova","branch":"stable/pike","topic":"do-not-merge","hashtags":[],"change_id":"I8610af4f93f4d87001a8ff6687b42f126b94a6d5","subject":"DNM: zuul fix check","status":"ABANDONED","created":"2020-09-11 17:55:05.000000000","updated":"2020-09-14 10:11:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"00895728336345548428901aa0d6452924543d9b","_number":751373,"virtual_id_number":751373,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-09-12 02:16:54.000000000","_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"date":"2020-09-11 17:57:15.000000000","_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"date":"2020-09-12 06:48:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"date":"2020-09-11 18:12:34.000000000","_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-09-11 20:11:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 17:57:15.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"REVIEWER"},{"updated":"2020-09-11 18:12:34.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"REVIEWER"},{"updated":"2020-09-11 20:11:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-12 02:16:54.000000000","updated_by":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-12 06:48:34.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25446b68cab22ae363c522ac60f50a6db6d66a7d","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-11 17:55:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b2da65b1c66e99d1e7d89afc4d989a3844b0863","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-11 17:57:07.000000000","message":"Topic set to do-not-merge","accounts_in_message":[],"_revision_number":1},{"id":"bab99b30d38d95083e6a1006bff28e4b49915b44","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-09-11 17:57:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/751373/1 : FAILURE in 1m 10s","accounts_in_message":[],"_revision_number":1},{"id":"2e84e0c66d737da4bd258869cadba4d5674a6f08","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2020-09-11 18:12:34.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- pmem-tempest-plugin-filtered-multi-node http://52.27.155.124/73/751373/1/check/pmem-tempest-plugin-filtered-multi-node/058e229/ : RETRY_LIMIT in 5m 22s\n- devstack-networking-ovs-dpdk devstack-networking-ovs-dpdk : ERROR Unable to find role in /tmp/tmpidyh3zm7/cd04ded6edb34945918988b3fe5397d7/ansible/pre_playbook_4/role_0/neutron","accounts_in_message":[],"_revision_number":1},{"id":"b782800b9eea6e52fe4d0dab6b6d05cbfc95e361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 20:11:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d923efe1db4fbc83aeb1c5a35e2044 : FAILURE in 8m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/84704fa288e54812a41f0e14e5fc68c8 : SUCCESS in 12m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b790a3672d3744fa9be2f411e9fda7c0 : SUCCESS in 13m 09s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/52de2a1dd7a043ac9105cce19c0e2536 : SUCCESS in 9m 41s\n- tempest-full https://zuul.opendev.org/t/openstack/build/afe6b840b04046439a1d29bfc090a13b : TIMED_OUT in 2h 14m 12s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/f7a26ffb97274529adeca42f282f04d3 : FAILURE in 28m 05s (non-voting)\n- ironic-tempest-dsvm-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eab3705c3be5449e8b31f2f8981d26ae : FAILURE in 36m 03s (non-voting)\n- nova-grenade-live-migration https://zuul.opendev.org/t/openstack/build/04c9a651047d47a497c31d84da94adbd : FAILURE in 37m 26s (non-voting)\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2b031224c51d408481223c50eb3791c5 : SUCCESS in 44m 39s\n- tempest-slow https://zuul.opendev.org/t/openstack/build/a0b376f7ab52405b86cb0bd938baab74 : SUCCESS in 1h 49m 05s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/27798a0143174f2583bc634b89d7b231 : SUCCESS in 21m 04s\n- nova-tox-functional-py35 https://zuul.opendev.org/t/openstack/build/77b837aee4b44941b15996f82dfd4837 : SUCCESS in 17m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdbf783108ce47a1a9d70375caabedc1 : FAILURE in 37m 55s (non-voting)\n- tempest-multinode-full https://zuul.opendev.org/t/openstack/build/d5580372c0464aa3923c695a495999a3 : SUCCESS in 1h 56m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b33ccf76f311738a0e3e71a3164323d57cd87930","author":{"_account_id":14595,"name":"z Systems KVM","email":"zkvm-ci@linux.vnet.ibm.com","username":"ibm-zkvm-ci","tags":["SERVICE_USER"]},"date":"2020-09-12 02:16:54.000000000","message":"Patch Set 1:\n\nTesting failed ubuntu-bionic-s390x. For rechecking only on the ubuntu-bionic-s390x CI, add a review comment with recheck-zkvm. Contact info: zkvm-ci@linux.vnet.ibm.com. For more information, see https://wiki.openstack.org/wiki/ThirdPartySystems/IBM_zKVM_CI\n\n- check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x http://extbasicopstackcilog01.podc.sl.edst.ibm.com/zkvm_test_logs/production/751373/1/check-tempest-dsvm-neutron-full-ubuntu-bionic-s390x/baf20b1 : FAILURE in 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"4ae5ed67919e6338685e01b4dfc3c3f72a1f0407","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-09-12 06:48:34.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- legacy-tempest-dsvm-full-xenial https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/751373/1/check/legacy-tempest-dsvm-full-xenial/cb117d3/ : TIMED_OUT in 2h 37m 06s\n- legacy-tempest-dsvm-full-xenial-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/73/751373/1/check/legacy-tempest-dsvm-full-xenial-py3/8d59b13/ : FAILURE in 1h 15m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00895728336345548428901aa0d6452924543d9b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-09-14 10:11:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46476fa06854b55b28ce63f171fabbb91ee8cf36","revisions":{"46476fa06854b55b28ce63f171fabbb91ee8cf36":{"kind":"REWORK","_number":1,"created":"2020-09-11 17:55:05.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/73/751373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/73/751373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/751373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/751373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/73/751373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/73/751373/1"}}},"commit":{"parents":[{"commit":"550c3db939a85aac1e13b480a6cd91ac0569bbea","subject":"Merge \"Fix os-simple-tenant-usage result order\" into stable/pike","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/550c3db939a85aac1e13b480a6cd91ac0569bbea"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-09-11 17:50:21.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2020-09-11 17:53:32.000000000","tz":120},"subject":"DNM: zuul fix check","message":"DNM: zuul fix check\n\nDepends-On: https://review.opendev.org/#/c/751313\nChange-Id: I8610af4f93f4d87001a8ff6687b42f126b94a6d5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46476fa06854b55b28ce63f171fabbb91ee8cf36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/46476fa06854b55b28ce63f171fabbb91ee8cf36"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
