)]}'
{"id":"openstack%2Fnova~767251","triplet_id":"openstack%2Fnova~master~I66c209baf11c37ffebca52764263daae9e1dd50b","project":"openstack/nova","branch":"master","topic":"bug/1908336","hashtags":[],"change_id":"I66c209baf11c37ffebca52764263daae9e1dd50b","subject":"Improving the description for unshelve request body","status":"MERGED","created":"2020-12-16 01:50:49.000000000","updated":"2021-01-06 00:01:11.000000000","submitted":"2021-01-05 23:59:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"767251-bug/1908336","meta_rev_id":"d06bf089fba3a53d575cce249b817028cf9d9a33","_number":767251,"virtual_id_number":767251,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-05 23:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-05 18:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":2,"date":"2021-01-05 12:22:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-05 18:53:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-16 01:55:02.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-12-16 01:55:02.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2020-12-16 01:55:02.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2020-12-16 01:55:02.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2020-12-16 04:19:24.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2020-12-16 15:11:47.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-16 18:48:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-24 08:41:39.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2020-12-24 13:23:58.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-24 13:58:53.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2020-12-25 09:49:47.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-12-25 14:35:21.000000000","updated_by":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"reviewer":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"state":"REVIEWER"},{"updated":"2020-12-28 14:19:41.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-12-29 07:36:36.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"0d54b053fd86dc199ea080cacbcff071c80b53d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-16 01:50:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4898dffd1f2cc52baa7c4c0bb2172894cefae0c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-16 02:20:38.000000000","message":"Topic bp/1908336 removed","accounts_in_message":[],"_revision_number":1},{"id":"64d631cde3d14f89a3ccf9c0e1c565541293dd4c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-16 02:20:57.000000000","message":"Topic set to bug/1908336","accounts_in_message":[],"_revision_number":1},{"id":"2d31df2f05680e51f052bf13bd0e125cf201d0fb","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-12-16 04:19:24.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/767251/1 : FAILURE in 2h 27m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ddbb8ec6952ef4fa69206b0fac3f52833fc517e1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-16 15:11:47.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/767251/767251/1/check-not-vote/ext-nova-zuul/4d7fed9 : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8bf627b8ed7a4dea1dd3e235f21b498a0db6c0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 18:48:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/27fe5a2c1df94f34af5a78aaabbc54d2 : FAILURE in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abdea4a8fc54c7a9f530e77287ca269 : SUCCESS in 7m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e48ce7ae41cc40c9a09af00712483866 : SUCCESS in 10m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a381b87e0ed1a6c9b1538c0267e9138da247a46f","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-17 05:48:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6b2fbc5c35034c0125a5e931b0151b19f36a7a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 06:02:41.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c74a5f23e28740d6972fbd0693b07d4f : FAILURE in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42309ac587d44f6fbaeb2c8467882ff8 : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db6790ae1050441d9deda3c91597110b : SUCCESS in 10m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b99a53e32df9faac9ce82a424e565a336131f428","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-17 07:58:01.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/767251/767251/1/check-not-vote/ext-nova-zuul/832a982 : SUCCESS in 2h 07m 55s","accounts_in_message":[],"_revision_number":1},{"id":"9139aa026c69514b45bfb11c1e0fa01e01818f3b","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-12-17 08:21:10.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/767251/1 : SUCCESS in 2h 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"241d1d98e28e007fdae88df8529eb1973db59607","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2020-12-24 08:41:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddc71a4741e3427dd56e3765b0fa5571bbfd768d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-24 09:49:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d7a760e0002957892d8c369552b2a8e38eed650","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-24 13:12:20.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"4bfc2a8b55ecc931b6297a40d1d438a1801395fd","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-24 13:12:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"163e596d041b92a8f64723ca20bed778f1605968","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-24 13:23:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0dd10c25dd6c4fe3824a96e21404c97e : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbae567daec74353bcb7476730e79230 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5899ac95bf48f6b87b132a101a4b3b : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":2},{"id":"a4a37db4c8af08f4813c94f76898597e5bd02f83","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-24 13:23:58.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/51/767251/2/check/EMC_PowerFlex_NOVA/9ad44d0/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"6c31abb74b0306dcd213326033478d21775fb235","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-12-24 13:58:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nQuick nit but thanks for documenting this!","accounts_in_message":[],"_revision_number":2},{"id":"f623a8f09277f97064857fade4890d0d19af6aca","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-12-24 15:40:15.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- nova http://cloudbase-ci.com/nova/767251/2 : SUCCESS in 2h 26m 39s","accounts_in_message":[],"_revision_number":2},{"id":"1f50328c303170d067a2b583dc24a4f09030d846","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-24 16:13:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"885e912c77db8e1f9af23249c2d80cab800de0a9","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-25 00:40:40.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThanks gmann, lyarwood. addressed your comments.","accounts_in_message":[],"_revision_number":2},{"id":"ef25ebabd03a69a250f23f54048344b76ba2288f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-25 00:40:54.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4cc970569321dbd07390a6916da4adc82a77c108","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 00:55:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/57a58603874a4a89a9683c2164ef12d1 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ba943e88b794f3b8b36c43d8cbe7050 : SUCCESS in 8m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ce02ad2cbe54f1da616fc62dd756efe : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":3},{"id":"638fe509f833e7f2f6ff11f751278540047d0d2a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-25 00:55:48.000000000","message":"Patch Set 3:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/51/767251/3/check/EMC_PowerFlex_NOVA/b26879c/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"5478a8d3d9935e0bc4a9205d373ef5030a16483b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 01:34:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\n1 comment","accounts_in_message":[],"_revision_number":3},{"id":"5ecbbfd44eca5932cf506f3ae2a6aac9a4a5234e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-25 01:45:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7653a3039c4c701fb0b8f76cf510c6683edc4f57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-12-25 01:52:13.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a1d4a1a7fdb26296763eb6fdbe88232be9d859af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-25 02:06:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fa2289e89539417cb24d2585d7848b26 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29ed88c3c8c4ac39653ea25dceafa8d : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1bcb0a7071945809adf336ee63b33a2 : SUCCESS in 9m 31s","accounts_in_message":[],"_revision_number":5},{"id":"5144bdbdf86f902b168c642d2b16d45807bb1d36","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2020-12-25 02:07:13.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/51/767251/5/check/EMC_PowerFlex_NOVA/ee7835f/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"d902cd3c6229eb3977cfce9f098dc670a5e1cd7e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-12-25 02:12:59.000000000","message":"Patch Set 5: Code-Review+2\n\nthanks, lgtm","accounts_in_message":[],"_revision_number":5},{"id":"ba0f17a72b376329cfd826a4a369d3c455e06cde","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-12-25 04:30:50.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/767251/5 : FAILURE in 2h 37m 31s","accounts_in_message":[],"_revision_number":5},{"id":"2e52d5e6b910c365a70251aacd2e24e0809df311","author":{"_account_id":32067,"name":"Linhui Zhou","email":"df.some@foxmail.com","username":"zhoulinhui"},"date":"2020-12-25 14:35:21.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"cce98ca97c4926ff4afaea2889b75567a0dad94d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2020-12-28 13:13:55.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/767251/767251/5/check-not-vote/ext-nova-zuul/d8b4f48 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":5},{"id":"35a7bfa3ee23d6030645fa22036660f75934d1ac","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-12-28 14:19:41.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"bd325c286e790bab8c73f63dc5e4f5184dd4e5b3","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2020-12-29 07:36:36.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8591adc377f06b01a3f66c129c3cb18b859bdec7","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-01-05 12:15:34.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\n-1 due to the example needs to be valid json","accounts_in_message":[],"_revision_number":5},{"id":"7d58dbcc319fdf4ec924375b605427531253b87d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-01-05 12:21:10.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"aa0bb4f870353d15926f329f90497e770b2f59f8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-01-05 12:22:40.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks. Looks good.","accounts_in_message":[],"_revision_number":6},{"id":"6f43a6270cad52cd569a5c3e88953018ffb1a067","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-01-05 12:22:48.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nThanks gibi.","accounts_in_message":[],"_revision_number":6},{"id":"ecc359ccef51644017d44d3014cc1ffdb090586d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 12:35:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/35e0e4df71cd4501b14162261bb5dc6c : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a658605ac19c4d4da640672b393213cf : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ace46d7293794463ba9dc0f0dff0698c : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":6},{"id":"a86c8bbe33373365c61dac6053a700d76e25e2e9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-01-05 12:36:10.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/51/767251/6/check/EMC_PowerFlex_NOVA/7ce4ec7/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"b368c47964f8832ddfd4d38b76587ad253edd73c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-01-05 13:27:02.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/767251/767251/6/check-not-vote/ext-nova-zuul/04198ff : SUCCESS in 1h 04m 15s","accounts_in_message":[],"_revision_number":6},{"id":"d597f47f80cba92227db6e636e1da2a3cf5a8b1e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-01-05 15:05:09.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/767251/6 : FAILURE in 2h 42m 43s","accounts_in_message":[],"_revision_number":6},{"id":"5b6a4eb1da001ee8e6a937650b210621397187b1","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-05 18:53:30.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nlgtm.","accounts_in_message":[],"_revision_number":6},{"id":"50cb0954c9259a78adf19096cf707aa16a0dccd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 18:54:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d4aa6171770d9863a2817a558c11250d6d6e6cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 23:59:14.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f158d0b18f8445c089e91140b36273a8 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec33e04a78b54d6cae39956bb5529cee : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5126138f8a054656a340d2aabea0b542 : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":6},{"id":"5c23153a4cd34aa5a51081b360d3e460d63a76e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-05 23:59:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d06bf089fba3a53d575cce249b817028cf9d9a33","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-06 00:01:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c0c964c38fd44a77ab3075094efee774 : SUCCESS in 58s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d98c8554a741e6a901b9368db17e9e : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b4f560d4e193f457d2cd5724c62d2b6adcaf332b","revisions":{"361137647688cb77ca7e7db343bad44520b34df7":{"kind":"REWORK","_number":1,"created":"2020-12-16 01:50:49.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/51/767251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/767251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/767251/1"}}},"commit":{"parents":[{"commit":"6c96602864403123f99995773990287a256c833e","subject":"Merge \"libvirt: make cross cell resize spawn from snapshot image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c96602864403123f99995773990287a256c833e"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:07.000000000","tz":0},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:10.000000000","tz":0},"subject":"Improving the description for az of unshleve API","message":"Improving the description for az of unshleve API\n\nThe \u0027availability_zone\u0027 description in unshleve api-ref is confusing,\nadding NOTE for it:\n\nSince microversion 2.77, allowed request body schema are {\u0027unshelve\u0027:\nnull} or {\u0027unshelve\u0027: {\u0027availability_zone\u0027: \u003cstring\u003e}}, and {\u0027unshelve\u0027:\n{}} request body is not allowed.\n\nCloses-Bug: #1908336\nChange-Id: I66c209baf11c37ffebca52764263daae9e1dd50b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/361137647688cb77ca7e7db343bad44520b34df7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/361137647688cb77ca7e7db343bad44520b34df7"}]},"branch":"refs/heads/master"},"9ee8db454a36185a2a2d45923a5af6ad0acdd095":{"kind":"REWORK","_number":2,"created":"2020-12-24 13:12:20.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/51/767251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/767251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/767251/2"}}},"commit":{"parents":[{"commit":"6c96602864403123f99995773990287a256c833e","subject":"Merge \"libvirt: make cross cell resize spawn from snapshot image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c96602864403123f99995773990287a256c833e"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:07.000000000","tz":0},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-12-24 13:10:18.000000000","tz":0},"subject":"Improving the description for az of unshleve API","message":"Improving the description for az of unshleve API\n\nThe \u0027availability_zone\u0027 description in unshleve api-ref is confusing,\nadding NOTE for it:\n\nSince microversion 2.77, allowed request body schema are {\u0027unshelve\u0027:\nnull} or {\u0027unshelve\u0027: {\u0027availability_zone\u0027: \u003cstring\u003e}}, and {\u0027unshelve\u0027:\n{}} request body is not allowed.\n\nCloses-Bug: #1908336\nChange-Id: I66c209baf11c37ffebca52764263daae9e1dd50b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ee8db454a36185a2a2d45923a5af6ad0acdd095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9ee8db454a36185a2a2d45923a5af6ad0acdd095"}]},"branch":"refs/heads/master"},"9918ede0d056edcca03c9c6dd0f1a390376c4c18":{"kind":"REWORK","_number":3,"created":"2020-12-25 00:40:54.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/51/767251/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/767251/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/767251/3"}}},"commit":{"parents":[{"commit":"6c96602864403123f99995773990287a256c833e","subject":"Merge \"libvirt: make cross cell resize spawn from snapshot image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c96602864403123f99995773990287a256c833e"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:07.000000000","tz":0},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-12-25 00:39:55.000000000","tz":0},"subject":"Improving the description for az of unshleve API","message":"Improving the description for az of unshleve API\n\nThe \u0027availability_zone\u0027 description in unshleve api-ref is confusing,\nadding NOTE for it:\n\nSince microversion 2.77, allowed request body schema are {\u0027unshelve\u0027:\nnull} or {\u0027unshelve\u0027: {\u0027availability_zone\u0027: \u003cstring\u003e}}, and {\u0027unshelve\u0027:\n{}} request body is not allowed.\n\nCloses-Bug: #1908336\nChange-Id: I66c209baf11c37ffebca52764263daae9e1dd50b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9918ede0d056edcca03c9c6dd0f1a390376c4c18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/9918ede0d056edcca03c9c6dd0f1a390376c4c18"}]},"branch":"refs/heads/master"},"945fd6e13399372bcdcd46a835177e51cd5897ab":{"kind":"REWORK","_number":4,"created":"2020-12-25 01:45:30.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/51/767251/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/767251/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/767251/4"}}},"commit":{"parents":[{"commit":"6c96602864403123f99995773990287a256c833e","subject":"Merge \"libvirt: make cross cell resize spawn from snapshot image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c96602864403123f99995773990287a256c833e"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:07.000000000","tz":0},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-25 01:44:23.000000000","tz":0},"subject":"Improving the description for unshleve request body","message":"Improving the description for unshleve request body\n\nThe \u0027availability_zone\u0027 description in unshleve api-ref is confusing,\nadding NOTE for it:\n\nSince microversion 2.77, allowed request body schema are {\u0027unshelve\u0027:\nnull} or {\u0027unshelve\u0027: {\u0027availability_zone\u0027: \u003cstring\u003e}}, and {\u0027unshelve\u0027:\n{}} request body is not allowed.\n\nCloses-Bug: #1908336\nChange-Id: I66c209baf11c37ffebca52764263daae9e1dd50b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/945fd6e13399372bcdcd46a835177e51cd5897ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/945fd6e13399372bcdcd46a835177e51cd5897ab"}]},"branch":"refs/heads/master"},"bb66f1e1f49216c724c672d04bcfb74f07922886":{"kind":"REWORK","_number":5,"created":"2020-12-25 01:52:13.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/51/767251/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/767251/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/767251/5"}}},"commit":{"parents":[{"commit":"6c96602864403123f99995773990287a256c833e","subject":"Merge \"libvirt: make cross cell resize spawn from snapshot image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c96602864403123f99995773990287a256c833e"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:07.000000000","tz":0},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-25 01:50:50.000000000","tz":0},"subject":"Improving the description for unshleve request body","message":"Improving the description for unshleve request body\n\nThe \u0027availability_zone\u0027 description in unshleve api-ref is confusing,\nadding NOTE for unshelve request body:\n\nSince microversion 2.77, allowed request body schema are {\u0027unshelve\u0027:\nnull} or {\u0027unshelve\u0027: {\u0027availability_zone\u0027: \u003cstring\u003e}}. A request body\nof {\u0027unshelve\u0027:{}} is not allowed.\n\nCloses-Bug: #1908336\nChange-Id: I66c209baf11c37ffebca52764263daae9e1dd50b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb66f1e1f49216c724c672d04bcfb74f07922886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bb66f1e1f49216c724c672d04bcfb74f07922886"}]},"branch":"refs/heads/master"},"b4f560d4e193f457d2cd5724c62d2b6adcaf332b":{"kind":"REWORK","_number":6,"created":"2021-01-05 12:21:10.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/51/767251/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/51/767251/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/51/767251/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/51/767251/6"}}},"commit":{"parents":[{"commit":"6c96602864403123f99995773990287a256c833e","subject":"Merge \"libvirt: make cross cell resize spawn from snapshot image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c96602864403123f99995773990287a256c833e"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-12-16 01:47:07.000000000","tz":0},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2021-01-05 12:21:01.000000000","tz":0},"subject":"Improving the description for unshelve request body","message":"Improving the description for unshelve request body\n\nThe \u0027availability_zone\u0027 description in unshelve api-ref is confusing,\nadding NOTE for unshelve request body:\n\nSince microversion 2.77, allowed request body schema are {\"unshelve\":\nnull} or {\"unshelve\": {\"availability_zone\": \u003cstring\u003e}}. A request body\nof {\"unshelve\":{}} is not allowed.\n\nCloses-Bug: #1908336\nChange-Id: I66c209baf11c37ffebca52764263daae9e1dd50b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4f560d4e193f457d2cd5724c62d2b6adcaf332b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b4f560d4e193f457d2cd5724c62d2b6adcaf332b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
