)]}'
{"id":"openstack%2Fnova~768736","triplet_id":"openstack%2Fnova~stable%2Ftrain~I9d8d8c3a30cf6da7e8fb48374347e069ab075df2","project":"openstack/nova","branch":"stable/train","topic":"bug/1841932","hashtags":[],"change_id":"I9d8d8c3a30cf6da7e8fb48374347e069ab075df2","subject":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"","status":"MERGED","created":"2020-12-29 08:42:26.000000000","updated":"2021-03-12 20:52:33.000000000","submitted":"2021-03-12 20:49:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"768736-bug/1841932","cherry_pick_of_change":722187,"cherry_pick_of_patch_set":9,"meta_rev_id":"9fdaa2555e22913aed56b39402d26680a18e1d5b","_number":768736,"virtual_id_number":768736,"owner":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 20:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-01 16:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2021-03-12 16:29:55.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":1,"date":"2021-03-12 16:29:55.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"}],"reviewers":{"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}],"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-29 10:09:22.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2020-12-29 10:16:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-29 11:11:54.000000000","updated_by":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"reviewer":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"state":"REVIEWER"},{"updated":"2020-12-29 13:51:14.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-01-20 16:04:33.000000000","updated_by":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"reviewer":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"state":"REVIEWER"},{"updated":"2021-02-03 14:06:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-03-12 16:29:55.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"2e6bab6902bebbbf8ae4797472e087b5fa9c101d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"date":"2020-12-29 08:42:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* nova/api/validation/extra_specs/hw.py\n* nova/api/validation/extra_specs/null.py\n* nova/tests/unit/api/openstack/compute/test_flavors_extra_specs.py","accounts_in_message":[],"_revision_number":1},{"id":"94b60d909207ca62b9b8708909a882cd72737ae6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-29 10:09:22.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/36/768736/1/check/tempest-dsvm-full-bionic-py3/8093405/ : SUCCESS in 1h 19m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1f7b43c0288a0b92c0f29e1c4b1493dfac8c4a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 10:16:52.000000000","message":"Patch Set 1: Verified-1\n\n(39 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/02913d5e65ff421fac0a4ee24803cd98 : SUCCESS in 1h 03m 47s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f07b2e17b0af46bdb16ad5f73b924e9b : SUCCESS in 51m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e87d8e66239b4c159fc804b9bbbf7b7d : FAILURE in 9m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df95d0653f36489ca3b348d808f0eda1 : FAILURE in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/844701764c6d47d8a12b95a1f0f54816 : FAILURE in 16m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bb947ceecf8a44a690efba4dc3ec8d98 : FAILURE in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2d6deb87de34577b3587c86c59e1e35 : FAILURE in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a8b9097781f476c943360841a1c3131 : FAILURE in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/123009a3ee764514b0b7a61559141619 : SUCCESS in 11m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f7ed10f05734e3383ed684ac944187e : SUCCESS in 10m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/95094691c06445fb9c6e3c7286a2f245 : SUCCESS in 1h 01m 51s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/7d3201e4b62f4b3e9173f07c8794f16f : SUCCESS in 59m 14s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f5c2101f6244433b8fc2af2b91340761 : SUCCESS in 1h 10m 35s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/59ccbb79fd2647eda4104ea3db7b4bec : SUCCESS in 50m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/396a90e4ff7b4a48aad85c009ecf11cc : SUCCESS in 40m 22s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/8a3a89df995a427aa117b8fde54f5521 : SUCCESS in 36m 47s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/b21e7a4795dc4522994036beeaee44dc : SUCCESS in 1h 28m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/69ff39d75bc7474f9f6c5cdb72af4d62 : SUCCESS in 1h 06m 17s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/ce18cb445652412393a646f8139980f4 : FAILURE in 4m 39s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/233cee1a9c224457a078320f8410d92a : SUCCESS in 16m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6f1b066c352849a2befc27634cc9d50d : SUCCESS in 1h 27m 59s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/e9a597aa34d64d50ba99a6c0ba119c7b : FAILURE in 20m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ef97a871ed240df88844884032d605e : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"11b9a0d24302e6a77278d44a20b59dc5cbe06011","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2020-12-29 11:11:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bfac0f04a38875b18856f2d5deb14dc9fd21f93b","author":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"date":"2020-12-29 11:17:47.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"3438af79862db48a163cacf366dcde5aff1d8767","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2020-12-29 12:57:00.000000000","message":"Patch Set 2:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/36/768736/2/check/tempest-dsvm-full-bionic-py3/350d2f5/ : SUCCESS in 1h 40m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3c4f1dce4c33a3ac21a584e3435707139a219ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 13:27:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2dc3771db5cd42a0a32fa5b850231a12 : SUCCESS in 55m 29s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0597c2e3a58c41ae83a53bf131b5f706 : SUCCESS in 1h 14m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/407943218d624a818f861190189d7462 : SUCCESS in 15m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/665572018b2c4e71b9dc42701725cf97 : FAILURE in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b8371a1261442498f4c656716aa91f : SUCCESS in 10m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d8e978b5ee05495482b7e98d50cf57bd : FAILURE in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe09faf75e5340a5bb621c97b6703feb : SUCCESS in 12m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/773fb454103a492e939ba884b42656b0 : SUCCESS in 12m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9f1082265244585996ff3a37926bf50 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27720ba8603d4e73b603821f0473fc13 : SUCCESS in 15m 06s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b8f137cfdf384bdd9ff4444405e8dae1 : SUCCESS in 59m 53s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/f817a1c484bb4ff9bef3fda9a1e4d6ff : SUCCESS in 58m 43s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/2d4c9cdcf4a04a9898300f6f25fc2027 : SUCCESS in 1h 22m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a5b96c2baeb34f7abe84d2ecacd82ba0 : SUCCESS in 56m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8bc064f8d19e4aec8635aef506447ef5 : SUCCESS in 59m 02s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/46bc66dc952743129c96af383204acb8 : SUCCESS in 49m 38s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/05b92e1459ac4779a1d0dc34a6853405 : SUCCESS in 1h 06m 06s\n- nova-next https://zuul.opendev.org/t/openstack/build/2b6038a0138e4040be06a7536bb337f5 : FAILURE in 1h 23m 10s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/d4a99d730ca543169d4bdb43dd93a152 : FAILURE in 3m 57s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2041bb666d540c4a50fbef05de245bb : SUCCESS in 18m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0205c4d6c85f4991ae1b5a54d01f8a97 : SUCCESS in 2h 07m 33s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/c96333a16adb4bd89b1c241eaf0c109a : FAILURE in 22m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d24c43950e14e51be07e868586a100a : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":2},{"id":"903262c60c7037cdec0cd1da0858c970a27edd0e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2020-12-29 13:51:14.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/768736/2 : FAILURE in 2h 38m 09s","accounts_in_message":[],"_revision_number":2},{"id":"db3697745f8f08d460ab781dc7649afba81c1224","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2020-12-31 02:40:33.000000000","message":"Patch Set 2:\n\nwaitting for https://review.opendev.org/c/openstack/nova/+/766171 is merged","accounts_in_message":[],"_revision_number":2},{"id":"3573be8328f54ec817ea267e13e6c3bb3ee2d00c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-01-19 08:40:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"c28b3870afa9af10f0d126293605770223150751","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-01-19 10:16:02.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/36/768736/3/check/tempest-dsvm-full-bionic-py3/2d3d835/ : SUCCESS in 1h 31m 06s","accounts_in_message":[],"_revision_number":3},{"id":"7c888d490fad1cb6f8beadf8d2cffb082509673c","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-01-19 11:29:40.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/768736/3 : FAILURE in 2h 48m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a88bd43cdda751b00bd7b08602561d5bd8f6c642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:40:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/132a7391e5a84b149667842ce44a6ae4 : SUCCESS in 1h 02m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/094f9a421bda4548a293cfc621bd836c : SUCCESS in 1h 10m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/165bba5d7d6e48808f738b8bf01c161c : SUCCESS in 17m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b13948cae51a40c28d94d189bc986d2d : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25afe699913b40cf99619be7dd5bc67f : SUCCESS in 11m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dea570d522543568dab35eb35a31a4b : SUCCESS in 13m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fba18a4b49034666b7cd1e2aa93774c3 : SUCCESS in 12m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/be9ce103c57d43898d1362670f9ff8d4 : SUCCESS in 12m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2df19fcbd294498bb2f63ce65d78526 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f50ad8f22065472fa17beeffc5ca98c9 : SUCCESS in 16m 10s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cdc0bd02b4e747b0ac1ba1528b6bfd37 : SUCCESS in 1h 08m 55s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/667f5862c78e4050a336de539f01e568 : SUCCESS in 1h 19m 00s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/f42531c511804c8eb97ef6190a1e8994 : SUCCESS in 1h 04m 26s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/dc6fb642310e45c4b9b472a185675f23 : SUCCESS in 1h 03m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2d3bf3cf6b4242449b8cf3448d32b35d : SUCCESS in 1h 05m 33s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/eafe5963ed074ce78c101a5b34cd1ed3 : SUCCESS in 44m 19s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d8b239cea15449ad937426ccf3690d63 : SUCCESS in 1h 17m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/7bd1502c68be4fe0bd4ccbf0ebaced13 : SUCCESS in 1h 16m 07s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/deb6527ae9804e5d8ded004303da19b6 : SUCCESS in 21m 29s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39e5f8d6a04445dc9952fb71cff063d4 : SUCCESS in 20m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8966f1f1d1dd4e77a967652bcfdf5d5a : SUCCESS in 2h 00m 10s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/5dcbc1f8682e48dfbb6601469d8748be : SUCCESS in 43m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a7bc836de8343938fd959baeac00d14 : SUCCESS in 1h 00m 22s","accounts_in_message":[],"_revision_number":3},{"id":"6a65452b7eb24a45f02e388c330521ff91d5d824","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2021-01-20 16:04:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7f5adc36ad03b2d96952f0f80707a3a074d69a2c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-02-03 14:06:47.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)\n\nIf I\u0027m not mistaken, this bug is valid in Train (so is the bugfix). Anyway, there is an issue with the commit message as a \u0027cherry picked from\u0027 line is missing (see my comment inline). Except this, the patch looks OK to me.\n\nOne additional comment: usually nova developers add a \u0027note\u0027 with explanation for the conflicts to make the reviewing a bit easier.","accounts_in_message":[],"_revision_number":3},{"id":"a42fc386a4b625617a4b720bba494b07b5ac215a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-03-01 02:58:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d571b73115b3d997638be650ec627e3457ea10f4","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-03-01 03:01:46.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e If I\u0027m not mistaken, this bug is valid in Train (so is the bugfix). Anyway, there is an issue with the commit message as a \u0027cherry picked from\u0027 line is missing (see my comment inline). Except this, the patch looks OK to me.\n\u003e \n\u003e One additional comment: usually nova developers add a \u0027note\u0027 with explanation for the conflicts to make the reviewing a bit easier.\n\nThanks, Elod Illes. Now the commit message looks much better.","accounts_in_message":[],"_revision_number":4},{"id":"cbdf051696974d8b7435bb2ac425112f04b90038","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-01 04:35:04.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/36/768736/4/check/tempest-dsvm-full-bionic-py3/89b1cba/ : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":4},{"id":"39981572309226483d45c50b9026c948cc37d1e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 04:45:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/95700a59e5d744fab8c683086b4eff08 : SUCCESS in 1h 35m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c470be9aa76b43c9b0c33163aa91e468 : FAILURE in 26m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8668d979e15c4b99842bdca67fe1d0cc : SUCCESS in 23m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1f6e0152594df98816db758497684b : SUCCESS in 15m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d6ae6a9e7b45419b83d2d82885554d34 : FAILURE in 29m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/590ac3d21d7348d3bd7dd04697679677 : SUCCESS in 22m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/522bbbe95e1847e9860f9bae45965550 : FAILURE in 30m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64216b21d5a34c759fe1c688f8fbbdaa : SUCCESS in 20m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9adb65824ebb4277ba2c5f791c3ef578 : SUCCESS in 12m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e093aa509ade4511a0c0b25d4d48cb49 : SUCCESS in 55m 56s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/e59b7cd8a2174060b03ed3792a6364a1 : SUCCESS in 1h 19m 59s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cd8b07be2964ae0be76da3bea7ed5b3 : FAILURE in 42m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6652ac2bb2ee4c16a3e9900001894dfd : FAILURE in 50m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fcdc1568b0704a61831dccf5f451604d : SUCCESS in 49m 00s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/a5b8e0eedf1f4a87b52cbe9485d87578 : SUCCESS in 59m 19s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6daed4acf659404fab1d4b27b4b68b25 : SUCCESS in 1h 19m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/1cbe620ac2244621bfe66c3077455f31 : SUCCESS in 1h 12m 02s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/e8b2cb7fb19342489271f45ef7e39a58 : SUCCESS in 16m 35s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3a39674437634feba95bab60e4738dcb : SUCCESS in 23m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e822ba93083f4681a4ed64b353135ae8 : SUCCESS in 1h 44m 28s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/fd52ee075edc42d1bfb04fcf52a7c191 : SUCCESS in 47m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2d7ebe9911ef46c2bcafe7359fd2557b : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":4},{"id":"3e1f47a5f082d7ff1e60a923bbdc80d21758ef0a","author":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"date":"2021-03-01 07:22:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"78843e485d71d26a38804d373b3191ab41d440fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 09:11:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0d951f6b2ac44c6495c1969a5430c09b : SUCCESS in 1h 19m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d311b5ec7daf4c10aec0f1e02deae20d : SUCCESS in 16m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/70f9f7aab3484e2fabce58fb37c9cb63 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d648577967ac4bd29cc566bb34acadc7 : SUCCESS in 10m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/867fdaeb23b3464d9fdf473f51aa247c : SUCCESS in 13m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8364e75bcfe24294a43a7bf5f4bc1404 : SUCCESS in 11m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/859b4a6d349c4da09b2c299355a5ffe2 : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa6943b41ce444593e78c34a8e4cb3f : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c91823cde11847009b68f46d2dedd30b : SUCCESS in 14m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/726ae6fa4fd541e2844f63dd6eae8713 : SUCCESS in 1h 08m 38s (non-voting)\n- devstack-plugin-ceph-tempest https://zuul.opendev.org/t/openstack/build/1d0eb58be9ff4f22bf683470027d2f91 : SUCCESS in 1h 24m 32s (non-voting)\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6d92b0dc092475aab80ce359bd9450e : SUCCESS in 1h 00m 33s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/acffcd608e434d51adbf31c68a97ccc7 : SUCCESS in 1h 03m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/25c01a72f305495db010b42a134a5c3b : SUCCESS in 43m 36s\n- nova-lvm https://zuul.opendev.org/t/openstack/build/d968c751f08e40168cfbfe9502241f9a : SUCCESS in 1h 05m 17s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f386d970b4eb4ce08e1f43ecac26b37f : SUCCESS in 1h 19m 26s\n- nova-next https://zuul.opendev.org/t/openstack/build/a131d445d7b841eda0a22483777ddb32 : SUCCESS in 1h 11m 11s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/4c0fb651439d4b978ed9538c09846d23 : SUCCESS in 19m 38s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2ff41b81da8a4ab1b721f036ee8fadfb : SUCCESS in 17m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f56f0caf0dd34b70b11fdaf8aa0dc1f4 : SUCCESS in 1h 47m 48s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/3a7d9fd52597491393a26f4db47ef1ce : SUCCESS in 56m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df0f132d573b4e358f94bd0fa9600e7f : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":4},{"id":"9706ae7e32ba67b5f7f0a8db241e696f197e543a","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-01 11:28:02.000000000","message":"Patch Set 4:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-bionic-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/36/768736/4/check/tempest-dsvm-full-bionic-py3/f1764a3/ : SUCCESS in 1h 44m 12s","accounts_in_message":[],"_revision_number":4},{"id":"c81a7ac3365fc354f5d9eff4e1a15871a4ebd0af","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-01 16:19:53.000000000","message":"Patch Set 4: Code-Review+2\n\nJust a note: the previous test run contains unit test failure of \u0027test_walk_versions\u0027 (bug: [1]) at multiple unit test jobs. This made me wonder how could that happen, as that issue is intermittent and usually rare... Anyway, I couldn\u0027t connect those failures to this patch, so the backport looks OK to me now. Thanks Xing for updating the commit message! (patch merged on ussuri, conflicts noted)\n\n[1] https://bugs.launchpad.net/nova/+bug/1823251","accounts_in_message":[],"_revision_number":4},{"id":"9f355a773676481dcdc267e807c68ffe4688473e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-12 16:29:55.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"fc34edbf032aeb72a90a83596867cc8b1bff4afe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 16:30:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"657bab0e52bdadc21bae6d3019f7c7830a51f1dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:49:43.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4cfa4f92746e4ab998b7f162c68fb394 : SUCCESS in 1h 24m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de3af1a3aad5432fb33f6a68a957187d : SUCCESS in 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b184ea31be44cd88b70ed1169030c98 : SUCCESS in 9m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/476a8b6dd73b4037ad2137ad3336a0bf : SUCCESS in 13m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0cd6d9698f04513bd7120c70c4f875f : SUCCESS in 16m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cda111c340394d8ebd7dccf5d8fcc571 : SUCCESS in 16m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e91cbc0a18e459b8b9c4f08d6396c51 : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bc82821b3954bad971471cfad6c7fdc : SUCCESS in 16m 30s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb9c36120da141838ba0a002375bfb3c : SUCCESS in 51m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/90bdb03e54e64a69b8c52cd667320ab6 : SUCCESS in 50m 06s\n- nova-tox-functional https://zuul.opendev.org/t/openstack/build/ebde383829af43c184628cb135a9c1cf : SUCCESS in 26m 08s\n- nova-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b88f0274f705461c8cf970ef3188cd49 : SUCCESS in 20m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/407b380f850549b092148b5809fa3045 : SUCCESS in 1h 37m 19s\n- nova-next https://zuul.opendev.org/t/openstack/build/89945004edb64eefbd028d6e056f5ec9 : SUCCESS in 1h 22m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5f140c03149b49d9832a9464d88a5e99 : SUCCESS in 2h 03m 12s\n- grenade-py3 https://zuul.opendev.org/t/openstack/build/b90f6800c89b4863a8c3b85207391630 : SUCCESS in 45m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f4e1c0200dd412fb4eae51f854f5e07 : SUCCESS in 43m 33s","accounts_in_message":[],"_revision_number":4},{"id":"db348e2acf732f21ee0bb0779f9dd69c32606495","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:49:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9fdaa2555e22913aed56b39402d26680a18e1d5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 20:52:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c81708e2daa54368829500f47f9eb896 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa615e3d34a6415993af3e03c32ea69a : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f60253677830483e63f73bcfd554f5e165cfc850","revisions":{"2117fabeea2a97e9e7c1e3625a6c5ba402cc20a8":{"kind":"REWORK","_number":1,"created":"2020-12-29 08:42:26.000000000","uploader":{"_account_id":24849,"name":"yao ning","email":"zay11022@163.com","username":"mslovy"},"ref":"refs/changes/36/768736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/36/768736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/36/768736/1"}}},"commit":{"parents":[{"commit":"81a3f4b22b867b372b94af81872b547e7500005c","subject":"Merge \"Change default num_retries for glance to 3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81a3f4b22b867b372b94af81872b547e7500005c"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-04-22 13:33:22.000000000","tz":480},"committer":{"name":"yao ning","email":"zay11022@163.com","date":"2020-12-29 08:42:26.000000000","tz":0},"subject":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"","message":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"\n\nWhen we use the flavor extra_specs \"hide_hypervisor_id\" in\nAggregateInstanceExtraSpecsFilter, then will retrun False.\nSo we need correct the extra_specs.\n\nChange-Id: I9d8d8c3a30cf6da7e8fb48374347e069ab075df2\nCloses-Bug: 1841932\n(cherry picked from commit bf488a8630702160021b5848bf6e86fbb8015205)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2117fabeea2a97e9e7c1e3625a6c5ba402cc20a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2117fabeea2a97e9e7c1e3625a6c5ba402cc20a8"}]},"branch":"refs/heads/stable/train"},"30f28d3ba469983dae296edc9e49115ce9c704ec":{"kind":"REWORK","_number":2,"created":"2020-12-29 11:11:54.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/36/768736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/36/768736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/36/768736/2"}}},"commit":{"parents":[{"commit":"81a3f4b22b867b372b94af81872b547e7500005c","subject":"Merge \"Change default num_retries for glance to 3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/81a3f4b22b867b372b94af81872b547e7500005c"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-04-22 13:33:22.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2020-12-29 11:11:39.000000000","tz":480},"subject":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"","message":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"\n\nWhen we use the flavor extra_specs \"hide_hypervisor_id\" in\nAggregateInstanceExtraSpecsFilter, then will retrun False.\nSo we need correct the extra_specs.\n\nConflicts:\n\tnova/api/validation/extra_specs/hw.py\n\tnova/api/validation/extra_specs/null.py\n\tnova/tests/unit/api/openstack/compute/test_flavors_extra_specs.py\n\nChange-Id: I9d8d8c3a30cf6da7e8fb48374347e069ab075df2\nCloses-Bug: 1841932\n(cherry picked from commit bf488a8630702160021b5848bf6e86fbb8015205)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30f28d3ba469983dae296edc9e49115ce9c704ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/30f28d3ba469983dae296edc9e49115ce9c704ec"}]},"branch":"refs/heads/stable/train"},"224fb5f76d6f93ad0afc4a3db74f8d5e60060a26":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-19 08:40:06.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/36/768736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/36/768736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/36/768736/3"}}},"commit":{"parents":[{"commit":"b2037fc4e356b55949339a1358c16431a9ab8930","subject":"[stable-only] Cap bandit to 1.6.2 and raise hacking, flake8 and stestr","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b2037fc4e356b55949339a1358c16431a9ab8930"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-04-22 13:33:22.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-01-19 08:40:06.000000000","tz":0},"subject":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"","message":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"\n\nWhen we use the flavor extra_specs \"hide_hypervisor_id\" in\nAggregateInstanceExtraSpecsFilter, then will retrun False.\nSo we need correct the extra_specs.\n\nConflicts:\n\tnova/api/validation/extra_specs/hw.py\n\tnova/api/validation/extra_specs/null.py\n\tnova/tests/unit/api/openstack/compute/test_flavors_extra_specs.py\n\nChange-Id: I9d8d8c3a30cf6da7e8fb48374347e069ab075df2\nCloses-Bug: 1841932\n(cherry picked from commit bf488a8630702160021b5848bf6e86fbb8015205)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/224fb5f76d6f93ad0afc4a3db74f8d5e60060a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/224fb5f76d6f93ad0afc4a3db74f8d5e60060a26"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"f60253677830483e63f73bcfd554f5e165cfc850":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2021-03-01 02:58:56.000000000","uploader":{"_account_id":21757,"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","username":"angeiv"},"ref":"refs/changes/36/768736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/36/768736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/36/768736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/36/768736/4"}}},"commit":{"parents":[{"commit":"e39e622cb004b7f8e986efe2ee814a3526205b82","subject":"Merge \"Disallow CONF.compute.max_disk_devices_to_attach \u003d 0\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e39e622cb004b7f8e986efe2ee814a3526205b82"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-04-22 13:33:22.000000000","tz":480},"committer":{"name":"Xing Zhang","email":"angeiv.zhang@gmail.com","date":"2021-03-01 02:35:26.000000000","tz":480},"subject":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"","message":"replace the \"hide_hypervisor_id\" to \"hw:hide_hypervisor_id\"\n\nWhen we use the flavor extra_specs \"hide_hypervisor_id\" in\nAggregateInstanceExtraSpecsFilter, then will retrun False.\nSo we need correct the extra_specs.\n\nNOTE: The first two files do not exist in stable/train due to\nextra specs validators (patch Ib64a1348cce1dca995746214616c4f33d9d664bd)\nwas introduced in Ussuri.\nThe last one file has conflict due to the same\nbp/flavor-extra-sepc-validators feature\n(patch: I8da84b48e4d630eeb91d92346aa2323e25e28e3b)\nadded in Ussuri.\n\nConflicts:\n\tnova/api/validation/extra_specs/hw.py\n\tnova/api/validation/extra_specs/null.py\n\tnova/tests/unit/api/openstack/compute/test_flavors_extra_specs.py\n\nChange-Id: I9d8d8c3a30cf6da7e8fb48374347e069ab075df2\nCloses-Bug: 1841932\n(cherry picked from commit bf488a8630702160021b5848bf6e86fbb8015205)\n(cherry picked from commit 9d28d7ec808469ec129b66c69b9e63cd9537a63f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f60253677830483e63f73bcfd554f5e165cfc850"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f60253677830483e63f73bcfd554f5e165cfc850"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
