)]}'
{"id":"openstack%2Fnova~774904","triplet_id":"openstack%2Fnova~master~I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c","project":"openstack/nova","branch":"master","topic":"bp/allow-secure-boot-for-qemu-kvm-guests","hashtags":[],"change_id":"I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c","subject":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","status":"MERGED","created":"2021-02-10 17:15:36.000000000","updated":"2021-03-01 19:47:28.000000000","submitted":"2021-03-01 19:43:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"774904-bp/allow-secure-boot-for-qemu-kvm-guests","meta_rev_id":"5c10bb63b5a1e1eebf400e1b59ca5e8c8679e912","_number":774904,"virtual_id_number":774904,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 19:43:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2021-03-01 16:15:10.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-22 16:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-03-01 16:15:10.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 19:09:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-10 19:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-11 00:49:58.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-02-12 17:59:44.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 20:00:03.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-17 14:52:50.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-02-22 16:59:51.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-03-01 16:15:10.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"c44dc51a691cd65e70ba011ec5858ae686c2b7f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-10 17:15:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e802fd69dac5e71de3aa7259a921bdb4896a50c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:09:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c73978c0c35b3626c1bbed8ed7b47af6ebd92322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 19:17:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35e1ea512dea4c5fabdb70d105eed403 : FAILURE in 14m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/64eb15fb2a314904a090a0b96af999fa : FAILURE in 17m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98cf4b92ad3645f3ab9c8a757f147cd9 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffec766526e041a9913902c58248c156 : FAILURE in 13m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0336625a493f44cf9e754bd58601fc38 : FAILURE in 12m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dff36f8f70404f93ba0081a91ed1320a : FAILURE in 16m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fe6007bc3b4f3bb76d76379935b60e : SUCCESS in 15m 51s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04f68e239f4f4d8a82ab173994cc573e : SUCCESS in 18m 09s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e73283f431a74ec7b3506a2cac6ab137 : SUCCESS in 18m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0ffc6589962f3628ef2c436a15a0b94dd6a5666","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-02-11 00:49:58.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/774904/1/iscsi : FAILURE in 3h 05m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a25432ee34c500d7e2fd0ccf0c2f483ba97d50dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-12 16:43:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"495b0b831d8fc5d616046e71a191ccb0a77d1771","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-12 17:59:44.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/774904/774904/2/check-not-vote/ext-nova-zuul/0a13405 : SUCCESS in 1h 13m 14s","accounts_in_message":[],"_revision_number":2},{"id":"567a8dd0559e463c2a0c491df4a223a76140e2e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 18:10:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c088e3c36234e01937ec5ade605a694 : SUCCESS in 41m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7220ce7bf5946629981ee12a3795f9c : SUCCESS in 43m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e10c10d84a8fb20c56f8786bb34cf8464023638d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-02-12 19:31:26.000000000","message":"Patch Set 2:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/774904/2/iscsi : SUCCESS in 2h 45m 36s","accounts_in_message":[],"_revision_number":2},{"id":"bb677d3f48c5b52eb5ffaafc4173f078fe4f1256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 19:59:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dc07ec7374e4aa2b989960c288db281 : SUCCESS in 18m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16ef9b58711e4ebb90131199b2ffb403 : SUCCESS in 14m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7ba45cfee5c4b13a4a24f54058f33c4 : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c2c94ac993e489e8b52ad6d479d8ae1 : SUCCESS in 13m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fd1f6415ab741259e2c677317c7fe91 : SUCCESS in 13m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b265040653cb454b8d8c1ed522d40e23 : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950874c1a5e0424d9814219549f0d12b : SUCCESS in 10m 14s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ceacc101da07470db38c1c59927c8364 : SUCCESS in 17m 30s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a381d8f10de44728fca5ef0e4c67f6e : SUCCESS in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0d60c1e55ab491189c26d2389a4b9e86271074c2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-12 20:00:03.000000000","message":"Patch Set 2:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/04/774904/2/check/EMC_PowerFlex_NOVA/508d348/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":2},{"id":"1fa91f387837a13209b1f8a0dc10b34bf3741ee8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-16 11:58:11.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ea03925cb73d9556c3bcf8816ae8b64bb84a0c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-16 11:59:14.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"ae429ed2c7c0a771b4568189cfbaf3adf7b32de2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 12:41:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a22be24631a94afb866d104ba39c8ab4 : SUCCESS in 29m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/201bc1b9c6cb40678e4ba70bc8e5f2db : SUCCESS in 31m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2d6f29a675ae4842de8ce4897a07d587d8d46d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 13:50:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2696745deb9b4d7584f86f47242e924e : SUCCESS in 18m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/29e3886c690249f4bb36ca953104a7e5 : SUCCESS in 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74de5a454a644c32b2768d2afc6be077 : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1634749951d440e6beee2bd45507b93c : SUCCESS in 14m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/400fb369ad6f4789a3e0dc84826b6cfe : SUCCESS in 12m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc418d87708947edab2622527a582cd9 : SUCCESS in 15m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cc77d236dc64818a7e65eaca9ac1e24 : SUCCESS in 10m 46s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/91de75a95ac245518166dbb8df07ff7e : SUCCESS in 20m 54s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3a3fb35a6bc4b50b7c940f145197f96 : SUCCESS in 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1584f3e8a56931fa4b1fbcc5b96543e31d7391cb","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-16 13:51:06.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/04/774904/4/check/EMC_PowerFlex_NOVA/c3f7db5/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"69c6aeea579d0c3bf6b0870980f430bcba35729d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-02-16 14:56:29.000000000","message":"Patch Set 4:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/774904/4/iscsi : FAILURE in 2h 53m 12s","accounts_in_message":[],"_revision_number":4},{"id":"86bc1a25e923e720c68dd3eebbcbb5578938a4a1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-17 14:03:04.000000000","message":"Topic set to poison-os-uname","accounts_in_message":[],"_revision_number":4},{"id":"9b20ba03cc63e6936c122c465eeec93550589b63","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2021-02-17 14:52:50.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)\n\n+1 but as a follow up i think we should see if we can use the fixture in the other cases that currenlty dont","accounts_in_message":[],"_revision_number":4},{"id":"c6aa310a14128e14a6d03d2f0e04511085f2d0d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-19 17:16:34.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"29fbd369a65d201fcb96e604d39f4bf61dee80b1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-02-19 17:22:59.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/774904/5/iscsi : FAILURE in 1m 59s","accounts_in_message":[],"_revision_number":5},{"id":"3e0277325bb97c87ea775ef1bf03bdb9ccf18bb8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:20:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4079342241247bc9a333b3c93bd2eb8 : SUCCESS in 38m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b178c00bae6d4b7181f6ea6a66ba3f1b : SUCCESS in 40m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ae8903ffc6276cc0a7d06ead611d4373ebe9ff70","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 19:03:41.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/774904/774904/5/check-not-vote/ext-nova-zuul/a977e6b : SUCCESS in 1h 04m 17s","accounts_in_message":[],"_revision_number":5},{"id":"689142d1883533efe8c9b5e70265d5a9950a0500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:10:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1448c5823f36467bb8abbfa48227fffb : SUCCESS in 16m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ec92d762d678433fb2357cc4c2dd8d3c : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/085d3a638cde41748228d5e3ca0d9df4 : SUCCESS in 10m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97812097990d416190d2e760060d3331 : SUCCESS in 14m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0edc94330e34f92a83900a6fbe16b59 : SUCCESS in 12m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94c90feccb2b485583093d486594b9b6 : SUCCESS in 15m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e392eb85db48feb1f5080ac8f183a3 : SUCCESS in 9m 46s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/818524be64644ef68ca54982134c70f3 : SUCCESS in 19m 56s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c173a6ec2964710a5825e0765572222 : SUCCESS in 19m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8a92f2ebdcc69441f82fa2397c105ce14b667d56","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-19 19:11:24.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/04/774904/5/check/EMC_PowerFlex_NOVA/082bbc9/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"360e7553901805c7d71ff67f4e48db8ebfe11223","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-02-20 15:37:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"67ef14666f7e634655098a607fdf4c05d7f04ebe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 16:04:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9aef8d5cd174fbdafe4fe618e7d0c92 : SUCCESS in 15m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0791608b8f11491c894e6d77369b8994 : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8942dd4dea204e4fb8c97060ac2bf871 : SUCCESS in 10m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0df0ed50c98c4c50bb3705c308f94d94 : SUCCESS in 13m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a234325ecd94c8cb13c9aabdaf6e183 : SUCCESS in 13m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34b135bb803f45feb4b63b6e19f6c9b1 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859148202df44c6284316d45aacd6eef : SUCCESS in 10m 40s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e55ff1f5eecd40988f9705e83ff4e1a7 : SUCCESS in 18m 54s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ab70f68429843538c887f02ecb9d1b6 : SUCCESS in 18m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"89947bb33e86d17bea04db0ce53f371fd63cbf9f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 16:04:44.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/04/774904/6/check/EMC_PowerFlex_NOVA/ccd8571/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"310a5b9adcdc36d3b72a6cc7d35d0c563e37abac","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-02-20 16:24:44.000000000","message":"Patch Set 6:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/774904/6/iscsi : FAILURE in 44m 02s","accounts_in_message":[],"_revision_number":6},{"id":"7933c678cb145169a935c01c223e0004b7effe09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 16:38:26.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c79ef29aa2247ccb99aa2d32837ec89 : FAILURE in 49m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb58705463a9463a9fdd860c5d9847c9 : SUCCESS in 31m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ec986a7a910dc95c3fc31b63938064b6e4dae1e7","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-02-20 16:38:52.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/04/774904/6/check/EMC_PowerFlex_NOVA/8718c05/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"068efda8b9d2976158b9c553ca7ff33381db0bcd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-20 17:03:12.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/774904/774904/6/check-not-vote/ext-nova-zuul/dc4e0ef : SUCCESS in 1h 21m 56s","accounts_in_message":[],"_revision_number":6},{"id":"86c209d21b006f415daa57dd72daeb066ba521ef","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-02-22 16:59:51.000000000","message":"Patch Set 6: Code-Review+2\n\nYeah, sounds to me a better place to provide this stub.","accounts_in_message":[],"_revision_number":6},{"id":"d4d34d137c579482850c412531763af97905f57e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-01 16:15:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLGTM, makes sense.","accounts_in_message":[],"_revision_number":6},{"id":"4b1007bbb4ed1427242f413e1122dc9c9bc3f338","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 16:21:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6734dffcb7bd29ff7f972d44b3ebdf4c7563388f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:43:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e23144f622a4aab880ed9ab5eee0d78 : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d530a3a06cc4481b0bfce4494eac570 : SUCCESS in 11m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e85676cf758420ea19bd389d9213141 : SUCCESS in 15m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22d01543b8e34e04bbe051ff3128849c : SUCCESS in 13m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87509ddcb3d34b2d8cdbe2fdeb296548 : SUCCESS in 12m 03s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/80773eae840c46f9a2588d36d4a7b3e8 : SUCCESS in 21m 32s","accounts_in_message":[],"_revision_number":6},{"id":"598526c62556a1124248608e392891bb018de008","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:43:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"5c10bb63b5a1e1eebf400e1b59ca5e8c8679e912","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:47:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a445ed9559544228df88c68d71c363b : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1053ae0b61a6c253b5684096b96ccad7a39343ac","revisions":{"c9d33a6362f7dbc15b49587da21e481d349baa7a":{"kind":"REWORK","_number":1,"created":"2021-02-10 17:15:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/774904/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/774904/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/774904/1"}}},"commit":{"parents":[{"commit":"21af4dd4bbd79e8049d3c61b7667f1fa433b5b88","subject":"libvirt: Stub out \u0027nova.virt.libvirt.utils.get_arch\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/21af4dd4bbd79e8049d3c61b7667f1fa433b5b88"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 16:53:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 17:14:17.000000000","tz":0},"subject":"tests: Move \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","message":"tests: Move \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n\nBy moving this to the \u0027LibvirtFixture\u0027, we ensure virtually every test\nthat validates a libvirt thing will have this stubbed automatically.\n\nChange-Id: I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c9d33a6362f7dbc15b49587da21e481d349baa7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c9d33a6362f7dbc15b49587da21e481d349baa7a"}]},"branch":"refs/heads/master"},"a4b298446be01424bbab9ad44ed151b3c7e2b730":{"kind":"REWORK","_number":2,"created":"2021-02-12 16:43:55.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/774904/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/774904/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/774904/2"}}},"commit":{"parents":[{"commit":"5b52db409b00e6257e83598efa0d8528ebe380e0","subject":"Merge \"libvirt: Always enable USB controller on PPC64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b52db409b00e6257e83598efa0d8528ebe380e0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 16:53:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-12 16:39:01.000000000","tz":0},"subject":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","message":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n\nBy add this to the \u0027LibvirtFixture\u0027, we ensure virtually every test\nthat validates a libvirt thing will have this stubbed automatically.\n\nChange-Id: I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4b298446be01424bbab9ad44ed151b3c7e2b730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a4b298446be01424bbab9ad44ed151b3c7e2b730"}]},"branch":"refs/heads/master"},"7f12ef78958ff3d4c2d07e29e21962902f55766b":{"kind":"NO_CHANGE","_number":3,"created":"2021-02-16 11:58:11.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/774904/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/774904/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/774904/3"}}},"commit":{"parents":[{"commit":"5b52db409b00e6257e83598efa0d8528ebe380e0","subject":"Merge \"libvirt: Always enable USB controller on PPC64\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/5b52db409b00e6257e83598efa0d8528ebe380e0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 16:53:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-16 11:57:55.000000000","tz":0},"subject":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","message":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n\nBy add this to the \u0027LibvirtFixture\u0027, we ensure virtually every test\nthat validates a libvirt thing will have this stubbed automatically.\n\nChange-Id: I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f12ef78958ff3d4c2d07e29e21962902f55766b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/7f12ef78958ff3d4c2d07e29e21962902f55766b"}]},"branch":"refs/heads/master"},"bd71cfc1152f551ffa50588cdfc1ca0c32d73a49":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-16 11:59:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/774904/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/774904/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/774904/4"}}},"commit":{"parents":[{"commit":"a7dd1f8881484ba0bf4270dd48109c2be142c333","subject":"Merge \"vmware: Use cookiejar from oslo.vmware client directly\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a7dd1f8881484ba0bf4270dd48109c2be142c333"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 16:53:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-16 11:59:04.000000000","tz":0},"subject":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","message":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n\nBy add this to the \u0027LibvirtFixture\u0027, we ensure virtually every test\nthat validates a libvirt thing will have this stubbed automatically.\n\nChange-Id: I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd71cfc1152f551ffa50588cdfc1ca0c32d73a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/bd71cfc1152f551ffa50588cdfc1ca0c32d73a49"}]},"branch":"refs/heads/master"},"2a59bfe34c983d97ef2b3c0e71abeec0fbb4e9af":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-19 17:16:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/774904/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/774904/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/774904/5"}}},"commit":{"parents":[{"commit":"f7975d640ce1e9fa06d045d35177f07451716f0c","subject":"Add regression test for bug 1914777","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f7975d640ce1e9fa06d045d35177f07451716f0c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 16:53:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-19 14:22:59.000000000","tz":0},"subject":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","message":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n\nBy add this to the \u0027LibvirtFixture\u0027, we ensure virtually every test\nthat validates a libvirt thing will have this stubbed automatically.\n\nChange-Id: I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a59bfe34c983d97ef2b3c0e71abeec0fbb4e9af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2a59bfe34c983d97ef2b3c0e71abeec0fbb4e9af"}]},"branch":"refs/heads/master"},"1053ae0b61a6c253b5684096b96ccad7a39343ac":{"kind":"REWORK","_number":6,"created":"2021-02-20 15:37:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/04/774904/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/04/774904/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/04/774904/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/04/774904/6"}}},"commit":{"parents":[{"commit":"ede0147abd54fdd043e7aabd33c01f88e4a01e51","subject":"Merge \"Change API unexpected exception message\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ede0147abd54fdd043e7aabd33c01f88e4a01e51"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-10 16:53:48.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2021-02-20 15:32:14.000000000","tz":0},"subject":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture","message":"tests: Add \u0027nova.virt.libvirt.utils.get_arch\u0027 stub to fixture\n\nBy add this to the \u0027LibvirtFixture\u0027, we ensure virtually every test\nthat validates a libvirt thing will have this stubbed automatically.\n\nChange-Id: I03febf6ad7d76c7eec818d3b16a3ef8b26dcd84c\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1053ae0b61a6c253b5684096b96ccad7a39343ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1053ae0b61a6c253b5684096b96ccad7a39343ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
