)]}'
{"id":"openstack%2Fnova~783275","triplet_id":"openstack%2Fnova~master~I15c3119857970c38541f4de270bd561166334548","project":"openstack/nova","branch":"master","topic":"bug/1865223","hashtags":[],"change_id":"I15c3119857970c38541f4de270bd561166334548","subject":"[neutron] Get only ID and name of the SGs from Neutron","status":"MERGED","created":"2021-03-26 11:30:26.000000000","updated":"2021-04-21 06:00:24.000000000","submitted":"2021-04-20 23:57:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"783275-bug/1865223","meta_rev_id":"8d2c226190c8bafe33fc6457079fd54a1c7e2b19","_number":783275,"virtual_id_number":783275,"owner":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-20 23:57:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"all":[{"value":1,"date":"2021-03-26 11:39:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-03-26 18:58:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":1,"date":"2021-03-26 12:07:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":2,"date":"2021-04-13 15:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-04-13 12:56:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":1,"date":"2021-04-13 15:59:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 11:39:21.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-03-26 12:07:19.000000000","updated_by":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2021-03-26 12:27:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 12:46:49.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 14:28:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 14:29:30.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 18:51:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-26 18:58:24.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2021-04-13 12:56:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-04-13 15:59:07.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-04-14 09:31:22.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-04-20 21:40:51.000000000","updated_by":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"reviewer":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"state":"CC"}],"messages":[{"id":"280fe6d902a989ac482b3bab318aecc83251957b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-03-26 11:30:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1245ae4104c45cb121b76e0f37b2ce73d4e9cce4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-03-26 11:39:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bcdba41dd0281f393746ba7569c8efb33bc5d4b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2021-03-26 12:07:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad2bd203dfb7ce730a749e6d3afaa02a4ce7cfd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 12:27:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/268ab405c3b74b6685973d91b2a9d181 : SUCCESS in 23m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/09b411283a2743e7ae18e63979cd8073 : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"494135f74f638f8abaeb299e32bc5c8e93ac3e86","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 12:46:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"ed0c2b823f4d1f85fe37b0fbee2bf39fae398f8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 14:28:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8ec384f0c5ad45d990c930a2d25034c2 : SUCCESS in 1h 18m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53ebdd648dc044daa0f5ce1d9449bc57 : SUCCESS in 17m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/136897d465944d50bbceb4e1ad4a0d6d : SUCCESS in 16m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/780b9d472813475c84ca8a56135ee4c1 : SUCCESS in 9m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8219ef943b934369a826df341f3fd333 : SUCCESS in 12m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/212ababdf4f34bb49d194db36bb5dee0 : SUCCESS in 14m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1edca3a33c7e4099894a5fcc7b2426c1 : SUCCESS in 20m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca437ac1d0f24d6b97ab4a11205fd171 : SUCCESS in 9m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c697264cee14447585e1784788dd5148 : SUCCESS in 1h 33m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/052a1a6bd10a403aa1e1c685633ea28b : SUCCESS in 1h 18m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/17afd249742a42568d834be3456438cc : SUCCESS in 1h 04m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/4d8352de6ed24cd4adc0a2a121f83024 : SUCCESS in 1h 14m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1e3415e6c36e43dab38de004b2324d92 : SUCCESS in 54m 22s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0557580619c44af1bd2223dbdc7a8fd4 : SUCCESS in 1h 40m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/7a160f17b90c43729a82756ece0b191b : SUCCESS in 1h 30m 12s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0954455c6f304df7bd2b3b97a5a3a1fb : SUCCESS in 20m 57s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/06840b5662d844c08fff17464a8a83ee : SUCCESS in 19m 32s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/e85aef7b923c45d59ca4fcceaf4a70f9 : SUCCESS in 1h 02m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a5efade7a77470c900dca22ed49db46 : SUCCESS in 59m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d9e7782ed034a719fb659b38e7e6b30 : SUCCESS in 1h 00m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b84b39247e3d45a9bdfe8a0b2f2e8af7 : SUCCESS in 36m 17s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/56b8a2025a894dd08d5a470b98db31e5 : SUCCESS in 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f895fa5055818c63eb1d62b1470ebfd50505522","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-03-26 14:29:30.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/eb2600f/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"c361724bfb9f586043780cfc232029a61b543c98","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 15:44:17.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"66e430c819f1aff4f7682cc3f440f81bf18bd5ea","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-26 18:51:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/66b403adb30740b082b3903887b5be8a : POST_FAILURE in 1h 07m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fc7b448f585ce27191822f374123fd3d6fe9dfb4","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-03-26 18:58:24.000000000","message":"Patch Set 1: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":1},{"id":"1b553c62a6141c5130feec520c836b674705d057","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-03-26 20:51:01.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7b1432bcac84e2e9424769d987eb422a536f67cc","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-04-13 12:56:16.000000000","message":"Patch Set 1: Code-Review+2\n\nI don\u0027t see any unit test coverage for _process_security_groups() :/ So I proposed a separate patch to add it top of this.","accounts_in_message":[],"_revision_number":1},{"id":"0ab391b75dd9c1c3d76455c044e61d0b6b3c9093","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-04-13 15:59:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"19c15254fd7356bd5fcccb22fc49e424f00d6884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:59:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"006558a18fe4483a590bded4751d6c543efface3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 23:26:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/1651aaab406b4ca1a44df6fab21b9da5 : SUCCESS in 1h 11m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f174bb190cc442d98ad3faabe07d255c : SUCCESS in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73d167cf64634b2c8e23282ed44d71ae : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b7af7e219a543c09023d7933b5af246 : SUCCESS in 11m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d4c806dbf3d44f8b0d921d23d588815 : SUCCESS in 12m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5210b1feb9a64360b7f7aa076d543b33 : SUCCESS in 10m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/952867a947b1400aa45f35f9cd00553a : SUCCESS in 1h 09m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4bef890d213340918cf3ed25e9ee25d5 : SUCCESS in 46m 07s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ec570a8b8189419d99b209a16e3f9a24 : SUCCESS in 46m 58s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/00390fba063146f9bcb8bb9740798bd3 : SUCCESS in 17m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/49b2cf19dfa342e5b818bf32d239813e : SUCCESS in 1h 13m 40s\n- nova-next https://zuul.opendev.org/t/openstack/build/8e89f307beac4153b2383ae693c132ce : FAILURE in 1h 21m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63f53cb192de4bfb825512e6fea12e97 : SUCCESS in 1h 09m 00s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/4f904073bca44633bf36ea3507cc528d : SUCCESS in 51m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/96629685000f46dc8fbfbf619f0c8cf1 : SUCCESS in 1h 02m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d555284fb334403949dc53478f79309 : SUCCESS in 52m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65bdd8f3c3a64908ac261c8159b6f654 : SUCCESS in 47m 50s","accounts_in_message":[],"_revision_number":1},{"id":"941d8ba1baa41ea1f4c644cbd72303e7ddaede3f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-14 06:46:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cd8a743596c77dc7ca9cbc59c73d2f488619929e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:08:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed616624434346ddab2cb9efc7ef3d60 : SUCCESS in 27m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/922c6ec4c6784de88251422d7533a5bc : SUCCESS in 26m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"834a9bc2e9d9f936dc3217b8026503fda8b8fbd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:50:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b555cf48c30940159728c3b30f743554 : FAILURE in 49m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c64e564f448e40daa43d96991a835ce1 : SUCCESS in 16m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6eb666cea1964552997febb1bbbd69ac : SUCCESS in 14m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525428a5504c45a7947021074c9bbe5a : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fff4a2a08f94cce916cc6f228f035a7 : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf82e5fbb68b4b5e9894dce02ec0e1ae : SUCCESS in 15m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdd7ddf8ef864d6da3028d60be2074e4 : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b91a79fcb5194c9186562e5a3dca0657 : SUCCESS in 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/406f2fe5edb74217b82255e32b20fb1d : SUCCESS in 1h 19m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d459db5ad9a4e41b2eac607ec59f27b : SUCCESS in 1h 10m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fdf01a8d1c784b919c7e68715732f397 : SUCCESS in 1h 05m 58s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8552c7e59fcb4ea1bf8a80693af50ba5 : SUCCESS in 1h 06m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aa83981b257e4af8a001b22435565cdd : SUCCESS in 46m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2b010ed6caa34d95a8fe7d86caf44c8e : POST_FAILURE in 46m 47s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3712c89ff6024673a049bed7b031ca41 : SUCCESS in 1h 07m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/fd238af2e2d4487a98bbf58471af2e72 : SUCCESS in 1h 16m 39s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/827e215dbda24638b3bbc0d454668414 : SUCCESS in 20m 02s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e8dfa335b614c1c9140694ed4f8e807 : SUCCESS in 17m 57s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/2751d0eb7ee64df7b8e4717bcc856a95 : SUCCESS in 1h 01m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59ff4c42118f440682ca4924ef249d88 : SUCCESS in 53m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26682434b31642209f534f33ac7e3712 : SUCCESS in 50m 45s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fbcc78115d234f8eb0529c5762309e21 : SUCCESS in 26m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/de62b16cd0f04536b0b04b0f488fa1e3 : SUCCESS in 35m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59056cdae76095a30543434228995d52f896b039","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 09:05:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"02f359ed15a63533ffb7b88f79dfe03c0f5fee96","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-14 09:29:34.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/da23259/ : SUCCESS in 1h 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7f39747c5480c0d49b8a1cf7fdbc35438e11d5fd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-14 09:31:22.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 43m 43s","accounts_in_message":[],"_revision_number":1},{"id":"df29693937655402feddba3fc3baf472a70ec610","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-14 14:10:40.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"339f0cbc6691f1368a401993303fc4633096df8b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-15 06:52:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"acc8c132ace4b4a1b3f05d1d31fcfec0c9874988","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 07:52:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8541be53d99f4f35b7638330d944300c : FAILURE in 56m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae94f67bdc6d4af9a68e236bf2091cb8 : SUCCESS in 21m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e119752bab040c81c7b4a99a43c282e933e1f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 09:00:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5080187d74f34e259d3b918dd24dadbf : SUCCESS in 1h 38m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc72120025514d79bdfe4116d847ae85 : SUCCESS in 17m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd29305794fc4ef6bd64176c0273ba22 : SUCCESS in 15m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6dfa33eef8649e380db37633ada5a67 : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11e83655d0084fbf8cdf67d12164ec9f : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e69774abdc46388007c2bff6ef6093 : SUCCESS in 13m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/449769a70efa4aa88dac658ca91d4ca8 : SUCCESS in 21m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8297003dc045c8aa52734839fd9ba0 : SUCCESS in 9m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f4da8ebf53cd4a3db63ba8f51bcba8d7 : SUCCESS in 1h 21m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3249795a177460f8f36cdcfec417eb8 : TIMED_OUT in 2h 04m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2006f48b4da3422e8d2bab77c91ee8bf : SUCCESS in 57m 31s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/af9441a9193f4f5ab4c5c9fb2a3864a0 : SUCCESS in 1h 11m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/90c19870947249bf9daef9bdfb8b08b8 : FAILURE in 46m 39s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2bae3ba9455b429e996aaf333bf0115f : SUCCESS in 48m 44s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/aaf9313bf9564f72992815e4e192fc44 : SUCCESS in 1h 28m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/32a7c790fdd244faa3d773ca91f51dcc : SUCCESS in 1h 21m 47s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d808b218bde4f2aa78ba0771e74be8e : SUCCESS in 16m 58s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2da16e9fcbef4653bb0f36da67fa1958 : SUCCESS in 17m 17s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/0151c72686ea42828761f58c6495f25f : SUCCESS in 1h 02m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcb291ef894f4e3e88446548e7bab2b1 : SUCCESS in 1h 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2503a3629ba436f94cb464b424c175f : SUCCESS in 40m 08s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5024acc94f6d47188f14892c7b541f66 : SUCCESS in 22m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d159898bbbdc4dd3ae9cb3a7dd5dbe43 : SUCCESS in 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f274117e7a489f80c49b17adc901df698835257","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-15 09:14:38.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/643c986/ : SUCCESS in 2h 07m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6ebf133a4cd6f743343c3ef6c34b63622d876624","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-15 09:28:25.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 34m 51s","accounts_in_message":[],"_revision_number":1},{"id":"319cb30cd34e2bd1e7ee48e60e7adde6f60a492a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-15 16:14:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : FAILURE\n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"3c14163dae01f2f8d18a2719082dccfb024d7fe9","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-04-16 20:59:07.000000000","message":"Patch Set 1:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":1},{"id":"ce467c710db43ecb46912bda278915b2cd3df475","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 21:24:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3160667e20c7496ba7e8292fbec936a7 : SUCCESS in 18m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7954e11cf9b34e83bd5d5a536e810587 : SUCCESS in 20m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd8224fc6c93fcf83e5996bea21d9756ba133667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 22:42:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c6001d25ab6747fb9f0f9bdeeced17aa : SUCCESS in 1h 36m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ef1219bc8164309aa14ce67ac7ce2d0 : SUCCESS in 17m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d844f19df604d4baf4e00302729bea7 : SUCCESS in 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af260195ca9486c96e9932fa8f351f5 : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/914ecbc39aca40a9bb6d68295637d17a : SUCCESS in 11m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5f046005f3d40d2a06167f7cfb62f28 : SUCCESS in 14m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12a1cd42632042b2b18d10bc2fc2fe11 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f72131345cd0457e947053beff836367 : SUCCESS in 12m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8ca25d13d357404bb0415a409b120759 : SUCCESS in 57m 54s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ddf907ed3385415e93b49b1db5aedad7 : SUCCESS in 1h 07m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/c894082b7980402f942a6df6436438a4 : SUCCESS in 44m 03s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/739501930c1540b58ad12e1f95dc1a89 : SUCCESS in 1h 02m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a26c2d2ef51a4c6a88a23a80ab0a34e2 : SUCCESS in 44m 53s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/29ba707a7c4e41d89411664c04e674c3 : SUCCESS in 47m 44s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5604d1a12d484627a0cf984a2b5b0cf3 : SUCCESS in 1h 05m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/24a42094cf14489b92444529741cb7ca : FAILURE in 1h 10m 11s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39d64e0e36d346d0b484cf1ff82c617a : SUCCESS in 17m 38s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6aca63c519c74798a0cd63e4178e2e1e : SUCCESS in 17m 20s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/4e0d512b89be45f18429fa774ce5d3fa : SUCCESS in 46m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40509b067fa942faa1fc3eeb1fb4712b : SUCCESS in 41m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32b16fbd176f4ba5ac064e709d0b1719 : SUCCESS in 42m 49s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a0b58bed115a4d7fbcc3c18cac5db018 : SUCCESS in 22m 30s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6ed0a9111f954cf184edbac21c834fb8 : SUCCESS in 24m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b72ea8fb88f607589bd88ed622520bc5c78bccdb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-16 22:58:54.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/27f44516e30f4a2ca127d369c71dac2d : POST_FAILURE in 1h 50m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a21bf6462823dca607ed066d575175d4fc371af5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-16 22:59:16.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"28d02317bf501cdc8b7d8f9bb29aa67566fe8246","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-16 23:26:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 26m 54s","accounts_in_message":[],"_revision_number":1},{"id":"a8fcb471c89b97232013acc5e2eaf65e9c1a7da6","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-04-17 05:33:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b55fda24c2126709f211b5e61dc57cba66899ec0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-17 05:34:24.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : ABORTED\n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"e801bee26b691f885926dac52344739d44f2424e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 06:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0da1cc2386704d9b94e99bb41ec8147e : SUCCESS in 22m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce58ab084446459b818d3c02f7473543 : SUCCESS in 42m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"346434940fc9c83b1ea393b09d63a5c29993ae59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 07:31:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9ccedc47262c46ae9233021fc256dc92 : SUCCESS in 1h 11m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dc287f56b0a4c978c4c3fea675b1179 : SUCCESS in 19m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd765d6e269f4fcc816ab6fe053cdba6 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fb6291c00ee46df8dc7d87e15afca9c : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/775491e65e1e4e9986df5a2b67a37af6 : SUCCESS in 26m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87a14ef3484c4a2382b0eb0f7fe12040 : SUCCESS in 12m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b1a6b9aeb0445b894d79c75558d2e95 : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32dbe06e95cc44e799c580ef7a8b37c5 : SUCCESS in 9m 40s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e92d44bd56904907b9556639044b6876 : SUCCESS in 1h 02m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a8803b250f7240638464d9605bf58a3f : SUCCESS in 1h 15m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6beb1ea8ee5e4f5fa3dd2d12c076d846 : SUCCESS in 48m 56s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7d2ddba7f9bd433fa1b8f02765d04ab3 : SUCCESS in 1h 11m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2912889f58974a49a791253216c930c8 : SUCCESS in 43m 03s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/35cfb3f91cec471f91fd752793b19a00 : FAILURE in 39m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5f5a84ca85144b7f9f9efaf8dcecfe9a : SUCCESS in 1h 19m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/446308ba21764d9cbfe7b3759b74a552 : SUCCESS in 1h 51m 14s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a4eeeb743a424ebeb4785e84cdf85cf9 : SUCCESS in 22m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/264aa946d32846049168e668a2106d32 : SUCCESS in 18m 16s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/6b718a868b08470a913921bd13c888d8 : SUCCESS in 57m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/954f6c691d6f4202b34d384892ae12d0 : SUCCESS in 51m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36b81e8015c642be90c7f0370705be37 : SUCCESS in 43m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b0d369d424e64f95b2924551b1c4836a : SUCCESS in 28m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0f415ede065e450bbb375631bddbdeb3 : SUCCESS in 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cb98f50ff8fe69f2d5aee0c3f80036a248712743","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 07:31:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdccaaf81477c82ab4f6ace75d3de29d0b1c80ea","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-17 07:31:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/7f336ee/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"732e647f315e4502d471e82bba41b853fb496ef0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-17 07:32:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"b680902e171f49b876593862b5c180eb3affa55f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-17 07:56:26.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 21m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c5973556c441895b9d637fdfdc50d4e1675a9dde","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-17 08:43:16.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/25fc90ad297c461ba603767ebed3eabf : TIMED_OUT in 3h 04m 53s","accounts_in_message":[],"_revision_number":1},{"id":"13a631a9a94bd4468b2beba1efb4e17556b887d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 09:47:39.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b903361d65084fae879e553ad9a4618d : SUCCESS in 55m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb5b19f2da3b4d7caefada98643f40e1 : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7632f57b7c54519b78c52604e818a53 : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/979cad8f8b31411694cd6ec17aafc56e : SUCCESS in 17m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e010c334d5884994bfa7ed6267cfae4f : SUCCESS in 19m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8013a51de1b4eb6a1fff13651610d46 : SUCCESS in 13m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/52aea4aec972470bb9b9a3df596cc6cb : SUCCESS in 1h 35m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/51c8b553ab784f5eb75f2d70d8baf7ab : FAILURE in 40m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/bd3b9f6462604994818bcebe6ac0ea7c : SUCCESS in 1h 05m 58s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/feab7c5517004b89bcf69f55f017dcfd : SUCCESS in 21m 16s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/af123c014e744d6da01629eebe25ef77 : SUCCESS in 1h 42m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/53384b5376a44deb98dec4707024afb4 : SUCCESS in 1h 37m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd57c0439a0c461d84c772474be132b9 : SUCCESS in 1h 14m 12s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/a0c08439956048d3ab8f4adab3b86905 : SUCCESS in 1h 02m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/3ca48333b23c46ffb4f469ce51ed97ed : SUCCESS in 1h 08m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7de902ad6584b9886692bf3fc88e610 : SUCCESS in 55m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b6d4b235bb44f2d9154dab89a4cd138 : SUCCESS in 52m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bfa51f82422903149e1eaff200c3ed3b5fde2154","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-17 11:34:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8e16791c2ad8ab9853f1a21691dcb5f16e53491","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-17 11:36:04.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"05129d3f179627e458a9d2b7ad265cd3cbe3630a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 12:12:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8aefaa01d26a41a0bb4643d05032f41c : SUCCESS in 34m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd0d6037030d4002b9a909850918e2a5 : SUCCESS in 23m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f9cb9ed4e0b9921962828f1c2339a792c7b3524","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-17 12:51:09.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/390a6b5ed0ef4caa9335752b7c96bd3a : POST_FAILURE in 1h 11m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f51eaaade939ac5513310500fd10aeda13126429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 13:18:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6d2ef412a627493e9a6feb2da118ae00 : SUCCESS in 1h 18m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef61420a19414f658060f8b20de1e78f : SUCCESS in 18m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0bb58423c084f37875293ce0fc62724 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e0b922ac93043789321d6dbbbd4324a : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df758a23b04c4a26801b74fd3c23f7ed : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a26fbac091d41ae815b47ea62004568 : SUCCESS in 14m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4282da31d6144bcb0f25ec6b09ad186 : SUCCESS in 13m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790c8dc1bb8447b790f6f55b0228f9b6 : SUCCESS in 9m 02s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8795f31229bf448b8630b73a1b4ab739 : SUCCESS in 1h 20m 30s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e67d9d0d9904eebb52c4d000470766c : FAILURE in 1h 14m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/de9e51df067c43f9af910603cdc509d1 : SUCCESS in 54m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2e1b3d0c60d43ca88ca75a83a814bbf : SUCCESS in 1h 24m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/66c0f90209214b80b67ffae9c28be925 : SUCCESS in 48m 11s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6b252b2260fb4d8dbd1d7959f92819d1 : SUCCESS in 53m 07s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/769b38deab2f46ff8fad2cffdc4811b8 : SUCCESS in 1h 33m 01s\n- nova-next https://zuul.opendev.org/t/openstack/build/14bcc2e5457d44429641d03f69021bda : SUCCESS in 1h 37m 25s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb04b8117b3045b687a6c55f517ef1b7 : SUCCESS in 19m 54s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58a1330c20ed4bfc824700aac147f44a : SUCCESS in 18m 27s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/d44d3bb9cd2c4c45938ecb43ee4e49f5 : SUCCESS in 1h 05m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a089db56f1e94a398efef4d551130f33 : SUCCESS in 1h 00m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df3062caf7da4f81b9c9c2c22963a3be : SUCCESS in 49m 56s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7eae475bcaae43349cf0f6158af21175 : SUCCESS in 28m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d7bd1e164a8041b998c88fe8c730cde0 : SUCCESS in 1h 06m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e593aeb6a1612940a09f91df9991c5dab45a3bb1","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-17 14:10:31.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 34m 42s","accounts_in_message":[],"_revision_number":1},{"id":"516772cf63b2aab7f12a110756508968156e40d1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-17 15:30:13.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d9267676b4c9d1d80bd90c7c53f9631b94578529","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-17 19:13:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bbaeafae14742938bb7592d0f468a83292dd587a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 19:44:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11854519c81b4e098c2f568aa5451937 : SUCCESS in 26m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2c692f6847e46b2a21aeceb169ba54e : SUCCESS in 28m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"369c4ebe284fcc592ce5bda650c2345110730052","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-17 20:31:59.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 finger://zuul-v3-executor.novalocal/682f485398344feab5dd1b75c38b47da : POST_FAILURE in 1h 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"53d92d42d4193172e2b6864b5854b618072c3d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 21:23:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6ee0a891e2a64712aeb7e57150d39236 : FAILURE in 1h 20m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35fa019e22da4fe9b75af08b0e77d37e : SUCCESS in 16m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cfe5213f67149f682aa6366fd0ebe75 : SUCCESS in 18m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8aff2284be4672b7100414facc344a : SUCCESS in 12m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a4f502c96bb142f0838c8775643ba3b0 : SUCCESS in 18m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eda836a22f82444eb77e974487ba9d08 : SUCCESS in 22m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acc6c6e1fca24d4e81fb0e28e4638d0f : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0ad824d6bc402dad9f845ee8c84817 : SUCCESS in 15m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/5a513523d7ec4857a9748575d9ae2217 : SUCCESS in 1h 13m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ded6cb1f013463bb4a5a94989418dd4 : SUCCESS in 1h 33m 45s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/67291fc221ff4a26ba563d416c92970f : SUCCESS in 53m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f4dd5dff04d841fdba0fe8685802cc07 : SUCCESS in 1h 58m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4de7786d67464419a39cc23d36bd9faf : FAILURE in 45m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/032ba11e28be47118e972ed65f4181d8 : FAILURE in 46m 16s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/feac74f9716c46b4b6f1e4ddf9790848 : SUCCESS in 2h 06m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/d9af2a9e0b4a4e84a8a359bee8f96261 : SUCCESS in 1h 21m 30s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5514964c5fc2487ca39fe143b78aae6f : SUCCESS in 16m 37s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/194042c840484e228f33a67a1607c15c : SUCCESS in 16m 52s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/27a2cd4f8a5a4253b533413ef1be9d47 : SUCCESS in 1h 36m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/820a99ef33c14a238ddfd2da2d647fb2 : SUCCESS in 59m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d610c8b087d748a894eea837feb026b6 : SUCCESS in 44m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0d33ec314e1d4aaf9bbc1362eaddc893 : SUCCESS in 18m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a26071fdf834e9f843c3a96d992b813 : SUCCESS in 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4ff82c8ee8482ff5cc7206b7f4afc2834b8ee58","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-17 21:42:57.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 28m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f655a76634a000f0adff4730642a3fd98970a163","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-18 00:02:11.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"b45a93ec2fddedc7f7fb2537b79e6facb440f9b1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-18 06:03:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1a7750c4133be7d315023c12891fba828a04a3a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 06:54:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/06a522290a854501b0bd4c0c33099a45 : SUCCESS in 41m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82dd775770af4aa2ac9d9f52f81188fb : SUCCESS in 45m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5e387edd62b98bd0d819162614397aa36145115","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-18 07:47:10.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/0b6465b/ : SUCCESS in 1h 37m 56s","accounts_in_message":[],"_revision_number":1},{"id":"dbde10afe506d27578f314606d84ca8fefea01f1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-18 07:48:27.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"3aa12de3eac7fe4d6e1a82260bc6d702e304ab25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 08:10:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/fade1f143806457f8f6bf918862a9e3a : SUCCESS in 1h 09m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eba0be2bf7c24e719c712764dd5a1ed0 : SUCCESS in 19m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8b18178b2b364438b1d739d3ce294265 : SUCCESS in 14m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6214c136b51546da9a2326437f2e98a0 : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2538a2d373294c1d81da260040fe817f : SUCCESS in 16m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e28827fe7e247c3901ec1fb40635179 : SUCCESS in 14m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f9c182930bf4d228faa24b59ea2226f : SUCCESS in 14m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349dfe94d4754e88892f4bd222ce05e2 : SUCCESS in 12m 24s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ece0de02e94f4813a8d34c67fc49113b : SUCCESS in 1h 25m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8714f435d97b4cb4ab8f2911d0666c89 : SUCCESS in 1h 12m 38s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9d7f1de73ca447aa91d1385927619d17 : SUCCESS in 1h 06m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/351d047e4cbc4d74a363652c893f1e52 : SUCCESS in 1h 01m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a5f89535a5074b479dd74dd25f1fd461 : FAILURE in 40m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/09715d6b969f45b6a3f41ac68eaee8ad : SUCCESS in 52m 11s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/58cc9e6fd1f847e680e3e74d55aea7b4 : SUCCESS in 1h 29m 31s\n- nova-next https://zuul.opendev.org/t/openstack/build/a2e5ac96d2d84b618570449e93abda63 : SUCCESS in 1h 20m 18s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4632b65fad54482e8372e7a3bf37c840 : SUCCESS in 22m 59s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd08553843f3417cbe2980c929974f30 : SUCCESS in 21m 34s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/ce5504b3258446ca92010511c61f9ae0 : SUCCESS in 1h 11m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c3dc272020c47128ae0e556a0d5db47 : SUCCESS in 57m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28d185a9f0ae44ae86a758b37964605e : SUCCESS in 1h 00m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/7e075611ef8d4f06ac529b1e678607be : SUCCESS in 32m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bbed9f13fd564079b19201ac901f3771 : SUCCESS in 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ce310f9866f4686f2bcb6268b32704a565a63f8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-18 08:25:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"982be5c8c4b86f2ce2a0a01a70c7f9b8545f2b4b","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-18 13:26:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"727a4d36b21744a6db2fb4517c9d14163fccff8a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-18 13:27:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"df7cc688885d3f7d7c78d7072f72b4f67d13ff68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 13:54:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81be1db35e294ba7a0c6e1dfe2897d78 : SUCCESS in 18m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7305daff522747b0b6f3b3f5555c63f8 : SUCCESS in 21m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4dfe4eadfa66bfda8e8525eb86f3036d89b0dc23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 14:48:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c5c506420bbe45bfbd50fada6dd6c58c : FAILURE in 55m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8668b9a9385c4ce58b9da429941d942c : SUCCESS in 15m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7205f85cdced42fc8b680452201fafd3 : SUCCESS in 14m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66d366a509a240efb92ebca4e33100c6 : SUCCESS in 8m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f7c0930bae74e94933839ae776d9d27 : SUCCESS in 13m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/181690d846c742c5999734bb416f70f5 : SUCCESS in 14m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0357718a03ae4a8ca40caad146683d79 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc5c88d9ee444bfac7148ec04987c7e : SUCCESS in 8m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2978b7f3edfa4642a7c72af7735308b4 : SUCCESS in 1h 11m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f2c7fdcc1a824db5b62aa5cd16ba320b : SUCCESS in 1h 05m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/380f54d8c3df446d9e4aab55c19ecbe1 : SUCCESS in 57m 00s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b756a41c263d428ab5ceb70546f53dc2 : SUCCESS in 1h 08m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c7a549593343439aa47a1771640fa875 : SUCCESS in 44m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/57c4176ebea4421b9605a8be2b046528 : SUCCESS in 45m 03s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/bdfce5b347ad4a599437420941e2cd32 : SUCCESS in 1h 16m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/ab55c2cf9fdd464e9c8dd369862fe62f : SUCCESS in 1h 21m 02s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c1d0cd0f39ff4a6293ace61c3d7d2b80 : SUCCESS in 20m 49s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c6d0c2a76c343eb9c9dc7b94dadfafe : SUCCESS in 20m 45s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/3b98d4508f6043549bff3586ac56bb6c : SUCCESS in 57m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cd787ab110f41f39b5345a6df7c5985 : SUCCESS in 46m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c82712a4d56c4a2984b935cf32d22547 : SUCCESS in 45m 44s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/475093119c484019a8265aa2742d21a7 : SUCCESS in 25m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/18a99917cb0340348032d0f47ef9e1e9 : SUCCESS in 34m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f1da405daf59863fe8a2854a7cc7c3552558ce4","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-18 15:40:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/02d739a/ : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"be2f644fd9ad3c5b63d08d7b1b89498d11566887","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-18 16:05:36.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d1bb5308f74cc0965f5f6e31916f0cbab03354d4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-18 18:18:50.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"7fb2db79fb3c7e0cfebdf99d543bd438b87dd1f0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-18 19:05:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aaf96f8377c3301aabdcfb54de10da0c57fc2c8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 19:34:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7272ccfe9009484baefb9ab8cace8008 : SUCCESS in 20m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/597bc336aa984564a5c3752738df0737 : SUCCESS in 23m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e81d8af7887ceddf4c2d45a6dab4d037207c80fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 20:48:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/c2f931a4a9e44972883eb8217f885ed2 : FAILURE in 1h 13m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/330cb2a2720340599558798f9b14ca39 : SUCCESS in 16m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9271eaeef375498c9b3bda048440eff7 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/611fad9c50ed485a81884be2c5888e66 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29fb84d9882f4bdfa367d74b28ca7cc7 : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c11beba12c024d0c906f3e63199c7bb2 : SUCCESS in 13m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/334b7705083d4b7ea2fb18eeca6640d3 : SUCCESS in 13m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633f78de75f1454ea4301058bf536fd6 : SUCCESS in 10m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4ecfae22342c40538fbb1031f2b57588 : SUCCESS in 1h 24m 19s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2c1a12f7551409fbb078c826f9d29ee : SUCCESS in 1h 10m 30s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/e9b96b02f8de454dabd542687a2792d7 : SUCCESS in 1h 02m 42s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/249543cb182a4cf18413ae56b2e73a85 : SUCCESS in 1h 26m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/49411a8b5af74535a23204e1a2e79c4e : SUCCESS in 53m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f536c96926ae4cf3a59ac815f7d4da24 : SUCCESS in 58m 29s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c5254e69729d4c1fa5c13ea2a43e8efc : SUCCESS in 1h 33m 23s\n- nova-next https://zuul.opendev.org/t/openstack/build/7ea0b75095584c2f8dbf7eb7b037bbf6 : SUCCESS in 1h 36m 35s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/768fb9d7eb694c469b1b54c0ebed9724 : SUCCESS in 19m 14s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6c39d5adf1a482d9f803ce7cee44e8e : SUCCESS in 20m 04s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/c0fe90bb0de24d57af639082e67082e3 : SUCCESS in 1h 11m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9bb39a5216a49e6852bb0b2efaf4d68 : SUCCESS in 1h 03m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a6a9cec674f3407fa958b11cbb83a1b0 : SUCCESS in 53m 50s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/09b6259260bb4b10a2378a1db02ae206 : SUCCESS in 30m 14s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c3679f012ce84a99a4880c9b2461d437 : SUCCESS in 42m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4464b593adae803bcaea0ce8f86eefd3274a4602","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-18 21:05:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"397f10e56139e4fe26b5553ae995cff1414a3544","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-18 21:33:18.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 26m 56s","accounts_in_message":[],"_revision_number":1},{"id":"69a09641006575e105f272f6a5b01964f33df9c1","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-18 21:33:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"9ec8a0947b86cce791fa90363df9f9fbb034c6e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 21:34:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c3af56d01d147c4b53f8faf1edbc88f : SUCCESS in 20m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4dfd83a85d9e4193aa1dbbb18e63ae18 : SUCCESS in 23m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42603e8ba19c890e8f67b7e331d964273655b7f0","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-18 22:10:00.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/274788e/ : SUCCESS in 1h 28m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d46c724caac116368d37846aa6c5c5a904aa276d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 22:32:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9451e1e3959a46ee8a49401a99070c4e : SUCCESS in 1h 06m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a20938c57fa4653b55dde0b13ff6d5f : SUCCESS in 17m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/199dab0dcf7c4c41a33fd0f05b7ece4d : SUCCESS in 13m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1bd9a856a54a8d9455b0a16f3c3df9 : SUCCESS in 8m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/470073e15ca04232895c63a0433d91f1 : SUCCESS in 12m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4f863584e7247d9b7e3453641087b87 : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bc24958144b47d89ebbe71970008487 : SUCCESS in 12m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d7f953efc00423baae004e698b810fa : SUCCESS in 9m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ae90336d1d1a4df4a4e67d0831d4325c : SUCCESS in 1h 06m 45s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07474c74acb343389bec7eab13944bcb : SUCCESS in 1h 01m 24s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1d2a551db002429d86017468b5daecb1 : SUCCESS in 51m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/258f7514507440cbb281382d8bef17da : SUCCESS in 54m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fe0cfd031fa8442ba8844241bf74ebaf : SUCCESS in 44m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c03f2b3ccd274997891cfceac50869a1 : FAILURE in 35m 06s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/17e96aaee25441caa0f90c75b0c02797 : SUCCESS in 1h 24m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/f1a2fc3dc3a1473ab1cf1298d19b9e41 : SUCCESS in 58m 58s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79b9fd7f6e5b40d1bad4dc871dbc1379 : SUCCESS in 14m 34s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c86738f7aa845159220e8dc811f168d : SUCCESS in 20m 24s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/5be8b35455f64c9ca2954396a50c7918 : SUCCESS in 44m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/feb1b26dc8944d21a10933abc497caa4 : SUCCESS in 1h 00m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48079cb790e84ca9b7164ed83cdbe594 : SUCCESS in 34m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c539acef73bb4357a5849a5c54261e00 : SUCCESS in 24m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/041c7ea13fef4671951754657e27f747 : SUCCESS in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0e2afce89812778f94ba991a31dc85f610739af6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-18 22:32:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18323c24b7ad390e75265adcadfff864b2040880","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-18 22:32:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/2baf553/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"b9d447af54db194312fbed65250c8a44f2969813","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 00:43:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d2350f97cc0b4f458f1a2241ab8b699c : SUCCESS in 1h 10m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe31f1e037fb46ad9584652a0f70a6c9 : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe0bba1387ae4d3ab8c3acc6ace18f24 : SUCCESS in 9m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f30e18a901b348c58ef5ea2e200ec78c : SUCCESS in 13m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66f41c6282c94ef4aaaf09ff6f1c1660 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af66059bd2bc4f8886bc016e2ee18f5e : SUCCESS in 9m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/aa5e610cb477449db00eea64af9df412 : SUCCESS in 1h 30m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e216a523e5ad4169a1fecc6bd780981f : FAILURE in 56m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e54df23da34e4244a1894ada1afa7552 : POST_FAILURE in 31m 22s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d2ba6b3840634e05bb0a5c6c0624d235 : SUCCESS in 19m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a571883a851e473ab38256c6a514385b : FAILURE in 27m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/56f488e3ad5848cd8f80b2b6f918d11f : SUCCESS in 1h 41m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/df212515b18d4b77ac4347b936e1f80f : FAILURE in 1h 01m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/2f7ad60e010b40669a417fe483bcd85a : SUCCESS in 1h 05m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/7cc205f12d324e2c9290bfbac4fe3084 : SUCCESS in 1h 07m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4fac061eea3d4cdc8e5553844c45db8b : SUCCESS in 1h 02m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fabe1c89a8643449946c3c76f8779c9 : SUCCESS in 1h 07m 04s","accounts_in_message":[],"_revision_number":1},{"id":"570292dd220032da25829fd71122678b4cc92a9f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 01:05:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d0e46835abf30d57152a7e6e723ff1823f116751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 01:41:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef01920cbaef4a4698852c6b74ee90b1 : SUCCESS in 19m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6372a1a7c9543da8000d2a96c964141 : SUCCESS in 30m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86c741f064c6df4099b344c20d6927580dbabaa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 02:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4849ea4a2d6243eaa0947db3fb430d32 : SUCCESS in 1h 16m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a8220e3627b4d59b6a8a879edf7fd53 : SUCCESS in 16m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/846fd92f60e747759a69d82f74a29924 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8e08c72448c4d3d9cf65b2a177a375b : SUCCESS in 10m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/343e6276f6c549b1b11237fa7eeac7f1 : SUCCESS in 15m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/041bc84ae6c643d194a06d37d80838d9 : SUCCESS in 15m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da8a7d8bee3448008c1b12e6704befe1 : SUCCESS in 15m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0229d03a6e3c49cab3b8734a870c63c3 : SUCCESS in 10m 58s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f0ccf1f579544b96a5634e17a7a6f3b1 : SUCCESS in 1h 22m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a6ede82f9954fc3985ef1fea2b6ab07 : FAILURE in 24m 37s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/1a98b002a17947f0a25698e81f1f6e31 : SUCCESS in 1h 17m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/a491da7055254cb3892aff7dc4e8ce51 : SUCCESS in 1h 16m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e0a1212b28b349249a74263709250108 : SUCCESS in 48m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c58bc1b6224d4512b45b7b476190acd8 : SUCCESS in 53m 14s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ca02549bf75248b0aa4a53585510182a : SUCCESS in 1h 12m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/7eb3f3f958e54358857a6344bfff6881 : FAILURE in 1h 21m 18s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bad5719541564263bacd4959b0f2bfd5 : SUCCESS in 23m 40s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27dc5e4848a8487e83b74bd22350d7e6 : SUCCESS in 25m 06s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/ab4cdeef138c448399f3b880c82af270 : SUCCESS in 1h 06m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3bdf97a56ea4417b90aa9ddf0e1cccc : SUCCESS in 49m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e75e91da33834d448fa192d6dd22f51f : SUCCESS in 1h 03m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a0b7e8bf56e7433aa245b5859a759de2 : SUCCESS in 24m 37s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a564863bd5234116853f8396e439f0a1 : SUCCESS in 40m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bfcaa4f2782c7f7b9c5c73fde15f3470dab42b0","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 02:35:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1f89e662dee58dab42b11628df455e2408983ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 03:06:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fca2899a8ef247f2b5ccdd7ecde0f9da : SUCCESS in 23m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b731a229d1d48cf8094ce6a2c77fec1 : SUCCESS in 29m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"151d688aec6b22d19ca22739054604cb375493cd","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-19 03:22:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"366af8528eb3d48e76a591cdda1cd1c436fd41ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 04:25:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5a94ea60acc84cc285be6f8c1eeaa199 : SUCCESS in 1h 20m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c892a22b61c43bfa2cfb511613263b0 : SUCCESS in 18m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2311815410644be2ac3c6b4c08e008cc : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aebcf2e8035540e4907527f5179878ef : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/feeb1feadfa444cfa4ff99ae329b3462 : SUCCESS in 13m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40fbda250338450193a23f113715f061 : SUCCESS in 13m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cdf254c7f4843c186625c9d338b7dd9 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5205dc877a3b4535a7e14fc8a3f3f151 : SUCCESS in 9m 15s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3c577d062ab74cbbb12051dbeddecfbe : SUCCESS in 1h 25m 14s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cdf84781fa764220a3856ad055b20e8c : SUCCESS in 1h 20m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/6b94e65d3e3b4f01a2300e344ce2e7b0 : SUCCESS in 1h 01m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/583f0ece8ec446ce8e40ab2f9539a51f : SUCCESS in 1h 25m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/02c7c8d0b7074b0d9ec983a45660ef0d : FAILURE in 54m 34s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/92c3349bda034f82b608275dca3fdaec : POST_FAILURE in 1h 03m 50s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9a7b61dd61c643a1926a774c1f09ed25 : SUCCESS in 1h 43m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/053f0b97db50493da88e70b457b513a8 : SUCCESS in 1h 29m 56s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7381e88ac5b4b99bbac705b2dc976f6 : SUCCESS in 17m 38s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f10273f44139484196885ad56a60f794 : SUCCESS in 18m 53s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/0408575c6d6e443db7479689e08815c3 : SUCCESS in 1h 00m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/adad2b131b5b469780cab7dbfcb37c83 : SUCCESS in 54m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ce33b73fc8d4ba39d638e9e21cddf86 : SUCCESS in 55m 57s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4cfcf404e4384503b81a4cd329fe4a15 : SUCCESS in 25m 28s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5d19908c688249d6a85618c60ad7bd1b : SUCCESS in 33m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6974864a7a9acbd02e0c36124e7490c3ef144d18","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 04:35:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5da823f6006a0901d6cfcaeaabf6a69f86d54a73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 05:28:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aba611a7f10845b59490960c1eb6a06d : SUCCESS in 18m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0be91ef756454463a6cdb39ded38980d : SUCCESS in 31m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"471a1d500f12b65cef7ad7c04d0d7d5ac148b809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 05:57:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b18981ea2d1d4129af6902afc5dd1e32 : SUCCESS in 1h 02m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d4a53ab4edd457295169e3988ed83ce : SUCCESS in 15m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dae7a7e13d0b41a7bd0202299e94e187 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c819edb5ec48e59252547e590256bb : SUCCESS in 8m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc56acc7756c431999a7b3e531676e1c : SUCCESS in 10m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de7b602677be40dd942d9bb71d521521 : SUCCESS in 13m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad86abca65b34bc186e8a691e83afd5e : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab46fd83b514bf2955cb0803e991a96 : SUCCESS in 9m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/34154265e9714a1db8c21627b23dd631 : SUCCESS in 53m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6c9284b1521e4bcb996ecc9f8a2ae9b4 : SUCCESS in 1h 01m 39s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7148bea57bac43b485b5d5951b5d4401 : SUCCESS in 39m 32s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/3287d5c5c8fe4733ba13fede8f6b2f6c : SUCCESS in 1h 09m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/baed3ab958e5458882242ef6a1c77074 : SUCCESS in 44m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/01cccab32bab47cdaed879941bffb4e5 : FAILURE in 43m 57s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/95d4cd0370b24ca0909f90bbd186a60b : SUCCESS in 1h 19m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/e2f321c34beb405ba3ef5adb6752be8f : SUCCESS in 1h 12m 56s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f05ecedf78da4e3d91633a963c04521e : SUCCESS in 16m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae7b47ad8cd946e187f2f510d74e633b : SUCCESS in 15m 21s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/ffaee11fadf145edaf7feef318c82d7d : SUCCESS in 58m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bbebe49002f4a978bc7b7b94d50aa6d : SUCCESS in 40m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00ac434b5d0341f5a1872a8918c2218d : SUCCESS in 40m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/15e728ac975342b5ad34f9d245e63b4e : SUCCESS in 24m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bb8a905ecaaa4c64849134c52f5ff5d0 : SUCCESS in 27m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"26286c9ec7c45a29760c63df22b69f048edc18b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 05:57:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"86a2a335bf246b251a2be0fb0cae7e786074d3bf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-19 05:58:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/dce99a7/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"8a7277e2538219ea9044ef7418748d2fea6f6ea7","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-19 06:56:18.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/f623afc/ : SUCCESS in 1h 42m 17s","accounts_in_message":[],"_revision_number":1},{"id":"86e8a0b174e664002caa2b74752a99d570080647","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-19 06:59:54.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 23m 06s","accounts_in_message":[],"_revision_number":1},{"id":"dc09f1d0b3f803780fe72fefde0d55849e799124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 07:55:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9e85898f28e545998e635481a492f508 : SUCCESS in 1h 15m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/204950648066422db23cadb8a5997ece : SUCCESS in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fd79ea625d64fe7a80b2195df53a37b : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac563da825c147c78f7734a42eec00bd : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46f8c2455f704a8dac07a30b2dc43665 : SUCCESS in 14m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5ee9148d4548e68c22aba25fb5dc9a : SUCCESS in 10m 24s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2390cc0eb2dc43c58915bcf6ba103b13 : SUCCESS in 1h 10m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f672d97d37fc498aaa3fcb354350a435 : POST_FAILURE in 56m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3450c1e3f7e64a37988e10e26df5f664 : SUCCESS in 55m 07s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa8dae92331b40a5a6f04133375077e0 : SUCCESS in 16m 56s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/3e8f3bae8def449dab596515c8a1cb69 : SUCCESS in 1h 17m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/43915a41ab0642d3b1516d4286c6dcfd : SUCCESS in 1h 55m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e82446c905649338820e2f7bfc4bd48 : SUCCESS in 1h 03m 23s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/76f93c67234142e9aba9ce5157338940 : SUCCESS in 53m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3577596766114792b6085a7843e85500 : SUCCESS in 59m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f95200f609e64630bfb4c0e5ba57feb9 : SUCCESS in 52m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/733adc3f8dde41fcb219ce25927747d5 : SUCCESS in 48m 46s","accounts_in_message":[],"_revision_number":1},{"id":"56c6de8fb9463e74cc1989a14e2e757b8e2bac52","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 08:25:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"958a91c0261ebce704b23419d6866db0eb0cb6a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:05:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e320c409d8142cdaa0cd3441bb244d5 : SUCCESS in 28m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2131759a586b4c98bf524e7425f12f4f : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"647a26a9f62477322788289781df511d85607c41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 10:10:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4d372a6441c242f0824fb9442d1d659b : SUCCESS in 1h 12m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0e9a33ce7e34b1c9670318621fc5ceb : SUCCESS in 18m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d1d92b118914c2399be393f5bf07c23 : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/924d1bcfdea1455d93a10c0af201bd22 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6fe43149641455c8a458c1337551420 : SUCCESS in 12m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e60e181cbd243559ca1f08a49764ea6 : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb7101c1ba4a499997313e395a98a21b : SUCCESS in 14m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ae84c01549a48a3b98e74b89e7c2ba1 : SUCCESS in 10m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/136080e2b807465ab8114155476cfc98 : SUCCESS in 1h 15m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/459e23d16649442c80c5c9a0e85b73d0 : SUCCESS in 1h 11m 32s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/595aae85777443049faa4313fc6968b9 : SUCCESS in 52m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a812e66f69f49d6b82f7e2c56bae6c3 : SUCCESS in 1h 20m 13s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3786cba4ca5445798aa09001551352b0 : SUCCESS in 53m 32s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3cb560fe31bd4424a306134c90192886 : SUCCESS in 49m 47s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/62e866d8d4d14b169f58673c698dbc63 : SUCCESS in 1h 29m 38s\n- nova-next https://zuul.opendev.org/t/openstack/build/41354c91d50840bf869fbc4b1e57c90b : SUCCESS in 1h 38m 42s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e9e1a55293de4476b9a3d6429486abb2 : POST_FAILURE in 20m 25s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c552b96d72b4bc89a321b8f33be02d4 : SUCCESS in 21m 41s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/1e7ff2fcd0a04cd7968745674480af79 : FAILURE in 1h 06m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27253396215d4dff8d8d329087bfb04d : SUCCESS in 1h 00m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0de66952fe84d6d8ee7de3da5d3a732 : POST_FAILURE in 19m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/aba8e476c6354be4890d8d4c8f126ab7 : SUCCESS in 32m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/def530c99bb2418e97a695286a6aca76 : SUCCESS in 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bba1517288c48d1aa0796751188baed18761edaa","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 10:25:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84622b40b5183d0949b708e458d827ffb4528690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 10:55:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90b4bbcedcea4ff1b6716d5dd0c23038 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/367ae4dd237d4288b18bb26db8c78aa8 : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e1516f7df891190ff77995ea43c1e6bfa16319e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-19 11:41:25.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"bef72c06df55f4f5c741bc1cfa643e124ff63f54","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-19 11:47:55.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 3h 21m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fcabe8c648157dcda6b56063739284573f6a385c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:54:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/734fb5ecd0104b37b098714705dfd570 : SUCCESS in 1h 16m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfff0c8572004de4911583e1c35152e4 : SUCCESS in 19m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c74af8424d604c828580910f9b46fe2e : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ac6256e9654f358485643ca86ce07c : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/582dd211a4104455ba8c89f210b4839b : SUCCESS in 15m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04d534ba7844467a940efb2623252f22 : SUCCESS in 13m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8eb7e2ca535d4db2a5cf9c8d3f8d9d2f : SUCCESS in 18m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3951738efb304dddbd7bfa3e2c86cc90 : SUCCESS in 11m 32s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/15f9b29f9040480cbf01cebec6173e19 : SUCCESS in 1h 11m 51s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fcf27fb827164206b89e51265fa547fa : SUCCESS in 1h 10m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/061770f2918e4114b0aff1d60dd6f5a3 : SUCCESS in 47m 04s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/42192b2cebd04458b7791e36627055f4 : SUCCESS in 1h 13m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7afff8d9e29346e2b8b06469e17424e3 : SUCCESS in 48m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/6782eb110a574cea9185ea01eb1210a4 : SUCCESS in 48m 54s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1da0f881526a4ae0aaef146dc00a2047 : SUCCESS in 1h 26m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/f850d2f430b140df9091fa5f6e04a7d8 : SUCCESS in 1h 24m 52s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b58a5c8a01f04b55b6d2ba8f6df9b976 : SUCCESS in 21m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/229c4525764045ecbc592848f7390efa : SUCCESS in 24m 09s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/3f709b3015474a6bbf3c87909993f24a : SUCCESS in 1h 06m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dffe4fccf4ec4a35ac55a446d762df02 : SUCCESS in 54m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1deeb984b8fd43dab7ceccf1de19ace5 : SUCCESS in 48m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a9ce6568de82475d8d15a7af6ae122a6 : SUCCESS in 29m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a483446d69e54474847748951895b5ac : SUCCESS in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3b59fdd86cc2e024513dadaa0afcafbfcb9b9554","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:55:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2f5868e7a073bf97436f0a2e01e73b254639263","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-19 11:55:03.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/d7eabaf/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"2bc6cd34e08cd95820b2f1ea5b25d7447db72a70","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-19 14:47:33.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/7d357f7/ : SUCCESS in 3h 00m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5bfdb9c297384147917fa578d87ee91b45f6c62a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:57:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4438c594fc684f9c9cc15303bfcae898 : SUCCESS in 1h 18m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbc26c4fdfae4423b77cc9f998b6a61a : SUCCESS in 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64058b54028421aaf7278e6f5eef5c3 : SUCCESS in 9m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc30a6e42c854ea9a0351e5df59f1369 : SUCCESS in 13m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33033e35669f4bef954b886a11eb9347 : SUCCESS in 13m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49929d4d94547078f4a7c7930a9d7a6 : SUCCESS in 10m 08s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/93aad81f5f5c425eac50d65cc47fd3bd : FAILURE in 1h 32m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9c06260ef14942aca8599b2ca7a13320 : SUCCESS in 58m 25s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/acf389f23a1d4346a6220362d9549348 : SUCCESS in 57m 55s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/170df1a850664e0586abb22471343470 : SUCCESS in 18m 29s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c8755a0921f94fccb354c42433855e6e : SUCCESS in 1h 42m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/c2fe30e0c7d946e9937abe24768dadfa : SUCCESS in 1h 47m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cefa550a40f4cf3af4b0d826afab33e : SUCCESS in 1h 20m 52s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/da93ec696b3f4cdc984ca4fd1d86646f : SUCCESS in 1h 01m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/cd554e7d74344374adb376c8baabbe8b : SUCCESS in 1h 09m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a30d2d0184154f6b87324388aefba253 : SUCCESS in 53m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/840071d8b4dd4bd0918cfcb147e8d322 : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f7bd1d51309374e41c4ae70c823457610b9a836f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 19:47:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b480d34dcdfafc3ceb4487f0ad90e2e8c00a5d4d","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-19 19:48:44.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"53795ff25b2bfb5861e3917c5ebd16b9e0e191a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 20:21:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6c4f0bee3db418cb74a2e8768729d77 : SUCCESS in 25m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7567886651354250b08becff5c7e237b : SUCCESS in 29m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c15680e821170e3d894a086b2d167e4fb2e3483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 21:14:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/11f29de5a4f44628968ba9f42727a46c : SUCCESS in 1h 17m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48daf05c24db461b96dc5c272c3da791 : SUCCESS in 14m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14d047428943427ea1e9703a8f6fe0f6 : SUCCESS in 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b49d06b58dd64e16832c14e6e3951608 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f491194e72b47ffb624c710595d8323 : SUCCESS in 12m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4193f11b7c44ec88864c2c03cdbc26a : SUCCESS in 10m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/718bf651890b4a85b8fef71c1d530282 : SUCCESS in 14m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9344a4d7b034b8086316930dc354094 : SUCCESS in 9m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c1f98d61abae49a5a8d3fb5d37985520 : SUCCESS in 1h 19m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1262960289f4326bb27c16d3ffc124a : FAILURE in 1h 04m 17s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7c6c0229a84a4be8834ec4b00dc5b503 : SUCCESS in 46m 41s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1c70a74512054266a69d0c722e2c4249 : SUCCESS in 1h 13m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/584e7e202a88450a9bc2dda68e392774 : SUCCESS in 46m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a4817787e2784069a7ac781a81b1779f : SUCCESS in 48m 02s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2da119f21b0048ae9a4e3177b56b6fb2 : SUCCESS in 1h 23m 52s\n- nova-next https://zuul.opendev.org/t/openstack/build/afe17393bc6142c692019e27ed7a2e27 : SUCCESS in 1h 04m 58s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c004699191f457abbe7e65325ecabe9 : SUCCESS in 18m 55s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59808010817b4f04a01be08c106ecbd6 : SUCCESS in 18m 13s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/fa8a742bef064de5abbcd93e4f52f550 : SUCCESS in 58m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce609d7f0fe24022885485144d694b5d : SUCCESS in 47m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ef1e2458a61b433e9166f30f3478f23e : SUCCESS in 58m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9b28894b9e4d42318c3d5d8e15a82980 : SUCCESS in 24m 11s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/a0fef36710974bd4aac5e04d4d452c18 : SUCCESS in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b61cf3017387401a9a2cb6d8aec51f55924d74c9","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 21:18:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"371a07a20313494f49a77843a63a5f698b3d995e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 21:46:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7bfbb5a8410b41ba82d3ce37bc79d37d : SUCCESS in 19m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18c7e6b0900e44d3848ccf3e82001ca3 : SUCCESS in 22m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f1ff8da978b10efc6fb88a2d8dc4f1a0b0bcfa1d","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-19 22:17:06.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 28m 09s","accounts_in_message":[],"_revision_number":1},{"id":"4ff3767d2438bf554877a0510ea9e26139499307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 22:46:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/678c61c8d93146c0907079c9c774772d : SUCCESS in 1h 11m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/782813ccad7e41d5a5b5824f738686e3 : SUCCESS in 18m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/176fc04001fd4b31b7b0b27bec61efaa : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/789b000d2f144e2eb6cdf9bd94a1cb6b : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c586feb72a74fdcbf7e016b63a20fd0 : SUCCESS in 13m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c8eebb0abd14e179277f24134d065da : SUCCESS in 14m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1d4bf99c21a41b383b5959d3a10f6a6 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/265193722dbf4cf7a61553f5100dc082 : SUCCESS in 11m 01s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0d74c0909643434b9657dc430c76bbb0 : SUCCESS in 1h 11m 07s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a64d911e07b4e9485dc1a7bf2988156 : SUCCESS in 1h 25m 16s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/7f0fc26e5aeb4aed9cd2dc35f9057177 : SUCCESS in 59m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/556e132643d74309a22434cd34fb430c : SUCCESS in 1h 22m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e6002c4c07fe4a078f33710d336627ef : POST_FAILURE in 49m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4ac5467bed4040c4ba71c831d14f9342 : SUCCESS in 49m 00s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5ffc4fa758f64573a12eeffd5996e865 : SUCCESS in 1h 09m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/261324890cac4270bbaa2a43098a7e55 : SUCCESS in 1h 14m 00s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/243183d9d83b49ba93a946813fe0963c : SUCCESS in 18m 31s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/558141c0cf7b46888c21f721106bf384 : SUCCESS in 14m 15s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/eb6bcc06d5ac4a1989b7158c2ced32f6 : SUCCESS in 1h 15m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4396b291e40438e96edbb4a72324bf6 : SUCCESS in 44m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95e2551909bc40f499f6fef9e28351e0 : SUCCESS in 38m 28s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8aee44d6bf2d4c12b73e9cd026c70217 : SUCCESS in 25m 21s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/bcbd698dd6f548d099ca47d8020f8884 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aca4041049165b3c6df555fd85607a554fa7187c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-19 22:48:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15f60c126317c55dae2826fb09a6cc5df01b507b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 23:27:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/21d1aeeafb654c3a99220525a51d3626 : SUCCESS in 27m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e38ef99c73644eedb7baefb5d2f3faff : SUCCESS in 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"38ae38834f8bfc702b16bd2e9defd9472a6a34dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 00:38:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/26ed768fac0e4caa84886105a1d16c54 : SUCCESS in 1h 18m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/764d099d41c042edbd198577afaea6ce : SUCCESS in 24m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f55e9bcbc744badbe9baea4fcbc262d : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a353f9c12ef44420aebd5432468bdbb2 : SUCCESS in 10m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11e17e4e94db4fe6b41a7d749731b581 : SUCCESS in 12m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/550f2bc3a8dc4b47b02e419a55d6d583 : SUCCESS in 13m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cf99f191afe4e828d5e2a8e97f944fe : SUCCESS in 15m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca37bfb5b104da586a8d575425aa683 : SUCCESS in 9m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/320a106f4cbf4885b32a65bb9d5981f3 : SUCCESS in 1h 17m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/88c41987b4724f3ca7f45b8e071dc3c5 : SUCCESS in 1h 13m 43s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/b83643686aa849bb88b39925ceac64e3 : SUCCESS in 1h 02m 26s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/163e402f15804ac4a111eedcaac8e9d6 : SUCCESS in 1h 18m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/95f19ee133c74648b02678ebf51393c7 : SUCCESS in 56m 06s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4d16bb89be5e418d8a9e8b9bd6b5d46a : POST_FAILURE in 22m 22s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7efa9fc62a04430ea8c9b78a10f10b01 : SUCCESS in 1h 34m 20s\n- nova-next https://zuul.opendev.org/t/openstack/build/f931a547e9ea4264a3e8171cd0427019 : SUCCESS in 1h 44m 36s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bcd69476c9b0491b980baa10eed8be4d : SUCCESS in 20m 45s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c942676973de4799a3d44835573587f2 : SUCCESS in 22m 14s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/b5c11ba08e8b456e80a07b7102af141b : SUCCESS in 1h 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/870a2b74854f4aef99478d79a7951162 : SUCCESS in 57m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8df950301dc4f0ab2d13b358b4723af : SUCCESS in 54m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0d20076a4e0c411db8f55500ce9a1b71 : SUCCESS in 32m 09s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/3265c27866d4439d9888a5a7b2ff0e1c : SUCCESS in 44m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1b6f3c90597defb1e1531882f5eb47d3f2d07bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 00:38:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e888a8d403acc2ca024da7ee8cf7594996875042","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 00:38:59.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/34575ea/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"224153d28106fc06c6528bf70dc1fec181089789","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 01:15:39.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 26m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6dfe1ea92466418cb7a251c03c0138f526fd8053","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:14:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8056702c8cb24b0ea2636d3a00caa407 : FAILURE in 1h 14m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d9e553cebfe478bbf1bf0dd17d91a57 : SUCCESS in 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759d2b402eae4f5faf578824279497fb : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f3ebb38dbd849fe8e63fbdb72d38849 : SUCCESS in 11m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f48c16bf9304a8e9ae7b72dec8b7804 : SUCCESS in 13m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26b4646c877d4338ba549fe24e744bde : SUCCESS in 12m 18s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c74710cb700146cebf2ab0f3cd162e59 : SUCCESS in 1h 18m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cf6140d27cf94c9ab8bdaa72c8903c77 : FAILURE in 56m 37s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/92d1c6aa0a7e4874a4b3baa2ca362e4e : SUCCESS in 47m 59s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c83829d670ca4cfebde044dc4bfebb31 : SUCCESS in 13m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9e75650bfdc347e0b03ea40269727fcf : SUCCESS in 1h 26m 21s\n- nova-next https://zuul.opendev.org/t/openstack/build/61a6f3bcd1234c1094b730fa279eb2ce : SUCCESS in 1h 11m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/237827c4ea414e39bf09b34d9af412ee : SUCCESS in 52m 50s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/d53680027ecb4435a1e1223e06e6b9cf : SUCCESS in 54m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/45143d1eb27f41709ab414c9870b712f : POST_FAILURE in 1h 05m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0a8394c16f44f57a661b69c8d0b7739 : SUCCESS in 40m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0b8d379d5db46009a18123099a43994 : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4d6b832ae682eccc4e99720bca0a27ac614d7970","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 02:18:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f8069742f416569a91eb3ccdcafa49bd470cb859","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:49:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f42c4295d764959b3ea2a32c7779551 : SUCCESS in 24m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c107211a043d40f3aa87cb14392e4d2f : SUCCESS in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8fcb94cd08784749ba09f65a696065aaade8d2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 04:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/02b25d6c71ef401c848e8a0bebab37ee : SUCCESS in 1h 29m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e06109152e34d86acb49c2355dd9f61 : SUCCESS in 16m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f66eabf282a4cb0be85546d8b1d0ffa : SUCCESS in 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e2eca7aad2485392665fdbf4e386bb : SUCCESS in 9m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb948a9e4857460e8b8fd4c4f817db62 : SUCCESS in 15m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/562891dfbeaa4d4b9ae6885694c34594 : SUCCESS in 12m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe40f6e2b073428c94e917adcfac0521 : SUCCESS in 16m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b1750414e8494db5c3336f54b9cd49 : SUCCESS in 11m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3153d8703db649ec9b9573a2fe824682 : SUCCESS in 1h 24m 56s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bebfeeff30fa482f8861f3d7e910c28a : SUCCESS in 1h 15m 22s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/04e08ed1b2c54c85aba50be051e9a4ed : SUCCESS in 53m 39s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/52555fb30abb4558bdd2de09348ee68d : SUCCESS in 1h 31m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6f5e642601064d64a166a69e73ecd754 : SUCCESS in 48m 24s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/3e0fbeba0ea149689fe6ab0d7a07454c : SUCCESS in 55m 45s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7053f5b1eb244bcb9a60ad1192822d33 : SUCCESS in 1h 28m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/53d386277d8e4e248e51639085fac058 : SUCCESS in 1h 36m 04s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85b1b9055d6947c9ae241a83737d2859 : SUCCESS in 19m 58s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6324648aa810420786624deced36723d : SUCCESS in 18m 10s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/4dd889087b914938a8be58a0bb55e914 : SUCCESS in 1h 02m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/884425e01e064a84aa3d625a5fa61fbb : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29288ef6af194966bd8abecf40eb7aed : SUCCESS in 49m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dc1a938735624c13a4e8313d85e187e1 : SUCCESS in 26m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c245fa7c8e2645e68527ef5403f4541f : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"264b0f070f410f0ee452cde5f91be985ff0c2537","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 04:00:40.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/3739ddc/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"963f3aa4ee10930a4c28129f1754a2fde1bcaa20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 04:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e70290c26794d0f04b8c55c6a69dd4df8b996782","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 04:40:02.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c6c4a471e4b66a6a5009600a0ec8d038b4e01851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 07:12:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/59876271034640ecb07d66b1fc1ee452 : SUCCESS in 1h 26m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3369a5370f34777821563e1e8b17f9e : SUCCESS in 16m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a0135e758c9438083007abdff94b6ad : SUCCESS in 9m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81909f46b8464867ad854a5575c43f83 : SUCCESS in 12m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c4dc57a5b6f4485ac7b452ac2ffb1fc : SUCCESS in 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6a2905fc13413692d37404246b59e3 : SUCCESS in 11m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/ca7c868fbed240799572df0c2ad0efaf : SUCCESS in 1h 28m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/644aa648b5ca4db1b2eb114fe12caea8 : FAILURE in 42m 14s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e79483e6ace94222b1cb16e14e261334 : SUCCESS in 56m 05s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f1b4ce74140c45568d4aa393176669f4 : SUCCESS in 17m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c53c21b819c046a4bd08baa5a8fc8968 : SUCCESS in 1h 21m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/4f6833344862443ea1c710bd0828d495 : SUCCESS in 1h 23m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5e7bad80de6424885e094aa5571898a : SUCCESS in 1h 05m 15s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/ebc5df88937d4f9e89fd102d9c6987b8 : SUCCESS in 50m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/7b5b7e7ea0894d56a36b73189382418b : SUCCESS in 1h 05m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/819dcac87bef44bfab73766253da74fb : SUCCESS in 48m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85cbe31d042e4eb9960ca2e3b09ed03e : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a66aaaf14b48bb5a2039cd36fb96d4120ff03ff1","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 07:18:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6ff26a9930f076abfaa160e03a5e3f59c433187c","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 07:19:19.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"06ffd392fae25265f2271f6840f2fc481ef51b9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 08:02:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fef0ba2a3fc148a080b268dd4b1ee4ce : SUCCESS in 39m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a9351bd8b814c67ac7b4fc5ea3ed589 : SUCCESS in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9a87113dfa137f4e51efeeabc551fa4876e28aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 09:04:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/290970cd3ccd420fb04ce637b848f9ad : SUCCESS in 1h 23m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18d75e5a73264390b1d4270e617de520 : SUCCESS in 13m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7084270726ff414f94f422e043c4fa2a : SUCCESS in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad1524c7702741df92fd145a58e3a7eb : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/449605618d4e4add92227fe40bc523ef : SUCCESS in 11m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/556494251f5c4ec293fba36d32d6add0 : SUCCESS in 10m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85c506196d7f4d32b4fd5dfd4035f5e4 : SUCCESS in 11m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/820762d6ab2c4861a217452fb997fbab : SUCCESS in 10m 16s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1ca55401339043618373d49e191e1140 : SUCCESS in 1h 14m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5278d7e71b2457f8df0c475e509e9b0 : SUCCESS in 1h 09m 42s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/45467174118a490e8f616a64d1c5ab90 : SUCCESS in 59m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/778929224571473ebef26af07c463759 : SUCCESS in 1h 17m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9ed780c91de340089286244642d0b3d0 : SUCCESS in 53m 08s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/019a235ba810484883a410a67ee9fd0a : SUCCESS in 1h 14m 19s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e30f02bdadea4514ae06ce4ad285e452 : FAILURE in 26m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/1bce77e506784fac85987f76dfa12dbe : SUCCESS in 1h 28m 31s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9d44fba7347f4f8fb875a9d2524e1aeb : SUCCESS in 22m 28s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5238535a94ae4893b234a70604626aa1 : SUCCESS in 19m 47s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/10457bc1c1254d049550596cb2236d32 : SUCCESS in 1h 05m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5f7ea3733db4ea4b8fa1516094fbae5 : SUCCESS in 56m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5346ce54201947b1bd2073530273e9aa : SUCCESS in 56m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/df1225fa6f6948d0ac668712c595134d : SUCCESS in 25m 41s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c596b30c87444ccbbc34ed2826b4cbfe : SUCCESS in 43m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c14ccbe0355b7bb216d9f701de4ed9a4ff7f7114","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 09:18:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b808e7763d51239350c5f548c079aef25f4ab09","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 09:36:53.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 17m 17s","accounts_in_message":[],"_revision_number":1},{"id":"820122bcc7ae4bdac16a55c4b9da40f7b0438e74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 09:45:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49d3e96bba58483998e575f3da51a3d5 : SUCCESS in 18m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74d50929d037472e9b5acece70ba959d : SUCCESS in 20m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4edc1d0b60e26538489397e43b38efdf39a93086","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 09:54:23.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"c7c5a67738598812bdb60361b8ef5ce98b81a930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 10:41:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d281bd09116b42a6b6162beb5209308c : SUCCESS in 1h 16m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/761edef2a5464ac7ad846ba57f63c8aa : SUCCESS in 17m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46d64bced3c24c9e90b713a7eb1ad0e6 : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/455cd35313a94c1c9be22233283d28fa : SUCCESS in 11m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2103ef3d4fa4be29a2eceb87fe2d8ef : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa0999371077494397fe60712d7d35f0 : SUCCESS in 16m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/322440c9c10745c38be09a4df418c164 : SUCCESS in 13m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533c314a73fb43e4998a0fa14a75645c : SUCCESS in 10m 59s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b72d87b3229448b0b8d9a815430e8949 : SUCCESS in 1h 15m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5863614bca74504af5d44bb730e7418 : FAILURE in 1h 05m 08s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/812ab16f66c3413e84a9643e47bd35e5 : SUCCESS in 1h 00m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fa76c60aecd542a98ed577c480e66654 : SUCCESS in 1h 17m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c2a50ac43119411ab8be96da7302f43a : FAILURE in 46m 02s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/13210833395149e08c842994709ac8af : FAILURE in 51m 14s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/449c1253460542bb8cd7e6a797c67935 : SUCCESS in 1h 18m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/d90c140a043c48ad97b4756622867f7b : SUCCESS in 1h 20m 09s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1e657adff2b04466bc128b24155d2941 : SUCCESS in 16m 17s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/731aee636e3b4c13a79f6983b380c024 : SUCCESS in 16m 09s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/54bccb5eb0f44ea5a688ce95e25b992c : SUCCESS in 1h 08m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/904bd7357c994f1bbf8dd78fa1735f65 : SUCCESS in 56m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1576d12f22b4ff3b73db8304414b6e8 : SUCCESS in 45m 26s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/99194824ecfc45b4a0a885f5ac956599 : SUCCESS in 40m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c27682d74a83433a9bc8102f8212439c : SUCCESS in 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcfbebbe0e5ed9e63a4372917f857f3f638c6d23","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 10:49:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa208b7803c263f85fe3819b8547824f15f191b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 11:30:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/163ad63c6a3f488ba9ed843c156fee0b : SUCCESS in 18m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5de7a1428141401e972b62f8d0974a02 : SUCCESS in 37m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af40873372a2798a15f7dce48c5e91665339a410","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 12:53:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d30934863f944bfab0d41afc8bc3e08d : SUCCESS in 1h 11m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23e508bba75b486b92fd9fab8ba981d8 : SUCCESS in 16m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90a45570af4b4b54a698809ab66940b6 : SUCCESS in 14m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4acd2c169ee4dd8ac71161cc45d84a4 : SUCCESS in 8m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de02a5d36e6247a8867b07ac5586333c : SUCCESS in 13m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4f6dd5e90b64febaea5603113d33bc9 : SUCCESS in 13m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83405b1f5389423491cd63ffc7773064 : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe131ab5abd455e8e8de4061823cea2 : SUCCESS in 10m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/eba8ea7d9c8949b398cc3277b85b9fa0 : SUCCESS in 1h 17m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d8103d4f53e4614999b0bee182294fa : FAILURE in 26m 49s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/17fcfab8e8354a868774889a46d200cf : SUCCESS in 1h 00m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd7ec4b02a85421eb5c67ec6540c1186 : SUCCESS in 1h 33m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3614e50fe1d74ee583b11acdbdfea730 : POST_FAILURE in 21m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c9f69824dcd14a11b732e84de43a019f : POST_FAILURE in 1h 12m 54s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ceb801b4fd9e4aa1a67fa9862d973ae7 : SUCCESS in 1h 31m 16s\n- nova-next https://zuul.opendev.org/t/openstack/build/3301769b944945e78fc3679c6aea78db : SUCCESS in 1h 48m 34s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb3939f8bd1042128ef60c0e138f79e5 : SUCCESS in 18m 21s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22e9a8f66ddd42acb5c1b873b2d3b2d1 : SUCCESS in 21m 47s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/6475a4365aea4cf497195c69b2912ee3 : FAILURE in 16m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5228be65ea7b4bbabbc1531dacb3f76b : SUCCESS in 1h 03m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2050ebce9e74915baab9218eba0031d : POST_FAILURE in 22m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c322f332abe74a218f3e6111b1510626 : SUCCESS in 39m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2ad4993733d34de9984996b857f463bb : SUCCESS in 29m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a2046506115710429df58e54bf9c6282b5f23da","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 13:04:13.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"8072ca5b47fc00d0e8e1721c0cd47652d1a9bbb8","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 13:15:20.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 25m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7c5b3a1de33a893e03cbbc5f7156161a0f3d7a9c","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 13:19:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"091596cba5a736439bec5b6b1d724d532b646f3e","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 13:21:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 36s","accounts_in_message":[],"_revision_number":1},{"id":"e792f7de51a0095bcc9b18e9aaac000e6fd591c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 13:57:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/063eb990d1ab4cc6ab10a9f6f3452cc1 : SUCCESS in 29m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bde24ae5b73e4755894feff728b9f8ad : SUCCESS in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77fd4e3e0eafcde7f68ee20e15f912fd3ae7a383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 14:40:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/00a66d143e8f4019a6fab263bba3977d : SUCCESS in 1h 10m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80d694c146a549dd9cd050bc73a6e6e3 : SUCCESS in 14m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/811f8e5bcfb540dca2deeda6795d9633 : SUCCESS in 12m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57dc47a3eed64149a78a9108b2b5533f : SUCCESS in 10m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/589221ba993043a98dc08eb5939f1d64 : SUCCESS in 13m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24b78bccf568425ca46a8f9ce83596f2 : SUCCESS in 13m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/018f9a9ea5c04e52a7d536392a55677d : SUCCESS in 12m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03b18b0a18e4b5f9fc8305066abd3ee : SUCCESS in 8m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0d4ab36058e44e5aa7e48cd7c5323219 : SUCCESS in 1h 01m 41s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5a3a3a04f8340b5b1df1e07e44ba8c5 : FAILURE in 1h 14m 25s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/fc5f93688a764433b937f28fefc25488 : SUCCESS in 48m 05s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6231978ad37143fab5e586a0c3b99b21 : SUCCESS in 1h 15m 45s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9938507ea1c3408883876e694a03f94e : FAILURE in 46m 21s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2a764600370e4d7691f75d5dea75fe2d : SUCCESS in 45m 13s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/084419c59f6441158daf86d30df41f03 : SUCCESS in 1h 11m 02s\n- nova-next https://zuul.opendev.org/t/openstack/build/8fa057b1e193401c8cd0ee8cdde1d4c8 : SUCCESS in 1h 06m 34s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a6f9ea64b8564073aaa7ae49b1ded249 : SUCCESS in 21m 13s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28e2a6fc440248ec8ca7f6925c92d648 : SUCCESS in 19m 42s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/873ef9de7c8240e9b10a4d5ed16edbe4 : SUCCESS in 1h 03m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaaedce7c31744e5bb6c9582c6af416e : SUCCESS in 53m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cbee9dc73a8c480d9ea614ef25d1c60e : SUCCESS in 51m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/a7b9d4cf97bf463a85cc49b5e0450b3f : SUCCESS in 25m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/6a16bf52e02d410c9a92665a200488d1 : SUCCESS in 34m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63956d3d557c4ed59e4761964e31b338a691696d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 14:49:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"670949898c3a154c4a2b60684e17a8bfba5fdb70","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 14:49:39.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"fe9399e39773e3b4e8da79dc968e49cf32acf93f","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 15:24:02.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"f3c07a2f2b800ca43c9fe87b487479d98f9ae1df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 15:32:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72592a872eb34b629fbeca903a1cfafe : SUCCESS in 19m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc1f998a2dd449ffbf11b4da12564c7f : SUCCESS in 37m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3e127db22b1f508644c2371cac7f4b710b83868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 16:35:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/32f400baf0ea43c784cbf529637c8a64 : SUCCESS in 1h 15m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8469072e60a847e0a2a5b1219fad20e5 : SUCCESS in 15m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36c78758920f4e1e8c6b35fd89927323 : SUCCESS in 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76bacd1783b14ef7ac6ffa7874122f82 : SUCCESS in 10m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a69350996b04620a86bd508a62174ad : SUCCESS in 13m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22399f5b222c4a52a94e5040b1314cda : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff32ffbfd221453eb7eee4fccc5c5cef : SUCCESS in 15m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0bfd685e9234047949eacff3bb9876d : SUCCESS in 11m 07s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7876832883df49da9fa5ba26216333c2 : SUCCESS in 1h 09m 28s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e38b129f73ec4fe98e9da4a2199dca61 : FAILURE in 1h 08m 09s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/79000f55bc674ac0b75e45f028559e99 : SUCCESS in 58m 45s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/bad8eb61e84849b8832edb69bb338f5a : SUCCESS in 1h 22m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/99caff34aabd403599293e02b2e433ae : FAILURE in 56m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1a7b49931d0849ae96649eec72cdfd93 : SUCCESS in 57m 16s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f520cfe97fe34a20b613affd2fc668df : SUCCESS in 1h 38m 53s\n- nova-next https://zuul.opendev.org/t/openstack/build/6e693582ee1741f288de33ddb62b05bf : SUCCESS in 1h 37m 40s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/00becaedc2c741b2bca8cfcab83e412a : SUCCESS in 21m 24s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ccd0f02d15d4f4890e8911ab163885b : SUCCESS in 20m 10s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/d1e48d825a5b4f76823978879086e683 : SUCCESS in 1h 11m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75f5ecbfb8bc4ff7abbc80d924e4c6ed : SUCCESS in 1h 00m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1e39b0d857d4975a46ab79319cb4487 : SUCCESS in 52m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/8fa916994bc24a6c9ce5102aad44bfa9 : SUCCESS in 32m 00s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1b57d1fad5a6467591fbc5b1139cadcb : SUCCESS in 41m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d671ff9c0944078274827e1b802501371db17d94","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 16:49:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d170492977770209d297074ef6b18bd0b1e6c96","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 16:59:05.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"3e2ff652ad30d4acf4e31c65747ec2ed5afdb36f","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 17:25:31.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : FAILURE in 2h 35m 02s","accounts_in_message":[],"_revision_number":1},{"id":"24641ee03f9e8e4ce39d112eda7a6d5b1e7aa6b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:59:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3036133b9a2a418d830f32cafb29e005 : SUCCESS in 35m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1cf81ce3037c48188574995622c964df : SUCCESS in 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2247cbac1ddb1b55b490d6bc27ad6eaef05d9f2e","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 18:01:57.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : FAILURE\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"57089fb03adc80ee554f57cb8f41d020acbc3058","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 18:41:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e0abcd4e4db543819a275cc47a6ed1fb : SUCCESS in 1h 28m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9e54e9432434063a8fa5aea3dab978d : SUCCESS in 20m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2e5a908cf3b4e64afc5bbc6de6e2f74 : SUCCESS in 16m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a8230f4657c4cc3ac188fd2fee1ad59 : SUCCESS in 13m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddad4c8a2c9a4948bf34622845ab31bb : SUCCESS in 12m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f14de975e37d458a9087f6139c70c694 : SUCCESS in 16m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b906f2f28c794f39b4139d92494e1c31 : SUCCESS in 14m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a5e61dc6b84e1793b87ac941e25b5a : SUCCESS in 12m 41s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/4cde3be25fb8411584b4fbd104d97740 : FAILURE in 1h 38m 47s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/902c8f1ac23e4f3ea2044bae7e2f3771 : SUCCESS in 1h 08m 31s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/18a65db1fe854c0b96a5ebbfb7cb283a : SUCCESS in 1h 02m 19s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/312f49608426438697105c8040bebda7 : SUCCESS in 1h 26m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/795193be0aed432bba243b216635a55b : FAILURE in 51m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/28868e2ee35f416593d25f1752da921b : SUCCESS in 48m 54s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/41506c48e5b64acab6103bfde3c6fda8 : SUCCESS in 1h 32m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/e80ee4c8329544debf4c844b0674a55b : SUCCESS in 1h 35m 09s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2192cd124ad74bd6a9a590aa725884f8 : SUCCESS in 20m 10s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2647f89d0cd84a15bf9976f2d4f7c03e : SUCCESS in 20m 34s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/76de3dffe6b14a5d9257e0083705eb15 : SUCCESS in 1h 03m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/507c1c7bb8a54734af226a0e0bdc8440 : SUCCESS in 56m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e980ca5b74446de94a1d3cf78fe2674 : SUCCESS in 51m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/97c6d7c49cef4d6299c460e18ab63eea : SUCCESS in 26m 42s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d40f551d834f4da8b8bcf3794e59b3d1 : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a89aca26195eeee9220a220e3909168b6b9dc89","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 18:49:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7571559189ac433c352ee66cc27706f4baccd59d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6db6088f10cd48378ca166c1f67248c8 : SUCCESS in 20m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dff82d85b87f4da7894856ca5b6fc4a0 : SUCCESS in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9840fd9ef80b088d933d25d7b7600b6eb47d016d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 20:34:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/34200c7b74bb4d4a819312d3b535424c : SUCCESS in 1h 18m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1bd956e95d0e4273931c022100ae01c8 : SUCCESS in 16m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed47a6fc99254f62b3147a5deb6fadc0 : SUCCESS in 20m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7961b9a849ba41509babf24d4ccc3fa4 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/334e0e2ce6a54620bb9909ac9c6c7eb4 : SUCCESS in 13m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35049a3aaab04025998138ffbc5ccc55 : SUCCESS in 12m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85bdccea78974367a7a8e596c7d6385e : SUCCESS in 18m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06522bdc6bc14aee8f32959d42cbcdd2 : SUCCESS in 9m 28s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/9fff676b758d4977836ca5d7a39c65e4 : SUCCESS in 1h 10m 21s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/177961a515c84d719e41bcb24c320d4c : FAILURE in 1h 10m 48s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/9f128a8e7c2b47598636d870dce8a5cf : SUCCESS in 54m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f650f15071f741eebb93093b3dc3762a : SUCCESS in 1h 26m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/529cc154b6094ef2ac2e3bba5ab2764b : FAILURE in 56m 22s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b88e2fd9bdc04d4d82cad8768748406d : SUCCESS in 59m 21s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6eeaa5b80b55484fba66e5c62179c34e : SUCCESS in 1h 41m 50s\n- nova-next https://zuul.opendev.org/t/openstack/build/65c0477bf9ab4437b05e97129d4e1cb8 : SUCCESS in 1h 35m 15s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40c69773c6564aec865d6feea218c426 : SUCCESS in 17m 18s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f648cc12d9740e1b05b2618ed41c895 : SUCCESS in 24m 35s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/f646a58250a24b3cab19cc8204f72e1a : SUCCESS in 1h 03m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d8a3d8896d3453cb8a8138edadaaefe : SUCCESS in 1h 01m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7200a1ef41fe4b3e8f52f017369d7e40 : SUCCESS in 47m 04s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/ed442e76ba01495cb052ef1fc8d8f516 : SUCCESS in 38m 36s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/909633958b0c4d0081ad7d32693eb7ee : SUCCESS in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"70c067e16387bea5d5f5cf39863f8cf6abced544","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-04-20 20:46:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"97504ac5e65f230732785edec9cd117dff3feb08","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-04-20 21:14:24.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/783275/1/iscsi : SUCCESS in 2h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"efc7852c7fc1a378230ace59062fd6ede8081b59","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-20 21:21:49.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"f9be7e2d51cd6801c718937b91a9a6be6014f1a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 21:35:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a7f517ab14bd4e9c94c7254e5713920a : SUCCESS in 26m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cad3c4c2ebb04bd4acec3f1869c4d013 : SUCCESS in 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49cb3c0c0c6dd0302fe8359fcaefa04bf494ac15","author":{"_account_id":29963,"name":"Intel_Zuul","display_name":"Intel Corporation CI","email":"intel-openstack-ci@intel.com","username":"Intel_Zuul"},"date":"2021-04-20 21:40:51.000000000","message":"Patch Set 1:\n\nBuild Succeeded","accounts_in_message":[],"_revision_number":1},{"id":"142ef7735f2b7e7296419af94ebd85f2974cb59b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 22:22:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ef19edbb638f477f970098f8f34fbdda : SUCCESS in 1h 22m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72668191af8142f69b8df8384a8dd8ac : SUCCESS in 18m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67c60f87dafb4a8fb0ba51b160b335d5 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd67f08f16904b24a4f642becb836eba : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c0e2bdbf89440aa684e4d68ddf65a8 : SUCCESS in 16m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/747a3476882648228dd75fa77a109bdf : SUCCESS in 13m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a3739a2b31d4e9c88d18dfadfceca11 : SUCCESS in 16m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619467acfd514fc6be0930ae803557da : SUCCESS in 10m 03s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b757280748d94b6080e6585d92d24a75 : SUCCESS in 1h 21m 36s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/95ee853544d74a519c0f728c88925adb : SUCCESS in 1h 14m 06s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/3e80d3dc2d4943aebd7ec10135cd7e13 : SUCCESS in 1h 01m 06s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/230ed065d991458c957482c9192a9b32 : SUCCESS in 1h 12m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0fc49b7f30b04cc5bc3f88781f2593c5 : SUCCESS in 48m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/eca6655ef043408eb5fde6c17643cfb9 : SUCCESS in 47m 57s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d3deeeee3b7d4c62a4981630ef013cc5 : SUCCESS in 1h 19m 32s\n- nova-next https://zuul.opendev.org/t/openstack/build/cbeeacecaf384b3baa607c4a001eb6d9 : SUCCESS in 1h 28m 45s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b064b0d8587b49f4975d5a422bf69bc3 : SUCCESS in 16m 44s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbbe4e6c9e7740c59e7ac6aafe47ebe0 : SUCCESS in 16m 49s (non-voting)\n- grenade https://zuul.opendev.org/t/openstack/build/448fc507c57648f181823784d9f5b2d2 : SUCCESS in 1h 05m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1ac562712ab4077b6c681de743e6765 : SUCCESS in 51m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1a98943a10e4fbfb9835e37d7024b3e : SUCCESS in 46m 24s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/fc9be5c7a597450cadff04ed05dd2efb : SUCCESS in 33m 25s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/227c8cd36b584ba3b39ad6d40f65e827 : SUCCESS in 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53cace887c64ef4ac067a1f7244d66dad2cb4de4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 22:23:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7071a3aff76ccbea5a2824688dbab9d7280df3b9","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-04-20 22:23:36.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/75/783275/1/check/EMC_PowerFlex_NOVA/3813237/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"690163bac8ee60a0cfdca9f33f195853d7cf942e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 23:57:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5553e08f4ea34a5aaff17baf8d687ce1 : SUCCESS in 1h 07m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20b74c264eab4ec9a19c1bc22aa162c3 : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2a252e206b463dbe3a2713f39ca505 : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ec85179eab04743a0783e25fbb340da : SUCCESS in 12m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44f169e4443a4a788b1a88eaab9c26ad : SUCCESS in 12m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d72100dcb1324b74b63232678818a8db : SUCCESS in 10m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f8cd6d92a144a6f933098c7542856fd : SUCCESS in 1h 17m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6034a915fe4d4d7fae1e9eb9c1037ba7 : SUCCESS in 46m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/1ee9115298f849e99948cee31a8b3247 : SUCCESS in 53m 43s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/69f48bb1ca464c0d83bdfd57e70d8d0b : SUCCESS in 19m 23s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/da2bae76e4d44f14bbbf3a60fbd75363 : SUCCESS in 1h 14m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/b18b1f16b5094ab5a2da69ed18c1240f : SUCCESS in 1h 30m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd76d92cc026468ab56d8236c819eb1a : SUCCESS in 1h 12m 03s\n- neutron-tempest-linuxbridge https://zuul.opendev.org/t/openstack/build/bd3ffd6b43a1487397fce0bed221ca85 : SUCCESS in 1h 00m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/9d74d84fccd048d681063bff9fe49468 : SUCCESS in 1h 02m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd0000c02c874b7a9b8b873103f25b3e : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b66f9839a6e4e35999fe857279b8140 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":1},{"id":"bb55977b9554fcf6e7cee65ee3f36667894c220c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 23:57:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0c10fc4d17f9a5eaf6069f9c59f7e32f4f0fe7e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 23:58:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/357ddfa8411b4e2fb4edd5f00d05cf0b : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"fcae0abc5a88955efe395f49d70a1d96bdc62205","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-04-21 00:44:07.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/75/783275/1/check/tempest-dsvm-full-focal-py3/baf9307/ : SUCCESS in 1h 46m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c17058d32dec4e86cd00512c8d90969b7817fd42","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-21 00:44:48.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : ABORTED\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : ABORTED","accounts_in_message":[],"_revision_number":1},{"id":"8d2c226190c8bafe33fc6457079fd54a1c7e2b19","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-04-21 06:00:24.000000000","message":"Patch Set 1:\n\nBuild Successful \n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/75/783275/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"388498ac5fa15ed8deef06ec061ea47e4a1b7377","revisions":{"388498ac5fa15ed8deef06ec061ea47e4a1b7377":{"kind":"REWORK","_number":1,"created":"2021-03-26 11:30:26.000000000","uploader":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"ref":"refs/changes/75/783275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/75/783275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/783275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/783275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/75/783275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/75/783275/1"}}},"commit":{"parents":[{"commit":"ded25f33c734ebff963f06984707a99fe76a9ee1","subject":"Merge \"Remove short call timeout from func test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ded25f33c734ebff963f06984707a99fe76a9ee1"}]}],"author":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-26 11:18:37.000000000","tz":60},"committer":{"name":"Slawek Kaplonski","email":"skaplons@redhat.com","date":"2021-03-26 11:30:17.000000000","tz":60},"subject":"[neutron] Get only ID and name of the SGs from Neutron","message":"[neutron] Get only ID and name of the SGs from Neutron\n\nDuring the VM booting process Nova asks Neutron for the security groups\nof the project. If there are no any fields specified, Neutron will\nprepare list of security groups with all fields, including rules.\nIn case if project got many SGs, it may take long time as rules needs to\nbe loaded separately for each SG on Neutron\u0027s side.\n\nDuring booting of the VM, Nova really needs only \"id\" and \"name\" of the\nsecurity groups so this patch limits request to only those 2 fields.\n\nThis lazy loading of the SG rules was introduced in Neutron in [1] and\n[2].\n\n[1] https://review.opendev.org/#/c/630401/\n[2] https://review.opendev.org/#/c/637407/\n\nRelated-Bug: #1865223\nChange-Id: I15c3119857970c38541f4de270bd561166334548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/388498ac5fa15ed8deef06ec061ea47e4a1b7377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/388498ac5fa15ed8deef06ec061ea47e4a1b7377"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
