)]}'
{"id":"openstack%2Fnova~783654","triplet_id":"openstack%2Fnova~stable%2Fvictoria~I2881d77d52bcbde9f3ac6a6ddfb4a22a9bd45c8a","project":"openstack/nova","branch":"stable/victoria","topic":"bug/1908075","attention_set":{},"removed_from_attention_set":{"10135":{"account":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"last_update":"2024-03-05 18:59:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2881d77d52bcbde9f3ac6a6ddfb4a22a9bd45c8a","subject":"api: Reject volume attach requests when an active bdm exists","status":"ABANDONED","created":"2021-03-29 14:08:56.000000000","updated":"2024-03-05 18:59:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5889fda52219b653e1fac4a8bc175f9153684c00","_number":783654,"virtual_id_number":783654,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-30 10:36:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"recommended":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"all":[{"value":2,"date":"2021-08-31 15:22:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-16 12:28:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 16:16:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 16:48:12.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 12:28:22.000000000","updated_by":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"reviewer":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"state":"REVIEWER"},{"updated":"2021-08-31 15:22:07.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"8728ceb70240bdccdb78c15039e2256ecd13b539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-29 14:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29c3b52ac2b796a2fe6e201a3f880b13563150f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 16:16:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/b90abf9db3fc4a5cbeae16458caa0c74 : SUCCESS in 1h 45m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77d6a015a1fe48c6ae95b057a2f1a296 : SUCCESS in 15m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75e82e8a39a440f0af507149098d9689 : SUCCESS in 15m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ec08c9b2084020a7c8971bcbd4a5b3 : SUCCESS in 10m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d66b31e581347dfa0891ebbda11ddac : SUCCESS in 15m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bcefc8c032e4f7f92765ef3f27acc66 : SUCCESS in 13m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5947086d06044a285e0b118c3697abc : SUCCESS in 10m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ec93dc115020491c80dc82b67d0dbee3 : SUCCESS in 1h 08m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38335c946a3f47978b6ba8b342970005 : TIMED_OUT in 2h 03m 57s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5e9a4dbec40b492d914d0038d5720405 : SUCCESS in 1h 05m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cab967561b4f4ea78398cab71e9d270e : SUCCESS in 56m 49s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6ace8a7d478d454aaa5614944a3dca5a : SUCCESS in 1h 55m 30s\n- nova-next https://zuul.opendev.org/t/openstack/build/53f6ebaee41a4abe807c341f401167a3 : SUCCESS in 1h 45m 40s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c9e26c25141446a9e25a214b3cdd6c7 : SUCCESS in 17m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/b837514f40f74e60ad520101f8c37e41 : SUCCESS in 1h 05m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef9d9e6355c64018b84b918b85037cb1 : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98fcd92fb9eb41d6b117a815562a6036 : SUCCESS in 57m 09s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dd3dbb40a33d443ea6a75d32901742f8 : FAILURE in 33m 55s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b39525b225ac43fdacf4051a0ed82c94 : SUCCESS in 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f6c91ee4c2add9ffa0cf3a880ad708941bcf67f1","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-29 16:48:12.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/783654/1/check/tempest-dsvm-full-focal-py3/fcf2d95/ : FAILURE in 1h 23m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7689b0be9dea4833256e50f6f96a358aeb27153e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-30 09:04:23.000000000","message":"Patch Set 1:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":1},{"id":"753e733e5512140467f1e69308484774856b0d99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 10:36:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8a6cdd358df344b98984abf62b689682 : SUCCESS in 1h 23m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7ab42b6f414455b69eb85ac45417f0 : SUCCESS in 17m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3e2502ea6b645169ff968ad57ce8308 : SUCCESS in 14m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1aa772769b472792f7619b82254c19 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba3c515958ce4e4386eac70db397bd3e : SUCCESS in 13m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13305a890e584786933e2b9448fb13f0 : SUCCESS in 13m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/febe0a1a6207488bb959a4e9eb98101b : SUCCESS in 12m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/39eb3342334049eeb3d176e18f3c90db : SUCCESS in 1h 14m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89008e1db5c845beac9874aa0fcf2e9d : SUCCESS in 1h 13m 02s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6d33b6350c8a4686b606c76760c7cebb : SUCCESS in 1h 10m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/57bbcb1949a74149a78938a30c18ef15 : SUCCESS in 50m 33s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/02c95cfbc0db4731bc2ac593140aa579 : SUCCESS in 1h 23m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/aa5cbd5db8874b19afb6a42aeb935c4c : SUCCESS in 1h 29m 36s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4215607db32745cca08c112fe310334d : SUCCESS in 20m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/b8ecd5106d8547bb9be67a2bd6fd3570 : SUCCESS in 57m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/773677a47c31448f895aafb72b13794a : SUCCESS in 55m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/632e81a20644442382f1755361c954f5 : SUCCESS in 54m 40s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5244b54c95de4bf5b850fe3a6321f5a8 : FAILURE in 29m 46s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/1b18e295c82e44a88f8f4a1ebf23a96b : SUCCESS in 39m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3046eb89ed02ece33b8c1bdc9a48a9279d1a39b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-03-30 11:47:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/54/783654/1/check/tempest-dsvm-full-focal-py3/734014d/ : SUCCESS in 2h 19m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fbbd69ec1160794b4af0c84878451cffd0be1339","author":{"_account_id":9373,"name":"Vlad Gusev","email":"vlad.esten@gmail.com","username":"s10"},"date":"2021-07-16 12:28:22.000000000","message":"Patch Set 1: Code-Review+1\n\nClean backport.","accounts_in_message":[],"_revision_number":1},{"id":"61df0c9126b3b6df4c0c8436a5da3a414c96cae3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-31 15:22:07.000000000","message":"Patch Set 1: Code-Review+2\n\nThis looks valid for backporting. Merged in stable/wallaby (master at the time of merge) \u0026 clean cherry pick. LGTM.","accounts_in_message":[],"_revision_number":1},{"id":"c911af7b49fd49e899cf789e4f524323bac505fa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-08-31 15:22:20.000000000","message":"Topic set to bug/1908075","accounts_in_message":[],"_revision_number":1},{"id":"5889fda52219b653e1fac4a8bc175f9153684c00","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 18:59:11.000000000","message":"Abandoned\n\nstable/victoria branch of openstack/nova is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/victoria if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75cdbdd9c64353e8b73c36b184ea2140f0a4861d","revisions":{"75cdbdd9c64353e8b73c36b184ea2140f0a4861d":{"kind":"REWORK","_number":1,"created":"2021-03-29 14:08:56.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/54/783654/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/54/783654/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/783654/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/783654/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/54/783654/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/54/783654/1"}}},"commit":{"parents":[{"commit":"8e2a3dd2f08a2f0c420bc1da9c02ff6b128c06c5","subject":"Add regression test for bug #1908075","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8e2a3dd2f08a2f0c420bc1da9c02ff6b128c06c5"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-12-24 11:50:43.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-29 14:06:50.000000000","tz":60},"subject":"api: Reject volume attach requests when an active bdm exists","message":"api: Reject volume attach requests when an active bdm exists\n\nWhen attaching volumes to instances Nova has previously relied on checks\ncarried out by c-api to ensure that a single non-multiattach volume is\nnot attached to multiple instances at once. While this works well in\nmost cases it does not handle PEBKAC issues when admins reset the state\nof a volume to available, allowing users to request that Nova attach the\nvolume to another instance.\n\nThis change aims to address this by including a simple check in the\nattach flow for non-multiattach volumes ensuring that there are no\nexisting active block device mapping records for the volume already\npresent within Nova.\n\nCloses-Bug: #1908075\nChange-Id: I2881d77d52bcbde9f3ac6a6ddfb4a22a9bd45c8a\n(cherry picked from commit 1252588d4e48da1d3a753639a8a4d937acf3e037)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75cdbdd9c64353e8b73c36b184ea2140f0a4861d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/75cdbdd9c64353e8b73c36b184ea2140f0a4861d"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
