)]}'
{"id":"openstack%2Fnova~801610","triplet_id":"openstack%2Fnova~master~I0051c0f8694699865bc16636953908d60ba456d2","project":"openstack/nova","branch":"master","hashtags":[],"change_id":"I0051c0f8694699865bc16636953908d60ba456d2","subject":"Bump oslo.db to 10.0.0","status":"MERGED","created":"2021-07-21 10:24:00.000000000","updated":"2021-08-07 15:07:09.000000000","submitted":"2021-08-07 15:05:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801610","meta_rev_id":"7ea8ba4675db3e2c68e5335609d476f71c190dd2","_number":801610,"virtual_id_number":801610,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-07 15:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"recommended":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"all":[{"value":1,"date":"2021-07-28 14:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":2,"date":"2021-07-22 09:22:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-07-26 10:23:57.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-07-26 10:23:57.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}],"REVIEWER":[{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-21 10:25:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-07-21 10:47:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-21 10:49:58.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-21 11:49:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-21 12:48:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-22 11:25:11.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-07-26 10:23:57.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-07-28 14:09:09.000000000","updated_by":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"reviewer":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"state":"REVIEWER"},{"updated":"2021-07-29 12:47:35.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-30 07:44:05.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"CC"}],"messages":[{"id":"c588d185475033c35e2dc7aff91ae23a05a15b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-21 10:24:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9296a7c34e3ddc7cce8d60675e9a85b80c94627b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 10:47:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32a5dc38f30a48a89cd910fa59046b12 : SUCCESS in 22m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c7efd29433234b08a681a3e0e5c339f9 : SUCCESS in 18m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"530d55bb1ba3f4bce8e33c0d15ec414b7f400661","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-21 10:49:58.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"60a08abb98da26b2eba2768cda01c2c68dc51673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-21 11:49:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/057a77d24d7f447e9305f3fbef4ddc61 : SUCCESS in 5m 23s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9142af610183428eac6c6e90d03d8518 : SUCCESS in 1h 03m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac9d52dbb5c64a6ab6bb72cc27554897 : SUCCESS in 14m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aff457783e7434eaecc6506cc1543be : SUCCESS in 15m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2569804d3ce41d0a3f4ce8374aebd72 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a0250c22dd648018c47c3760e8b28e4 : SUCCESS in 13m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86c61db754ca4e20b36b8e2f3ffdb679 : SUCCESS in 13m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02ca20739fe346448edffef2166f6e57 : SUCCESS in 10m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3208ca3694594b3bad3722cb2c1f0a52 : SUCCESS in 10m 51s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1a17b192b92c4fda906ea61c24965626 : SUCCESS in 1h 07m 24s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed3f957ed9f64e43a76771b7ae4b6c1b : FAILURE in 30m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9f1b0902e0ae45a09fb561f01a1446ee : POST_FAILURE in 46m 13s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f0841de9567640b3b394738740c50e10 : POST_FAILURE in 33m 56s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/18b933db1ba34d9099785a64255c3287 : FAILURE in 23m 43s\n- nova-next https://zuul.opendev.org/t/openstack/build/11a1303c972b4a66b7bb0c05b6330cf9 : SUCCESS in 1h 21m 27s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/311bc4c150fc4dbea87471066e2eaa5a : SUCCESS in 6m 37s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/0367322301f8452796f9eda280a88e32 : SUCCESS in 22m 57s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a7aebce847be470fb8b9ea68fd10d17f : SUCCESS in 18m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f8f88348b5b4fa894d6c6d6d1ea5660 : SUCCESS in 18m 11s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0e30d8a451a14f4ba1f31db19fc9d574 : SUCCESS in 1h 04m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6705c753710499997b87585343e65e3 : SUCCESS in 49m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a02c28c758b44ac6acb36383e195a23f : SUCCESS in 44m 02s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/1b590958eda147b8a6b1f09099a77a06 : SUCCESS in 31m 27s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/d1924b020e4e4dda89608ef220b22bb0 : SUCCESS in 40m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a3cb636dc8900499926ef5b66f8657d09cf178ec","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-21 12:48:06.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/e748dcd : SUCCESS in 56m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2c115eba01a7c2bc7ef4db51a4fbfc5ed0aaf169","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-22 09:22:17.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck failures look unrelated","accounts_in_message":[],"_revision_number":1},{"id":"1f3d6c7e0a8a46f132d87874d6e1418b77254bc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 09:48:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea8506e2589141248f12375b81f409a5 : SUCCESS in 17m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d219cc227984446fb2b2608e65616cc5 : SUCCESS in 20m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5e8ae9e93699b14a20539db5654119289c2a9179","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-22 09:48:19.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"83536c4445e0642a41fcef26d19f825bb432ac24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 10:57:25.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1360dc4a4e0342b9ad3cbab6025dd065 : SUCCESS in 5m 08s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6b4100dfbed744849924876dcd5e693a : SUCCESS in 1h 13m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98fecea32faa4be5b584800eb6736893 : SUCCESS in 17m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f909c5e926d4ff28bfc63247e4f2bd2 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d4479081b7481d89688d6dc80adfae : SUCCESS in 9m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aed0b0f0a0b947e2a3a5aa6812e05269 : SUCCESS in 13m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a72dc1b29de9498788c8d98ee9ef6405 : SUCCESS in 14m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4fa57e6b8464e88bb2aa2a8913d18bd : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c081a7a5f824f7581539b6155adf9db : SUCCESS in 11m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d4a4b621f5444e0b85a9fafed3fcae19 : SUCCESS in 1h 11m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9d8ddb37827d4cbb814a7f7ea49455dc : SUCCESS in 1h 03m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/effa8db45d16493c8e7a4abeb635a917 : SUCCESS in 56m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0c8145c7e3ae403996f03e02de452ecb : SUCCESS in 46m 29s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6d6aaf7221be432ea85af9a25a2e6e0a : SUCCESS in 1h 30m 29s\n- nova-next https://zuul.opendev.org/t/openstack/build/7a856df3629a4b16a03017914e441787 : SUCCESS in 1h 20m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/7d60149581b448058e26161dec323a82 : SUCCESS in 6m 53s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/8bdafea904ca42b6a544f41edf36b57d : SUCCESS in 21m 14s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac91fefc3c1d41faae22bc27596d2b85 : SUCCESS in 18m 12s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb75a4f1a6a7457bb4f4cfe258c7d71c : SUCCESS in 17m 04s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9746aa937d2447c5b78e55ac3cf79a96 : SUCCESS in 1h 14m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f3334799b9a4159a4bdbfec0354ec53 : SUCCESS in 51m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57a6a7826f8f4e7e8d2d0a2b5a651844 : FAILURE in 42m 47s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0c87fe1e90f14d25ba080e9a075128d9 : SUCCESS in 31m 59s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/de0d53a8d05d4f9e81233a60f86a5d86 : SUCCESS in 31m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b9980be7a71d3f15c8b6a191745b4c95d8820fa1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-22 11:25:11.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : SUCCESS in 2h 01m 57s","accounts_in_message":[],"_revision_number":1},{"id":"de7e367066fdc1c4ff0fb0fe5cdfece2e6722df6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-22 12:12:37.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/522ace9 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9eca9809560fcadd5abab340576f8e5fb3089b98","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-07-23 09:48:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"010433877885f26e790eb10c2092699cc27c74d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 10:27:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46c715bb25d14a6fbc8bc03cfdca57cf : SUCCESS in 26m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/faa7021e28274ad98d78073bfe875184 : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"36a5ec623f84c5cca6b2902c9d7b96538fa60580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 11:22:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/547a664e0f6f4a79b4a1e6f0f8858ad0 : SUCCESS in 5m 26s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3a55f964321e49aea2b66ec4440182e5 : SUCCESS in 1h 09m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4ccef1b18f04b30bc663809394b9368 : SUCCESS in 15m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7ff275b4f0e4a29b1056ab7087a4984 : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f514d98dd74895868290410e600743 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/491d232c7e9545e783317368bd4e2353 : SUCCESS in 11m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f9897bddd8c44a0940ee9b90d950e00 : SUCCESS in 12m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30b91786b17e4bbb80bef3ca70f7265c : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b2eddcd6b74bcfb6f51631521da730 : SUCCESS in 9m 46s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/45aca02f87204df1b1f25caef9e1041c : SUCCESS in 1h 20m 58s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/daa41472028f447da2d5a259fc30993b : SUCCESS in 1h 09m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b772d841fcf54c5a8084d97de3e6780e : SUCCESS in 50m 30s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/43b1b878ce8247259a3f162b7ca641a7 : SUCCESS in 45m 41s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8645d3112c064981a11ef78c351f6787 : FAILURE in 1h 11m 34s\n- nova-next https://zuul.opendev.org/t/openstack/build/33902392031047f5a99a6debc1929b16 : SUCCESS in 1h 28m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/756328afab7446d0b26c1cf537744db6 : SUCCESS in 5m 56s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/2e826893e77d479e898652eac43f67db : SUCCESS in 18m 14s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca65078229904156a3670c21433d61c1 : SUCCESS in 17m 31s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f81ca0937c444d6baef59f09d5a55492 : SUCCESS in 18m 00s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2fa1bbb0abb14f538f38a68ba1f1017d : SUCCESS in 1h 08m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2c44171c79f49919ffc359dd9ce9a12 : SUCCESS in 50m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3802cbf95524fe1b3370b0dc20f1460 : SUCCESS in 53m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4c3af2eeee6a4b79a6fb546f37942034 : SUCCESS in 33m 22s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/22bd23359e0b489c99cb8824d12a8a0d : SUCCESS in 37m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"03b5550e5f03a20feda87f4817da3f1345bcc81f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-23 11:49:03.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/801610/1/check/tempest-dsvm-full-focal-py3/20033c4/ : SUCCESS in 1h 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"476a6fb6f15380d4a79573343d569c07bdd00e98","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-23 11:53:15.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : FAILURE in 2h 04m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a26c74ebce13e4cfc3dfdbf1cdd84a8154385daf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-23 13:42:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d1f26917c9d03ab9072c02710ff9315780770eb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 14:10:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c9c43b3e269407893861c79e843ab47 : SUCCESS in 16m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5eee1fcf9ba742669819e1ddfad3e1b5 : SUCCESS in 25m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fca35e529af174f4b21ec97389b15ab31a3c2870","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 15:18:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5232c371de1340c888155895609761a6 : SUCCESS in 4m 49s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/07ccf531b7344d63b81659be9bc10c8f : SUCCESS in 1h 09m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9336a9d54bab45a9965f6e95ef25434d : SUCCESS in 16m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7978ade20be42fdb0653d01f9804481 : SUCCESS in 16m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69125bbd62b04612bbc89628a074fb1a : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e544d539cdb4c98a6480d1a70739d2c : SUCCESS in 12m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4706f5d30f47475e936e403e01f458de : SUCCESS in 12m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfb1f9e27d48440093c421c244e978fa : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faa1fc3b53cc45dbb81c014ffd5878a3 : SUCCESS in 9m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/95e4f1a65efa4718af4609d587b951a6 : SUCCESS in 1h 29m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/adc9f5f49a0a4cd48193490997e52334 : SUCCESS in 1h 13m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e4116342e1f41838542cb0facaca8f6 : SUCCESS in 50m 46s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/a697a0c5b95e4e849bf91c22fa82b626 : SUCCESS in 47m 32s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6e47ba75016241949ba797b22096367e : SUCCESS in 1h 20m 25s\n- nova-next https://zuul.opendev.org/t/openstack/build/0fc02b6c19ac416583721e154ead2bd9 : SUCCESS in 1h 26m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8529bf3f90d04a769c896d4176aee38e : SUCCESS in 5m 41s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/5091a480304a47d6af481b336a1ab5c3 : SUCCESS in 17m 42s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a49c1cdbc974981bb8f94bd0518b16c : SUCCESS in 17m 01s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1766b30f856144b88d81554e8072bfe7 : SUCCESS in 18m 42s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/39b197ac8abc4f34b36e14a75314ef45 : SUCCESS in 1h 08m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28c49b9b667b4829955b879fb0f1674e : SUCCESS in 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bab4ccef51b54df292d8916c0020330e : SUCCESS in 44m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/69e2fc4b7a9f413fb9508ebaef79c396 : SUCCESS in 28m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/916693738a1d4327865dde835ea2d93b : SUCCESS in 37m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"09a13009203b27709364273577a64bdc3f6c6799","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-23 15:25:57.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/5ead580 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ea431565ef466bec1d0653a0ba1d1e15aa8b6713","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-23 15:48:50.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/801610/1/check/tempest-dsvm-full-focal-py3/85d0aca/ : FAILURE in 2h 02m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e053a7277f98692ee15acc6cf6f8eca9b4850d37","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-23 15:52:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : FAILURE in 2h 07m 50s","accounts_in_message":[],"_revision_number":1},{"id":"070fe2d09e4cad9860707448f9b99145781a9103","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-07-26 10:23:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"a19aaf93eaffecdd37f04e0c4928d6e6b98295d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:24:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca00d77424eeec0f7c1d273cdcacca77cf98db1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 12:56:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/89c7dc7e1a4f4b2e94228fa8cef01c51 : SUCCESS in 4m 32s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6e766ab0c9684d389c0ffe3270f667bf : SUCCESS in 1h 07m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb154009baa44deaa0d06a70421de2be : SUCCESS in 14m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b5435075e494eb1995835fc30002b84 : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63fd16c9bc9449799fce38efa0c59943 : SUCCESS in 11m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19b7f6bfc6ca4c068650743e865654e5 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4272ac82e87645eeb37bf9da5f774d37 : FAILURE in 8m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7134ae56bfec4a3e9e1035d647bcc2b8 : SUCCESS in 46m 16s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4ac893d3792b4fbdb44a5143986318fa : SUCCESS in 46m 42s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/1fe7aedc9e2b4380920e22cdff300b28 : SUCCESS in 23m 37s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39126d7af7154aaca6553c492a832e5a : SUCCESS in 17m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/7586e33bbb7b400da7692279883d1291 : SUCCESS in 1h 35m 36s\n- nova-next https://zuul.opendev.org/t/openstack/build/c051f9751b05480ba5cf916132ef7793 : SUCCESS in 1h 18m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/d5c1dc77fcaf4d3abe7d106d0b1bc1e6 : SUCCESS in 6m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/793cc04644cf480bbb54edc1139ce61e : SUCCESS in 1h 10m 34s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f86030c3518e45b6a2a4e50c22c5e47e : SUCCESS in 1h 10m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c73acc9aebf64c75bd9b0398ce021a8f : SUCCESS in 54m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/749b37fa01cc44299b9c1e8e7168e6ac : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"9fe016b3aa8919a45cb6f510d4d55e7d1efafa2a","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-07-28 14:09:09.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"081aa2d851ee695c435756b1439e791eca5e5610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 14:49:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6cb83b7816d24353aac38f2dceecd7ab : SUCCESS in 36m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50b993538acf40ea87d3f125b4baee48 : SUCCESS in 38m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3de8297e358b3d2895c438ab6a4aa7237e17cd5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 15:40:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d1bfc548422461396d1de93cda52cbb : SUCCESS in 4m 28s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/54d1e815509643c880749325b565cddb : SUCCESS in 1h 15m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06dc8a92846f4f2a8dc08b2b4b9239ce : SUCCESS in 12m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9735414920e94eb992a3521ed8b26c77 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6347329ac02439fab7909290912644f : SUCCESS in 8m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88e714ee3a954264942e9045b868f1a7 : SUCCESS in 12m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03241f7291de4b6da49f9907d391f089 : SUCCESS in 12m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/777ddf27505d43078cddf78b9952b12f : SUCCESS in 13m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8b7298bc0de4f458d3f94e193677fc0 : SUCCESS in 10m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0c60b97e91e14983bad375bb6b6f685b : SUCCESS in 1h 17m 48s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5bc0baf85e824dae9a1dfbdaac435142 : SUCCESS in 1h 13m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7994d4b4b370457e816860da0062cb21 : POST_FAILURE in 23m 10s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/9bb2d4c13b584acc9a7524e29e3d08b1 : SUCCESS in 49m 02s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ffa1eafb0f5142dc88b0af59cc58dfc4 : SUCCESS in 1h 25m 07s\n- nova-next https://zuul.opendev.org/t/openstack/build/5c98d05ddc764944a5db00eb0ba8dd42 : SUCCESS in 1h 23m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/86cd03e008b84b659b2ac689155b00f4 : SUCCESS in 6m 29s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/655055b68cf94fdf984cd5e752ca435b : SUCCESS in 18m 06s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7fb0d1f34948445da6a9e0f8c3265c18 : SUCCESS in 17m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cd6d6595711b4e5797fe3344422f5b0f : SUCCESS in 18m 57s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/7178e06aa9d24565a9a97a60b386da80 : SUCCESS in 1h 10m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1f8e37a559b4a888c8fdcca3c176155 : SUCCESS in 44m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3afa030e66e3410cbabd1bc9527815f1 : SUCCESS in 44m 22s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2f840fc2f6b542409b3a3bba4811596b : SUCCESS in 27m 12s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/fa3e54b8098847988a734e386383aa48 : SUCCESS in 41m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5d1abc81c251ebe84d010cee9312e2e25ae7f624","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-28 17:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/b336baf : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"604c7d2af813df84afd629ca215df2bf3bfbb7be","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-28 17:22:46.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : SUCCESS in 2h 01m 58s","accounts_in_message":[],"_revision_number":1},{"id":"29f24be2ac49356054e7faaafeda30ce64cc2e32","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-07-29 11:14:49.000000000","message":"Patch Set 1:\n\nrecheck post failure","accounts_in_message":[],"_revision_number":1},{"id":"c252c5d58cf89d030d15d05a1074f247372cdda9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:25:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73938b10a9114e38b54ccaf65ef3c8f3 : FAILURE in 54m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f7f76432c9e4fdfbe7f4a2320c71997 : TIMED_OUT in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2fb399f061fb4cca78512db1c9620515ea57dfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:46:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1af70a5ddfb4f8c99970fe12cf7f07b : SUCCESS in 6m 36s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a79614d5f01449fd82ac78605ce4f3db : SUCCESS in 1h 06m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c52d626875d648b982dbc50d062ef606 : SUCCESS in 16m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c939b290936440b9e8e4f7ce5400586 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ae08f4ab482439fad389d9802fa3b7d : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c42e217fe2a047689d0e132ea0af99a5 : SUCCESS in 12m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e7d6334c6c84655aca0ff06bfe38df7 : SUCCESS in 12m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3af05e4135f4cf29f03c9cd35bceabb : SUCCESS in 15m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5110e117f646aba70db3b376d03a7d : SUCCESS in 10m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1b2421ec27394bc18c878cbbd59b159f : SUCCESS in 1h 19m 05s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/42748f5ccf894db388b75be6eed7ebf2 : SUCCESS in 1h 12m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a774e5548af84eee8050b11b7d941233 : SUCCESS in 45m 56s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f826b294a25844149c1b817181fa5a7c : SUCCESS in 52m 47s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/a70ea6f1e03c4c829169388ff203a192 : SUCCESS in 1h 14m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/78b62c114f2c438dba5e6766b2c3c425 : SUCCESS in 1h 26m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ca5a77fe3a184ed7a76c058765c42665 : SUCCESS in 5m 57s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/66c861f882864da69d18d20e6add66a9 : SUCCESS in 19m 10s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41a96fe86fe047418c00002d0d445b40 : SUCCESS in 17m 53s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47ad7fe9f551431f93ea33ff2a7f7305 : SUCCESS in 17m 00s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1d463f73ac3e45c688dcf40385bf1300 : SUCCESS in 1h 15m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf277b1e3a9745a9b6dcf48cca64022d : SUCCESS in 50m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bfb8c3786e541069c0bb27042a1ecbd : SUCCESS in 47m 37s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/5c3eefa6632e45a0977c9445014a5798 : SUCCESS in 32m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/30130cf522ca43558bac54730644e6a1 : SUCCESS in 35m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"126ee0ba88be006cf97b15a3dcbd186698642154","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-07-29 12:47:35.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/10/801610/1/check/EMC_PowerFlex_NOVA/9f4727f/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"c900773516388af6287a9536e059fe7e4a7a69cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:47:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"626fbc5b1f3c0e80b04e43ad52a7d98b08c7d537","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-29 13:35:13.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/801610/1/check/tempest-dsvm-full-focal-py3/2aca93e/ : FAILURE in 1h 58m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2d5a752cec0e8887d9d2f749ce81a2ad28150768","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-29 13:48:49.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : SUCCESS in 2h 31m 58s","accounts_in_message":[],"_revision_number":1},{"id":"672e478df367be37207428930dfd6abc45a77df0","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-29 15:04:04.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/9d68daf : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":1},{"id":"790a8a508c662c58d6adf563c8763ff2ce5e8982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 16:12:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a515411acb540d7b930dd8d7588c9f3 : SUCCESS in 4m 48s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/7879bc07ef39451c9dc539086931d313 : SUCCESS in 1h 18m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b395d9ce28ae431b864ce7baa5b31a31 : SUCCESS in 15m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c1a75df6ab464d9ed3ed7408463781 : SUCCESS in 9m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee32fc25bdf410fbd38436e40ab377f : SUCCESS in 14m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f004f04cc8f44bc98ae766c38ffa565d : SUCCESS in 16m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33063efc19604d6db80ec83872aadf02 : SUCCESS in 12m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6291e9eb16834e4db75a277c1e02cece : FAILURE in 58m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/66d85aaa25694f03881a645a9421f59c : SUCCESS in 1h 11m 09s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/37bf59cc5d854e15b2a1905edce6c074 : SUCCESS in 22m 05s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7073590be53e4c3b8fd13a694e54a3ec : SUCCESS in 21m 31s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0fad006ccb3f44ca94deece01671671a : SUCCESS in 1h 18m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/21d831b3c3494fefa21eedc5b8672890 : SUCCESS in 1h 31m 38s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/1b40d660c32b42288a40ed6f973d9050 : SUCCESS in 6m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8e14491e0544f268660bad30d91d6ce : SUCCESS in 1h 12m 44s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e2ce0eafb16422488fc6f4253815c8e : SUCCESS in 1h 26m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5d382b8b4474b6db1531ad56c9f3c3b : SUCCESS in 1h 00m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9c042c4525d44999eff259f498ccd4d : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":1},{"id":"10a0224767b691457d1698421e69c0bc7031a9ab","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-30 07:44:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c60948ea530b0da22305a2464f09e00cc4bb8e16","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 08:13:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28d1fa0053854da09a8810acefb37143 : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ba0982688154d25b0342b38646b1f6b : SUCCESS in 27m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d70afe6f2820bff76d9bc0d1a351a436daa7669b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-30 08:37:15.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/7c8200d : SUCCESS in 51m 37s","accounts_in_message":[],"_revision_number":1},{"id":"27345035a9b5bb81b2d69dadb680b4d8c840fe24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 09:12:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/372f2f70e2f44fbf95fc9618bdb96daf : SUCCESS in 4m 41s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/6cc92f688c644ab791f7ba6dbbd17cd5 : SUCCESS in 1h 15m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9008667b364244ac922a11bde4558e44 : SUCCESS in 16m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d58271c86c004282bbebf272849318ff : SUCCESS in 14m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a34997345a94664a81e3ac28b248136 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e204a9ad4c4d4ccc84b0f164b28680da : SUCCESS in 12m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f517e1e85be4cc7912d04f5e2816077 : SUCCESS in 12m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c3a71302ae94dfd99ca06e6fa8649f5 : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c26f746d754ce2aad985d86145c900 : SUCCESS in 12m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/07d52c5920be4f4dacc9fa4b4a2124da : SUCCESS in 1h 16m 04s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78930618918b45b0b1aeac31aad42fcb : SUCCESS in 1h 14m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bb492f2a81ba4adfbf8d38e159e1828b : FAILURE in 54m 52s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/02c6fa95f16a420a980a58c043fc60f5 : SUCCESS in 47m 14s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6d23e4cc137c44f0b615c7cfd46e0ad2 : SUCCESS in 1h 17m 10s\n- nova-next https://zuul.opendev.org/t/openstack/build/f18dcf3996f841a2a3d40637e7c4a545 : SUCCESS in 1h 18m 57s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4ff3476c1dbd4c31857308c4a51eeda8 : SUCCESS in 6m 35s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/5dbc746e2c3440638733e8c546228bdc : SUCCESS in 19m 46s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b45f800c05342e1a531dcbe69f2efac : SUCCESS in 17m 42s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/46c6efbbf0354cc1a63652daed869321 : SUCCESS in 18m 38s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/23732f70829e4de482ed9a87c64d0d5a : SUCCESS in 1h 22m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d84585e620da47489aee267a31bec0ce : SUCCESS in 49m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8eabddf6c1241a7937ba8737176b515 : SUCCESS in 43m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/31c350ffc2f040178cbcb62729daabd2 : SUCCESS in 27m 24s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/8d587e1f53b04153ac1f2271f7ba2bb7 : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42473cbcdc08b12f1287b04b91c6925637558ebd","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-30 09:50:21.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/801610/1/check/tempest-dsvm-full-focal-py3/8a39fde/ : FAILURE in 1h 52m 49s","accounts_in_message":[],"_revision_number":1},{"id":"42e137b578ee7e8b568700de23d529c7ee777842","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-30 09:53:58.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : SUCCESS in 2h 08m 55s","accounts_in_message":[],"_revision_number":1},{"id":"b3afd36c01924dfd796994543f75a43c9fbe539c","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-07-31 00:44:36.000000000","message":"Patch Set 1:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":1},{"id":"acceb392d278a8f9807235a5988d2319a18bb382","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-07-31 00:58:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b217519b7792baeae03767cc958dcbba647717bb","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-07-31 01:10:20.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 tempest-dsvm-full-focal-py3 : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"c130dca351aac9856c9f8b23f94a7613468d1fea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 01:14:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03e8e308fe5f4cfda0d16681f9c016b7 : SUCCESS in 18m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b077b293c754ea2a9d66ee188a367ed : SUCCESS in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79f3e773f318d520b8e7a1b8459c253c544a8762","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-31 01:41:38.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/0b8a828 : SUCCESS in 55m 11s","accounts_in_message":[],"_revision_number":1},{"id":"6f072edad29a1bd19e51fd4ecb7674784f8133fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 02:42:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/8c6415a32d0e4f06bd2d311d95b9d87e : SUCCESS in 4m 44s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ac1bfc4c095d4c22b25c1170fe4853c4 : FAILURE in 1h 54m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a4cbd354aa74792a98ba7157418e65b : SUCCESS in 16m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22165feeeedc4c83a6507ccd3e641f9b : SUCCESS in 17m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94a1db8c777492c8d89ee7548e62db5 : SUCCESS in 11m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f919efe99e4f53b13ea88614fde0a9 : SUCCESS in 19m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd747dad57b940cdaeeaae4accd6048e : SUCCESS in 25m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daa7498f77014d1caa395384dd7e5457 : SUCCESS in 21m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f00a7f8dc03d4e2aaa54ba4fff8607ba : SUCCESS in 15m 25s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/95e88a7c6b844a06ab372a7976bcf448 : SUCCESS in 1h 23m 30s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e4d820829374e3ca1f88d582d411907 : SUCCESS in 1h 11m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc8769c8c25c48c28319613043f988d5 : SUCCESS in 52m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b6628ea0b0fa4229943e9b3dfdea47e1 : SUCCESS in 1h 18m 04s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ee886eebde2d449586ab8e0daa07d705 : SUCCESS in 1h 18m 18s\n- nova-next https://zuul.opendev.org/t/openstack/build/931feedc15754318b41274bc4b85583d : SUCCESS in 1h 10m 43s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/77cedc08303a48f098d47d100026b0a0 : SUCCESS in 7m 48s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/ee370124d6724a66a9610147d1c92d63 : FAILURE in 19m 21s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/83395821c56d48f4a49de9be3c4bfd78 : FAILURE in 17m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd5a185ec0da4162999abaa194990a00 : FAILURE in 17m 30s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d38e660165944a19997c84f3d974f594 : SUCCESS in 1h 10m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65b198c0976d4ddf991975626de8fcda : SUCCESS in 49m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52b26dc96eb04155b5838c968039a906 : SUCCESS in 45m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f90f5f6dd33a4a5eb2981cc9563d86c9 : SUCCESS in 30m 44s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ae5761ab1ceb4dd98df2ccb3c78fc036 : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7383aa3d0cb8d77f078a518af90ac06fff264ab5","author":{"_account_id":7634,"name":"Takashi Natsume","email":"takanattie@gmail.com","username":"natsumet"},"date":"2021-08-07 11:12:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ac86beb5192c5fe72cfe2d4ca3270284f24578f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-07 11:49:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b16dbb11510648eeaaf08bd443e16903 : SUCCESS in 19m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/420cc7090d8a4231b0c78ed9ff221c6c : SUCCESS in 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"77dc678ee8b0e36f7328263033a0ddb4e911e1f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-07 12:47:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/78d65f5bbe484c779bface09fc7fa02c : SUCCESS in 4m 39s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/0a8d7e88275e4e7fa7e21443b1865347 : SUCCESS in 1h 13m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/295e006053b040379ee14df9d19dd29c : SUCCESS in 14m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77c6153d17844ff1bd101de360546f47 : SUCCESS in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d954282944461d8cce37420d936511 : SUCCESS in 8m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/256a0e31eaa84d63bb6e03f851a6b795 : SUCCESS in 11m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8bc3b026ed54480b68bccd8e57a69ab : SUCCESS in 13m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b5c7176ad554167982798b82c468025 : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d44b2e5dec485dab481a77bb807131 : SUCCESS in 9m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/85bed4ef47df4f0eb3b88d30f1a1d3c0 : SUCCESS in 1h 07m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1233f5ffef03447b9ac11ab48f12eaea : SUCCESS in 1h 08m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/780913cee1ac476f9663e783e3df59ca : SUCCESS in 56m 50s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b099d4283225401cab1408959cc12e37 : SUCCESS in 51m 19s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/64c36e84fe2748cea322b59a8fd5da8a : SUCCESS in 1h 20m 54s\n- nova-next https://zuul.opendev.org/t/openstack/build/a01e3c46c23d4912aed21011bb8a7183 : SUCCESS in 1h 28m 55s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6b63ae80113443f39230a8ca59302898 : SUCCESS in 7m 38s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/f1441a6daffb44a78703264662ce2d04 : SUCCESS in 16m 52s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/582759211ff04239bbbd412c104b50de : SUCCESS in 17m 47s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af3d4d6068b04dcfa86c7f1eb82b0145 : SUCCESS in 21m 12s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f30eb49262a74598aec488fd2b57c8d0 : SUCCESS in 1h 16m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e88b7d4af3244f73a441a83804e3d291 : SUCCESS in 51m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cf03a53be124de9a70ca61f5a62eab1 : SUCCESS in 40m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/14f6ec1b3e3f4f14b3c3afda3e461e85 : SUCCESS in 23m 54s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/df0006f952cd4e00b5ce5cce12868370 : SUCCESS in 39m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"003964d9ef4da42516da535ed771640e9c598f91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-07 12:48:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01d7d88575d1a96da4a55ff65694208b25ff23ef","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2021-08-07 13:23:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/10/801610/1/check/tempest-dsvm-full-focal-py3/e40dfba/ : FAILURE in 2h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"f0506257268be2b28080bc5c68fb666075ba4d54","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-08-07 13:32:08.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/801610/1/iscsi : SUCCESS in 2h 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6322180dd677b55f5ba0d1c90b20cd6b5fc51372","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-07 15:01:36.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/801610/801610/1/check-not-vote/ext-nova-zuul/42d3c5f : SUCCESS in 57m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e4cb6cf0bf37a61fb1515fc875b217436464b852","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-07 15:05:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3651f9f0cea4b2fb0bf2a860412b10a : SUCCESS in 4m 56s\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/030b997c881641979c6f2f7272cfe3af : SUCCESS in 59m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa1e60caf1154fb6a636d9ec40b628f2 : SUCCESS in 12m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6cb4808e67b4502bdd58f77c852f7a6 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dadd1fb9c0cd480f98a5d1abe2af5634 : SUCCESS in 10m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06df65d938cd497b8e22a7ae49208ee1 : SUCCESS in 12m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5815531b529451d9fe0bc2ae95d6be3 : SUCCESS in 9m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/809a3e07f54248b783f977e2a648e7ae : SUCCESS in 45m 49s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2a9258a1988f4147825405e0c0c34a0f : SUCCESS in 51m 33s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/661dd7e6d1aa4bfd8dabfd4b1705da1a : SUCCESS in 15m 22s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79ede9ae296f4ebeaadb32e922545c03 : SUCCESS in 17m 11s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e40aeb334b2d43ba99a74560cb2340da : SUCCESS in 1h 13m 58s\n- nova-next https://zuul.opendev.org/t/openstack/build/6af40379e12b4ac2a399acc51f6fc002 : SUCCESS in 1h 10m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/bb1f524d94b4491e818ded71d4c67fac : SUCCESS in 6m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/22b9dc10334845a9897b7e1fa6f8d443 : SUCCESS in 1h 07m 48s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e9dc814bca94c9faccd048d0f4c8433 : SUCCESS in 1h 09m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f633dbf80624de4860f5e0bb115eb0d : SUCCESS in 42m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87ee64a46a0946e997005fa3752fc3eb : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7f1086cb443ae0e7ff96d24bfde52dbb853bf3ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-07 15:05:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ea8ba4675db3e2c68e5335609d476f71c190dd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-07 15:07:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d8f60ef18164a9c9e723dc6d2aff4c0 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a65ea33824edc570e04acf994cea26edf874bf48","revisions":{"a65ea33824edc570e04acf994cea26edf874bf48":{"kind":"REWORK","_number":1,"created":"2021-07-21 10:24:00.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/10/801610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/10/801610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/801610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/801610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/10/801610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/10/801610/1"}}},"commit":{"parents":[{"commit":"3545356ae3a719442833cb8c3c911408d4bd3c15","subject":"Merge \"tests: Silence SQLAlchemy warnings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/3545356ae3a719442833cb8c3c911408d4bd3c15"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-07-21 10:20:22.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-07-21 10:20:22.000000000","tz":120},"subject":"Bump oslo.db to 10.0.0","message":"Bump oslo.db to 10.0.0\n\nThis makes it possible to partially revert\nIf48dd949ec4d69a09c87178f16d56a2517e21fd8 as one of the warning silenced\nthere now fixed in oslo.db.\n\nChange-Id: I0051c0f8694699865bc16636953908d60ba456d2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a65ea33824edc570e04acf994cea26edf874bf48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a65ea33824edc570e04acf994cea26edf874bf48"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
