)]}'
{"id":"openstack%2Fnova~805657","triplet_id":"openstack%2Fnova~master~If71620e808744736cb4fe3abda76d81a6335311b","project":"openstack/nova","branch":"master","topic":"bug/1936849","hashtags":[],"change_id":"If71620e808744736cb4fe3abda76d81a6335311b","subject":"Avoid modifying the Mock class in test","status":"MERGED","created":"2021-08-23 15:41:16.000000000","updated":"2021-08-23 22:23:57.000000000","submitted":"2021-08-23 22:21:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805657-bug/1936849","meta_rev_id":"e38eb76eb5ac919bb718cc2a711fd8fecee15298","_number":805657,"virtual_id_number":805657,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 22:21:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2021-08-23 15:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2021-08-23 16:22:22.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2021-08-23 16:22:22.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 15:43:09.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-08-23 15:43:09.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-08-23 16:09:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 16:09:54.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-23 18:02:38.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-08-23 20:06:27.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"7c235d4a68bd67b0db2931a22863683a5344059f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-23 15:41:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bb0afac28439f0d76d59b295487e63add86686a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-08-23 15:55:37.000000000","message":"Patch Set 1: Code-Review+2\n\n🥳 Good job on spotting this. I hadn\u0027t a clue","accounts_in_message":[],"_revision_number":1},{"id":"acc6d0812a4f589865597b68475b43ee6c2d4e6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:09:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/393b1dfd5f254c9699598913367472da : SUCCESS in 17m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56d72788ed3d4ebbb860226e2d9f7d59 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e8587f19458429cb44553bf123f8c24 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efddf2b442a542cb9cc2a9ae82a1ef73 : SUCCESS in 12m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34b135f748ad48bfa174de91419943ac : SUCCESS in 12m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8f87b4ee7ad4055a9b18d5b862da86d : SUCCESS in 12m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58710d6effc24bb8a20ebb17566bf540 : SUCCESS in 9m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/2a05b748885e49f985ed956f76622bf7 : SUCCESS in 5m 57s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/2070a1ff972442a0a67c0af5d925dfc3 : SUCCESS in 22m 26s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/561a23e5f9fc44bc9aeb7ac4746c75a4 : SUCCESS in 18m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ee06c9e41a54451e8d9f6c55942bbb45 : SUCCESS in 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0dadd90dfda41ea2463cb3342cd5c039e339342","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-23 16:09:54.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/57/805657/1/check/EMC_PowerFlex_NOVA/f2f61b3/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"c24f70f3ffc462e5ad602301a89e1172cee7d27b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:12:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88474e597fb64c588e2087ce241150a8 : SUCCESS in 26m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0079a5d009745dda8723facb49f5e9d : SUCCESS in 28m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"62b74c9f71c629dc0245564362e1d3aef9ea00b8","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-23 16:12:39.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/57/805657/1/check/EMC_PowerFlex_NOVA/e055887/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"9df304e76ae2a60ba6d5c3b39fe192d6a6bc607a","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-23 16:22:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM yup great catch!","accounts_in_message":[],"_revision_number":1},{"id":"d037b45b1475501813814d6a8bd4101617e5138d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:23:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34d9bbc2622e5507456fc3fdce0f06a62a2aaa98","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-08-23 18:02:38.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/805657/1/iscsi : FAILURE in 2h 17m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ac5c789c327ae80e6844bc5eba99c0bbd22f442b","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-23 20:06:27.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/805657/805657/1/check-not-vote/ext-nova-zuul/1ec3439 : FAILURE in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"94ac972619df02578e77e4235837930548322bc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 22:21:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33a837cf3eb24bddbba7168a7d5d776a : SUCCESS in 15m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8237ea56304291a26f3fe54bfd6865 : SUCCESS in 14m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3caf205ac2e4f2f8c82338679215016 : SUCCESS in 12m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acc441710c8a46c99f71d72b3bbc95a9 : SUCCESS in 16m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2ffad25fb904458bd0e380c3b26c50e : SUCCESS in 10m 53s\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/9d216fbba245487d87b7a7515ff0e0a4 : SUCCESS in 21m 56s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37f6310dbc4e41308c333e8582c86eef : SUCCESS in 19m 50s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/53f92400fe674ed1a794e0df51e328e5 : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":1},{"id":"1732409a96ff205bacae802142d099100e8f0bd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 22:21:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e38eb76eb5ac919bb718cc2a711fd8fecee15298","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 22:23:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b3e185bfbc494f8a7aceb66d24289e : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"930b7c992156733fbb4f598488605825d62ebc0c","revisions":{"930b7c992156733fbb4f598488605825d62ebc0c":{"kind":"REWORK","_number":1,"created":"2021-08-23 15:41:16.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/57/805657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/57/805657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/805657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/805657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/57/805657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/57/805657/1"}}},"commit":{"parents":[{"commit":"10b1dc84f47a71061340f8e0ae0fe32dca44061a","subject":"Merge \"fup: Move _migrate_stub into LibvirtMigrationMixin\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/10b1dc84f47a71061340f8e0ae0fe32dca44061a"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-23 15:13:14.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-08-23 15:36:00.000000000","tz":120},"subject":"Avoid modifying the Mock class in test","message":"Avoid modifying the Mock class in test\n\nThe rdb unit test defines a shutdown field on the Mock class\nunintentionally. This causes that a later test in the same\nexecutor expecting that mock.Mock().shutdown is a newly auto\ngenerated mock.Mock() but it founds that is an already used\n(called) object. This causes that the later test fails when\nasserting the number of calls on that mock.\n\nThe original intention of the rbd unit test was to catch the\ninstantiation of the Rados object in test so it set Rados \u003d mock.Mock()\nso when the code under test called Rados() it actually called Mock().\nIt worked but it has that huge side effect. Instead of this a proper\nmocking of the constructor can be done in two steps:\n\nrados_inst \u003d mock.Mock()\nRados \u003d mock.Mock(return_value\u003drados_inst)\n\nThis makes sure that every Rados() call will return rados_inst that is a\nmock without causing Mock class level side effect.\n\nChange-Id: If71620e808744736cb4fe3abda76d81a6335311b\nCloses-Bug: #1936849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/930b7c992156733fbb4f598488605825d62ebc0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/930b7c992156733fbb4f598488605825d62ebc0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
