)]}'
{"id":"openstack%2Fnova~805882","triplet_id":"openstack%2Fnova~master~I8c052a6526819834d79d2b5027050f1bdf36638f","project":"openstack/nova","branch":"master","hashtags":[],"change_id":"I8c052a6526819834d79d2b5027050f1bdf36638f","subject":"Reproduce bug 1941005","status":"MERGED","created":"2021-08-24 19:56:14.000000000","updated":"2021-08-25 11:31:56.000000000","submitted":"2021-08-25 11:30:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"805882","meta_rev_id":"aa32fc69af093823e7207524fa1d992d05a6096f","_number":805882,"virtual_id_number":805882,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 11:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":2,"date":"2021-08-25 09:52:09.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":2,"date":"2021-08-25 09:33:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":1,"date":"2021-08-25 09:52:09.000000000","_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 19:59:34.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-08-24 20:20:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 20:20:55.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-24 22:11:42.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-08-25 00:51:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-08-25 09:33:02.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"}],"messages":[{"id":"18a42bcde1d6b86b448b17e91545b9ab8caf58b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-08-24 19:56:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"420e514b5439194acd110f4b24fa1f3bdc276981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:20:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3362a19cc4464591ba036b6b6259ef5b : SUCCESS in 19m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3466b558c53b452f905170f73d6376bb : SUCCESS in 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5527a4b93cd541d7b666bc4be747b91e : SUCCESS in 8m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c56fbdbc7464711bb6628824af31354 : SUCCESS in 14m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e5063d71c3e4c2a9d698c150c28876e : SUCCESS in 14m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29bc4b73035e4374877794645fc553d9 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d75ba94adc1442abdc3c4465c834c60 : SUCCESS in 11m 10s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/59f806596d8244e38efc3047835cf1c3 : SUCCESS in 5m 47s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/f2ca1419946d43e68505efc825da9d18 : SUCCESS in 18m 57s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab7ad564e16f42a2b5104f01cbeeb2a2 : SUCCESS in 17m 34s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/36b2680a2e4c4fbf9001862d053aa6f9 : SUCCESS in 17m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a478e5b83e6b5ef49c79204ac6678723007be338","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-24 20:20:55.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/82/805882/1/check/EMC_PowerFlex_NOVA/73db0b5/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"a8a732c6a6815245167ddecb33a3bfcc97d65c80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:22:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6826bbbc84a4343bd6bab06f1009d94 : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8aea042ccac945b2b4aae472eded4ec3 : SUCCESS in 25m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"536525149ba6d9ecb57903913f3ed9d56c4b11a2","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-08-24 20:23:20.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA http://publiclogs.emc.com/82/805882/1/check/EMC_PowerFlex_NOVA/1961f41/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"e7ca280526b4f56c7385e9bafcfd4b4422e34960","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-08-24 22:11:42.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/805882/1/iscsi : FAILURE in 2h 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"d6efca0ecdf41d032d4ae8c8de84538939f51803","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-08-25 00:51:43.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-nova http://207.189.188.190/logs/805882/805882/1/check-not-vote/ext-nova-zuul/ffe61f0 : SUCCESS in 54m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b37ebf146039beddebb61b2f0422b84f158814d4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-08-25 09:33:02.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks valid, the test log shows that the PciPassthroughFilter filters out the compute even though the new flavor has no pci request any more.","accounts_in_message":[],"_revision_number":1},{"id":"964bb1cc1372742ed4e3a13eb72f607685b0de90","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-08-25 09:52:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(2 comments)\n\nLGTM I was about to question why this wasn\u0027t in a regression test but I get that it\u0027s sometimes better to group these tests alongside others in the same area. I\u0027d like to see the fix add some asserts about the host we end up on but that\u0027s not critical enough to respin this initial change and can wait.","accounts_in_message":[],"_revision_number":1},{"id":"1fa27c49251724302b4310300f49fbcbca372996","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 09:52:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"666224b021e3ccbb230a44bb29344213f0b9750c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:30:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7099e56beb784867a8651924134df3f6 : SUCCESS in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa33b571e9e342f2ad18e5927ca9b76d : SUCCESS in 10m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff21c6c0d87f48e6b14dbb5d0c15678b : SUCCESS in 19m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0566c75419724acf9fb345f2929415de : SUCCESS in 16m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb90967c4d24881aebf8a525decb351 : SUCCESS in 15m 20s\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/68c15d1f14b043e9b4d07d41cbdad375 : SUCCESS in 18m 58s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6536f04dccfc481a858476393714e986 : SUCCESS in 22m 14s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ee88db8c4b104b87bbcc8cf8f49a4403 : SUCCESS in 12m 12s","accounts_in_message":[],"_revision_number":1},{"id":"9740c08cfc0650b63b975e226a291793aa67920b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:30:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa32fc69af093823e7207524fa1d992d05a6096f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 11:31:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4986b9b3175452096e93435321f0317 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38603a4d09a77882298423a8b20c67b283b99ad0","revisions":{"38603a4d09a77882298423a8b20c67b283b99ad0":{"kind":"REWORK","_number":1,"created":"2021-08-24 19:56:14.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/82/805882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/82/805882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/805882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/805882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/82/805882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/82/805882/1"}}},"commit":{"parents":[{"commit":"25151c3b46ddbb2ce475dcc77e79a9647e763b48","subject":"Merge \"conf: Remove deprecated aliases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/25151c3b46ddbb2ce475dcc77e79a9647e763b48"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-24 19:52:02.000000000","tz":-240},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-08-24 19:56:08.000000000","tz":-240},"subject":"Reproduce bug 1941005","message":"Reproduce bug 1941005\n\nWe don\u0027t update instance.request_spec when we resize to a new flavor.\nIf the flavor\u0027s extra specs change, this is not reflected in\ninstance.request_spec.\n\nThis generally goes unnoticed when *adding* stuff to an instance. The\nscheduler will still find a host based on the old request spec. The\nresource claim on the destination uses the flavor and image data\ndirectly (not the request spec) so the claim will correctly fail if\nthe destination does not have the new things that the instance wants.\n\nWhen *removing* stuff form an instance this becomes problematic. The\nscheduler will attempt to schedule based on the old request spec,\nwhich could end up with NoValidHost if no host exists that has the\n*old* stuff the instance wanted, despite the new flavor not wanting\nany of that stuff.\n\nRelated bug: 1941005\n\nChange-Id: I8c052a6526819834d79d2b5027050f1bdf36638f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38603a4d09a77882298423a8b20c67b283b99ad0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/38603a4d09a77882298423a8b20c67b283b99ad0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
