)]}'
{"id":"openstack%2Fnova~813443","triplet_id":"openstack%2Fnova~master~I5669e20bf8e693df073ce6a2b1c098f9825e13f4","project":"openstack/nova","branch":"master","topic":"bug/1946546","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2021-10-18 15:01:33.000000000","reason":"norman shen replied on the change"},"29071":{"account":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"last_update":"2021-11-08 12:34:25.000000000","reason":"Lee Yarwood replied on the change"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I5669e20bf8e693df073ce6a2b1c098f9825e13f4","subject":"Add a timeout for waiting volume snapshot ready","status":"NEW","created":"2021-10-11 14:26:16.000000000","updated":"2021-11-08 12:34:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":7,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"3fa7611b3c57511c095b7db5a03a661ea282fe22","_number":813443,"virtual_id_number":813443,"owner":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-11 16:01:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"all":[{"value":1,"date":"2021-10-12 07:00:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":-1,"date":"2021-11-08 12:34:25.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"}],"CC":[{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-11 14:38:24.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-11 15:14:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-11 16:01:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-11 16:01:48.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-10-12 07:00:47.000000000","updated_by":{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},"reviewer":{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},"state":"REVIEWER"},{"updated":"2021-10-18 15:01:33.000000000","updated_by":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2021-11-08 12:34:25.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"}],"messages":[{"id":"98ead996e6a49246e3726cfde3fe7d765eff6be1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"date":"2021-10-11 14:26:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa03e01bb8c27e5a2084cdfaeec7181acf42b2fb","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-11 14:38:24.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/43/813443/1/SRIOV-nova-macvtap : FAILURE\n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/43/813443/1/SRIOV-nova-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"bd11abb35c3b3c8b232d43cdc6da190bdd4bf69a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 15:14:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/85c78b5704d54856bddda8de6b096112 : FAILURE in 37m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80174bc2cfee485b90d90246cb885367 : FAILURE in 35m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd3235471c91b8d6320e07c14b5f5e41d88ddb91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 16:01:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/311458c2da314b41ac8782405af8cbe4 : SUCCESS in 1h 25m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47487c96ee7645c8b05376c7e75a5380 : SUCCESS in 15m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0cf302e53ae24701b1cd25ea12538b8f : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40b4908d4b447df9e22270a08cd2474 : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3403f1eb63504a3d81d79184d22111e5 : SUCCESS in 15m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/494980592f144185b88134cd25926b3c : SUCCESS in 11m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1726c16bdc3b4f3a8dc1628bfd883405 : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/925b94b47dd1445f904caf249b37b318 : SUCCESS in 9m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/615be941fd8e434bb36df89e53bc854b : SUCCESS in 1h 22m 59s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/77a13e995ff8494d9f84be59f829da16 : SUCCESS in 1h 10m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aaa8070c031047a3b7fef6ec37851f0b : SUCCESS in 48m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/232cda6c93304f50a388eaebe772cad1 : POST_FAILURE in 45m 26s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/beddcb8ab139490c84f4d9997141d8c5 : SUCCESS in 1h 15m 49s\n- nova-next https://zuul.opendev.org/t/openstack/build/558d5ed04e8f4bd1a1050dc0e6b622af : SUCCESS in 1h 23m 40s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4f9ab2993d1c44e6b049ec2f621ae713 : SUCCESS in 5m 57s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/622c30190adb49a099e623eefa9d2a81 : SUCCESS in 20m 34s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8f4543e3942b4ede8a27c55c67367f09 : SUCCESS in 21m 54s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/89ad071808c149d9b90dfe4e4aebe945 : SUCCESS in 20m 00s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/eb0492ba97c6496d9a2354afab8ccc18 : SUCCESS in 1h 08m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c75a893cbda4a15a9261034382dfeda : SUCCESS in 49m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de0ef99e0a094923ab653e0138681fa4 : SUCCESS in 49m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/53703b26c7074114b36b58d73cda4eb9 : SUCCESS in 29m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/15df69b434544cfa91b2eb1b0e42eaeb : SUCCESS in 35m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a61fc1419a3580f348cd5ecd7905b459abb6482f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2021-10-11 16:01:48.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/43/813443/1/check/EMC_PowerFlex_NOVA/4e2869e/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"3bfb8a97219e2f365fe7706738064c5c5c6e9af3","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2021-10-11 16:19:09.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/43/813443/1/SRIOV-nova-macvtap : FAILURE\n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/43/813443/1/SRIOV-nova-direct : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"b33e12f2a5b3bd1ed6be89d2a8d307f49eeb7081","author":{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"},"date":"2021-10-12 07:00:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fa7611b3c57511c095b7db5a03a661ea282fe22","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-08 12:34:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2da9a20d09af4812a81f4760bf8366662ec50b82","revisions":{"2da9a20d09af4812a81f4760bf8366662ec50b82":{"kind":"REWORK","_number":1,"created":"2021-10-11 14:26:16.000000000","uploader":{"_account_id":29071,"name":"norman shen","email":"yshxxsjt715@gmail.com","username":"ushen"},"ref":"refs/changes/43/813443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/43/813443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/813443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/813443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/43/813443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/43/813443/1"}}},"commit":{"parents":[{"commit":"fdfdba265833d237e22676f9a223ab8ca0fe1e03","subject":"Merge \"Update min supported service version for Yoga\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/fdfdba265833d237e22676f9a223ab8ca0fe1e03"}]}],"author":{"name":"ushen","email":"yshxxsjt715@gmail.com","date":"2021-10-10 03:15:58.000000000","tz":480},"committer":{"name":"shenjiatong","email":"yshxxsjt715@gmail.com","date":"2021-10-11 14:25:59.000000000","tz":480},"subject":"Add a timeout for waiting volume snapshot ready","message":"Add a timeout for waiting volume snapshot ready\n\nEndlessly waiting for a volume snapshot being\nready is not user friendly. This PS adds a timeout\nand quits if a user defined timeout has met.\n\nCloses-bug: #1946546\nChange-Id: I5669e20bf8e693df073ce6a2b1c098f9825e13f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2da9a20d09af4812a81f4760bf8366662ec50b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/2da9a20d09af4812a81f4760bf8366662ec50b82"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30498,"name":"weisongf","email":"songwei-8@163.com","username":"NealSong"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review Priority","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
