)]}'
{"id":"openstack%2Fnova~818515","triplet_id":"openstack%2Fnova~stable%2Fxena~Ie904d1513b5cf76d6d5f6877545e8eb378dd5499","project":"openstack/nova","branch":"stable/xena","topic":"bug/1946729","attention_set":{},"removed_from_attention_set":{"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2021-11-22 20:04:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie904d1513b5cf76d6d5f6877545e8eb378dd5499","subject":"Add a WA flag waiting for vif-plugged event during reboot","status":"MERGED","created":"2021-11-19 08:47:45.000000000","updated":"2021-11-22 20:09:39.000000000","submitted":"2021-11-22 20:04:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818515-bug/1946729","meta_rev_id":"2b1718e0c475373a737bbb4f4fff66fbabd6986c","_number":818515,"virtual_id_number":818515,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-22 20:04:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-11-19 21:01:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-19 10:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2021-11-19 21:01:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}],"REVIEWER":[{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 09:13:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-19 10:01:55.000000000","updated_by":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"reviewer":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"state":"REVIEWER"},{"updated":"2021-11-19 10:34:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 11:00:17.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2021-11-19 21:01:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"fb2ea52999e239a52e95426d202f6f7cae1ac2e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-19 08:47:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3815a0f6ff91e25878ecb0c4c7fc60e96aba29d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 09:13:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5be9be3adedc4488824a2fc037406fc0 : SUCCESS in 18m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e23db59ac2cc447d880c58d487f06321 : SUCCESS in 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"69c3241cee3bfd7e86095da8f3da195b3f9c5a05","author":{"_account_id":7166,"name":"Sylvain Bauza","email":"sbauza@redhat.com","username":"sbauza"},"date":"2021-11-19 10:01:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1eef32095ac1c9b3bcff68fc16fd7e64ac136c53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 10:34:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ef4b3c65b18f43f2a6348511a4da8114 : SUCCESS in 1h 10m 12s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/85bb10c3d9ce4599bd7f0cd0c690f3a4 : SUCCESS in 1h 19m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c8b55ea900487ab817c02b94869e25 : SUCCESS in 14m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/965cf310f6a84473b69aa2d3df5ba035 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7634162da2ac408bba0e0d2f5d29c6d6 : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/438c77fdfee348aeae2b49cc0a8e2f9c : SUCCESS in 11m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc3d1307388a4293a318cdfaa4f4a898 : SUCCESS in 13m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d4eb84f995740808abb6abbc1fab492 : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c62a79ea27ae4c5db2af2a478e683833 : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5153cf320b5740829a298552e0e32529 : SUCCESS in 19m 18s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/6a92e46d55e346a184dd14af29de650e : SUCCESS in 1h 21m 27s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1bda204e7b241fe9ee6ecb56d052663 : SUCCESS in 1h 24m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5783f58e8b1c4477b6b81771d4f37079 : SUCCESS in 46m 33s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/201209ef364c4affb6ff07a34332311e : SUCCESS in 48m 18s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/00f05e1a4a6c4c478f5461d44a41cc6a : SUCCESS in 46m 10s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/f5f3f32f0fff4e52a7b19f2f6f82125c : POST_FAILURE in 1h 27m 51s\n- nova-next https://zuul.opendev.org/t/openstack/build/124a0ce944c0402c9d6e45201abb5b15 : SUCCESS in 1h 37m 49s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b4916496973d4dff8806be7c9b768ab8 : SUCCESS in 6m 17s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/981328aba61f4da0814dc9443774a162 : SUCCESS in 19m 52s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ab7bcde699e47aa8f39e94069803a08 : SUCCESS in 22m 11s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78c02725c22342f884e532c5fb409ef2 : FAILURE in 23m 20s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/15e19f48a84a4aa1a1270d97651876e8 : SUCCESS in 59m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11bb3a645fc643d2b73a6c56e5a31639 : SUCCESS in 47m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12d3c29b03df434b83ff9ed21a0392f8 : SUCCESS in 45m 16s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/dfa72009249b49fd93b3849855d56d9c : SUCCESS in 28m 13s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/403798b338e74295bfc6f9d49a626653 : SUCCESS in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93b953e888c4fbee4e1bc9730e7ddbdbb558c1ab","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-19 11:00:17.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : SUCCESS in 2h 11m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f94b853ec8b028d53688aaf0362e8d4ad3c7fdd4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-19 14:26:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6efa559245766884ddfca44acacc6b5e200a52c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 14:52:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/210a6ed6ac224483923b7ff601342d43 : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/713ee023d8924376ac5c793c54b96041 : SUCCESS in 20m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f862bcda4dce8829f469df493fbabbe8f0ea821d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 15:55:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/61e2dc4504dc445f8cd951f4367080f6 : SUCCESS in 1h 16m 52s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/c9b78d1a194347248cff95de1f5a11bc : SUCCESS in 1h 25m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9550ef856de45538041848aba796fe2 : SUCCESS in 15m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9632c1d91244578ae9bbdbf53082b56 : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd0e47815a114edba48ccbede3ff4b65 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c73a9c12f86b402b903e76c3810e6ab7 : SUCCESS in 13m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faad99e97a6242f7aeacb2e1ebca75d1 : SUCCESS in 11m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cea438ad8254458ad9b137e5d94b7bb : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c91582c60f724dcbbb5167dd3da32d6d : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7434c29298b8468d94b95771041347eb : SUCCESS in 15m 36s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/dc859b59e0cb4911b4220d6bab1da7d0 : SUCCESS in 1h 10m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d33b91cb07e243d083e7818bb9381540 : FAILURE in 12m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e908f87489545659441de30cd1ed51b : SUCCESS in 40m 34s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4fb26c8fd6944e08ad958ff947f3dcc8 : SUCCESS in 43m 54s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/9fd0066f128d45e78234961b11c47368 : SUCCESS in 34m 20s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/e2010983dc8849e49eeb4198047aa066 : SUCCESS in 1h 10m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/83a87bf8fbc7406ebe48955a403797f3 : SUCCESS in 1h 17m 41s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3d2108e4323148efb5295283e0e2de5a : SUCCESS in 6m 23s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/487d942e17bc4f238d97c75ff2c5a0fb : SUCCESS in 21m 58s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8e0ada351cc14a0fa79c84fb9ff30ae2 : SUCCESS in 18m 50s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c495f848f9b24e2dbaf70f47983e3bbe : SUCCESS in 17m 15s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e5d46162dcb043b48dd636c983892db5 : SUCCESS in 56m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0dd59e4f29a4c27b805eb8b571e94eb : SUCCESS in 44m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/667e4c3f917f4c21a48a3084f028c30d : SUCCESS in 38m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/08f20869a88443f2be80d27cfef9ece1 : SUCCESS in 33m 49s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c719aea065174bf5a40b9f3d62ef3d0d : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8cba691acf7b572745c445e7c2b057d9315b4fda","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-19 16:28:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : SUCCESS in 2h 01m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d93ad22e7b51f253d96ef108dfa265736488b62d","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-19 17:15:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91cb07f7ffc9ac430a66da8fccdd6726e2834ae9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 17:50:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7f08e369cf34ccd9b705b8f8e18951d : SUCCESS in 30m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82a6b1234c634599913721e0e95531ca : SUCCESS in 28m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"57f35d1f3d61b0b01729abf01b92bb02be14ab88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 18:59:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/cf7f956d95f245daa6f1873c91002393 : SUCCESS in 55m 47s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/61d77799dfd74c4f8fe671d0db319efb : SUCCESS in 1h 17m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0169974bb550486c8100936a5fde619d : SUCCESS in 14m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ac29a723bd642eeb090fa938b33d7fc : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da585605e4ee48c88b2c3fccf36b6d55 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f8a780b52d64b1598c2a821b6428b44 : SUCCESS in 12m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37e12d0a30664885a974d92e5c564460 : SUCCESS in 13m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9899b5c5f35648f9b9b19d562df6fb4f : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a90fec1b5c047d69d5bcc5c34405774 : SUCCESS in 10m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e39fba8f854997a393cf4a1e574096 : SUCCESS in 19m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/3befe6c3cb7041eca95547236edcf8ff : SUCCESS in 1h 26m 24s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c825fd91e49404588430457cc519f48 : SUCCESS in 1h 19m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/aee944fcd1de49c3b1ceca0d53eee519 : SUCCESS in 46m 48s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/72b6247818d441fe80739edf28b48299 : SUCCESS in 54m 28s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/f8eae274581444e7925b7cf5153c83c5 : SUCCESS in 54m 06s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/19f96aabef7244d1a9fd715fcb8043ae : SUCCESS in 1h 32m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/a3ef1ffea52d4171b7918c934c919f19 : SUCCESS in 1h 14m 42s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/9c9daafc265c40e4b6ca138f1c798944 : SUCCESS in 6m 11s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/1d2b8351a7a2482889f117c63d40fdb5 : SUCCESS in 23m 37s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0a230c70a49c49f094f8fdc7cc9fed65 : SUCCESS in 19m 22s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/70cc7ef000bf40d98ce1192594b89087 : SUCCESS in 21m 22s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5fe18128e492416d9310b5037e788110 : SUCCESS in 1h 11m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2994a45a63b84976a20e9e847d229f80 : SUCCESS in 42m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab2ecba5070462a9b63e844d0786995 : SUCCESS in 48m 21s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/36a1d385dda84a758df861c702836f48 : SUCCESS in 34m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/dfa8a6b6d5bd4e01b2109a0bd0f70a67 : SUCCESS in 38m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc8f1715bb59d628a649eb53e97daa6fd70778c1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-19 19:46:41.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : FAILURE in 2h 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"75e51cf22c4ed743c7ef9f1cd028b820c2c6f743","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-19 21:01:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b7b83217264ff6f9c036aa7d2d496b74a70a89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 21:02:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7009d52d78f9d06778c59f8f31b21844493092e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 22:25:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e1f9569e4208498ebe9e6743f34d9b77 : SUCCESS in 1h 19m 07s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/f991589a4e5e409597e223f32ec17923 : SUCCESS in 1h 14m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a68ffd907674e9fba30a66193faa63f : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c91e7bb3a2349ab81037ed9e8368b40 : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d96d046d92cb4b8fb56d95c51a91f9d3 : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f830ae3c21f4574beaa7c3b942b1d7f : SUCCESS in 13m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59d06bf098004a5c9ebb54b3b229d22b : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b9c2516faf6481b8f6fa4783a42e8a1 : SUCCESS in 20m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/af5fbb4b4e214f6b9296ad01ef6f5c26 : SUCCESS in 52m 18s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c6db57b53fc74832b466fd16e45507e4 : SUCCESS in 48m 01s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/03e1a680ed9741328a9fa14bd747acc2 : SUCCESS in 17m 27s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1aab83da15424129b4bae0438be3a896 : SUCCESS in 22m 01s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/0c715c26d6ac424393902430becb9c9e : SUCCESS in 1h 11m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/5a94b42443824f87918d76af6d80e2d7 : FAILURE in 1h 22m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ee685d39a1a94fe58216ce52bac9bd0a : SUCCESS in 6m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/842a7bae63eb42138f133e7b4e96c13c : SUCCESS in 1h 13m 28s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5047665913f74d678af493ca2ce22abc : SUCCESS in 1h 07m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebd9e0eb50ee49db990b665e6cd642b0 : SUCCESS in 48m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/279a715c880d46e5bb89ecfd8d00db68 : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4bb2ea0c4f9843efe12ea51987726c298b9ce792","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-20 14:47:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2009ff0fa9ef48f9eccd0f760ef85f15517f98a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 15:21:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c065b21ada5a4bfb95e34d7f62d1be04 : SUCCESS in 26m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2080627ad1c04af69b0374f09c3e9d1d : SUCCESS in 31m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c70a895ff910188c377342c4494ce9185fe5d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 16:25:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e4c76e4304d74fd49b725791fca22da6 : SUCCESS in 1h 22m 30s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/809c549adc84490fb9dcf69744bee02a : SUCCESS in 1h 22m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85811470b33a497e97ee11f534513375 : SUCCESS in 16m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b10798eb3cb4696a386b31b61601e80 : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3f4c0831af480b83a77106f62df577 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b5b703b98f44cb7be85bea83698530a : SUCCESS in 13m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/544b72db93144300a43c492bd2ebb14d : SUCCESS in 12m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/084f17e206174feaa6f45d09517e2904 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aff3c8e8dda4df1aa81cc5421a9a8e2 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/074bd079331c478589e2c105d678fc95 : SUCCESS in 24m 04s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/02a680dfae564d0da4882448cccb3812 : SUCCESS in 1h 29m 44s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eaf6546ea2b049e281fd2d2ca16f2512 : SUCCESS in 1h 24m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bd43e04831c8451fa334acf9067cbe1c : SUCCESS in 48m 58s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/25a5945d784b45aab817a8a07010179d : SUCCESS in 55m 38s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/cbfa381dd9d748ba98f22d78568667e5 : SUCCESS in 1h 05m 48s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2c70be7783b345bbb1d1a74a36fba957 : SUCCESS in 1h 08m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/7ea23726a73d4706bd3a621d65585021 : SUCCESS in 1h 00m 28s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/27d32897574943ada72e7ef5b6d583cd : SUCCESS in 7m 41s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/e6e65141917547d7aee8c533feb9edcf : SUCCESS in 18m 07s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0dd9674d7624743b17cef6bee2ca58b : SUCCESS in 15m 00s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3166a4030ce9405b9139d098b1df0dbf : SUCCESS in 16m 33s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c622759488014c88b460d70484024883 : SUCCESS in 1h 02m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c73307f916574732bfcbe044df94483f : SUCCESS in 39m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1766c6d8ba8d483791beff1c84760a77 : SUCCESS in 47m 10s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/53364027302543078a7ab28180e01553 : SUCCESS in 19m 45s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b2a3e5d00e8549bfa71c442542ffc510 : SUCCESS in 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac2f51a332c99edefff15e22ca5edcef22b92684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 16:26:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1639352e276f6ba8af965f9d0cafdb4caef4214f","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-20 16:55:34.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : FAILURE in 2h 06m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7aa4c83ba392b4797cbb2d5776b6cdb30e33fb65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-20 18:14:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/48f3aa0a736a49459b2eb95f917dbc75 : SUCCESS in 1h 07m 58s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9a00e3208f64b2db4a69830bd27ccbd : SUCCESS in 1h 26m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c053dd130ae44b1ca17d39adb4c43f3c : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7be27c80af448a18db2c126ee41ee29 : SUCCESS in 8m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13a5a063a2b74f24b010d2bdf6630b18 : SUCCESS in 13m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61e513839bcd4c7e93a7dcda70299752 : SUCCESS in 8m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66fb24a944cb4ac39286fc60e9d9f108 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce421c7fd348448fbb2ecc6a3386f619 : SUCCESS in 22m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4baae6f10f584faa902d3eccad9eb7ab : SUCCESS in 38m 04s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/2f1e243c83684cf89479e6a5caf60d08 : SUCCESS in 58m 55s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/c31474cdec0846da83e82171939fa4ec : SUCCESS in 19m 25s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/432d01800fed4b5d95a1fe32abff9fed : SUCCESS in 19m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5572e885ba434549bdb49dd14e5df6cf : FAILURE in 1h 19m 55s\n- nova-next https://zuul.opendev.org/t/openstack/build/e102cd0ba07a4deb96a4016f8747c851 : SUCCESS in 1h 41m 34s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b19524e2091e41ca9a6305ceb96dd7e5 : SUCCESS in 5m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7059aea6c1e54f00bb36532805698183 : SUCCESS in 1h 19m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/b98adb904025400d9bfa3378c6fe0398 : SUCCESS in 1h 12m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/038b6af7e5934696a30aee7aec710f39 : SUCCESS in 49m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/637861e1c0b04519849211291d840cbb : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"01886cd017f9b7f285e08237bbc9d96768922de4","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-22 09:42:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad32bf7c9c9eaf5e9d63723cf2534a16e4987e79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 10:18:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e968de079eea4e2b9d904b8524791c4a : SUCCESS in 33m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b7d923f863f45cba19a21e94b9d9b32 : SUCCESS in 25m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e69de2b2cd19244daf6f5e86c419c284a21ed26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 11:04:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/5bef637a88214722b032ac922dfa5d1d : SUCCESS in 1h 06m 07s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb57c82f148f43a29d6c6102f4b15b71 : SUCCESS in 1h 06m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/628c5915c3514d318f676a44bde33783 : SUCCESS in 14m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e80b2bb012074737b10ee7b03b681636 : SUCCESS in 14m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b34d01306be4b35b4dd2906ae9d5ed5 : SUCCESS in 10m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5825e21c8c114cb780b956e2d1459026 : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aad9b0b309e4daea3637619c592dafd : SUCCESS in 12m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cf9f0f219d74baf8bfdfe494a7e94b1 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6d22c266584447b62eee99cb154ff2 : SUCCESS in 10m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08b37423f9b448dc8c205333cbdda6c5 : SUCCESS in 20m 45s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a0e735a1fbf3412ba9ce324700ebbd1f : SUCCESS in 1h 07m 26s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/95e736eed88f44debbb933ef9b592189 : SUCCESS in 1h 07m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d6283ddfd4754378b9fcb32c9f8f221d : SUCCESS in 48m 58s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/ead5d95af2b9466190ff0f09381dacb3 : SUCCESS in 49m 29s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/704e367730894e318e6347770c9e1208 : SUCCESS in 53m 00s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/1b62dfa3d6764b40ab7d72e6417af615 : SUCCESS in 1h 19m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/a2064a7fb64b4c9eac122fe7dfb13fec : SUCCESS in 1h 12m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/aec8a82efdb44292bacc43846f928332 : SUCCESS in 5m 03s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/1223ee9d1a564544bac1f1e3c2a69522 : SUCCESS in 19m 25s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3eb2bff9fa14c58a8ee79a27af7a1c0 : SUCCESS in 22m 06s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c7a167f5e48456faa38d384be681852 : SUCCESS in 19m 55s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/c2f73f37755d488ca61dfb3afd7d1366 : SUCCESS in 59m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ade6cc638c5941caa721646cd6870cc9 : SUCCESS in 44m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8dbc4c39ef8848939ac02e760f645692 : SUCCESS in 42m 46s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b19ba5b98cc54f10babcac3c084f5fd3 : SUCCESS in 26m 02s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ad452917e0f8465fbf093413809e5e65 : SUCCESS in 36m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9794c32bb24befc882fe8584bc5e28fb2d1e45de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 11:05:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"178c0a2613c4cc8206c40a25120540a5644d945a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-22 11:48:48.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : SUCCESS in 2h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4732d5b2cfca01ed0e659516f5a20cfb65a25004","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 12:52:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/ef70ec37de194bef897fa4452f8e9552 : SUCCESS in 1h 11m 13s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/feccc2543f2d499b80839164fbdd32fa : FAILURE in 55m 18s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/464be2dde88e485a800bd7f4e70f0cb3 : SUCCESS in 13m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2018101536a49d88b49d0352e51d488 : SUCCESS in 12m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a73de5681b3495d8fa2ff1e599f64b9 : SUCCESS in 11m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9800bed100204a3d8a92dbef19befe9d : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d215a896da8c408cb0c294ac8a192a52 : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d2356c9cb0343a5ab3e2245945f7596 : SUCCESS in 18m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4cb5d1a3c50542e085d962f8d6c41d0b : SUCCESS in 47m 13s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/128fdf2137f4411aa282696c99e5828c : SUCCESS in 57m 49s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/1acaa7fb167d437daa913b8aa6d54726 : SUCCESS in 17m 00s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/11681f937f654d858741ae001f30f25c : SUCCESS in 17m 09s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/11c84d0b2ed2449c83dc1d1ac0e29b0e : SUCCESS in 1h 17m 35s\n- nova-next https://zuul.opendev.org/t/openstack/build/56723e10026c426ab7bf4af1f1b30a4f : SUCCESS in 1h 22m 17s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cd353549654a4f0fb391af0d250a2cec : SUCCESS in 6m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/35a1f2d60e1747f1b0f3d428313179a7 : SUCCESS in 1h 45m 21s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/f2f15fcc1479475c8e25b94348eaee65 : SUCCESS in 1h 00m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd4c51957c264576aff89c8c6fa84d5e : SUCCESS in 39m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95021daf2cc04d628bdd7e45b9f80ffe : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"13b71c9881b27b99ebfabd0fc524009e6e77b291","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-22 13:09:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34e65c4f4aee50954cc80c8034fa216466405ac9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 13:42:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/981a7b63f30d43f997823624ee8d2889 : SUCCESS in 23m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1142a39ab8da49f5b7b51baa8232e4b3 : SUCCESS in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a0d92f61bbd96a4fc388e4b83b58330226ba5e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 14:51:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/a5559d83367f491eb8345bcddec0ae67 : SUCCESS in 1h 17m 44s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/04b6ccc55ab34dc2a5d4001268784e4b : SUCCESS in 1h 34m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e25f74d4fecf43a19e53378716cdd8e3 : SUCCESS in 18m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e788df2b103a4a459bd1097725c29f54 : SUCCESS in 15m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdf1997c726447349eef491ed1a5bbeb : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16aa51ca0c084fe280e21c03efe1566d : SUCCESS in 14m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26c4ab36751e4992a40a34e08937deb7 : SUCCESS in 13m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7782463d6bc9406a9c6cd1687caa3919 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2d319e684d4e9dbd304e0e018545b9 : SUCCESS in 9m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/513daa6a4bd04139ab54fa1ebc684c4b : SUCCESS in 24m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/f4a025b2832a41168ae0bafcf5992c70 : SUCCESS in 58m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1375e56a89d34898baeb63a1f375db33 : SUCCESS in 1h 05m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/55321927c40343c2a2a5eefdecfb9217 : SUCCESS in 56m 13s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0c309bb0f5b14df6b37d664ea35ca58b : SUCCESS in 37m 41s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/b94e58aecac941ee85d2240cafa52f15 : SUCCESS in 1h 00m 02s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/762acd73b0d24cd8b27a45e3586125ff : SUCCESS in 1h 18m 48s\n- nova-next https://zuul.opendev.org/t/openstack/build/a704f9c4c23c4a75a998d0dc62abf3c9 : SUCCESS in 1h 17m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/c09019b06d124c899c1bfb3dde9aa530 : SUCCESS in 5m 26s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/e2a248c300d74dfca5fd15009968a00c : SUCCESS in 18m 38s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/80abd1c868494bb4a07bd1f8593f278e : SUCCESS in 19m 18s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c700323bf23e4381a1d99e4495cb260e : SUCCESS in 18m 03s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e7c8ea5ba9444eb99152af0402bdb569 : SUCCESS in 1h 12m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a4a7ec29ec4e068e1333930a2692ed : SUCCESS in 40m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f8d3ce1476140dbb33a7135d91602db : SUCCESS in 41m 42s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/f63a69d119a04a13875e4de61333a20d : SUCCESS in 30m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/b491695b38e145e2b949821cf908f16f : SUCCESS in 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ffbe224c09f41e4d87d016ac3308740b94c8247d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 14:53:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"826292ddc3d0b935a189fb11adda8220ea00eae1","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-22 15:23:36.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : FAILURE in 2h 13m 20s","accounts_in_message":[],"_revision_number":1},{"id":"d23d081433e66c1b3faf11c6e6b6ae1b5f3ece72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 16:33:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/36656f5aa0654a4bb2377512f57a238d : SUCCESS in 1h 03m 15s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c54cd95e6f048b09b892a4922331130 : FAILURE in 1h 06m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a67678fa5e2245af840e9c6826e784ca : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/994513e7fffa4c4a8cd11f97a501203f : SUCCESS in 9m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d916c893e5d49fcbaaf389b9d47f699 : SUCCESS in 12m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c5d17e23694fc58c67cb4964cf990d : SUCCESS in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08da8e5c12d644d29ac3e9ba129a826e : SUCCESS in 10m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04eb9c445e35483ebadb25415e663c61 : SUCCESS in 20m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/572ddf0928504e8eac4305a328513dd4 : SUCCESS in 48m 47s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e5db17f698c1451cab3cc927bf64bc1d : SUCCESS in 54m 35s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/c318af29478d4161b66627b997e0480b : SUCCESS in 24m 31s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/700dd0a986b04be488460e40be8fab26 : SUCCESS in 23m 45s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2e153f4aea2647c7b80a35bce5aed118 : SUCCESS in 1h 33m 59s\n- nova-next https://zuul.opendev.org/t/openstack/build/92e300a7fb7b491c902f0a66fb4aba45 : SUCCESS in 1h 26m 03s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/522d2573c97d495b8c0560c310347485 : SUCCESS in 7m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9b7fbaee3dee46b2a1b78b96ada924f2 : SUCCESS in 1h 15m 38s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/477df452cbff461c933c26889e08d129 : SUCCESS in 1h 04m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/774c02cea515457a89d6f04f6ece25ef : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bb8ab9ea0d0424fa65b124e54a0db93 : SUCCESS in 52m 32s","accounts_in_message":[],"_revision_number":1},{"id":"da100e3ae8c27753d20816678f284da444dff2b3","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-22 16:43:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3b006828fc525000fe1b1a2ccbf35cd3b79f81e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 17:16:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1356f0f904b4e7bbcc0779eb65d075a : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8aca3f5ba8fa4b5789354bd533679186 : SUCCESS in 30m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56c5cc741333c243e665ce67fc53095ae067e7ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:23:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/d4936cccfea44a268f9c54df99739dce : SUCCESS in 1h 37m 37s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/81647a084af5461db0e2d5b3f0a7267f : SUCCESS in 1h 22m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67b2b49426ba402bbd20a138cf28a8db : SUCCESS in 15m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dbd7c7134a074b378691e63cac2c138b : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6fe438bfdc421dbd01d8c19665331b : SUCCESS in 10m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbe16fd0bd1740939ce633bd0da75072 : SUCCESS in 12m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e2e7d9c41c4b509c7efa1a90c204d5 : SUCCESS in 14m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0d7ef45463c4965bcf81342387a2d59 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b2d48a5632345f28f698c6a5d0e0df7 : SUCCESS in 10m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04ff7ed0d7c4468285b63f24f74fde25 : SUCCESS in 20m 34s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d1138bfd7edc4558a26d51c6c89cc259 : SUCCESS in 1h 21m 11s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9febc67446844bcc897945a4f6f0e2e5 : SUCCESS in 1h 08m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b12b2fd6314142d987f135f2aca442c0 : SUCCESS in 47m 09s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/c8f8d57189f049839cf106e518ede3e4 : SUCCESS in 51m 12s (non-voting)\n- nova-lvm https://zuul.opendev.org/t/openstack/build/1d592720873045be8f9db7fdf3cfa0d6 : SUCCESS in 53m 47s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9d044a658d0044d8b4f88dc079c2f996 : SUCCESS in 1h 25m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/f636fb749520419690589d211da67c82 : SUCCESS in 1h 12m 54s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3a71867059af45d88536272aa0bf3638 : SUCCESS in 6m 46s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/441559ca755f4464b66e699e3a1b5dbc : SUCCESS in 19m 17s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d20fc5078d2b4159a1e39c347f98375a : SUCCESS in 18m 01s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7716627747e5474695b2ad27640cc1cd : SUCCESS in 18m 28s (non-voting)\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/51bacc7e9f354e489fb6685785e66502 : SUCCESS in 1h 17m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd557fe9f33a4adf8f61c64e1f241ffe : SUCCESS in 44m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/919e8e242d6d4e8a83accdb9c478304a : SUCCESS in 42m 29s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/c88d72b3abac4a8784e36e087d678a76 : SUCCESS in 30m 32s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/0a8213813c3e422ba9aca789d328e4b2 : SUCCESS in 37m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"402d89706585d9311f6d4ef1f804d17ea7f9b14c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 18:26:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8f9c0cde1364021d2b5692e0e41d0f597eb1a81d","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2021-11-22 19:23:11.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/818515/1/iscsi : FAILURE in 2h 39m 10s","accounts_in_message":[],"_revision_number":1},{"id":"17f446bf71b7223db445f1f6c32c96a00adec5b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 20:04:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/453a6eb677a44357b8673c66e65aa2f2 : SUCCESS in 1h 32m 24s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/d27b03bb94184943ba9b37fc00770172 : SUCCESS in 1h 18m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43d597720cf24f81a32f8b7c1efb312e : SUCCESS in 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a98c6b4eb0444aae3545674edbce73 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21cb241e985040209045338bcfddd939 : SUCCESS in 14m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8451a4188bf9445597a3d66268d82350 : SUCCESS in 12m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0afe10a7a74a425a99709ec8f36712b0 : SUCCESS in 9m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202ee1f672174c0ca220aa7e98c745a4 : SUCCESS in 20m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b69021ae5f564291aa308f1e73d94232 : SUCCESS in 51m 44s (non-voting)\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/4fc4b342446346629387e6ac998ac1f9 : SUCCESS in 48m 37s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/30d7ac7d14e84e71b196f23796ab462c : SUCCESS in 19m 52s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9c02c9b26667447c8ba6ce1139d66306 : SUCCESS in 18m 38s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/52d53220a68442babe416f380b9d11ba : SUCCESS in 1h 22m 11s\n- nova-next https://zuul.opendev.org/t/openstack/build/3ae18862b5bd48ff9a26508a55dc4685 : SUCCESS in 1h 20m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/61d3088210ac420caad1da1006829222 : SUCCESS in 6m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd38c9df297e4bc7a2c0744507f3743a : SUCCESS in 1h 12m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/429acb8196654e9aa3e31c8c44736c97 : SUCCESS in 1h 13m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8a53021aabb4eae8dadf8669e4d080b : SUCCESS in 47m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea189386313843ddbcc57d29032cfdfe : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b9fb7028181087e057793c166959e8dc7af8d671","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 20:04:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2b1718e0c475373a737bbb4f4fff66fbabd6986c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 20:09:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cd6b366dbb144418e6e5b387012ccdc : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82afe90b1c7b411fafaaae877a4dbf88 : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c41bfb8c5c60f1cc930ae432e6be460ee2e97ac","revisions":{"0c41bfb8c5c60f1cc930ae432e6be460ee2e97ac":{"kind":"REWORK","_number":1,"created":"2021-11-19 08:47:45.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/15/818515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/15/818515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/818515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/818515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/15/818515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/15/818515/1"}}},"commit":{"parents":[{"commit":"64f5c1cfb0e7223603c06e22a204716919d05294","subject":"Define new functional test tox env for placement gate to run","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/64f5c1cfb0e7223603c06e22a204716919d05294"}]}],"author":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-10-11 12:41:37.000000000","tz":120},"committer":{"name":"Balazs Gibizer","email":"balazs.gibizer@est.tech","date":"2021-11-19 08:41:33.000000000","tz":60},"subject":"Add a WA flag waiting for vif-plugged event during reboot","message":"Add a WA flag waiting for vif-plugged event during reboot\n\nThe libvirt driver power on and hard reboot destroys the domain first\nand unplugs the vifs then recreate the domain and replug the vifs.\nHowever nova does not wait for the network-vif-plugged event before\nunpause the domain. This can cause that the domain starts running and\nrequesting IP via DHCP before the networking backend finished plugging\nthe vifs.\n\nSo this patch adds a workaround config option to nova to wait for\nnetwork-vif-plugged events during hard reboot the same way as nova waits\nfor this event during new instance spawn.\n\nThis logic cannot be enabled unconditionally as not all neutron\nnetworking backend sending plug time events to wait for. Also the logic\nneeds to be vnic_type dependent as ml2/ovs and the in tree sriov backend\noften deployed together on the same compute. While ml2/ovs sends plug\ntime event the sriov backend does not send it reliably. So the\nconfiguration is not just a boolean flag but a list of vnic_types\ninstead. This way the waiting for the plug time event for a vif that is\nhandled by ml2/ovs is possible while the instance has other vifs handled\nby the sriov backend where no event can be expected.\n\nChange-Id: Ie904d1513b5cf76d6d5f6877545e8eb378dd5499\nCloses-Bug: #1946729\n(cherry picked from commit 68c970ea9915a95f9828239006559b84e4ba2581)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c41bfb8c5c60f1cc930ae432e6be460ee2e97ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0c41bfb8c5c60f1cc930ae432e6be460ee2e97ac"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
