)]}'
{"id":"openstack%2Fnova~824458","triplet_id":"openstack%2Fnova~master~Ifbc45baf7a52fcf0ce564b55c556ca56ca9e0ee7","project":"openstack/nova","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-04-29 12:08:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifbc45baf7a52fcf0ce564b55c556ca56ca9e0ee7","subject":"List auth plugin parameters for [keystone] section","status":"MERGED","created":"2022-01-12 16:32:29.000000000","updated":"2022-04-29 12:10:48.000000000","submitted":"2022-04-29 12:08:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824458","meta_rev_id":"6c44d3ca6994124a56e5457243962d2381dc7cb3","_number":824458,"virtual_id_number":824458,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-29 12:08:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2022-04-29 10:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2022-02-21 18:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2022-04-29 10:25:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-12 16:32:37.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 16:32:38.000000000","updated_by":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"reviewer":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"state":"CC"},{"updated":"2022-01-12 17:07:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 18:52:40.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2022-01-12 18:56:32.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-12 21:24:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 04:51:45.000000000","updated_by":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-02-21 18:54:47.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2022-04-29 10:25:38.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"e76880a8ddb7c6cde78eb3e3274f7eef6235e59f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-12 16:32:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d09ab13fa8071ba1ab0a5ae9b0459048a3e19b8","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-12 16:32:37.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41218/","accounts_in_message":[],"_revision_number":1},{"id":"7fa351cd06fc7f44e6d7f433eb24f893a1620ddf","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-12 16:32:38.000000000","message":"Patch Set 1:\n\nBuild started","accounts_in_message":[],"_revision_number":1},{"id":"a2efd02fd22f1133ffcb25b6870950ea4cd43117","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-12 16:34:27.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41218/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"505344dbb952a81ca6d635fff28d2d419bb6029a","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":34420,"name":"openstack lightbits","display_name":"Lightbits CI","email":"openstack@lightbitslabs.com","username":"openstack_lightbits"},"date":"2022-01-12 16:39:59.000000000","message":"Patch Set 1:\n\nBuild SUCCESS. log path: https://raw.githubusercontent.com/LightBitsLabs/logs/main/12_01_2022_16:39:57","accounts_in_message":[],"_revision_number":1},{"id":"1e881272d4874d856f860574f80f2a69a320d9df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 17:07:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43257b19be9b4582b492f5d81af9f677 : SUCCESS in 24m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6828f5d5d1be4480b08a93dea524ae10 : SUCCESS in 19m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d35b8e366d8d5ad309e3564920dbd4570ee2d3f0","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-12 18:52:40.000000000","message":"Patch Set 1:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824458/1/iscsi : SUCCESS in 2h 19m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a70161364db03699688adec91d742b6f4fd16ff3","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-12 18:56:32.000000000","message":"Patch Set 1:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/58/824458/1/check/tempest-dsvm-full-focal-py3/b933e91/ : SUCCESS in 2h 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c750f9d3595c8119c0f8380bc5b1ef6a0a216e92","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-12 19:30:57.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/824458/824458/1/check-not-vote/ext-nova-zuul/27b9a5c : FAILURE in 1h 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"19315e766a0722ab03bb71ffdf4543138b1cacee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 21:24:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/f12acfac663740e1a2a1f3dd7b72e6b2 : SUCCESS in 1h 22m 57s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3f6529a00d6441790e600621a45342c : SUCCESS in 1h 40m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a661529b73e1440b96d9d5a1c556afa6 : SUCCESS in 52m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc8b8198f0a1493a97c3931f6844ceee : SUCCESS in 16m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6fea6ae75d8344d381fa288f8394213a : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2893b599a23b46b792e001ae83e21ae0 : SUCCESS in 13m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a28d8d2136c9435f9eaf7bee12d08fb1 : SUCCESS in 13m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/007be2f92a624e1bb8a8b7cf8da6269e : SUCCESS in 13m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7f430b753341bb97b2661be2931e34 : SUCCESS in 11m 12s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a65dd214fc4842f6a95110d6ba6f9883 : SUCCESS in 1h 23m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4649b8794f9b4cec972ee477cc3c8bb7 : SUCCESS in 1h 13m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b3790f9752b24b768372a4b22e6ff366 : SUCCESS in 54m 48s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/531a55d8213c445ba0f9507d6f06c0b2 : SUCCESS in 46m 24s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fc7bcc802444495e88847db9b6c563f7 : SUCCESS in 1h 42m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/e14269194fb14023bcce6ec7ae8cccd8 : SUCCESS in 2h 00m 32s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f144aa04df864151801fa664a6edced8 : SUCCESS in 9m 14s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/6bbed567e5f147caa11f9990c38811c1 : SUCCESS in 20m 42s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97a2baad4bc249dbadd42207e42e91b5 : SUCCESS in 24m 57s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f94048d6ac54b489f91ed0cd22ab9a7 : SUCCESS in 25m 25s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/77e9c3bce7df4ffd9923c0b49a8617b0 : SUCCESS in 1h 12m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c26dcd1da9e443149ff106a1dfea84d3 : SUCCESS in 51m 05s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/83ae4680c4454d3e8fe8727beb6e944b : SUCCESS in 29m 47s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e7396f7d2c4c473f8fb60a3f3242c01d : SUCCESS in 39m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ac291fe70b66e2944ca5f5a1a9f8f23166e75e20","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9732,"name":"Mellanox CI","email":"mlnx-openstack-ci@dev.mellanox.co.il","username":"mellanox","tags":["SERVICE_USER"]},"date":"2022-01-13 04:51:45.000000000","message":"Patch Set 1:\n\nBuild Failed \n\n* SRIOV-nova-direct-cloud http://13.74.249.42/refs/changes/58/824458/1/SRIOV-nova-direct-cloud : FAILURE\n\n* SRIOV-nova-direct http://13.74.249.42/refs/changes/58/824458/1/SRIOV-nova-direct : SUCCESS\n\n* SRIOV-nova-macvtap http://13.74.249.42/refs/changes/58/824458/1/SRIOV-nova-macvtap : SUCCESS","accounts_in_message":[],"_revision_number":1},{"id":"d16b2c5bd5d8cb71cca2aef89f237d2878923184","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-21 18:54:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb745eae731ae8975e23e8ddedeee8547e4f38f3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-02-21 18:54:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"db0df6ae1cccafa66498c961647ee28bffe6f31c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-04-29 10:25:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbe7c813cd9d735fda84708037499d35e0fb4d05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 10:26:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"351ce93f5571268c1621a24a1e80b2598c1f7111","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:08:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/895dc390418948be8aaa4f753603fd5f : SUCCESS in 1h 23m 43s\n- tempest-integrated-compute-centos-9-stream https://zuul.opendev.org/t/openstack/build/2388946581604eb3a97da92436ad053f : SUCCESS in 1h 33m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80e5d13425e7462f988dd4083328dac7 : SUCCESS in 54m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/54d28c3b3c344eb9b46b5be53c6aafe0 : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b976469bfb6440f9b82eacea7910697 : SUCCESS in 14m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eccd195d21c94bada5f999cfcbad2c77 : SUCCESS in 14m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e576e0235a640df980d54ae6ca12d55 : SUCCESS in 14m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41c544869c094de1b50dd4cdfff17f03 : SUCCESS in 10m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ce5e99dadff944238a455983a22ffebe : SUCCESS in 52m 36s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7cad5a7f0e3945ccba3737ff5208c846 : SUCCESS in 54m 12s (non-voting)\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3ead83d7a36845008f64309fe31f6269 : SUCCESS in 20m 58s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2109390f937345768deae3407b94557d : SUCCESS in 21m 24s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c1436396bc124933807092d47aaa20ad : SUCCESS in 1h 20m 22s\n- nova-next https://zuul.opendev.org/t/openstack/build/aff622f25cf74eb881699af1bae47a98 : SUCCESS in 1h 37m 33s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/293406547e194ee28bdce15d161384fa : SUCCESS in 7m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dca676273b9b4871838cbceae04d1da4 : SUCCESS in 58m 07s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/392baee208e1477fa589be2fd38ebd62 : SUCCESS in 1h 04m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61d6c9e3788f4eea811afec3a0fa956c : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":1},{"id":"12c8dc0e4fd197c92d573be149510f825713d585","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:08:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c44d3ca6994124a56e5457243962d2381dc7cb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-29 12:10:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4e051df10a141a6b45402165fb8f281 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d021194753b03ee5cf58cdc88c8029419701928","revisions":{"8d021194753b03ee5cf58cdc88c8029419701928":{"kind":"REWORK","_number":1,"created":"2022-01-12 16:32:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/824458/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/58/824458/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/824458/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/824458/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/58/824458/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/58/824458/1"}}},"commit":{"parents":[{"commit":"8c714c76ae07ff6dc7acb6f44ecbd4c086354d2f","subject":"Merge \"Enable min pps tempest testing in nova-next\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8c714c76ae07ff6dc7acb6f44ecbd4c086354d2f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 16:30:43.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-12 16:30:43.000000000","tz":540},"subject":"List auth plugin parameters for [keystone] section","message":"List auth plugin parameters for [keystone] section\n\nThis change ensures parameters of auth plugins are included in\nthe config file generated by oslo-config-generator, so that users can\nmore easily understand the parameter they need to set.\n\nFollowing the existing implementation for the other services like\nneutron, the only password plugins are covered.\n\nChange-Id: Ifbc45baf7a52fcf0ce564b55c556ca56ca9e0ee7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d021194753b03ee5cf58cdc88c8029419701928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/8d021194753b03ee5cf58cdc88c8029419701928"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
