)]}'
{"id":"openstack%2Fnova~824833","triplet_id":"openstack%2Fnova~master~I83a128a260acdd8bf78fede566af6881b8b82a9c","project":"openstack/nova","branch":"master","topic":"2021-09-10-off-path-net-backends-dep","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-02-08 12:43:30.000000000","reason":"sean mooney replied on the change"},"13686":{"account":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"last_update":"2022-01-27 10:07:47.000000000","reason":"Frode Nordahl replied on the change"},"24824":{"account":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"last_update":"2022-02-09 16:27:30.000000000","reason":"Dmitrii Shcherbakov replied on the change"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-02-08 12:12:34.000000000","reason":"gibi replied on the change"}},"hashtags":[],"change_id":"I83a128a260acdd8bf78fede566af6881b8b82a9c","subject":"[yoga] Include pf mac and vf num in port updates","status":"MERGED","created":"2022-01-16 20:56:26.000000000","updated":"2022-02-09 21:03:31.000000000","submitted":"2022-02-09 20:59:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":76,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824833-2021-09-10-off-path-net-backends-dep","meta_rev_id":"2219f6a49d03fa5573f9fe1160dbabca936e0abd","_number":824833,"virtual_id_number":824833,"owner":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-09 20:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2022-02-09 10:36:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":2,"date":"2022-02-08 12:43:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-08 12:12:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-09 16:13:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{" 0":"Default Priority","+1":"Contributor Review Promise","+2":"Core Review Promise"},"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"}],"REVIEWER":[{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-16 20:56:48.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-16 21:25:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-16 22:01:10.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-16 23:00:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-16 23:00:38.000000000","updated_by":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-01-16 23:47:27.000000000","updated_by":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"reviewer":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"state":"CC"},{"updated":"2022-01-17 09:29:08.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"state":"CC"},{"updated":"2022-01-17 09:29:08.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2022-01-17 09:29:08.000000000","updated_by":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-02-09 10:36:03.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"}],"messages":[{"id":"e0d0f15e517d2187acd1b26d0f90f2074da10667","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-16 20:56:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33c41c3705e2b45d0ed1e583a19233cf518f6c7b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-16 20:56:48.000000000","message":"Patch Set 1:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41278/","accounts_in_message":[],"_revision_number":1},{"id":"3a8feabd0ad721a6d75fd7c27b0f74864af301e6","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-16 20:58:27.000000000","message":"Patch Set 1:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41278/ : FAILURE","accounts_in_message":[],"_revision_number":1},{"id":"7257a1e6d3acc555af9888897c78634c51ad6ad6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-16 21:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23f71f5ab7bb4e83990e1cb8acc2801f : SUCCESS in 26m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e5870415ae4408ab7dcabfe02e2443c : SUCCESS in 17m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cfe70e291a7f2ebfc59ad70837cbd19adf858084","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-16 22:01:10.000000000","message":"Patch Set 1:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/1/check/tempest-dsvm-full-focal-py3/8a0ab2b/ : FAILURE in 55m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9afccd6ac52729722dbd50237d65f80f79743cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-16 23:00:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4da18b41256f4866a9650fbbf39f6bab : SUCCESS in 1h 12m 33s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/2b750632a05649e3b08329e3c55be805 : FAILURE in 1h 08m 46s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc9e5026320e430bb5ed8c5e113ef9b5 : SUCCESS in 46m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bff0f7bfc0fc446784da87b26aaf4ad4 : SUCCESS in 19m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57a2b3e993f04c9ebac3b974af099a9f : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0532fcecf2c422396912282829ac99a : SUCCESS in 11m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/293da714a8d24ac08d8ba4bd4fef6279 : SUCCESS in 14m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc1f89c2e10745e7a4a51bf5387170b6 : SUCCESS in 14m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f53ff93bdbb4ed98aa86c0928cf3ffe : SUCCESS in 12m 42s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8a5052d2a1ba4fd784be51cfabc525c8 : SUCCESS in 1h 08m 01s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/555577188c77498a8bae51ac0ecb36df : SUCCESS in 1h 15m 56s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/3254950a22d34d1abbc96d29c42c53a0 : SUCCESS in 52m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7b25d83877a740d28b97a0ed3b948b97 : SUCCESS in 44m 41s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7112806a541c4f43859c09a8b19e2388 : SUCCESS in 49m 42s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/9917ce7f28ed4c37a8770b342a7b7ea3 : SUCCESS in 1h 56m 08s\n- nova-next https://zuul.opendev.org/t/openstack/build/9b47e1e3ccec4a33acd177f35fcfa350 : SUCCESS in 1h 15m 21s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f88ebb62db944632b96b8f7d252c8384 : SUCCESS in 6m 53s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/e45d5d59dbfa40bf917d7a05a49200cd : SUCCESS in 18m 56s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32f4e72cbcdf46acbb6a71f1c27b1d9b : SUCCESS in 25m 15s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/945d0253e5484076a002030cfccda6c6 : SUCCESS in 25m 11s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/034a02d56e4f408bb8a9e4f056cb46f7 : SUCCESS in 1h 21m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab6c80232c6c46af9d5deecf0982296f : SUCCESS in 45m 33s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/9ecd9aea68e44fc99045d1c2701f6969 : SUCCESS in 26m 26s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f101600bba39440cbada261687b5ffaf : SUCCESS in 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"beb9650ab2ee775b775774c019e2f29b01006b5f","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-16 23:00:38.000000000","message":"Patch Set 1:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/33/824833/1/check/EMC_PowerFlex_NOVA/ba57110/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":1},{"id":"bd2f8cdb23f7154cfd8b11efa2e16a0175cd4488","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-16 23:47:27.000000000","message":"Patch Set 1:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/1/iscsi : FAILURE in 2h 46m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d607ebd330cabab7318418de99245e9c7901c828","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-22 15:58:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d6e7af9424217d2ec5d65c2106816d1efd76785","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-01-24 11:51:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e0e3e4e1f78c46d4179537e66bfd5cfe0c4921d5","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-24 22:52:58.000000000","message":"Patch Set 1:\n\n(13 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b0323f099dcc3aa1f28c125963aa0841620cd962","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-25 06:37:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f6b52ca3c99ff9b30f6099ef01154c25cce5930e","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-25 15:13:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cad414e107af67b99f5b30d176c2f040ca61705d","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-26 11:23:49.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42a3eff85681355231522eeb6303e2768e8b578f","author":{"_account_id":13686,"name":"Frode Nordahl","email":"fnordahl@ubuntu.com","username":"fnordahl"},"date":"2022-01-27 10:07:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0e7977ef79ed18b343228bd84b326a165f1581ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-27 11:14:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93962860b75214df436ad8709c46c74d501a6d2a","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-27 11:30:12.000000000","message":"Patch Set 2:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/2/iscsi : FAILURE in 14m 06s","accounts_in_message":[],"_revision_number":2},{"id":"d9935f8b7169e6760b478f5252621c3b7bdab8eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 11:46:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a83ed54aaf714865adf70c4623cd5a07 : FAILURE in 23m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb7b70a695804b3685917029c80b045a : FAILURE in 21m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"438b4c2010058b51478a3d568f1ac944981464a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:20:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/9d738d7b01bd45a2874280461d2a3430 : SUCCESS in 1h 07m 01s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/25e2e02f0c0b4d75a3238ca3f643914c : SUCCESS in 1h 17m 04s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8df7d53c8cf464ca2d2eb0ffab11cc6 : SUCCESS in 39m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba4698d407144c195bc41c2019c5a46 : FAILURE in 13m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2512565166f340fea8b0f720d7a37874 : FAILURE in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a41406ef76f049a6847e6a71d1e1cb26 : SUCCESS in 12m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0015096018a43e8b95b1266c6d9c74a : FAILURE in 14m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/663c92f881574a268680b9f84e3afdaa : FAILURE in 13m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251b08e5a9b345c39dacc1a7f0c62256 : SUCCESS in 9m 31s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/a1a149ecda5b49959b456574595bc185 : SUCCESS in 1h 02m 22s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ec7344b9c61403987b4c78599985466 : FAILURE in 21m 21s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/2d34d8e94f5047d286e365b510241d37 : SUCCESS in 59m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5fd399208c784917a210beda303e6855 : SUCCESS in 56m 31s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/b427e557e66045c0849410c5730faffc : SUCCESS in 51m 22s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/2d26586cafcc40bfb63f4659dc6b3cac : SUCCESS in 1h 32m 12s\n- nova-next https://zuul.opendev.org/t/openstack/build/901dd3ed04be476390927e5589a85c6b : SUCCESS in 1h 57m 09s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/cc1fcad739f747d0b580bec2680d01c4 : SUCCESS in 6m 27s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/14a4da607a3c4317a821633f3fc87072 : SUCCESS in 22m 28s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6695ca5c358d4676bc136ef7c55e7260 : SUCCESS in 20m 15s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc88db1bc0fa4d09bb66277c82cb860b : SUCCESS in 21m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/2db559a5740d4d64ae32225095b548e9 : SUCCESS in 1h 10m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e835c300f024141b4dbb85f3b2115ae : SUCCESS in 56m 41s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/d564bc05915c4a1e9304ac68a37cfa43 : SUCCESS in 32m 52s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/11b25c3dbddd4944a64bfa3c4151b13e : SUCCESS in 43m 03s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/4ef680aa4800441b806a84f349eaf638 : SUCCESS in 1h 41m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"836bcca69f903470d48c9b7f2643b576cade6b44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-27 13:20:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3127731aa9dc7a44f9ca31d7ce9372c8c27164c","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-27 13:35:40.000000000","message":"Patch Set 3:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/3/iscsi : FAILURE in 13m 58s","accounts_in_message":[],"_revision_number":3},{"id":"91c1a23ebc79351515a9cb3c3b8d4529128cbe0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 13:49:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/046f1f5156d6484facc35dee8c641d5f : SUCCESS in 20m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3572c86def64b0d8db1d85b3bc13bf7 : SUCCESS in 26m 52s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5f5fd50fd7a0216b8fea192b7064f71d78fc00d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 15:05:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/93c191e1eeef4180a4adbae3bac78fca : SUCCESS in 1h 24m 08s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/889729d0e5014bb68bbdf941a7d6b2d9 : SUCCESS in 1h 40m 34s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42139fd9a75e44c193eeb11c8847ba8b : SUCCESS in 49m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfcdac83f1cd467d910f5d8ae7f14f9f : SUCCESS in 16m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f1d5f621c9f4cd3bda63e4c3ad4d8e5 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af153c4748c54c9ebde7b386f6e3213d : SUCCESS in 14m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/850199dbfafe46ffa53987816d1fb1c3 : SUCCESS in 12m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daf00a97b54f4e349ef609240a956347 : SUCCESS in 13m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c890a29b0ab54853b4e29c93afe08522 : SUCCESS in 9m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8c4258c04b854a79be7375d47aeb240b : SUCCESS in 1h 22m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b46a663fbb54c19bf6c9cba04c1624c : SUCCESS in 1h 18m 54s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/aba7c978ffc94f60b50837269176a87d : SUCCESS in 58m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bbdee5cfa5564f20a53989af8df9b043 : SUCCESS in 47m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/933aa6c9263d45edabc7770dfc973846 : SUCCESS in 48m 58s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ebd0ecef964d49659cdd08c43e4f1006 : SUCCESS in 1h 32m 56s\n- nova-next https://zuul.opendev.org/t/openstack/build/6225dfb79f7a4701983f98976699b7d4 : SUCCESS in 1h 07m 11s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/42f0d3a78812480384c4062c7a5e5e6f : SUCCESS in 6m 02s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/6a53e67675e74ef981d3c0574ac0d494 : SUCCESS in 22m 21s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0d9e3bf5ffdf465a941d1e43f80769e7 : SUCCESS in 18m 55s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0caaa21434c84be4a83362b0fb8465a5 : SUCCESS in 20m 55s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/9956a456b66a451a8a1594d60b524721 : SUCCESS in 55m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e724b37b8df4fa09706e69e69d195ad : SUCCESS in 51m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/e2d944dd3f5b43a69241dbb32a612a5a : SUCCESS in 31m 31s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ef2e75085c454f54b1f3a802bcfb5a2f : SUCCESS in 28m 31s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/4030ba6c85f84362958cb22799ff6a7e : FAILURE in 1h 24m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2b88423f3725ba5d06d791e1883a5c80e3bbb582","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-27 15:34:02.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/824833/824833/3/check-not-vote/ext-nova-zuul/0eb44db : FAILURE in 45m 44s","accounts_in_message":[],"_revision_number":3},{"id":"e41441dbea4c2f417ee9a7357a321903b82c0786","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-27 15:38:14.000000000","message":"Patch Set 3:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/3/check/tempest-dsvm-full-focal-py3/f3d8583/ : SUCCESS in 2h 12m 19s","accounts_in_message":[],"_revision_number":3},{"id":"ad3e459b1610608a7376aa763074dcd8916a59ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-27 21:25:49.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"0d7c2375e8ad2132107afe1574167b64f23b8f13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 23:00:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c74e280ecdb41fd88edc6713e518b8e : SUCCESS in 28m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4791202d82194891af48e1607d381131 : SUCCESS in 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"77bd7991c020aa9cbd4e3c582a3f715d2cba6c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 23:12:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3b9433d8b7c24c9eaab42976f120e3d3 : SUCCESS in 1h 27m 44s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/b664839046e94e45ba50a269d46c058a : SUCCESS in 1h 23m 35s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d0011ee670e47ca8d093b99fd512e40 : SUCCESS in 46m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5b0575d6ac049668a0040b357de3c1c : SUCCESS in 15m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab01f04ce51a41b2822458beff792e24 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6abff2698bf7415794e015bad8bf263d : SUCCESS in 12m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cb81f6a8e7544549634c87e8d21a162 : SUCCESS in 15m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46c6485369fc410e988d6a99fabd7160 : SUCCESS in 13m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31bb269a05194807a6a8aa768e2e20f6 : SUCCESS in 12m 17s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7746425d01104cbeb2d2cb992c860a8b : SUCCESS in 1h 17m 50s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/02b6bb5d44b94f70ae790ac48a1e98fe : SUCCESS in 1h 33m 21s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/9a2ecc5ec0aa4bf38f09ef5f9ec05de7 : SUCCESS in 55m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dbd31a4fafc64714a2f911c5bf705e88 : SUCCESS in 44m 32s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/0d42947c3bdb49a59bcec134ceb12e1e : SUCCESS in 48m 30s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/6d79aef449fc41b39945d544af105328 : SUCCESS in 1h 23m 47s\n- nova-next https://zuul.opendev.org/t/openstack/build/94ef6347516b43918fbc61cecacc0e23 : SUCCESS in 1h 22m 30s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b32ccded86244eeb998cd899ef68b281 : SUCCESS in 6m 50s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/50b75712d5b6493ea0cb063ec3ce654d : SUCCESS in 22m 15s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7cd117ada1064ccd8ececcc052f5269b : SUCCESS in 20m 33s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efd9fbf15460485699d3f998958d0a9d : SUCCESS in 20m 36s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/6826602d36b342da95eb8552bf31322f : SUCCESS in 1h 06m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/470f1d9f955247258b42ff35ae770ac2 : SUCCESS in 59m 39s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/396b5fe28ee84a51b028b662404d16c7 : SUCCESS in 27m 40s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/5a2bf6cf496340dea8fd9a3795261b38 : SUCCESS in 40m 51s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/89f773ad76034d758679edf744b744e9 : SUCCESS in 1h 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5aa1027de4516ef0b924de7d01e2f6f650eda7ae","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-27 23:13:16.000000000","message":"Patch Set 4:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/33/824833/4/check/EMC_PowerFlex_NOVA/742f8b9/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":4},{"id":"c215f7a286e4585c649cc4d5e2cc95878b2d7cdc","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 01:52:08.000000000","message":"Patch Set 2:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":2},{"id":"17e75ede790ac2825000523658a9789d0f7525f9","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 02:21:28.000000000","message":"Patch Set 3:\n\nNo Builds Executed","accounts_in_message":[],"_revision_number":3},{"id":"633e0cb327b9093fff046eb1505d9311ba960aca","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-28 02:21:43.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":3},{"id":"60553e95d8d0f86c766803aa99e4f22c793b7578","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-28 05:33:18.000000000","message":"Patch Set 4:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/4/iscsi : SUCCESS in 2h 38m 13s","accounts_in_message":[],"_revision_number":4},{"id":"bf4f03bfb647641f5b1b5b42d85843f9ae32ab5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-28 18:16:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2dae2d9eb6b407812a5d00d5151d49bebe6eed2b","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 18:16:34.000000000","message":"Patch Set 5:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41414/","accounts_in_message":[],"_revision_number":5},{"id":"152adf2e0aa497fe5ff5349b76bd2f92ebb702c2","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-28 18:16:48.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b7db48ca7e45eb990958b8840d6e2f55b87efea5","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 18:17:56.000000000","message":"Patch Set 5:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41414/ : FAILURE","accounts_in_message":[],"_revision_number":5},{"id":"d38e058739a635733aa784cdf351395208956428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 19:29:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c7b7c72d4094577839e2a1dcfb5c0c3 : SUCCESS in 22m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d597c94e0d144738bc59c6ca186188f0 : SUCCESS in 27m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c676132a982b0dec14e0371bc9efdab028f97331","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-01-28 19:41:23.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-nova http://207.189.188.190/logs/824833/824833/5/check-not-vote/ext-nova-zuul/fb0d51a : FAILURE in 46m 41s","accounts_in_message":[],"_revision_number":5},{"id":"f0b8ef83fef6d213dcf4c95e2b5500bfee84a8c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:15:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/e6250bfde7784e2887170a29a2e0e049 : SUCCESS in 1h 07m 37s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/260e3394783f4ec2a4f6b280e45a6347 : FAILURE in 1h 13m 06s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c2f00b0128a440f7b8beabbe3efe66e1 : SUCCESS in 47m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4c926743b5244f2bde7f3f5e2254285 : SUCCESS in 21m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ec115d337454a38a9764e8665a1c870 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a23fa8f2b75d49189a81e15598158916 : SUCCESS in 13m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b47c7b52dba4effaac098a7576c8bee : SUCCESS in 14m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa1bd8d306a24e8aade0e4aa2e2c87bc : SUCCESS in 14m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5302b849930643f69e154e47c07df394 : SUCCESS in 10m 37s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/688e8050992d421a966a5eda3b3a510b : SUCCESS in 1h 09m 53s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/52fe5eea59ec4421ad85291ae3ebb489 : SUCCESS in 1h 09m 38s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/baea6fba0340467188dc5131e1ec9269 : SUCCESS in 49m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dda2fa9ced134b03bf54a2a6b96f4ebc : SUCCESS in 45m 44s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5137feec26fa49379f6f2491ad17e6a8 : SUCCESS in 50m 57s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/cff9b5f2ed924b76baff78fb0e1738b6 : SUCCESS in 1h 19m 37s\n- nova-next https://zuul.opendev.org/t/openstack/build/aef19989a9934471a9d1fbb7f588319e : SUCCESS in 1h 25m 18s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/3136eb5b1e334f789fc2682172218961 : SUCCESS in 8m 49s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/503e022a219649b1b9735b91a0a559b5 : SUCCESS in 20m 03s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d9f7eef74d51455b98391d4fad5e34d8 : SUCCESS in 18m 57s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d553797a8d5a447aae8b9b23aafd1c3c : SUCCESS in 22m 52s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/501c426e9ccd4fd69c8ebcb6c96651b7 : SUCCESS in 1h 04m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6719a167b79d4556af994c48ac634f1e : SUCCESS in 52m 03s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/949cf9972605440287338e9380790415 : SUCCESS in 30m 34s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/ba4658fc14e74151bb752fcaf9f258d2 : SUCCESS in 36m 40s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/597013aabd4c4ccfaf1d94d3feb00011 : FAILURE in 1h 13m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"aec60c14b9cddefa1267c1a3422fb0085355edbf","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-28 20:15:56.000000000","message":"Patch Set 5:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/33/824833/5/check/EMC_PowerFlex_NOVA/2d1ba60/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":5},{"id":"25169f56968cbb22e311824e0d9b9dc6d8a56c01","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-28 20:35:00.000000000","message":"Patch Set 5:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/5/iscsi : FAILURE in 2h 17m 01s","accounts_in_message":[],"_revision_number":5},{"id":"5b2c55e5c2be1fe995fc5361a6851229f46b71f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-01-31 15:53:38.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"919b65d2c484b919c2e22ec165c236c93b29f5df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 17:01:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76972e2e8d964f7a86bb0d6887349a6b : SUCCESS in 42m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9c0416178f34862840b56e4c4eb8c26 : SUCCESS in 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d70a54cf25f34df5eb762da7694f6aec8f9a7a32","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-01-31 18:23:50.000000000","message":"Patch Set 6:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/6/iscsi : SUCCESS in 2h 25m 27s","accounts_in_message":[],"_revision_number":6},{"id":"41e04438d9a634df2ea2320ab737a8e448426d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-31 18:24:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4979be8d4b2944fa8b1295a33a35e4d2 : SUCCESS in 1h 15m 36s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/b5e503d5a78d4abab30b43f2f45a81da : FAILURE in 12m 44s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa65b589b98f4e75b72e8f2dda5f84a5 : SUCCESS in 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e598ce65170d41448b27c297acb7e548 : SUCCESS in 18m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55bd2f65f3a543d69b32ea4bbee881ba : SUCCESS in 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4733de70b88d445b98ea931ce61ebc92 : SUCCESS in 14m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/228564f4db7d4a94b31496f785b24685 : SUCCESS in 15m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb46c75599744b7ab5a5a75efa5110f8 : SUCCESS in 18m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f910d4a47d4837bc602ad0d2df7f24 : SUCCESS in 14m 38s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/e69a39d82ed94696b939505b61bd55f9 : SUCCESS in 1h 06m 06s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38cc97860a584eaf94471bb9f1b74812 : SUCCESS in 1h 08m 15s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/bf4b889881254350abd93e9b8c582c72 : SUCCESS in 57m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e6f880c4381f44128fb4fe10ff1abda8 : SUCCESS in 42m 04s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/febcc8a2efae45249504c5d8c3bff641 : SUCCESS in 52m 22s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/10a8072354e74e909c1e392bd0a4958a : SUCCESS in 1h 09m 03s\n- nova-next https://zuul.opendev.org/t/openstack/build/cfedd0d0ce594c4f943fe6f1cbaf9107 : SUCCESS in 1h 34m 02s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/cc94ab0c905f421c9c4c3d1c3dedf203 : SUCCESS in 52m 46s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/6b17ddf4c355433ca0e7d04d95bf08d1 : SUCCESS in 10m 30s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/56897f863cff4db39803c809f1f1f363 : SUCCESS in 23m 44s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0854ba9d03174475bbfd38b85322e5cb : SUCCESS in 22m 46s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4efe6c17cd5c48bba4bcda0f7a0575d8 : SUCCESS in 23m 10s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/97452848fe8246dc854387ef7534f26b : SUCCESS in 1h 09m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/063d6fa3a8584ae2aa0582c4f165ee2d : SUCCESS in 49m 34s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/b5a59c2a15ec4dd48f810ab78cf6c0d9 : SUCCESS in 33m 03s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/2389e842d3b54341a69407c72b576300 : SUCCESS in 37m 07s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/f398e636f9a44ec9a3568bb3f529d155 : FAILURE in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e4619e5d566805d092301a3e4f8cc38ad876d68f","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-01-31 18:28:40.000000000","message":"Patch Set 6:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/6/check/tempest-dsvm-full-focal-py3/5805675/ : FAILURE in 1h 41m 10s","accounts_in_message":[],"_revision_number":6},{"id":"96e219e996e65e27ddfae730f5b1ca6878ff3da6","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-01-31 19:11:53.000000000","message":"Patch Set 6:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/33/824833/6/check/EMC_PowerFlex_NOVA/73e63e7/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":6},{"id":"88b09a089ba147b55e05b5aa5256a61a5acdedd8","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-01 13:35:48.000000000","message":"Patch Set 6: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"72b78b7bd4bb61ec27bb8bfa80c5ee3706955dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-01 14:58:13.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"835daa4b41435b5d4ea7bc6a29f63414fbf215a0","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-01 14:58:32.000000000","message":"Patch Set 7:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41450/","accounts_in_message":[],"_revision_number":7},{"id":"803bdd0d0d497b931c4993e1c17562bd29395524","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-01 14:59:40.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"03d4754479f64b536e0d93a49cb3b839081fc133","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-01 15:00:39.000000000","message":"Patch Set 7:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41450/ : FAILURE","accounts_in_message":[],"_revision_number":7},{"id":"a67b67f5760f564221de17f3fe57204aa81e3e4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 15:57:39.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df050330e96640e9b520b908ea188cce : FAILURE in 37m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7fde78f88484c428c4adc43f5a7f020 : SUCCESS in 41m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1983e11f6422b48006e43e458467f8b7045d7e53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 16:42:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3d36ed102acb4c5db1ca4adfdee15f99 : SUCCESS in 1h 19m 08s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/c20053793894482586f4102c8cae4f0e : FAILURE in 10m 14s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea0111946b1748fab89e1b185d2ca822 : SUCCESS in 39m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fea8acbb28d8416389a167a8b79c1b79 : SUCCESS in 17m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98911741a04d447d99cc7854d5ee8a91 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b570f3119446ad8e11aac2c7958143 : SUCCESS in 12m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/578ebd173d5047b4b11a5efa2870f567 : SUCCESS in 13m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9370c8eb59df473e8ee91a3a970f0447 : SUCCESS in 14m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698d6da8e9d04c70828630d1be1c3066 : SUCCESS in 11m 47s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fbcbd545c8ad440ba3d2d1bd1d5c0380 : SUCCESS in 1h 06m 24s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78d3b66eb51a438985d5eb7fd87d1312 : SUCCESS in 1h 10m 55s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/21261410d3af4e82ac783bd5d1d7d517 : SUCCESS in 1h 05m 08s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c56636acbb7a41ee80fa9cc15c477dbc : SUCCESS in 51m 57s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/52f2ef1be6da4b5ab427b9fff19314f8 : SUCCESS in 53m 21s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/c5d5147f91784e12b60f5d42c3183d59 : SUCCESS in 1h 06m 44s\n- nova-next https://zuul.opendev.org/t/openstack/build/9888583cd530442988320e58faf95635 : FAILURE in 1h 19m 22s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/b69708707ae745e483aa9f1889442911 : SUCCESS in 48m 52s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/b8c12c4b0f034a099eb74c7da068d52c : SUCCESS in 6m 56s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/1332532749d845e984166375ac365ce0 : SUCCESS in 21m 44s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62117ab8a2214ff688a84a7cf1f33146 : SUCCESS in 19m 48s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca6f2640c3444a65ba209f78667120df : SUCCESS in 29m 12s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/1cfaeb1dd33144b58fdb2ce3492c0b33 : SUCCESS in 1h 05m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/987061736c544650a1bb88ae64171550 : SUCCESS in 51m 30s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/4125ac46d38d4f109aa3e1c733bbfa35 : SUCCESS in 27m 51s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/f0dd8f344d6549f580ec364d6e62ed2b : SUCCESS in 37m 33s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/2b6e934bf8a34ff5bc6906ce376393a8 : FAILURE in 12m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ce05d3aacf9f32bfdcc702d19cec4e64f2330ad6","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-01 17:07:40.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/7/check/tempest-dsvm-full-focal-py3/17ea7eb/ : SUCCESS in 1h 55m 35s","accounts_in_message":[],"_revision_number":7},{"id":"1cfcfdcb33cd3ec1b5f6b70a31d07cf6808900dd","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-01 17:21:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fdab10073297ade4a59c7f793fbeb156b9106e8e","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-02-01 17:28:08.000000000","message":"Patch Set 7:\n\nFor rechecking only on the Cloudbase Nova Hyper-V CI, add a review comment with run-Cloudbase Nova Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/7/iscsi : FAILURE in 2h 26m 28s","accounts_in_message":[],"_revision_number":7},{"id":"6264b5591d8cd5db43bce52222c6783ce1ba2951","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 17:54:44.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/241ed0b115f048879ce4982636947f23 : SUCCESS in 30m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/564d1b0076a347209e8098a13f6c641f : SUCCESS in 18m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"af887d1b8dacac1d8042c64ceabe4cb8c8ee5d3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 19:03:13.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/761a831b92ed4c78978bd77e3267e29f : SUCCESS in 1h 22m 36s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e048808a7e84b92887133319bac5db4 : FAILURE in 13m 10s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b6df5540c1d4807a51212dfd665488e : SUCCESS in 46m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a02d39cff35480ba677c9fc387f6332 : SUCCESS in 15m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ade04d1da4c34f9cb4ffc9b072b7d8bb : SUCCESS in 17m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e68e0a62e66463696adb4b3cb398e1d : SUCCESS in 14m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31fb280abd074b18acd024405bfcd9de : SUCCESS in 13m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e9de007606944bab781701ba2a628e1 : SUCCESS in 14m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e81cb46a8c024449846f5cbf2712cc72 : SUCCESS in 13m 35s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/d8ea86e382c447009df8fb02ac5a02b3 : SUCCESS in 1h 18m 03s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d52cad03a9b4d73b1a2a153b3cf1328 : SUCCESS in 1h 25m 05s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/5a5aa0f6ac614fa6bd283bb3532a3e64 : SUCCESS in 1h 06m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6fdd9bcb0dfe424aa81d9be2a62a793f : SUCCESS in 54m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/f2eec82c76414ea1b6c6edcaab00de0e : SUCCESS in 49m 57s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/5ae4562da98044e787734984b9bbe199 : SUCCESS in 1h 34m 28s\n- nova-next https://zuul.opendev.org/t/openstack/build/7e0634e56cde49d1a58a079be439321a : FAILURE in 1h 24m 18s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/2844456a15144ee3b04bcb6f6d4f42ea : SUCCESS in 51m 07s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/ac649affe9ad4f2193ecb5ca8679c177 : SUCCESS in 6m 00s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/bb684347ac5c44598c5842f07e6eddd3 : SUCCESS in 21m 38s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa518f5591f84c2091abb34b328373a8 : SUCCESS in 18m 42s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/426d887cdf1b40e48f301eb91dcfde4a : SUCCESS in 24m 14s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/00df45ceb33a4f41ab41c47320e9a554 : SUCCESS in 1h 05m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81c816968bfe4204b1963c1b564ed1df : SUCCESS in 48m 15s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/0154b4be5cf5479bb5f03b4d0887f785 : SUCCESS in 22m 04s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/675fcfd1f91e4eb6b9755f2bdedaffc3 : SUCCESS in 40m 35s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/4ba5cc9e84cf4563a8a0914f324195ad : FAILURE in 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"09311aa3429def2155d7ea4237fe84aeaf0fe12b","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-01 19:55:07.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/7/check/tempest-dsvm-full-focal-py3/026a108/ : SUCCESS in 2h 21m 03s","accounts_in_message":[],"_revision_number":7},{"id":"213cd3e32d00d05ff36daedbafaf3b22a2a1f19d","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-01 20:29:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"10670908f59aa6ea786d1b18ed8d9fc8a7297b82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:02:09.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee8e35ec98ea410abf056c52df8adaac : SUCCESS in 29m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ab9ce5ade5c440782fd49a411d75e50 : SUCCESS in 17m 09s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"1b9b076de3de84d081eb2f281f3a222232665161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 21:49:01.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/2118e6f29a1b4eacaf8fc0d70e943c26 : SUCCESS in 1h 14m 53s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/eeeb75c0767b4b88a18862a66802dc96 : FAILURE in 9m 02s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/093c112a94ad476da1fc5f6b0db20240 : SUCCESS in 44m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0371c9c8c56444058595ff57444c135c : SUCCESS in 16m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07de63863317437f814fda42925e3171 : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e27f9f9d7ce441797a4600361730fac : SUCCESS in 12m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a84575711044efc85ff6b2bd8fb8ef1 : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2c91e4f1db94c6099d3d8355c59f3d4 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a0aaf936eb47249f3098238562c07b : SUCCESS in 10m 14s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/166e33e5822e47cc9181619bd422dae0 : SUCCESS in 1h 09m 00s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/98666e58189c4727b0e998e00e2c227b : SUCCESS in 57m 13s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/721985bf03874117ae93732b9d8bc019 : SUCCESS in 48m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bdb1238ca9e7429e9f68129d0072965e : SUCCESS in 49m 28s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/779e9a0d35774c748b65418c94cef9b1 : SUCCESS in 56m 52s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/ac80e3028fd74d7e88798eab199b7a68 : SUCCESS in 1h 10m 24s\n- nova-next https://zuul.opendev.org/t/openstack/build/1e9119302cd241098354f750403fd6c9 : FAILURE in 1h 06m 23s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/68b7f5fd67f044bd8f7fc2efc503474a : SUCCESS in 47m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/4c7f22a48c8d4cd0884ed3f41ab526d1 : SUCCESS in 6m 00s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/3553cd2d02b34318bc93f7a82fc08f8f : SUCCESS in 20m 51s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb2d5ee018e246c6b7fef591abf2b4b5 : SUCCESS in 24m 16s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a2772291be843a2810c808cb30485e0 : SUCCESS in 21m 46s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/e53978ccd83b4e2c8049fddcec30ce8f : SUCCESS in 1h 03m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7906d32ad1914d27b9cc97a235d6d6fb : SUCCESS in 45m 27s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/2db1f94e7e464b9bbe9ff98eccc841ae : SUCCESS in 33m 07s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/32b3bb04e7f746f9a4f26a3a0e194b70 : SUCCESS in 32m 09s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/91a196e5f4c6452ab1a065894b2b70f4 : FAILURE in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"5170da2125af043b9f723cef47e302ae66f9762d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-01 22:40:33.000000000","message":"Patch Set 7:\n\nBuild failed. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/7/check/tempest-dsvm-full-focal-py3/4e09452/ : FAILURE in 2h 04m 06s","accounts_in_message":[],"_revision_number":7},{"id":"92c12c7558e91ddb9b33121346191da0e3593d30","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-02-01 23:22:13.000000000","message":"Patch Set 7:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/7/iscsi : SUCCESS in 2h 49m 06s","accounts_in_message":[],"_revision_number":7},{"id":"77405d8ed2d02df4e6996a60fb2d2ca1f77ad2b6","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-02 12:46:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"2b12add851b1bc9e24b5812339bc8692f9243e07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 13:18:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e65c30c84bd24100a16c7422743370da : SUCCESS in 26m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7e964a76267d450eae06ba31375de365 : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a1808a1bbc87910df0e40cc310513ae17c700e56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 14:34:46.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/4ed3a43ce61049c3856e15c0eddf9f8e : SUCCESS in 1h 13m 05s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/ed19abce4c5747ac933b8aee351f8051 : SUCCESS in 1h 21m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b3a8c1eee0d479eb89bf24a48ea8281 : SUCCESS in 45m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f382225bff6f46899d7043dd539d44b7 : SUCCESS in 17m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c4630d73c7d499c8642490ac0cd3e9f : SUCCESS in 13m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e06524a36b5f4a2494eda195f21c886e : SUCCESS in 11m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46f467286fb34c40ab474d65fe5c7182 : SUCCESS in 14m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad9610c8ac484be5a57925cbf5597d6f : SUCCESS in 12m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc74c96b09e54981856ffdb3688c23ff : SUCCESS in 11m 20s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1e17e34e8a014df9a7ce1ef2e79a38ea : SUCCESS in 1h 12m 31s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de520c1cc0014b8391b16537b8a38250 : SUCCESS in 1h 12m 56s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/2edbf5d66f74492ea69ec336e33ac37c : SUCCESS in 51m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4ac1b2d976ac454d824464ccae786889 : SUCCESS in 45m 09s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/e9110fa494f742dc9fa471774e1ad831 : SUCCESS in 46m 12s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8b34d936e14d4a19b1673db543f9c21b : SUCCESS in 1h 40m 41s\n- nova-next https://zuul.opendev.org/t/openstack/build/f1fc3b37e45b47d6b047aa884caa3910 : FAILURE in 1h 30m 31s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/e65b35c8c0b943ba91afc5e27a606793 : SUCCESS in 54m 12s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/709935e8f5464e50855a66f16dd69f08 : SUCCESS in 5m 55s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/6d2e56b4890b433c89814a0947102b04 : SUCCESS in 18m 58s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0df0c449be3a42c0be7a318f8dfb30fb : SUCCESS in 19m 32s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7cca6247e794c478e880d784250ad9c : SUCCESS in 18m 09s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/40cfa4f5b4284b3d9c4db6bbab647cc8 : SUCCESS in 1h 05m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50ab8bb580b441bb8b34a6d136d233a2 : SUCCESS in 54m 00s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/3c56093eea04498aba3bbf7807f1144e : SUCCESS in 27m 12s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/c95bf5fe22bb45468e33f534f6b934dc : SUCCESS in 39m 41s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/adb03f8e9a374c42a68938869e96cf36 : SUCCESS in 1h 30m 38s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e444ad79bf5a0a7fcec7ca43bb087f59a9712891","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-02 16:16:29.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/7/check/tempest-dsvm-full-focal-py3/1367a3f/ : SUCCESS in 2h 50m 28s","accounts_in_message":[],"_revision_number":7},{"id":"a50f14a50f739ff1315829a4279e256dd3e90b77","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-02 17:19:11.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ceac4439c2e621a43fd93caa72a06d89d00afbed","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-02 17:19:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bf416b94ab03f4aa1db43220468d1405f511ce1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 17:55:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d0a3d2afe974e43923c25689475a02b : SUCCESS in 33m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9505ed486613497b9ed47f318883c63d : SUCCESS in 20m 10s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"726ea362b1d7c68120e59c5443b9196face00055","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-02-02 18:02:44.000000000","message":"Patch Set 7:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/7/iscsi : SUCCESS in 3h 07m 32s","accounts_in_message":[],"_revision_number":7},{"id":"1ce4d82557f1ba3d60438d01483e7fffb82594e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-02 19:13:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/8e394f9ec99341e79624482ee0b36b51 : SUCCESS in 1h 20m 35s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/7aeb819591284d53a0bc271c03a67487 : SUCCESS in 1h 39m 40s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac3557f807e64a018286fd4f25799b1d : SUCCESS in 58m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50ca60d835f3410d8b46dfcd0d53301e : SUCCESS in 16m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0624f2e8d41b4f5c8325d7b01ef781e7 : SUCCESS in 19m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88232f0bc4164c0aa52d4cd963df93e3 : SUCCESS in 11m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a706e3bf35544e9da0d2130b3b1b770f : SUCCESS in 13m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/386d33ea29414375a659938b8e9bf351 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c514bc26e6a4265aa2e471c2db95380 : SUCCESS in 10m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0b6c6754054c47f999135b9c9659dfc9 : SUCCESS in 1h 23m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d6a9ef5e82d410793ea04ff7a12067a : SUCCESS in 1h 15m 47s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/ea7731edb6ca46ac9fbec4620a7db712 : SUCCESS in 1h 01m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/603e5d76dd344aa8b4edcdd349db960b : SUCCESS in 45m 05s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/5ada39fb55ca4bd896cdf28d35ef7099 : SUCCESS in 47m 46s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/60a7d42f0e6c4170818fabb298806e66 : SUCCESS in 1h 26m 33s\n- nova-next https://zuul.opendev.org/t/openstack/build/d7033b0b72d74e33a0ad3c044e1f186d : SUCCESS in 1h 41m 17s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/c76c9556fd404fd4be649b0fbbb4bf9a : SUCCESS in 55m 53s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/87250efd6117412d9aaf25a7aee5d5cb : SUCCESS in 6m 48s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/2aa1507e4a7b4486b223fb9661660863 : SUCCESS in 28m 30s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a758ae6bf25468ca82debfa2d209dca : SUCCESS in 20m 27s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0133f3ac1513472bad7859953a2c3350 : SUCCESS in 20m 49s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/0026fca6224245879180b429b6f6e12f : SUCCESS in 1h 08m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d5b53e4cc9747e3a54fb939b0580961 : SUCCESS in 40m 55s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/46a1a9112b3d4b21ad3f0b249a690a62 : SUCCESS in 31m 53s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/e14dd94c8402419d932af56608c2c239 : SUCCESS in 36m 15s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/cb03309bc68b4db48ee9e21b61ad9acf : SUCCESS in 1h 26m 29s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4f1d28d5222224c529e8443d8931fae0e8223d0a","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-02 19:13:40.000000000","message":"Patch Set 7:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/33/824833/7/check/EMC_PowerFlex_NOVA/bfc1879/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":7},{"id":"5ed76fc8bd33fde1bd64f1878c0a5e8ee3bb6e1d","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-02 20:23:09.000000000","message":"Patch Set 7:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/7/check/tempest-dsvm-full-focal-py3/730a64a/ : SUCCESS in 2h 23m 51s","accounts_in_message":[],"_revision_number":7},{"id":"b37c79830039d192e31240daf9c5b448b5df5097","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-03 13:25:41.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"edcf75c5ff0cb81506af59abf129c9a7870fb289","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-02-07 13:39:05.000000000","message":"Patch Set 7: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"69fe92e84733ed420d059d2c074b79fc04d35006","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-02-07 13:43:25.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"07ebb0196631d121f6c79040d0158139af13b5a0","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-07 16:25:32.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1a4eb47bd99229058044102e7b3c778998996c88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-08 11:33:26.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"62a02b37dbffcddb70ec412acddaac3849ae96e4","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-08 11:33:39.000000000","message":"Patch Set 8:\n\nBuild Started http://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41503/","accounts_in_message":[],"_revision_number":8},{"id":"b6c0f83b731aec4342064b5c1134c90d0125fc20","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-08 11:33:40.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e6f13d300fcc2d5448f166f7f4bb990e08d4daa2","tag":"autogenerated:jenkins-gerrit-trigger","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2022-02-08 11:36:02.000000000","message":"Patch Set 8:\n\nBuild Failed \n\nhttp://p3-ci-int.eng.vmware.com:8080/job/ext-nova-zuul/41503/ : FAILURE","accounts_in_message":[],"_revision_number":8},{"id":"4f8de416d1280d02d53684a6a27f767b46c9f0ef","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-08 12:12:34.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"04469172f2c70df1feafa8c62ed7b5d200ac9a7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 12:33:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca96664760f64ab38cb1e37d01d3bd6f : SUCCESS in 35m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/041aa29b0a4b41bda6fbcc20599b9707 : SUCCESS in 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"65d5ae38cd7982d4518a5fff8286c4e55fa071d8","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-02-08 12:43:30.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fbcc39bf0f23da55c2bf957160d4132ea330c557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-08 13:24:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/3551d2db047c4b6f8d813124842ca602 : SUCCESS in 1h 24m 36s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a0eaaa9440347aca5b383e6bf113182 : FAILURE in 1h 35m 02s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/53cd2ed1caff48789f86a7bd1ec574af : SUCCESS in 1h 01m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c38599075b69483d8570f9935e14599b : SUCCESS in 16m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61c92fdaaa8845acacbf795e6c2a9182 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b0b7af01e3b4470a161d1afeecb2309 : SUCCESS in 13m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/926a9bc833df486eabc2acd8b6dd2951 : SUCCESS in 11m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/586f86bd0b1a4ada9ba2be09c6554294 : SUCCESS in 13m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e70c514bf364c30ad182bd2addc6048 : SUCCESS in 10m 33s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/ee9392dff28f4f21bd79b8ac82790892 : SUCCESS in 1h 31m 18s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/88546d578d024c36a0ae4c609c93b7b1 : SUCCESS in 1h 11m 24s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/899697e2649a4cf9a24d02316451373d : SUCCESS in 54m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6fd420f9f1534acbb71127e9ea820f9e : SUCCESS in 48m 51s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/7bed7fb97bf8434fa6de90313c966de5 : SUCCESS in 47m 25s (non-voting)\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/d4d6df0dd09745ccaa0b5301d7526d8a : SUCCESS in 1h 20m 46s\n- nova-next https://zuul.opendev.org/t/openstack/build/2cef20e2260743f1afba99be14f3b4b1 : SUCCESS in 1h 32m 59s\n- nova-ovs-hybrid-plug https://zuul.opendev.org/t/openstack/build/5d03243661f343368e14b247b8773c42 : SUCCESS in 48m 39s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/8eec9b98ff9e4f6784f8e3c93d2d5caa : SUCCESS in 6m 03s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/6d233350244a4e2e9069367f1d12771f : SUCCESS in 28m 47s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9951eff3e7ae42db8189cc8fb70c14c5 : SUCCESS in 18m 10s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3efc83363f224011aaa5e1f78fe5e108 : SUCCESS in 21m 47s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/d6881385d36246579b030558043da92b : SUCCESS in 52m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df25a09f6a104ef09f921acea8481d17 : SUCCESS in 51m 53s\n- cyborg-tempest https://zuul.opendev.org/t/openstack/build/561ce20e83c0420896eb833f7bca5ec1 : SUCCESS in 30m 56s (non-voting)\n- barbican-tempest-plugin-simple-crypto https://zuul.opendev.org/t/openstack/build/9bcbf3e5b9854f28ad05019877586492 : SUCCESS in 35m 21s (non-voting)\n- tempest-centos8-stream-fips https://zuul.opendev.org/t/openstack/build/a2221893f590454da8f8778cd61a20b9 : SUCCESS in 1h 43m 29s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a3ed20454fcb3ebec6c316c87d98d5caf06a59ee","author":{"_account_id":15941,"name":"DellEMC PowerFlex CI","email":"emc.scaleio.ci@emc.com","username":"emc-scaleio-ci","tags":["SERVICE_USER"]},"date":"2022-02-08 13:26:56.000000000","message":"Patch Set 8:\n\nBuild failed.  For information on how to proceed, see https://docs.openstack.org/infra/manual/developers.html\n\n- EMC_PowerFlex_NOVA https://elab-os-logsrv.delllabs.net/33/824833/8/check/EMC_PowerFlex_NOVA/b4fa7e2/EMC_PowerFlex_NOVA/None : NOT_REGISTERED\n\nLeave a comment with \u0027run-DellEMC PowerFlex CI\u0027 to trigger a recheck. For more information about CI, please see https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_PowerFlex_CI","accounts_in_message":[],"_revision_number":8},{"id":"6562e9dae47c9e8114e773a46a7d95274eb231de","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2022-02-08 13:27:51.000000000","message":"Patch Set 8:\n\nBuild succeeded. Test completed on IBM PowerKVM platform. For rechecking only on the IBM PowerKVM CI, add a review comment with pkvm- recheck. For contact and more information, see https://wiki.openstack.org/wiki/PowerKVM\n\n- tempest-dsvm-full-focal-py3 https://oplab9.parqtec.unicamp.br/pub/ppc64el/openstack/nova/33/824833/8/check/tempest-dsvm-full-focal-py3/b782bd1/ : SUCCESS in 1h 48m 19s","accounts_in_message":[],"_revision_number":8},{"id":"b0d39c759776429919eda75c4a927c197445bdd9","tag":"autogenerated:cloudbase-ci","author":{"_account_id":26515,"name":"Cloudbase Nova Hyper-V CI","email":"nova_hyperv_ci@cloudbasesolutions.com","username":"nova_hyperv_ci"},"date":"2022-02-08 14:14:45.000000000","message":"Patch Set 8:\n\nBuild succeeded. For checking Nova with Ceph, add a comment with run-Cloudbase Nova Ceph Hyper-V CI\n\n- nova http://cloudbase-ci.com/nova/824833/8/iscsi : SUCCESS in 2h 40m 07s","accounts_in_message":[],"_revision_number":8},{"id":"6f220af8b631de6d2e2360f83fc514e6a6c920d7","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2022-02-09 10:36:03.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"127571bda53700982c0170f70c50d9be356daee6","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-02-09 16:13:25.000000000","message":"Patch Set 8: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"986db362b5a7d9f501841c8cb3f060d00ee2bc8c","author":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"date":"2022-02-09 16:27:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"92c6139c2064f7c4db6e42a917af3afa409dbcfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 16:33:31.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c14a50314833429efae3e635d4991a2a4daedc00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:59:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- tempest-integrated-compute https://zuul.opendev.org/t/openstack/build/499dc2fd24944ba8a1bd92fbe5f1cf0f : SUCCESS in 1h 15m 17s\n- tempest-integrated-compute-centos-8-stream https://zuul.opendev.org/t/openstack/build/88e07d8de0454349a9e97fe35f58060f : FAILURE in 1h 29m 18s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08b390cd56ad42b3a99c8980f702e576 : SUCCESS in 46m 00s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efc1a316d6bb4e948d91642e0cca89d5 : SUCCESS in 14m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a039515c589d423ebf0a0f39dda044de : SUCCESS in 16m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14013a9f8d894d6a95535d0e52a51817 : SUCCESS in 13m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00171ec88f634f3dbc9499768c2bfaec : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d094b06bda524ea98059fbcd7c314b83 : SUCCESS in 10m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d5a344b69fa54f868dd6e24859dae697 : SUCCESS in 51m 55s\n- nova-live-migration-ceph https://zuul.opendev.org/t/openstack/build/05828f94d5b34aefab52903de9349a4d : SUCCESS in 46m 56s (non-voting)\n- nova-tox-functional-centos8-py36 https://zuul.opendev.org/t/openstack/build/79b4ca6ab2ef4f59b4963f4941904de9 : SUCCESS in 23m 32s\n- nova-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b195eac9f5b84a668e2aca6f4cfc22f2 : SUCCESS in 23m 15s\n- nova-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8323a73d1af4b3fb33a22b690d9d453 : SUCCESS in 20m 21s\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/fe8b0cc62e7544c38f759e4adcde626b : SUCCESS in 1h 37m 27s\n- nova-next https://zuul.opendev.org/t/openstack/build/ea302bad5ba24bd0ad6ac62e7fd4dd49 : SUCCESS in 1h 43m 23s\n- nova-tox-validate-backport https://zuul.opendev.org/t/openstack/build/f1a9a648083347569839972aa4f7e2ea : SUCCESS in 7m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0ca1a31cb2e4211948471b564a5fc97 : SUCCESS in 1h 27m 46s\n- neutron-linuxbridge-tempest https://zuul.opendev.org/t/openstack/build/694f53194cee46938d3c27ee9c0b25b0 : SUCCESS in 52m 17s\n- nova-grenade-multinode https://zuul.opendev.org/t/openstack/build/5f4019e1e1264c349a63face375515a4 : SUCCESS in 1h 06m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d08021a391f440ea1cf05352b46a870 : SUCCESS in 56m 58s","accounts_in_message":[],"_revision_number":8},{"id":"e96beb41a79bae4d17033ae31a4487f7536bf6ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 20:59:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"2219f6a49d03fa5573f9fe1160dbabca936e0abd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-09 21:03:31.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31e8a8bf484466b809b8415d7f3f884 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"1f71696ecc2cd1abfc30f2f03f3c4857e51b9fbf","revisions":{"dd141c79024a55c5fe536c9596616bb490057002":{"kind":"REWORK","_number":1,"created":"2022-01-16 20:56:26.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/1"}}},"commit":{"parents":[{"commit":"c2929f74dd2800a1d3c870aff796421908753b7e","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2929f74dd2800a1d3c870aff796421908753b7e"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:56:38.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nDepends-On: I6445433142286728a8c7efadcf80d07082d60bc3\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd141c79024a55c5fe536c9596616bb490057002"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/dd141c79024a55c5fe536c9596616bb490057002"}]},"branch":"refs/heads/master"},"c4e31fb0e73c734c0aadcc3db21121117f7c28fb":{"kind":"REWORK","_number":2,"created":"2022-01-27 11:14:33.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/2"}}},"commit":{"parents":[{"commit":"c2929f74dd2800a1d3c870aff796421908753b7e","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2929f74dd2800a1d3c870aff796421908753b7e"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-26 14:36:52.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4e31fb0e73c734c0aadcc3db21121117f7c28fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c4e31fb0e73c734c0aadcc3db21121117f7c28fb"}]},"branch":"refs/heads/master"},"0696bd112bdb02678c77f9c4216e9bc3bdbe9262":{"kind":"REWORK","_number":3,"created":"2022-01-27 13:20:11.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/3"}}},"commit":{"parents":[{"commit":"c2929f74dd2800a1d3c870aff796421908753b7e","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/c2929f74dd2800a1d3c870aff796421908753b7e"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-27 12:49:17.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0696bd112bdb02678c77f9c4216e9bc3bdbe9262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/0696bd112bdb02678c77f9c4216e9bc3bdbe9262"}]},"branch":"refs/heads/master"},"a695662932d5bcce0a0b32813071c20f6c0ffd5c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2022-01-27 21:25:49.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/4"}}},"commit":{"parents":[{"commit":"e43f72c90861da600bc2cd7bbebce4353cf241a6","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e43f72c90861da600bc2cd7bbebce4353cf241a6"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-27 21:21:59.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a695662932d5bcce0a0b32813071c20f6c0ffd5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/a695662932d5bcce0a0b32813071c20f6c0ffd5c"}]},"branch":"refs/heads/master"},"89966c1219dda4963d9b1bbc47d3acc6b742d85b":{"kind":"REWORK","_number":5,"created":"2022-01-28 18:16:12.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/5"}}},"commit":{"parents":[{"commit":"e43f72c90861da600bc2cd7bbebce4353cf241a6","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/e43f72c90861da600bc2cd7bbebce4353cf241a6"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-28 18:07:12.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89966c1219dda4963d9b1bbc47d3acc6b742d85b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/89966c1219dda4963d9b1bbc47d3acc6b742d85b"}]},"branch":"refs/heads/master"},"f766125471dca45dc4501ae90ed4140da8c9c87a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2022-01-31 15:53:38.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/6"}}},"commit":{"parents":[{"commit":"6c6cacc8df5f66c25f25fdc877a948dc9c47a485","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/6c6cacc8df5f66c25f25fdc877a948dc9c47a485"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-31 15:50:29.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f766125471dca45dc4501ae90ed4140da8c9c87a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/f766125471dca45dc4501ae90ed4140da8c9c87a"}]},"branch":"refs/heads/master"},"00b9afd7772d2331eb0b92873b4ae18eebdb36d8":{"kind":"REWORK","_number":7,"created":"2022-02-01 14:58:13.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/7"}}},"commit":{"parents":[{"commit":"ab49f97b2c08294234c7bfd3dedb75780ca519e6","subject":"[yoga] Add PCI VPD Capability Handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/ab49f97b2c08294234c7bfd3dedb75780ca519e6"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-01 14:57:33.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nAdds methods to retrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00b9afd7772d2331eb0b92873b4ae18eebdb36d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/00b9afd7772d2331eb0b92873b4ae18eebdb36d8"}]},"branch":"refs/heads/master"},"1f71696ecc2cd1abfc30f2f03f3c4857e51b9fbf":{"kind":"REWORK","_number":8,"created":"2022-02-08 11:33:26.000000000","uploader":{"_account_id":24824,"name":"Dmitrii Shcherbakov","username":"dmitriis"},"ref":"refs/changes/33/824833/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/nova","ref":"refs/changes/33/824833/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/nova refs/changes/33/824833/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/nova refs/changes/33/824833/8"}}},"commit":{"parents":[{"commit":"b6fe7521afa8d42febc68f5f79782f7bcc3b568f","subject":"Merge \"docs: Follow-ups for cells v2, architecture docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/b6fe7521afa8d42febc68f5f79782f7bcc3b568f"}]}],"author":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-01-16 18:53:49.000000000","tz":180},"committer":{"name":"Dmitrii Shcherbakov","email":"dmitrii.shcherbakov@canonical.com","date":"2022-02-07 20:38:41.000000000","tz":180},"subject":"[yoga] Include pf mac and vf num in port updates","message":"[yoga] Include pf mac and vf num in port updates\n\nRetrieve PF mac and VF logical number at runtime for\na given VF PCI address and include them in port updates to Neutron.\n\nImplements: blueprint integration-with-off-path-network-backends\nChange-Id: I83a128a260acdd8bf78fede566af6881b8b82a9c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f71696ecc2cd1abfc30f2f03f3c4857e51b9fbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/nova/commit/1f71696ecc2cd1abfc30f2f03f3c4857e51b9fbf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
